)]}'
{"id":"openstack%2Foctavia-lib~896055","triplet_id":"openstack%2Foctavia-lib~stable%2F2023.2~I9079030a5fef9dc44da242adab3c568666777451","project":"openstack/octavia-lib","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-09-26 15:32:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9079030a5fef9dc44da242adab3c568666777451","subject":"Fix a possible receive timeout","status":"MERGED","created":"2023-09-21 12:31:12.000000000","updated":"2023-09-26 15:34:04.000000000","submitted":"2023-09-26 15:32:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896055","meta_rev_id":"4aafe907983e8639f44ccf7534a9db911011d130","_number":896055,"virtual_id_number":896055,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 15:32:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2023-09-21 12:53:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2023-09-26 14:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-26 15:21:08.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-26 15:21:08.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 12:31:12.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-09-21 12:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 12:53:56.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-09-26 14:09:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-26 15:21:08.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"a25a9d139eace0442db2689e2c30f8f54844d056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-21 12:31:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c4a836e8bdbf6f95581a765afab8d6323e6089c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 12:42:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/440e8006f7f54a5e9b9e8cb2af0f4317\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6423be01a57f47059d19773b67d0bba3 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eec770cdf134ee3b99b4173a20bc484 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aaf4f0c0e274db78cd8dfc9af4a39c9 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c493f5bb2be048058ba335e25f73ee28 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9365266e29e840b0b2d81d4cff476694 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3d1a03241eb4755898b99f766b7fd59 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6921f2f365e4084be4be246e727ab24 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2183318e7aa844d7b44acb146eacb1c9 : SUCCESS in 2m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/513f3a30ddec495f8637ef9884f30694 : SUCCESS in 6m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0c5c7d3d58e847c3a091146ab44219d2 : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a58b9aff002c7d0946db887bf6067aa6ff2e7976","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-09-21 12:53:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"044323500d064fbf7ab99fa52548ca4f85d7b0a8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-26 14:09:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"127283f47fb9d2df693193d6f8c2eac5b4b6017b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-26 15:21:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2bed50ce22dd32fc34ca8dfdc82d61630adbed3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:21:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b19749b97250e69a43a1053140335ab4aa4bd373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:32:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee523e612b6549b6b5a900d94ad5fde1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4db889b140747259824be0063791e33 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61270422a8cf4f9493fcac8fba9d47e3 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddb7f923b92d40fcae4b2ab0801106b7 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/104885abd6b149d5bc54e605a4c3f05d : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084d967652c04cc5aa7431d6f59b2379 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36ab3ae15a5e42cd9adab0138f9cbd54 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"254fd42c0dc05fa5558664cdec559e5f8d1c2d9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:32:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4aafe907983e8639f44ccf7534a9db911011d130","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:34:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95508dcac3064c2caf11c67d10e181dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8059ae1e9fb74df3a1fdde3b0daec3b9 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd0a737006b34ae1b8cea05819fb89c5 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1142a99a982cc6919143734680b3b0736b1c0b1","revisions":{"f1142a99a982cc6919143734680b3b0736b1c0b1":{"kind":"REWORK","_number":1,"created":"2023-09-21 12:31:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/55/896055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/55/896055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/896055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/896055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/896055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/55/896055/1"}}},"commit":{"parents":[{"commit":"17bef334d137837965831341378e4733f96393b1","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/17bef334d137837965831341378e4733f96393b1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 00:13:39.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-21 12:31:11.000000000","tz":120},"subject":"Fix a possible receive timeout","message":"Fix a possible receive timeout\n\nThis patch fixes a possible receive timeout caused by a slow response from the\ndriver agent. For example if the database is very slow.\n\nCloses-Bug: #2032890\n\nChange-Id: I9079030a5fef9dc44da242adab3c568666777451\n(cherry picked from commit 2a84a218ef1f683d5584784b7d9607e7453b0729)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f1142a99a982cc6919143734680b3b0736b1c0b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f1142a99a982cc6919143734680b3b0736b1c0b1"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
