)]}'
{"id":"openstack%2Foctavia-lib~896059","triplet_id":"openstack%2Foctavia-lib~stable%2Fxena~I9079030a5fef9dc44da242adab3c568666777451","project":"openstack/octavia-lib","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-09-26 15:33:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9079030a5fef9dc44da242adab3c568666777451","subject":"Fix a possible receive timeout","status":"MERGED","created":"2023-09-21 12:32:06.000000000","updated":"2023-09-26 15:35:05.000000000","submitted":"2023-09-26 15:33:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896059","meta_rev_id":"3a2ce362dae2061cb70084eb00ad8e289110a62b","_number":896059,"virtual_id_number":896059,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 15:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-09-26 14:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-26 15:22:10.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-26 15:22:10.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 12:32:06.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-09-21 12:53:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 14:11:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-26 15:22:10.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"23d9a6bf84908d523c66d71bb795720a4755ad19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-21 12:32:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4fc68bdab8f52da2565009612c3c6c39f3b7050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 12:53:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96d4935c0a084e0fb4f47e9caf1d6673\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83863c1aa429450cb9085e668e6122e7 : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/078248f9ea3342f2bcc35a4667da66b5 : FAILURE in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0679d70e91894ab78b02150f0b7ae676 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/231bf7a54ac34a7daff3a8de0fa3deb1 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7a18f9928aa49fd8ee635e7a764ec2c : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd669d5ad44b423d8b44dc21d5ed7b2b : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f580544a2f4fed88aebd93dd08cab3 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69d8fb1c50894bb998a05bad512d6957 : SUCCESS in 2m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e52d0dfae9244e138319b7f6ec26f3a9 : SUCCESS in 6m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b8c1042175424c118c677106e87a380e : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"279e590b4750a3c41d7d35a4a44d2fbceb6284cd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-21 15:20:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35849a545fd18b67f4904eeebdecd1dbd3ec5b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:34:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52cf38e01a114897bb93f7bf824139d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fb21c1e5af148e7b76667be40e8e768 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2503bffe0fd148c2b0521833480e0f72 : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e24a1c8df73c40f1b3d2fcafe0a1badd : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d70e1f84485041fabf66394c02550a02 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7bc1beb08784f829cb4974645ecd056 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f852e282cf4068983a54ba609b62f2 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddc00bd741704e9982b4d48f8de23211 : SUCCESS in 2m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aac938c649ff4a9ba08afa2d90c70651 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/710bc8b84ab74f648853498f0726dcad : SUCCESS in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ed67572be0588bbdb0aedec45a2312b499015b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-26 14:11:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30886259bdd7161aecb2e2aaeb26cb35114f52ab","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-26 15:22:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20d289a14e2a1d2713bfa0ecdaf1a5098201f69d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:22:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b2eb906cdb1e2d3052856c197b7eb8a7ee4cf18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:33:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cad04f74f31404299d723fc3348a73d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e52696925e241658ff7d055774c1ae2 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2360cd952d64aedbc86caced2d2edd3 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9ad0876e4c84c39b80c0dc56dbea8d8 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4bc5eaf41d46aea17bf1294bcf57e2 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d4111ae06c9413baeb178fcf3bf42f0 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6ed6a7e082923405fab2f640262aaf548114ae0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:33:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a2ce362dae2061cb70084eb00ad8e289110a62b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:35:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d779caba935a454f8f2d8a6e8d2cb631\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218d58e223814c3dadb23c8c0b269f3a : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2dd1992dd1e411691ac770729287178 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c2d5cdba50763c2f37bf173b9113f36dedef4ad","revisions":{"7c2d5cdba50763c2f37bf173b9113f36dedef4ad":{"kind":"REWORK","_number":1,"created":"2023-09-21 12:32:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/59/896059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/59/896059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/896059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/896059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/896059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/59/896059/1"}}},"commit":{"parents":[{"commit":"d7eeb68642638f74d48cbe67398085a630c51369","subject":"Fix new pylint warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/d7eeb68642638f74d48cbe67398085a630c51369"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 00:13:39.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-21 12:32:03.000000000","tz":120},"subject":"Fix a possible receive timeout","message":"Fix a possible receive timeout\n\nThis patch fixes a possible receive timeout caused by a slow response from the\ndriver agent. For example if the database is very slow.\n\nCloses-Bug: #2032890\n\nChange-Id: I9079030a5fef9dc44da242adab3c568666777451\n(cherry picked from commit 2a84a218ef1f683d5584784b7d9607e7453b0729)\n(cherry picked from commit f1142a99a982cc6919143734680b3b0736b1c0b1)\n(cherry picked from commit 5ec74a4dbeda9b65fbe3cdae078fca713a1dad74)\n(cherry picked from commit 3f764c5d2b17568aebc5a551de08f9d489d62519)\n(cherry picked from commit 450c3b9f4739d80f859fa7b7ecc64fda35cb983d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/7c2d5cdba50763c2f37bf173b9113f36dedef4ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/7c2d5cdba50763c2f37bf173b9113f36dedef4ad"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
