)]}'
{"id":"openstack%2Foctavia-lib~919722","triplet_id":"openstack%2Foctavia-lib~master~I1c3554d31b9066c5abdc958985de58f6f5a35d32","project":"openstack/octavia-lib","branch":"master","topic":"custom_sg","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-11-06 08:35:41.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-11-06 08:35:41.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-11-06 08:30:26.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I1c3554d31b9066c5abdc958985de58f6f5a35d32","subject":"Add vip_sg_ids to the data model","status":"MERGED","created":"2024-05-15 11:19:43.000000000","updated":"2024-11-06 08:36:39.000000000","submitted":"2024-11-06 08:35:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"919722-custom_sg","meta_rev_id":"b4407461b9a4229004652c448786d1e36909ad5c","_number":919722,"virtual_id_number":919722,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-06 08:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-11-06 08:30:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-11-06 08:30:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 11:45:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-23 15:40:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-10-28 12:29:11.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"cb460089ed425a86640db5ec10a22cc5b9cb670e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-15 11:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f859acb8d8923e6243ef5699266db999ffbb940","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-15 11:19:57.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"da48c94468ca5342eaf636fa6ceab90338a2a414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 11:45:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1d7eacff45b48c3b7be9d7e9b6943d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20d6d7443e564aeebd376ae2b33fc03d : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc43ea437074283bdce287e4e6b7dfd : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bfaa37a85b340a9bb3acd515fbf0177 : FAILURE in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14e85484d66a4ed99561b253ae230c4e : FAILURE in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cddac3510c04f03a55be07795153a96 : FAILURE in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3c46a553724584bc8fee8515430147 : SUCCESS in 4m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ca64a59a2dd540c981e2ec7f0bc1d873 : SUCCESS in 5m 46s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b2325e19d58b41b0b4a84ae4ecb4c2be : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9c33ae09eb5aa5d017b84831765ce8ac43d6ad5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-24 08:20:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6942dc96593d79ede37e3a0f761766f49c2f0219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 08:31:40.000000000","message":"Patch Set 2: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7ab34feb4024729b77d4e0e93d62d3a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/041abc7034d84e02b8c96b54a78d6f59 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/984b6e0c8f7c4132b81a9b2c70a63c32 : FAILURE in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98f01de2221f49189994c89021f9706e : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15a38b4e0592460486cdc70c5d178640 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f53f5017861248c0ad410d9f61fcc12d : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b5e8c9c90a447c91c9cf1d2dbc647c : SUCCESS in 4m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0cc3f60debdf4131845359b59c582daa : SUCCESS in 5m 19s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0feffb214bba4420ae451a826e116f99 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d55fea6579de7d7d16effecc10ad39f7bc454406","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-11 07:58:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ff8e542b66c90d696f0b328545d0820691ef7f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 08:15:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34d7be8e5c884b13b69dd8992f230938\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a44525d74cb140a89c7292cce727b81b : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b440d3b048469bae3cb4b6815a2776 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1749fd26666431a9e0363544766bfef : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3068bb5265a46438b1d67a78d712b5c : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f6e2a0c7cb45bdbbf799cd10d9ffe4 : SUCCESS in 5m 07s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a00d5edcbfa74897badda1aaeccdaf3a : SUCCESS in 7m 00s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/64f6ca69337a4a23a50acbaec00b0bb2 : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":3},{"id":"3e4fa7b264411dd329e50121075feab7337b5d72","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-14 08:33:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"791ce36ce11b7caaa4f3417cf450879cc5a41222","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-14 08:45:44.000000000","message":"Topic set to custom_sg","accounts_in_message":[],"_revision_number":3},{"id":"c042077303b9dad55e42e04f649c5dc5960e02c0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-10-23 15:40:28.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"37b8a07259dfceb364ec0c6a44eb21a181ab1d66","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 12:29:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0c849f5470c3e2c5cba7f268fa9b315ae36f90bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-04 10:59:17.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"539c295d9c668f61bf252e3b0b927a8aee633115","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-04 10:59:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7bcb3ef3adad3843b2e57c0cc5b8b65303327260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 11:49:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae22a7fc2ff4e34890902c2c20d5126\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31582c623f7043ceb470def5a8b3c89c : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387550ccabf44af5bb3ce24838918399 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de928065117a4f5bb95feaa069cb62b1 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49d6839ac8a34c9895b6efd7c30d824e : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3016a01e5c41888ee932a404ce1fbc : SUCCESS in 4m 22s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/39f6a363519045ee98eb4312f972d697 : SUCCESS in 3m 59s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5ecbc265dfb94f678c46582c8217012d : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":4},{"id":"4fe7d0aef5a26c435ac52fae812690a06a0979ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-04 18:37:40.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e86855758dbd3e9d0715940e285be3928c8760cd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-05 20:34:24.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"df1fa32833e546fe7a3177260a472c1d2b606222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-06 06:45:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"497c33496ef5e6eb4c1ac6e9326c4e62015f90f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 07:22:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e60766ec06274561ae21ac3319dbfbd8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2d6fcc2740d4afe921fe06c684be09e : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69d3d9b22424d7a8b7b84fb591fb856 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bcca94c758046d5be1f18ecf78bc429 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9ac63c374484185b876328e9ce67f24 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6fb69484d04d078cc8db87e4a37bc0 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c523f3b2836d49e58f0fba8ba722011d : SUCCESS in 2m 15s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6419cbe7bdb74133a4dd13682cedc9f2 : SUCCESS in 6m 47s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/15d467d966ab4fb7ba3d657ea270596c : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":5},{"id":"42351075febc1b9e51679523df939db72c4db1ea","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-06 08:30:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"070f89ee51c77c1c6481084571a053a2d0198bc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 08:30:49.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0a5ca4ca28defa8f8f64c4593cd60d61028e7876","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 08:35:41.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/242c42c0578e4a4fb12c13511ae104f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b7c452949e48d3a13780703281449c : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c3fc8d5899143b5901ea483a2fa4044 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fc53edb47e5437484e399e4c1134cae : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1147b858b8f34eaf9ffcb68f839b2113 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebc405ac7fcd4db6a8b27d63dcb5b727 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":5},{"id":"6abdd37e043e47b515639b706f8259e3484bc291","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 08:35:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b4407461b9a4229004652c448786d1e36909ad5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 08:36:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a97e51d9c3f64dc2a0d7560ea131b33c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d8352dcf6f4397aacb826be385c024 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/294b09cbb406427e9e65dd6323b1139b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0c86807d0045f9619f9016459eac0ea84f7e6723","revisions":{"e7aeafbdec4234c7353653a228e835ec3988d801":{"kind":"REWORK","_number":1,"created":"2024-05-15 11:19:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/919722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/22/919722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/1"}}},"commit":{"parents":[{"commit":"c7479143dfb4a5ccebd5c7492111b5776cb19d75","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c7479143dfb4a5ccebd5c7492111b5776cb19d75"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-15 11:19:12.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-15 11:19:36.000000000","tz":120},"subject":"Add vip_sg_ids and vip_sgs to the data model","message":"Add vip_sg_ids and vip_sgs to the data model\n\nRelated-Bug: #2065767\nChange-Id: I1c3554d31b9066c5abdc958985de58f6f5a35d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e7aeafbdec4234c7353653a228e835ec3988d801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e7aeafbdec4234c7353653a228e835ec3988d801"}]},"branch":"refs/heads/master"},"8c3d6dff27693cfbd726814028894478e5f4455c":{"kind":"REWORK","_number":2,"created":"2024-09-24 08:20:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/919722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/22/919722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/2"}}},"commit":{"parents":[{"commit":"aa833df58cdfdeb96a7da482c37b5654c60f2d52","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/aa833df58cdfdeb96a7da482c37b5654c60f2d52"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-15 11:19:12.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-09-24 07:36:10.000000000","tz":120},"subject":"Add vip_sg_ids and vip_sgs to the data model","message":"Add vip_sg_ids and vip_sgs to the data model\n\nRelated-Bug: #2065767\nChange-Id: I1c3554d31b9066c5abdc958985de58f6f5a35d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/8c3d6dff27693cfbd726814028894478e5f4455c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/8c3d6dff27693cfbd726814028894478e5f4455c"}]},"branch":"refs/heads/master"},"ef09717d53e1c7c9eb150e209925eabf50c5c931":{"kind":"REWORK","_number":3,"created":"2024-10-11 07:58:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/919722/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/22/919722/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/3"}}},"commit":{"parents":[{"commit":"5cbf95034354f0e037e48d37f90aa3223da5e341","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/5cbf95034354f0e037e48d37f90aa3223da5e341"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-15 11:19:12.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-11 07:59:17.000000000","tz":120},"subject":"Add vip_sg_ids and vip_sgs to the data model","message":"Add vip_sg_ids and vip_sgs to the data model\n\nRelated-Bug: #2065767\nChange-Id: I1c3554d31b9066c5abdc958985de58f6f5a35d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/ef09717d53e1c7c9eb150e209925eabf50c5c931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/ef09717d53e1c7c9eb150e209925eabf50c5c931"}]},"branch":"refs/heads/master"},"c7a3380b5e4c4a8d9a865d7ac27e584fb1a0931c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-11-04 10:59:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/919722/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/22/919722/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/4"}}},"commit":{"parents":[{"commit":"5cbf95034354f0e037e48d37f90aa3223da5e341","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/5cbf95034354f0e037e48d37f90aa3223da5e341"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-15 11:19:12.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-11-04 10:59:20.000000000","tz":60},"subject":"Add vip_sg_ids to the data model","message":"Add vip_sg_ids to the data model\n\nIt represents an optinal list of Neutron Securit Group IDs that can be\napplied on the VIP port of a Load Balancer.\n\nRelated-Bug: #2065767\nChange-Id: I1c3554d31b9066c5abdc958985de58f6f5a35d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c7a3380b5e4c4a8d9a865d7ac27e584fb1a0931c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c7a3380b5e4c4a8d9a865d7ac27e584fb1a0931c"}]},"branch":"refs/heads/master"},"0c86807d0045f9619f9016459eac0ea84f7e6723":{"kind":"REWORK","_number":5,"created":"2024-11-06 06:45:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/919722/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/22/919722/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/22/919722/5"}}},"commit":{"parents":[{"commit":"5cbf95034354f0e037e48d37f90aa3223da5e341","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/5cbf95034354f0e037e48d37f90aa3223da5e341"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-15 11:19:12.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-11-06 06:45:23.000000000","tz":60},"subject":"Add vip_sg_ids to the data model","message":"Add vip_sg_ids to the data model\n\nIt represents an optinal list of Neutron Securit Group IDs that can be\napplied on the VIP port of a Load Balancer.\n\nRelated-Bug: #2065767\nChange-Id: I1c3554d31b9066c5abdc958985de58f6f5a35d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0c86807d0045f9619f9016459eac0ea84f7e6723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0c86807d0045f9619f9016459eac0ea84f7e6723"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
