)]}'
{"id":"openstack%2Foctavia-lib~936863","triplet_id":"openstack%2Foctavia-lib~master~I09b21e2163c01ff9be7e9c5e3bb86604e0067cfb","project":"openstack/octavia-lib","branch":"master","topic":"bug/2090782","attention_set":{"37513":{"account":{"_account_id":37513,"name":"Aslan","display_name":"Amir Aslan Aslani","email":"amiraslanaslani6@gmail.com","username":"amiraslanaslani"},"last_update":"2025-01-21 16:10:44.000000000","reason":"\u003cGERRIT_ACCOUNT_34451\u003e replied on the change","reason_account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I09b21e2163c01ff9be7e9c5e3bb86604e0067cfb","subject":"Add a limited graph retrieval option to the driver library","status":"NEW","created":"2024-12-02 17:23:16.000000000","updated":"2025-01-21 16:10:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6ae99e7d2f276afecccd684c64377757535d625","_number":936863,"virtual_id_number":936863,"owner":{"_account_id":37513,"name":"Aslan","display_name":"Amir Aslan Aslani","email":"amiraslanaslani6@gmail.com","username":"amiraslanaslani"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-02 17:42:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2025-01-21 16:10:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 17:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-21 16:10:44.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"}],"messages":[{"id":"3ef3dd0671e1de4dff9a2beb9566558e1c4c212e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37513,"name":"Aslan","display_name":"Amir Aslan Aslani","email":"amiraslanaslani6@gmail.com","username":"amiraslanaslani"},"date":"2024-12-02 17:23:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"195dfc1f6f44939bf98560f059b30a15bab4b890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 17:42:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3647ce33b44441bd9d1048fd21477798\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75e8856c1002479fb36272212cb3bf43 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe73a3ce14e47bab76813fbca34c989 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a92b24c43ba24c0bb4502b5299f202c3 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9391e6591f9c42dd8d213f09f3a65384 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e27232a6dc254a3e9d755f25deb1dde7 : SUCCESS in 4m 51s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2bad267f8ab2477c9166077b1ec8c6b3 : SUCCESS in 4m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/85678d2085c242c088fefec2501a1dd3 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c6ae99e7d2f276afecccd684c64377757535d625","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-01-21 16:10:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a39d5b9ca1dd4801b15b50af6283eb1efe46a2d","revisions":{"7a39d5b9ca1dd4801b15b50af6283eb1efe46a2d":{"kind":"REWORK","_number":1,"created":"2024-12-02 17:23:16.000000000","uploader":{"_account_id":37513,"name":"Aslan","display_name":"Amir Aslan Aslani","email":"amiraslanaslani6@gmail.com","username":"amiraslanaslani"},"ref":"refs/changes/63/936863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/63/936863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/63/936863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/63/936863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/63/936863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/63/936863/1"}}},"commit":{"parents":[{"commit":"270efb60cf35bf5a9a4d90b761bf52d8ae4825fc","subject":"Merge \"Add vip_sg_ids to the data model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/270efb60cf35bf5a9a4d90b761bf52d8ae4825fc"}]}],"author":{"name":"Amir Aslan Aslani","email":"amiraslanaslani6@gmail.com","date":"2024-12-02 17:22:43.000000000","tz":210},"committer":{"name":"Amir Aslan Aslani","email":"amiraslanaslani6@gmail.com","date":"2024-12-02 17:23:13.000000000","tz":210},"subject":"Add a limited graph retrieval option to the driver library","message":"Add a limited graph retrieval option to the driver library\n\nObject retrieval from the driver library now includes an optional argument in all existing get methods to limit the number of retrieved nodes in the model, reducing performance overhead. This approach is suitable for specific cases where full graph information is unnecessary.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/936857\nRelated-Bug: #2090782\nChange-Id: I09b21e2163c01ff9be7e9c5e3bb86604e0067cfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/7a39d5b9ca1dd4801b15b50af6283eb1efe46a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/7a39d5b9ca1dd4801b15b50af6283eb1efe46a2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
