)]}'
{"id":"openstack%2Foctavia-lib~942691","triplet_id":"openstack%2Foctavia-lib~master~I82ada3f3e8b58687ef0492f1968949daabcf7974","project":"openstack/octavia-lib","branch":"master","topic":"rate-limiting","attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-04-08 12:21:30.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I82ada3f3e8b58687ef0492f1968949daabcf7974","subject":"o-lib changes for rate limit support","status":"NEW","created":"2025-02-25 13:34:53.000000000","updated":"2026-02-12 20:45:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"98e98a4ada26f82de9249397ad76b64ac376d150","_number":942691,"virtual_id_number":942691,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-08 12:44:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2026-02-12 20:45:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-25 13:48:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 20:45:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5293b9390fbf9d8d8ed4a7fd6f01273f31cac27a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-25 13:34:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf9b1def448551dfd253bea75c7fe127f4f2063f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 13:48:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42dc226f45b546e3955525182358668c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9139903ab6ec41a6b1789494778de6af : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9163d50bba26401cab28b53e830e83b1 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64baff960ca7427ab83d808aef4ed08d : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28c14b8f775742e18bf453b960da83d0 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8302228bc24d9bb49c3f73a8477edd : SUCCESS in 6m 01s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/92ebcc0b5917492bb32237e4b5f8761e : SUCCESS in 4m 55s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/978114a964384228b86b5eea9416f89b : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cfd002df9553ceaf89d8812e2780a5d7ab9f9721","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-01 14:12:41.000000000","message":"Topic set to rate_limit","accounts_in_message":[],"_revision_number":1},{"id":"3c87a81c407406b58861f43f6404c0f2cb1876ab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-01 14:12:52.000000000","message":"Topic rate_limit removed","accounts_in_message":[],"_revision_number":1},{"id":"47b1c91a35055a2e68865a2155e32297d1209486","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-01 14:12:59.000000000","message":"Topic set to rate-limiting","accounts_in_message":[],"_revision_number":1},{"id":"e607374b13b6c9d0a9b12e4f6461a94b171cc8a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-01 17:47:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bdfb998e71a21302e164fc4886afcc26004820ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 18:00:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebe692e8e68d4a1ea56a9dd6c90975c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2eb14817714068979a7cc7b238a420 : FAILURE in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc45b148608745a88f5ca3a649b2894d : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15418aabf4894496b4894e4f1f7dee44 : FAILURE in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6080cbfbb982436584fa34a83ebb6786 : FAILURE in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a178fe7abdc46b0aed8c8d8ed3002d7 : SUCCESS in 5m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4457f9140260462289602e0ddb97e1e4 : SUCCESS in 4m 01s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/efe76f4405154f1d803022df91701e77 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d5f951d1f1761eb9e43a46d802d4f25026414231","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-02 14:54:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a167e42fe461165e1c08596520b4cc562ed835c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 15:20:44.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/483a28e8b6694f8e8616d5e659aa67d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/242d6181290342a3a867924884a148ce : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d27a138c8294b99a0a296df9c5db9e1 : FAILURE in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40c9f89824e6401c847675d925c904bf : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f0482d096f44487b93482b8fb6bc4d9 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd9fef449ea4601a83347d7200f69e3 : SUCCESS in 7m 19s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/561e484b0fee4bd3854ed5a8faab5fbb : SUCCESS in 4m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c4653acebfcc443f8ba234465910d596 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":3},{"id":"f6a894bf2df60ea27346b0d8585db8ae561fd479","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-02 16:19:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"274a6cc838f224c851c9824082f1869960955529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 16:33:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/403f44f427c947578bf54e1def9eea3a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c12208747db54b75b7dc41f811fa63ef : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31259294064b4f418c1fd5b416519a6a : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/399eaa731d2a417d9de9af9dd2ade722 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6f114b82c784ef883b9e3c715c35c6e : SUCCESS in 1m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4a109ce0794932adab469651b73fcd : SUCCESS in 5m 45s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/46cc2ee5cc0545f18820fdd13bd055be : SUCCESS in 4m 38s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6d8c455395fa4536a02736cc4712b819 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":4},{"id":"0fc04aee4cc05930ec82f712d9334f4af9fc754a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-02 16:45:51.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5a0c4d3c709adc96a6e397f8cf73e4cfdb448571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 17:01:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feafccd547f34f1283d7046f3c14b99a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74a8c4542fc5422286835dcfcf7c95ef : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15bb48b603d4b52910ed3af715176cd : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/051b6849eb1f4697aacf7d9ed331fdf1 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11c64528f6ec44d49a76f9c700bcf52f : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f0e31e2a8f4bf4b0e32d712daa0b70 : SUCCESS in 7m 21s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4ebe24bb80cf4a8db7dd63fde1963437 : SUCCESS in 4m 04s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dee26c631c164f38995ac75302c4fb0c : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":5},{"id":"f6d44f04151f05f764ce94b4158f46ed5f02480b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-08 11:04:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"47f9c5ec8a55b0239920c081545dc0f9b35146ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 11:13:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebcd27d733e44e45a75ded03427ffacc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d82dfb4264a54dda84f2f318c36169bf : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0bbd9108d6741669f5073dd7a8e262b : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ef12f218fee42ebb34b78cd4a987960 : SUCCESS in 2m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd19f291fc454a369ecf9ded189d97ab : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5e4060d420343d58e9a53e3abe13dc2 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30aadcb7c224cd2b6ecfd6be777b249 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b75ab75e4904673a5dcda709bb55d1d : SUCCESS in 2m 05s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4266c0296871452d9b5a0a1650ea206e : SUCCESS in 4m 02s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1ca0bfb4fde549c9825904ab3d496bb4 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":6},{"id":"26688104022a53defecac7b6b3957e7fda66afe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-08 12:08:52.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"fb9eba4975f3a2f12d133767913c9b6807a7c41b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 12:21:30.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3f524103cea4107978512516055cf57\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/254cd8787a564991a48624d3b44ba0c4 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4033556ae643268dc24ed083060a24 : FAILURE in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c79ce87cf7142608df642549c66c0ed : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9785438782b4112bcc718fa2d723f92 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8057b5f78c94274bfcd2d5cace85630 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f10033486aaf4786a52d5c326c4c23c3 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6db826e325a4f389887e197b2ea8190 : SUCCESS in 2m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8cf2f74a69a64a4cbad9249017ebd29b : SUCCESS in 6m 34s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8e5730e2f69944a481abf4338a2b1cdb : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":7},{"id":"83e9b5f67ab3cb0e1585ea1cf79d61173b0b3ac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-08 12:30:19.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"26f85ee6cd4db7ded618e9cb9ba8532845c7bbbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 12:44:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b4ec8286b5d43ea9f62923087f0b2ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c580fd550ef46399282ddf73dce0a35 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e0c50652154488a9a3971d8e616e9c : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73610885d59245b2b07dfff63a9a0e4c : SUCCESS in 1m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d83ffb5ea674bc0b48fa257affd72a4 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/229ee60b58344400909ba5646721daf5 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fbfa8ab23a240f38235c19d70c4646c : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75b2d022dc544a98a6e9e9e0a9f17d6d : SUCCESS in 3m 02s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6f5f0a1940174bc980e20f466dd9aa1f : SUCCESS in 4m 37s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/82d2780142d042fd963f9b0ff190a6ae : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":8},{"id":"98e98a4ada26f82de9249397ad76b64ac376d150","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-02-12 20:45:18.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"0c6931cc12a99939e492b3bce6dac2cc14db7920","revisions":{"0741711a7bdf0e7a3d73f85f0369a91bbd1d9c03":{"kind":"REWORK","_number":1,"created":"2025-02-25 13:34:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/942691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/91/942691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/1"}}},"commit":{"parents":[{"commit":"e473f6f4b368cdb97c725f29fa070a05d283848d","subject":"Update default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e473f6f4b368cdb97c725f29fa070a05d283848d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"subject":"O_lib changes for rate limit support","message":"O_lib changes for rate limit support\n\nChange-Id: I82ada3f3e8b58687ef0492f1968949daabcf7974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0741711a7bdf0e7a3d73f85f0369a91bbd1d9c03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0741711a7bdf0e7a3d73f85f0369a91bbd1d9c03"}]},"branch":"refs/heads/master"},"d57472ed0d48fbdec408c8e955e3ef34a57fce14":{"kind":"REWORK","_number":2,"created":"2025-04-01 17:47:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/942691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/91/942691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/2"}}},"commit":{"parents":[{"commit":"e473f6f4b368cdb97c725f29fa070a05d283848d","subject":"Update default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e473f6f4b368cdb97c725f29fa070a05d283848d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-04-01 17:47:07.000000000","tz":120},"subject":"O_lib changes for rate limit support","message":"O_lib changes for rate limit support\n\nChange-Id: I82ada3f3e8b58687ef0492f1968949daabcf7974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/d57472ed0d48fbdec408c8e955e3ef34a57fce14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/d57472ed0d48fbdec408c8e955e3ef34a57fce14"}]},"branch":"refs/heads/master"},"c80dfd49e75bed0fc17f05e041f490d510494dce":{"kind":"REWORK","_number":3,"created":"2025-04-02 14:54:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/942691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/91/942691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/3"}}},"commit":{"parents":[{"commit":"e473f6f4b368cdb97c725f29fa070a05d283848d","subject":"Update default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e473f6f4b368cdb97c725f29fa070a05d283848d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-04-02 14:53:51.000000000","tz":120},"subject":"o-lib changes for rate limit support","message":"o-lib changes for rate limit support\n\nChange-Id: I82ada3f3e8b58687ef0492f1968949daabcf7974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c80dfd49e75bed0fc17f05e041f490d510494dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c80dfd49e75bed0fc17f05e041f490d510494dce"}]},"branch":"refs/heads/master"},"9f241c93189f04a7eb7dbee03c877b58899506c6":{"kind":"REWORK","_number":4,"created":"2025-04-02 16:19:00.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/942691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/91/942691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/4"}}},"commit":{"parents":[{"commit":"e473f6f4b368cdb97c725f29fa070a05d283848d","subject":"Update default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e473f6f4b368cdb97c725f29fa070a05d283848d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-04-02 16:18:47.000000000","tz":120},"subject":"o-lib changes for rate limit support","message":"o-lib changes for rate limit support\n\nChange-Id: I82ada3f3e8b58687ef0492f1968949daabcf7974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9f241c93189f04a7eb7dbee03c877b58899506c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9f241c93189f04a7eb7dbee03c877b58899506c6"}]},"branch":"refs/heads/master"},"e9f521da4448c5fff5cfde5c9bd90815c3b8a281":{"kind":"NO_CHANGE","_number":5,"created":"2025-04-02 16:45:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/942691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/91/942691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/5"}}},"commit":{"parents":[{"commit":"e473f6f4b368cdb97c725f29fa070a05d283848d","subject":"Update default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e473f6f4b368cdb97c725f29fa070a05d283848d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-04-02 16:20:23.000000000","tz":120},"subject":"o-lib changes for rate limit support","message":"o-lib changes for rate limit support\n\nChange-Id: I82ada3f3e8b58687ef0492f1968949daabcf7974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e9f521da4448c5fff5cfde5c9bd90815c3b8a281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e9f521da4448c5fff5cfde5c9bd90815c3b8a281"}]},"branch":"refs/heads/master"},"79a09c31bf09d31c499a3d6b7626ed4bf4e434ed":{"kind":"REWORK","_number":6,"created":"2025-04-08 11:04:09.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/942691/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/91/942691/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/6"}}},"commit":{"parents":[{"commit":"e473f6f4b368cdb97c725f29fa070a05d283848d","subject":"Update default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e473f6f4b368cdb97c725f29fa070a05d283848d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-04-08 11:03:57.000000000","tz":120},"subject":"o-lib changes for rate limit support","message":"o-lib changes for rate limit support\n\nChange-Id: I82ada3f3e8b58687ef0492f1968949daabcf7974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/79a09c31bf09d31c499a3d6b7626ed4bf4e434ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/79a09c31bf09d31c499a3d6b7626ed4bf4e434ed"}]},"branch":"refs/heads/master"},"9e145b67b362c828621338568c1699560a25a8d9":{"kind":"REWORK","_number":7,"created":"2025-04-08 12:08:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/942691/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/91/942691/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/7"}}},"commit":{"parents":[{"commit":"e473f6f4b368cdb97c725f29fa070a05d283848d","subject":"Update default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e473f6f4b368cdb97c725f29fa070a05d283848d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-04-08 12:08:43.000000000","tz":120},"subject":"o-lib changes for rate limit support","message":"o-lib changes for rate limit support\n\nChange-Id: I82ada3f3e8b58687ef0492f1968949daabcf7974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9e145b67b362c828621338568c1699560a25a8d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9e145b67b362c828621338568c1699560a25a8d9"}]},"branch":"refs/heads/master"},"0c6931cc12a99939e492b3bce6dac2cc14db7920":{"kind":"REWORK","_number":8,"created":"2025-04-08 12:30:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/942691/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/91/942691/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/91/942691/8"}}},"commit":{"parents":[{"commit":"e473f6f4b368cdb97c725f29fa070a05d283848d","subject":"Update default envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e473f6f4b368cdb97c725f29fa070a05d283848d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-25 13:34:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-04-08 12:30:11.000000000","tz":120},"subject":"o-lib changes for rate limit support","message":"o-lib changes for rate limit support\n\nChange-Id: I82ada3f3e8b58687ef0492f1968949daabcf7974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0c6931cc12a99939e492b3bce6dac2cc14db7920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0c6931cc12a99939e492b3bce6dac2cc14db7920"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
