)]}'
{"id":"openstack%2Foctavia-tempest-plugin~1002232","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Idd7931910b97a91e9d57f6b82d11b0b55370a6ba","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-08-26 15:46:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idd7931910b97a91e9d57f6b82d11b0b55370a6ba","subject":"Disable pypi in zuul config, use only local mirrors","status":"ABANDONED","created":"2026-08-25 06:17:30.000000000","updated":"2026-08-26 15:46:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b70e454969742abe38d31aa6e0e1800dbc53c9be","_number":1002232,"virtual_id_number":1002232,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-25 08:05:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-25 08:05:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bffc7bed8b313a22c0c645267ade6d59d19526c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-25 06:17:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d77440f612ab10665948a218e82d134631c8baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 08:05:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/102146890ea84156ab85472645646fe5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8b001c6f1d4539a169ea040ff181b0 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d68bb110a17b4c6d9e7e4887b67f51a3 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74fcd0f87b5f418c9815cda64c6f05ba : SUCCESS in 3m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6000e29f326432da3eaac46daf361ca : SUCCESS in 1h 10m 16s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/416131743051425db6fa24a6c60ea2dc : SUCCESS in 1h 01m 49s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/fff541ab038c466d925b06bae75b582f : SUCCESS in 1h 18m 35s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/f5450963169d44909aef617cabfe7e54 : SUCCESS in 1h 01m 02s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/6fdca84fc939476f8e2012d255b09c53 : SUCCESS in 1h 13m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/05196c8908a1425a950947d9b640235c : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f508d94d93424b54b058c7b685ca22af : SUCCESS in 57m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/76faf170af7f4052bb982f27ef4da2cf : SUCCESS in 1h 45m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/4d4e4f92ff7b425783f4f0079b7bd5fc : SUCCESS in 1h 09m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/d3cfa9368d004793922aad2db47f0179 : SUCCESS in 1h 45m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/b1570e1f693543b6b19daf0807e35dd5 : SUCCESS in 1h 11m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/02453093919a4bde8dbcb72de7d54cc2 : SUCCESS in 1h 45m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/f087f09bb9cf49258210223ddb5a491b : FAILURE in 22m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf1e4a6e243b4a8d817828ad8c6ac94e : SUCCESS in 44m 22s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/20e9aa9ba8dd44b2b6bad3e3dd3b3baf : SUCCESS in 37m 10s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/f8fcc236c3274658bf0ac4f0ca0c5b23 : SUCCESS in 36m 46s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/705ed4340b1e4f54a4575306bc888f67 : SUCCESS in 37m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cabf593da15f4ef1bba94e6f5b21ab2b : SUCCESS in 1h 23m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/1e630f4871bd44cd8ddcedc6cb244979 : SUCCESS in 1h 13m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/694512543f16478aaf1fdfd24fcd725c : SUCCESS in 54m 36s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/c6bd11aa21ae4211a4dec9738e579a6f : FAILURE in 20m 32s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/599046b093a0436da08ed3713ec55737 : FAILURE in 24m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/210fb897116b4024aab3be0d1fa03239 : SUCCESS in 1h 04m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b401fcc44274ce2a415af6cbe5373c1 : SUCCESS in 29m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/ae40218fdc184e1094347086aa975dd1 : SUCCESS in 43m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/8f4c87f11b0246f98da88d40875beeba : SUCCESS in 31m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/61d491b1829d49618f5af2e9f1603429 : SUCCESS in 32m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a1a0c9db130f47c4a02167100ef20c84 : SUCCESS in 34m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d0cfe3f2a22d41c694861560d2eec697 : SUCCESS in 1h 05m 25s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/55456c9d3fef4bc8a90908325f409bfb : SUCCESS in 53m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b70e454969742abe38d31aa6e0e1800dbc53c9be","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-26 15:46:02.000000000","message":"Abandoned\n\ndidnt work","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bf43aa573097639bd74e9efd292ff9b53a366a6","revisions":{"4bf43aa573097639bd74e9efd292ff9b53a366a6":{"kind":"REWORK","_number":1,"created":"2026-08-25 06:17:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/1002232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/32/1002232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/32/1002232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/32/1002232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/32/1002232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/32/1002232/1"}}},"commit":{"parents":[{"commit":"3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f","subject":"Drop python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-08-25 06:16:46.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-08-25 06:16:50.000000000","tz":120},"subject":"Disable pypi in zuul config, use only local mirrors","message":"Disable pypi in zuul config, use only local mirrors\n\nCI jobs have been failing due to connection issues to pythonhosted.org\nwhile local mirrors were already configured.\nRemove pypi index from pip config and force the use of local local\nmirrors with DIB_NO_PYPI_PIP\u003d1\n\nChange-Id: Idd7931910b97a91e9d57f6b82d11b0b55370a6ba\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4bf43aa573097639bd74e9efd292ff9b53a366a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4bf43aa573097639bd74e9efd292ff9b53a366a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
