)]}'
{"id":"openstack%2Foctavia-tempest-plugin~1005310","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I86c2bf63c47473ded9321f0efca0364f48d60395","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"2026.2-job","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-11 17:14:11.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I86c2bf63c47473ded9321f0efca0364f48d60395","subject":"Add stable/2026.2 job","status":"NEW","created":"2026-09-11 17:07:27.000000000","updated":"2026-09-11 18:50:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3c3288a700631247906b3bd5e607c70ed27c1bcb","_number":1005310,"virtual_id_number":1005310,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 18:50:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 17:08:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17bf8e04d9a9a91e7449496624f3f9ad6c578bbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:07:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ceec0846258d7ea529331ad7c145359eccd241ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 17:08:13.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  Configuration item is not a dictionary.  Each zuul.yaml\n  configuration file must be a list of dictionaries, for\n  example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with one\n  key (in this example, \u0027job\u0027 and \u0027project\u0027).\n  \n  The incorrect values are around:\n  \n  job: null","accounts_in_message":[],"_revision_number":1},{"id":"eda34486cd088475861395622c5bdaec188e77e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:11:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3c3288a700631247906b3bd5e607c70ed27c1bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:50:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9381403a5c049b9a4ed7db5c8e3cfae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd1a5113fcb4fabbf9b165c81c37819 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e3354cdb7a4c5c892b2a189331d4c4 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/421aa3b9907b4fde957284414b1aa839 : SUCCESS in 3m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6c16aaed3de42968f308c585b224ff1 : SUCCESS in 1h 06m 02s\n- octavia-v2-dsvm-noop-api-stable-2026-2 https://zuul.opendev.org/t/openstack/build/ade86c770dba4690a5cbcfc762e49da1 : SUCCESS in 56m 26s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/53a55fd7fc6947968b9a224f45ea7df7 : SUCCESS in 1h 00m 51s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/782547256a484c2b8d8f281b85d735b4 : SUCCESS in 58m 51s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/ae48b600014f4504b9e2309cbd7c4989 : SUCCESS in 1h 08m 15s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/0187f3d632df40ab8290de2998fd4bf9 : SUCCESS in 1h 23m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3814c3afb616400986ec76e269d298c7 : SUCCESS in 1h 22m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/343e1a83cbf94bae9b2e0d42c3fb31d7 : SUCCESS in 1h 02m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-2 https://zuul.opendev.org/t/openstack/build/1359228e06614fa0bd286d5f5b501c59 : SUCCESS in 1h 21m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-2 https://zuul.opendev.org/t/openstack/build/f34eaa7e7f754adab715a4f1df568fd6 : SUCCESS in 1h 05m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/3d9d1999bef84ef8a2f647c43ad8b8d0 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/68e4f61b049e4665ae8cb90a4c9c2688 : SUCCESS in 1h 06m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/9f40139889194f7abc647472b6ac96a4 : SUCCESS in 1h 35m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/4ddc4cf93ad14593967b38114e832c23 : SUCCESS in 59m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/d27a69a2c44844538d0830af7e1e4475 : SUCCESS in 1h 32m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2ffddb2183b3421889d6d71c5b77b141 : SUCCESS in 1h 00m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85ec1671c9ff4baf8c271b78d2317d4e : SUCCESS in 36m 02s\n- octavia-v2-dsvm-tls-barbican-stable-2026-2 https://zuul.opendev.org/t/openstack/build/f2c34d1629684b48b73b408d9652dc67 : SUCCESS in 35m 41s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/72df5d4e718b4086b5654b78195343dc : SUCCESS in 38m 38s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/4d9a21e772aa4c6bbe4d3244b0cc42cb : SUCCESS in 37m 45s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/e6c18325d6cd4c8695ac1cf888ce7b00 : SUCCESS in 36m 59s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/26ffe6f7e0294351bc2238297690fec7 : SUCCESS in 1h 21m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/74c64b6a611946309db56b8ff1bdeed9 : SUCCESS in 1h 13m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80fea9d0bf244e71b911fd101da54e28 : SUCCESS in 1h 00m 36s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/ccea11ff367d41b4a1745b36761dddd8 : FAILURE in 20m 42s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/65e547e4e617494d84520342be2fd936 : FAILURE in 20m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/a57c457fe42042a0b155a99bc2957298 : SUCCESS in 1h 34m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d75e8475de724439a7cec62f07dec9bb : SUCCESS in 1h 02m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-2 https://zuul.opendev.org/t/openstack/build/67c846de284e4ef6b2a67d7babf72a23 : SUCCESS in 32m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/f6ae102df1f44dd482c798003b2b819e : SUCCESS in 1h 05m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/3005fb9aabdf40c0b4ab511eac2845de : SUCCESS in 1h 04m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/37c8703e7d21484b8cd19e7a3acc960c : SUCCESS in 32m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3b35079b97a44e3a98c763c4767cac40 : SUCCESS in 33m 05s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2f1622b6d7d04c528e9c2546e50e3d52 : SUCCESS in 1h 20m 24s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/333b5e2d5b44470a860c515bc7eeb04b : SUCCESS in 49m 23s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c63f8a5ebe7e0c2d7840120b6029f4260064fe9c","revisions":{"d2e70e01c5b8518214778527944371bb60a4b780":{"kind":"REWORK","_number":1,"created":"2026-09-11 17:07:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/1005310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/10/1005310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/1005310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/1005310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/1005310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/1005310/1"}}},"commit":{"parents":[{"commit":"3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f","subject":"Drop python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:07:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:07:17.000000000","tz":540},"subject":"Add stable/2026.2 job","message":"Add stable/2026.2 job\n\nThe branch was created so we should start testing it.\n\nChange-Id: I86c2bf63c47473ded9321f0efca0364f48d60395\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d2e70e01c5b8518214778527944371bb60a4b780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d2e70e01c5b8518214778527944371bb60a4b780"}]},"branch":"refs/heads/master"},"c63f8a5ebe7e0c2d7840120b6029f4260064fe9c":{"kind":"REWORK","_number":2,"created":"2026-09-11 17:11:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/1005310/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/10/1005310/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/1005310/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/1005310/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/1005310/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/1005310/2"}}},"commit":{"parents":[{"commit":"3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f","subject":"Drop python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:07:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:11:15.000000000","tz":540},"subject":"Add stable/2026.2 job","message":"Add stable/2026.2 job\n\nThe branch was created so we should start testing it.\n\nChange-Id: I86c2bf63c47473ded9321f0efca0364f48d60395\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c63f8a5ebe7e0c2d7840120b6029f4260064fe9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c63f8a5ebe7e0c2d7840120b6029f4260064fe9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
