)]}'
{"id":"openstack%2Foctavia-tempest-plugin~1005728","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I9d9657dc81c64342fc1a4ae83c7f6a0eb9603cd3","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{"28258":{"account":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"last_update":"2026-09-15 15:32:13.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9d9657dc81c64342fc1a4ae83c7f6a0eb9603cd3","subject":"Add test for LB/amphora failover when vip-subnet is full","status":"NEW","created":"2026-09-15 13:46:33.000000000","updated":"2026-09-15 15:32:13.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e86382d3ea71d04eff974ad154377b27a77a8100","_number":1005728,"virtual_id_number":1005728,"owner":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-15 15:32:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 15:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86bba06dd15eb3f3526864bb3064f82c6ab48bcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2026-09-15 13:46:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e86382d3ea71d04eff974ad154377b27a77a8100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 15:32:13.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8a7d782d7b540339dc4d3175e67a343\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1ba8ca388c4e06b39e6ccea786b39c : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef6132c7d50549b0a2a3c3a4fbd33e82 : FAILURE in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e80ee3efd17f4a988a5498cf48229620 : SUCCESS in 3m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d63046c55c44424954690f9143d71a2 : SUCCESS in 1h 04m 55s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/1beefa46d32d49f8916327438caa6c9a : SUCCESS in 1h 01m 12s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/78fae91b180641afa43c78b2d76ee46b : SUCCESS in 1h 11m 16s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/9ef2b3b70c5744899ffd7380e448d429 : SUCCESS in 59m 04s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/045d8c917ac3427084c0ce05ca662e25 : SUCCESS in 1h 02m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8e38a32786f1461b8832ce18bec39826 : SUCCESS in 1h 24m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/607258991d8147cdb7e6f7d90a9fff18 : SUCCESS in 1h 07m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/06f863c7050c45b7ad60982e2478a7ad : SUCCESS in 1h 36m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/8439bb2398834fe8a000d2c5257b79fe : SUCCESS in 1h 01m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/7fb7b6e6f9cb41e5a0064d9a04193a42 : SUCCESS in 1h 37m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/edfe05aef468463cb01177cf51561ef9 : SUCCESS in 1h 05m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/8ce45d137fe947b89ba8d972d8937052 : SUCCESS in 1h 34m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/5c3071fc33bb434e9bfe3a5894b049ab : SUCCESS in 1h 02m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53466855f2614a098d78c8b8e491c11a : SUCCESS in 37m 16s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/f675b91fdff24d54821447f82981994a : SUCCESS in 40m 06s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/acfd5fec3af1484db925e8e0bbd98140 : SUCCESS in 38m 17s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/dac4f0d340644838b656d71161c233a1 : SUCCESS in 38m 47s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10ef56827f1241949d6b56e0a5bdafeb : SUCCESS in 1h 21m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/9df508097b2a4056a285b018e2dae2d9 : SUCCESS in 1h 14m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7e9ba6fa827d4625ae742caaf2197560 : SUCCESS in 1h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/dd32b12b27a84b1795fae2a49141d877 : FAILURE in 19m 14s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e1e40f3ae5f64372a65a5badb7566f97 : FAILURE in 20m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/da75a75f8e4741a6bd355de5e9e96fe9 : SUCCESS in 49m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67860909e3de47e4834d0f5101f59870 : SUCCESS in 33m 50s (non-voting)\n- octavia-v2-act-stdby-vrrp-subnet-failover-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06e7089ad9474a2298e596daf9a915d2 : SUCCESS in 31m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/82a44e85d60e48c480c5aaf19445e66f : SUCCESS in 34m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/7fc6977ae8fb4b99ac157bad11cee157 : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/92599bb7ab4942f29351c31ef29f5cb5 : SUCCESS in 33m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/28fd10b5d6f54225ae9aefb5077594f7 : SUCCESS in 34m 24s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/3b585af91f104dd2948faab8bbcddcf7 : SUCCESS in 1h 03m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7476e17f41d5462386227ff6dc82ec6e : SUCCESS in 51m 52s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff0c13040ccf27099be3152c56db90acd0b039dc","revisions":{"ff0c13040ccf27099be3152c56db90acd0b039dc":{"kind":"REWORK","_number":1,"created":"2026-09-15 13:46:33.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/28/1005728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/28/1005728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/28/1005728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/28/1005728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/28/1005728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/28/1005728/1"}}},"commit":{"parents":[{"commit":"3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f","subject":"Drop python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-09-15 13:39:56.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-09-15 13:39:56.000000000","tz":-240},"subject":"Add test for LB/amphora failover when vip-subnet is full","message":"Add test for LB/amphora failover when vip-subnet is full\n\nAdd a scenario test covering selection of an alternate subnet\nwhen the primary VIP subnet does not have an available address.\n\nDepends-On: If9c515056a39236a7b1266159731eba9ce20d18a\nChange-Id: I9d9657dc81c64342fc1a4ae83c7f6a0eb9603cd3\nSigned-off-by: Austin Russell \u003ckenrusse@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ff0c13040ccf27099be3152c56db90acd0b039dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ff0c13040ccf27099be3152c56db90acd0b039dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
