)]}'
{"id":"openstack%2Foctavia-tempest-plugin~1005784","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ic893e96014133f710edf13bd285b60e1e1f9bcad","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ic893e96014133f710edf13bd285b60e1e1f9bcad","subject":"Split listener API tests by protocol","status":"NEW","created":"2026-09-15 18:46:03.000000000","updated":"2026-09-15 19:58:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"54ad3d1d9c427edd0e36639cb89cb089d119b28f","_number":1005784,"virtual_id_number":1005784,"owner":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-15 19:58:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 19:58:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a99957525d2d4eea09a9b16a6fa3dd465ab973b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"date":"2026-09-15 18:46:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54ad3d1d9c427edd0e36639cb89cb089d119b28f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 19:58:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dff4144562949fea9e41f6a5dccd684\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151ca4acf68d46719f3425d515c83a83 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26d58772d1a84dc4970b63584d1d901d : SUCCESS in 2m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0896c87689ce467c8c628f28611a0b7f : SUCCESS in 3m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b435c7000804fe0908a095cdcc2b49d : SUCCESS in 41m 57s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/508e5469485d46479fd78b6ea69aba3f : SUCCESS in 1h 11m 03s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/7aa17d24a3944e5eab836349c3b3bba1 : SUCCESS in 1h 03m 42s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/6225feed5c844abe8bf75ef02ceb481f : SUCCESS in 1h 02m 33s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/5e2e129621b543ee9c4f01be24274a99 : SUCCESS in 1h 03m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc201c53cc3d40889c5489e982dcbf44 : SUCCESS in 33m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/690e4c5a60d44a89976e2d24fbbfff43 : SUCCESS in 34m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/c5536931bf014c9fb3c1ebd35d05d168 : SUCCESS in 34m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/a4616295f9bf49d4912893b68c429dc5 : SUCCESS in 33m 41s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2e0b14ac203b4d54bf86823a561fe558 : SUCCESS in 52m 37s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a094244cf8344fa7a874a1958cb9e10d : SUCCESS in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf959f8d659e7743ec343b69a010a33ef07f8294","revisions":{"bf959f8d659e7743ec343b69a010a33ef07f8294":{"kind":"REWORK","_number":1,"created":"2026-09-15 18:46:03.000000000","uploader":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"ref":"refs/changes/84/1005784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/84/1005784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/1005784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/1005784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/1005784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/1005784/1"}}},"commit":{"parents":[{"commit":"3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f","subject":"Drop python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3d0aefdebb4de7e8ef80821fe8e7ce91a982e92f"}]}],"author":{"name":"Oleksandr Kononenko","email":"okononenko@mirantis.com","date":"2023-10-10 07:24:32.000000000","tz":180},"committer":{"name":"Anna Arhipova","email":"harhipova@mirantis.com","date":"2026-09-15 18:45:16.000000000","tz":120},"subject":"Split listener API tests by protocol","message":"Split listener API tests by protocol\n\nAllow jobs to filter or skip one protocol without running the\nfull ListenerAPITest class. TLS Barbican setup stays on the\nTerminated HTTPS and multiprotocol classes.\n\nListenerAPITest is now ListenerAPIBaseTest. Tests live in\nListenerAPITestTCP, UDP, HTTP, HTTPS, Prometheus, SCTP,\nTerminatedHTTPS, and Multiprotocol.\n\nChange-Id: Ic893e96014133f710edf13bd285b60e1e1f9bcad\nSigned-off-by: Hanna Arhipova \u003charhipova@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf959f8d659e7743ec343b69a010a33ef07f8294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf959f8d659e7743ec343b69a010a33ef07f8294"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
