)]}'
{"id":"openstack%2Foctavia-tempest-plugin~626406","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Id9004adf8952d6ef82b7bda6f2d3b509655f1c69","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"log-offloading","hashtags":[],"change_id":"Id9004adf8952d6ef82b7bda6f2d3b509655f1c69","subject":"Save amphora logs in gate","status":"MERGED","created":"2018-12-19 23:02:14.000000000","updated":"2019-06-17 22:34:48.000000000","submitted":"2019-06-17 22:34:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"626406-1560810888256-b95397b7","meta_rev_id":"1d157746e2a58d1968d3c244c4645fd7b832e1b7","_number":626406,"virtual_id_number":626406,"owner":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"date":"2019-06-17 18:03:24.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-06-17 22:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-06-17 15:11:14.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-06-14 23:11:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-06-15 01:54:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2019-06-17 15:11:14.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 00:16:13.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-06-14 01:36:06.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-14 02:15:34.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-06-14 23:11:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-15 01:54:24.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-06-17 15:11:14.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-06-17 18:03:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-17 22:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6f702917d9d3758853ad9acf3a20f77cecdb8f9","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2018-12-19 23:02:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"922e3520cfed6bf3d62586fad265079b948e4135","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 01:18:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/1/check/openstack-tox-docs/7fafdc3/html/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/1/check/openstack-tox-pep8/dcacd2c/ : SUCCESS in 4m 06s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-api/0b1127d/ : SUCCESS in 41m 29s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api/df01e70/ : SUCCESS in 41m 11s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/3c5deaf/ : SUCCESS in 36m 37s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/a0887c1/ : POST_FAILURE in 41m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario/da19664/ : SUCCESS in 1h 36m 05s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario/5b37761/ : SUCCESS in 1h 27m 03s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/a9e34bf/ : SUCCESS in 1h 13m 08s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/0e0f700/ : SUCCESS in 1h 19m 53s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-centos-7/f30b965/ : SUCCESS in 2h 11m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/591c3c7/ : FAILURE in 36m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-two-node/fd42f63/ : SUCCESS in 1h 37m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-two-node/433b354/ : SUCCESS in 1h 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb9aca8c075cc4e7340051ae82cbff02024182bd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-12-20 01:20:12.000000000","message":"Patch Set 1:\n\nrecheck\n\n2018-12-19 23:48:18.708621 | TASK [fetch-subunit-output : Generate testr_results.html file]\n2018-12-19 23:48:19.233374 | controller | ERROR\n2018-12-19 23:48:19.234021 | controller | {\n2018-12-19 23:48:19.234134 | controller |   \"msg\": \"[Errno 13] Permission denied\",\n2018-12-19 23:48:19.234239 | controller |   \"rc\": 13\n2018-12-19 23:48:19.234340 | controller | }","accounts_in_message":[],"_revision_number":1},{"id":"1dd77a60f0c9973489ec79e6b3e6136dcb738fe6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 03:53:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/1/check/openstack-tox-docs/3a869c1/html/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/1/check/openstack-tox-pep8/66c36b5/ : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-api/38671ee/ : SUCCESS in 37m 06s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api/05f4a00/ : SUCCESS in 34m 48s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/dc56d50/ : SUCCESS in 36m 54s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/9a2524e/ : SUCCESS in 36m 05s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario/19ac448/ : SUCCESS in 1h 37m 23s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario/110867a/ : SUCCESS in 1h 10m 48s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/fe7ea04/ : SUCCESS in 1h 20m 20s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/c5cbeca/ : SUCCESS in 1h 30m 53s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-centos-7/5da6155/ : SUCCESS in 2h 25m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/1011e1a/ : FAILURE in 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-two-node/d786d64/ : SUCCESS in 1h 30m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-two-node/4a7fb90/ : SUCCESS in 1h 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e819d931a991088166544275592317d74512b7c8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-12-23 12:31:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31fed1790410343fb324c2e98189d94c2452277c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-01-16 20:15:41.000000000","message":"Patch Set 1:\n\nrecheck\n\nTo test against latest dependent patch set.","accounts_in_message":[],"_revision_number":1},{"id":"5ddf8903249d83ec5fe255389435d499e7cee2f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-16 22:51:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/1/check/openstack-tox-docs/8d86707/html/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/1/check/openstack-tox-pep8/65260ca/ : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-api/314650f/ : SUCCESS in 33m 01s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api/569c83e/ : SUCCESS in 38m 42s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/b36f5f2/ : SUCCESS in 39m 33s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/8d17e7f/ : SUCCESS in 38m 39s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario/61d68e6/ : SUCCESS in 1h 31m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario/157af23/ : SUCCESS in 1h 34m 59s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/a3efc9c/ : SUCCESS in 1h 28m 12s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/e634ecc/ : SUCCESS in 1h 41m 59s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-centos-7/49bdf54/ : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/67f65e8/ : FAILURE in 51m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-scenario-two-node/64ddbf8/ : FAILURE in 1h 34m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/1/check/octavia-v2-dsvm-py2-scenario-two-node/ee52f1f/ : FAILURE in 1h 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a03de2f657f8ac8ced16a4c6e2f3117baf71200","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-02-13 13:36:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8a57bb490bb9a3895330c6d3e162e8865434c48d","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-02-13 13:37:18.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Patch Set 1 was rebased\n\nrebased to double check the recent ps of https://review.openstack.org/624835\notherwise LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"d98e960b23d651ea5aff0990f61bf3a8bb9223ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 16:13:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/2/check/openstack-tox-docs/917f150/html/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/2/check/openstack-tox-pep8/4917f1c/ : SUCCESS in 4m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-noop-api/c86568f/ : SUCCESS in 42m 54s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-noop-py2-api/7cfbabb/ : SUCCESS in 41m 43s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/499fdff/ : SUCCESS in 41m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-noop-py2-api-stable-queens/d51e7a3/ : SUCCESS in 44m 13s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-scenario/bc9471f/ : SUCCESS in 1h 57m 47s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-py2-scenario/380993d/ : SUCCESS in 1h 35m 57s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-py2-scenario-stable-rocky/0a535f4/ : SUCCESS in 1h 28m 14s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-py2-scenario-stable-queens/16e24df/ : SUCCESS in 1h 55m 25s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-scenario-centos-7/b614dd0/ : SUCCESS in 2h 30m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ba205e5/ : SUCCESS in 1h 46m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-scenario-two-node/3aae6b8/ : SUCCESS in 1h 44m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-py2-scenario-two-node/31fea34/ : SUCCESS in 1h 36m 26s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/06/626406/2/check/octavia-v2-dsvm-tls-barbican/98a0423/ : SUCCESS in 52m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6010c40f13948505a98317ee3549819be80d4729","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-02-13 20:35:09.000000000","message":"Patch Set 2: Code-Review-1\n\nThis doesn\u0027t seem to be working, at least not for the octavia-v2-dsvm-scenario gate","accounts_in_message":[],"_revision_number":2},{"id":"28f117b290564ed7c63e37f445aeed362749c334","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-13 20:41:06.000000000","message":"Patch Set 2: Code-Review-1\n\nI am sure I had seen the log file when I reviewed it PS 1, hence my up-vote. My vote was then carried to PS2 because it was just a rebase. I can\u0027t find the log now either.","accounts_in_message":[],"_revision_number":2},{"id":"441dcaf7e27e781777d419b511a4bda4db701421","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-02-13 22:05:13.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"163d7a1c31ad2fb2fa8cc1808107e25481a483a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 00:34:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/3/check/openstack-tox-docs/a35cefd/html/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/3/check/openstack-tox-pep8/65cf78a/ : SUCCESS in 4m 16s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-noop-api/808fd42/ : SUCCESS in 48m 59s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-noop-py2-api/b8ce5e6/ : SUCCESS in 40m 19s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/4a27039/ : SUCCESS in 42m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-noop-py2-api-stable-queens/ed042f9/ : SUCCESS in 46m 23s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-scenario/04c566d/ : SUCCESS in 1h 27m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-py2-scenario/669db0b/ : SUCCESS in 1h 36m 20s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-py2-scenario-stable-rocky/34007f9/ : SUCCESS in 1h 27m 21s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-py2-scenario-stable-queens/325686b/ : SUCCESS in 1h 21m 12s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-scenario-centos-7/6769872/ : SUCCESS in 2h 18m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b0ff30e/ : SUCCESS in 1h 39m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-scenario-two-node/d18fc1a/ : SUCCESS in 1h 43m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-py2-scenario-two-node/6c72382/ : FAILURE in 1h 18m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/06/626406/3/check/octavia-v2-dsvm-tls-barbican/3952aa9/ : SUCCESS in 54m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a5ec23702172f6f3ecb92f91a03e16de446f5703","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-02-14 15:48:04.000000000","message":"Patch Set 3:\n\nnone of the scenario links open for me?","accounts_in_message":[],"_revision_number":3},{"id":"d81665589326d0b1db161d37aa358d446b1a7e98","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-03-11 17:12:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f7ced5d5fe4784db3e5fa68789c9a8c4658cf292","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 19:16:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/4/check/openstack-tox-docs/dcbc333/html/ : SUCCESS in 6m 41s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/4/check/openstack-tox-pep8/e0cf0aa/ : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-noop-api/2f0da25/ : SUCCESS in 48m 04s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-noop-py2-api/ab78ea7/ : SUCCESS in 47m 11s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/9269542/ : SUCCESS in 44m 20s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-noop-py2-api-stable-queens/8464368/ : SUCCESS in 46m 18s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-scenario/ec2e757/ : SUCCESS in 1h 23m 28s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-py2-scenario/130d104/ : SUCCESS in 1h 43m 12s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-py2-scenario-stable-rocky/d957d9a/ : SUCCESS in 1h 24m 18s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-py2-scenario-stable-queens/c35eedd/ : SUCCESS in 1h 34m 51s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-scenario-centos-7/dfc1755/ : FAILURE in 1h 56m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/40318fe/ : SUCCESS in 1h 38m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-scenario-two-node/ec44a09/ : SUCCESS in 1h 48m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-py2-scenario-two-node/125e89c/ : SUCCESS in 1h 54m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/06/626406/4/check/octavia-v2-dsvm-tls-barbican/c7a0fbc/ : SUCCESS in 49m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f662ccbc45db003701d4dfb480ad63c2d2e7c833","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-10 23:46:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"64e37c191813f812b60e6173cc26cbb68186b2a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-10 23:46:30.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"195ecab3f8fafce1923abfa89f9a11dfece67473","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 02:11:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/5/check/openstack-tox-docs/36eb36d/html/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/5/check/openstack-tox-pep8/0d4f1eb/ : SUCCESS in 5m 38s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-noop-api/daca86a/ : SUCCESS in 36m 28s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-noop-py2-api/fccbf82/ : SUCCESS in 56m 28s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-noop-api-stable-stein/c3425e7/ : SUCCESS in 42m 35s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/08add0e/ : SUCCESS in 48m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-noop-py2-api-stable-queens/19e4a88/ : SUCCESS in 48m 52s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-scenario/7825def/ : SUCCESS in 1h 23m 11s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-py2-scenario/bbbe1c6/ : SUCCESS in 1h 44m 02s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-scenario-stable-stein/4bfdd13/ : SUCCESS in 1h 23m 57s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-py2-scenario-stable-rocky/8e7cf08/ : SUCCESS in 1h 52m 02s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-py2-scenario-stable-queens/72b9ca3/ : FAILURE in 1h 43m 55s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-scenario-centos-7/08d51e1/ : FAILURE in 2h 06m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/46a6ada/ : SUCCESS in 1h 52m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-scenario-two-node/d2c81a8/ : SUCCESS in 1h 20m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-py2-scenario-two-node/1aedce4/ : SUCCESS in 1h 25m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-tls-barbican/4f18c8e/ : FAILURE in 9m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-tls-barbican-stable-stein/6011785/ : SUCCESS in 59m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-tls-barbican-stable-rocky/c706881/ : SUCCESS in 56m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/06/626406/5/check/octavia-v2-dsvm-tls-barbican-stable-queens/d1e8288/ : SUCCESS in 55m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8a1bad8b3e47a3685280af53ee1b30229d2e8982","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-11 23:56:07.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"be2473843922b2dde0679f296e7153e3e3d69ce3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-11 23:58:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dfc15d000fab44c8a96c27f0dab300fef5e70fd7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 00:55:49.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"3c49e786b7d53f68728bbb57db412bcda7577072","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 03:30:54.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"c162090b58251fdb82ad3128bf75fddc239248ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 06:27:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/8/check/openstack-tox-docs/7bb2a51/html/ : SUCCESS in 6m 56s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/8/check/openstack-tox-pep8/0187871/ : SUCCESS in 5m 06s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-noop-api/fcb1144/ : SUCCESS in 41m 13s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-noop-py2-api/b5864d5/ : SUCCESS in 44m 43s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-noop-api-stable-stein/51cb8d4/ : SUCCESS in 44m 20s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/f0f356f/ : SUCCESS in 50m 22s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-noop-py2-api-stable-queens/117097c/ : SUCCESS in 41m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-scenario/50a1aed/ : FAILURE in 2h 04m 48s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-py2-scenario/30e303f/ : SUCCESS in 1h 28m 34s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-scenario-stable-stein/8ce621d/ : SUCCESS in 1h 40m 07s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-py2-scenario-stable-rocky/61c9913/ : SUCCESS in 1h 18m 20s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-py2-scenario-stable-queens/01570e9/ : FAILURE in 1h 31m 27s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-scenario-centos-7/05e9f56/ : FAILURE in 2h 35m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-scenario-ubuntu-bionic/89e2962/ : SUCCESS in 1h 31m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-scenario-two-node/e0273da/ : SUCCESS in 1h 49m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-py2-scenario-two-node/75aec47/ : SUCCESS in 1h 45m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-tls-barbican/5a5cfe9/ : SUCCESS in 57m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-tls-barbican-stable-stein/1273adb/ : SUCCESS in 46m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-tls-barbican-stable-rocky/206f646/ : SUCCESS in 1h 01m 16s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/06/626406/8/check/octavia-v2-dsvm-tls-barbican-stable-queens/1565b27/ : SUCCESS in 50m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"84ae01c7baf01caa8f0eda3c7b5ae393d7a0f431","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 16:59:44.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"7c7ee3757dc4e9d7907cc9e836fd41b220fa465f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:32:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/9/check/openstack-tox-docs/82b448b/html/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/9/check/openstack-tox-pep8/f0fdc15/ : SUCCESS in 3m 59s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-noop-api/d0c5007/ : SUCCESS in 43m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-noop-py2-api/9532433/ : SUCCESS in 41m 20s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-noop-api-stable-stein/f1d7bbf/ : SUCCESS in 43m 47s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/7c7561f/ : SUCCESS in 42m 55s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-noop-py2-api-stable-queens/3524abc/ : SUCCESS in 43m 37s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-scenario/240afba/ : SUCCESS in 1h 35m 39s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-py2-scenario/e0bad7f/ : SUCCESS in 1h 27m 47s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-scenario-stable-stein/044eb03/ : SUCCESS in 1h 32m 31s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-py2-scenario-stable-rocky/73672f8/ : SUCCESS in 1h 36m 31s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-py2-scenario-stable-queens/a0bccd9/ : SUCCESS in 2h 00m 00s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-scenario-centos-7/218826a/ : FAILURE in 2h 05m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-scenario-ubuntu-bionic/89436d0/ : SUCCESS in 1h 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-scenario-two-node/259922a/ : SUCCESS in 1h 52m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-py2-scenario-two-node/becad05/ : SUCCESS in 1h 30m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-tls-barbican/595d87b/ : SUCCESS in 51m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-tls-barbican-stable-stein/d48a3df/ : SUCCESS in 56m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-tls-barbican-stable-rocky/abf2f45/ : SUCCESS in 53m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/06/626406/9/check/octavia-v2-dsvm-tls-barbican-stable-queens/067b918/ : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"90e51dbffe166cff49209b145a669d9f64070b08","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-14 00:16:13.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"5ba2ed5769a630f090e3212c03bab69f82683220","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-14 01:36:06.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"dce4fe2b34dda92e1c78e7ca3ecc74af9bb803ac","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-06-14 02:15:34.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"029df34dfc8373ac373424ff7ff6eea256edab3f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-14 11:33:59.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cee3fb332bd4624a81f80be133267400abc9e99e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-14 16:08:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"239eaadc09d2f9e8892b195751eaf0dc20517992","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-14 16:10:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0912b96689095bfe935e32e68c8bc6573f4e77c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 19:52:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/11/check/openstack-tox-docs/92027e2/html/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/11/check/openstack-tox-pep8/012fd8d/ : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-api/475ec89/ : SUCCESS in 40m 16s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-py2-api/4f04664/ : SUCCESS in 41m 29s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-api-stable-stein/94ee02d/ : SUCCESS in 45m 05s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/97691fc/ : SUCCESS in 41m 15s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-py2-api-stable-queens/664c724/ : SUCCESS in 43m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario/0436b0a/ : SUCCESS in 1h 24m 36s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-py2-scenario/5ffda32/ : SUCCESS in 1h 41m 16s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario-stable-stein/7b3996e/ : SUCCESS in 1h 35m 53s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-py2-scenario-stable-rocky/f453ae5/ : SUCCESS in 1h 34m 24s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-py2-scenario-stable-queens/9d28125/ : SUCCESS in 1h 35m 08s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario-centos-7/6b4135c/ : FAILURE in 1h 58m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario-ubuntu-bionic/84404a0/ : SUCCESS in 1h 51m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario-two-node/e410bb7/ : SUCCESS in 1h 24m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-py2-scenario-two-node/e093319/ : SUCCESS in 1h 52m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-tls-barbican/28577d6/ : SUCCESS in 1h 02m 26s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-tls-barbican-stable-stein/0548ce3/ : SUCCESS in 53m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-tls-barbican-stable-rocky/94e47f0/ : SUCCESS in 56m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-tls-barbican-stable-queens/daaa24b/ : SUCCESS in 55m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"62b3de01efa5711268e57702be3ad78d5a8872a9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-14 23:11:03.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b5aaf18a7cd41d5ec2cecb087c8f034eaf058b49","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-06-15 01:54:24.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"e208b47edacfffc5c11d1254a0c1e2a99a206869","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-06-17 15:11:14.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"943b0ea599469df844ed5a21705b58c908aadf69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 15:11:30.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5320aa00b28f0043cba35964e2a1aa3141cb0cd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 17:28:45.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/11/gate/openstack-tox-docs/ad9457b/html/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/11/gate/openstack-tox-pep8/e966d14/ : SUCCESS in 4m 58s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-api/03926cd/ : SUCCESS in 43m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-py2-api/3575a08/ : SUCCESS in 44m 55s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-api-stable-stein/f57972f/ : SUCCESS in 51m 45s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/1e1134b/ : SUCCESS in 41m 00s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/19a97f9/ : SUCCESS in 57m 07s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-scenario/330b0b4/ : SUCCESS in 1h 42m 09s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-py2-scenario/b011aca/ : SUCCESS in 1h 39m 40s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-scenario-stable-stein/ca8c1b5/ : SUCCESS in 1h 42m 21s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/2c17ea2/ : SUCCESS in 1h 42m 04s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-py2-scenario-stable-queens/a097721/ : FAILURE in 1h 09m 52s","accounts_in_message":[],"_revision_number":11},{"id":"fe7470ff83cedcf7ca941ccd17cf14a39cef4147","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-17 18:03:24.000000000","message":"Patch Set 11:\n\nrecheck\nNova no hosts error","accounts_in_message":[],"_revision_number":11},{"id":"4a8cbc96036be34eada8c01dd8a6b39c625aedfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 20:26:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/11/check/openstack-tox-docs/1c44df2/html/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/11/check/openstack-tox-pep8/b5fdf07/ : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-api/e4c3b6c/ : SUCCESS in 43m 18s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-py2-api/8d9b351/ : SUCCESS in 40m 58s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-api-stable-stein/781fa4e/ : SUCCESS in 48m 23s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/5428b40/ : SUCCESS in 37m 56s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-noop-py2-api-stable-queens/956de2d/ : SUCCESS in 39m 24s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario/4fa4983/ : SUCCESS in 1h 34m 38s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-py2-scenario/e4d7f26/ : SUCCESS in 1h 41m 46s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario-stable-stein/b0f196e/ : SUCCESS in 1h 29m 53s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-py2-scenario-stable-rocky/43cc7ee/ : SUCCESS in 1h 28m 37s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-py2-scenario-stable-queens/c76fe2b/ : SUCCESS in 1h 30m 06s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario-centos-7/d41a8f1/ : FAILURE in 2h 06m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario-ubuntu-bionic/21ad79d/ : SUCCESS in 1h 43m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-scenario-two-node/a5a68be/ : SUCCESS in 1h 54m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-py2-scenario-two-node/83c6291/ : FAILURE in 2h 05m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario http://logs.openstack.org/06/626406/11/check/octavia-v2-act-stdby-dsvm-scenario/92fc1d4/ : SUCCESS in 1h 00m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario http://logs.openstack.org/06/626406/11/check/octavia-v2-act-stdby-dsvm-py2-scenario/788886f/ : SUCCESS in 1h 08m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-centos-7 http://logs.openstack.org/06/626406/11/check/octavia-v2-act-stdby-dsvm-py2-scenario-centos-7/dacd5a8/ : FAILURE in 58m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein http://logs.openstack.org/06/626406/11/check/octavia-v2-act-stdby-dsvm-scenario-stable-stein/727cc2c/ : SUCCESS in 1h 03m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/11/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky/1ad724a/ : SUCCESS in 1h 03m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/11/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens/c5a9ea5/ : SUCCESS in 1h 06m 11s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-tls-barbican/d13654c/ : SUCCESS in 59m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-tls-barbican-stable-stein/0008fb7/ : SUCCESS in 1h 02m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-tls-barbican-stable-rocky/fac2d6b/ : SUCCESS in 56m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/06/626406/11/check/octavia-v2-dsvm-tls-barbican-stable-queens/017d768/ : SUCCESS in 1h 00m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"97fdf9288a9a6b98e51e0f28add305a30cb98e97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 20:27:00.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"f5efe7e6f8e49462e71e1aa578dad91c2fce3073","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 22:34:48.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/626406/11/gate/openstack-tox-docs/2b02bf5/html/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/06/626406/11/gate/openstack-tox-pep8/f408818/ : SUCCESS in 4m 23s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-api/2784ff5/ : SUCCESS in 55m 37s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-py2-api/ce635f5/ : SUCCESS in 37m 53s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-api-stable-stein/8d9e451/ : SUCCESS in 41m 13s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/31a3b9d/ : SUCCESS in 40m 33s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/6945ec4/ : SUCCESS in 41m 45s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-scenario/bbfb159/ : SUCCESS in 1h 27m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-py2-scenario/9bb67ac/ : SUCCESS in 1h 25m 09s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-scenario-stable-stein/e194a56/ : SUCCESS in 1h 30m 46s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/74a1f3c/ : SUCCESS in 1h 32m 34s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/06/626406/11/gate/octavia-v2-dsvm-py2-scenario-stable-queens/8eb1d4a/ : SUCCESS in 1h 41m 54s","accounts_in_message":[],"_revision_number":11},{"id":"6b98a385cd61974cb582a39a912c2e9cc07119a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 22:34:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"070297f2881e8328e151f2deb88ef4f23a22d88b","revisions":{"1bff5ef09e755429129d58f57c21965fe3bb6f1d":{"kind":"REWORK","_number":1,"created":"2018-12-19 23:02:14.000000000","uploader":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"ref":"refs/changes/06/626406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/1"}}},"commit":{"parents":[{"commit":"38f574dcb0e1c1601b95eda372520fe9d1b7ae1d","subject":"Merge \"Adds a mixed IPv4/IPv6 members traffic test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/38f574dcb0e1c1601b95eda372520fe9d1b7ae1d"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 23:02:13.000000000","tz":-480},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.openstack.org/624835\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1bff5ef09e755429129d58f57c21965fe3bb6f1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1bff5ef09e755429129d58f57c21965fe3bb6f1d"}]},"branch":"refs/heads/master"},"57aa4e630966d96d553d9357beb2ab584c112a4e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-02-13 13:36:36.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/06/626406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/2"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-02-13 13:36:36.000000000","tz":0},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.openstack.org/624835\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/57aa4e630966d96d553d9357beb2ab584c112a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/57aa4e630966d96d553d9357beb2ab584c112a4e"}]},"branch":"refs/heads/master"},"c0a5a8b568ed4b7805289e355f5599a7bcf5d4b2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-02-13 22:05:13.000000000","uploader":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"ref":"refs/changes/06/626406/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/3"}}},"commit":{"parents":[{"commit":"c4ccd81154959ca0dba6b7796a866fb82f8ba36b","subject":"Use cls.skipException and raise it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c4ccd81154959ca0dba6b7796a866fb82f8ba36b"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2019-02-13 22:05:13.000000000","tz":0},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.openstack.org/624835\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c0a5a8b568ed4b7805289e355f5599a7bcf5d4b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c0a5a8b568ed4b7805289e355f5599a7bcf5d4b2"}]},"branch":"refs/heads/master"},"7c638d0a86a9b10552fa3fd357ca3dd8d2a966a6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-11 17:12:38.000000000","uploader":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"ref":"refs/changes/06/626406/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/4"}}},"commit":{"parents":[{"commit":"1f0f4ef028cdd629575aed7082eec558054ee89a","subject":"Merge \"Fix barbican service_available check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f0f4ef028cdd629575aed7082eec558054ee89a"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2019-03-11 17:12:38.000000000","tz":0},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.openstack.org/624835\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7c638d0a86a9b10552fa3fd357ca3dd8d2a966a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7c638d0a86a9b10552fa3fd357ca3dd8d2a966a6"}]},"branch":"refs/heads/master"},"c9cd60ff752a2ff38654437a224adc36981d8925":{"kind":"REWORK","_number":5,"created":"2019-06-10 23:46:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/626406/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/5"}}},"commit":{"parents":[{"commit":"735c71d5099c0ad693bb95511536db622b23fe9b","subject":"Merge \"Fix the IPv6 tests for no ipv6-public-subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735c71d5099c0ad693bb95511536db622b23fe9b"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-10 23:46:02.000000000","tz":-420},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.openstack.org/624835\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c9cd60ff752a2ff38654437a224adc36981d8925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c9cd60ff752a2ff38654437a224adc36981d8925"}]},"branch":"refs/heads/master"},"48b99793682393d5cee08148a05ea23dfd05db35":{"kind":"REWORK","_number":6,"created":"2019-06-11 23:58:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/626406/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/6"}}},"commit":{"parents":[{"commit":"735c71d5099c0ad693bb95511536db622b23fe9b","subject":"Merge \"Fix the IPv6 tests for no ipv6-public-subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735c71d5099c0ad693bb95511536db622b23fe9b"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-11 23:58:48.000000000","tz":-420},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.openstack.org/624835\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/48b99793682393d5cee08148a05ea23dfd05db35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/48b99793682393d5cee08148a05ea23dfd05db35"}]},"branch":"refs/heads/master"},"fe42bee76e9f9b765f141c5b9e5f2dfab4b44641":{"kind":"REWORK","_number":7,"created":"2019-06-12 00:55:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/626406/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/7"}}},"commit":{"parents":[{"commit":"735c71d5099c0ad693bb95511536db622b23fe9b","subject":"Merge \"Fix the IPv6 tests for no ipv6-public-subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735c71d5099c0ad693bb95511536db622b23fe9b"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 00:55:46.000000000","tz":0},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.openstack.org/624835\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fe42bee76e9f9b765f141c5b9e5f2dfab4b44641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fe42bee76e9f9b765f141c5b9e5f2dfab4b44641"}]},"branch":"refs/heads/master"},"985286acba9fe9de3197d67763f55320ac561a82":{"kind":"REWORK","_number":8,"created":"2019-06-12 03:30:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/626406/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/8"}}},"commit":{"parents":[{"commit":"735c71d5099c0ad693bb95511536db622b23fe9b","subject":"Merge \"Fix the IPv6 tests for no ipv6-public-subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735c71d5099c0ad693bb95511536db622b23fe9b"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 03:30:49.000000000","tz":0},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.openstack.org/624835\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/985286acba9fe9de3197d67763f55320ac561a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/985286acba9fe9de3197d67763f55320ac561a82"}]},"branch":"refs/heads/master"},"1e6ea2b79912bc0b11887daf2d7fbadc92f8fb64":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-06-12 16:59:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/626406/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/9"}}},"commit":{"parents":[{"commit":"735c71d5099c0ad693bb95511536db622b23fe9b","subject":"Merge \"Fix the IPv6 tests for no ipv6-public-subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735c71d5099c0ad693bb95511536db622b23fe9b"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 16:59:42.000000000","tz":0},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.opendev.org/#/c/624835/\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1e6ea2b79912bc0b11887daf2d7fbadc92f8fb64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1e6ea2b79912bc0b11887daf2d7fbadc92f8fb64"}]},"branch":"refs/heads/master"},"749e23769841138b180519939d722e1a3cc7287e":{"kind":"REWORK","_number":10,"created":"2019-06-14 16:08:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/626406/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/10"}}},"commit":{"parents":[{"commit":"735c71d5099c0ad693bb95511536db622b23fe9b","subject":"Merge \"Fix the IPv6 tests for no ipv6-public-subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735c71d5099c0ad693bb95511536db622b23fe9b"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-14 16:08:53.000000000","tz":-420},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.opendev.org/#/c/624835/\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/749e23769841138b180519939d722e1a3cc7287e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/749e23769841138b180519939d722e1a3cc7287e"}]},"branch":"refs/heads/master"},"070297f2881e8328e151f2deb88ef4f23a22d88b":{"kind":"REWORK","_number":11,"created":"2019-06-14 16:10:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/626406/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/06/626406/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/06/626406/11"}}},"commit":{"parents":[{"commit":"735c71d5099c0ad693bb95511536db622b23fe9b","subject":"Merge \"Fix the IPv6 tests for no ipv6-public-subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735c71d5099c0ad693bb95511536db622b23fe9b"}]}],"author":{"name":"German Eichberger","email":"German.eichberger@rackspace.com","date":"2018-12-19 22:59:45.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-14 16:10:18.000000000","tz":-420},"subject":"Save amphora logs in gate","message":"Save amphora logs in gate\n\nDepends-On: https://review.opendev.org/#/c/624835/\nChange-Id: Id9004adf8952d6ef82b7bda6f2d3b509655f1c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/070297f2881e8328e151f2deb88ef4f23a22d88b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/070297f2881e8328e151f2deb88ef4f23a22d88b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
