)]}'
{"id":"openstack%2Foctavia-tempest-plugin~633295","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I4b1a1f48d2f619b883619811539ddb262d6b5f45","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"amp-update-api","hashtags":[],"change_id":"I4b1a1f48d2f619b883619811539ddb262d6b5f45","subject":"Add amphora update service client and API test","status":"MERGED","created":"2019-01-25 23:46:08.000000000","updated":"2019-08-05 04:59:38.000000000","submitted":"2019-08-05 04:59:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"633295-1564981178415-c5f855de","meta_rev_id":"1aa153baaaefbc3aed5cc1edb91b2935023e836c","_number":633295,"virtual_id_number":633295,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-08-05 04:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-31 15:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-07-31 15:20:06.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-05 03:12:17.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-07-31 15:20:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-07-31 15:20:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-07-31 15:20:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2019-07-31 15:20:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2019-07-31 15:20:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-08-05 03:12:17.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 20:03:36.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-06-13 08:18:32.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-13 08:24:26.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-06-13 09:05:16.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-14 08:20:32.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-06-14 11:50:50.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-06-14 14:37:23.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-07-26 21:37:45.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-05 03:12:17.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-08-05 04:59:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"782126dcd7b100e944f54e5d5321c98e1e400c7c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-01-25 23:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec9c6b955633c3a507a41755257d855a48fd39a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 23:46:43.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"96211be13ab11dcc5153449534118e757d3abd7e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-01-26 00:09:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d3efb710c7b00f5864d2ccc37a5f11e653ee4e48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-01-26 01:37:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f5dca93ec5b556edb918a241fe270a43d1dd289a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-26 01:37:28.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"309032a2f6725c6c2d889916293c96dd350fcdd2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-01-26 01:38:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7267b6b56e222ca4ea243d0391f1f0124ffc56c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-26 01:38:59.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"1812909e3ec000fc9072843146776731a05ef165","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-01-26 01:39:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4d24fe2e3b8b2ec4efabb9ab48b5629685549ac5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-26 04:21:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/4/check/openstack-tox-docs/7f2246d/html/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/4/check/openstack-tox-pep8/35e8f13/ : SUCCESS in 4m 11s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-noop-api/e739e2f/ : SUCCESS in 41m 53s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-noop-py2-api/47accc3/ : SUCCESS in 43m 27s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/9799775/ : SUCCESS in 43m 17s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-noop-py2-api-stable-queens/f2a2778/ : SUCCESS in 41m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-scenario/d308eae/ : SUCCESS in 1h 33m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-py2-scenario/8bf9a78/ : SUCCESS in 1h 32m 05s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-py2-scenario-stable-rocky/800519f/ : SUCCESS in 2h 01m 21s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-py2-scenario-stable-queens/ccd7b10/ : SUCCESS in 1h 25m 52s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-scenario-centos-7/974b22f/ : FAILURE in 2h 35m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/53552c0/ : SUCCESS in 1h 34m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-scenario-two-node/aaf51e0/ : SUCCESS in 1h 44m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-py2-scenario-two-node/63614c6/ : SUCCESS in 1h 45m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/4/check/octavia-v2-dsvm-tls-barbican/b722dd4/ : SUCCESS in 52m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"70891cff2bb5d9466f409b28717dfcf87dbaee79","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-01-28 22:41:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"c278db91a8d596c667da514e72f8ca176262cad3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 01:13:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/5/check/openstack-tox-docs/89cb67c/html/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/5/check/openstack-tox-pep8/8d210e9/ : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-noop-api/de7c76a/ : SUCCESS in 47m 03s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-noop-py2-api/dbb9f95/ : SUCCESS in 40m 52s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/9ca6b4f/ : SUCCESS in 39m 54s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-noop-py2-api-stable-queens/455690c/ : SUCCESS in 45m 47s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-scenario/7f2f5c5/ : SUCCESS in 1h 44m 55s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-py2-scenario/efb51da/ : SUCCESS in 1h 28m 13s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-py2-scenario-stable-rocky/6bf2a8a/ : SUCCESS in 1h 41m 08s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-py2-scenario-stable-queens/ac0753e/ : SUCCESS in 1h 28m 27s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-scenario-centos-7/291eb03/ : SUCCESS in 2h 21m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b05e87f/ : SUCCESS in 1h 43m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-scenario-two-node/1f871ee/ : FAILURE in 1h 47m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-py2-scenario-two-node/f9c8b77/ : SUCCESS in 1h 51m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/5/check/octavia-v2-dsvm-tls-barbican/5748ffe/ : SUCCESS in 54m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c75898dde077b7f775ad2cf5eaa4aade640a84d2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-05 23:59:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"95c8d26afcdf6abd143aad3ed801d9412206de73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-06 02:16:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/6/check/openstack-tox-docs/650c43c/html/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/6/check/openstack-tox-pep8/ba6c332/ : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-noop-api/fdfbe8a/ : SUCCESS in 43m 32s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-noop-py2-api/1c23aba/ : SUCCESS in 41m 38s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/84c3221/ : SUCCESS in 41m 11s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-noop-py2-api-stable-queens/8ca3920/ : SUCCESS in 44m 08s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-scenario/3ca5353/ : SUCCESS in 1h 39m 21s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-py2-scenario/7550f16/ : FAILURE in 1h 38m 51s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-py2-scenario-stable-rocky/4590b66/ : SUCCESS in 1h 18m 07s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-py2-scenario-stable-queens/033277e/ : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-scenario-centos-7/2bf2b86/ : FAILURE in 2h 12m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/af62997/ : SUCCESS in 1h 39m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-scenario-two-node/c6006ac/ : FAILURE in 1h 46m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-py2-scenario-two-node/f77ba53/ : FAILURE in 1h 43m 24s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-tls-barbican/db51aa4/ : SUCCESS in 49m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cd1724befc14d13945c4e787a17b8dc9544048cf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-11 00:32:22.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"466862944caa41f291b684a8c3be7ecddfb57001","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 03:17:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/6/check/openstack-tox-docs/5053d81/html/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/6/check/openstack-tox-pep8/fcb86c1/ : SUCCESS in 4m 18s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-noop-api/3b2e8fd/ : SUCCESS in 59m 02s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-noop-py2-api/f8e7e41/ : SUCCESS in 38m 21s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/96c2901/ : SUCCESS in 54m 07s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-noop-py2-api-stable-queens/bffff0f/ : SUCCESS in 35m 00s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-scenario/4c8a9b7/ : SUCCESS in 1h 27m 38s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-py2-scenario/1e86b52/ : SUCCESS in 1h 34m 14s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-py2-scenario-stable-rocky/918a408/ : SUCCESS in 1h 29m 51s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-py2-scenario-stable-queens/8a06ce4/ : SUCCESS in 1h 34m 36s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-scenario-centos-7/152ce43/ : SUCCESS in 2h 30m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f3a887f/ : SUCCESS in 1h 30m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-scenario-two-node/95571d8/ : FAILURE in 1h 43m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-py2-scenario-two-node/ba8433e/ : SUCCESS in 1h 35m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/6/check/octavia-v2-dsvm-tls-barbican/ac562e3/ : SUCCESS in 50m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3886001e3ebe8458cfcb10df43c2e61d185c169c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-17 18:35:44.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6a74a12bb2cc53d712c4f8da07db5145cd667905","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-22 23:09:38.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4ed495ffd334971c3cdc82f61e46f2708da5f7a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-22 23:11:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bad635960d62e51c0bbdffdfda2b4df581045fa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:40:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/7/check/openstack-tox-docs/78e244d/html/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/7/check/openstack-tox-pep8/8e83382/ : SUCCESS in 4m 40s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-noop-api/2e9bc09/ : SUCCESS in 40m 32s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-noop-py2-api/808c8f7/ : SUCCESS in 45m 06s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-noop-api-stable-stein/cba12be/ : SUCCESS in 39m 28s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/7c595b4/ : SUCCESS in 40m 50s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-noop-py2-api-stable-queens/04cb5d9/ : SUCCESS in 42m 09s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-scenario/acc81f6/ : SUCCESS in 1h 51m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-py2-scenario/bdc6fad/ : SUCCESS in 1h 41m 06s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-scenario-stable-stein/a8e5167/ : SUCCESS in 1h 33m 27s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-py2-scenario-stable-rocky/da3eeac/ : SUCCESS in 1h 44m 56s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-py2-scenario-stable-queens/b6c6ff9/ : SUCCESS in 1h 28m 06s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-scenario-centos-7/dd91365/ : FAILURE in 35m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ae351a3/ : SUCCESS in 1h 40m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-scenario-two-node/61e6f8e/ : FAILURE in 2h 01m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-py2-scenario-two-node/efc254a/ : FAILURE in 1h 41m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/7/check/octavia-v2-dsvm-tls-barbican/8f903b2/ : SUCCESS in 46m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"de904a7e5d06641b0183ea1874f4c86a6bf9e569","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-23 06:41:48.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"da254a33a27e414424231b7ec0dc88fc6927bff8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-23 14:31:43.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"d81bc2db7185904441a388e75cf1f1d6d9176bf4","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-04-28 20:03:36.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2883fc105611fb509da7d91857318387ca94e7c5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-16 15:06:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"295054fa95b3bca44e039b4a4d36813e7e303c57","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-04 18:41:20.000000000","message":"Patch Set 7:\n\nNir: ping","accounts_in_message":[],"_revision_number":7},{"id":"802c936cfd30c457ebf50384f5796a3439ec0399","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 17:07:37.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"d27292593dda693db90d0415219e649bc043f5dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 21:38:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/8/check/openstack-tox-docs/e0cdf95/html/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/8/check/openstack-tox-pep8/1b42157/ : SUCCESS in 4m 18s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-noop-api/ea3aba1/ : SUCCESS in 42m 50s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-noop-py2-api/ed67d2b/ : SUCCESS in 48m 00s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-noop-api-stable-stein/e7a0093/ : SUCCESS in 46m 45s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/8de981b/ : SUCCESS in 46m 25s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-noop-py2-api-stable-queens/174b873/ : SUCCESS in 41m 20s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-scenario/70d0cde/ : SUCCESS in 1h 42m 11s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-py2-scenario/fb00cdf/ : SUCCESS in 1h 33m 21s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-scenario-stable-stein/92b1a5e/ : SUCCESS in 1h 37m 59s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-py2-scenario-stable-rocky/1545137/ : SUCCESS in 1h 30m 18s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-py2-scenario-stable-queens/a7a4804/ : SUCCESS in 1h 43m 15s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-scenario-centos-7/f1ffbc2/ : FAILURE in 2h 20m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b26eca2/ : FAILURE in 2h 01m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-scenario-two-node/e8d34f7/ : SUCCESS in 1h 42m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-py2-scenario-two-node/477c44b/ : SUCCESS in 1h 51m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-tls-barbican/cbc0497/ : SUCCESS in 1h 01m 44s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-tls-barbican-stable-stein/c9b4fd4/ : SUCCESS in 49m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-tls-barbican-stable-rocky/c48767e/ : SUCCESS in 56m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/95/633295/8/check/octavia-v2-dsvm-tls-barbican-stable-queens/c63c4d4/ : SUCCESS in 49m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8d4fbdc044dadb2c0e5987f8a609b5654dc1e625","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-13 08:18:32.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"7d58691bc8051a218846dbb57ee18d99df68a15d","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-06-13 08:24:26.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"2be9363aa3b791bc9ca79310778b2710ef6cd5fb","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-13 09:05:16.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"40e6a6ed615e84c0b7f2310168e2e70b7f8c0774","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-06-14 08:20:32.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"5db41b86e3c85d36b2781427902ad503653d1700","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-06-14 11:50:50.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"0d4678b4bb2d52cadb4dfb952b5102692be8b3db","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:37:23.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"6efc2095facfc5320df083c51017833beb96b712","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-18 09:05:22.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"b7f303d4c760f75ff54c8c9f94958fab5af5e84a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:05:34.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"10b6085926ec395d669d4516c30fc21e7fc38c75","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-16 05:02:53.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"02384b2f501e891567ec439531197adc701b08e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 05:03:07.000000000","message":"Patch Set 10: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  Job octavia-v2-dsvm-scenario-two-node not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-v2-dsvm-py2-scenario-centos-7:\n              voting: false\n          - octavia-v2-dsvm-scenario-ubuntu-bionic:\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-two-node:\n              voting: false\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n          - octavia-v2-dsvm-scenario-amphora-v2:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora28:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":10},{"id":"93461c828219e0525290787f19ceeb503d1f79bb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-16 05:09:54.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"4f343119f4c219fab8b369f58ac1d6c68d72b48c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 07:08:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/11/check/openstack-tox-docs/661313a/html/ : SUCCESS in 2m 50s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/11/check/openstack-tox-pep8/83e4fd4/ : SUCCESS in 4m 06s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-noop-api/64bb6ab/ : SUCCESS in 43m 06s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-noop-py2-api/9237bde/ : SUCCESS in 46m 21s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-noop-api-stable-stein/2ecffd5/ : SUCCESS in 43m 59s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/407e8a2/ : SUCCESS in 41m 34s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-noop-py2-api-stable-queens/903f64d/ : SUCCESS in 42m 59s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-scenario/1dd1e6e/ : SUCCESS in 1h 31m 39s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-py2-scenario/7ac2acf/ : SUCCESS in 1h 22m 38s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-scenario-stable-stein/ae1199f/ : SUCCESS in 1h 33m 40s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-py2-scenario-stable-rocky/202c795/ : SUCCESS in 1h 34m 14s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-py2-scenario-stable-queens/6aa0c85/ : SUCCESS in 1h 27m 10s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-scenario-centos-7/7e0629a/ : POST_FAILURE in 1h 53m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9293bb3/ : SUCCESS in 1h 35m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node http://logs.openstack.org/95/633295/11/check/octavia-v2-act-stdby-dsvm-scenario-two-node/6841708/ : SUCCESS in 1h 41m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/11/check/octavia-v2-act-stdby-dsvm-py2-scenario-two-node/85d29ac/ : SUCCESS in 1h 39m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario http://logs.openstack.org/95/633295/11/check/octavia-v2-act-stdby-dsvm-scenario/67f445e/ : SUCCESS in 1h 01m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario http://logs.openstack.org/95/633295/11/check/octavia-v2-act-stdby-dsvm-py2-scenario/e40650c/ : SUCCESS in 59m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-centos-7 http://logs.openstack.org/95/633295/11/check/octavia-v2-act-stdby-dsvm-py2-scenario-centos-7/011beb1/ : FAILURE in 1h 02m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein http://logs.openstack.org/95/633295/11/check/octavia-v2-act-stdby-dsvm-scenario-stable-stein/dbaeb04/ : SUCCESS in 1h 08m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/11/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky/79d6ee8/ : FAILURE in 1h 24m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/11/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens/0294c38/ : SUCCESS in 1h 00m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-tls-barbican/dc99d7c/ : SUCCESS in 54m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-tls-barbican-stable-stein/c0c4af5/ : SUCCESS in 57m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-tls-barbican-stable-rocky/31f238a/ : SUCCESS in 57m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/95/633295/11/check/octavia-v2-dsvm-tls-barbican-stable-queens/b2b38fb/ : SUCCESS in 56m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"02c8300841d7ee865a66551e38b4aa70a9461f86","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-26 21:37:45.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"166b489cf47647d1b6b4c6a4ad837b3ee12c440e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-30 16:53:03.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"a365db37b70967847bdd8c296a007bc92139f9d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 18:46:08.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/12/check/openstack-tox-docs/3b19a06/ : FAILURE in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/12/check/openstack-tox-pep8/6821bed/ : SUCCESS in 3m 42s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-noop-api/9f69321/ : FAILURE in 38m 22s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-noop-py2-api/863d1f4/ : FAILURE in 36m 33s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-noop-api-stable-stein/010358f/ : FAILURE in 31m 19s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/e970e34/ : FAILURE in 34m 38s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-noop-py2-api-stable-queens/86c1229/ : FAILURE in 29m 30s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-scenario/773cfcd/ : FAILURE in 38m 55s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-py2-scenario/fa16899/ : FAILURE in 39m 01s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-scenario-stable-stein/522d79b/ : FAILURE in 50m 41s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-py2-scenario-stable-rocky/2081a02/ : FAILURE in 54m 10s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-py2-scenario-stable-queens/3a09722/ : FAILURE in 39m 26s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-scenario-centos-7/e56d759/ : FAILURE in 1h 00m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/4274063/ : FAILURE in 45m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node http://logs.openstack.org/95/633295/12/check/octavia-v2-act-stdby-dsvm-scenario-two-node/45dd210/ : FAILURE in 55m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/12/check/octavia-v2-act-stdby-dsvm-py2-scenario-two-node/626a243/ : FAILURE in 47m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/95/633295/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario/4c79ed5/ : FAILURE in 42m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/95/633295/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/abd9476/ : FAILURE in 38m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/95/633295/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/8c4cccf/ : FAILURE in 53m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/95/633295/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/22dbd2c/ : FAILURE in 42m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/8759769/ : FAILURE in 34m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/da1183a/ : FAILURE in 40m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-tls-barbican/448e675/ : FAILURE in 47m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-tls-barbican-stable-stein/36e484a/ : FAILURE in 46m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-tls-barbican-stable-rocky/1ad994e/ : FAILURE in 46m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-tls-barbican-stable-queens/c84c66f/ : FAILURE in 45m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-spare-pool/992fe6d/ : FAILURE in 41m 07s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-py2-spare-pool/27adfd9/ : FAILURE in 40m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-spare-pool-stable-stein/2d28ff9/ : FAILURE in 41m 02s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/06331e6/ : FAILURE in 42m 00s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/95/633295/12/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/3619fc5/ : FAILURE in 31m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a20cdac021b1ba291d3c2098d47f63220ae2054c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-31 15:20:06.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"6cfc7b1bfb9cd6f5a757306d3731ecf15a3dc221","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 17:30:21.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633295/13/check/openstack-tox-docs/2873272/html/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/95/633295/13/check/openstack-tox-pep8/412b8df/ : SUCCESS in 4m 34s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-noop-api/7bf8d3f/ : SUCCESS in 46m 10s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-noop-py2-api/ed5e817/ : SUCCESS in 47m 53s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-noop-api-stable-stein/9c33dd5/ : SUCCESS in 49m 04s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/8ee7a98/ : SUCCESS in 43m 47s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-noop-py2-api-stable-queens/d3ba42a/ : SUCCESS in 48m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-scenario/df4ca7b/ : SUCCESS in 1h 36m 23s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-py2-scenario/0308147/ : SUCCESS in 1h 46m 24s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-scenario-stable-stein/407c481/ : SUCCESS in 1h 26m 39s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-py2-scenario-stable-rocky/8d569d8/ : SUCCESS in 1h 34m 35s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-py2-scenario-stable-queens/a573b8b/ : SUCCESS in 1h 50m 53s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-scenario-centos-7/63b239a/ : FAILURE in 1h 55m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3bcabd4/ : SUCCESS in 1h 52m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node http://logs.openstack.org/95/633295/13/check/octavia-v2-act-stdby-dsvm-scenario-two-node/a689324/ : SUCCESS in 1h 21m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node http://logs.openstack.org/95/633295/13/check/octavia-v2-act-stdby-dsvm-py2-scenario-two-node/e67043a/ : SUCCESS in 1h 48m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/95/633295/13/check/octavia-v2-act-stdby-iptables-dsvm-scenario/0d9d303/ : FAILURE in 1h 17m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/95/633295/13/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/c11398f/ : FAILURE in 1h 23m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/95/633295/13/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/65b0f1a/ : FAILURE in 1h 38m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/95/633295/13/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/75da748/ : SUCCESS in 1h 15m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/95/633295/13/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/a6f23df/ : SUCCESS in 57m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/95/633295/13/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/9a1dad6/ : SUCCESS in 1h 02m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-tls-barbican/7a208e8/ : SUCCESS in 46m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-tls-barbican-stable-stein/2fe84a8/ : SUCCESS in 53m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-tls-barbican-stable-rocky/20653e2/ : SUCCESS in 51m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-tls-barbican-stable-queens/a960281/ : SUCCESS in 57m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-spare-pool/1b20bf9/ : SUCCESS in 1h 11m 48s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-py2-spare-pool/4aa6399/ : FAILURE in 1h 06m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-spare-pool-stable-stein/4a860ca/ : SUCCESS in 1h 11m 11s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/e7cf024/ : SUCCESS in 1h 04m 53s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/95/633295/13/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/3260a00/ : SUCCESS in 56m 18s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ed17ec7dbe1edf88cc478816fd462f181eda0de2","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-05 03:12:17.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"f4cdb1583f71bc27d4d506f1f574450e691ec33e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 03:12:30.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"a7398fdb13b20b09c2636fbd9722d2efaf3bfc64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 04:59:38.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/95/633295/13/gate/openstack-tox-docs/a98ce2d/html/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://logs.opendev.org/95/633295/13/gate/openstack-tox-pep8/21ac745/ : SUCCESS in 3m 30s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-noop-api/bc14208/ : SUCCESS in 46m 31s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-noop-py2-api/9d11483/ : SUCCESS in 44m 14s\n- octavia-v2-dsvm-noop-api-stable-stein https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-noop-api-stable-stein/c883519/ : SUCCESS in 44m 37s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/36d9912/ : SUCCESS in 40m 19s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/3b400e8/ : SUCCESS in 48m 52s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-scenario/d9155bf/ : SUCCESS in 1h 41m 05s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-py2-scenario/11857f2/ : SUCCESS in 1h 42m 05s\n- octavia-v2-dsvm-scenario-stable-stein https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-scenario-stable-stein/b708fdd/ : SUCCESS in 1h 38m 28s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/7963321/ : SUCCESS in 1h 32m 15s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://logs.opendev.org/95/633295/13/gate/octavia-v2-dsvm-py2-scenario-stable-queens/fbb08d7/ : SUCCESS in 1h 37m 14s","accounts_in_message":[],"_revision_number":13},{"id":"359048c1a653b6943518c2897a0fe464e3e27d36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 04:59:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"2b10e0a447fe997720a41cc7e06ca85520b2f04a","revisions":{"ffd4517a922601fe904c47f32a6ea2abbd835424":{"kind":"REWORK","_number":1,"created":"2019-01-25 23:46:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/1"}}},"commit":{"parents":[{"commit":"7007274a616d1cc46845eed2dd76d599aa3b9158","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7007274a616d1cc46845eed2dd76d599aa3b9158"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:45:57.000000000","tz":-480},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ffd4517a922601fe904c47f32a6ea2abbd835424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ffd4517a922601fe904c47f32a6ea2abbd835424"}]},"branch":"refs/heads/master"},"48a289c2cbcc8e0556ed7d4da18907a8592bb5ee":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-01-26 00:09:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/2"}}},"commit":{"parents":[{"commit":"658139223f6808194a4f1bb911e2df6aff6e45a7","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/658139223f6808194a4f1bb911e2df6aff6e45a7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-26 00:09:14.000000000","tz":0},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/48a289c2cbcc8e0556ed7d4da18907a8592bb5ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/48a289c2cbcc8e0556ed7d4da18907a8592bb5ee"}]},"branch":"refs/heads/master"},"2d390932a3cf0d530f72af19439e1e339d747138":{"kind":"REWORK","_number":3,"created":"2019-01-26 01:37:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/3"}}},"commit":{"parents":[{"commit":"7007274a616d1cc46845eed2dd76d599aa3b9158","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7007274a616d1cc46845eed2dd76d599aa3b9158"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-26 01:37:03.000000000","tz":-480},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nDepends-On: https://review.openstack.org/#/c/632842/\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d390932a3cf0d530f72af19439e1e339d747138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d390932a3cf0d530f72af19439e1e339d747138"}]},"branch":"refs/heads/master"},"363991275822fd6b3214c8940655909f847bf874":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-01-26 01:39:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/4"}}},"commit":{"parents":[{"commit":"658139223f6808194a4f1bb911e2df6aff6e45a7","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/658139223f6808194a4f1bb911e2df6aff6e45a7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-26 01:39:13.000000000","tz":0},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nDepends-On: https://review.openstack.org/#/c/632842/\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/363991275822fd6b3214c8940655909f847bf874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/363991275822fd6b3214c8940655909f847bf874"}]},"branch":"refs/heads/master"},"ab3e86198ac9f554f20c98d66920d1fa55dae99e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-01-28 22:41:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/5"}}},"commit":{"parents":[{"commit":"02240bdb20f76753ff5f262f9cf6b2b3e3bc7ef3","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/02240bdb20f76753ff5f262f9cf6b2b3e3bc7ef3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-28 22:41:03.000000000","tz":0},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nDepends-On: https://review.openstack.org/#/c/632842/\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ab3e86198ac9f554f20c98d66920d1fa55dae99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ab3e86198ac9f554f20c98d66920d1fa55dae99e"}]},"branch":"refs/heads/master"},"4f8dbc453d97e008864e70412a3938c04d959e52":{"kind":"REWORK","_number":6,"created":"2019-04-05 23:59:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/6"}}},"commit":{"parents":[{"commit":"bedf84fae88388db94ee30115cbdcaf71834ea38","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bedf84fae88388db94ee30115cbdcaf71834ea38"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-05 23:59:25.000000000","tz":-420},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nDepends-On: https://review.openstack.org/#/c/632842/\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4f8dbc453d97e008864e70412a3938c04d959e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4f8dbc453d97e008864e70412a3938c04d959e52"}]},"branch":"refs/heads/master"},"153c0aed24a415e267ae1925c525bb8886bb2896":{"kind":"REWORK","_number":7,"created":"2019-04-22 23:11:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/7"}}},"commit":{"parents":[{"commit":"bedf84fae88388db94ee30115cbdcaf71834ea38","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bedf84fae88388db94ee30115cbdcaf71834ea38"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-04-22 23:10:57.000000000","tz":-420},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nDepends-On: https://review.openstack.org/#/c/632842/\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/153c0aed24a415e267ae1925c525bb8886bb2896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/153c0aed24a415e267ae1925c525bb8886bb2896"}]},"branch":"refs/heads/master"},"bb77e9564106f243c7a92752645f894b607e4b53":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-06-12 17:07:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/8"}}},"commit":{"parents":[{"commit":"bedf84fae88388db94ee30115cbdcaf71834ea38","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bedf84fae88388db94ee30115cbdcaf71834ea38"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 17:07:35.000000000","tz":0},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb77e9564106f243c7a92752645f894b607e4b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb77e9564106f243c7a92752645f894b607e4b53"}]},"branch":"refs/heads/master"},"7d3b33e8e208ea4b42bc20b714beaaccc89c0f2d":{"kind":"NO_CHANGE","_number":9,"created":"2019-06-18 09:05:22.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/95/633295/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/9"}}},"commit":{"parents":[{"commit":"7714859ed2afbf128dd9b9ca160cca706f0fd151","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7714859ed2afbf128dd9b9ca160cca706f0fd151"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-18 09:05:22.000000000","tz":0},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7d3b33e8e208ea4b42bc20b714beaaccc89c0f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7d3b33e8e208ea4b42bc20b714beaaccc89c0f2d"}]},"branch":"refs/heads/master"},"2000504f05bd301bd27d07caa6dcee012ca56c30":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-07-16 05:02:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/10"}}},"commit":{"parents":[{"commit":"6e356de597ac1816c1190ecd566ef41e3487fcab","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6e356de597ac1816c1190ecd566ef41e3487fcab"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-16 05:02:36.000000000","tz":-420},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2000504f05bd301bd27d07caa6dcee012ca56c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2000504f05bd301bd27d07caa6dcee012ca56c30"}]},"branch":"refs/heads/master"},"def44aa80072748139a80e2032ed2ac61a203dcb":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-07-16 05:09:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/11"}}},"commit":{"parents":[{"commit":"1a254d130e95097213acd80f824a16d499552fb1","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1a254d130e95097213acd80f824a16d499552fb1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-16 05:09:54.000000000","tz":0},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/def44aa80072748139a80e2032ed2ac61a203dcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/def44aa80072748139a80e2032ed2ac61a203dcb"}]},"branch":"refs/heads/master"},"2c7369cb8d62cbf78fee0054513aa49772777b1b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-07-30 16:53:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/12"}}},"commit":{"parents":[{"commit":"20ff3317b3e63356ea6b7dcaf04540ed59e66ce8","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/20ff3317b3e63356ea6b7dcaf04540ed59e66ce8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 16:53:03.000000000","tz":0},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2c7369cb8d62cbf78fee0054513aa49772777b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2c7369cb8d62cbf78fee0054513aa49772777b1b"}]},"branch":"refs/heads/master"},"2b10e0a447fe997720a41cc7e06ca85520b2f04a":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-07-31 15:20:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/633295/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/633295/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/633295/13"}}},"commit":{"parents":[{"commit":"2734ffe0ea8568f19d9d15cda3329df3de35871d","subject":"Add a flavor to the load balancer CRUD scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2734ffe0ea8568f19d9d15cda3329df3de35871d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-01-25 23:42:13.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 15:20:06.000000000","tz":0},"subject":"Add amphora update service client and API test","message":"Add amphora update service client and API test\n\nThis patch adds a service client for the amphora agent configuration update\nAPI and adds an API test for the amphora configuration update API.\n\nIt also adds the service client for amphora failover and cleans up some\nclient credintials in the amphora scenario tests.\n\nChange-Id: I4b1a1f48d2f619b883619811539ddb262d6b5f45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2b10e0a447fe997720a41cc7e06ca85520b2f04a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2b10e0a447fe997720a41cc7e06ca85520b2f04a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
