)]}'
{"id":"openstack%2Foctavia-tempest-plugin~634988","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I87249017453628e3d2cd320a02677d81397f1b26","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"spare-pool","hashtags":[],"change_id":"I87249017453628e3d2cd320a02677d81397f1b26","subject":"Add a scenario test for spare pool","status":"MERGED","created":"2019-02-05 17:18:30.000000000","updated":"2019-07-19 22:12:06.000000000","submitted":"2019-07-19 22:12:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"634988-1563574326857-8d65fae4","meta_rev_id":"a5de7974bb4570cc1930262a470116ace9eaa804","_number":634988,"virtual_id_number":634988,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-07-19 17:45:23.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-19 22:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"recommended":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":1,"date":"2019-07-14 02:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-07-17 03:44:52.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-07-13 16:24:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-05-27 08:25:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-06-18 17:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-05 10:46:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-07-17 03:44:52.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-03 12:33:16.000000000","updated_by":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"reviewer":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"state":"REVIEWER"},{"updated":"2019-04-08 09:16:32.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-04-16 07:32:22.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-05-27 08:25:01.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-06-05 10:46:35.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-06-18 17:45:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-13 16:24:17.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-14 02:34:19.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-07-17 03:44:52.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-07-19 22:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d6acc3a4426bfdd3afa855dfd0e5a4bddaf9a18","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-05 17:18:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e94b379568a4acc5158c650bf200c2db791cc2b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-05 17:20:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a3685a756e0858587defa6f7966d3d52ecaa1bda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 19:24:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/2/check/openstack-tox-docs/6fefbda/html/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/2/check/openstack-tox-pep8/581c8f9/ : SUCCESS in 4m 53s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-noop-api/ff65257/ : SUCCESS in 48m 19s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-noop-py2-api/77e0a5b/ : SUCCESS in 47m 00s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/7a7a719/ : SUCCESS in 46m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-noop-py2-api-stable-queens/5ef3e04/ : SUCCESS in 43m 37s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-scenario/7ecc021/ : SUCCESS in 1h 32m 18s\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-spare-pool/792a7f8/ : FAILURE in 58m 28s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-py2-scenario/eb6e750/ : SUCCESS in 1h 28m 01s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-py2-scenario-stable-rocky/a5d52f3/ : SUCCESS in 1h 26m 50s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-py2-scenario-stable-queens/3b501c9/ : SUCCESS in 1h 33m 52s\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-py2-spare-pool/48da9b8/ : FAILURE in 54m 22s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-scenario-centos-7/e845d65/ : FAILURE in 32m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/5e4aff4/ : SUCCESS in 1h 29m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-scenario-two-node/240d7d8/ : FAILURE in 1h 51m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-py2-scenario-two-node/79031c0/ : FAILURE in 1h 54m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/2/check/octavia-v2-dsvm-tls-barbican/a3141c4/ : SUCCESS in 41m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"091410980bf23d0f4a4d62195acb56fee85ab944","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-05 21:36:35.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9104c63b07c4abec726d5b3fcc3ff3feefde777d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-02-05 21:49:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae6b8b91837d79547fef282a572eb8decaeb9cac","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-05 21:55:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0e8971ecb345e1f7bace1417825c7e1582900de8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 22:49:22.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"52e573a1990c43fb24c70c95b36ef1a3e6b903a1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-06 05:43:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d8fddcfe9e8f6ab979ca7e6d67ae829b491abc4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 07:33:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/4/check/openstack-tox-docs/ea9b110/html/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/4/check/openstack-tox-pep8/0f50500/ : SUCCESS in 4m 32s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-noop-api/e912fc7/ : SUCCESS in 41m 12s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-noop-py2-api/95762e4/ : SUCCESS in 38m 06s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/7cd5561/ : SUCCESS in 39m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-noop-py2-api-stable-queens/57d506d/ : SUCCESS in 40m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-scenario/969c7c0/ : SUCCESS in 1h 24m 56s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-py2-scenario/5a2fbf0/ : SUCCESS in 1h 15m 06s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-py2-scenario-stable-rocky/55b3791/ : SUCCESS in 1h 22m 17s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-py2-scenario-stable-queens/1146c81/ : SUCCESS in 1h 12m 01s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-scenario-centos-7/3a05145/ : FAILURE in 31m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/70ffff4/ : SUCCESS in 1h 48m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-scenario-two-node/0ff1502/ : SUCCESS in 1h 09m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-py2-scenario-two-node/f291d1f/ : FAILURE in 39m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-tls-barbican/d3fe63f/ : SUCCESS in 47m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-spare-pool/fe6a4d6/ : SUCCESS in 48m 15s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/4/check/octavia-v2-dsvm-py2-spare-pool/968158f/ : SUCCESS in 41m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"92e4572241eba0cf962017e3a8ea0045daf8bfd0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-03-13 17:11:40.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"619830530f310e565b2d2cbfcb58ccf5e6b003aa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-03-13 20:40:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"49855fbff9d37c67a788e81f98dfb7e8807e6432","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-03-18 09:41:57.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"076100eca611daa4c74eae26ddbc1a9acb6e0226","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-03-18 09:42:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4ea445e660b81cac760c585bb686ec9afebced5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 11:42:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/5/check/openstack-tox-docs/e2110e3/html/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/5/check/openstack-tox-pep8/964819e/ : SUCCESS in 6m 04s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-noop-api/8176b59/ : SUCCESS in 38m 35s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-noop-py2-api/dc993ee/ : SUCCESS in 41m 31s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/4547005/ : SUCCESS in 48m 22s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-noop-py2-api-stable-queens/5e7accb/ : SUCCESS in 41m 20s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-scenario/5a5b016/ : SUCCESS in 1h 28m 17s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-py2-scenario/54c5bf2/ : SUCCESS in 1h 46m 36s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-py2-scenario-stable-rocky/d3e6503/ : SUCCESS in 1h 41m 16s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-py2-scenario-stable-queens/3774f9a/ : SUCCESS in 1h 27m 51s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-scenario-centos-7/3de5bb8/ : FAILURE in 1h 56m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b423b7f/ : SUCCESS in 1h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-scenario-two-node/b111634/ : SUCCESS in 1h 44m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-py2-scenario-two-node/e154c77/ : FAILURE in 1h 21m 14s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-tls-barbican/3180e20/ : SUCCESS in 50m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-spare-pool/a03dac3/ : SUCCESS in 56m 36s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/5/check/octavia-v2-dsvm-py2-spare-pool/2540091/ : SUCCESS in 1h 07m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e24ec633e4340ce8e46494a62b7236a8a57e9f9b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-03-26 09:26:30.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2ea488f4521c4192001b5904b0ae95cc19f68f4e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-03-29 18:43:59.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"1d3dcda71d2b6bb1f42bb35974fdb18e1cde84f3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-02 15:57:28.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e3ba501c71c63d94fdc502e2cfff1d7f6383b465","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-02 17:22:20.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"864443b8ac29519411d2f9f5f8ba1d9d0f10b341","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 19:57:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/6/check/openstack-tox-docs/4821a76/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/6/check/openstack-tox-pep8/16e9353/ : SUCCESS in 4m 07s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-api/120ba37/ : SUCCESS in 52m 11s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api/d5dfd7a/ : SUCCESS in 41m 27s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/035d3e2/ : SUCCESS in 44m 19s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api-stable-queens/07b3c23/ : SUCCESS in 41m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario/857de53/ : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario/9895360/ : SUCCESS in 1h 37m 39s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-stable-rocky/0e4a248/ : SUCCESS in 1h 47m 32s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-stable-queens/558fe60/ : SUCCESS in 2h 24m 55s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-centos-7/6c3b520/ : FAILURE in 1h 54m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b853c06/ : SUCCESS in 1h 41m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-two-node/4f7f395/ : SUCCESS in 1h 41m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-two-node/1114d43/ : FAILURE in 2h 08m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-tls-barbican/940ad13/ : FAILURE in 10m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-spare-pool/71dee75/ : SUCCESS in 1h 01m 54s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-spare-pool/7d835c7/ : SUCCESS in 54m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e585e6e5458aaf82de5a3e55943ef59ee9d3994e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-03 17:28:45.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"7da2020d2585a9ae7c1e5d33941d33036b8f40af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 19:43:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/6/check/openstack-tox-docs/600850b/html/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/6/check/openstack-tox-pep8/7a05293/ : SUCCESS in 5m 05s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-api/860d372/ : SUCCESS in 49m 08s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api/52d3786/ : SUCCESS in 45m 49s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/35147f8/ : SUCCESS in 44m 37s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api-stable-queens/5d8f2a3/ : SUCCESS in 47m 06s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario/57584de/ : SUCCESS in 1h 29m 45s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario/2ef3ec5/ : FAILURE in 1h 55m 44s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-stable-rocky/aefdd49/ : SUCCESS in 1h 33m 14s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-stable-queens/4089624/ : SUCCESS in 1h 26m 41s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-centos-7/b4fdf16/ : FAILURE in 2h 04m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/d6a9be1/ : SUCCESS in 1h 41m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-two-node/f011af3/ : SUCCESS in 1h 34m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-two-node/1a455d1/ : SUCCESS in 1h 32m 56s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-tls-barbican/b23e96b/ : FAILURE in 11m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-spare-pool/16fe66b/ : SUCCESS in 56m 41s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-spare-pool/6602681/ : FAILURE in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ab4396b0acadfd58a308fe100453c518557b8683","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-03 20:23:30.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e490839fa5f00fe0847bad826a2f7419cebb5074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 21:14:46.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"df609ffcdbed298cbd2fcade9059fbba07d71532","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-03 21:15:11.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6483f0c9f5f6e31fa24fd57455a80b5423ce7660","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-03 23:49:11.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"51f0d843c54cd68d5dfaaa3c104ca4149ef8846c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 01:20:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/6/check/openstack-tox-docs/31ea415/html/ : SUCCESS in 6m 15s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/6/check/openstack-tox-pep8/315aeba/ : SUCCESS in 5m 47s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-api/fcd24a7/ : SUCCESS in 45m 28s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api/0afc174/ : SUCCESS in 46m 05s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/7df21cc/ : SUCCESS in 43m 51s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-noop-py2-api-stable-queens/f52593d/ : SUCCESS in 49m 39s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario/0ec3cdc/ : SUCCESS in 1h 30m 02s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario/c0c30aa/ : SUCCESS in 1h 37m 44s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-stable-rocky/14b35b5/ : SUCCESS in 1h 37m 47s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-stable-queens/f7743eb/ : SUCCESS in 1h 37m 17s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-centos-7/d6eaaf5/ : FAILURE in 2h 09m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c1db89d/ : SUCCESS in 1h 26m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-scenario-two-node/0aa682d/ : FAILURE in 2h 31m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-scenario-two-node/43d59c2/ : FAILURE in 1h 50m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-tls-barbican/7123452/ : FAILURE in 12m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-spare-pool/cf856b5/ : SUCCESS in 56m 16s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/6/check/octavia-v2-dsvm-py2-spare-pool/e82e228/ : SUCCESS in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"76da04f8e0d06f5f677472ca12bacaac9ae4eb95","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-05 21:04:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ad1656eb5460cc4c579004aab281e0226988e3ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 21:04:45.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error\nwas:\n\n  Job octavia-v2-dsvm-py2-spare-pool-stable/rocky not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - publish-openstack-docs-pti\n        - tempest-plugin-jobs\n      check:\n        jobs:\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-noop-py2-api-stable-rocky\n          - octavia-v2-dsvm-noop-py2-api-stable-queens\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-v2-dsvm-py2-scenario-stable-rocky\n          - octavia-v2-dsvm-py2-scenario-stable-queens\n          - octavia-v2-dsvm-scenario-centos-7:\n              voting: false\n          - octavia-v2-dsvm-scenario-ubuntu-bionic:\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-two-node:\n              voting: false\n          - octavia-v2-dsvm-tls-barbican:\n              voting: false\n          - octavia-v2-dsvm-spare-pool:\n              voting: false\n          - octavia-v2-dsvm-spare-pool-stable-stein:\n              voting: false\n          - octavia-v2-dsvm-py2-spare-pool:\n              voting: false\n          - octavia-v2-dsvm-py2-spare-pool-stable/rocky:\n              voting: false\n          - octavia-v2-dsvm-py2-spare-pool-stable/queens:\n              voting: false\n      gate:\n        queue: octavia\n        jobs:\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-noop-py2-api-stable-rocky\n          - octavia-v2-dsvm-noop-py2-api-stable-queens\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-v2-dsvm-py2-scenario-stable-rocky\n          - octavia-v2-dsvm-py2-scenario-stable-queens\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":7},{"id":"42215d8cd19a3733347be7294de42dbd17d36262","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-05 21:05:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c621c22b33a00538ccbec63ff18ba9baeeeb042c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-05 21:20:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7f6fbdca022aaf7237c74c59ff764684987d3955","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 23:45:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/9/check/openstack-tox-docs/6b43ac4/html/ : SUCCESS in 7m 34s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/9/check/openstack-tox-pep8/7469e75/ : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-noop-api/ead1a15/ : SUCCESS in 55m 31s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-noop-py2-api/eb008a9/ : SUCCESS in 46m 03s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/16d316c/ : SUCCESS in 51m 04s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-noop-py2-api-stable-queens/d1cceee/ : SUCCESS in 41m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-scenario/626c346/ : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-py2-scenario/893d9da/ : SUCCESS in 1h 26m 10s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-py2-scenario-stable-rocky/2c1039d/ : SUCCESS in 1h 40m 16s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-py2-scenario-stable-queens/db8f954/ : SUCCESS in 1h 39m 27s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-scenario-centos-7/a76efb5/ : FAILURE in 2h 16m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-scenario-ubuntu-bionic/8b6cfa2/ : SUCCESS in 1h 41m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-scenario-two-node/df3168b/ : SUCCESS in 1h 39m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-py2-scenario-two-node/f6e1249/ : SUCCESS in 1h 40m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-tls-barbican/15ad03f/ : SUCCESS in 55m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-spare-pool/bbb0066/ : SUCCESS in 1h 17m 25s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-py2-spare-pool/ff38d81/ : SUCCESS in 56m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-spare-pool-stable-stein/7095dfb/ : SUCCESS in 1h 10m 18s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/33dde00/ : FAILURE in 1h 05m 37s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/9/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/100d3df/ : SUCCESS in 52m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a526c6ee580a9072f481c33fa315873f5e94fa94","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-06 00:20:29.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"519a5b7f6e52ba596ca02e115e8170f03a73239f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-08 09:21:48.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"da0ffad209fc556d8c3ece6cdaa4c8d85986931f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-04-16 07:32:22.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"24774e14ec7ab840018366b6f499df2acbb3fd53","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-02 17:49:25.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5b661b9e28dc8992f0d86952ee77b995baa950a9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-08 09:32:53.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f4ee707b17aa355fa5c0993970cf83787b624ead","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-08 09:32:55.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bc464a4e3d4ac204c80bc292a54c2a657447d96a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 11:57:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/10/check/openstack-tox-docs/d7178f6/html/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/10/check/openstack-tox-pep8/b334bbd/ : SUCCESS in 4m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-noop-api/e2bd3c6/ : SUCCESS in 40m 47s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-noop-py2-api/f50818a/ : SUCCESS in 41m 57s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-noop-api-stable-stein/0af4e9f/ : SUCCESS in 37m 04s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/f6ebaaa/ : SUCCESS in 44m 49s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-noop-py2-api-stable-queens/9048af7/ : SUCCESS in 41m 33s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-scenario/2275205/ : SUCCESS in 1h 35m 41s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-py2-scenario/709a20f/ : SUCCESS in 1h 44m 01s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-scenario-stable-stein/3a4ab82/ : SUCCESS in 1h 28m 34s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-py2-scenario-stable-rocky/49ebfdf/ : SUCCESS in 1h 27m 51s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-py2-scenario-stable-queens/742a5b2/ : FAILURE in 2h 21m 20s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-scenario-centos-7/63a02d8/ : FAILURE in 1h 59m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-scenario-ubuntu-bionic/589ee96/ : SUCCESS in 1h 41m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-scenario-two-node/950de5c/ : SUCCESS in 1h 22m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-py2-scenario-two-node/c380ac2/ : FAILURE in 1h 45m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-tls-barbican/29a416c/ : SUCCESS in 54m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-spare-pool/df50161/ : RETRY_LIMIT in 5m 03s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-py2-spare-pool/984b171/ : RETRY_LIMIT in 4m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-spare-pool-stable-stein/bac3228/ : RETRY_LIMIT in 7m 35s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/8b00b88/ : RETRY_LIMIT in 5m 50s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/10/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/d103521/ : RETRY_LIMIT in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"45632de0b49406f47cfa8134b405d0dfc22c4323","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-14 15:31:52.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ebe5176cbdd4e17fdd23f43bb16801e9ed39f911","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-05-14 16:06:08.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"eed00416d34e78f71e48e10205a9c8acfc761471","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-14 16:20:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"fb38fb31a876b62ddffa061a156fc33527e46e2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:43:17.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/check/openstack-tox-docs/5705509/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/check/openstack-tox-pep8/1c148bd/ : SUCCESS in 4m 39s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api/e0a2e7b/ : SUCCESS in 43m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api/e1cc52d/ : SUCCESS in 43m 05s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api-stable-stein/9688952/ : SUCCESS in 50m 39s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/c5571e7/ : SUCCESS in 48m 21s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-queens/bfc36fe/ : SUCCESS in 40m 58s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario/54c14e7/ : SUCCESS in 1h 40m 16s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario/b1df243/ : SUCCESS in 1h 39m 02s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-stable-stein/a75b6e5/ : SUCCESS in 1h 32m 30s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-rocky/ae0be15/ : SUCCESS in 1h 37m 32s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-queens/d731fa1/ : FAILURE in 1h 28m 19s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-centos-7/55f8d59/ : FAILURE in 2h 16m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/fd699dd/ : SUCCESS in 1h 50m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-two-node/a310c94/ : SUCCESS in 1h 58m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-two-node/aa0e5c5/ : SUCCESS in 1h 56m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican/4915fcb/ : SUCCESS in 1h 03m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-stein/3b79830/ : SUCCESS in 58m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-rocky/21f2621/ : SUCCESS in 59m 19s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-queens/05d7db6/ : SUCCESS in 1h 00m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool/47f9517/ : SUCCESS in 1h 00m 19s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool/d63f454/ : SUCCESS in 57m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool-stable-stein/0019d04/ : SUCCESS in 54m 31s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/7b3b8e4/ : SUCCESS in 56m 40s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/19c3543/ : SUCCESS in 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"643ff989328395f5b081dc10ed35d3f831c56c22","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-14 19:17:25.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"74249fe471b44944741f00c2ffbe7699b6ac4ecd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 21:55:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/check/openstack-tox-docs/b56003c/html/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/check/openstack-tox-pep8/2f668ec/ : SUCCESS in 4m 58s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api/962b497/ : SUCCESS in 40m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api/d9a7c24/ : SUCCESS in 42m 36s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api-stable-stein/f2e2cbd/ : SUCCESS in 59m 47s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/28f0c3b/ : SUCCESS in 40m 38s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-queens/eeb9971/ : SUCCESS in 1h 02m 38s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario/6403694/ : SUCCESS in 1h 29m 34s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario/07f6dcc/ : SUCCESS in 1h 28m 45s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-stable-stein/bcaaf1a/ : SUCCESS in 2h 02m 47s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-rocky/7ee7422/ : SUCCESS in 1h 58m 26s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-queens/4c84ebc/ : SUCCESS in 1h 20m 47s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-centos-7/c6b587c/ : FAILURE in 2h 05m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/0d06174/ : SUCCESS in 2h 10m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-two-node/93778a5/ : FAILURE in 1h 52m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-two-node/5d74fe6/ : TIMED_OUT in 2h 36m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican/7a5173a/ : SUCCESS in 1h 21m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-stein/aa4c463/ : SUCCESS in 55m 24s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-rocky/430bd3f/ : SUCCESS in 1h 00m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-queens/f5626a8/ : SUCCESS in 49m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool/aef6112/ : FAILURE in 27m 44s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool/bb25085/ : FAILURE in 39m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool-stable-stein/4574142/ : FAILURE in 39m 28s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/d00876a/ : SUCCESS in 45m 36s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/edcdcf5/ : SUCCESS in 55m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8aff9a648c5e3b1a0301b474c4cf6a481aa70823","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-15 11:43:02.000000000","message":"Patch Set 12:\n\nrecheck\n\nCould not resolve host: bootstrap.pypa.io","accounts_in_message":[],"_revision_number":12},{"id":"78c8641643289a5ac381121200422d7048ce5988","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:12:36.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/check/openstack-tox-docs/3978578/html/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/check/openstack-tox-pep8/a0aaa64/ : SUCCESS in 4m 02s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api/785a765/ : SUCCESS in 1h 05m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api/a7b1c60/ : SUCCESS in 38m 50s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api-stable-stein/a2f6fd5/ : SUCCESS in 35m 04s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/8c7a35e/ : SUCCESS in 45m 31s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-queens/ba14095/ : SUCCESS in 1h 00m 01s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario/6ec6ac5/ : SUCCESS in 1h 27m 29s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario/cc342a8/ : SUCCESS in 1h 48m 57s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-stable-stein/5d0dd20/ : SUCCESS in 1h 48m 45s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-rocky/927edb9/ : SUCCESS in 1h 35m 27s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-queens/99d752e/ : SUCCESS in 2h 24m 13s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-centos-7/e231aa1/ : FAILURE in 2h 09m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/33e2e33/ : SUCCESS in 1h 44m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-two-node/4a0525b/ : SUCCESS in 1h 49m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-two-node/240769a/ : SUCCESS in 1h 43m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican/50a4ed9/ : SUCCESS in 52m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-stein/e641a33/ : SUCCESS in 52m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-rocky/d439436/ : SUCCESS in 59m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-queens/4a4c412/ : SUCCESS in 2h 00m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool/3f85268/ : SUCCESS in 1h 00m 13s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool/2837500/ : SUCCESS in 1h 02m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool-stable-stein/20eb3fc/ : SUCCESS in 1h 11m 53s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/8d7d857/ : FAILURE in 1h 01m 18s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/690c93c/ : SUCCESS in 1h 55m 38s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"627212f38be3c8b3d3456ebb7f486ca740d30aa7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-05-27 08:25:01.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"ba593cbc470db202ba2af725ef4f558bf128ebf0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-04 21:35:31.000000000","message":"Patch Set 12:\n\nIn the rocky test, the housekeeping was unable to build a spare instance due to a nova error. Are we handling that appropriately? It caused the test to fail with a timeout.","accounts_in_message":[],"_revision_number":12},{"id":"05d25b89d916f638e258f93ade8b3067fe5ca1c2","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-06-05 10:46:35.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"fa9122a085a222a6dc76dc2c2c833559c493fc60","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-07 09:34:59.000000000","message":"Patch Set 12:\n\nThe first attempt to create a spare amphora failed, yes, but the housekeeping retried on a second periodic check and it was built okay. The test timed out on a much later phase (wait for the amphora failover to start, L212-217).\n\nThe test timed out around 2019-05-15 12:42:11,323. Looking at logs of health manager, it can be seen that it indeed succeed in failing over the amphora but ended after the test timed out.\n\nhttp://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/8d7d857/controller/logs/screen-o-hm.txt.gz?#_May_15_12_43_57_635477\n\nWe could consider increasing the check_timeout from 60 seconds to something higher in CI.","accounts_in_message":[],"_revision_number":12},{"id":"37b8907cc55229b1a7b596f4ce1ceea8d4f51dbd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 15:07:26.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"1c88e5f9a211cfa20ef5e2a99c5469ddb2a9b42e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 17:15:43.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/check/openstack-tox-docs/626c789/html/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/check/openstack-tox-pep8/af05a84/ : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api/5b9c552/ : SUCCESS in 43m 52s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api/28f2393/ : SUCCESS in 44m 04s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api-stable-stein/55513dd/ : SUCCESS in 42m 04s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/ddbb362/ : SUCCESS in 47m 37s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-queens/0a7b735/ : SUCCESS in 38m 47s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario/bc45dd9/ : SUCCESS in 1h 31m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario/3957285/ : SUCCESS in 1h 34m 54s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-stable-stein/f46db8a/ : SUCCESS in 1h 29m 39s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-rocky/08293dc/ : SUCCESS in 1h 15m 34s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-queens/53bb640/ : SUCCESS in 1h 32m 22s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-centos-7/110844e/ : FAILURE in 1h 58m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/1ee025d/ : SUCCESS in 1h 24m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-two-node/1482fad/ : FAILURE in 40m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-two-node/a0efe9b/ : SUCCESS in 2h 01m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican/a674d43/ : SUCCESS in 57m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-stein/36ea6b8/ : SUCCESS in 49m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-rocky/d52ea42/ : SUCCESS in 55m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-queens/71c6c82/ : SUCCESS in 54m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool/3215b7a/ : SUCCESS in 1h 10m 29s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool/f6947d3/ : SUCCESS in 1h 00m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool-stable-stein/85d121f/ : SUCCESS in 1h 11m 00s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/170828f/ : SUCCESS in 57m 34s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/dfdb388/ : SUCCESS in 49m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0dd09c30dad77f525a8dbc0e5e0a3f4869b4dca5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-18 17:45:35.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"89efcfdb884dc8fb53be1fae75d1414981afc81a","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-13 16:24:17.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"46a8ccd1e14113fc2fd0a016751f21606b016c0e","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-07-14 02:34:19.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"5c4c0f8ea304ffe9353e8d588c12e30de290f47a","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-07-17 03:44:52.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"bd919aae4783719e7571be16888b32de66049e37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 03:45:06.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"64aa1b5ca34a9c5cc8a6e9f8b001a8c335c968b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 05:49:30.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/gate/openstack-tox-docs/79bebf1/html/ : SUCCESS in 3m 03s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/gate/openstack-tox-pep8/c823873/ : SUCCESS in 2m 43s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-api/cbf3005/ : SUCCESS in 46m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api/5af6c1f/ : SUCCESS in 52m 17s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-api-stable-stein/4a27472/ : SUCCESS in 45m 20s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/40cea5d/ : SUCCESS in 47m 08s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/cd4bda9/ : SUCCESS in 1h 22m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-scenario/1213736/ : FAILURE in 1h 54m 43s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario/c5aca5a/ : SUCCESS in 1h 50m 09s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-scenario-stable-stein/4f1d18b/ : FAILURE in 2h 02m 37s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/eca77a2/ : SUCCESS in 1h 23m 48s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario-stable-queens/14a9fb5/ : FAILURE in 1h 56m 14s","accounts_in_message":[],"_revision_number":12},{"id":"81b787e156c68f21339c247e812bffbd05674f5a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-17 10:18:50.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"1d6c72c142f264821ac9138248690462c5d14254","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 12:44:46.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/check/openstack-tox-docs/dafae8f/html/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/check/openstack-tox-pep8/b6dd2f4/ : SUCCESS in 3m 35s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api/1f2bba4/ : SUCCESS in 43m 57s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api/8350a9c/ : SUCCESS in 51m 47s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api-stable-stein/15223f8/ : SUCCESS in 49m 30s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/7fe68a9/ : SUCCESS in 42m 17s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-queens/0afd5ce/ : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario/baf1117/ : SUCCESS in 1h 36m 14s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario/c5dd9c6/ : SUCCESS in 1h 17m 39s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-stable-stein/5a19629/ : FAILURE in 1h 37m 01s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-rocky/b42a392/ : SUCCESS in 1h 31m 12s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-queens/683656d/ : SUCCESS in 1h 38m 03s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-centos-7/e5d6d56/ : FAILURE in 2h 18m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/561a8cd/ : SUCCESS in 1h 49m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-two-node/3a47492/ : SUCCESS in 1h 21m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-two-node/e94b4da/ : SUCCESS in 1h 50m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario/a79c1b8/ : SUCCESS in 1h 03m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/e6233ce/ : SUCCESS in 1h 00m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/9161f2e/ : FAILURE in 1h 04m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/22a0cd2/ : SUCCESS in 58m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/a0015c4/ : FAILURE in 1h 26m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/5f03894/ : SUCCESS in 1h 14m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican/c359669/ : SUCCESS in 1h 04m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-stein/bcb97c1/ : SUCCESS in 1h 10m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-rocky/a9756c1/ : SUCCESS in 1h 10m 49s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-queens/8fae772/ : SUCCESS in 57m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool/a122dce/ : SUCCESS in 1h 06m 33s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool/1e20128/ : FAILURE in 56m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool-stable-stein/f3d5c0e/ : SUCCESS in 55m 54s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/94a60cc/ : SUCCESS in 58m 57s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/18c9ef9/ : SUCCESS in 47m 42s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"39b6397174d8bb04e5396f9c6bc974760f2f9129","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-17 16:28:15.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"ed43684bd5f851e596c8238c0fea8a3687072edd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:46:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/check/openstack-tox-docs/57fdcb0/html/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/check/openstack-tox-pep8/5e38873/ : SUCCESS in 4m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api/88ed0b3/ : SUCCESS in 53m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api/4856ac9/ : SUCCESS in 53m 46s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api-stable-stein/58fd8d1/ : SUCCESS in 50m 18s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/05d1ed6/ : SUCCESS in 46m 53s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-queens/74e93dc/ : SUCCESS in 1h 12m 13s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario/c549599/ : SUCCESS in 1h 28m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario/0f4a893/ : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-stable-stein/98809ed/ : SUCCESS in 1h 31m 06s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-rocky/b2d2029/ : SUCCESS in 1h 39m 33s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-queens/41d23f3/ : SUCCESS in 2h 00m 25s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-centos-7/7d97b73/ : FAILURE in 2h 11m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/0565849/ : FAILURE in 1h 49m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-two-node/92774d1/ : SUCCESS in 1h 28m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-two-node/24248c9/ : FAILURE in 1h 40m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario/744afbb/ : FAILURE in 1h 21m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/3625949/ : SUCCESS in 1h 11m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/50236cc/ : FAILURE in 1h 10m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/d9a137f/ : SUCCESS in 1h 03m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/83cf2f6/ : SUCCESS in 1h 06m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/face065/ : FAILURE in 1h 00m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican/9b24863/ : SUCCESS in 57m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-stein/057504e/ : SUCCESS in 54m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-rocky/9d6d1ff/ : SUCCESS in 59m 14s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-queens/43bb177/ : SUCCESS in 56m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool/0a89ef8/ : SUCCESS in 52m 33s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool/3075a49/ : SUCCESS in 1h 01m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool-stable-stein/c410ea8/ : SUCCESS in 1h 16m 16s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/9ca29b4/ : SUCCESS in 1h 07m 51s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/0a36a9b/ : SUCCESS in 1h 03m 23s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8502ec2cb52c6bfad55519e2cdc638913f6f493f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:46:41.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"950e9da628d80e38f016b5e6d34b52546b7d59ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 20:45:55.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/gate/openstack-tox-docs/3a6ad75/html/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/gate/openstack-tox-pep8/c78dcc2/ : SUCCESS in 3m 46s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-api/bae4637/ : SUCCESS in 53m 20s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api/41e050a/ : SUCCESS in 50m 59s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-api-stable-stein/a3bc407/ : SUCCESS in 45m 42s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/ab6ecbf/ : SUCCESS in 51m 33s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/f30beab/ : SUCCESS in 43m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-scenario/51de8aa/ : SUCCESS in 1h 32m 44s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario/4fbee75/ : SUCCESS in 1h 38m 52s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-scenario-stable-stein/e6bdd5c/ : SUCCESS in 1h 30m 16s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/086c431/ : SUCCESS in 1h 53m 34s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario-stable-queens/0d90595/ : FAILURE in 1h 38m 17s","accounts_in_message":[],"_revision_number":12},{"id":"31a5e1b64fe2233d233e3f1165486f4b1ae7bc11","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-19 17:45:23.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"9283f92ccace77b30ceab7e15453ef0ed793b452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 20:15:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/check/openstack-tox-docs/3cffdd0/html/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/check/openstack-tox-pep8/dc5c656/ : SUCCESS in 3m 30s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api/8eb1e2a/ : SUCCESS in 41m 44s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api/acc90f9/ : SUCCESS in 43m 18s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-api-stable-stein/f00dff7/ : SUCCESS in 45m 42s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/e324a64/ : SUCCESS in 52m 19s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-noop-py2-api-stable-queens/1c73fa6/ : SUCCESS in 53m 17s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario/79a8d41/ : SUCCESS in 1h 39m 20s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario/35fe814/ : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-stable-stein/872d5bd/ : SUCCESS in 1h 38m 56s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-rocky/33b92b1/ : SUCCESS in 1h 28m 12s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-stable-queens/f234722/ : SUCCESS in 1h 47m 09s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-centos-7/cf36ceb/ : SUCCESS in 2h 22m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9eeed47/ : SUCCESS in 1h 37m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-scenario-two-node/eb64bc4/ : FAILURE in 1h 31m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-scenario-two-node/905aba4/ : SUCCESS in 1h 52m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario/e828f87/ : SUCCESS in 56m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/48aeffe/ : SUCCESS in 58m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/0a04317/ : SUCCESS in 1h 24m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/8ea7eae/ : SUCCESS in 1h 01m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/efc85cb/ : SUCCESS in 56m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/0d64e85/ : SUCCESS in 1h 07m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican/167d641/ : SUCCESS in 50m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-stein/b2bbc68/ : SUCCESS in 56m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-rocky/66c153f/ : SUCCESS in 1h 00m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-tls-barbican-stable-queens/a6f4664/ : SUCCESS in 1h 02m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool/8c2ab99/ : SUCCESS in 1h 05m 40s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool/bdcc426/ : SUCCESS in 1h 07m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-spare-pool-stable-stein/786c618/ : SUCCESS in 1h 09m 41s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/791c5d3/ : SUCCESS in 1h 04m 36s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/88/634988/12/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/9eda218/ : SUCCESS in 1h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2bab634e9231ecf0939ef183e3c2c5daf031b632","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 20:15:26.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"640839fa4203fbfdbcef99a07e853a6fc4f37a01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 22:12:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"a4c7bbbba93b7fba5582b565c60213f9eb357ac0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 22:12:06.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/634988/12/gate/openstack-tox-docs/bdaa881/html/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/88/634988/12/gate/openstack-tox-pep8/475f7e8/ : SUCCESS in 3m 31s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-api/b6e5dcb/ : SUCCESS in 50m 26s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api/7b187f8/ : SUCCESS in 59m 18s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-api-stable-stein/a954e32/ : SUCCESS in 43m 04s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/1de88a8/ : SUCCESS in 46m 58s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/72efa78/ : SUCCESS in 46m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-scenario/647960e/ : SUCCESS in 1h 50m 25s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario/2c1e4ec/ : SUCCESS in 1h 36m 56s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-scenario-stable-stein/3b64c6b/ : SUCCESS in 1h 36m 43s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/0fa3370/ : SUCCESS in 1h 27m 25s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/634988/12/gate/octavia-v2-dsvm-py2-scenario-stable-queens/1869f42/ : SUCCESS in 1h 38m 54s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"5b963c4c219131b45992f906c4d2919c035d8984","revisions":{"77afa54c29c98f01aea70f29dd42cd958f0c43e8":{"kind":"REWORK","_number":1,"created":"2019-02-05 17:18:30.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/1"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-05 17:17:24.000000000","tz":60},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/77afa54c29c98f01aea70f29dd42cd958f0c43e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/77afa54c29c98f01aea70f29dd42cd958f0c43e8"}]},"branch":"refs/heads/master"},"bf6586519a66dc97b0703c65d949e0a75d64b818":{"kind":"REWORK","_number":2,"created":"2019-02-05 17:20:44.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/2"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-05 17:20:29.000000000","tz":60},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf6586519a66dc97b0703c65d949e0a75d64b818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf6586519a66dc97b0703c65d949e0a75d64b818"}]},"branch":"refs/heads/master"},"7d8aea7d8ef7bf025c569af7bca2fa84f7aa206f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-02-05 21:36:35.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/3"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-05 21:36:05.000000000","tz":60},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nDepends-On: https://review.openstack.org/#/c/632594/\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7d8aea7d8ef7bf025c569af7bca2fa84f7aa206f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7d8aea7d8ef7bf025c569af7bca2fa84f7aa206f"}]},"branch":"refs/heads/master"},"c99e7bf22bab59fb79817d35d249a9450d3871f8":{"kind":"REWORK","_number":4,"created":"2019-02-05 21:55:19.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/4"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-05 21:55:10.000000000","tz":60},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nDepends-On: https://review.openstack.org/#/c/632594/\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c99e7bf22bab59fb79817d35d249a9450d3871f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c99e7bf22bab59fb79817d35d249a9450d3871f8"}]},"branch":"refs/heads/master"},"ec857f12d462b4829618b3542feaf9263ca72e45":{"kind":"REWORK","_number":5,"created":"2019-03-18 09:42:05.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/5"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-03-18 09:41:40.000000000","tz":60},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ec857f12d462b4829618b3542feaf9263ca72e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ec857f12d462b4829618b3542feaf9263ca72e45"}]},"branch":"refs/heads/master"},"538fa9eea06c9ff5feed5eca7a60a85ed925cb47":{"kind":"REWORK","_number":6,"created":"2019-04-02 17:22:20.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/6"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-02 17:22:13.000000000","tz":0},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nDepends-On: https://review.openstack.org/#/c/649381/\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/538fa9eea06c9ff5feed5eca7a60a85ed925cb47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/538fa9eea06c9ff5feed5eca7a60a85ed925cb47"}]},"branch":"refs/heads/master"},"98481e5a53fdbf70ef97632bf2996f3fe0e0baee":{"kind":"REWORK","_number":7,"created":"2019-04-05 21:04:30.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/7"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-05 21:03:13.000000000","tz":120},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nThis patch adds a scenario test for testing amphora spare pool feature.\nIt adds new check jobs to test against master as well as stable\nbranches.\n\nDepends-On: https://review.openstack.org/#/c/649381/\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/98481e5a53fdbf70ef97632bf2996f3fe0e0baee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/98481e5a53fdbf70ef97632bf2996f3fe0e0baee"}]},"branch":"refs/heads/master"},"de5a3b9479bd0de75c1e2e0994993bca89ef0d2a":{"kind":"REWORK","_number":8,"created":"2019-04-05 21:05:47.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/8"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-05 21:05:38.000000000","tz":120},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nThis patch adds a scenario test for testing amphora spare pool feature.\nIt adds new check jobs to test against master as well as stable\nbranches.\n\nDepends-On: https://review.openstack.org/#/c/649381/\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/de5a3b9479bd0de75c1e2e0994993bca89ef0d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/de5a3b9479bd0de75c1e2e0994993bca89ef0d2a"}]},"branch":"refs/heads/master"},"03fe897e47f7fceab13e21cf5ab74f1ed96aa8a0":{"kind":"REWORK","_number":9,"created":"2019-04-05 21:20:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/9"}}},"commit":{"parents":[{"commit":"ebbe4ef9698d0873337d2176bbaf0df32c846ca7","subject":"Merge \"Adds provider API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebbe4ef9698d0873337d2176bbaf0df32c846ca7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-05 21:20:34.000000000","tz":120},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nThis patch adds a scenario test for testing amphora spare pool feature.\nIt adds new check jobs to test against master as well as stable\nbranches.\n\nDepends-On: https://review.openstack.org/#/c/649381/\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/03fe897e47f7fceab13e21cf5ab74f1ed96aa8a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/03fe897e47f7fceab13e21cf5ab74f1ed96aa8a0"}]},"branch":"refs/heads/master"},"69b92b82daad1b463067479f55ea40a1e6dbea73":{"kind":"REWORK","_number":10,"created":"2019-05-08 09:32:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/10"}}},"commit":{"parents":[{"commit":"52edc5ee1145a29bd111e3531fa017149bae65b2","subject":"Merge \"Drop requirement on octavia-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/52edc5ee1145a29bd111e3531fa017149bae65b2"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-08 09:31:52.000000000","tz":120},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nThis patch adds a scenario test for testing amphora spare pool feature.\nIt adds new check jobs to test against master as well as stable\nbranches.\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/69b92b82daad1b463067479f55ea40a1e6dbea73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/69b92b82daad1b463067479f55ea40a1e6dbea73"}]},"branch":"refs/heads/master"},"508d3d5e3b64e30349a3ca62fa3f82e1fae36bdd":{"kind":"REWORK","_number":11,"created":"2019-05-14 15:31:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/11"}}},"commit":{"parents":[{"commit":"3c0bde3d192bcbda1598fa2b7f3f7b945266c060","subject":"Merge \"Add TLS Barbican jobs against stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3c0bde3d192bcbda1598fa2b7f3f7b945266c060"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-14 15:31:36.000000000","tz":120},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nThis patch adds a scenario test for testing amphora spare pool feature.\nIt adds new check jobs to test against master as well as stable\nbranches.\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/508d3d5e3b64e30349a3ca62fa3f82e1fae36bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/508d3d5e3b64e30349a3ca62fa3f82e1fae36bdd"}]},"branch":"refs/heads/master"},"5b963c4c219131b45992f906c4d2919c035d8984":{"kind":"REWORK","_number":12,"created":"2019-05-14 16:20:37.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/634988/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/634988/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/634988/12"}}},"commit":{"parents":[{"commit":"3c0bde3d192bcbda1598fa2b7f3f7b945266c060","subject":"Merge \"Add TLS Barbican jobs against stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3c0bde3d192bcbda1598fa2b7f3f7b945266c060"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-03 14:37:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-14 16:20:29.000000000","tz":120},"subject":"Add a scenario test for spare pool","message":"Add a scenario test for spare pool\n\nThis patch adds a scenario test for testing amphora spare pool feature.\nIt adds new check jobs to test against master as well as stable\nbranches.\n\nChange-Id: I87249017453628e3d2cd320a02677d81397f1b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5b963c4c219131b45992f906c4d2919c035d8984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5b963c4c219131b45992f906c4d2919c035d8984"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
