)]}'
{"id":"openstack%2Foctavia-tempest-plugin~666601","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ie61f42f0d5565455a766b321055f0b854820a865","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"act-stdby","hashtags":[],"change_id":"Ie61f42f0d5565455a766b321055f0b854820a865","subject":"Rename iptables-based active-standby jobs","status":"MERGED","created":"2019-06-20 14:55:26.000000000","updated":"2019-07-17 07:22:29.000000000","submitted":"2019-07-17 07:22:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666601-1563348149357-0ae53db0","meta_rev_id":"da6209c6a61213e1df3511d2c9a96999762f0bdd","_number":666601,"virtual_id_number":666601,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-21 13:04:12.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-17 07:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"recommended":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-07-14 02:34:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-07-17 03:47:09.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-07-13 16:24:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-06-21 13:01:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-07-16 04:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-07-17 03:47:09.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-21 13:01:53.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-07-13 16:24:09.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-14 02:34:38.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-07-16 04:38:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-17 03:47:09.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-07-17 07:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"734858980d9fbd357dcbff5d66539d63aa30a9b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-20 14:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a877f85ba8a3e1bc19d1578903bdfd0b1685803b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 17:40:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/01/666601/1/check/openstack-tox-docs/6ec3afa/html/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/01/666601/1/check/openstack-tox-pep8/90b22f5/ : SUCCESS in 9m 31s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-api/69aad7f/ : SUCCESS in 1h 38m 41s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api/f3769fc/ : SUCCESS in 49m 36s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-api-stable-stein/1a054bb/ : SUCCESS in 52m 13s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/4042b52/ : SUCCESS in 52m 55s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/c609211/ : SUCCESS in 50m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario/8a7184b/ : SUCCESS in 1h 33m 59s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario/f5b05b4/ : SUCCESS in 1h 22m 00s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-stable-stein/bac342b/ : SUCCESS in 1h 29m 56s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/1b5b06d/ : TIMED_OUT in 2h 34m 11s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/4e17876/ : TIMED_OUT in 2h 33m 58s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-centos-7/0b8e8f4/ : POST_FAILURE in 2h 04m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b512512/ : FAILURE in 2h 06m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-two-node/e0069bd/ : SUCCESS in 1h 53m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-two-node/a402702/ : SUCCESS in 1h 51m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/599857a/ : SUCCESS in 1h 04m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/56d5bd7/ : SUCCESS in 1h 01m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/2a9b959/ : FAILURE in 1h 04m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/69447c7/ : FAILURE in 1h 21m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/6d51d9a/ : FAILURE in 1h 33m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/74bb876/ : SUCCESS in 59m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican/2efd508/ : SUCCESS in 57m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/aab949c/ : SUCCESS in 58m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/c96b75c/ : SUCCESS in 1h 04m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/6a91c0c/ : SUCCESS in 2h 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54a08602c552c72f8ece1de91011317843aa45c9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-20 18:23:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b42f2030e0f62317d786929506018d06f9526a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:09:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/01/666601/1/check/openstack-tox-docs/e16da50/html/ : SUCCESS in 10m 11s\n- openstack-tox-pep8 http://logs.openstack.org/01/666601/1/check/openstack-tox-pep8/d005a9b/ : SUCCESS in 9m 43s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-api/4ab81dd/ : SUCCESS in 45m 30s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api/46bf6a1/ : SUCCESS in 46m 34s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-api-stable-stein/9b063b6/ : SUCCESS in 46m 05s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/77a9a1c/ : SUCCESS in 44m 33s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/8353451/ : SUCCESS in 42m 03s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario/6a1854b/ : SUCCESS in 1h 36m 18s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario/0afb12c/ : TIMED_OUT in 2h 33m 11s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-stable-stein/08d28a9/ : SUCCESS in 1h 28m 49s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/64ce47b/ : SUCCESS in 1h 27m 53s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/840e269/ : SUCCESS in 1h 31m 10s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-centos-7/ff65290/ : FAILURE in 2h 11m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/11df4cd/ : SUCCESS in 1h 49m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-two-node/0a94aa1/ : FAILURE in 1h 22m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-two-node/7cf4a83/ : SUCCESS in 2h 00m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/bc1182f/ : SUCCESS in 1h 06m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/c3c64dd/ : FAILURE in 1h 32m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/5f7128a/ : FAILURE in 1h 15m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/dcc98de/ : SUCCESS in 1h 05m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/f9e6cfa/ : FAILURE in 2h 31m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/ca084b5/ : SUCCESS in 2h 18m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican/2138579/ : SUCCESS in 55m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/e8cfc88/ : TIMED_OUT in 2h 33m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/278bb9d/ : SUCCESS in 2h 18m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/be4fe3a/ : SUCCESS in 2h 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bebf33753c0eeab441146d75831b3d96745f81ee","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-20 22:24:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d5608f90d6efbfdeff66f3b8bd537230926bf5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 01:06:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/01/666601/1/check/openstack-tox-docs/30796be/html/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/01/666601/1/check/openstack-tox-pep8/1cb651a/ : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-api/cd6dbd4/ : SUCCESS in 1h 36m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api/09538ba/ : SUCCESS in 44m 28s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-api-stable-stein/f9bc597/ : SUCCESS in 41m 16s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/07ed266/ : SUCCESS in 49m 13s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/493d278/ : SUCCESS in 1h 07m 24s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario/d4ded3b/ : TIMED_OUT in 2h 33m 02s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario/81f6894/ : TIMED_OUT in 2h 33m 27s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-stable-stein/e042674/ : TIMED_OUT in 2h 33m 13s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/94ea4c8/ : TIMED_OUT in 2h 34m 12s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/4fe0bb1/ : SUCCESS in 1h 40m 31s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-centos-7/e767b62/ : FAILURE in 1h 56m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3c12ddf/ : SUCCESS in 1h 38m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-two-node/3bc47a2/ : SUCCESS in 1h 54m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-two-node/d1a44b9/ : SUCCESS in 1h 43m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/d822f3e/ : SUCCESS in 1h 01m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/a64f962/ : SUCCESS in 1h 08m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/7a11b61/ : FAILURE in 1h 14m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/7ab1b61/ : FAILURE in 1h 45m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/0c4f5a0/ : SUCCESS in 1h 05m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/90c9f9c/ : SUCCESS in 1h 01m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican/493466b/ : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/b8e2da7/ : SUCCESS in 50m 14s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/33049a0/ : SUCCESS in 1h 04m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/b058b0f/ : SUCCESS in 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41cb3dc204889c5408e36d7fa27ac32a13d4f534","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-06-21 13:01:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"973100eff1673e2c866198b184d33c1b597f42ef","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-21 13:04:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa6800237df7a5f46f1c9395032b04dea115ca09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 15:47:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/01/666601/1/check/openstack-tox-docs/a3f99b4/html/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/01/666601/1/check/openstack-tox-pep8/731e90d/ : SUCCESS in 10m 01s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-api/9d63a36/ : SUCCESS in 40m 42s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api/fdc8d17/ : SUCCESS in 47m 17s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-api-stable-stein/fef4033/ : SUCCESS in 42m 33s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/f5ebf15/ : SUCCESS in 42m 56s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/e67bf6b/ : SUCCESS in 42m 47s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario/7dd7de3/ : SUCCESS in 1h 30m 43s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario/b8f5361/ : SUCCESS in 1h 23m 32s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-stable-stein/dc6ee57/ : SUCCESS in 1h 43m 01s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/59f85d3/ : SUCCESS in 1h 34m 25s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/2e7e045/ : SUCCESS in 1h 23m 52s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-centos-7/421f5f6/ : FAILURE in 1h 52m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f311442/ : FAILURE in 1h 41m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-scenario-two-node/7d5bc0a/ : TIMED_OUT in 2h 33m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-py2-scenario-two-node/b3ac8ea/ : SUCCESS in 1h 56m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/594f076/ : SUCCESS in 1h 10m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/6620ecb/ : SUCCESS in 1h 05m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/6f81c60/ : FAILURE in 59m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/3e877e8/ : SUCCESS in 1h 05m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/0722016/ : SUCCESS in 1h 02m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/f9af9a0/ : SUCCESS in 1h 06m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican/e11adc1/ : SUCCESS in 59m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/1206778/ : SUCCESS in 1h 47m 26s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/2e8a531/ : SUCCESS in 56m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/01/666601/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/7ad5769/ : SUCCESS in 1h 00m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93a38e31993244cb807ea3d08222426493c91e00","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-13 16:24:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a13bfcc50c077304b6a79476b5d5e1495d416faf","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-07-14 02:34:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"652ffef8ef7cffdbe512fb79aca00231c2caa13c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-16 04:38:27.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"935a38cec4737842e123d011ac9c41d1f03acc63","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-07-17 03:47:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"177d48b152b76a0663bd70933f2688496a0dc77d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 03:47:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5b508aed11395d503a605c502a6c63cdd68e028","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 07:22:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c12aa9603ab9a7403a1fa082c8b3497aab8a49ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 07:22:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/01/666601/1/gate/openstack-tox-docs/2d8ffad/html/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/01/666601/1/gate/openstack-tox-pep8/200f95b/ : SUCCESS in 3m 11s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-noop-api/b4a4965/ : SUCCESS in 41m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-noop-py2-api/b324127/ : SUCCESS in 49m 07s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-noop-api-stable-stein/a40399f/ : SUCCESS in 44m 14s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/8994994/ : SUCCESS in 43m 30s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/87e4152/ : SUCCESS in 40m 10s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-scenario/4760216/ : SUCCESS in 1h 38m 33s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-py2-scenario/e6f65df/ : SUCCESS in 1h 27m 35s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-scenario-stable-stein/2c8ebeb/ : SUCCESS in 1h 28m 19s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/5272e42/ : SUCCESS in 1h 29m 37s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/01/666601/1/gate/octavia-v2-dsvm-py2-scenario-stable-queens/daea5a6/ : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"531b129ba205dd382a6b0aa65c1097dc1c888b3f","revisions":{"531b129ba205dd382a6b0aa65c1097dc1c888b3f":{"kind":"REWORK","_number":1,"created":"2019-06-20 14:55:26.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/01/666601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/01/666601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/01/666601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/01/666601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/01/666601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/01/666601/1"}}},"commit":{"parents":[{"commit":"2261d112b2c5a9c25e3eceb701fe131a9782ddf5","subject":"Merge \"Save amphora logs in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2261d112b2c5a9c25e3eceb701fe131a9782ddf5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-20 14:53:23.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-20 14:53:23.000000000","tz":120},"subject":"Rename iptables-based active-standby jobs","message":"Rename iptables-based active-standby jobs\n\nRename jobs so that it does not conflict with amphora stats-based\nactive-standby jobs being added in\nIbcd5552a67cea650edc72bfaa986357267ca2407 and will become default\nactive-standby jobs from Rocky.\n\nChange-Id: Ie61f42f0d5565455a766b321055f0b854820a865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/531b129ba205dd382a6b0aa65c1097dc1c888b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/531b129ba205dd382a6b0aa65c1097dc1c888b3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
