)]}'
{"id":"openstack%2Foctavia-tempest-plugin~666988","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I9eba013fb0b87afc44187e259b594f9b93bcc5d9","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I9eba013fb0b87afc44187e259b594f9b93bcc5d9","subject":"Set irrelevant files on CI jobs","status":"MERGED","created":"2019-06-23 13:27:10.000000000","updated":"2019-08-03 01:59:04.000000000","submitted":"2019-08-03 01:59:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666988-1564797544263-d227193f","meta_rev_id":"fbae5b52aeb6a381ac7edb83d95c591c01865c29","_number":666988,"virtual_id_number":666988,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-07-26 12:08:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-03 01:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-03 00:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-01 16:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-08-03 00:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 16:36:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-03 00:19:59.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-03 01:59:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"684bc424ec9e72cca3bc9ba34c31b46c52a69e8c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-23 13:27:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e67cb1d2eef194df0cdceef31542235e0a8f15c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 13:27:23.000000000","message":"Patch Set 1: Verified-1\n\n(15 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error\nwas:\n\n  expected str @ data[\u0027irrelevant-files\u0027][0]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: octavia-v2-dsvm-scenario\n      parent: octavia-dsvm-live-base\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n                api_v1_enabled: False\n        tempest_concurrency: 2\n        tempest_test_regex: ^octavia_tempest_plugin.tests.scenario.v2\n        tox_envlist: all\n      irrelevant-files:\n        - *common-irrelevant-files\n        - ^octavia_tempest_plugin/tests/(?!scenario|.*\\.py)\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 319, column 3","accounts_in_message":[],"_revision_number":1},{"id":"883cdecfe2c096f6b2ccb97d9f2056e1ec5368e1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-23 14:39:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bec78207142fa4d830fda84832d853dc794ad8eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 16:51:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/666988/2/check/openstack-tox-docs/fe19b29/html/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/88/666988/2/check/openstack-tox-pep8/0654b99/ : SUCCESS in 4m 04s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-api/ac9bc6b/ : SUCCESS in 42m 32s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-py2-api/65fcdbe/ : SUCCESS in 47m 34s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-api-stable-stein/017faed/ : SUCCESS in 43m 11s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/1be3cd3/ : SUCCESS in 45m 47s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-py2-api-stable-queens/fb6b60f/ : SUCCESS in 45m 31s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario/378702e/ : SUCCESS in 1h 54m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-py2-scenario/500d7be/ : FAILURE in 1h 48m 42s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario-stable-stein/724a30b/ : SUCCESS in 1h 22m 02s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-py2-scenario-stable-rocky/ec129c7/ : SUCCESS in 1h 22m 30s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-py2-scenario-stable-queens/e3e0bc8/ : SUCCESS in 1h 16m 45s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario-centos-7/1279453/ : FAILURE in 2h 08m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/923afd1/ : SUCCESS in 1h 29m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario-two-node/dfdfd62/ : SUCCESS in 1h 46m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-py2-scenario-two-node/9890414/ : SUCCESS in 1h 51m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-scenario/8d63b67/ : FAILURE in 1h 20m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-py2-scenario/4c70f5c/ : SUCCESS in 1h 00m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-py2-scenario-centos-7/22a9344/ : SUCCESS in 1h 11m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-scenario-stable-stein/a7427a7/ : FAILURE in 1h 52m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky/3bb2123/ : FAILURE in 1h 19m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens/0df6de3/ : SUCCESS in 55m 56s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-tls-barbican/b25530b/ : SUCCESS in 57m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-tls-barbican-stable-stein/c8132f8/ : SUCCESS in 51m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-tls-barbican-stable-rocky/80dfd53/ : SUCCESS in 55m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-tls-barbican-stable-queens/fdede19/ : SUCCESS in 51m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e74361c14ca346ec6d06c0baeed823eba0d9d2c1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-23 20:10:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cac3a33c4cc54a3430feca6d5ab539e072a27f4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 22:10:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/666988/2/check/openstack-tox-docs/7d1c661/html/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/88/666988/2/check/openstack-tox-pep8/f98011f/ : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-api/d903736/ : SUCCESS in 40m 25s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-py2-api/0762d39/ : SUCCESS in 48m 42s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-api-stable-stein/c566e2e/ : SUCCESS in 42m 46s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/dd68df2/ : SUCCESS in 43m 27s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-noop-py2-api-stable-queens/1f86957/ : SUCCESS in 44m 39s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario/70fc7e7/ : SUCCESS in 1h 40m 02s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-py2-scenario/86e4307/ : SUCCESS in 1h 41m 25s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario-stable-stein/74f27cc/ : SUCCESS in 1h 42m 50s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-py2-scenario-stable-rocky/1836aa6/ : SUCCESS in 1h 32m 05s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-py2-scenario-stable-queens/16d2450/ : SUCCESS in 1h 23m 38s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario-centos-7/d408536/ : FAILURE in 1h 54m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/973d10a/ : SUCCESS in 1h 51m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-scenario-two-node/82400b4/ : SUCCESS in 1h 47m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-py2-scenario-two-node/8db3441/ : SUCCESS in 1h 13m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-scenario/b5555eb/ : SUCCESS in 1h 01m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-py2-scenario/1b37747/ : SUCCESS in 1h 00m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-py2-scenario-centos-7/064f71c/ : FAILURE in 1h 04m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-scenario-stable-stein/0b62ed9/ : SUCCESS in 1h 00m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky/8d24934/ : SUCCESS in 58m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/666988/2/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens/c707eb8/ : SUCCESS in 58m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-tls-barbican/71e87fe/ : SUCCESS in 52m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-tls-barbican-stable-stein/baa0ad2/ : SUCCESS in 50m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-tls-barbican-stable-rocky/c9419aa/ : SUCCESS in 52m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/666988/2/check/octavia-v2-dsvm-tls-barbican-stable-queens/714a391/ : SUCCESS in 48m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b040851a6ed83f21e5bcbbba81364514e81557ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-26 16:28:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4fd1ac0a1ae009829b9db72060da2e152d8dcb9d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-26 16:30:59.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"866ed7ef80db6b4d14022ffbfcd11bc020437065","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-26 16:36:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7fc9db75334d2b6245a4a6f5f974a690461d3a70","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-01 18:24:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8dba0dacd8f3776f7b017b509117131a6ce443f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 20:45:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/666988/3/check/openstack-tox-docs/8507550/html/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 http://logs.openstack.org/88/666988/3/check/openstack-tox-pep8/e396105/ : SUCCESS in 3m 03s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-noop-api/d74247b/ : SUCCESS in 49m 30s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-noop-py2-api/7b90b33/ : SUCCESS in 46m 23s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-noop-api-stable-stein/0779c77/ : SUCCESS in 46m 52s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/53ea570/ : SUCCESS in 48m 05s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-noop-py2-api-stable-queens/61511ea/ : SUCCESS in 43m 15s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-scenario/9d76075/ : SUCCESS in 1h 28m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-py2-scenario/41150ff/ : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-scenario-stable-stein/2a70310/ : SUCCESS in 1h 34m 21s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-py2-scenario-stable-rocky/8f569ad/ : SUCCESS in 1h 39m 47s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-py2-scenario-stable-queens/d949d5c/ : SUCCESS in 1h 51m 19s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-scenario-centos-7/a7ed6d8/ : FAILURE in 2h 09m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3cb789f/ : SUCCESS in 1h 48m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-scenario-two-node/3d2aa83/ : SUCCESS in 1h 50m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-py2-scenario-two-node/e9de74a/ : SUCCESS in 1h 35m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario http://logs.openstack.org/88/666988/3/check/octavia-v2-act-stdby-dsvm-scenario/4b09f25/ : SUCCESS in 56m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario http://logs.openstack.org/88/666988/3/check/octavia-v2-act-stdby-dsvm-py2-scenario/54dfa4c/ : SUCCESS in 1h 00m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-centos-7 http://logs.openstack.org/88/666988/3/check/octavia-v2-act-stdby-dsvm-py2-scenario-centos-7/3f3f5bc/ : FAILURE in 1h 10m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein http://logs.openstack.org/88/666988/3/check/octavia-v2-act-stdby-dsvm-scenario-stable-stein/8a2822e/ : SUCCESS in 52m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/88/666988/3/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-rocky/a54a7b4/ : SUCCESS in 58m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens http://logs.openstack.org/88/666988/3/check/octavia-v2-act-stdby-dsvm-py2-scenario-stable-queens/11bb8f1/ : SUCCESS in 1h 12m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-tls-barbican/06111fe/ : SUCCESS in 59m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-tls-barbican-stable-stein/b59d74a/ : SUCCESS in 53m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-tls-barbican-stable-rocky/0861085/ : SUCCESS in 1h 04m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/88/666988/3/check/octavia-v2-dsvm-tls-barbican-stable-queens/ffcf78a/ : SUCCESS in 56m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52cfe4ac0e144ef8ecaff0ca2d8da237b720a9da","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-22 20:27:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThese changes are too specific and will miss changes that may impact the tests.","accounts_in_message":[],"_revision_number":3},{"id":"d93ba79e5813eba94aee71599550c5630fa9dc53","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-26 12:08:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"41c274c77e72f05f0e59abe9c96c339965105e83","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-01 16:27:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9aefb0eeff29df62529d828bdc4a421fbae9cf69","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-01 16:36:16.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nOk, I see now. LGTM","accounts_in_message":[],"_revision_number":3},{"id":"60a61d95e154facb8635302fc1ee5989e5967ef7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:19:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"04aa9c127370976a66e53c7e080ef3e1574439a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 00:20:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"15507b67ea260f9d5c6e3f3e4fc908c7e7bc80b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 01:59:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/88/666988/3/gate/openstack-tox-docs/7cf7b3e/html/ : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://logs.opendev.org/88/666988/3/gate/openstack-tox-pep8/ae77ea3/ : SUCCESS in 2m 31s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-noop-api/eca7821/ : SUCCESS in 37m 31s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-noop-py2-api/56a673a/ : SUCCESS in 40m 39s\n- octavia-v2-dsvm-noop-api-stable-stein https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-noop-api-stable-stein/c83259f/ : SUCCESS in 37m 14s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/fb1bcf4/ : SUCCESS in 49m 54s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/903c9d9/ : SUCCESS in 52m 42s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-scenario/aca1d88/ : SUCCESS in 1h 37m 55s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-py2-scenario/b8b61d1/ : SUCCESS in 1h 36m 42s\n- octavia-v2-dsvm-scenario-stable-stein https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-scenario-stable-stein/dd662da/ : SUCCESS in 1h 18m 17s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/dd569a5/ : SUCCESS in 1h 20m 16s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://logs.opendev.org/88/666988/3/gate/octavia-v2-dsvm-py2-scenario-stable-queens/84fc587/ : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":3},{"id":"fbae5b52aeb6a381ac7edb83d95c591c01865c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 01:59:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"531c5062c4cff602e455522c90fd9ce546f5ade1","revisions":{"879971c8f10bcdb1a04c2aac933e2832cc31057f":{"kind":"REWORK","_number":1,"created":"2019-06-23 13:27:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/666988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/666988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/1"}}},"commit":{"parents":[{"commit":"2261d112b2c5a9c25e3eceb701fe131a9782ddf5","subject":"Merge \"Save amphora logs in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2261d112b2c5a9c25e3eceb701fe131a9782ddf5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-23 13:18:06.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-23 13:18:06.000000000","tz":120},"subject":"Set irrelevant files on CI jobs","message":"Set irrelevant files on CI jobs\n\nAll jobs are triggered regardless of files changed, except for files\nalready defined in octavia-dsvm-base. This means many jobs that not\nimpacted by a given change will be run in CI. Not only this is a waste\nof resources but the job may also fail (e.g. timeouts) and block changes\nfrom merging.\n\nThis patch defines a more granular set of irrelevant files per job. For\nexample, it prevents active-standby jobs from running when a change is\nlimited to normal scenario files.\n\nChange-Id: I9eba013fb0b87afc44187e259b594f9b93bcc5d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/879971c8f10bcdb1a04c2aac933e2832cc31057f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/879971c8f10bcdb1a04c2aac933e2832cc31057f"}]},"branch":"refs/heads/master"},"770e933577c005a1bfcb1342ce97522be5129a6a":{"kind":"REWORK","_number":2,"created":"2019-06-23 14:39:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/666988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/666988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/2"}}},"commit":{"parents":[{"commit":"2261d112b2c5a9c25e3eceb701fe131a9782ddf5","subject":"Merge \"Save amphora logs in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2261d112b2c5a9c25e3eceb701fe131a9782ddf5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-23 13:18:06.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-23 14:37:22.000000000","tz":120},"subject":"Set irrelevant files on CI jobs","message":"Set irrelevant files on CI jobs\n\nAll jobs are triggered regardless of files changed, except for files\nalready defined in octavia-dsvm-base. This means many jobs that are not\nimpacted by a given change will be run in CI. Not only this is a waste\nof CI resources but undeeded jobs may also fail (e.g. timeouts), thus\nblocking changes from merging.\n\nThis patch defines a more granular set of irrelevant files per job. For\nexample, it prevents active-standby jobs from running when a change is\nlimited to normal scenario files.\n\nChange-Id: I9eba013fb0b87afc44187e259b594f9b93bcc5d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/770e933577c005a1bfcb1342ce97522be5129a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/770e933577c005a1bfcb1342ce97522be5129a6a"}]},"branch":"refs/heads/master"},"531c5062c4cff602e455522c90fd9ce546f5ade1":{"kind":"REWORK","_number":3,"created":"2019-07-01 18:24:55.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/88/666988/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/666988/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/666988/3"}}},"commit":{"parents":[{"commit":"2261d112b2c5a9c25e3eceb701fe131a9782ddf5","subject":"Merge \"Save amphora logs in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2261d112b2c5a9c25e3eceb701fe131a9782ddf5"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-23 13:18:06.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-01 18:24:43.000000000","tz":120},"subject":"Set irrelevant files on CI jobs","message":"Set irrelevant files on CI jobs\n\nAll jobs are triggered regardless of files changed, except for files\nalready defined in octavia-dsvm-base. This means many jobs that are not\nimpacted by a given change will be run in CI. Not only this is a waste\nof CI resources but undeeded jobs may also fail (e.g. timeouts), thus\nblocking changes from merging.\n\nThis patch defines a more granular set of irrelevant files per job. For\nexample, it prevents active-standby jobs from running when a change is\nlimited to normal scenario files.\n\nChange-Id: I9eba013fb0b87afc44187e259b594f9b93bcc5d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/531c5062c4cff602e455522c90fd9ce546f5ade1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/531c5062c4cff602e455522c90fd9ce546f5ade1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
