)]}'
{"id":"openstack%2Foctavia-tempest-plugin~667038","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ifa7e58304e8d2cc5b34fe8ba5fff580e0cb0e823","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ifa7e58304e8d2cc5b34fe8ba5fff580e0cb0e823","subject":"Sync Sphinx requirement","status":"ABANDONED","created":"2019-06-24 08:06:27.000000000","updated":"2019-07-22 20:20:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0ba923658baecf073f3a126b1ad188fe34c42288","_number":667038,"virtual_id_number":667038,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-24 08:17:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 08:17:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"442c3ac670a8e38afa8476d5ae6845795878a5e6","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-24 08:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4ccb4e7c7888b4eaa34e9fe794013e08f95f51c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 08:17:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/38/667038/1/check/requirements-check/27e9cc8/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/38/667038/1/check/openstack-tox-docs/a728a74/html/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/38/667038/1/check/openstack-tox-pep8/1dfbba0/ : SUCCESS in 4m 13s\n- build-openstack-releasenotes http://logs.openstack.org/38/667038/1/check/build-openstack-releasenotes/a5837f4/html/ : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0ba923658baecf073f3a126b1ad188fe34c42288","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-22 20:20:36.000000000","message":"Abandoned\n\nDuplicate to https://review.opendev.org/#/c/666316/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f62ce5a6b25c42d12326f239793edc574bf1a81","revisions":{"5f62ce5a6b25c42d12326f239793edc574bf1a81":{"kind":"REWORK","_number":1,"created":"2019-06-24 08:06:27.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/38/667038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/38/667038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/38/667038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/38/667038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/38/667038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/38/667038/1"}}},"commit":{"parents":[{"commit":"2261d112b2c5a9c25e3eceb701fe131a9782ddf5","subject":"Merge \"Save amphora logs in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2261d112b2c5a9c25e3eceb701fe131a9782ddf5"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-06-24 08:05:16.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-06-24 08:05:16.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: Ifa7e58304e8d2cc5b34fe8ba5fff580e0cb0e823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f62ce5a6b25c42d12326f239793edc574bf1a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f62ce5a6b25c42d12326f239793edc574bf1a81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
