)]}'
{"id":"openstack%2Foctavia-tempest-plugin~672799","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ic3f620dcf02c756395da760f8b73874b40318cb6","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ic3f620dcf02c756395da760f8b73874b40318cb6","subject":"Sync enabled_provider_drivers config help message","status":"MERGED","created":"2019-07-25 17:35:13.000000000","updated":"2019-08-07 02:34:58.000000000","submitted":"2019-08-07 02:34:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672799-1565145298527-62203689","meta_rev_id":"34a5ab30879de8f21d98dd87b75ce6716e768674","_number":672799,"virtual_id_number":672799,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-07-26 20:21:06.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-07 02:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-03 00:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-07 00:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-08-07 00:43:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 00:32:49.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-07 00:43:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-07 02:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d505a3cd385a0bcf72f5f4b541e58fefdd4daf94","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-25 17:35:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b6e470626a3085cfb1dfca09185c13cac6367d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 19:46:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/99/672799/1/check/openstack-tox-docs/614784e/html/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/99/672799/1/check/openstack-tox-pep8/384bd9d/ : SUCCESS in 4m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-api/52992bc/ : SUCCESS in 45m 55s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-py2-api/2b7ecb1/ : SUCCESS in 39m 52s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-api-stable-stein/9daf77c/ : SUCCESS in 39m 48s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/c0b0202/ : SUCCESS in 47m 01s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/871f7bd/ : SUCCESS in 51m 32s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario/b85d88b/ : SUCCESS in 1h 28m 48s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-scenario/859b419/ : SUCCESS in 1h 54m 53s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario-stable-stein/eca923f/ : SUCCESS in 1h 35m 32s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/78c3d49/ : SUCCESS in 1h 36m 44s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/f207c47/ : FAILURE in 1h 35m 12s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario-centos-7/2e15b15/ : FAILURE in 2h 00m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2502695/ : SUCCESS in 1h 36m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario-two-node/3aa4c03/ : FAILURE in 1h 16m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-scenario-two-node/6bcfda5/ : SUCCESS in 1h 43m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/8a67efe/ : SUCCESS in 1h 09m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/7e05ee0/ : SUCCESS in 1h 07m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/e83ce92/ : SUCCESS in 1h 29m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/84e19d2/ : FAILURE in 1h 23m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/24238e2/ : FAILURE in 1h 28m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/ebfb6ba/ : SUCCESS in 1h 13m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-tls-barbican/08d768d/ : SUCCESS in 56m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/64e11c5/ : SUCCESS in 54m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/a5d32a6/ : SUCCESS in 57m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/72da84b/ : SUCCESS in 57m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-spare-pool/3631dae/ : SUCCESS in 54m 18s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-spare-pool/bfe34c3/ : SUCCESS in 54m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-spare-pool-stable-stein/e19b1db/ : FAILURE in 1h 04m 42s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/396fdc3/ : SUCCESS in 54m 01s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/6df189a/ : SUCCESS in 58m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"515409cd1565754f96343b904fe5d8d937f9d6ed","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-26 20:21:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"440d1834066c4abed29d3fc14a5f6720cd8bfeae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 22:20:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/672799/1/check/openstack-tox-docs/bbfcd1a/html/ : SUCCESS in 2m 53s\n- openstack-tox-pep8 http://logs.openstack.org/99/672799/1/check/openstack-tox-pep8/67f9fc3/ : SUCCESS in 5m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-api/08a05d6/ : SUCCESS in 43m 41s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-py2-api/a71ef6c/ : SUCCESS in 47m 49s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-api-stable-stein/e693014/ : SUCCESS in 55m 28s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/4a4008c/ : SUCCESS in 52m 07s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/356b0a3/ : SUCCESS in 51m 42s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario/d2fa363/ : SUCCESS in 1h 46m 36s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-scenario/7e00938/ : SUCCESS in 1h 51m 36s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario-stable-stein/309ccf2/ : SUCCESS in 1h 56m 16s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/7f04f85/ : SUCCESS in 1h 44m 01s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/99e9461/ : SUCCESS in 1h 38m 22s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario-centos-7/8b0af82/ : FAILURE in 1h 54m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/537385d/ : SUCCESS in 1h 39m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-scenario-two-node/b28ecc2/ : SUCCESS in 1h 41m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-scenario-two-node/20d22b3/ : SUCCESS in 1h 42m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/b659dc5/ : SUCCESS in 1h 20m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/a29de51/ : SUCCESS in 1h 04m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/9ad8009/ : FAILURE in 57m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/09f6a81/ : FAILURE in 1h 25m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/b2f885d/ : SUCCESS in 1h 04m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/b2560e7/ : SUCCESS in 1h 09m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-tls-barbican/fa99d8b/ : SUCCESS in 52m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/96ba278/ : SUCCESS in 47m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/7ffc076/ : SUCCESS in 1h 06m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/64afb96/ : SUCCESS in 51m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-spare-pool/ef408d3/ : FAILURE in 45m 33s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-spare-pool/543b6df/ : SUCCESS in 1h 05m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-spare-pool-stable-stein/e74b938/ : SUCCESS in 1h 12m 40s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/308a142/ : FAILURE in 1h 13m 08s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/99/672799/1/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/77c635f/ : SUCCESS in 48m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82f2644e845281425fe10feb7e4d04d3130c9232","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:32:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8fc5dc3e4f3c5c391aa919221dc2e4616483797","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-07 00:43:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"f7be4f066aa840eb9993cf18607c1f378d24a13c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:43:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29600a66f1bc28178be1ddf42f9c122f95341f37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:34:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/99/672799/1/gate/openstack-tox-docs/143823f/html/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://logs.opendev.org/99/672799/1/gate/openstack-tox-pep8/c1ed47e/ : SUCCESS in 3m 05s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-noop-api/e6ea187/ : SUCCESS in 46m 57s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-noop-py2-api/85b27cc/ : SUCCESS in 49m 08s\n- octavia-v2-dsvm-noop-api-stable-stein https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-noop-api-stable-stein/0ecf53d/ : SUCCESS in 39m 39s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/711bba6/ : SUCCESS in 42m 54s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/03c2d79/ : SUCCESS in 40m 11s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-scenario/945602f/ : SUCCESS in 1h 36m 36s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-py2-scenario/250e552/ : SUCCESS in 1h 43m 33s\n- octavia-v2-dsvm-scenario-stable-stein https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-scenario-stable-stein/9eccec8/ : SUCCESS in 1h 21m 56s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/455c5cb/ : SUCCESS in 1h 30m 59s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://logs.opendev.org/99/672799/1/gate/octavia-v2-dsvm-py2-scenario-stable-queens/bf8c23c/ : SUCCESS in 1h 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"34a5ab30879de8f21d98dd87b75ce6716e768674","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:34:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"decfc35aa862128228196c1113923b54d57ceb96","revisions":{"decfc35aa862128228196c1113923b54d57ceb96":{"kind":"REWORK","_number":1,"created":"2019-07-25 17:35:13.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/99/672799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/672799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/672799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/672799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/672799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/672799/1"}}},"commit":{"parents":[{"commit":"b6fb3d1993033aba38bb090cf82445d561b1d1cc","subject":"Merge \"Add a scenario test for spare pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b6fb3d1993033aba38bb090cf82445d561b1d1cc"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-25 17:31:52.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-25 17:31:52.000000000","tz":120},"subject":"Sync enabled_provider_drivers config help message","message":"Sync enabled_provider_drivers config help message\n\nUpdate the help message to match the same help message in octavia.conf.\nUsers were confused because the example was in Python dictionary format.\n\nChange-Id: Ic3f620dcf02c756395da760f8b73874b40318cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/decfc35aa862128228196c1113923b54d57ceb96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/decfc35aa862128228196c1113923b54d57ceb96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
