)]}'
{"id":"openstack%2Foctavia-tempest-plugin~673168","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ic887c726dfe151ea582a1bd935ae0b30582f14ea","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"story/2005259","hashtags":[],"change_id":"Ic887c726dfe151ea582a1bd935ae0b30582f14ea","subject":"Fix positional formatting and add skip check","status":"MERGED","created":"2019-07-28 11:51:13.000000000","updated":"2019-10-23 04:15:13.000000000","submitted":"2019-10-23 04:13:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673168-1571803996586-5cb63f8a","meta_rev_id":"a6e96198287b56045d321b037cae6cab41ffde98","_number":673168,"virtual_id_number":673168,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 04:13:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-10-23 02:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-13 12:48:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-10-09 22:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-10-23 02:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 12:48:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-10-09 22:37:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-23 02:18:38.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-10-23 04:13:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c642c3f8168cac0992c1ca52143582d9fb4e0849","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-28 11:51:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1225cea6fbc4d7a67dcc9faf80ad788fa47f6f9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-28 11:52:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d883776ccc241dbf91dfec0c9f02794b421b6d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 14:07:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/673168/2/check/openstack-tox-docs/5509d47/html/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/68/673168/2/check/openstack-tox-pep8/cd22d57/ : SUCCESS in 3m 02s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-noop-api/0e887c4/ : SUCCESS in 46m 09s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-noop-py2-api/1e43056/ : SUCCESS in 40m 38s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-noop-api-stable-stein/53c1586/ : SUCCESS in 44m 26s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/5ed7aee/ : SUCCESS in 38m 39s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-noop-py2-api-stable-queens/b2bb532/ : SUCCESS in 41m 09s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-scenario/b2e9d1f/ : SUCCESS in 1h 25m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-py2-scenario/f17de16/ : SUCCESS in 1h 21m 31s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-scenario-stable-stein/355c079/ : SUCCESS in 1h 23m 02s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-py2-scenario-stable-rocky/fb1b97c/ : SUCCESS in 1h 28m 48s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-py2-scenario-stable-queens/3b16c73/ : SUCCESS in 1h 16m 39s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-scenario-centos-7/e488f6e/ : POST_FAILURE in 2h 08m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/fd54dc9/ : SUCCESS in 1h 32m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-scenario-two-node/256f87f/ : SUCCESS in 1h 58m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-py2-scenario-two-node/107950b/ : SUCCESS in 1h 49m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/68/673168/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/6445787/ : SUCCESS in 1h 20m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/68/673168/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/0ac5cff/ : FAILURE in 1h 02m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/68/673168/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/192b827/ : SUCCESS in 1h 18m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/68/673168/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/60c9c22/ : SUCCESS in 1h 01m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/68/673168/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/7df884b/ : SUCCESS in 1h 28m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/68/673168/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/74655bf/ : SUCCESS in 1h 23m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-tls-barbican/2638fb3/ : SUCCESS in 57m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-tls-barbican-stable-stein/1126fd5/ : SUCCESS in 1h 02m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-tls-barbican-stable-rocky/54ac63c/ : SUCCESS in 54m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-tls-barbican-stable-queens/e1c0857/ : SUCCESS in 1h 11m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-spare-pool/f402071/ : SUCCESS in 56m 46s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-py2-spare-pool/beef516/ : SUCCESS in 1h 06m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-spare-pool-stable-stein/175a029/ : SUCCESS in 54m 51s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/3b34d53/ : FAILURE in 52m 30s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/68/673168/2/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/989763a/ : SUCCESS in 54m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef7657e0f0e2c890777b41609288886e12b85b98","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-02 09:30:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"20da13b47752bb7922224fd13f472882209c706c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 12:02:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30294814cada13655b60e8ab415dcc132e19726c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 12:02:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f2a25bc54a3515c31fd40fc2dca80c2557f069e5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-13 12:15:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e967afb0a47badf2e19d7548a5957ca95ca62c9e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 12:39:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60137d83795024982a728542b5ef564cea2268cf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 12:39:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5d5f345c06016d5174320117daafc6b5da4bac66","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-13 12:48:38.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ea7db43ececf48c98b310ef76fe34ed743ed5a53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:58:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe6f52708ce64a95a6e7c7552c9fc651 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a737fdd78ec47ab8ab911201b57ae88 : SUCCESS in 2m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4f3764caa904eca9757e18a58629ffd : SUCCESS in 47m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9b40562c0eea42fc866e93be1766848c : SUCCESS in 49m 12s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f506a7431c10455dab3b5905d008d966 : SUCCESS in 43m 04s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/95d4f241252a4a749448575d4f1141c6 : SUCCESS in 42m 38s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/039d0bd662fc4fa8ae8e63264f192f3a : SUCCESS in 45m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80739a902c5548ba82d9c6727cc0ce9e : SUCCESS in 1h 44m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5d5f818664aa4413a90d2f8ebc4f7785 : SUCCESS in 1h 48m 53s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fedbff355e524b46bc66134cd71f3b2b : SUCCESS in 1h 54m 15s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/7c425e956f4f448fba938bec7ef8a359 : SUCCESS in 1h 40m 27s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/525025965c8c4f8ebcb113e55ffca524 : SUCCESS in 1h 37m 21s\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4cf5c5d05c3e4fb891e73999207048f4 : SUCCESS in 2h 09m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1202fceb04b4303a3c841f84816f588 : SUCCESS in 2h 03m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5b752aedaa44297bfc6f51bfb7d9706 : SUCCESS in 1h 19m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/83aa03b5b6b649658f4d611a0c2ac60e : FAILURE in 37m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a9b467f33c943d5b5f69212fb5da366 : SUCCESS in 1h 09m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f96cd45e7572498bb574cc0d2ec081b7 : SUCCESS in 1h 18m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/84bfb7a9fe054cecb62e25b88bea0bfa : FAILURE in 58m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/470ae1168e1643d3b5d10ce8c900f10b : FAILURE in 1h 05m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/a5cbdc430b4b48a8b4e90304c198d900 : SUCCESS in 1h 06m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/fb5839fc139440ccbaa70fc349130010 : SUCCESS in 1h 07m 49s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b10dd2456304170a992105c40396e5f : SUCCESS in 54m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/bba810ffcb3449e2902558d6febf9a86 : SUCCESS in 58m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/19b5e487cf4e4a2e9ee038abb423876d : SUCCESS in 57m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/69c92d7881f3412eb681ddbb63c9672e : SUCCESS in 1h 05m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/664cecb7820942299953e0575cba47c0 : SUCCESS in 59m 57s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/83ae6df3ff14486c89fd4515d9d28bbd : SUCCESS in 1h 01m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f3eb6b22c1b24109be2b906192db7117 : SUCCESS in 1h 06m 58s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/16aeaf151c21486985895bf871d0ddad : SUCCESS in 54m 12s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/5d90bc1bcec84f7985b4a71163511e66 : SUCCESS in 59m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d7abd28ddb5f77013c13afe4b2d065ce4b35cb29","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-09 22:37:31.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"5368d0904c1630e241f645f5d852072beb1d6127","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-10-23 02:18:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"909b9fe48c2441052f787130e499ac1f86de3967","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 02:18:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"740a477bdf56bc4f146e8cc0b742c1ae313206ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 04:13:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217a3aa5f25b498585d169ecc81ecf2a : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3ee3285a892460993e2d5779c56ff34 : SUCCESS in 2m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22c124de6b944ead9b008f657cc91c4d : SUCCESS in 42m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/96e6734149fd40ae955e9f73f1a2c2f7 : SUCCESS in 44m 28s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/37b244fe34f449c48b222d4d002f6a0b : SUCCESS in 46m 38s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/ad879f2c17c44ca7b742cc5779b21125 : SUCCESS in 43m 52s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/b8b4b91deb51484e947177e11041fd03 : SUCCESS in 46m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2101ab888d34020a790671ed2da0275 : SUCCESS in 1h 28m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a8b14b73920e49a5bfdbd0ae42f34d7f : SUCCESS in 1h 52m 47s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6f6b4ce23c994fe0bf8167445a5a84eb : SUCCESS in 1h 46m 14s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/b7612e347fc0482ba4e3beeacae5ace3 : SUCCESS in 1h 23m 27s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/6e19de2f1d0d4494aa8ef316aece9a70 : SUCCESS in 1h 44m 07s","accounts_in_message":[],"_revision_number":4},{"id":"4d495f9e6fa69cf12f84991c55feda067c9ce81a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 04:13:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f63d9730e0c3168ca4d0b6c478554d5ef48753cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 04:15:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21511c4c43194b2cb5fc779e9199b8cd : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9891de0ebb5c18de0a056ef37ee61642d0995898","revisions":{"9fed9f11a7c8c3480017622b0afdc04b3386a347":{"kind":"REWORK","_number":1,"created":"2019-07-28 11:51:13.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/68/673168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/673168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/1"}}},"commit":{"parents":[{"commit":"65e7c5fcd47d04059cb2203776f84255d9177ea2","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/65e7c5fcd47d04059cb2203776f84255d9177ea2"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 11:37:33.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 11:50:04.000000000","tz":120},"subject":"Fix positional formatting and add skip check","message":"Fix positional formatting and add skip check\n\nThis patch fixes an offset positonal argument in a string format that\nwas throwing an IndexError exception. It also adds a skip check to check\nif the required amphora SSH key file is readable.\n\nAdditionally, the iptables-based active-standby test has now a better\namphora_ssh_key default value to match with the default path set by the\nOctavia Devstack plugin.\n\nStory: 2006259\nTask: 35876\n\nChange-Id: Ic887c726dfe151ea582a1bd935ae0b30582f14ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9fed9f11a7c8c3480017622b0afdc04b3386a347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9fed9f11a7c8c3480017622b0afdc04b3386a347"}]},"branch":"refs/heads/master"},"23156003d9d49b98f10487627f3d7474e8ffaf35":{"kind":"REWORK","_number":2,"created":"2019-07-28 11:52:49.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/68/673168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/673168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/2"}}},"commit":{"parents":[{"commit":"65e7c5fcd47d04059cb2203776f84255d9177ea2","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/65e7c5fcd47d04059cb2203776f84255d9177ea2"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 11:37:33.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 11:52:41.000000000","tz":120},"subject":"Fix positional formatting and add skip check","message":"Fix positional formatting and add skip check\n\nThis patch fixes an offset positonal argument in a string format that\nwas throwing an IndexError exception. It also adds a skip check to check\nif the required amphora SSH key file is readable.\n\nAdditionally, the iptables-based active-standby test has now a better\namphora_ssh_key default value to match with the default path set by the\nOctavia Devstack plugin.\n\nStory: 2006259\nTask: 35876\n\nChange-Id: Ic887c726dfe151ea582a1bd935ae0b30582f14ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/23156003d9d49b98f10487627f3d7474e8ffaf35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/23156003d9d49b98f10487627f3d7474e8ffaf35"}]},"branch":"refs/heads/master"},"08eb643c1234f8eacac2337db887f656385955ea":{"kind":"REWORK","_number":3,"created":"2019-09-13 12:02:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/68/673168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/673168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/3"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 11:37:33.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-13 12:01:56.000000000","tz":120},"subject":"Fix positional formatting and add skip check","message":"Fix positional formatting and add skip check\n\nThis patch fixes an offset positonal argument in a string format that\nwas throwing an IndexError exception. It also adds a skip check to check\nif the required amphora SSH key file is readable.\n\nAdditionally, the iptables-based active-standby test has now a better\namphora_ssh_key default value to match with the default path set by the\nOctavia Devstack plugin.\n\nStory: 2006259\nTask: 35876\n\nChange-Id: Ic887c726dfe151ea582a1bd935ae0b30582f14ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/08eb643c1234f8eacac2337db887f656385955ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/08eb643c1234f8eacac2337db887f656385955ea"}]},"branch":"refs/heads/master"},"9891de0ebb5c18de0a056ef37ee61642d0995898":{"kind":"REWORK","_number":4,"created":"2019-09-13 12:39:27.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/68/673168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/673168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/673168/4"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 11:37:33.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-13 12:37:51.000000000","tz":120},"subject":"Fix positional formatting and add skip check","message":"Fix positional formatting and add skip check\n\nThis patch fixes an offset positonal argument in a string format that\nwas throwing an IndexError exception. It also adds a skip check to check\nif the required amphora SSH key file is readable.\n\nAdditionally, the iptables-based active-standby test has now a better\namphora_ssh_key default value to match with the default path set by the\nOctavia Devstack plugin.\n\nStory: 2006259\nTask: 35876\n\nChange-Id: Ic887c726dfe151ea582a1bd935ae0b30582f14ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9891de0ebb5c18de0a056ef37ee61642d0995898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9891de0ebb5c18de0a056ef37ee61642d0995898"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
