)]}'
{"id":"openstack%2Foctavia-tempest-plugin~673726","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"log-offload","hashtags":[],"change_id":"I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2","subject":"Add an optional amphora admin log offloading test","status":"MERGED","created":"2019-07-31 04:33:03.000000000","updated":"2020-08-13 15:15:59.000000000","submitted":"2020-08-13 15:13:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673726-1597331616462-5080a3ed","meta_rev_id":"1822e776ef0d09d9eb7e3d84db6b63ed0edbf014","_number":673726,"virtual_id_number":673726,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 15:13:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-13 10:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-08-13 10:13:17.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-13 10:13:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-08-13 10:13:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-08-13 10:13:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-13 15:38:11.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-13 18:07:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-07 06:57:27.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-13 10:13:24.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-13 15:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92bf5284f5094a5a402d5c41ec8390849062f8f9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-31 04:33:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d9c731583fa2e559cda4358a20f24cd8c182265","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 06:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/26/673726/1/check/openstack-tox-docs/a188ba0/html/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/26/673726/1/check/openstack-tox-pep8/e531d2f/ : SUCCESS in 3m 58s\n- build-openstack-releasenotes http://logs.openstack.org/26/673726/1/check/build-openstack-releasenotes/1eb7248/html/ : SUCCESS in 3m 47s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-noop-api/8596ef9/ : SUCCESS in 42m 06s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-noop-py2-api/df4341a/ : SUCCESS in 40m 36s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-noop-api-stable-stein/332e339/ : SUCCESS in 41m 26s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/14c79e7/ : SUCCESS in 44m 06s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/09b8567/ : SUCCESS in 38m 32s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-scenario/bf2db9b/ : SUCCESS in 1h 32m 21s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-py2-scenario/dfa8a47/ : SUCCESS in 1h 43m 03s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-scenario-stable-stein/21b9c46/ : SUCCESS in 1h 24m 36s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/8b68e54/ : SUCCESS in 1h 44m 02s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/7fde05e/ : SUCCESS in 1h 32m 48s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-scenario-centos-7/3b25994/ : FAILURE in 1h 57m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/8872f14/ : SUCCESS in 1h 48m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-scenario-two-node/e5bf303/ : SUCCESS in 1h 13m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-py2-scenario-two-node/3d6599f/ : FAILURE in 1h 21m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/26/673726/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/a94a39b/ : FAILURE in 1h 01m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/26/673726/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/f3e3bbc/ : FAILURE in 1h 18m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/26/673726/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/b25ab42/ : FAILURE in 1h 35m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/26/673726/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/9f3b97e/ : SUCCESS in 1h 02m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/26/673726/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/cd28885/ : FAILURE in 1h 19m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/26/673726/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/55857e4/ : SUCCESS in 59m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-tls-barbican/1f36eaa/ : SUCCESS in 55m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/93e5206/ : SUCCESS in 51m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/c1377b9/ : SUCCESS in 50m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/666f8b3/ : SUCCESS in 55m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-spare-pool/d96b9c6/ : FAILURE in 1h 01m 37s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-py2-spare-pool/980440a/ : SUCCESS in 53m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-spare-pool-stable-stein/ca1fa69/ : SUCCESS in 56m 07s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/b858e51/ : SUCCESS in 54m 25s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/26/673726/1/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/f788eb0/ : FAILURE in 1h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b774a28fef504fd2722fbca5c3075c9e02d7cd95","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-31 17:08:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2b6e05b05d08a52c971f5ba2d299ac212e45ca41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 19:24:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/26/673726/2/check/openstack-tox-docs/8f398c4/html/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/26/673726/2/check/openstack-tox-pep8/cb7e31b/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/26/673726/2/check/build-openstack-releasenotes/27b606a/html/ : SUCCESS in 2m 51s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-noop-api/3a7bc3c/ : SUCCESS in 50m 23s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-noop-py2-api/ed3fd44/ : SUCCESS in 47m 28s\n- octavia-v2-dsvm-noop-api-stable-stein http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-noop-api-stable-stein/51609f8/ : SUCCESS in 51m 23s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/a705e99/ : SUCCESS in 47m 53s\n- octavia-v2-dsvm-noop-py2-api-stable-queens http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-noop-py2-api-stable-queens/db7e190/ : SUCCESS in 43m 00s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-scenario/23e0262/ : SUCCESS in 1h 52m 47s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-py2-scenario/ca4f473/ : SUCCESS in 2h 01m 02s\n- octavia-v2-dsvm-scenario-stable-stein http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-scenario-stable-stein/e9cd0bb/ : FAILURE in 1h 26m 09s\n- octavia-v2-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-py2-scenario-stable-rocky/278adcf/ : FAILURE in 1h 45m 36s\n- octavia-v2-dsvm-py2-scenario-stable-queens http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-py2-scenario-stable-queens/b970e39/ : FAILURE in 1h 33m 18s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-scenario-centos-7/a69acc4/ : FAILURE in 1h 58m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/1ea46a9/ : SUCCESS in 1h 33m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-scenario-two-node/302c25b/ : SUCCESS in 2h 06m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-py2-scenario-two-node/7e6472b/ : FAILURE in 1h 30m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/26/673726/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/d609e12/ : FAILURE in 1h 18m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/26/673726/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/b14251c/ : FAILURE in 1h 28m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/26/673726/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/7369338/ : FAILURE in 1h 06m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein http://logs.openstack.org/26/673726/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/451e184/ : SUCCESS in 1h 05m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky http://logs.openstack.org/26/673726/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/ed3e610/ : FAILURE in 1h 24m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens http://logs.openstack.org/26/673726/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/d3b5d3d/ : SUCCESS in 1h 04m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-tls-barbican/4a4c504/ : SUCCESS in 1h 01m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-tls-barbican-stable-stein/b7824b4/ : SUCCESS in 1h 02m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-tls-barbican-stable-rocky/587aa1d/ : SUCCESS in 54m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-tls-barbican-stable-queens/cc6fefd/ : SUCCESS in 55m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-spare-pool/450e79e/ : SUCCESS in 1h 00m 21s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-py2-spare-pool/326ca54/ : SUCCESS in 1h 10m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-spare-pool-stable-stein/5acd3a8/ : SUCCESS in 1h 06m 52s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/6579720/ : SUCCESS in 1h 00m 19s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens http://logs.openstack.org/26/673726/2/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/e2aacbb/ : FAILURE in 51m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e9d4c62b5d9cfac4fe483503a6262f147657f5fb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-09 21:51:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"cf6ae871e0ad6268722498f2188381ed0fca2e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 00:00:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e58273fac29481b88417c4e67d4a3e2 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d9e7bf4fb4474eb2b918fa3bacd301 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ed5f285e9414e078385722965b28bde : SUCCESS in 5m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d3796fbd5b441678e0600ccaa223cac : SUCCESS in 45m 45s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/aa7b61ab6d9e4a728fd77492e1661598 : SUCCESS in 42m 03s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/2378dae119b5472682cc3e2386e35608 : SUCCESS in 1h 01m 41s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/6a583c4688f747e59c12c14e18bd75de : SUCCESS in 37m 33s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/49e75558daf34f5694b2ca5bacdb2620 : SUCCESS in 40m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0567fdf4bed47a1b0983c3e0f71e658 : SUCCESS in 1h 38m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b162d6174c1d4e809bbfa21b71cda525 : FAILURE in 1h 36m 16s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a6ea9b246e25446394bb5dea13de8ed0 : FAILURE in 1h 18m 07s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/8a79f36d28c84e47b5ac0a46a2861197 : FAILURE in 1h 37m 42s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/b5e4589d4cf04050b58ecac57076b9cc : FAILURE in 1h 56m 20s\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/80d03d1699ce4dcfa514bd2b3917a85c : FAILURE in 2h 06m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48cbbfd4df9a4721880a1bf23d83524b : SUCCESS in 1h 35m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fce00db5ebf8448c91b41d74ac4190a7 : SUCCESS in 1h 12m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c6114c2e88148a8aa59265a6240c6c3 : FAILURE in 1h 27m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/acfc95d682a640beaac85b9aac00fe27 : SUCCESS in 1h 02m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cedede06bc9f4842bdfc1076a0929ee3 : SUCCESS in 1h 04m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5f308b6e821b4583914907bce554ab23 : SUCCESS in 1h 18m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/cde7e4c29cdf4244a3d8892bff92d8bf : FAILURE in 1h 34m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/60572fb51df2494a9e0270980e12107b : SUCCESS in 58m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/3c4eb243c25645f8af89dbe32dcd93b3 : SUCCESS in 1h 00m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7b3873fbcbc47ae93c4b491f54b941a : SUCCESS in 1h 03m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cab08fb4564d4d789cb97df1df97d3c9 : SUCCESS in 59m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6a8b95dbc128485990be05eebe8bb4f3 : SUCCESS in 1h 02m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/085855741b784c46ac682794a6203fce : SUCCESS in 54m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/b5545ec2c12a42e9b815c293ae892785 : SUCCESS in 49m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/74009cc8d98d48aeada47522202a77da : SUCCESS in 54m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/49612112b13745af8537799f11765cae : SUCCESS in 50m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e547055fd36413c8df5a774c0d73a14 : SUCCESS in 57m 30s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/bae9b809d50f441b8515e31f9580f82c : SUCCESS in 50m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/ab8227fac90442829dfb1a58ba82aaa7 : SUCCESS in 1h 02m 25s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/e479cedf4dcb4495a947509be17784ba : SUCCESS in 52m 49s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/0aff367d009046788faaec98b4f55b3e : SUCCESS in 1h 37m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1112173664254d2ea5e62c597619111e : SUCCESS in 1h 37m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c2aceea3a4e7008af27289a7624dba1a24ff51ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-10 15:57:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"df4351d0fa70073e085b2d1a0ccfae66b185e58b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 18:14:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203824a5dd2447709b8a9cf4b63ea43d : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d92900278ec425e813362203b666133 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d662cca001b47268032e50a4d2ad954 : SUCCESS in 4m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7cd46ce229384dc39eda8f212eb602d2 : SUCCESS in 50m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/129ec6d73b60408f8728fc551cd6ab6e : SUCCESS in 46m 15s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/fdd08ca149a043faadc23b7f19c94cd5 : SUCCESS in 44m 49s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/e2ebc65df8974873a5c442c042c1dc8c : SUCCESS in 40m 51s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/7cb154c8ee264bc4ab2e0c22fd15b237 : SUCCESS in 43m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d7d63a715ee4aa19575743b51d9ebe1 : SUCCESS in 1h 50m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/539affc9f2414d5699194095d9e84997 : SUCCESS in 1h 50m 18s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c9f5def0b802469a8b4401675764b6ef : SUCCESS in 1h 51m 33s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/aa0cd95b9d2a49df8ec13258f720fe76 : SUCCESS in 1h 57m 54s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/5c2eccfc10684eceb1ebd365b5aa18a5 : FAILURE in 1h 23m 54s\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9e584f887a1d45818b1dffdf9222fc20 : FAILURE in 2h 10m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a8db914dfae461bbfe1f46e06922d67 : SUCCESS in 1h 37m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1d352a0d4b740d5bc610ec246d78d04 : SUCCESS in 1h 58m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6cb750d9b17e4c1e8695a07ecdf9ebb1 : SUCCESS in 1h 58m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6e51803f14649849ce0a3ad35c78697 : SUCCESS in 1h 11m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cb7a74321ab34916b551e4a7b39a4c29 : SUCCESS in 1h 24m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6524a57dec7848bf9f1db1d49571d872 : SUCCESS in 1h 23m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/249afd674ef74de7bee27d92fe59b500 : FAILURE in 1h 46m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/5c070e3797bb4cc293c0e467a7f6d363 : FAILURE in 1h 44m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/41dfd71094dd4852a583591d1def1980 : SUCCESS in 1h 24m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78a8912741a440e5bb718cbcea310a5d : SUCCESS in 59m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fb860ab94b844ee29914a825d082521c : SUCCESS in 1h 05m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/5f6551b053f74b059c44fa1adfa5d5c2 : SUCCESS in 1h 09m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e87b7720267d4a1b8777f9931ddb6f61 : SUCCESS in 57m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/50a1d38d046e494b8cfd28870f5eb9db : SUCCESS in 55m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/062f242bf75e4e20a50937fda9246cce : SUCCESS in 1h 01m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/4001edf5856a4716aba0c9f58a7efa86 : SUCCESS in 52m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/480b0c4d06cc4cc9ba22e2be12541fb1 : SUCCESS in 1h 06m 43s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/a2048fe67a6847c9bfda822ed56fb291 : SUCCESS in 1h 01m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0c5c1eba712849e3b8e904ebdbc95f6b : FAILURE in 1h 18m 27s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/3837b32117cd4ffe836b269818687445 : SUCCESS in 1h 03m 49s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/f6f9a631fd5c4868af9ab25b3a8c95d6 : SUCCESS in 58m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e64a407cb3d644e181c8d20336a62365 : SUCCESS in 1h 54m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f4dd92361619ea46f0501999da6f7237ca8352b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-11 23:23:27.000000000","message":"Patch Set 4:\n\nrecheck\npick up parent depends-on","accounts_in_message":[],"_revision_number":4},{"id":"b330c35af79188c4b1ad7304a83f7d4b0f08f1f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 02:09:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2deaa393bd034fc6956b7a5351204eb3 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1344cb4db8c944db8935bf3e2863fbff : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06b3b6b325254774afeca29d2393c1c7 : SUCCESS in 2m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03c656d304584f8dbfa75fe2333c689f : SUCCESS in 44m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cd9744657b904b20b4dafe5ebd648c6a : SUCCESS in 42m 54s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/ada1b88be3664bd79e6beb4b313fbda8 : SUCCESS in 43m 42s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/d8a79d89cd3647338a2e0ea50cdaa038 : SUCCESS in 41m 12s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/80a1f66f4f8c402d9d2a1da08fe60ae7 : SUCCESS in 41m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc67366727aa4cadab8c186d59744475 : SUCCESS in 2h 01m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b89d7a6c62ed459b9c58e3452acddaf0 : SUCCESS in 2h 19m 40s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4837d1b497d64f4ebf592bf69ea69ff5 : SUCCESS in 1h 51m 19s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/d34d372ca844413a92fc0ad08ed8d417 : SUCCESS in 1h 37m 44s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/08dc6158d01743ffa022fb2612ecbd4d : SUCCESS in 1h 44m 37s\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/42bd06ee93404ce3aa0fb30cbae6d3c7 : TIMED_OUT in 2h 35m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbd34df0536d4297b58961fe4dfed264 : SUCCESS in 1h 49m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb33ba3812284e0ab1dd8f4cadfa7a7f : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/dbd7ba84d2464b3e8c3734f1a091b009 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2979aec705c043b484a56eeb2066f39e : SUCCESS in 1h 22m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9b0c4196d6d54f8b83e1b523dc64c5cd : SUCCESS in 1h 09m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0c5a9a0a9505473ebccb262434f27750 : FAILURE in 1h 25m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7aefe68916074eb58d199f0be5d39ae0 : SUCCESS in 55m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/86882f026556459794f19a2effad7949 : FAILURE in 1h 13m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/add627d35a1648ee9091532a2070d086 : SUCCESS in 1h 04m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cc0d3c2f00c4c3f91596772c1bacf53 : SUCCESS in 1h 09m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fde5066efc214323b97d16c174131509 : SUCCESS in 1h 09m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/f3c7d8d80079450e8f6f76317a2f1cf8 : FAILURE in 1h 06m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f5e480b68bb5432fb554c2e1efc72ff6 : SUCCESS in 1h 06m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/7281189162964b51905e06e3c1c6ecda : SUCCESS in 1h 25m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/cec8c8c20e7145b88e6711f329f57511 : SUCCESS in 56m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/c20c07215d5a4cee8934aae90c7e6757 : SUCCESS in 53m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/46261381f06d41fd9d394d7264ba6856 : FAILURE in 1h 41m 39s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/c763a8d314344a7fba56edb3f8f564a7 : POST_FAILURE in 59m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/2cd6f27d5f0440ca9971288605d70cc9 : SUCCESS in 57m 54s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/460ca93bb463400ba3b408e2ed249f8c : SUCCESS in 58m 15s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/9e12924cda2c42769190725b1dba3608 : SUCCESS in 1h 03m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6ee091f014234e56b1522d8c21e3e4f0 : SUCCESS in 1h 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1304a1e0072814035e642a1d2df0858c056612f8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-08 22:34:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"73a9f1431a11bd61f41b07bb1960d2491aaebbd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 00:57:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129da5377cde40cbbffa8b51efaeb249 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6907ff91e748aea55e9b2c13dde488 : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79e7405eaecb4a09ae80c1284d2e7c71 : SUCCESS in 3m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/529fb7441f3e4b82becb29b8e8786b63 : SUCCESS in 45m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/13ae8740952d4a4ca4b45aee11de6088 : SUCCESS in 53m 06s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/0a28320843a94262bbc0c19d7c3468c9 : SUCCESS in 47m 40s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/177d8a3521744d8ba3aa1f4d7380e1aa : SUCCESS in 42m 54s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/682b5f26cc3940f48493a6b3058c13f5 : SUCCESS in 40m 29s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/511ee5a5235642b38f00412d93527b45 : SUCCESS in 48m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3841161b3b1f4b45a6aacc49c8c7dfb3 : SUCCESS in 1h 46m 57s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7c9638c53ebc4008a57b4904d9e76cf5 : SUCCESS in 1h 43m 01s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/66737bf594944bf1acf19f8a90d366c5 : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/2ba26d114b7448bf84ad5e2ca24a53d9 : SUCCESS in 1h 44m 09s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/6c0526ff543b4154a834662714867720 : SUCCESS in 1h 51m 22s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/97a86b886186447293288304e9f77c87 : SUCCESS in 1h 43m 16s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7d82c200c740b6ae1e5b0b32b7e6f4 : FAILURE in 37m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9005afc2fe8a44888b7f365b3f9dd671 : SUCCESS in 2h 15m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b2d7fdd6b3341f5b5e0ed22c518aecf : SUCCESS in 1h 50m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/360f381346ba4e28aa48192a87bbb2aa : FAILURE in 40m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2eab196eabd245cd9d9dd97db18cd226 : SUCCESS in 1h 41m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/0b636ce2216d4654b255d8c5697b6f8b : SUCCESS in 1h 10m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/daa739b3b0b943248e0a5c338c3dff9d : SUCCESS in 1h 08m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b426090370b04b5bba666acc18402dd1 : SUCCESS in 1h 07m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7485f2e4deb44499a3a71fcc925e03ce : SUCCESS in 1h 04m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/280dc3988a5e497bb1e5bd431ab1fb36 : SUCCESS in 1h 05m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b1a53852aa854ee584e048854a714cfe : SUCCESS in 1h 00m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b45e0405a72547f18eec9462d25b0847 : SUCCESS in 47m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/1a5dda32d4ef49d8bb8a48a0b73f6d94 : SUCCESS in 53m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/3b194bdc99d64f0ba414252bfd18081e : SUCCESS in 57m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/7af890de48524baea84664f53ec5bb16 : SUCCESS in 52m 11s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/462fd81473764f03891801efaa9b8612 : SUCCESS in 1h 00m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cb9b98d3f9d946cca21e24df70a67eec : SUCCESS in 1h 05m 47s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/b8d7257ce1154bfeb975a1f74f549bb1 : SUCCESS in 1h 03m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/1489177e17bc4962a1006431599b9d11 : SUCCESS in 1h 02m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/80dfe2b6c33f4d0cabfb767e951a357f : SUCCESS in 56m 55s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/5864a651957e4186af76e489f4ae1432 : SUCCESS in 56m 59s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/f60a9b1357384d12be756dfebc786b23 : SUCCESS in 1h 04m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/734f6a7140b642e794627f6a51cac534 : SUCCESS in 1h 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bf6822558d68e2b2e2bc2dd1183dac997e5b8a63","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 11:03:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"90db066d5ae6d7ccd8b4b8ef4ab02f6bfd4b5e5f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-03 21:16:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a1927098aa7633c9d72770785c65d8995c1bcd3f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-03 21:19:46.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"ab2a8e51b76bc4e66e8b9c0366e060332a96ade3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 23:56:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1911874ec3a4a51be71a24e6713b0d0 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0a33cd4e7b64a7da814edc1d0a5121c : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4966f52b39e34b3b837d83440317e94e : SUCCESS in 3m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/801695534ad24714b514e5ed2723c940 : SUCCESS in 41m 28s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/b4c3a3d8df744c478d8a863aff9f5cf0 : SUCCESS in 43m 56s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/0457acf3f8a7402d80426d24620678b3 : SUCCESS in 46m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4664e794e4840058ea9aa407c88c7f3 : SUCCESS in 1h 43m 54s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6e5277d7faeb4eb8b5745c18e5f62e9e : SUCCESS in 1h 41m 45s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1b9707ce13514bb68c1850bdc22c82d7 : FAILURE in 43m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d8680dfaed7f4330bcbf7033734ce41e : TIMED_OUT in 2h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a65b82bfd3e4434d81c968354fcb110a : FAILURE in 1h 43m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/292a8b0c78c6478d963a31f3b735ff6d : SUCCESS in 1h 52m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43dab66c4ac84f2995a4389be0ba6912 : SUCCESS in 53m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d4d264bed80e404db5607d1f16cd428b : SUCCESS in 59m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b33c9c3d259d4aa2b531370beed579df : FAILURE in 27m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62c978d5c662470895fa6a6a80b67b51 : SUCCESS in 49m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ca1af450c10c45d4973bc383c5df554d : SUCCESS in 1h 09m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/1b722989f2ee4c5db91568c503d1d146 : FAILURE in 31m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e0b3fb0fac624e5e9894923cb93b1bef : SUCCESS in 1h 02m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/eb85ee1f3902415aa681d5ea340e187e : SUCCESS in 51m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/1b509f169ca14e2dab7ee258878cca73 : FAILURE in 41m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/02f1f51e52b04c2b838bdff518379497 : SUCCESS in 1h 48m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0c467d9c49eecdc31951414ad552fd78cdaf3c05","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 17:42:48.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"a017ac02f174a5f167dcfc913dcdaa4f41b766d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 20:31:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e4645def1144a4a14849c660d14e9a : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3931881873b14bcca0df7be6717f5583 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/303583dc4a9d400bb755e8564012f411 : SUCCESS in 3m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40179f7a8f09495191d4f5f5fe3270b5 : SUCCESS in 47m 38s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/6bbe8e5cd6544cc5bf22db6811e2cce1 : SUCCESS in 53m 56s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/507160c1f4154a71b5e5e02bb98e7b21 : SUCCESS in 41m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5285792d5efb48dca7d8a73443c8474f : SUCCESS in 2h 02m 11s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/52e73453882540abaaf239dabab2ac1a : SUCCESS in 2h 00m 13s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/391c2a9672f44895905da3afcc508d2b : SUCCESS in 1h 57m 46s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/476b9e0649e64131a7739e775f81878b : TIMED_OUT in 2h 34m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e2c7624e003c448195e0ffb57c21e77c : TIMED_OUT in 2h 34m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e15e8239b4494d2bb3a38218f809aa4c : SUCCESS in 2h 12m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc6f48d848664ad7aae2a84db369a317 : SUCCESS in 59m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6c30cebdb5ad48ad93beb8a287787752 : SUCCESS in 1h 06m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6056d53e466e4b4990b577df70daf3d5 : SUCCESS in 58m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31e6bbec30ca4ff2befab77f1769a577 : SUCCESS in 55m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/2e09a8b998234fc79e849fb6f0273494 : SUCCESS in 1h 11m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/c030ae6ec6ee4835bcf2d65595ef9ba1 : FAILURE in 59m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7435c66c38df488abea9946573c353d9 : SUCCESS in 55m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/7d1c1ca41f144160a93c9554a511e449 : SUCCESS in 54m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/9e541a6bf5674645a288918317ab16a4 : SUCCESS in 53m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07be04d5728f4e87ab34ce9b34e6b8fd : SUCCESS in 1h 47m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"132f4d54a1f023f99b0019114ea32398cbe2d409","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-24 14:08:27.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"f2082a495146b462e8411b9b98057d8677f55e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:59:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9363c93c5d2a41048d0d5e0a92114783 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07995bc5bd24c239b7fe0dc10d1a4fa : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93b8d44c5a0643c3bd176b45fddebe8f : SUCCESS in 10m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53f90f2bc3dc4dfd972fce9ee1780ae7 : SUCCESS in 1h 13m 52s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/37e23f0bae7b4456a19f5d7cee744659 : SUCCESS in 1h 24m 20s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/06485b163c0e4c45a2728ee1defc84a8 : SUCCESS in 1h 24m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d47c66e04e8a4b07989cb8e1af8d5531 : SUCCESS in 1h 45m 07s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/fdeba44a9b864763817b4b693faf644c : SUCCESS in 1h 44m 34s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/18f2c8b260a84567b7005e8cd8dd2196 : FAILURE in 1h 15m 37s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/264d998397dc49bfa5c630b6992418df : TIMED_OUT in 2h 35m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e2e814885e6849f58d9a7af3850a52fc : SUCCESS in 1h 03m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea1624290a354f4eb2e7c9ac852afa3f : SUCCESS in 2h 02m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/242f86def0b64cf687aed427910d8480 : SUCCESS in 1h 08m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e35a3bf85c4b41569ed77807ef93a8be : SUCCESS in 55m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b1116242ea9e4a5e99f9fe33314c0b37 : SUCCESS in 57m 14s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7594eb801864515ae2b23d90c45edfe : SUCCESS in 1h 05m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/8b94836ad07242f3ac9b51357ff006a9 : SUCCESS in 1h 06m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/6a602a49583c45ad8fadf9b61b5037d4 : SUCCESS in 1h 10m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f39747fe173740678a0a5b785d73498e : SUCCESS in 58m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/10eafb26fe714db0b09d4b7fda48f0b2 : SUCCESS in 1h 00m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/75ea987365d94698825f65579a58692c : SUCCESS in 1h 04m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e843699ad054647a1adede53041922a : SUCCESS in 2h 02m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f36c4b985272edfe5d4a1c407dc9f9c40ee92ba7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-30 18:54:05.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"d8d39eb2bf968aaac1af84c414ddb70315de422b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:30:29.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03b895959c04e958e600c3f9346d7f9 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c7b0f54f07452f86a2935bcc9a9982 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/926f6dd169d74a5d9917299189c32761 : SUCCESS in 4m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec0021438111471a87decc0837415f7c : SUCCESS in 44m 13s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/19699132a2f64f128c7b85f7c0c2888a : SUCCESS in 43m 57s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/e236d93de1a5432bb889710d89e5b2bf : SUCCESS in 41m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8d62575ff4146aa9d88d00a6097e930 : SUCCESS in 1h 41m 35s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/aa493aad785a45509eca79b3277a597b : SUCCESS in 1h 30m 48s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/5907410a9ff54e5690557e6be4060947 : SUCCESS in 1h 32m 44s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/821ccbcfe6f347d3b602bc816f395114 : TIMED_OUT in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b2e91a3a75944090a4bc5876a9b733dd : FAILURE in 26m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/53374304d6544f0a9b48441e9b5648a0 : SUCCESS in 1h 52m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec6679357ee84f65bf3b0ff2aa0dd37d : SUCCESS in 1h 01m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b03eec397a884d49b22395d25b7deabe : SUCCESS in 1h 01m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/aa82fd7ae665421e91e65530d4be7a89 : SUCCESS in 1h 02m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f8d5a7ddd0034b9dba42ee0f49649c05 : SUCCESS in 56m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/fc75ff37fc3948fdb27f423c98a91231 : SUCCESS in 52m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/43a12b0f006c423391690e8fb59e7286 : SUCCESS in 53m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b8fcb86232fc4ba7aae69c268058b7aa : SUCCESS in 51m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/5b40bdc389234b129e04e52cbad81258 : SUCCESS in 59m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/d9da8adaf76d4fdebf431ea93d8918ad : SUCCESS in 59m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7774107612d94395b6f55689900a4800 : SUCCESS in 1h 58m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3a01c9cef1204a8d504c450cd5d6e4c73429817d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-13 14:08:19.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)\n\nOne minor issue with a comment. Otherwise LGTM","accounts_in_message":[],"_revision_number":10},{"id":"37cf6f0aa10797627044db3bfde092cdee9ecfd1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-13 15:28:16.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"d112d08b9a4fd96c6084a78be6b87a98432a1786","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-13 15:38:11.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"914d7d45ce91a516ed2b46dbbd334916a1dfcabd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 16:56:21.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c013a3b7d3448e84117ba6ebc5af67 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0eefbde33ec4c809f5d9a34a10d9fda : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f5272dcfdea4dcf8871983f9b9890ed : SUCCESS in 4m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8880c81b0dcb4c24b5a15d19a09a61b6 : SUCCESS in 42m 09s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/fb5a15da57954a7fa0e46dea53f4d149 : SUCCESS in 48m 55s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/1f4d34ce6b8f43a0a9e0883a28b693af : SUCCESS in 36m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ef3d868649d4605819e4faebd3931fa : SUCCESS in 57m 06s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c1c3e14d4424484bbfa251672b8c14e8 : SUCCESS in 1h 14m 21s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/145ce5b3a39c43f4a49d9054fdf9a3bf : SUCCESS in 1h 17m 45s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/20fae5ccea054305bd566a0c5a7c9d36 : SUCCESS in 1h 08m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/95c4d915bbf14d93af185c3deee88289 : SUCCESS in 55m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b78014fea18438498e97951477cf317 : SUCCESS in 1h 19m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/065259fcc9c349b686fe4d25f6d561b4 : SUCCESS in 1h 00m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8eab10fd61d24d2f9bdf9b222ae8321c : SUCCESS in 1h 11m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/982cfae7e0304cd29ecc128936f4c442 : SUCCESS in 1h 01m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4659500e03004ce588eeccb8c9f5fadd : SUCCESS in 56m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/5d32081f1b8940199a2fb8157e20fe4a : SUCCESS in 1h 05m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/ef75d47e3122432fac5e6d5fc5c393b5 : SUCCESS in 1h 05m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ca1dddf134844176977ba0c1fd27f66d : SUCCESS in 53m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/80c2235c282c431c8c7e95db0efcb64f : SUCCESS in 53m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/be03ccad3e4841179ba78be7e1d303f8 : SUCCESS in 1h 16m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd9931822a5443f992f7a40123f23651 : SUCCESS in 1h 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"de2dd5e3bae27ff34d97f81fb391b6005e930369","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-13 18:07:53.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"175245db133ff2dba8cc696e651bf9af5bda7403","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-07 06:57:27.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"7b90a915866e00dc1bfb1e6fadbb08ea0e210837","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-12 15:45:25.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"b194176e99335e4ce0043bc34a5707a00ea26f1a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-13 10:13:17.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"fd49ddac2b4f1c7524f32c3a0647d3195643d700","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-13 10:13:24.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"9b68d81c48b50a1c5f50b62a8fb625502438ce87","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-13 10:13:48.000000000","message":"Patch Set 12:\n\nRebased to fix merge conflict.","accounts_in_message":[],"_revision_number":12},{"id":"146689e214fe249812a2a05cf102bffaba125574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 11:53:49.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77712884e28048c796a3dbddf881f7e3 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/909e9755fd8640dbb5fd274cf932f92c : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33ae5ece4ad5494284b77b97dca19f87 : SUCCESS in 3m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb5863f8e85d42e69d480797402a34ee : SUCCESS in 46m 41s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/dca0627316a749ffbc5a6099518ef908 : SUCCESS in 41m 37s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/ee6cd530803c473ebb8af69ae7229165 : SUCCESS in 35m 12s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/2255bd0199a940529bdeab1a74c21fab : SUCCESS in 33m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2904d2cb908c4f2b88c8e5d4798269e5 : SUCCESS in 1h 28m 16s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5f212d324a3d484884b395100a9a0ad0 : SUCCESS in 1h 21m 15s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/241f9cfcc8cb4f6da1a87396805bceeb : FAILURE in 1h 01m 19s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/78ce8c989ef546a5830721f10667d580 : SUCCESS in 1h 02m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7000c18570b14c91af31eb13b293cee9 : SUCCESS in 56m 15s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/a306a31c095a4e73a08ce1c799fe9661 : SUCCESS in 32m 59s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/bf059811607b4986bc73f1b68041499f : SUCCESS in 32m 37s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/2df6e3e24ad143f48de552b24717531e : SUCCESS in 53m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8a80d9f896aa4baa9d05d873881789c8 : SUCCESS in 1h 03m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/12d0e80054af4581a0fbcfc199683f35 : FAILURE in 45m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8071dc25097b44a58d13277b9099fa71 : SUCCESS in 1h 28m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/660540cd48af41fc82904fad28084276 : SUCCESS in 32m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6da6b0944e80456c98da4ebbbe58bff2 : SUCCESS in 30m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6cbc56e0f4b3496f96a4fb4d92d9a580 : SUCCESS in 32m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/bce05e11f22e41bcad049665ba7e59cb : SUCCESS in 50m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/602aeaa8b90744cdbd77e7c4ef92a651 : SUCCESS in 28m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ab9685861e3f4a7cbf8b3d3235f7866c : SUCCESS in 1h 14m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/bcebc23dd2334a9db1709dca4e149bc2 : SUCCESS in 52m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/3dc96695973b4f5a908ae98a8c30a80f : SUCCESS in 53m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5dee36572d764d4d9818919ec817f877 : SUCCESS in 43m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3c9739096dfd70c6a26f43d333a664126b1dd888","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-13 12:06:59.000000000","message":"Patch Set 12:\n\nrecheck\n\nudp health monitor test failed","accounts_in_message":[],"_revision_number":12},{"id":"fe360c96a5fd863ed701ee05c3f64f13405ac210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 13:50:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/398694c367c441fea34075b1bb90e514 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb06a151e19b4b769fcf3ce2925d6d24 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47c3ecd24b29418abaedc816acce2bf7 : SUCCESS in 3m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22240a24859f4276901ecf87dcaa9589 : SUCCESS in 44m 11s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8a7eadbe97504f2c8e63e49a6fe90f64 : SUCCESS in 52m 55s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c10cb0e91f4648be994dc932b0c35191 : SUCCESS in 45m 33s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/627f9a088fd840b0b97172e797d31867 : SUCCESS in 49m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fb1ad4d376744b1994a026f7b6bb7b6 : SUCCESS in 1h 01m 07s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3e15eb338f4d4260939b2ff39582de30 : SUCCESS in 41m 58s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b1a0fa423298419ab012ba620bf39453 : SUCCESS in 1h 06m 08s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/570e468aac5b4ea3af35ccd30268eaa5 : SUCCESS in 47m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5157a9c80b0a499aa6014744fb37efa6 : SUCCESS in 31m 23s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/ec16310fd3d14601ba556ef57551972d : SUCCESS in 50m 39s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/5932e9f4079b48eb83dd042f05150c80 : SUCCESS in 33m 44s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/d72f943c436148e3aa5f2dfffea37b0e : SUCCESS in 54m 15s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bff72ece94294371b948d3ee72fec5c4 : SUCCESS in 1h 04m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3251f96ff0aa4d9385785b137d9c2041 : FAILURE in 43m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c573da4de6334a7d9250a2af41565147 : SUCCESS in 1h 14m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75f9533ebfad462b804a1e88e8e6df5d : SUCCESS in 59m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e26e456b01b64c3cb41c1439d09237cc : SUCCESS in 30m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b13865c74dc74317b0ef881004567230 : SUCCESS in 31m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1b44295b91f5410796f4982ee2102498 : SUCCESS in 35m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/486980bf32e94c40bbded0e05cd64a9c : SUCCESS in 30m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/99fdbdd083bc427da640cec43a87599e : SUCCESS in 49m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/6a08979ed0484d00b94714add6cc3828 : SUCCESS in 54m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/107aa01e1e924e2f8606ba39d8c4ea5e : SUCCESS in 33m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9788a1003128432683f537567ed8c295 : SUCCESS in 42m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"079e68fe552994deca5b622e3c5cdfc4a8f402ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 13:50:18.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"6a7b24f0c802d0cf70f5a44856cef4ddb8dd15b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:13:36.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b317ccc3ed94211a6735d4b59bec79e : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8d7534a25b244e897efaf1fe9207cf7 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0387204860144ccb8c6afe749b8ecabd : SUCCESS in 4m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e14bb0726f2b41f8a77710c7c7159903 : SUCCESS in 42m 46s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/56021819ae17440a846b0f1a09d1f8c2 : SUCCESS in 45m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/a782a5df75d348d68502517e53d1f9a0 : SUCCESS in 44m 40s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/e232a9ba9a64469b87218dac104c8cbc : SUCCESS in 49m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/694dedc8209640ab8df7a6282ef3c75c : SUCCESS in 39m 17s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5b9a2a04038648e796712639a47d516a : SUCCESS in 1h 11m 04s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a517626ced5543e8a7f640cce9424162 : SUCCESS in 1h 19m 56s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d2ab6986b5da4e1ba5759fa23b5db1fd : SUCCESS in 54m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/13dce8c93a6e440eb71ea7447dd9bb6f : SUCCESS in 42m 02s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/ddaa562dbd314daebe463a4cf3ad225c : SUCCESS in 1h 06m 04s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a82de28a90fc46209244aa28d66e05ad : SUCCESS in 1h 07m 22s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/e9b5f1f36a2e46bbbf6ddd23396ba9c1 : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":12},{"id":"99b693203e090f03b50ad7e6f713f0de12eb5c54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:13:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"0aaf5e8fdc39f3c33f5b0a08181f25d290b49fbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:15:59.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5c152b40204eb2a3089d2540a9dfce : SUCCESS in 1m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82cf2807ad4649cbb34a985a836e1e95 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"7bd2f97221a369f379e15f33a3d3294f220c039e","revisions":{"3775fed326ecaf6a5fc040163317359534203f5d":{"kind":"REWORK","_number":1,"created":"2019-07-31 04:33:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/1"}}},"commit":{"parents":[{"commit":"13c47d9ff6906c95606c72c8ee5b88e770eb6866","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/13c47d9ff6906c95606c72c8ee5b88e770eb6866"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3775fed326ecaf6a5fc040163317359534203f5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3775fed326ecaf6a5fc040163317359534203f5d"}]},"branch":"refs/heads/master"},"50f3e62eb25fdfaee104dc39ebe61dbdbcb60d4f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-31 17:08:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/2"}}},"commit":{"parents":[{"commit":"769b4d40b32f444d9f043dcbb78cdacd2703983e","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/769b4d40b32f444d9f043dcbb78cdacd2703983e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 17:08:19.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/50f3e62eb25fdfaee104dc39ebe61dbdbcb60d4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/50f3e62eb25fdfaee104dc39ebe61dbdbcb60d4f"}]},"branch":"refs/heads/master"},"c06c6d514640d01a68298038a8c96b3fe1c1401a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-09 21:51:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/3"}}},"commit":{"parents":[{"commit":"8e1ed4454681988dc2473dee4a91385f329c7118","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8e1ed4454681988dc2473dee4a91385f329c7118"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-09 21:51:48.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c06c6d514640d01a68298038a8c96b3fe1c1401a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c06c6d514640d01a68298038a8c96b3fe1c1401a"}]},"branch":"refs/heads/master"},"3db347f93f92b84055c2a1e7bf369f743c9a941f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-10 15:57:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/4"}}},"commit":{"parents":[{"commit":"0d0468afc61d926a76a91befd745d2b73a4b3ee5","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0d0468afc61d926a76a91befd745d2b73a4b3ee5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-10 15:57:02.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3db347f93f92b84055c2a1e7bf369f743c9a941f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3db347f93f92b84055c2a1e7bf369f743c9a941f"}]},"branch":"refs/heads/master"},"d91a820197a1d53c391ebd0a16102ce516d5c80e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-08 22:34:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/5"}}},"commit":{"parents":[{"commit":"d7526be7a1eb3204769ca36c73c6c2c75ef1d03e","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d7526be7a1eb3204769ca36c73c6c2c75ef1d03e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 22:34:42.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d91a820197a1d53c391ebd0a16102ce516d5c80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d91a820197a1d53c391ebd0a16102ce516d5c80e"}]},"branch":"refs/heads/master"},"18ed09d5af7a6fd17220627ead119cc4a5b7c9d9":{"kind":"REWORK","_number":6,"created":"2020-06-03 21:16:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/6"}}},"commit":{"parents":[{"commit":"a09bf3c43e6555a62e26a45ac00ea48ad5af075b","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a09bf3c43e6555a62e26a45ac00ea48ad5af075b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-03 21:16:30.000000000","tz":-420},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/18ed09d5af7a6fd17220627ead119cc4a5b7c9d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/18ed09d5af7a6fd17220627ead119cc4a5b7c9d9"}]},"branch":"refs/heads/master"},"f167fadfb71966fab5ef91c0851f43ae91fefb0b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-06-03 21:19:46.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/7"}}},"commit":{"parents":[{"commit":"2e704169458c7fb08d153b5574aab289edfc51a9","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2e704169458c7fb08d153b5574aab289edfc51a9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-03 21:19:46.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f167fadfb71966fab5ef91c0851f43ae91fefb0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f167fadfb71966fab5ef91c0851f43ae91fefb0b"}]},"branch":"refs/heads/master"},"adc55600654fb1abbdccc64d463a13889098bf0f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-06-11 17:42:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/8"}}},"commit":{"parents":[{"commit":"6d3b6542722ce575280d8d08d2ac541a00ee3ddc","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6d3b6542722ce575280d8d08d2ac541a00ee3ddc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 17:42:48.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/adc55600654fb1abbdccc64d463a13889098bf0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/adc55600654fb1abbdccc64d463a13889098bf0f"}]},"branch":"refs/heads/master"},"55c225e206ffc3675b810fbd6bf21b54c1d3ff72":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-06-24 14:08:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/9"}}},"commit":{"parents":[{"commit":"77535cce31ebfebf983612feb13dcad40c8b1a31","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/77535cce31ebfebf983612feb13dcad40c8b1a31"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-24 14:08:27.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/55c225e206ffc3675b810fbd6bf21b54c1d3ff72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/55c225e206ffc3675b810fbd6bf21b54c1d3ff72"}]},"branch":"refs/heads/master"},"76451d9c9a5696a1f2495c526c36765542305d17":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-06-30 18:54:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/10"}}},"commit":{"parents":[{"commit":"209d835988f49a9741b0cd87f38e22645725cc98","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/209d835988f49a9741b0cd87f38e22645725cc98"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-30 18:54:05.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/76451d9c9a5696a1f2495c526c36765542305d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/76451d9c9a5696a1f2495c526c36765542305d17"}]},"branch":"refs/heads/master"},"7b7bfdf2bef382f5cb38714c64b6bdc4cd946531":{"kind":"REWORK","_number":11,"created":"2020-07-13 15:28:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/673726/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/11"}}},"commit":{"parents":[{"commit":"209d835988f49a9741b0cd87f38e22645725cc98","subject":"Add an optional tenant flow log offload test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/209d835988f49a9741b0cd87f38e22645725cc98"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-07-13 15:28:14.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7b7bfdf2bef382f5cb38714c64b6bdc4cd946531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7b7bfdf2bef382f5cb38714c64b6bdc4cd946531"}]},"branch":"refs/heads/master"},"7bd2f97221a369f379e15f33a3d3294f220c039e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-08-13 10:13:17.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/673726/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/673726/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/673726/12"}}},"commit":{"parents":[{"commit":"bf6cc4095644f782f3731ae158e94b5a2e0a60e1","subject":"Merge \"Add an optional tenant flow log offload test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf6cc4095644f782f3731ae158e94b5a2e0a60e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-31 04:32:04.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-13 10:13:17.000000000","tz":0},"subject":"Add an optional amphora admin log offloading test","message":"Add an optional amphora admin log offloading test\n\nThis patch adds an optional amphora admin log offloading scenario\ntest.\n\nChange-Id: I90e3ed23cdd2aa1125de18dc332c30a9d9cca0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7bd2f97221a369f379e15f33a3d3294f220c039e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7bd2f97221a369f379e15f33a3d3294f220c039e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
