)]}'
{"id":"openstack%2Foctavia-tempest-plugin~674009","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I75c648026d43d533b21367f3adf77014b07e89a3","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I75c648026d43d533b21367f3adf77014b07e89a3","subject":"Add missing keystoneauth1 to requirements.txt","status":"MERGED","created":"2019-08-01 11:31:53.000000000","updated":"2019-08-03 04:46:23.000000000","submitted":"2019-08-03 04:46:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674009-1564807583590-e5490da1","meta_rev_id":"631ede487b212f3e52a586f20a9735c69301373f","_number":674009,"virtual_id_number":674009,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-08-01 14:24:38.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-03 04:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-03 00:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-01 16:15:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-08-03 00:18:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 16:15:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-03 00:18:49.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-03 04:46:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"673fadab56d1a5d890a767bd6b4cc6cea2552e4f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-01 11:31:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89ab3e53f887c5b5cc6f0988beed23360f944844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 14:13:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/09/674009/1/check/requirements-check/1c7c522/ : SUCCESS in 2m 10s\n- openstack-tox-docs https://logs.opendev.org/09/674009/1/check/openstack-tox-docs/b15d1b0/html/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://logs.opendev.org/09/674009/1/check/openstack-tox-pep8/3bb76e3/ : SUCCESS in 4m 28s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-api/0834203/ : SUCCESS in 47m 10s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api/65979e9/ : SUCCESS in 53m 20s\n- octavia-v2-dsvm-noop-api-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-api-stable-stein/ef475b9/ : SUCCESS in 46m 24s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/0fbb26d/ : SUCCESS in 47m 36s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/488faf2/ : SUCCESS in 48m 08s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario/e1cb0c1/ : SUCCESS in 1h 44m 02s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario/7b38505/ : FAILURE in 1h 54m 11s\n- octavia-v2-dsvm-scenario-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-stable-stein/dd95612/ : SUCCESS in 1h 33m 47s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/aa3ac84/ : SUCCESS in 1h 49m 28s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/86d6d70/ : SUCCESS in 1h 30m 09s\n- octavia-v2-dsvm-scenario-centos-7 https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-centos-7/a04d1fa/ : FAILURE in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2cea097/ : SUCCESS in 1h 48m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-two-node/3a446be/ : FAILURE in 1h 49m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-two-node/9b1e5f3/ : SUCCESS in 1h 59m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/f938306/ : FAILURE in 1h 40m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/748c2d1/ : FAILURE in 1h 18m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/4531bb0/ : FAILURE in 1h 07m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/c5e37bb/ : FAILURE in 1h 48m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/7cb9b6e/ : SUCCESS in 1h 04m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/f274b85/ : SUCCESS in 1h 05m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican/db53bb8/ : SUCCESS in 54m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/3dc380a/ : SUCCESS in 53m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/329f835/ : SUCCESS in 1h 07m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/cebd4f4/ : SUCCESS in 57m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-spare-pool/aecd970/ : SUCCESS in 1h 06m 55s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool/949dbef/ : SUCCESS in 56m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-spare-pool-stable-stein/204ab87/ : SUCCESS in 1h 04m 23s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/b41660c/ : SUCCESS in 52m 35s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/5cf5ae8/ : SUCCESS in 51m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"870c9d3c4273b283d32baeecfa92be31551a62c6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-01 14:24:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b193bfdf2933d4ac89ca7ca56defb294f43e423c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-01 16:15:44.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"9edc0ee58b72e55b44494ac06288d4e8efdacc33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 18:04:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/09/674009/1/check/requirements-check/9ea3fc2/ : SUCCESS in 2m 25s\n- openstack-tox-docs https://logs.opendev.org/09/674009/1/check/openstack-tox-docs/97808d1/html/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://logs.opendev.org/09/674009/1/check/openstack-tox-pep8/3ccaa9b/ : SUCCESS in 3m 32s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-api/9abe54f/ : SUCCESS in 45m 27s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api/f64481b/ : SUCCESS in 50m 55s\n- octavia-v2-dsvm-noop-api-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-api-stable-stein/be85eae/ : SUCCESS in 49m 58s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/c320c6e/ : SUCCESS in 56m 01s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/09f4688/ : SUCCESS in 45m 01s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario/44d0b19/ : SUCCESS in 1h 40m 44s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario/a4094e7/ : SUCCESS in 1h 46m 01s\n- octavia-v2-dsvm-scenario-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-stable-stein/239bbbe/ : SUCCESS in 1h 40m 42s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/2240f55/ : SUCCESS in 1h 34m 18s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/2aaab93/ : FAILURE in 1h 38m 02s\n- octavia-v2-dsvm-scenario-centos-7 https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-centos-7/2e916fb/ : FAILURE in 2h 31m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/696c14d/ : SUCCESS in 1h 48m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-two-node/6b44a00/ : SUCCESS in 1h 58m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-two-node/f2bad69/ : SUCCESS in 2h 07m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/3588c4f/ : FAILURE in 1h 21m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/f921d63/ : FAILURE in 1h 04m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/32068d9/ : POST_FAILURE in 1h 18m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/e53433c/ : FAILURE in 1h 38m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/ae528e8/ : SUCCESS in 1h 04m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/1fcac6a/ : FAILURE in 1h 29m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican/6a6fedb/ : SUCCESS in 1h 03m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/dc9416c/ : SUCCESS in 51m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/6fef074/ : SUCCESS in 1h 00m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/65bbeb6/ : SUCCESS in 59m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-spare-pool/580f2df/ : FAILURE in 37m 00s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool/b14aac6/ : SUCCESS in 1h 04m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-spare-pool-stable-stein/bbf0a21/ : SUCCESS in 57m 13s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/5cdc177/ : SUCCESS in 51m 57s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/9eeddb8/ : FAILURE in 1h 06m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d81c6c17de8dbbf63c145f8affa8068f7965cb5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:18:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13d25c0c63c689a241d9282a4204ff8459f326f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 03:16:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/09/674009/1/check/requirements-check/66c2cb4/ : SUCCESS in 2m 18s\n- openstack-tox-docs https://logs.opendev.org/09/674009/1/check/openstack-tox-docs/660995d/html/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://logs.opendev.org/09/674009/1/check/openstack-tox-pep8/5e409fb/ : SUCCESS in 3m 05s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-api/80f6aea/ : SUCCESS in 40m 11s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api/c161cf7/ : SUCCESS in 39m 24s\n- octavia-v2-dsvm-noop-api-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-api-stable-stein/d202330/ : SUCCESS in 39m 21s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api-stable-rocky/2af36cd/ : SUCCESS in 38m 15s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-noop-py2-api-stable-queens/561116e/ : SUCCESS in 42m 49s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario/9691b26/ : SUCCESS in 1h 40m 49s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario/959336f/ : SUCCESS in 1h 43m 33s\n- octavia-v2-dsvm-scenario-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-stable-stein/fa3d911/ : SUCCESS in 1h 30m 39s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-stable-rocky/1e6202c/ : SUCCESS in 1h 22m 34s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-stable-queens/8304b56/ : SUCCESS in 1h 26m 31s\n- octavia-v2-dsvm-scenario-centos-7 https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-centos-7/0bda112/ : FAILURE in 2h 28m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3d612e8/ : SUCCESS in 1h 26m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-scenario-two-node/c2d04a0/ : FAILURE in 1h 49m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-scenario-two-node/8bf53f2/ : SUCCESS in 1h 05m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/5302353/ : FAILURE in 1h 09m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/55f615d/ : FAILURE in 1h 17m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/9651d50/ : FAILURE in 1h 06m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein/52ffd2f/ : FAILURE in 1h 16m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky/150761a/ : FAILURE in 1h 34m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens/60c8ef2/ : SUCCESS in 1h 01m 22s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican/d99ab43/ : SUCCESS in 49m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-stein/cd568aa/ : SUCCESS in 50m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-rocky/bf3eca0/ : SUCCESS in 50m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-tls-barbican-stable-queens/5ca587f/ : SUCCESS in 50m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-spare-pool/bedd635/ : FAILURE in 49m 37s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool/d8aaca5/ : SUCCESS in 47m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-spare-pool-stable-stein/6a0b56d/ : SUCCESS in 53m 49s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool-stable-rocky/2c1fff1/ : SUCCESS in 48m 58s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://logs.opendev.org/09/674009/1/check/octavia-v2-dsvm-py2-spare-pool-stable-queens/6378fd2/ : SUCCESS in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd8b24a4925ba30784249c5a30eb0d85d4c97f4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 03:16:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7492b7183658db067b729322397d9b4b3af90402","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 04:46:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://logs.opendev.org/09/674009/1/gate/requirements-check/41587b2/ : SUCCESS in 2m 13s\n- openstack-tox-docs https://logs.opendev.org/09/674009/1/gate/openstack-tox-docs/d3b152e/html/ : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://logs.opendev.org/09/674009/1/gate/openstack-tox-pep8/36ca56e/ : SUCCESS in 2m 11s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-noop-api/b9d54b9/ : SUCCESS in 40m 50s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-noop-py2-api/fc248c6/ : SUCCESS in 42m 44s\n- octavia-v2-dsvm-noop-api-stable-stein https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-noop-api-stable-stein/b53332e/ : SUCCESS in 39m 12s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-noop-py2-api-stable-rocky/d26c5cb/ : SUCCESS in 38m 59s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-noop-py2-api-stable-queens/2063809/ : SUCCESS in 41m 45s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-scenario/c593e5f/ : SUCCESS in 1h 26m 19s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-py2-scenario/da5589c/ : SUCCESS in 1h 29m 05s\n- octavia-v2-dsvm-scenario-stable-stein https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-scenario-stable-stein/0b9786e/ : SUCCESS in 1h 14m 59s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-py2-scenario-stable-rocky/d2f946e/ : SUCCESS in 1h 17m 24s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://logs.opendev.org/09/674009/1/gate/octavia-v2-dsvm-py2-scenario-stable-queens/569fda1/ : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"631ede487b212f3e52a586f20a9735c69301373f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 04:46:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42e54aadd9cd2f83ac278741002ec6fe736087cd","revisions":{"42e54aadd9cd2f83ac278741002ec6fe736087cd":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:31:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/09/674009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/09/674009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/09/674009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/09/674009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/09/674009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/09/674009/1"}}},"commit":{"parents":[{"commit":"65e7c5fcd47d04059cb2203776f84255d9177ea2","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/65e7c5fcd47d04059cb2203776f84255d9177ea2"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-01 11:28:48.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-01 11:31:10.000000000","tz":120},"subject":"Add missing keystoneauth1 to requirements.txt","message":"Add missing keystoneauth1 to requirements.txt\n\nThe keystoneauth1 library is imported by the Barbican client manager but\nwas missing in the requirements.txt file.\n\nChange-Id: I75c648026d43d533b21367f3adf77014b07e89a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/42e54aadd9cd2f83ac278741002ec6fe736087cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/42e54aadd9cd2f83ac278741002ec6fe736087cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
