)]}'
{"id":"openstack%2Foctavia-tempest-plugin~681084","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ifc557949f83035a0636ac11793bef62ec07cdbd7","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ifc557949f83035a0636ac11793bef62ec07cdbd7","subject":"DNM: Testing netconsole","status":"ABANDONED","created":"2019-09-09 17:56:44.000000000","updated":"2019-09-13 16:34:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92fb96afcca1c13c55f696515835eca4c1dab008","_number":681084,"virtual_id_number":681084,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-09-09 20:43:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-09 20:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce80853c8ac92268161b61be48148dedb8583e7c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-09 17:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11fb66492eaa9226db0257e2ba7a4dcbfce2b591","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-09 17:59:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"26ebf3931d63248756a870cb1035f42156de936f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 18:07:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5f9987af404b7fb2fcf6ffd5f10a18 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096884ac4199441ead3a2d9c5fc786af : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ce23cf191eee47fc61372b71eaf8f50d14dddbc0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-09 18:11:44.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"63bc070fe30ae6f453c8a07c8eae5c5b5343e23e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 18:35:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa49088fc53458780d904eaf947ef87 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e741327282744884ac60189888e014cb : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":3},{"id":"5f2d0a598735f07c7dce585465e1a717ac8206a1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-09 18:57:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2f5ba548fc5acad8a0f49e0fec6c30d1e6af2618","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 19:09:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17f7b13356749ab94da8304d1a966e1 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce49e292876484dbb7a4d4eb2dc63d1 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":4},{"id":"370ba2e78e57a307d44994e4d286e2c782400b9c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-09 20:31:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0444edcfdd5d74e86061bcac35740a3ced5a84bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 20:43:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86371ff226fb4c83bf6d88b65bb69682 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d952883563944f2a8f46aa817a87971 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":5},{"id":"92fb96afcca1c13c55f696515835eca4c1dab008","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-13 16:34:44.000000000","message":"Abandoned\n\nWorks!","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5cec536380ead36383bd6a7e7b744f2fa70fae6e","revisions":{"71ac80ffef00d03949e493e290da46628ae163bb":{"kind":"REWORK","_number":1,"created":"2019-09-09 17:56:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/681084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/84/681084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/1"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Zuul","email":"zuul@review.opendev.org","date":"2019-08-16 19:26:24.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-09 17:56:43.000000000","tz":-420},"subject":"DNM: Testing netconsole","message":"DNM: Testing netconsole\n\nPlease don\u0027t make patches depend on this patch.\n\nChange-Id: Ifc557949f83035a0636ac11793bef62ec07cdbd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/71ac80ffef00d03949e493e290da46628ae163bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/71ac80ffef00d03949e493e290da46628ae163bb"}]},"branch":"refs/heads/master"},"f3efab81974006c9886948becda048d6fb94b023":{"kind":"REWORK","_number":2,"created":"2019-09-09 17:59:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/681084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/84/681084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/2"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Zuul","email":"zuul@review.opendev.org","date":"2019-08-16 19:26:24.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-09 17:59:22.000000000","tz":-420},"subject":"DNM: Testing netconsole","message":"DNM: Testing netconsole\n\nPlease don\u0027t make patches depend on this patch.\n\nChange-Id: Ifc557949f83035a0636ac11793bef62ec07cdbd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f3efab81974006c9886948becda048d6fb94b023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f3efab81974006c9886948becda048d6fb94b023"}]},"branch":"refs/heads/master"},"e658770a5de60a560e914e0510214c0108993772":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-09 18:11:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/681084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/84/681084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/3"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Zuul","email":"zuul@review.opendev.org","date":"2019-08-16 19:26:24.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-09 18:10:58.000000000","tz":-420},"subject":"DNM: Testing netconsole","message":"DNM: Testing netconsole\n\nPlease don\u0027t make patches depend on this patch.\n\nDepends-On: https://review.opendev.org/680901\nChange-Id: Ifc557949f83035a0636ac11793bef62ec07cdbd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e658770a5de60a560e914e0510214c0108993772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e658770a5de60a560e914e0510214c0108993772"}]},"branch":"refs/heads/master"},"8ab3b687e56816839f1b37619331b6a4163b7508":{"kind":"REWORK","_number":4,"created":"2019-09-09 18:57:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/681084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/84/681084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/4"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Zuul","email":"zuul@review.opendev.org","date":"2019-08-16 19:26:24.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-09 18:57:22.000000000","tz":-420},"subject":"DNM: Testing netconsole","message":"DNM: Testing netconsole\n\nPlease don\u0027t make patches depend on this patch.\n\nDepends-On: https://review.opendev.org/680901\nChange-Id: Ifc557949f83035a0636ac11793bef62ec07cdbd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8ab3b687e56816839f1b37619331b6a4163b7508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8ab3b687e56816839f1b37619331b6a4163b7508"}]},"branch":"refs/heads/master"},"5cec536380ead36383bd6a7e7b744f2fa70fae6e":{"kind":"REWORK","_number":5,"created":"2019-09-09 20:31:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/681084/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/84/681084/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/84/681084/5"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Zuul","email":"zuul@review.opendev.org","date":"2019-08-16 19:26:24.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-09 20:31:12.000000000","tz":-420},"subject":"DNM: Testing netconsole","message":"DNM: Testing netconsole\n\nPlease don\u0027t make patches depend on this patch.\n\nDepends-On: https://review.opendev.org/680901\nChange-Id: Ifc557949f83035a0636ac11793bef62ec07cdbd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5cec536380ead36383bd6a7e7b744f2fa70fae6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5cec536380ead36383bd6a7e7b744f2fa70fae6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
