)]}'
{"id":"openstack%2Foctavia-tempest-plugin~682726","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ieb74ef2276632f04d6324f835b164df3114b2613","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"ipv6-only-deployment-and-testing","hashtags":[],"change_id":"Ieb74ef2276632f04d6324f835b164df3114b2613","subject":"[train][goal] Define new \u0027octavia-v2-dsvm-scenario-ipv6-only\u0027 job","status":"MERGED","created":"2019-09-17 19:37:12.000000000","updated":"2019-10-30 07:16:31.000000000","submitted":"2019-10-30 07:12:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682726-1572419555370-e7a1f455","meta_rev_id":"8c863cb9208f87ab16e989121a62ceb5deabc6e8","_number":682726,"virtual_id_number":682726,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 07:12:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2019-10-30 02:18:14.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-10-29 21:03:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-29 21:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-10-29 21:03:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 21:01:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-10-29 21:03:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-30 07:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e34cc014a4cd98477b5e315ee0f3b718067b7e89","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 19:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6740822649b4b39f9a65fcbdeba6928295680be","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 19:37:23.000000000","message":"Topic set to ipv6-only-deployment-and-testing","accounts_in_message":[],"_revision_number":1},{"id":"8082d51307ebbd028c2b73776f78cd8ef5653051","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 19:37:26.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  found undefined alias \u0027base_irrelevant_files\u0027\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 322, column 23","accounts_in_message":[],"_revision_number":1},{"id":"bdd75b3262f711ee4729dde8a97014316865a9bb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 19:41:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4d21136537fab466d5fac96054b0e83d8a10957f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 19:41:45.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error\nwas:\n\n  The key \"\u003c\u003c\" appears more than once; duplicate keys are not permitted.\n\nThe error appears in the following stanza:\n\n  \u003c\u003c: *base_vars\n        \u003c\u003c: *base_vars_2\n        \u003c\u003c: *base_vars_3\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 319, column 7","accounts_in_message":[],"_revision_number":2},{"id":"db93cdd650fe9fdc855a2868375500cb2eb1024a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 19:44:42.000000000","message":"Patch Set 2: Workflow-1\n\nMore work to do here.","accounts_in_message":[],"_revision_number":2},{"id":"c8806677ec0946f85e0e4411f296d5a238e62b6b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 19:47:47.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Zuul encountered a syntax error while parsing its configuration in\n \u003e the\n \u003e repo openstack/octavia-tempest-plugin on branch master.  The error\n \u003e was:\n \u003e \n \u003e The key \"\u003c\u003c\" appears more than once; duplicate keys are not\n \u003e permitted.\n \u003e \n \u003e The error appears in the following stanza:\n \u003e \n \u003e \u003c\u003c: *base_vars\n \u003e \u003c\u003c: *base_vars_2\n \u003e \u003c\u003c: *base_vars_3\n \u003e \n \u003e in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line\n \u003e 319, column 7\n\nwe can copy paste the *base_vars_2 and *base_vars_3.","accounts_in_message":[],"_revision_number":2},{"id":"8f9642dc77166dfed930a13b4894e80632fb92ce","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 20:44:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e0e6e3aad2d57dd92fa238b0748f89f871e5b5e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:45:04.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  while constructing a mapping\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 124, column 9\nexpected a mapping for merging, but found sequence\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 38, column 24","accounts_in_message":[],"_revision_number":3},{"id":"c419ce2d8ffae18380373f928de0582c5ea225e2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 20:48:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c6ad45b486edc1616fb0ed3d35fcf8ec6eb893f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:48:21.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 119, column 5\nexpected \u003cblock end\u003e, but found \u0027\u003cblock mapping start\u003e\u0027\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 125, column 9","accounts_in_message":[],"_revision_number":4},{"id":"ce589de30d0727ad57a4f9cb81f24c6a9a9e990b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 21:14:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2f5105afeeff18f5ad11e8af5bbbb907997f2d9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:14:38.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 119, column 5\nexpected \u003cblock end\u003e, but found \u0027\u003cblock mapping start\u003e\u0027\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 125, column 9","accounts_in_message":[],"_revision_number":5},{"id":"3ad9155f7781b243ff39411957c5fb4a5aeec909","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 21:16:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"35b95f59d89b72be7f06da17ca1a6f5a5b996c4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:16:25.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  while scanning a simple key\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 125, column 9\ncould not find expected \u0027:\u0027\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 126, column 5","accounts_in_message":[],"_revision_number":6},{"id":"d5cb3d3441fc32b9b776def69895ea6481748577","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 21:23:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2fcd96b896970bf08233eff5b1ecd93474e3aeac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:23:25.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error\nwas:\n\n  The key \"\u003c\u003c\" appears more than once; duplicate keys are not permitted.\n\nThe error appears in the following stanza:\n\n  name: octavia-dsvm-live-base-ipv6-only\n      parent: devstack-tempest-ipv6\n      timeout: 9000\n      pre-run: playbooks/Octavia-DSVM/pre.yaml\n      \u003c\u003c: [*base_required_projects, *live_base_required_projects]\n      irrelevant-files: *base_irrelevant_files\n      \u003c\u003c: [*base_vars, *live_base_vars]\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 119, column 5","accounts_in_message":[],"_revision_number":7},{"id":"36cf5d6943f5ca7dccb6697193a93e5da6ea2d42","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 21:24:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bf525f0dcd1203f1165c1485612970c3b6882216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:24:59.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  while constructing a mapping\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 119, column 5\nexpected a mapping for merging, but found sequence\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 38, column 24","accounts_in_message":[],"_revision_number":8},{"id":"16888230d3cddce889f06b19811c1733973eea11","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 21:29:11.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"95f157777724fbd2735339704425c23da67754fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:29:27.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 124, column 26","accounts_in_message":[],"_revision_number":9},{"id":"d50ebb8515c04cfdf6ed4ed3bc085910a9b1f4f3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 22:06:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"25767c7d1b71dea1c14b047ed552cd54d6eca8be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 22:19:14.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"09131d00b185a17cdc2afa613ebbe9a6ece95744","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 22:20:48.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"a3af6d9fb7ab1a9448fcc10f8fd43b0627bb1e8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 23:30:02.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"db9fa4d650a875c4a4d05d7cd177ca274aeb78c7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 23:30:11.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"3596c0ffa15e6593d992f7cf71e368110964e1be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 00:16:36.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"2d562740cef7f1cd5d02f13691e013ae90229c3e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 00:20:42.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"70ec81078572631050f017cafaf9034dff78b1a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 01:42:55.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"c0adabf4c68e6ec4c32d7bd702da66622d3ae333","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 01:54:33.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"ba0e700f4504188556c344b81e452d6558cb83e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 03:03:32.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"3788c02f4b95d08ffc333b0d6ffaae3c62dfd66a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 03:03:52.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"5ded480a7e3da6b490b68f976baabb02bb05bbb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 03:52:46.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"c7996a173e54d33f2b38ea609fd525b23bd85ee7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 03:54:02.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"4ad12eaf687a4b336f65f0b5a388116071b282e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 04:18:14.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"51d6b5a6616414ff53dc2b0716ee8af7a39f1e3a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 04:19:30.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d92b3bfa4b486fc26612f3d97d051623a8376418","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 04:29:21.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"23a9221783a00943ac19aa931662f99e49d5b359","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:07:25.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"bad59c75079d4ec10d6f148ec8b81e2c558f8cd0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 05:08:01.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"d24fe4aee13bb05bb886472b2a23c5c49edec949","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:53:05.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"f727e1f4a55c7543017abdac2ff2b0e8025ce358","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 05:54:01.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"77e5f3df046f5622f300ad8b391d4b0a7928e5b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 06:40:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/468d28f6768a4d44aba96bbb21964f4e : SUCCESS in 44m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a2b034f0a43b4e16a39358ada1edaa5d : FAILURE in 32m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"911340028a2da6d474567434462d68813d8350d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 17:37:08.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0038cbe5e01a67d9932cae5404d5fa0d172ab4d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:32:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a0e1febc2f8468da21aa309be49595a : SUCCESS in 2m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/35e1e95461bf42acae8094f517022a47 : SUCCESS in 2m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/583b8971a73c4903ae251a42cb1224be : SUCCESS in 48m 16s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/890797a32d544ab2af88461001297700 : FAILURE in 41m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"805feb6874a1389b017bd9b089caaa25fd950e60","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 18:35:21.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"bc435bc811af2e9f41a5e5a8cdf22f233e0d06a1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 19:27:19.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"be6034612a5e6e638f9b4d6a9937dec7ead15057","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:27:22.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"f03e5e6ef64720ff70752abdf1b66c8e17ffbe40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:08:43.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/32041225de8f4400ac454425378071fa : FAILURE in 29m 27s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1ddd70ee6f2140849a0201fd92680977 : FAILURE in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"26703c674eba6bac3efae5c6caba2c9890e8c79a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 21:14:02.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"60700275f1a20b457ae66e25b96d5045fb6a5527","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 23:13:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72e2dccebdb342b3a9b266eeafb800b7 : SUCCESS in 1h 01m 20s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ebd7cf40a9e4adaacc805d84cd0c30d : FAILURE in 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d0722c712fd2c178699e346a04aacdbe72e2c064","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 23:55:58.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c41b03b322c504781bddc139b06b57403cf2db14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:41:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ba0cd5fb02944ff8851e01c46da4957 : SUCCESS in 43m 15s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d57d301d843b4b84a8408ff2cdee81ef : FAILURE in 25m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b4df12dcb95d2eba8f1ab4f4385297f312e906b6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 01:44:16.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"4e3de2b47b2af7c287a081fdc5260d0385636b46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 02:45:54.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/152144ed10a947b5abbe70d05eb7564e : SUCCESS in 53m 53s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ee5c6cc1ced24e4a9b88767791b87f52 : FAILURE in 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5ebc9d71aa75982cc9eaac91cc2dac2167c38530","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 04:58:18.000000000","message":"Patch Set 11:\n\nrecehck","accounts_in_message":[],"_revision_number":11},{"id":"9ba5760e086ea81426b2ccbd60cce60f23c4097c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 06:39:40.000000000","message":"Patch Set 11:\n\nrecheck\nNo job in zuul yet again","accounts_in_message":[],"_revision_number":11},{"id":"ca872c365a46773c952a403be6afb1d375b27ae5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:14:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0ffeb25c301b42848a5612d7b9c26e0f : FAILURE in 8m 55s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2e93ea1ae5eb478cbd6565a2a9473d8e : FAILURE in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5c874672fed758473a2df855051902aa1bbff7e5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 13:07:52.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"d78d8ec688fa6dff2c072ac0957abf2541185434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:20:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8e08ac07387407aba7dffdcd8c5af87 : SUCCESS in 47m 10s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ab6bb7aaaab4d209043a5f97e242ea0 : TIMED_OUT in 2h 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"77f1a8c4d27873bdee825e5474063680b34dce43","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 04:07:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d8e07cdffc9ea0d63cc6ddc7c6d6b1e788913dfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:18:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752f10d3f658450bae50aed5090e8570 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc386120a3b946ecbb77fd12de8355df : SUCCESS in 4m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/310504565e2a46a59b47876c494329cf : SUCCESS in 45m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/940c551488844027a096bc3d47019520 : SUCCESS in 44m 33s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/3a1aa604baf143759c0b6ba3e4c98169 : SUCCESS in 40m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/f6e7e92399b4458d9a5ea0ccbe7e237d : SUCCESS in 51m 30s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/1c0257a8decb4623b08c106331d52110 : SUCCESS in 49m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e47d0808d1194c23a4fc6d974a55d733 : SUCCESS in 1h 40m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0838ff1ec43543bf80db1efbfa6ba3a6 : SUCCESS in 1h 48m 17s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ca243839f1334438805e2fc9e9d0747e : SUCCESS in 2h 00m 48s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/4d2e16341ed84233afcb4017090ba929 : SUCCESS in 1h 38m 29s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/9c1166ae74ca4bf4bab73451fa5c1624 : FAILURE in 1h 29m 37s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6489e016e34a448eb8b5f6b63a6b4011 : SUCCESS in 2h 00m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/583688010ecb4d0c82d43023ca340b33 : FAILURE in 19m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcf8dfbfdb9649bdbdd479f2e37e31d1 : SUCCESS in 1h 46m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/539734ae333041319800393d8c1fe80d : SUCCESS in 1h 59m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bac91aa263844cc7b52532a9316565ba : FAILURE in 1h 31m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cf10af404824ff9a235aa68a91af33e : SUCCESS in 1h 04m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5e54d3c507554ff5bd77f95603ba23ec : SUCCESS in 1h 16m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/11c34c3cca544401a2e5db651c1c57dc : FAILURE in 22m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6e42190d58534d2bb7a8fffbd73a39fa : SUCCESS in 1h 19m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/41b47e0fb3a54090b025c32895c0821e : FAILURE in 1h 42m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/6303f12a16044cf8897c34545587d790 : SUCCESS in 1h 07m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa64e4ce1a9849f79598fa2596a12b28 : SUCCESS in 59m 16s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/8b60eee1b1984867b0a031701424ee2b : SUCCESS in 57m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/a453a5df72b34e15a594ea0ce12f411f : SUCCESS in 1h 01m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/d6023d47c81947f9bbb5a792def8bdd2 : SUCCESS in 1h 00m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e85b03b872f4e0e91ded0e16577c060 : SUCCESS in 58m 55s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/3b1648941d6e41af8874b04efce17958 : SUCCESS in 58m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/7f6bb7e4e0a94464a087f02a99a4c264 : SUCCESS in 1h 00m 21s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/26735bba5b0f4c1d8463d033ef7eccc3 : SUCCESS in 1h 02m 16s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/ddac6036e7974fd9adacf72525c4f971 : SUCCESS in 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"360c0ec00b99e70b212e7b99229eed55c981ca0e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 15:29:45.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"a34d898631d726727a0abdafa9511d838cd50550","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:30:30.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db8607d97d49491f90f0e35a7ea22781 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c905fde02c4b70b64c9c1b6bff2700 : SUCCESS in 3m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a5b22f96c6d480eaafbeee3054a50d5 : SUCCESS in 52m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a1882d58667f4291975923cc1f0f60ea : SUCCESS in 58m 34s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/24421334fdf1433f82cb6f682dbf0325 : SUCCESS in 56m 25s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/a299508dfe134a0da84025dd4eb74c7e : SUCCESS in 50m 32s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/3526117bb4cc478783a0a89ebc62fff9 : SUCCESS in 52m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9a2b3a23557412fbec6b7893299437a : SUCCESS in 1h 44m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6c457a820f1a49f5ac51d35dc68e519e : SUCCESS in 1h 38m 00s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/61c93efe25784f0786f27c2ec078960c : SUCCESS in 1h 55m 54s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/2e0aa39fed4846a4834ef82a768b9809 : SUCCESS in 2h 18m 50s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/6ff72c0c44e2439292fb6d4aed2ad699 : SUCCESS in 1h 33m 39s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3779204cf69d42cf8e853951258636ef : FAILURE in 1h 42m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f08d88cbea984ae39c748a63d8c08802 : SUCCESS in 2h 10m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4adebdb618c44abafa11679e570a87c : FAILURE in 2h 00m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6707db9d382f4b20aca9a73c2bb60534 : SUCCESS in 2h 24m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e628b9d20a5241b78b2fed251f5bed08 : SUCCESS in 1h 51m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba2c3860eaae4ff1870feb1030b8f8eb : SUCCESS in 1h 26m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/aa204146d0a447808e839070cf26b70d : SUCCESS in 1h 25m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8f50153da5174501a3060954551e9ce2 : FAILURE in 1h 13m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/739c6bea3f6649ceab9acea69df9e4b4 : SUCCESS in 1h 03m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/f3fb03073f554429a7ea3a867f356048 : SUCCESS in 1h 02m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/f7dc518e8d8e4cad8a7a6aad290010cb : SUCCESS in 1h 17m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c2552ca0eb749deb0576f468e4e1078 : SUCCESS in 51m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/b89633d4b7034094b293ad5a274ce1a4 : SUCCESS in 1h 03m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/b0548f8371a042d18164a3acfaad5fd9 : SUCCESS in 54m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/8e5a2bf31b7747cea02ef04aae95401d : SUCCESS in 58m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/47de21a46120464abfeaad58a80da6c3 : SUCCESS in 59m 19s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/7db7ddb3f56942ce87807b44173375da : SUCCESS in 1h 05m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/2dd54d59c4ed4786abb2560fa1fd2f81 : SUCCESS in 1h 11m 14s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/f79b57e9c9bd4effa0878dabe4ae87de : SUCCESS in 1h 04m 45s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/949b668c29b0496e94cdc79868215094 : FAILURE in 1h 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3298c331963900ee96778a39188f2f11d8d674b6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-07 22:50:04.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"edbe864c3403588fcc8fab42c5e04d7238511e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 01:17:26.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf3904b0223419694141b392eb88a84 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b720dca4c2884e619e0aa9727e8f6c37 : SUCCESS in 2m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d2e2c7f984c4e7789821208c9386b6f : SUCCESS in 39m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b19f4c97d74d4959a956d5f58b9e344d : SUCCESS in 45m 31s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6995cdd06be941ccacf30f733d4c996d : SUCCESS in 49m 00s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/7aa68359f8c742afb6a22914e8342359 : SUCCESS in 47m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/bf112b599f154418b21a409d31c37651 : SUCCESS in 48m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a4d28dfe34848298c4e9d4bb7604a0d : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c30b6b57b3b0433b9083486a71d64417 : FAILURE in 2h 05m 50s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fb5635d32e0b49d4bee8be400fe59f79 : SUCCESS in 1h 38m 36s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/de80c16ef0974a4db83f8670b37a18d4 : SUCCESS in 1h 23m 57s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/3f74ad48364c4839a631d38a90edab80 : FAILURE in 1h 42m 16s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7a4956b2cef34c4aaf0aee45b8487d76 : SUCCESS in 1h 28m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f9a9aa68aaf24541a349956c20738a1f : SUCCESS in 2h 07m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d2a3d7241d941aeb3d19c29f61ebacd : SUCCESS in 1h 34m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ed748bcee1f4a489ed0dd28ea6f5be8 : SUCCESS in 1h 20m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/11c1cfcdeee44bc79e3a5bc67abe945e : FAILURE in 1h 43m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31b1679a3b644223ac629695a09b4730 : SUCCESS in 1h 07m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ea2113725fa949efbb4af0e9148ce41c : SUCCESS in 1h 20m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/afcfc3eed005460fbddf478d5043919a : FAILURE in 1h 16m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/229c06107a6e4481be166b11498614be : SUCCESS in 51m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/250e252b2f234c4b97da814c9526160f : FAILURE in 1h 49m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/a93bff09549f4bbda1bc481cad1eb7ec : SUCCESS in 1h 14m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc261f106ea149ba8fcd85602984157d : SUCCESS in 1h 39m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f212bb80ac004e078930ba23ad247fad : FAILURE in 1h 31m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/532bd33f6f744500b82fc2489a87399f : SUCCESS in 1h 14m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/55af928ea7434b2a8776d77e7521b2f3 : SUCCESS in 49m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/1507df1cb33641709ca9010ef5997697 : SUCCESS in 42m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/122be345952946a695a04d2d4cf984a5 : SUCCESS in 1h 29m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/eeb5e33eb8894bc08d8b0161bac6f67a : SUCCESS in 1h 22m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/08ebf46b50e1485b998093f585e3c68e : SUCCESS in 1h 33m 31s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/772c5736736b48ae98957126a70db8b2 : SUCCESS in 52m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/550dfa2d964345e2a87fa2ac0916a2ea : SUCCESS in 1h 35m 20s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/06bdea398c924e18ad55b1716a295ccc : SUCCESS in 1h 33m 29s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/10dc97d12c7a44b795c62e9a27f0df2b : FAILURE in 1h 26m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0305409f520b417581d472c11d53a3a4 : SUCCESS in 2h 09m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c773afaa15f34d4e7c74b5a279a401695c326c86","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-09 18:29:40.000000000","message":"Patch Set 13:\n\nrecheck\nThe new job passed, but unrelated jobs failed...","accounts_in_message":[],"_revision_number":13},{"id":"c9a70c641a3fca8f7d42a112915cecec09125c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 21:07:21.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8028d95c0e744c7b5cbe5357d449b3a : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b432bb2f8abe407085a8358d61d3dcbe : SUCCESS in 3m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d2ef4c469364e7ea6f3e4b803983f22 : SUCCESS in 47m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4664286effc24bb88d997c1232e247d4 : SUCCESS in 1h 03m 28s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/b9a5cada99894a0c8c7cd5c91bb923dc : SUCCESS in 43m 23s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/324ad09842f64747986e06a00cfc9638 : SUCCESS in 46m 18s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/63f248487144439cbfa33d3e30858006 : SUCCESS in 37m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f4f9abbf17745259fdb31a1abed0144 : SUCCESS in 1h 45m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4b6e445dce124dc0ac0d60a604cd07b7 : SUCCESS in 1h 45m 46s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d4c84e84d3bd4da7bf692f902ba7e0af : SUCCESS in 1h 40m 42s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/c753f26843ce419db8ef8b7a8472cd7e : SUCCESS in 1h 46m 16s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/4a610edc7f1c40f2a7ad426975d4662b : SUCCESS in 1h 25m 28s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae8b696d6b5144e0b64862bd32dc2bf2 : SUCCESS in 1h 40m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4d9c1f760c66441fa6f6ed2a03360c98 : SUCCESS in 2h 06m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/effade60d7be405e8efd9d506bc9fb2f : SUCCESS in 1h 33m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed98e82433f9474d9ce04a5bb7a3bd01 : SUCCESS in 1h 13m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd0c92f1739d43a7ac08c935c845d60c : TIMED_OUT in 2h 34m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fbca787e250483ca8af9171a59e3280 : SUCCESS in 1h 11m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d671aadc70c94eeb996cc95e3e51cbec : SUCCESS in 1h 21m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c6a549065cce478e80f5f2cc4a0dfdcc : FAILURE in 1h 28m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3285caddb0494347b43d42a2e50c8a8f : FAILURE in 1h 42m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/380ac5e3fcad4934988d0031d43d5940 : SUCCESS in 55m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/3ffb4fbb95a2493d95eb5ed6c28185ce : SUCCESS in 57m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc7214204a7d4a9389eee02bed8d3f43 : SUCCESS in 1h 20m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8376844a25c742398061aaac0587e2d2 : SUCCESS in 1h 31m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c681d247856748feafbb673f01cee109 : SUCCESS in 1h 01m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd383849b69d4410b6cb41382ed94cb4 : SUCCESS in 55m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/a2a464a784e54673b7ba4c229c03d3ef : SUCCESS in 51m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/f53546615b314a95ad6528c36635ad1a : SUCCESS in 58m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/eaf9ff58471841468e1c2c5a717eb203 : SUCCESS in 52m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3762ed993749476d827730e96e34a091 : SUCCESS in 56m 15s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/b00b7e49c433474788822e6fcb8d488f : SUCCESS in 1h 00m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/5f0de258823a474a83ddb9683564f6e7 : SUCCESS in 58m 19s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/602059581a994174a509760d0097dc07 : SUCCESS in 1h 03m 11s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/1b71173513cc4cf59505611b2de7de38 : SUCCESS in 1h 03m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9bf632c5c09d4b7f9e9898fc3ad7e807 : SUCCESS in 1h 42m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1a2fc47bcdbdb226ca10ece33dce4cfe78ced8fb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-09 22:11:42.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"0c6b9362bfc4519b012834140ee61aba95172da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 00:26:58.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c943b493c6754ccbbf9772d3634fed48 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad7eed01d61487e89f54a33d540c7cd : SUCCESS in 4m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46978d1fac5449e28b352b8686611709 : SUCCESS in 43m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cc3e6cbf8def4866aa1df8a3188c5332 : SUCCESS in 48m 50s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/64937ecc981348dfbefa6d3f0370871c : SUCCESS in 41m 06s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/7f22fc000db84fa98e474c785d822a4c : SUCCESS in 45m 24s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/146fc134b3914584a21c76d6385e851c : SUCCESS in 42m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e57f7c9797ea40dca848d18e1cfb6877 : SUCCESS in 1h 35m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d238fa21bedf45ccbb46b42b2a997a1f : SUCCESS in 1h 50m 37s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/430d2aac064141f19ac20f752aec169b : SUCCESS in 1h 40m 46s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/265f37f240af4de48928da976aa69c60 : SUCCESS in 1h 19m 30s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/a5c02d92f84f4a818e19ff3dc882a508 : FAILURE in 29m 52s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/22442c702c4547fbb5cb83c59cd6c0e8 : SUCCESS in 1h 32m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/441d9326f51c496689447fdb562f6d5a : FAILURE in 1h 47m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9b25ef314574bcb8dcaee10ac7521a0 : SUCCESS in 1h 20m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/587bbc5822674964a20b8ac03dfd0384 : SUCCESS in 1h 17m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1d1dc7d709554bf4bf1aeaa8f888ef42 : SUCCESS in 1h 07m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd8aba422e1541e8ba7153cbb64237b0 : SUCCESS in 1h 12m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5e4f6b4282374eb6aee611d46af4d49e : SUCCESS in 1h 03m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/96158c4c8e7445aa88865cdcd758c823 : SUCCESS in 1h 18m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b229b10622f54342bd0a211108603f24 : SUCCESS in 1h 07m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/c632a6044c6147c1935b049fd2800ebd : SUCCESS in 1h 03m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/e2a397b70926487ebeb607989feee443 : SUCCESS in 1h 11m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/babbfef4e49746f68f0c6420d028b8c2 : SUCCESS in 1h 13m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8a9e6e8a758a4810854413f494fa74a9 : SUCCESS in 1h 00m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/e94e87542385491e8d69d51e774302c9 : SUCCESS in 59m 19s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/490d52854e2d4ea5a47fcd39b94c859b : SUCCESS in 50m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/116365a3835c4c8d8012dedb6af02816 : SUCCESS in 50m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/c00e2f812dde4ad598074c411ad5d8cd : SUCCESS in 56m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/39df3027048a4c7aa9d94572db6a2b14 : SUCCESS in 52m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/98097a72a9fa4d0696fb84a7af94c9bc : SUCCESS in 1h 00m 33s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/900eced1654d44e88457e8ff6082b44c : SUCCESS in 58m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/173c77ca80634eb3beb4a3ff200292b7 : SUCCESS in 55m 46s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/563d24cf5f8a450da1031293bd423e22 : SUCCESS in 51m 42s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/9f447b47aec943fd89ce319422d6558a : SUCCESS in 49m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/84d9b452906d46a5ad8896097b5edf58 : SUCCESS in 1h 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a4bb22b1e880e9cad92952382426e9c0ea38ddda","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-15 16:10:36.000000000","message":"Patch Set 14:\n\nrecheck\nThis job passed, other non-related failed.","accounts_in_message":[],"_revision_number":14},{"id":"d7651cd89603860fd8bc97453b1d22f15279dcc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:39:46.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb7a7f95cfd4356af5d43d0b2103a7f : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db2346a7224d4852b24a93f83fd82a41 : SUCCESS in 3m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0e21c9c044946d6bf9072f9b5f5ecf3 : SUCCESS in 56m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/981eb9b5dd3c44c8b596cba870033119 : SUCCESS in 54m 57s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/4793a9ac90044e639af6cf55ee39ad3d : SUCCESS in 43m 55s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/ff78275f682648d88a8e11fd7eee7307 : SUCCESS in 45m 26s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/a48858a95f3b4420bcef945da04a70ae : SUCCESS in 45m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/333c56943ab648faafe1570cf2c6c606 : SUCCESS in 1h 48m 57s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/27aceca7e6414cd6a285067f612a3c28 : SUCCESS in 1h 45m 06s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/624b5608b16546da9a29d9acc21f3602 : SUCCESS in 1h 40m 21s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/41191f317574420780f4b9dcde8cf89d : SUCCESS in 1h 33m 53s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/4e3a362b2237467fa9bc9fba93d972a7 : SUCCESS in 1h 39m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fa956e349be84380be0cd7bc1bb9d212 : SUCCESS in 1h 41m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8e33cf23afd14aeabbdbf83987500c6d : SUCCESS in 2h 22m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa57a4829ed842a29f950d9dcb2bb6f2 : SUCCESS in 1h 34m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/13120e8162264ccd9ea393defae52f84 : SUCCESS in 2h 08m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf91a22dd0db46a6875fcd584cda8530 : SUCCESS in 1h 36m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0784662960bf4a748c4a2db227b53a07 : SUCCESS in 1h 09m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b80a077ecbf74e19aec405393d5fb989 : SUCCESS in 1h 06m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9fbcc17d123244d18d1ae4b96c301bec : SUCCESS in 1h 33m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/038fcf488b0b44398579e1093aa064ba : SUCCESS in 1h 10m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/265a5a6a686e4e479b23854443017df0 : FAILURE in 1h 42m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/b3629c7b1d4e480ca376fa03c66cbbd3 : SUCCESS in 1h 10m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/390d060c76524611a02fc5fd450be617 : SUCCESS in 1h 06m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6df12a3747694198b56f327fdff9ba3d : SUCCESS in 1h 03m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/11c0f62e97d546e2acd31154f6362ab3 : SUCCESS in 1h 14m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/63ea3a185dac4a28824da4587c40c387 : SUCCESS in 1h 03m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4baca3bb980843cf86405da94bdc9223 : SUCCESS in 1h 01m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/a32427d381c1413ab05009f5691c7c5e : SUCCESS in 1h 06m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/1ea724db640a4e109186c05d9037adda : SUCCESS in 1h 12m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5de7b6807a944a7a96e49780d8ab025d : SUCCESS in 1h 06m 19s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/d291b714cae24343882fb71bc7fb7f8b : SUCCESS in 1h 02m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/46e0a021c3cb498db46d9c8ce8d36cc4 : SUCCESS in 1h 00m 33s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/d178e6df644e413fb381721102e372a8 : SUCCESS in 57m 01s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/880f05ceb68a4bfb957b0e1aeff6fb06 : SUCCESS in 54m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a2a6f9d748014aa99cf57e0c9113762c : SUCCESS in 1h 55m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b1d7a2245ec8e29be856cae336042d24a7769607","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-29 21:01:22.000000000","message":"Patch Set 14: Code-Review+2\n\nLGTM \u003cgrin\u003e","accounts_in_message":[],"_revision_number":14},{"id":"fd9dd07585489548a93dfc17579db0d749e34017","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-29 21:03:44.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"7431cd2056fb66fd2ba3ba209df602342c85bb34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 21:04:00.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"7e24f4b8e587e169573cc4f0ec2b851eeddf6d31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 23:51:34.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59cfe6403c0a4a139436fa711e6f943a : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd854b616af459bad893b52734d5d0b : SUCCESS in 3m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f3d2d20ef034357a7b8228837088d26 : SUCCESS in 49m 32s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/47dca6c9c28c4c54bfface1aaa7360df : SUCCESS in 56m 23s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6f8a69d280b24fc4a28d8c0ea5fdeb5f : SUCCESS in 49m 06s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/42a50471b94d49c896669b98724d76c6 : SUCCESS in 44m 16s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/d2b279f9fe2e46918ada6bc258a30b47 : SUCCESS in 41m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6459a4ca1d04f9da233b57c87c3be6e : FAILURE in 1h 43m 49s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a477d365416b4c2c9c17a65f422aab71 : SUCCESS in 1h 40m 05s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4dc073196fdb40c09e2ecb448089e90d : SUCCESS in 1h 16m 33s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/291ba417c4d84c0f95ec3c202b07b072 : SUCCESS in 1h 22m 55s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/f74989593bb34c83a57f20c59d02ac20 : SUCCESS in 1h 28m 14s","accounts_in_message":[],"_revision_number":14},{"id":"4be3bea0d624ffd62c23ebcf6e2f40a58ef9239c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-30 02:18:14.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"6f0800dd903578556b4af65f8f305f592a833895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 05:24:14.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2e87e2af1c4d33aa65d21985a09839 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/771e17cb82034f38afa0a6e87fc18cbe : SUCCESS in 2m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0cf1ba723bca487f962a91b19ae64d8f : SUCCESS in 41m 32s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7f0e510d812e4839a5cfd8c45ec845e6 : SUCCESS in 39m 29s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/dc4c96471487447f9af2cc5bc92a45d7 : SUCCESS in 43m 04s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/c5beee81e9d44b5199b32c35c61d4540 : SUCCESS in 53m 30s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/cb040d7eb53641318d4c13dc543f475d : SUCCESS in 38m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/003d24b750f5477d9b5cd4ed3263e130 : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fd13e4157c4d4164acea494d85054c9d : SUCCESS in 1h 26m 00s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/88a8a6c11b6c4424b3aec5f2fd2c2285 : SUCCESS in 1h 26m 06s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/4b21c06fbc1d4d65a4c7dca359ff23e0 : SUCCESS in 1h 23m 57s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/867e735a39ef4c3892a444b7fd382fd8 : SUCCESS in 1h 27m 52s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aad6958c9c2c4ee9910509fb543bd216 : SUCCESS in 1h 35m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a9edfd98c1284abf9f1df3e3af5c9376 : SUCCESS in 2h 01m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bcb6a16eabc465685a78f7def4bf827 : FAILURE in 32m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36f2734ee09140a2b37dc8d652a4a195 : TIMED_OUT in 2h 33m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/87470cd9514b4891b04de8fe7193abe1 : SUCCESS in 1h 47m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/526b926790fa41f890df9bd5a44501f3 : SUCCESS in 1h 25m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7810ab66ddad4a58b01b9dcfdf3c1a90 : FAILURE in 1h 08m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7795b5e3224440e2b4ca25a132efbed6 : FAILURE in 1h 15m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9136604d0f224fdb8ffeba4feffbf168 : SUCCESS in 48m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/4f63ac67d95a439aafc350ef88f29712 : FAILURE in 1h 54m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/2ae50ab37cc7449eb239afdd2b30961e : SUCCESS in 1h 23m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/884bad83feb34c96bccb9337bd64bede : FAILURE in 30m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/db9714e93c794a5e9c3ba75254c2f658 : SUCCESS in 1h 10m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/cf0781fabaef4243ab5f692d79092b93 : SUCCESS in 1h 07m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6bc16b89f17449078109d8dc7784eb83 : FAILURE in 32m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/673711ce8e6f441697fab7e835147c5e : SUCCESS in 51m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/144f996b7dea435ca417b0251c559d6f : SUCCESS in 59m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/a387cf077ab24d9caf8b6bbd14b0b704 : SUCCESS in 49m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a90c3376eb054e44a2f8d5e79776d78b : SUCCESS in 50m 12s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/e381bb5790214947998810c23303e5e1 : SUCCESS in 54m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/036b21443fac4f46b88d951533960d85 : SUCCESS in 43m 21s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/1f5faf5d1edf4f6d9bd8854b3bf98d8f : SUCCESS in 58m 02s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/69dc5192201b4755b671dffd494b943b : SUCCESS in 51m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4ae2a91556974c9d84f8ee98a5f5c379 : SUCCESS in 1h 48m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ff376ec87e606ad52fa65ab24cdbc0f27289c88e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 05:24:28.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"3049318a80c0cf1b84f0e6ff2163ba1ec50b9f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 07:12:35.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/552cb647900b4fa7847b378acffcd05b : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a8d081f5df4ab08b5896b2c935fac1 : SUCCESS in 2m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df64b1234a2a4aed822bbbedb4b16635 : SUCCESS in 49m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c3c6dd2902d34c199df6e3d3a5fdda81 : SUCCESS in 49m 12s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/bfb689bb0358498dba33400744c943de : SUCCESS in 48m 38s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/2eaf5046b579443d9cc66e20d780e951 : SUCCESS in 47m 16s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/8cb2e096112b4d6589505554f421d7fb : SUCCESS in 48m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4dc6a44ab58469bb9d9f19d827d26a1 : SUCCESS in 1h 45m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/44d5023d140b47f8a362fee6df5fc03b : SUCCESS in 1h 44m 11s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ccd2ab47372c4c8fa41f5b93bb470923 : SUCCESS in 1h 40m 03s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/05b6e02c19984211b2ed603bd6d544a6 : SUCCESS in 1h 35m 27s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/1570751b87f04035952d1c75dd10fd0c : SUCCESS in 1h 30m 55s","accounts_in_message":[],"_revision_number":14},{"id":"fdcea01acee7f501f19f26369662731194d45457","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 07:12:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"8c863cb9208f87ab16e989121a62ceb5deabc6e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 07:16:31.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6b47b4d9e249d5862f113a61a5bf42 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"cccc262c9190e825d2e91bd149da4c8c2aad8ae2","revisions":{"1e7f694cbea00e1c03b44b4f0a9ed8018672c3e7":{"kind":"REWORK","_number":1,"created":"2019-09-17 19:37:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/682726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/1"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:37:09.000000000","tz":0},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1e7f694cbea00e1c03b44b4f0a9ed8018672c3e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1e7f694cbea00e1c03b44b4f0a9ed8018672c3e7"}]},"branch":"refs/heads/master"},"d3eaa49829b4703dfff860574cea386541da4a8f":{"kind":"REWORK","_number":2,"created":"2019-09-17 19:41:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/682726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/2"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:41:25.000000000","tz":0},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d3eaa49829b4703dfff860574cea386541da4a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d3eaa49829b4703dfff860574cea386541da4a8f"}]},"branch":"refs/heads/master"},"6d013a947f5e03db3a26b44313a483cddb98b76c":{"kind":"REWORK","_number":3,"created":"2019-09-17 20:44:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/3"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 20:44:46.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6d013a947f5e03db3a26b44313a483cddb98b76c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6d013a947f5e03db3a26b44313a483cddb98b76c"}]},"branch":"refs/heads/master"},"f20ea063022b32483a0a0fa3271426ac2da9b6ce":{"kind":"REWORK","_number":4,"created":"2019-09-17 20:48:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/4"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 20:47:58.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f20ea063022b32483a0a0fa3271426ac2da9b6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f20ea063022b32483a0a0fa3271426ac2da9b6ce"}]},"branch":"refs/heads/master"},"36c8308a2551e7630038458656272d04c2fbaf16":{"kind":"REWORK","_number":5,"created":"2019-09-17 21:14:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/5"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 21:14:16.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/36c8308a2551e7630038458656272d04c2fbaf16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/36c8308a2551e7630038458656272d04c2fbaf16"}]},"branch":"refs/heads/master"},"54ab4f407df2a1ac396c257f838c08e40a5705f6":{"kind":"REWORK","_number":6,"created":"2019-09-17 21:16:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/6"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 21:16:04.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/54ab4f407df2a1ac396c257f838c08e40a5705f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/54ab4f407df2a1ac396c257f838c08e40a5705f6"}]},"branch":"refs/heads/master"},"d7439acf198060c7483624b679db90fe2670e264":{"kind":"REWORK","_number":7,"created":"2019-09-17 21:23:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/7"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 21:23:03.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d7439acf198060c7483624b679db90fe2670e264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d7439acf198060c7483624b679db90fe2670e264"}]},"branch":"refs/heads/master"},"553d61ae2bcdf7fb535383d00c608572d9299cca":{"kind":"REWORK","_number":8,"created":"2019-09-17 21:24:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/8"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 21:24:37.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/553d61ae2bcdf7fb535383d00c608572d9299cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/553d61ae2bcdf7fb535383d00c608572d9299cca"}]},"branch":"refs/heads/master"},"342721cb592bbb88fed93a72cb9adebf31180d60":{"kind":"REWORK","_number":9,"created":"2019-09-17 21:29:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/9"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 21:29:05.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/342721cb592bbb88fed93a72cb9adebf31180d60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/342721cb592bbb88fed93a72cb9adebf31180d60"}]},"branch":"refs/heads/master"},"00f2b08332e1fc2f7026251e5e67a855be05d25c":{"kind":"REWORK","_number":10,"created":"2019-09-17 22:06:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/10"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-17 22:06:01.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/00f2b08332e1fc2f7026251e5e67a855be05d25c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/00f2b08332e1fc2f7026251e5e67a855be05d25c"}]},"branch":"refs/heads/master"},"47f449870b7986579179078547b1cca749e81f8e":{"kind":"REWORK","_number":11,"created":"2019-09-18 04:19:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/11"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-18 04:19:22.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/47f449870b7986579179078547b1cca749e81f8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/47f449870b7986579179078547b1cca749e81f8e"}]},"branch":"refs/heads/master"},"3ff34be72a3ec62f606415ca92529fe608d4e7a8":{"kind":"REWORK","_number":12,"created":"2019-09-20 04:07:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/12"}}},"commit":{"parents":[{"commit":"714047991937361eca00b2932a729c44ebf81eb6","subject":"Merge \"Enable overcommit_memory in test server VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714047991937361eca00b2932a729c44ebf81eb6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-20 04:07:36.000000000","tz":-420},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3ff34be72a3ec62f606415ca92529fe608d4e7a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3ff34be72a3ec62f606415ca92529fe608d4e7a8"}]},"branch":"refs/heads/master"},"a9412cb944d443c9ad2e6c487ba7b64e4f93fa47":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-10-07 22:50:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/13"}}},"commit":{"parents":[{"commit":"f333d21ea3c76febd09f897587597213c7eedcb4","subject":"Merge \"Fix the health monitor traffic test member ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f333d21ea3c76febd09f897587597213c7eedcb4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-07 22:50:04.000000000","tz":0},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-noop-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a9412cb944d443c9ad2e6c487ba7b64e4f93fa47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a9412cb944d443c9ad2e6c487ba7b64e4f93fa47"}]},"branch":"refs/heads/master"},"cccc262c9190e825d2e91bd149da4c8c2aad8ae2":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2019-10-09 22:11:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/682726/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/26/682726/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/26/682726/14"}}},"commit":{"parents":[{"commit":"f333d21ea3c76febd09f897587597213c7eedcb4","subject":"Merge \"Fix the health monitor traffic test member ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f333d21ea3c76febd09f897587597213c7eedcb4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 19:34:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-09 22:11:35.000000000","tz":0},"subject":"[train][goal] Define new \u0027octavia-v2-dsvm-scenario-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027octavia-v2-dsvm-scenario-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027octavia-v2-dsvm-scenario-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027octavia-v2-dsvm-scenario-ipv6-only\u0027 will run the tests.\n\nThis scenario test will run with the following IPv6 additional settings:\n1. All OpenStack endpoints will use IPv6 addressing.\n2. The lb-mgmt-net will use IPv6 addresses.\n\nStory: #2005477\nTask: #35910\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nDepends-On: https://review.opendev.org/665103\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ieb74ef2276632f04d6324f835b164df3114b2613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cccc262c9190e825d2e91bd149da4c8c2aad8ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cccc262c9190e825d2e91bd149da4c8c2aad8ae2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
