)]}'
{"id":"openstack%2Foctavia-tempest-plugin~695291","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ifc2965152c1bca8e899cf3c84153203ddb85712f","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"octavia-availability-zones","hashtags":[],"change_id":"Ifc2965152c1bca8e899cf3c84153203ddb85712f","subject":"Test the Availability Zone [Profile] admin API","status":"MERGED","created":"2019-11-20 19:17:45.000000000","updated":"2019-12-09 21:09:47.000000000","submitted":"2019-12-09 21:08:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695291-1575925699652-a6c9ab22","meta_rev_id":"de68618a9dbbbe07342030fa961f9fcdc0995d07","_number":695291,"virtual_id_number":695291,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 21:08:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-09 00:44:38.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-09 00:44:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-12-09 00:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-12-09 15:33:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 05:40:11.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2019-12-03 01:10:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-12-03 17:26:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-12-03 17:26:33.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-09 21:08:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4dfaea2437fafad09e345dd67917a94eece98765","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-20 19:17:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86555983a1f59bb14366ef5a032d0e932e805600","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-20 19:21:13.000000000","message":"Topic set to octavia-availability-zones","accounts_in_message":[],"_revision_number":1},{"id":"bf7c035354b9e7c188322607bde4e77fc1759df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 21:48:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0172b60580a447e891d7c085917c0dd : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7afd96ff72f544e991c6843a6db5c30a : SUCCESS in 3m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1f113872a334051b20636ffc3e64a5c : FAILURE in 45m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6f0fec13f14a45b0a759fbaf84bc1011 : FAILURE in 48m 19s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/42145555aede47b080d9640d9f6e4d65 : SUCCESS in 46m 32s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/fa682e6b051e41e4aa87abf0f48065fc : SUCCESS in 46m 25s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/af3de700a5ef4f3e98ac7ce83267e1b4 : SUCCESS in 48m 19s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/9677c7fbc2f545c1884173bf991e280e : SUCCESS in 44m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04a39dc7ee98494ba90cb3c864e34615 : SUCCESS in 1h 52m 02s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0dbf21e117e54ec1bde19e45aba7aa34 : SUCCESS in 1h 42m 09s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5a23f6d3f35340e9ad53824ab8a7a88f : SUCCESS in 1h 40m 20s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8c68b0a6153e46508d5d45e936153223 : SUCCESS in 1h 43m 21s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/22bde8a5e38b4fafb341d39a94d39ff7 : SUCCESS in 1h 38m 23s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/402a9160267342d5942f362f73fc8475 : SUCCESS in 1h 48m 50s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/62cc7915eee146d989763c4243b443da : SUCCESS in 1h 45m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/09566d6f2a894cf6b80c4952cae07d5b : POST_FAILURE in 2h 22m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6319afefc5194b10906347dc0992931b : FAILURE in 41m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bc0855cd7cb343ba87d0d374c6864d24 : FAILURE in 37m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/508a943aa5a74228bf5ed49f1edb9ff4 : FAILURE in 1h 32m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/6ddaa7a7351349f7ad14f37d8f853a2a : SUCCESS in 1h 15m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/cf624fc8e3dc4074abcc3ca34fc31a57 : SUCCESS in 1h 14m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f24a2318f2b54804bf7adea8b8db02b2 : SUCCESS in 56m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/136523c9e2ac445a9e69d3ef22b7fe40 : FAILURE in 1h 24m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d72d78090d4e4c409a9cc3fefaf652ca : SUCCESS in 1h 11m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/39927d3deec943cdb034c3fdcfc06a4f : SUCCESS in 1h 18m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a38aa64eb06345ec8cc1f6144d24ff6b : SUCCESS in 50m 49s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/0d92b613a24f408e9d94de5f3c7d650f : SUCCESS in 1h 04m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/971e179186e44438a47d6b2ac79f9871 : FAILURE in 1h 19m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/de01431673794281b1c771bb710f8608 : SUCCESS in 56m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/b2ef72533a2e4db7b763ded8cc265061 : SUCCESS in 52m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fe670485c0fc4d0ab0bcbf2d1acd1d27 : SUCCESS in 52m 48s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/8966b74a941d493183436f8b47b92278 : SUCCESS in 53m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/e32574d6b1f74d0dbd1f29c73b060c89 : FAILURE in 56m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/6c81b1243dfc428f9c0dff86bfaafde6 : SUCCESS in 55m 35s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/ea87311c5b294ccd9471b333663523b3 : SUCCESS in 56m 51s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/8d371c980fc54eaeb6750bc2a00ad385 : SUCCESS in 52m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/372f6db27a8d4067a798c371816ecc70 : SUCCESS in 1h 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"654e4ef7e204bf15e9754df0c15cb780fe9b49fd","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-21 04:53:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4bc14071e36a71b7d7d19681e379f9509988cd8c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-21 05:04:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a6f6f454812443d78d9a2ed3c00b67b8c750984","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-21 05:40:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e5f46f4b908f911356aba589da7bb9a6052d1446","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-21 06:22:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b481243454bfa0e45c44b89a4d18fd710e669f5d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-21 06:30:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ebe5e9367c1159c4d76a5f3e57cac248c33dfa91","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-21 06:36:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"14d726985f0756603c9cbcaf70815c3f3d7d881f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 07:54:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135dde390e7f41979453e13b0dc7063d : SUCCESS in 4m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f22f72986dd4567a6727c86fbf50f3a : FAILURE in 1h 00m 37s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/91293e7609304b028091ea9228b3ffb6 : FAILURE in 49m 11s","accounts_in_message":[],"_revision_number":6},{"id":"c30188ea4753ab09dd704824ccfa35782ef5c627","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-21 08:10:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7b813246bf7726ed1f0c654eab9a8cb1d290c00a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 09:02:17.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576845f4ae824714a93ab7acff68c58a : SUCCESS in 2m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4735cb1298464c43818db4e3962e658b : FAILURE in 46m 18s","accounts_in_message":[],"_revision_number":7},{"id":"a537045fb3349164de9b051bd392665cdee6ac8d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-21 12:34:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b9a1728065db7c55a8b939c6fdc1ffb234afe549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 13:28:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ad05dfac9c4763b55d358d6e837114 : SUCCESS in 2m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea0b0e3946614edc917196e4a4b736aa : SUCCESS in 45m 58s","accounts_in_message":[],"_revision_number":8},{"id":"3d6fa0be2ed19c387727173987db561ae49b062e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 19:09:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"15975919e3fb75d8c7df38cbd6db45fbad0b55f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 19:54:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e66f40cea8541fc8e12dfc056155242 : SUCCESS in 3m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f8a266f3f694d1a87f59fb89e3561c3 : FAILURE in 43m 32s","accounts_in_message":[],"_revision_number":9},{"id":"8e3ddf865064373c3d2c1f3d3109b448ddc55d86","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 13:39:10.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"ffbe605dc634817a70070042e989b24080c9c819","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 16:57:25.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"694ad286442457039f4b620b5e86e74f77e01875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:25:17.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6cfd8e2f2b84d3fbbe5bfdb9ac28e2d : SUCCESS in 3m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/100135e919bc4335bffca6a91befe7bb : FAILURE in 56m 02s","accounts_in_message":[],"_revision_number":10},{"id":"6bff379533114fdbd6e7ee84ddbb9e71d295976d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 18:26:18.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8854e476d267a11367dea3d62c4162d4dcc45fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:00:34.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4eaf7f98c704a8dbfe7121aee6e2120 : SUCCESS in 4m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aa97c86f41c246cb88fc277c502a38cf : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":11},{"id":"6c37836004803904cbc5f5d4a944fbd3851d0657","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-28 05:17:08.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ad7b993b76f64e88c16804aa7e14ac9200aab589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 07:27:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f73adf95f6401cbd00348c96376de8 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99515b4c806745128a344ca226a97633 : SUCCESS in 3m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/57062cdd01f54c798da1a23078267e4c : SUCCESS in 46m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8fca6aa824c343b98fb277581a851351 : SUCCESS in 46m 06s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/371a8139922645dea0c39493acf85fbf : SUCCESS in 43m 56s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/4da44d3286a6426b8f940804f7dc80cb : SUCCESS in 44m 51s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/1074a969492f4faaa9057bf0f90fe978 : SUCCESS in 41m 49s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/8038b1453c234d499b73bb666c30f108 : SUCCESS in 45m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/617d37aaec0e4852b5280ab26d4e0a12 : SUCCESS in 1h 43m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f72f85b637f14a69952ac4c94f3785d3 : SUCCESS in 1h 40m 10s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d683423943304053bbaac01f680dfd65 : SUCCESS in 1h 41m 10s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/5f35a1e5b7384c58bc6609e24e3da320 : SUCCESS in 1h 24m 41s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/f19b798f0a1c4836b5e0b6eb2963173e : SUCCESS in 1h 30m 43s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/587002ab8ea2430f8edf764932e8a47d : SUCCESS in 1h 20m 38s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fd66bf5d1e6743a196b5aeef08702268 : SUCCESS in 1h 35m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4224fc49121d4b659f0e4c13e8562425 : SUCCESS in 2h 05m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/493ddf978e734413aa926dc9f3d5803d : SUCCESS in 1h 34m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cdbc7d3ff584597a09a8753486406d7 : FAILURE in 31m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c4d70872ee984f97994e8c7786700bab : SUCCESS in 1h 46m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/6a31a43a18254f208c9e91e32eae4819 : SUCCESS in 1h 03m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/7678fa0fa31e4dab8de0c4df6a5f15ad : SUCCESS in 1h 03m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/168d16c2e1e64e039e48535c6623e08a : SUCCESS in 1h 06m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f6bc69cb9add4c09ac146cbe495cade0 : SUCCESS in 1h 16m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c3b7bb82be4c4fd3a92afcd5c46945df : SUCCESS in 1h 06m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/de0d40c45b7144378bcff30b9cf92436 : SUCCESS in 58m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/27e961adbecf41eabf9410a639a29a94 : SUCCESS in 54m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/378405a5c05e47b89309e48891510189 : SUCCESS in 57m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/b8c1755b932e441d8a284cda1cbca4aa : SUCCESS in 56m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/357878b9bcaa4c8fb52f3764818aa11a : SUCCESS in 52m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/9ae1f63364cc43d0a3f94759b97d9519 : SUCCESS in 55m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0c7a15b479814276b972296daef1da7a : SUCCESS in 58m 03s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/5a687ed864bb49679952743a48da27cf : FAILURE in 1h 23m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/fbaf56ab6be844f3a9885079d1ef434c : FAILURE in 1h 25m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/4fd7ce82d3034d9d859db0dbe1c1e1c7 : SUCCESS in 52m 08s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/954e926dad734e228506b360bfe4d7b0 : SUCCESS in 56m 49s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/9fcf627b970749bfacde57ed7d98d944 : SUCCESS in 53m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9b811026572549c7a7bd8f516c7e6c4f : SUCCESS in 1h 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3736bfe5e42f81421ef7f6cd5a82343ec74381d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-03 01:10:05.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"e8c4855e1ba62dd48a8eef0c58cca1522a7e18a3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-12-03 17:26:10.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"e80009c6f3b5ce84f9b9956ae037a8ce82a60e7b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-03 17:26:33.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"896a5037404ac7920b5b22461acf2841ca390035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:26:46.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"e25c4897bcabc35596aa164e8c53b672858b9990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 20:06:29.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53fb4f58bd344c0aadc9c3bf56cec6a9 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f66639cb5947ebb7c0934350733df5 : SUCCESS in 5m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc9187b3e957420caad45ddc9f599ee7 : SUCCESS in 47m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/57cfa1d0e6af429991dc0b1136700473 : SUCCESS in 48m 14s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/2ec74f3811e04fb8b9aa90757f897372 : SUCCESS in 45m 07s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/d1bbf164999941a892b2fafdb9eef312 : SUCCESS in 46m 17s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/a6dab94064de48d59ffe12004cd39727 : SUCCESS in 44m 27s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/ef1cad8ebd3c4a95beceff2ccab5ebd4 : SUCCESS in 45m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06db945fca3b4766a3794c137faa4cf3 : SUCCESS in 1h 50m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f5f6744bd1a4452da22a301d73894d14 : TIMED_OUT in 2h 34m 24s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e76a8a716f514195a5926f95b1d0d31d : SUCCESS in 1h 37m 35s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/70c92a2a8b1847fd8efa62f855937262 : SUCCESS in 1h 46m 32s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/923a16d7ea224eadad087e232b639774 : SUCCESS in 1h 32m 27s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/05d2bafc94bb48278fe68eec92bc8944 : SUCCESS in 1h 29m 21s","accounts_in_message":[],"_revision_number":12},{"id":"97d80c80cc6fd92f6c14dc72136d0cde12040d76","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-04 23:31:09.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"1b0e9d54d89fbc20e9fdd8c398e6e02d25a94810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 02:11:28.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f94e0c9595040fbafeb6b4e953662ee : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79141412dfff480db27c4de42e26ba69 : SUCCESS in 3m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/568a2e65467f43d689a160c819975d28 : SUCCESS in 44m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dd22b58361f947709caccc16fb701af2 : SUCCESS in 49m 26s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/6a38be99b66a420687ffe71a18a526fd : SUCCESS in 43m 51s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f94f6b8d96484ca6bc3ca0bb2bbff3f7 : SUCCESS in 47m 54s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/32b88171bf99433e937876925a541b60 : SUCCESS in 38m 37s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/ac90859a10354df4bd9b8767caaeea2e : SUCCESS in 44m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/caa67cd17e374098927974507324b967 : SUCCESS in 1h 32m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1d7a25b8d0c44e589a10c1a4395eba94 : TIMED_OUT in 2h 33m 35s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/59ccca99b4aa482e8e5269c509245494 : SUCCESS in 1h 46m 57s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/37fafdda8a3e480aae21dcafd174dbc0 : SUCCESS in 1h 35m 33s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/53d91ccf8e264af69b6604385425d55f : SUCCESS in 1h 49m 39s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/3502deabc5584e6d93a756b82a0e934e : FAILURE in 1h 29m 17s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aeda382b961440aeb517fe0aee303606 : SUCCESS in 1h 38m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4171ee43756a41c68b5cc87291658b80 : TIMED_OUT in 2h 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cca759b773324b47aa16fafcf59621e7 : SUCCESS in 1h 42m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a1ad3647a5954a44a4a95c1254fb0850 : FAILURE in 37m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bde2bc0fb1aa4e6a9eabbc39a1a2570b : TIMED_OUT in 2h 33m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/2378eed5b6f8463da2932f876ccf4e17 : SUCCESS in 1h 01m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/05fde6641d5e4344ac802aa777da774f : SUCCESS in 1h 06m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75d8556c139e4962ae39965254834a51 : SUCCESS in 1h 01m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/587e16a0e9204bb4af9306f332b9462c : FAILURE in 1h 06m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0c68a04bae17493f95e1739810b44da4 : SUCCESS in 1h 06m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/84a42a0492264bf0b28abfb53a35d85a : SUCCESS in 1h 03m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ca7cad24e3a491d9b08a50529091402 : SUCCESS in 55m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/6e3c6723a40546cc82d18823c9d213be : SUCCESS in 56m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/85967ddd14cc49cb8b9d06a4697a0c1d : SUCCESS in 52m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/8769f86904474f92baa66d06e7e2f57a : SUCCESS in 54m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/86199cc4ef4443f4b471d6ebc1ce0b37 : SUCCESS in 43m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/48d96907a8114d438ef20efeeff3c65f : SUCCESS in 55m 46s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/b3d596545d4445fdbe80e57d39f93c8b : FAILURE in 54m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/83ae3404b6ba42129f2626d68833f72c : SUCCESS in 50m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/01901cb29bcd4f3ab571c68b5e244119 : SUCCESS in 1h 01m 24s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/c06d6cf1e8664b9e93e75106a3ab9960 : SUCCESS in 55m 13s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/e2950e4fb6014670a479faecd7b2d512 : SUCCESS in 52m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e2b7a8124c6e41688fddf471ef1e8467 : SUCCESS in 1h 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b9e9d24fab925f395c51e8f9d6cb15e9b7b5deb2","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-09 00:44:38.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"19b5766496e8dadc0f2f14f7023f2e901d172c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 02:25:47.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1c90b85e47456d9bcaaa51b5616131 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55541e44cd61452c93a7e68e4832a0b0 : SUCCESS in 2m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b9f87ced7414bfa8414d767fa5696bb : SUCCESS in 47m 09s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/b04073fff7a6401699c4231f337323df : SUCCESS in 46m 48s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/34b8518e848c484ca5f2de9e21ae7c28 : SUCCESS in 52m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/4c500c97b2f14961b0ca181975340ce9 : SUCCESS in 44m 12s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/bfe4e66466f348af9ccdd5188105334a : SUCCESS in 40m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47f76bad6b2e4d73a3a0d882cb81e707 : SUCCESS in 1h 29m 47s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6fbc0470de3b4bab94abacba2d64bf7d : SUCCESS in 1h 23m 57s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/dfbafe70250743db993fdc2f762f15d1 : SUCCESS in 1h 18m 58s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/417a1c6eca2a483a9f0a56fd7c64d12e : SUCCESS in 1h 15m 50s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/42c06a5c731144639a9af404152138bc : FAILURE in 1h 17m 06s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c5b759fc3775453c8a9518681393d516 : SUCCESS in 1h 32m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a280ef2141714141babfa8b5a58dead9 : SUCCESS in 1h 26m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e85dcc3320694de083c4399a3cd76949 : FAILURE in 8m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/d3d0c9c7925f4eae9cd86c1fd2fac18a : SUCCESS in 58m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/fe9830148f204f2c8b0d623b959f23f2 : SUCCESS in 59m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41dc43fdcfa24f3190de44bdc3533b64 : SUCCESS in 59m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9f3ad73fa3344a6c9de9abbd3d647052 : SUCCESS in 57m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6783d511103e454e96d30c22bf3ec4ba : SUCCESS in 55m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/75491f0ffe974f1087e2c7014e89b1bf : SUCCESS in 54m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/65ce430b0f854da9ba790b0c19ca5abc : SUCCESS in 54m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/6369b483a1d243f68d2b681d63250e31 : SUCCESS in 50m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/09b1b6d8b2794988abb1a1443e05c52e : SUCCESS in 59m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/47cc100505d74c7e972a2f97700c06cb : SUCCESS in 51m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8f04fb30dc4c499ab1885edb730ab816 : SUCCESS in 1h 05m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/510c9d54a8ac43e9960baeea02958a6a : SUCCESS in 59m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/317b85d3dd1649b69d68ea8d6a93c96c : SUCCESS in 50m 42s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/f666ee7ac4974259b193300c2e89d379 : SUCCESS in 48m 20s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/5ebd5f0b84e24f3dae624f7de4ae00cb : SUCCESS in 50m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/156da4be2a7f4eabb88aa7d80316053c : SUCCESS in 1h 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"dfe785bc92e1bbd42e57748fe80963f2b979b199","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-09 15:33:27.000000000","message":"Patch Set 13: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"2572b785850bfeff9fa5024dfdb0a87d1aa46c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:45:37.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af302ab19274c13bc79c347902f1024 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc7a24ce1a814093b503c5bb22c9158d : SUCCESS in 3m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/35aacf926da848bca418dba287ba70ad : SUCCESS in 45m 49s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/a2d1cd9587994a6fb2f00f024a398be2 : SUCCESS in 45m 39s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/9647ce5e42ac49edbf5c470b56a203fd : SUCCESS in 44m 23s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/be4492f8439b42c6a11160ec2c1a0ed8 : SUCCESS in 46m 10s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/d387768857504f10aacfce52016a4e5e : SUCCESS in 44m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1d32fc93a70455e88fecd99a9f00c6c : SUCCESS in 1h 48m 30s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/245d154d081a4483996d094482bee1ad : SUCCESS in 1h 51m 12s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/10699051a2e3467cae6d06e800cc752e : SUCCESS in 1h 33m 45s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/0f5065deaead4294a772ae1082dd204a : SUCCESS in 2h 08m 34s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/5e4893fe425143a5aeb6c77895518277 : SUCCESS in 1h 35m 44s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/698ae75e0e15490bbb80326ecd0cf56e : SUCCESS in 1h 36m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/116b913a7c224323bd48d91e75e18b32 : SUCCESS in 1h 39m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f63c189cbb9747aab6d7e73f78af44f8 : FAILURE in 16m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/228e6f516f774240b399a1a47d754093 : SUCCESS in 57m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/8852f149225f485cbd90d91a27066b4d : SUCCESS in 1h 00m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a526b07c72184ab5bf61f9824ee6b203 : SUCCESS in 1h 06m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d6a725d593484d93b23e8143af75c5f3 : SUCCESS in 1h 03m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/60026abfba9448389179cb45814ae484 : SUCCESS in 1h 08m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ce99e05ce654045ad1136b74bdf51a3 : SUCCESS in 1h 02m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a6f8afd83cc84d9f9b318e1e107da34c : SUCCESS in 55m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/28a6aca3301740bcb216cd951db22b04 : SUCCESS in 1h 11m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/3d8d8f74c99b4cebb3c56901fe312177 : SUCCESS in 54m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/8c8a7575bec04c3da3ec07eb103178eb : SUCCESS in 59m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d4c2bf0d1acc469e9bf457a32db0d7d9 : SUCCESS in 53m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/cf22f28dff274b6d890b461136d4c524 : SUCCESS in 1h 11m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/d30f3268e3374507a2f5eef5a3e717d1 : SUCCESS in 56m 14s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/6491c535696c49a8b799f6c2a3176075 : SUCCESS in 1h 01m 46s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/43db135954544c0aa7473558d7037238 : SUCCESS in 51m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54495e5a6f2e488e8ddfa19196b4d6d9 : SUCCESS in 1h 58m 22s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"528d55c76cd254ca43895a6be362209e762d7fdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:45:52.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"485a4a8b1d55690374e532e459ba01fba6380a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 19:13:25.000000000","message":"Patch Set 13:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"5aa617aa59954234c4f2b3c377402a6c23e7b36f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 21:08:19.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51181aab198469daab3c628b3762737 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbfc9d5181e433ca03d42577e17d822 : SUCCESS in 4m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/acc58c9dc77947688884e3ce07c42e98 : SUCCESS in 53m 59s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/53eb21448c3b47b7ac43107a6927d03e : SUCCESS in 46m 47s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/718f0303de85409d974a88c10ddf0180 : SUCCESS in 49m 50s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/44de9a3169f34c8f8b1192b06d7c5d34 : SUCCESS in 47m 12s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/ea2c2c37799448b1bcea74f8f54b4bf1 : SUCCESS in 43m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7aedeccdd1ca45ab83521f91d7b6e01c : SUCCESS in 1h 32m 21s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/125db4fd00a14e459932ccf012e0a819 : SUCCESS in 1h 47m 50s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ea6fd3d6524948fe9103acf04d0dff61 : SUCCESS in 1h 38m 57s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/c274bbeba6b646f1a779a442b0b7afde : SUCCESS in 1h 30m 55s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/79220514c7664f84a1a4a6f4e126c6d7 : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":13},{"id":"ec055f45285726637c908d45c77707c0b8975466","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 21:08:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"de68618a9dbbbe07342030fa961f9fcdc0995d07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 21:09:47.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b9a7b7c5984d77bcc94d9f2d8dd3d6 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328","revisions":{"f3d94f675d7d507dbf93ad560a822daae5c45576":{"kind":"REWORK","_number":1,"created":"2019-11-20 19:17:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/1"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"subject":"Test the Availability Zone [Profile] admin API","message":"Test the Availability Zone [Profile] admin API\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f3d94f675d7d507dbf93ad560a822daae5c45576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f3d94f675d7d507dbf93ad560a822daae5c45576"}]},"branch":"refs/heads/master"},"bcab0c17867689c720c34991e05d2c5d4aeaa0e7":{"kind":"NO_CHANGE","_number":2,"created":"2019-11-21 04:53:41.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/91/695291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/2"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-20 21:15:05.000000000","tz":660},"subject":"Test the Availability Zone [Profile] admin API","message":"Test the Availability Zone [Profile] admin API\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bcab0c17867689c720c34991e05d2c5d4aeaa0e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bcab0c17867689c720c34991e05d2c5d4aeaa0e7"}]},"branch":"refs/heads/master"},"488a3fff28e31e58d3dff98e750ab5b8bfcb5030":{"kind":"REWORK","_number":3,"created":"2019-11-21 05:04:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/3"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-21 05:04:16.000000000","tz":-480},"subject":"Test the Availability Zone [Profile] admin API","message":"Test the Availability Zone [Profile] admin API\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/488a3fff28e31e58d3dff98e750ab5b8bfcb5030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/488a3fff28e31e58d3dff98e750ab5b8bfcb5030"}]},"branch":"refs/heads/master"},"747fdf4db77238f6d3887f1210616d5f9221fcb3":{"kind":"REWORK","_number":4,"created":"2019-11-21 06:22:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/4"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-21 06:22:05.000000000","tz":-480},"subject":"Test the Availability Zone [Profile] admin API","message":"Test the Availability Zone [Profile] admin API\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/747fdf4db77238f6d3887f1210616d5f9221fcb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/747fdf4db77238f6d3887f1210616d5f9221fcb3"}]},"branch":"refs/heads/master"},"f8d35650fe46691abeb96e5582a37735f71d4c60":{"kind":"REWORK","_number":5,"created":"2019-11-21 06:30:34.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/5"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-21 06:29:58.000000000","tz":-480},"subject":"Test the Availability Zone [Profile] admin API","message":"Test the Availability Zone [Profile] admin API\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f8d35650fe46691abeb96e5582a37735f71d4c60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f8d35650fe46691abeb96e5582a37735f71d4c60"}]},"branch":"refs/heads/master"},"a83e4311d947d6f5977f726ea4cb70df27380d0c":{"kind":"REWORK","_number":6,"created":"2019-11-21 06:36:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/6"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-21 06:36:16.000000000","tz":-480},"subject":"[WIP] Test the Availability Zone [Profile] admin API","message":"[WIP] Test the Availability Zone [Profile] admin API\n\nTemporarily disable extra jobs to ease load on zuul.\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a83e4311d947d6f5977f726ea4cb70df27380d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a83e4311d947d6f5977f726ea4cb70df27380d0c"}]},"branch":"refs/heads/master"},"81fc9f68e40e91821c4479631b21f4125b386a24":{"kind":"REWORK","_number":7,"created":"2019-11-21 08:10:12.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/7"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-21 08:09:45.000000000","tz":-480},"subject":"[WIP] Test the Availability Zone [Profile] admin API","message":"[WIP] Test the Availability Zone [Profile] admin API\n\nTemporarily disable extra jobs to ease load on zuul.\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/81fc9f68e40e91821c4479631b21f4125b386a24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/81fc9f68e40e91821c4479631b21f4125b386a24"}]},"branch":"refs/heads/master"},"46a4f10133e17570da583301da8c34d68eb51a89":{"kind":"REWORK","_number":8,"created":"2019-11-21 12:34:30.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/8"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-21 12:32:32.000000000","tz":-480},"subject":"[WIP] Test the Availability Zone [Profile] admin API","message":"[WIP] Test the Availability Zone [Profile] admin API\n\nTemporarily disable extra jobs to ease load on zuul.\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/46a4f10133e17570da583301da8c34d68eb51a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/46a4f10133e17570da583301da8c34d68eb51a89"}]},"branch":"refs/heads/master"},"bf5161c63951639602578e56852d3171335f81df":{"kind":"REWORK","_number":9,"created":"2019-11-22 19:09:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/9"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-22 19:08:58.000000000","tz":-480},"subject":"[WIP] Test the Availability Zone [Profile] admin API","message":"[WIP] Test the Availability Zone [Profile] admin API\n\nTemporarily disable extra jobs to ease load on zuul.\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf5161c63951639602578e56852d3171335f81df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf5161c63951639602578e56852d3171335f81df"}]},"branch":"refs/heads/master"},"132714b49d24c2864dfea5b1b422452e5217c205":{"kind":"REWORK","_number":10,"created":"2019-11-27 16:57:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/10"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-27 16:56:15.000000000","tz":-480},"subject":"[WIP] Test the Availability Zone [Profile] admin API","message":"[WIP] Test the Availability Zone [Profile] admin API\n\nTemporarily disable extra jobs to ease load on zuul.\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/132714b49d24c2864dfea5b1b422452e5217c205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/132714b49d24c2864dfea5b1b422452e5217c205"}]},"branch":"refs/heads/master"},"b9cd994ddc52df1c30463faf503857b73740f369":{"kind":"REWORK","_number":11,"created":"2019-11-27 18:26:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/11"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-27 18:25:31.000000000","tz":-480},"subject":"[WIP] Test the Availability Zone [Profile] admin API","message":"[WIP] Test the Availability Zone [Profile] admin API\n\nTemporarily disable extra jobs to ease load on zuul.\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b9cd994ddc52df1c30463faf503857b73740f369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b9cd994ddc52df1c30463faf503857b73740f369"}]},"branch":"refs/heads/master"},"f86df83a8bd04d3211d1d905bc5dcd244b1aa90d":{"kind":"REWORK","_number":12,"created":"2019-11-28 05:17:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/695291/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/12"}}},"commit":{"parents":[{"commit":"9c7fe1f8771fe2d3493b15482d51417f6ed98c1c","subject":"Merge \"Remove act-stdby-iptables scenario jobs in Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9c7fe1f8771fe2d3493b15482d51417f6ed98c1c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-28 05:16:16.000000000","tz":-480},"subject":"Test the Availability Zone [Profile] admin API","message":"Test the Availability Zone [Profile] admin API\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f86df83a8bd04d3211d1d905bc5dcd244b1aa90d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f86df83a8bd04d3211d1d905bc5dcd244b1aa90d"}]},"branch":"refs/heads/master"},"c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-12-09 00:44:38.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/91/695291/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/91/695291/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/91/695291/13"}}},"commit":{"parents":[{"commit":"70c90ff66f58119f418c57f88358b8b4f42f84c3","subject":"Disable py2 jobs on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/70c90ff66f58119f418c57f88358b8b4f42f84c3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-20 19:15:07.000000000","tz":-480},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-09 00:41:58.000000000","tz":660},"subject":"Test the Availability Zone [Profile] admin API","message":"Test the Availability Zone [Profile] admin API\n\nDepends-On: https://review.opendev.org/#/c/693765/\nChange-Id: Ifc2965152c1bca8e899cf3c84153203ddb85712f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
