)]}'
{"id":"openstack%2Foctavia-tempest-plugin~695349","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I03776d0b7f19066e85034020f374fd8102a0018e","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"octavia-availability-zones","attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-02-24 14:27:17.000000000","reason":"\u003cGERRIT_ACCOUNT_32053\u003e replied on the change","reason_account":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"}},"3031":{"account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"last_update":"2022-10-31 16:03:45.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-10-25 11:09:35.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I03776d0b7f19066e85034020f374fd8102a0018e","subject":"Add new scenario test to create LB in specific AZ","status":"NEW","created":"2019-11-21 04:53:41.000000000","updated":"2023-03-27 17:14:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":22,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"5ad0f522346273483f29b9218010c4c6f60549d7","_number":695349,"virtual_id_number":695349,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-15 16:54:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2023-03-27 17:14:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-10-31 16:03:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2023-02-24 14:27:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},{"value":1,"date":"2022-10-27 00:59:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2022-10-26 11:30:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-10-25 11:09:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-06 08:43:58.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-12-17 20:22:14.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-01-09 14:45:15.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-01-09 17:26:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-07 06:49:27.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-03 18:43:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-12 21:09:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2022-10-25 11:09:35.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-10-26 11:30:51.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-10-31 16:03:45.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-12-27 10:03:38.000000000","updated_by":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"reviewer":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"state":"REVIEWER"},{"updated":"2023-03-27 17:14:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"688438b895fee304fbfc55197e1055dceba1cda8","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-21 04:53:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c366154863ababdcbc3954098d988401cc007fe7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-21 05:10:15.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nYeah this was what I had in mind as a start. If we can make this run, I\u0027d like to switch it all over to its own two-node test and get multiple AZs set up (and run only these AZ tests).","accounts_in_message":[],"_revision_number":1},{"id":"b6262c859ade6bf5cf4135cc9e740e6858492252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 07:16:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ffff6b958dc40129a2aeabad5ffd694 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a8c5cd4ad8c4b33b4e9f8b7f93c517a : SUCCESS in 2m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cdfbdf123034de58a01c0d09cb90c3b : FAILURE in 44m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/14d35c2cbb6a4655b303696177dd0431 : FAILURE in 45m 05s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/092983bd191e450ca7bd9d1f5b3c928f : SUCCESS in 43m 49s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/da6da633bff8482bbea68d547cf9584e : SUCCESS in 50m 18s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/879ba05ea88d46238fa3768df5391c68 : SUCCESS in 41m 39s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/dccdf831b01b4a2898c36fcf3fe11da5 : SUCCESS in 50m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26f2eb029aad44a18ca7a971c58ab2cb : FAILURE in 1h 47m 11s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2e01496fe7ba4fd8a2be9905d82d7c5d : FAILURE in 53m 30s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f7bde274de1b4fe192157e0aa505c9fb : FAILURE in 1h 20m 48s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1abd1e3fc29342caa4172bff022cff80 : FAILURE in 1h 28m 29s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/363485a65829428f844e07982bef621e : FAILURE in 1h 54m 12s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/75bbfcbdd5764ce69d826792c5ef16cb : FAILURE in 1h 45m 18s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/18dcaeec7af8486b9997409a2cdee0e7 : FAILURE in 1h 34m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/47070e167ab64025a8dbd60ac91b2393 : FAILURE in 2h 07m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/809d6fe02022410cbfb010a337736a07 : FAILURE in 48m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f7db8313a03f472ab603caa114f0f90a : FAILURE in 31m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ac9ade0e4674f61af167327dbf4e86a : FAILURE in 1h 52m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/0fc9a03b9fed4f33b182d5ff24a00894 : SUCCESS in 1h 02m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/3ca618ff588a4979a85d9a6667d55689 : SUCCESS in 1h 18m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f172caa3d32441f89010b192a2ee334d : SUCCESS in 1h 12m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/31f26e707ed74752a288b00146aa37b9 : SUCCESS in 1h 07m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6e632f4c49b44ffab380a6098354caa9 : SUCCESS in 1h 00m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ad684db758f14433b1fc41a1e4ca7d6d : FAILURE in 52m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c46c333a0544d0d9eba5b7c66735312 : SUCCESS in 49m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f606f1dbc132463bbc04aa05f25f7148 : SUCCESS in 48m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/a3a9229a9676486fa52ef0b7b4bfc2fd : SUCCESS in 50m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/978f605928c047a3b8c5af8109f38d8c : SUCCESS in 1h 07m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/18620a61c17c4ff59031b069e1049485 : POST_FAILURE in 49m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/53f78348bd6e4345849b0a28f6fe8299 : SUCCESS in 1h 03m 56s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool https://zuul.opendev.org/t/openstack/build/4091012928f147aeb27ae007f3690648 : FAILURE in 53m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d2f2c49b996b466481c2d248954daf26 : SUCCESS in 50m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0fa9a64731bb44dcb9c9984157176d0e : SUCCESS in 55m 26s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/cde14dc9154a4a55aa6a76c3ae3342dc : SUCCESS in 52m 53s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/36995ed53ff44b49b38dede27e78d8dc : SUCCESS in 58m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d98419c539b44f97bf1adb09b404fa22 : FAILURE in 1h 39m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d938ab23eafde09910c4b16360bba0b768525dc","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-21 22:31:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a273cc0a9082debbef016b88a94036131f6a35ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 23:35:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a851d695b82848fca251f11ff56d60c4 : SUCCESS in 2m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7a0fab557d7d4098b6ce5bd0e2bd12e6 : SUCCESS in 40m 32s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dafec56b65c84f8b8eb386470d15db75 : FAILURE in 44m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a3d2cd54df807fd8655c2a07e35db37283a5e032","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 06:39:37.000000000","message":"Topic set to octavia-availability-zones","accounts_in_message":[],"_revision_number":2},{"id":"451ea19b8853a446d7c352f2e0d028fee8b4276a","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-26 04:46:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f1f5845678d64045d146d2f3e3d376b7535c915a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 05:32:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd952dc6e529468894a7eee5ae4e5f70 : SUCCESS in 2m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aaf587479edb4eb7ab0ae9dfc185267e : FAILURE in 44m 25s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2eb35d9f5aed49a5aef7665abe8c4fb9 : FAILURE in 28m 09s","accounts_in_message":[],"_revision_number":3},{"id":"b69edf8dca2dede631908ee2685583b3845f99e3","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-27 22:22:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5d43dd3a4ec5ce8abd88a71c3186fc3db8953e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 23:03:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ba42a39d2d4f49a4bb6789ed60b95e : SUCCESS in 3m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fb1ff93b6bb84ba4b86c4092270f9144 : SUCCESS in 40m 58s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bea8ffed8ec34cd9b63c98902513a7a3 : FAILURE in 35m 27s","accounts_in_message":[],"_revision_number":4},{"id":"a0a6b32902f4abe2a3e581462a044688bea16149","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-02 00:55:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8680da1169f9f513928c1908a9e75de8311069b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 02:11:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c858bf558d479d9d07ec2d94f13bab : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3b47d03bbc48bca93a8dc6fef0008a : SUCCESS in 2m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ec50ba57b9f4f218dd79959cb569a82 : SUCCESS in 38m 05s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/713b00cf3d9d48ea865d03a014683ddf : FAILURE in 30m 50s","accounts_in_message":[],"_revision_number":5},{"id":"69315364a49359a460fc27f4379d023d1524d210","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-02 04:51:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"539f9bbf9680488bb4d17109284dda1e8c4f7bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 06:22:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c28e5c738b458f8c2f728f27dd3915 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aeebde77a064f0dad711242da327f80 : SUCCESS in 2m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3977d542958433eb02d0c7061d609b9 : SUCCESS in 39m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cccee0337aa94ac8a2a56ac423e52357 : SUCCESS in 1h 29m 12s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4025e83a3f374cf7856a393c9b77c33a : FAILURE in 29m 33s","accounts_in_message":[],"_revision_number":6},{"id":"c51c9dddf8d166698c20d38ebfd11f9024468242","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-03 05:18:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ba99f9804ae3663ac1607333904f3e8e2140b8a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 07:53:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db019c723ee04c9499bdc159fc0064da : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8797a1b8c5b24cd2b9dbada7ab8cd319 : SUCCESS in 3m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c271ac898c9744bbb105b1834ea9542a : SUCCESS in 42m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fa6c3a9650c4e1c83d4e2d066465a34 : SUCCESS in 2h 00m 48s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/14597de7c40b4b9583908a5d3987e528 : FAILURE in 30m 49s\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3b37351a6a4e43e992f7e18fcfb66c55 : TIMED_OUT in 2h 33m 36s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/295ef0ceb6b647d8995ead7d3b28acfd : FAILURE in 30m 31s","accounts_in_message":[],"_revision_number":7},{"id":"32986547db7595cef0e22bb3d3229c1fde0f29cd","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-03 21:46:34.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"64ef391e4adde5fb17e0a82212a2472c6d7c28d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 00:25:48.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a04b56467eb4fe8823239edcec657de : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6fc32336f2c42a0a5084df6ecd85127 : SUCCESS in 2m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1cfc66faadbb4cff8f8421b1abae8ff4 : SUCCESS in 45m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e99f3c473cad436c9bff723f6280b407 : SUCCESS in 1h 43m 21s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8aeb5c9da6c944f2ba5990f70d4ab432 : FAILURE in 32m 50s\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f011ca8a29c249a6aeba11b5678adc58 : TIMED_OUT in 2h 32m 56s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d3e57e98e5840eeb0de910ab6e815eb : FAILURE in 38m 13s","accounts_in_message":[],"_revision_number":7},{"id":"5b5a444d01003b19a4e5dc4a272f17a2e513b5cb","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-04 01:08:54.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"cafd1c8ab28f88d5686ce77d7ec0f65daa1fc7d0","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-04 01:10:24.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"9b0b76714356549d0a7017cbdb4cfee90aee9bcd","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-04 01:14:26.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"2180fd318933d673b0b43e636d65d5a7eade03f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:49:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f515977c854a4499c41411770cc773 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02eebb2f6843434cb2955875336ef619 : SUCCESS in 3m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc526bbb11844d4cb49fb14aeacd7b05 : SUCCESS in 42m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d0f09ffb9e04a9b80689cd02a73b08f : SUCCESS in 1h 29m 17s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74482b07831945a89e97037271b89f00 : FAILURE in 33m 40s\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5d05301731b4da0b15fef782473dcdc : TIMED_OUT in 2h 33m 29s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da378e95cf0f423ab99b8bfadf5f8800 : FAILURE in 31m 42s","accounts_in_message":[],"_revision_number":10},{"id":"3a4a00a209b948179b569158013d69d40ac5f0e7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-04 05:38:18.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"c1d295af8772649633cf1b6ab0d951d64621c6b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 08:14:43.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9628bb0f4f6144948ab118f6681ed399 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a90b84caa974d98939c2c2ace9e28bc : SUCCESS in 2m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3217712cd64a43e9b577c78742e66212 : SUCCESS in 40m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff655cde65aa4cac998763d70773755a : SUCCESS in 1h 59m 15s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aef752c85bdb44628b505d39918ef76d : FAILURE in 1h 58m 40s\n- octavia-v2-act-stdby-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c90f51c6e6b4b6a8b3da91abad3a533 : TIMED_OUT in 2h 34m 50s\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d7118ecba31b405495479db1cb18b30a : SUCCESS in 49m 43s","accounts_in_message":[],"_revision_number":10},{"id":"a7f8e081cd94b5261ca8d3c0481c7b404ab430e3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-12-06 08:43:58.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ad5579182c85672a313b036af2b24d836d896025","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-09 00:44:38.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f49597626021385fbab1f795b3dd4c8eabfe3459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 00:44:57.000000000","message":"Patch Set 11: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/octavia-tempest-plugin/zuul.d/projects.yaml@master\", line 60, column 19","accounts_in_message":[],"_revision_number":11},{"id":"9d11e042063253c1936669c5c5c9dd9e3b33db0f","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-09 00:47:30.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"80e56f4e478a975b3e4181191da725ed5e5c6850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 02:37:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a2e628b82548c8baeefa24f8440b80 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93160f25a6ee4b64947e9499be2df9fa : SUCCESS in 2m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c554a337058c44ed9a6931daf81d83cd : SUCCESS in 44m 12s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/51db64e712a74585b9bcc48d06da0223 : SUCCESS in 44m 46s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/e2d09269b9d0457b988ea25846441e29 : SUCCESS in 49m 24s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/726dbfe5f2244da29dcaec1ee2fbd8c0 : SUCCESS in 49m 00s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/1b0997df75a3418684f987430b21627f : SUCCESS in 39m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/207c4a37323c4a5caf012917c8bc389e : SUCCESS in 1h 30m 38s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a64e57ce6ba84c71a0092c5dff60ef44 : SUCCESS in 1h 48m 11s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4a4b51a8ba6a49398e2a6172eca1cb72 : SUCCESS in 1h 23m 57s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/6c2dc146bb834d3186058dfdcefd2aab : SUCCESS in 1h 30m 54s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/e453464b69bf463f873c8da2b731c841 : SUCCESS in 1h 24m 34s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e0a68efdf9334095a0ccaf5d49cd343f : SUCCESS in 1h 23m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06c1c95cea164cabb328c9051fae174d : SUCCESS in 1h 34m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b5da15188744004bd4aad62a8e61722 : FAILURE in 6m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/289dd6e20e9b498ab541865870d94c53 : SUCCESS in 51m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/f79f71ec1328477181c981ccf98afcab : SUCCESS in 1h 11m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbe0a5312b8f4e4a9e166694ce852de9 : SUCCESS in 1h 04m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f5c83d31d4284344b6c634047ff06ebd : SUCCESS in 59m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/bf6ad8bfa6bc4b55a719eed065b6f7bd : SUCCESS in 1h 00m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb0d96524fe44e9585e19da1bfbd058f : SUCCESS in 52m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/0468d10615b34c71afaf5c661c71abfb : SUCCESS in 53m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/9578e2da1bdf4704900ad0d05e0592b2 : SUCCESS in 50m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/be9c82beeff54a8cb72556f2bb430766 : SUCCESS in 53m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/4a25250835fc48e7944494d4bd4f0124 : SUCCESS in 54m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/63d9f4beadf64b09a754510e5728fe8d : SUCCESS in 55m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/9df00e1902c443c596b67ee747d597ee : SUCCESS in 51m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/379e8ed57e0045ccbfa50e321a320497 : SUCCESS in 49m 33s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/d8b334b8df9f43f88c6b5453bf71fcff : POST_FAILURE in 1h 14m 27s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/9e22b97011df4a05bf297cd1efd8de81 : SUCCESS in 54m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e597380e1f8a4782803cd3cbd249c440 : SUCCESS in 1h 41m 25s (non-voting)\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/75e7c8d90a514f68accf67d85935f79c : FAILURE in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"891f4977048c58b995273f8fd4f1b5575e165443","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-09 03:14:21.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"c5add0252378be28c020819daebeb7980c1a8b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 04:56:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8bea35aef664ddf8e916d3110184c8f : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e301a1a13744c93b717d4a500ef76b9 : SUCCESS in 3m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4abb341282b4e8a948986ddedd21e10 : SUCCESS in 41m 31s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/a5f598fc90bb41cabe377f60b3c76dd0 : SUCCESS in 43m 57s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/7820fe141dfc44a68e08cf520893d9ca : SUCCESS in 48m 13s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/0f5d25442b634cf7b1c61f45f03d98a6 : SUCCESS in 42m 40s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/203013808b1147449072f5dc8efa96d1 : SUCCESS in 39m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b248f3077e643b48276ce758692e756 : SUCCESS in 1h 40m 01s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/34748bf4b17348f3b296daad60aaa903 : SUCCESS in 1h 39m 17s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7720960a53e7439b9bf45f3530d7f4aa : SUCCESS in 1h 22m 10s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/aa5c9b996f7343baa44e23e3ece88ee1 : SUCCESS in 1h 31m 06s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/1f1a2c28457b49d8b648c31bbe8e8d95 : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9467eee79eac4ef99255e2b075386c89 : SUCCESS in 1h 36m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf63915da8014bb8b4be385ba00a7070 : SUCCESS in 1h 35m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b11f9ea3accc44f7bd1d603724ef7cfd : FAILURE in 8m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/8582a98696b943ec9b24abe1d5ba5033 : SUCCESS in 1h 03m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/d6d0b6ba23484c629881b9870bc97d3b : SUCCESS in 1h 01m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a8d643d7a4d48dfa036be45d3b1a63a : SUCCESS in 1h 04m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2be899c5d46e4e82b7410e6cf37c9b2b : SUCCESS in 1h 04m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4466a4da3c1740f7af52c5585844d15a : SUCCESS in 58m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e70df358c34d45318401ea937160246f : SUCCESS in 53m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3db7d32c26ab45c5bfc096c36fd6e39b : SUCCESS in 53m 49s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/dc589b6a211e4912a5696afae50d588b : SUCCESS in 52m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/381436cf4951441fa6a1426da6f3bd2d : SUCCESS in 58m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/ed9f93fabaf043afa579310487dbdb7a : SUCCESS in 50m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4b1d9184e3244d3abd2fb0b997a73727 : SUCCESS in 53m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/0fc258cb96be4dd18bdac6faafc051cb : FAILURE in 1h 13m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/4a0a193c61064645a6b7a5ebf0aa962e : SUCCESS in 51m 50s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/d4f9e446c1854e46842c6efcf706d782 : SUCCESS in 52m 42s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/307c4f9c179a468191df181d71322079 : SUCCESS in 50m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1a018f76627d4c2582db133ffbc4a776 : SUCCESS in 1h 34m 38s (non-voting)\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/632509248ca94d43b55b8f6fcdb58ab5 : FAILURE in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"90dca87735a89af9aba36d7350a0e653148c392a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-17 20:22:14.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a99683cfffcc5c94942c5e18950cd0b8ec18e340","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-17 20:28:10.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"e6882e1ec9b35888c66fd66b9eac665e543a87cd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-17 20:50:20.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"c61bf5beaf07c3ad457816669f490b89d97dde45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 23:31:07.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b49be1792f43648089d0b048bf6dbd : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2e25b55902424cadc5b43544bbac43 : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54997624f5e94a9ba5e7622402a2a710 : SUCCESS in 55m 11s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/f038978e30b54efa8fdba9c845d6f1ef : SUCCESS in 1h 09m 08s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6c799726a5d94140b830d8cf6005d5b5 : SUCCESS in 52m 11s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/504ea5f141694ec5954124eb208d199e : SUCCESS in 1h 03m 56s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/d723cd148ed64293b852f85498637691 : SUCCESS in 53m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac0a64db957640d59e3ec609eb678508 : SUCCESS in 1h 51m 20s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5605bff4105c4d4689b7cca6c7b8b4e3 : TIMED_OUT in 2h 34m 33s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1e17e2c10b6d4e33a53a50063ecdfda0 : SUCCESS in 1h 30m 34s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/4bd8fda42f1d475d904c9e4d99988e4a : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/7436af7f3d4f4581a5588ed6e232babc : SUCCESS in 1h 21m 39s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a83e6d72f5614356adbd07de40b343b0 : TIMED_OUT in 2h 34m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c273b075b964b16b8ddead275a58a6b : SUCCESS in 1h 41m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c659d7fc603344ddb84e6730d7df8301 : SUCCESS in 1h 58m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/7175eee19e3549048bbe3f9aef816ac3 : SUCCESS in 1h 04m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/088a5569832740618fd3b4779a64e5d9 : SUCCESS in 1h 16m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/719dc072b2794f7b93865853a12dce9e : SUCCESS in 1h 09m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/92adde39ba384a17907c5b64cd461372 : SUCCESS in 1h 11m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7d079fd12d6342c8a880c8686a183988 : SUCCESS in 1h 08m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/312f2ec382af4de1a4ed96cfd889a9dd : SUCCESS in 1h 02m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/21650dbb65764b69af26d04f2ece3071 : FAILURE in 1h 15m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4c521f662dab46b1af311925c0960b2c : FAILURE in 1h 04m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/3cb68301fdf64c12ac7131a3725f180c : SUCCESS in 1h 02m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/96cd28fa84854bb684c78a3f567d331b : SUCCESS in 1h 04m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/baa32dfe5a60473fa373fc7e8e654d69 : SUCCESS in 1h 02m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/3c5c48c25984452bb8ebed3cd92e726a : SUCCESS in 55m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/8f88655f1e6448eda583625ed5bbecd5 : SUCCESS in 1h 02m 36s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/af779dae47754624b4a5b7f8058f02de : SUCCESS in 1h 00m 58s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/ce8127eccc6e4b1492f757404f95e7a1 : SUCCESS in 58m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/34b901f11a0f4ad6917e66e1e78fe1aa : FAILURE in 2h 01m 23s (non-voting)\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/469dee4df6df479f8bb2da7f49e4ddb9 : TIMED_OUT in 2h 32m 56s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5a9815c6a3f93ec799e347391ab057149f854493","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-17 23:46:08.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"b50d2333c8bc63011744f4fd29ff373c9903e030","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-17 23:49:04.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"21db1fac1d1830fc83259b5e3177fac052a23c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:31:14.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d88b2f346e418abf86bdbb17a977a1 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d974ad2948e5443db926aad8bffdbab4 : SUCCESS in 3m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a844c09010d4e13842784da8b70ce7a : SUCCESS in 54m 45s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/189fd1eaa6a346b39f758873b05b089f : SUCCESS in 56m 26s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/fcdf653dc5484515a7cd6873fffe4e1f : SUCCESS in 51m 47s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/b30766ced46a43e695438e87b27b92ef : SUCCESS in 1h 06m 29s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/9a5160ddf26c4c3c8cc56bb93e19fe48 : SUCCESS in 1h 04m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/309d3ee028af47bd87d1c16ca8228ee6 : SUCCESS in 1h 46m 44s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b3decedcfcbf484a936841394c6d28f2 : TIMED_OUT in 2h 35m 37s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fd459534db964ee9b6baac4120a8fe96 : TIMED_OUT in 2h 35m 09s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/fef054497b374a7fb5c48169b9ce1638 : SUCCESS in 1h 31m 08s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/96dc6c2f5fe84689b67d631f297e85dd : SUCCESS in 1h 40m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/16446eaf5cce4dadadd50443ab6d4add : TIMED_OUT in 2h 35m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/865faca755de4ce9ad5a951b26fcde81 : TIMED_OUT in 2h 36m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2027c7fcea4d49638014a3d18d94a8a9 : TIMED_OUT in 2h 35m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/519b3829448a4054b7bf9cee37bcf99c : FAILURE in 2h 28m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/b72b3a634d19445e9a78f63e1ce46675 : SUCCESS in 1h 19m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/057235a7085a402283127ff51a892ace : FAILURE in 1h 05m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25b782c8e65642ada0ed09a67376d19c : FAILURE in 1h 59m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/48a8e0477806466aaffde97274b81ed5 : FAILURE in 2h 11m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/17293e5df1734446acfb629c77bc4e02 : SUCCESS in 1h 07m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b21fb7c4a54c43d4ae65b5d330688879 : SUCCESS in 1h 14m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/2ae5d61aff864f8580791c703fcf4009 : FAILURE in 1h 05m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/e7d065c1c7c44fe4a795fd6ae95f38ac : FAILURE in 1h 09m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/a0bf921bf62f4e88be69cd3e8f7cc1b8 : SUCCESS in 1h 00m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/2ccebc08fad14aeea32c04a8fda7a71a : SUCCESS in 1h 01m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f3d59d5209714aa884fbb843b424bf23 : SUCCESS in 1h 01m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/df1bce89a1fd46d0bb1d2fc5c5b8b145 : SUCCESS in 1h 06m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/7b4985b63da445b0a34a3a943797687c : SUCCESS in 54m 05s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/158c529ca33f4db1974fe98b4d7c8322 : SUCCESS in 54m 12s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/c00b521623444618b10ec521bf55b563 : SUCCESS in 54m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2a6162de82f24ff8bc611b064908941d : SUCCESS in 1h 45m 16s (non-voting)\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9cb076f326be43d69437195dfbf5080b : SUCCESS in 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a0c2c04c0873d9a9ef733fba1628e4d034688ae7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-18 17:11:35.000000000","message":"Patch Set 16: Code-Review-1\n\nI don\u0027t see any amphora on that second node.","accounts_in_message":[],"_revision_number":16},{"id":"340511bc9fc83a072596af077a4f51247577b87a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-18 23:21:07.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"de354f921ed85e10d75a9ae3d31c72265dd2ca0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 02:02:15.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322bf974b9c04ac1b8ddc530af7e6a6e : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aadc715cef7406887f4785e83432ac0 : SUCCESS in 3m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/210fc73ca0df4e6482b4288c25e2971d : SUCCESS in 48m 01s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/74a0f5058cbc4da099dc9738bdab91fe : SUCCESS in 52m 03s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/26eeae4ad156451e8d274ce67d0c8a3d : SUCCESS in 54m 07s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/e15595e7a515489694ca5917d1ad2245 : SUCCESS in 48m 10s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/5655ddb7c783481c85cb87b029dd7600 : SUCCESS in 1h 02m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/789c2911d1c24d4d8d021c7df1ee24c3 : FAILURE in 1h 50m 45s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0ce753da7f034a0e99138410b1fcdf97 : SUCCESS in 1h 40m 13s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/80c02333695d451ea69834b184106df3 : SUCCESS in 1h 29m 50s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/33428c53d9b2427a842ebe442094dd2e : SUCCESS in 1h 44m 03s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/860ef9eaf928449f9d8bdcad77f2e77b : SUCCESS in 1h 26m 20s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3695640ee2a7423da969d02e904dbe3a : TIMED_OUT in 2h 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ce1516f280f140fb88a47ed073224a52 : TIMED_OUT in 2h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bb1950e378243048f7c1d34c9c6eae7 : SUCCESS in 1h 56m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f34ef52db58c4c39a82b070e828de133 : SUCCESS in 1h 07m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/5ccb034472df4e18bb850c6f4e3864b6 : SUCCESS in 1h 09m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/09cd455cff1b40c586dffc135c4bd06f : SUCCESS in 1h 12m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd2add48a9f849cf82b2b367ca16917c : SUCCESS in 1h 18m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/41599d64c09c42f9bffaedac39fbc30f : SUCCESS in 1h 24m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8b2b80acbf9145dbb42211c9a7da51ec : SUCCESS in 1h 18m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6709fd7311f14286a9e44dab4e2eb9bc : SUCCESS in 1h 09m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/1501a2ee50ab470b9671b311afb9f64e : FAILURE in 1h 02m 16s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/1a96d631082f434f92a94c194b909e0d : FAILURE in 1h 03m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/dab402b261fd41e58a4795a26d9bd75a : SUCCESS in 1h 01m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/4a3a467dd26a439e954ce24d84e4f3d8 : SUCCESS in 53m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a453af8f9e064e03a4edd8c17c6638f1 : SUCCESS in 1h 01m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/e8818ed52ff64fee8533de778044f7b3 : SUCCESS in 1h 04m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0b13cf5ec6f347f4ac5423f2d3dbcb47 : SUCCESS in 55m 32s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/90cd53fdf3874b9b9740a98f808cd502 : SUCCESS in 54m 56s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/8a62738d479046babf186cc4efcbdddc : SUCCESS in 48m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e43fb2210ccb4b1081a9a2257f26814d : FAILURE in 42m 10s (non-voting)\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b4dbf1ba53614980bfa32d2a97cdcf61 : SUCCESS in 1h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"afd6fb89f6baed072e3c4d1a97889b56afb8c3fa","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-09 14:45:15.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"e1ffbf703fc23ad02b0b2c00c3412ce37edd00bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 17:26:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8bd80f5d3d490ab42a68ea69696a0f : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d64fcf629442c8abd3405046bca14c : SUCCESS in 5m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/263bd29dec664977beb080741af9c50f : SUCCESS in 47m 05s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/15e5fb8058654c0c9a3670c524d6b32f : SUCCESS in 51m 11s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/ca5dc127e6fa4fdca902c8a7734538a1 : SUCCESS in 47m 53s\n- octavia-v2-dsvm-noop-py2-api-stable-rocky https://zuul.opendev.org/t/openstack/build/987305b60f8d4f2e9ac3c07499e6962a : SUCCESS in 49m 02s\n- octavia-v2-dsvm-noop-py2-api-stable-queens https://zuul.opendev.org/t/openstack/build/f5722c9f4f0c47ad8c69bba8473120d0 : SUCCESS in 52m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c92b49c4f6948f38a60fd45ec33121e : SUCCESS in 1h 39m 26s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c18c69351e31419d8044ac6c3fc369b7 : SUCCESS in 1h 29m 54s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/59310117fb614afe9c2e9afde356c5e0 : SUCCESS in 1h 52m 42s\n- octavia-v2-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/3ef6096bb46b4b62a64e34a7e7cc836d : SUCCESS in 1h 31m 30s\n- octavia-v2-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/b29b92ca622e48a6a7dc8d5751cc7942 : SUCCESS in 1h 31m 04s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0adb1ad45d924f8da712ab408e07ac56 : TIMED_OUT in 2h 35m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/36411cd7dea0472991339294c371a383 : SUCCESS in 2h 22m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa7d3150cfd845248fd725369ff0abb9 : FAILURE in 1h 43m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/130d62f435224551a8fd3f5b2265c047 : SUCCESS in 1h 23m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-rocky https://zuul.opendev.org/t/openstack/build/8538cb96fd6843ff9f0edd93bc8f581f : SUCCESS in 1h 06m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-stable-queens https://zuul.opendev.org/t/openstack/build/c3c30f06f1c34155af881b43140bd8c1 : SUCCESS in 1h 10m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ff5fd6418774576b3c68c08c1160724 : SUCCESS in 1h 06m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/71219c021db441fcad4c0abe70645975 : SUCCESS in 1h 06m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/bf00092bec7645f19677629a82e24e4e : FAILURE in 1h 35m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7e1b1b7ade20463a97d864aa8be55c3d : SUCCESS in 1h 06m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/d32c005b3cf94f1cb8e4adbc6040c9c1 : FAILURE in 1h 07m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/3879e8f2a3fa44df96fb7b5df6c83aa3 : FAILURE in 1h 08m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-rocky https://zuul.opendev.org/t/openstack/build/dbe006e378ff4b8ebbcb4526876184fa : SUCCESS in 1h 01m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-queens https://zuul.opendev.org/t/openstack/build/2e8ed79047f043fbb13f0cc0b9e2bb66 : SUCCESS in 56m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7d518e14039e4646a8d829026debcb22 : SUCCESS in 56m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/75b8b3e8de604babaa68e4784a368240 : SUCCESS in 1h 00m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0a72836102f8453ba79fa15685960a77 : SUCCESS in 1h 07m 20s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-rocky https://zuul.opendev.org/t/openstack/build/00b89563c3dd4d5b8b10194b4c8efd49 : SUCCESS in 1h 03m 24s (non-voting)\n- octavia-v2-dsvm-py2-spare-pool-stable-queens https://zuul.opendev.org/t/openstack/build/99af8da290264dd7b23cc4ae9fb2354b : SUCCESS in 1h 00m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/31625a874bf749d1841c7868c695c4b2 : SUCCESS in 1h 42m 19s (non-voting)\n- octavia-v2-az-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/82ee47cced0d4810ba6e44584078ce15 : SUCCESS in 50m 04s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b4eade114ef6f25876914856eced7c0c879fc7b7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-07 06:49:27.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"868c1a8f714cf2833ddb317b14d08c98b2b903b9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-03 18:43:19.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b1df10a93d157ccb22a8ce85af58574553215d52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-03 07:37:34.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"3135dd643105ce411487f41269a0063311712640","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-03 08:52:29.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"c945febe99bd7e4d5b1428e9084571b4716b33e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:32:15.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f2221852be9445488007699e2606008 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d1f449c33e94f3fb0a9b8ec7a2cf3ab : SUCCESS in 2m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90c6353ed6a24494919651a5f4efece7 : SUCCESS in 1h 13m 53s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/d0bcee5ebc7645e59d7b5d91b0194321 : SUCCESS in 1h 13m 08s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/89ca76b1c44a46e795ed680334235178 : SUCCESS in 1h 12m 16s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/dd073e750a774b56b00fb7acc5355acc : SUCCESS in 59m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23243c913bb44bfeb2856f6342e0df00 : SUCCESS in 1h 36m 18s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0d3f501f2bed435eba64e7640408c19a : FAILURE in 23m 05s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e31d995dd35e4bcb9a6cd1da8c8770ad : FAILURE in 42m 51s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/afed4e5339834175b958bc905936f229 : FAILURE in 26m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/513f043f23914a37a6e8636b793cdbf6 : SUCCESS in 45m 01s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/405b727b61d94c9d9c25866033e9dd77 : FAILURE in 34m 12s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3ab743a472bd48a7963a8bb4f144f794 : FAILURE in 25m 40s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/e2a211c8b3d64fa8a58eb8b644beb5f2 : FAILURE in 26m 32s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c3fe7b9268464eccb47a3611009d4442 : SUCCESS in 2h 05m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4fe2382fba98412aa359cf8310648744 : FAILURE in 30m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2db78ecd07e84be281b311fa02a237b2 : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e33f7bcb91f4471688383d1cf46a5f56 : SUCCESS in 59m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/fc9887c2b24b46649f45a1c13cd1a458 : FAILURE in 46m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/17528d6619084aa38957b04abe06474d : FAILURE in 41m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1bc91208492f4dc5affa328d4c233090 : FAILURE in 38m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/47601d246d194ae8a8a2a181e138fa2c : SUCCESS in 53m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/231809bc270343cbaa94283796c4e75b : FAILURE in 58m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/03e7d39c9ce44c0a9a10c682dd1086f9 : FAILURE in 33m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/6028f264716e446ca2a5f620e9157447 : FAILURE in 32m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4da79ca136384fddb7c131aba54a8ffe : SUCCESS in 1h 57m 59s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b00ac2ddcd0817442c1a02011a72d446a1b6a040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-04 13:51:34.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"edceaf76aed57da62e933cf9daf33eae6ea0f1ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:12:17.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bbe4514e5e4acbadbc410067fdbc90 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44eac5d2796746c3abdc761baacec1a2 : SUCCESS in 3m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8839ca35c50640478becd7075489d5d1 : SUCCESS in 1h 22m 27s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/81b28d4f5ac541fcbe02a1eb2d520f7f : SUCCESS in 1h 24m 47s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/bb5f3754227844e580c02e7182e5a6e4 : SUCCESS in 1h 23m 24s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/37057367e93a4d96a1a735876eab78da : SUCCESS in 1h 02m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af385bd79a224a14b6782e9d51b8e277 : FAILURE in 2h 16m 27s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7d4e7cbd073148609728d4be073cb009 : SUCCESS in 1h 44m 11s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0d7bbc320a7b4ee1acea195f35d5921f : FAILURE in 28m 22s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4a043a8fd35d48eebb1ac2eecb059697 : FAILURE in 21m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3611060ef70a481e98adf2b12e973d5c : SUCCESS in 41m 15s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/d5d9714905324bf0ba75a88aed17bca1 : SUCCESS in 48m 34s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/7a0874f78bb94f29b039db8f66bc176d : FAILURE in 38m 13s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/8841a352600a4883852226d41cc47f95 : FAILURE in 27m 29s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/444ed8c3a0d949a69c71e296000aafc5 : SUCCESS in 1h 48m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/45e78abbb20a45fba2fa3907b8543c7b : FAILURE in 30m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/13676722701642c49e46217f35d46424 : SUCCESS in 2h 15m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1b8b6adc90b48f6ae3c88f060b6b19c : SUCCESS in 39m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/045bb86c18e9433daf95c4c4c494c2b5 : SUCCESS in 59m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f8154c66b40d403db26c4e8b90bd5e0b : FAILURE in 23m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7a032799749747c4a48c9fabf1470a85 : FAILURE in 29m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/079ae3831b18443f8e7a9f145401e332 : SUCCESS in 44m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/90f975201d86427a8f3cbf1407b72b88 : SUCCESS in 43m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/79acf836ab0a47e68878b1ac105b82e1 : FAILURE in 27m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f1a0bb26d3b049fc965e7d686fd99b48 : FAILURE in 26m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1da4188d55c54ad29f8be57d1b34f33f : SUCCESS in 1h 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"00ee009cf63eb03669b90a20bb33c2555eb83e73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-04 17:18:39.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"017da2f474638291f274d09827c79da1ff27f46d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-04 17:19:35.000000000","message":"Patch Set 21: Workflow-1\n\nwip: need to clarify in which jobs the test should be run","accounts_in_message":[],"_revision_number":21},{"id":"99b4c6b2cd1dfa86866e159f6d39c4aef034600a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 19:56:01.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80edda1d8b04d8c985641161fcae724 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e286006f6e24f1ca2858cdaa781378f : SUCCESS in 2m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc0d55c34b52479d8ac6045252ebc946 : SUCCESS in 1h 20m 02s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/b4565a2c19844296961ac9860256fe2d : SUCCESS in 52m 27s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/16c13a609c0b4a54b0fd4a50c3312bdc : SUCCESS in 1h 12m 11s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/a765720bab3241cd9497cb37f72e7879 : SUCCESS in 1h 06m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/041cd85c9c8243baa9ebc09cadf6e12c : SUCCESS in 1h 38m 32s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5657bc0aad444a5fab69e8ee0e904e0e : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3527e39a51104e4bb9f90699f440ab6d : FAILURE in 24m 16s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1caf6990eb994dfca1809834f39afbf5 : FAILURE in 26m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29393978ac1a4738b2ee6c4f85f1302c : SUCCESS in 41m 54s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f60da874474e4b5fba11430288beba50 : SUCCESS in 44m 59s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ef400c5a2cc4476b9f5054bfc759168f : FAILURE in 29m 54s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/447581b5d55c47528fabbbf5e3179143 : FAILURE in 29m 29s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/95a0a7902a8a43b1b7bd91aa30598571 : SUCCESS in 2h 06m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/513c2268914a4b1dbabfe048b5ab0378 : FAILURE in 32m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/480a312178744e3dbd7ca92db762cb5a : SUCCESS in 2h 07m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73b4f8f26ba448aea4300ab3732bc7ae : SUCCESS in 36m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b65f07a69f17485e82c9c5340c90ad0f : SUCCESS in 1h 00m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/af4ff906c21b432c8ab446d261b906b4 : FAILURE in 38m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/19854c86c124498ea9782320c2859218 : FAILURE in 25m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/60ff0e3e161d4aaa8f6e87011e42c6c1 : SUCCESS in 39m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/43a2eac8cc1340e59fd700898141b257 : SUCCESS in 39m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/a668521091694fab86fe453b2ddfe427 : FAILURE in 24m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/57e08fceaff642fcb3f9f716be56c678 : FAILURE in 24m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6528a9e7c3d479590718ece2f556d22 : SUCCESS in 1h 48m 13s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"1b82c537f1f3c3795622b923a6cdd2c50f28f3ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-05 07:37:31.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"cfb4ba455b70a489f6b4cda36295af0aefec3561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:01:42.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5654a9841d4c358c8475d5fb94da87 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd92c68d04134d5f957a3367c4312efb : SUCCESS in 3m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dac6cac1bb2d495fb679b0f15375561d : FAILURE in 31m 47s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/7dd66a834b394e00abf524084c766c46 : SUCCESS in 1h 20m 45s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/7d9cc09b9a5f4792b445265604404089 : SUCCESS in 1h 19m 58s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/069bfd5192cb4b79a2834f007e353f44 : SUCCESS in 58m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ab639e90a8a4e2ca72c268a3eefeef9 : SUCCESS in 2h 07m 52s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e19d4f43945a43e29bfb6885df8fda8a : SUCCESS in 1h 25m 44s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/41bdbd5aa1a143be8a1eb7a768cc45b2 : FAILURE in 23m 51s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/321173be8d164f16beec01786bea34b5 : FAILURE in 31m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17f1e8a5d48346939e62430890bc478c : SUCCESS in 36m 35s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/fc03a7de1e974907984724bb9e1b2192 : SUCCESS in 42m 09s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/82a1a83d68204245806b7b9ed3c7331f : FAILURE in 26m 15s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/8d065aef9896410ea741d5b6c20a2276 : FAILURE in 37m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1eb6c1d2692544b9a7b83cdff9c9bc73 : SUCCESS in 1h 23m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e66535061e70408d88eafb00f14ad9ae : FAILURE in 29m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37e782ae219247b7850c49fdc0e5d6c1 : SUCCESS in 2h 19m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a8de399b0c64c448a78754211f46e49 : SUCCESS in 39m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/4a5543e9e0654dc1a17016b79bdb0b35 : SUCCESS in 38m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f19ac4862cca48c0b5e80166e50b9c76 : FAILURE in 24m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/e21669be7207401cb8f954435bad4e9b : FAILURE in 23m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d577da52e58b40fc89dd07a8d0c29f72 : SUCCESS in 45m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/6f07804e992c4075849e240fb2e71ff3 : SUCCESS in 36m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/367e362f63654578b1e12fb945b1af9c : FAILURE in 24m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/9d56a2a80dc34a3dbe7ff939d511a9c1 : FAILURE in 22m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ac1e1e3d7cba4ff6b8b235fa49d291cf : SUCCESS in 1h 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7855e2b986f357301290abaf558e4c17214c69d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-05 12:00:36.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"d01faf76f954bcf72cd38dfe9ce35bb4ca7cc9b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 14:20:29.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc67fc0f234d4dae8ee757311ee6ef90 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d5fe10ef3824c07b282edc16d6fa62c : SUCCESS in 3m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f7c7d6693bf45ee91c08677379376d4 : SUCCESS in 1h 19m 30s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/0443651dce6042078732b8f07ad67587 : SUCCESS in 1h 20m 03s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/aca458b7ddde41b284d38ac0deec76b9 : SUCCESS in 1h 15m 55s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/31663e90cc244b20bc52ba0a99359931 : SUCCESS in 1h 21m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/480aeba0a1f94ab4857a5c847be43cf7 : SUCCESS in 1h 52m 18s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f260564de0e54b48b31bcf1287ba63f7 : SUCCESS in 1h 47m 31s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e96e3eb99d594e2e8d71b25b51f1eb3e : FAILURE in 22m 48s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/2005694e48ac46989db6715e167fe40d : FAILURE in 22m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea612f31de534534aab5eddaf3ac26b9 : SUCCESS in 40m 58s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/484a3c8e42c04196b89bbdd2e4622018 : SUCCESS in 33m 03s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/77689fa0e0a04da9bc92d3dc8923e60c : FAILURE in 30m 56s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/267a26fb5c314e60ac2d9e5b949ec810 : FAILURE in 25m 59s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8213abd4b8824e99ba225ce12350043e : SUCCESS in 1h 47m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1374ff01e93f4c89bd2e375a680f59cd : FAILURE in 33m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b0dbacaa5dac4b3b92c7d888e92c125e : FAILURE in 2h 14m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a31d940721b4e84b0a7751ef341fe01 : SUCCESS in 33m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8a8d2645ddd04863a6fa214068179246 : FAILURE in 51m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d021eced0d9142969cf5ffd397a8ebd8 : FAILURE in 29m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/304c0b3666aa475c95261989342d2d58 : FAILURE in 29m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a56d7565d92c423d8871e88d60644e5b : SUCCESS in 45m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/754e07ec60ac4449a66ebc6530b38983 : SUCCESS in 43m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/861109d93ff44e939d2a5839a8629ed6 : FAILURE in 27m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/57edf87b72914f5ebf6db85a6f22f063 : FAILURE in 23m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5fbbb22a369e4f96a92bc0a2f07b89aa : SUCCESS in 1h 53m 05s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"ee90aa54329b35b94b4f9a812f8f880e53f08ef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 10:34:32.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"b2b498f1de53333ebdc69f5191a659a3d1e8dd1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 12:45:30.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c00dda43dc4267a0ec3212b17b809d : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eb8a3f90b7f47239d4a9c162834b419 : SUCCESS in 3m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d56445d8896041b8959cff13d4768fa8 : SUCCESS in 1h 23m 55s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/50be990fe33f49b791c1a35408ccb797 : SUCCESS in 1h 14m 42s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/8ade36cf79c244b7818adb91133e7339 : SUCCESS in 1h 10m 45s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/011a6e12a5ee4b70be2aacaae668afac : SUCCESS in 1h 14m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68e63021742c4d4fab8609c7cd39f387 : SUCCESS in 1h 42m 21s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2ba57431e17640cfbc38233afcc2a33c : SUCCESS in 1h 20m 22s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b40e0888ed214a099531226c3476b3b5 : FAILURE in 51m 01s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/aa1d574718dd4f708d0771b00dbbf9c2 : FAILURE in 21m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/08e1e68ebcb9468bab40044eb30512e4 : SUCCESS in 1h 00m 39s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/e9138f956cd74179b5bc45b582a77876 : SUCCESS in 44m 47s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/caa1616066f34cd190433b350da6ce6a : FAILURE in 29m 58s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/e7fd39ab841d46c7bd128b2d63107074 : FAILURE in 24m 45s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b7a47835b1b546e1bbc0be53ba120f83 : SUCCESS in 1h 44m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9604ff8b00c54b1881e3fe56f5704795 : FAILURE in 30m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/842d793b835b43cfa1356bafa244436c : SUCCESS in 1h 54m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23d9d684a6744bcfbabe708a87112b55 : SUCCESS in 45m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2dfced227bc54a549f5f1c56a86bafde : SUCCESS in 40m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/308b583c70fa4c2ab77ca2f600a6cb3d : FAILURE in 25m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/56566deeee0041afa354a006bb14fb11 : FAILURE in 23m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/648ae45d17764a6d92be9dff476b4915 : SUCCESS in 37m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/77c40abbe7e84e35894bfd0d56e27b78 : SUCCESS in 1h 04m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/bd557304e58542819b6cb224e1afbf10 : FAILURE in 23m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/747721565cfc4e939aaecc1fb3d9deec : FAILURE in 26m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b12bc988bca648a9b7ba92474d240c9f : SUCCESS in 1h 55m 02s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"20aee5ce7e8fdb75be171e65130fb221e2f25248","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 12:56:16.000000000","message":"Patch Set 24: Workflow-1","accounts_in_message":[],"_revision_number":24},{"id":"0a65aa5bae46aa10b56be7642f77a75557f063fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 14:56:11.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"6e162bbc1e4218959fd4339d4c8d8868b8c6273d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:05:08.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec38a186f3c4f85ae17d2ec6b1f35c7 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10dfa52d5b42465ba22756605a95b4dc : SUCCESS in 3m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/73435d210a2a4e33b12968d299b19bfb : SUCCESS in 1h 12m 43s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/716ee698eb954aaf9ec8ff098318de22 : SUCCESS in 1h 11m 56s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/056eb9985bad43da8c1ff89ff9ac6780 : SUCCESS in 1h 12m 58s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/3198316a25f94069a769681b79494d1d : SUCCESS in 1h 00m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db86520b24644575bed836f9714e6ce3 : SUCCESS in 1h 56m 03s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ed49e87afe4e415390162176e42c44fc : SUCCESS in 1h 39m 50s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/678f84466879460f8624015c9f55e366 : FAILURE in 25m 39s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/5025f6c3fd0c4c4ca99cc3bf09f3eba2 : FAILURE in 24m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7e7668c704544515a7c0e952e439f807 : SUCCESS in 45m 28s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/761b8d8a1a73468f98aecf944a9f548e : SUCCESS in 39m 27s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/91cf390e8bf444f0807701f0ed78bbde : FAILURE in 30m 01s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4c26ebab4ca442c9b2536241cb94a4cf : FAILURE in 31m 11s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/306a84dfc61a41598abdb5f573f5b4d3 : SUCCESS in 1h 37m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3eb9b231ae244eef875914bad858e5a5 : FAILURE in 29m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c28b5e0ed804acfa57e55ac86ff5bea : SUCCESS in 2h 05m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e6e145e374240188075b686fd82661d : SUCCESS in 44m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/94c0e342df1a4cf58953617bb7574a04 : SUCCESS in 42m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/dae6cf6e76474c3ca31b087e29a10c17 : FAILURE in 21m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8733d7e0c5f449b5a1d560b3c3da0382 : FAILURE in 45m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b01b6021e928405cb2a9940104600ba9 : SUCCESS in 42m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/57d528d620f1440eb7afab842960bff3 : SUCCESS in 41m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/0984b74c9fcd4ba1aff422a81e3e51ba : FAILURE in 23m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/ee98fc1d034a4600bcc567963e30d9e2 : FAILURE in 23m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a973d913131240deb779df29f28d42c1 : SUCCESS in 2h 02m 02s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"23cd2682a8270caaa521e149dac3ef649914081c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-15 10:02:03.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"22b328d2154b03ac65f5abd2931261cbf47da6c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 12:43:44.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19a434c43bb4adbae0b80556ee3d2cf : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/585e1c20c80c44288728817d1916ccc4 : SUCCESS in 3m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d86722717944a98a3448dff09278626 : SUCCESS in 1h 12m 42s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/3f0e66987505463e865eac43b0a68234 : SUCCESS in 1h 09m 18s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/a766e44f1b19414da78d8775d505c9f0 : SUCCESS in 1h 11m 01s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/348fc6ae2fbc4415b2ad9587ac80f3be : SUCCESS in 1h 00m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/878a9c0d600e4dc78aca0b6cd13b9400 : SUCCESS in 1h 44m 29s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/eb1a38426bbb4e808fec2147f36d730f : SUCCESS in 1h 37m 01s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ab108f6b92ea41e5af1a08e3de9d9009 : FAILURE in 27m 27s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ab495d734885428eb4958755c715c00d : FAILURE in 22m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/22594889db04463b892eef387f4bea69 : SUCCESS in 41m 31s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/7a380da222b04a8890a6791635638bad : SUCCESS in 45m 52s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/8836901aa4bf4b9095b11644f16e6593 : FAILURE in 27m 02s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4395d7bb9f6b4d1cbb79a0b775f1760d : FAILURE in 26m 53s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5db838bfef964461a1b8263ef4ffcaa1 : SUCCESS in 1h 46m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/815c9c7884424d96bb78f48323ce55d6 : FAILURE in 30m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84bdf7d2cc6c4aecb8e1cbb5046b586a : SUCCESS in 2h 30m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aacb79547ffb4b42bb8177b4e3aba8fb : SUCCESS in 1h 58m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f4b398ca68948c995becf9fac36ea06 : SUCCESS in 35m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5723e1d6a3c844e8b047d270d137f461 : SUCCESS in 44m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9ea16126ee094893b7e2189754bad4ed : FAILURE in 22m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b489a8d2899d49ed9047842cdeaabeec : FAILURE in 32m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b274e4cbe0634ce5aa0a780e806d383a : SUCCESS in 38m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/4012dec1db6548bab91e5bc58d25d65c : SUCCESS in 39m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/fc5cf1f0210043b38a76c8ab159d0216 : FAILURE in 27m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/b514799d05f44c57b8ba28ef07e1e102 : FAILURE in 35m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/72343c40db144434b6136aed2a1dc0f5 : SUCCESS in 2h 07m 19s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"ad0e57ec2d648e08c0ff8c626f879f3ff9d93d74","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-16 09:56:41.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"4dd4212d9fc3e2a4054c598d517edb590f9b0002","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-19 15:35:46.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"33954afc27b0406fe9ced48d6f6b5360604240b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 18:03:52.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74b59ef437c4c6199cde846df6f5035 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d8a356f458440d5a08db95b1066c5e2 : SUCCESS in 3m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a03e6efbbea04cc2a1c82a12164a3c43 : SUCCESS in 1h 12m 29s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/26e235fdb2234dfaa7125ae68a156247 : SUCCESS in 1h 57m 25s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/ac274082e5004f03b328cc5b9c217d58 : SUCCESS in 1h 22m 27s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/399c01af14f6436da425caa3e96fbb3d : SUCCESS in 1h 00m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/142b4e07c8df4062aaca9c679e4ef965 : SUCCESS in 1h 48m 20s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7cc4cea728554428af16240517ec27f4 : SUCCESS in 1h 36m 39s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8d58489573924937adfc35ee2cc140c8 : FAILURE in 25m 21s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/08b1ba3040c94a5da6f8f8f3769ec44e : FAILURE in 27m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6b0d3bf599b24292b7bd8e4c39d94522 : SUCCESS in 47m 44s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f534b1c084124697a58fa7ab0e9983c8 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/0dd17e3f206a4641bc2b30e21276ce46 : FAILURE in 30m 11s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/69e2feb6051b44a7a94f46e691243f63 : FAILURE in 28m 56s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/12c33273ff2f44039d1bfa5fe6b495ac : SUCCESS in 1h 45m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fed5f8b1957b4995b8953aa3cd8975d3 : FAILURE in 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/83f4bbbd1aae47fcb128b4d2a6adad63 : SUCCESS in 2h 10m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37218cca2c4c430d9dc1bc880a9413ae : SUCCESS in 2h 13m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9e8626261344fbaa1e2cefa4fdf738e : SUCCESS in 38m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7fcf28506b2e4cf9aba3496677fbfae4 : SUCCESS in 42m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/30d16a0e50ac4c59a6be8afde2aed393 : FAILURE in 26m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a5fe675948c44fbc9cda744f52966ca5 : FAILURE in 38m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6101c4df1b474b13b2d8dd000b00c104 : SUCCESS in 58m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2cce16b393b748f6a002e77840dfebe9 : SUCCESS in 39m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/ab2e8f2708344f56b24e8141ad1ce131 : FAILURE in 26m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/4c8924b5389a4dd1a58d8bbcd7eafdc5 : FAILURE in 30m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/49b993c800e64eb8945d60c50514cd44 : SUCCESS in 1h 58m 14s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"5975e110df8a5a26cdc73e5df74a4e43169e70b7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-17 07:36:21.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"2fe00fde8d70e6f2439529903e2aee5d8b545bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 09:53:56.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/044cc368d64948f4a13ccb32332cc6bd : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e43514c84604d8f81d63cc44cbd5a56 : SUCCESS in 3m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c89d00a2844440c1b68a6833dacb5e52 : SUCCESS in 1h 22m 58s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/b4b7c1833b1842369422d796e4caaa2c : SUCCESS in 1h 22m 47s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/3d4cc8ec2eb04286b21b8a13ba8d3b31 : SUCCESS in 1h 20m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/186852db51e94da29cf959b82fb4165f : SUCCESS in 1h 47m 46s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3900fb6915334866be420b1b537b070b : SUCCESS in 2h 14m 21s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b4f42526c311469ab457633c543c48d3 : SUCCESS in 2h 05m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/69deab8456f54e63989f0c78dcb34e02 : SUCCESS in 45m 47s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b224ed496b0048adb12fdc6f7c494874 : SUCCESS in 39m 55s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/66d7f11f668749bf8e2f4e6fc2a3ffb2 : SUCCESS in 45m 14s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/40eb99bc2280433a94d5570094343979 : SUCCESS in 1h 38m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/09e2557a4ed044c6bd1f23ae736b7902 : FAILURE in 32m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0611816173a4fb8a7e1a358db591bdd : FAILURE in 43m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ca2cb981f994a2faed3236befea602c : FAILURE in 51m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28208310721641d6b29d243654888a7f : SUCCESS in 39m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a9fe8b87cd3f4708b702c49cee55b763 : SUCCESS in 37m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/66c085ed25a5452489329f21fd5a8023 : SUCCESS in 37m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/51ff3a2bbcd344c2b367f9a0ed50c943 : SUCCESS in 38m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/752239ea0dd64fcebafd721f87eba2f9 : SUCCESS in 38m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/9e2b3d6934b149f2a7c89e0ff68baca7 : SUCCESS in 38m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c85a186aa2ab4c479f88be417d19f604 : SUCCESS in 2h 00m 34s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/cf5360c13a2043feaba207f35335d18a : SUCCESS in 45m 04s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a9c2f70f7e9a46da950c2124e89e39de : SUCCESS in 57m 37s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"f857b44cd5cecf3497168c72b692d43e40bf4bff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 08:17:16.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"68e39d114d12ca9b4396c5654ce3d3e998ceb9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 10:36:32.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5692c868c6b3417ca253167abf6e0e0a : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb483455fe4433aab99e18780fff593 : SUCCESS in 3m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0468f8d89ef64138a3c5094ed1e3375e : SUCCESS in 1h 21m 29s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/faef161a989d4ef992130c68ddc68b1a : SUCCESS in 1h 13m 29s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/1f464c7cf04b44f1b0cad13c65be08e5 : SUCCESS in 1h 16m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3cf65774de14341ad8907238b75aa12 : SUCCESS in 1h 43m 51s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1a820b8a8121422d92504c25ee632a4a : SUCCESS in 1h 47m 05s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2e9f28ed3c844a26a3dbbb51379b5fed : SUCCESS in 2h 10m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68b7b38bca734f769b467d705acf434f : SUCCESS in 44m 42s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/fbd63a94fd004c4a930362993f5ef541 : SUCCESS in 43m 08s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/93dd9bff2fba4f2bae7d51b76395ccea : SUCCESS in 42m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d06a0f8ccb594ae49e6a757427afb53d : SUCCESS in 1h 41m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b27981b62dec45e9836f34d16109156b : FAILURE in 32m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/39dbab1d3eaf4077bdd346a1b2f85895 : FAILURE in 49m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/43552e6cdf194b3381e15fe527dac144 : FAILURE in 50m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84eb254a0d074750a3f3382a6297f694 : SUCCESS in 43m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/79117355a00c4a64a4e22ef9957d741e : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6581b228b54e4c34af6c32a3fac831c2 : SUCCESS in 42m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1dd065d74ade420cb09a7124ea51d207 : SUCCESS in 41m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/5890f7be28b64cba84d1a84ecef47d89 : SUCCESS in 42m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/22ae1e93f5914cc48837411de35e1120 : SUCCESS in 40m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cafe4b81bca545ea91b133950ad3f8a4 : SUCCESS in 1h 52m 10s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a2d81b31227b4e968c16f43e05391864 : SUCCESS in 42m 37s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/628118310fc14a9b9ebf7ca717ca9943 : SUCCESS in 51m 20s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"037cc5e9b2442a8739f04515331209ef825de497","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 10:45:27.000000000","message":"Patch Set 28: Workflow-1\n\ntwo-node job failed","accounts_in_message":[],"_revision_number":28},{"id":"29d90f8280c0fa5d1e2a6d04e48fffc0b048d206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 15:07:19.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"f7946b397c1e8cd1ee28b062d6444e55276dd112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 17:47:15.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81279657dbb48cdba8ad4d9af891006 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb2858465cc481d9afaac082c1854ed : SUCCESS in 5m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bdaab0fffcbe4872ba7b7aa3e9256e1b : SUCCESS in 1h 17m 35s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/3d68bbf8496145c79c784e11616b73bf : SUCCESS in 1h 16m 13s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/49e6bad516c743df900d9ed58d4bb5bd : SUCCESS in 1h 16m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d53cf2635aea47c187fe41a993195b0b : SUCCESS in 2h 07m 47s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/81ab5c480ed841e4a713027b270059ce : FAILURE in 1h 58m 27s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bc6a794e26f54e3f8fa75329236d84fa : SUCCESS in 1h 47m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b6d12589ded4c31bbdfa8c74192845e : SUCCESS in 45m 29s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f85972dc54824943853f3b8bf600aa6b : SUCCESS in 45m 14s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/9a977a184cd2452a99dab42ffc67d644 : SUCCESS in 43m 50s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e44cfa0c60d344a1acd21c5d18b7aab6 : SUCCESS in 1h 48m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ffa3715057424634b0756c2b9f55f5e4 : FAILURE in 29m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f03c6022b1d4dd49ba3484933700b15 : SUCCESS in 2h 11m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0dca2b29364247459c64e076a5cdc34e : SUCCESS in 2h 22m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c186d322487440baaa5b2d3f0cd36e6 : SUCCESS in 38m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e9c1ca642984476784f9354aaa2eb124 : SUCCESS in 38m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0d8e4d398ab54c5a83bd31e0cd2a5ae6 : SUCCESS in 56m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/46c15f0b6ec346eaba47861d53314de4 : SUCCESS in 34m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/817ec67f168248108e8ea0b8807be71c : SUCCESS in 39m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/a27e5fb1e30a4fbd85b061502792cb3d : SUCCESS in 36m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a427a1d3a85f4897bce4a81f4402cc04 : SUCCESS in 1h 46m 25s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5db2c607461d4e5ba3a576e27216dc73 : SUCCESS in 45m 04s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/650a67e775cd47f6ab01a3eec2fea08f : SUCCESS in 54m 22s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"27df6195b8ea0d61709aa8b5bb372cb787b11fb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-24 09:58:57.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"2bef61a86713601763078458e20e1936ee7b4603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 12:41:45.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f9ce31f33e4642a8bdf8392485e224 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88d17928f28544b1911fe0670ad2af7f : SUCCESS in 4m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/061b104d9ccc48e49057f2b10f13264a : SUCCESS in 1h 22m 46s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/2159863f26924abf803765effe60908c : SUCCESS in 1h 22m 21s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/189e7a15a13c4198828eabc6310208b0 : SUCCESS in 1h 17m 29s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e4ca4e423e4a448ab2da225556b2ceaa : SUCCESS in 1h 17m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a480924cfdb430a9efbbea42a908249 : SUCCESS in 1h 48m 02s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/286f263bb5e24a1490d2b3a6a46a76c1 : SUCCESS in 1h 45m 56s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/96f4b91b67544935a70bc2c4d25318ac : FAILURE in 1h 50m 25s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/811351ba6a6e4608af13cb291cfa2250 : SUCCESS in 1h 53m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d0396d5a57845efaa2a6169818c3377 : SUCCESS in 42m 28s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/230ac75db0154330a2444764079fd8be : SUCCESS in 34m 49s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/e007b895ef7a470b97f054c02ce5691e : FAILURE in 20m 09s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/281215d908504890a44c1badacb7b60b : SUCCESS in 52m 31s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/49d6a9e431b24ec9b39c5384888709f8 : SUCCESS in 1h 32m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f5bf6cde568042a9ad46e12055f980e7 : FAILURE in 41m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f363fdd4ff946b39be00ad88ddd24ac : FAILURE in 43m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8fe3972ecff0418ba868ab85838560f1 : TIMED_OUT in 2h 34m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c90ad5131a446e7baad375a15788708 : SUCCESS in 46m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/1685f48c40e047e7a8ebe7c87b5a407a : FAILURE in 22m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a2d83b753c8347f789fa742ef84ebe95 : SUCCESS in 42m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/afc78b82cbe34e5f81e0643ea960cfcb : SUCCESS in 44m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4c6d6ffb42a149c48b92747e9d787e4f : SUCCESS in 55m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/5f181462f0734caa94e7f041b42005fb : SUCCESS in 42m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/42a70e20ced64ad0bc442b80ea944fb6 : SUCCESS in 34m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/cac4444ea8ff410595f535766ad45d6a : SUCCESS in 43m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b266e861c2f4b758864619d6c69bc81 : FAILURE in 38m 41s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b474871761234ba3921e901633817e46 : FAILURE in 23m 04s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/8640823e3181458f9766a10fbd407e4c : FAILURE in 27m 18s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"d35764ed0c429ad8b0114f38b329f5374cf534ad","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-01 14:14:50.000000000","message":"Patch Set 30:\n\nrecheck\n\nipv6+tcp+LC issue","accounts_in_message":[],"_revision_number":30},{"id":"03b21124c5bcf0e9991deb06eea39d3d1579da33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:44:34.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e4f835ca2d482caaaadfca778550b4 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb985c9f13834b368d82c443bd624768 : SUCCESS in 4m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/338cab0da3ce4aa79faf947753b0e722 : SUCCESS in 1h 44m 28s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/eaf7c50b03ee4ad2a2f9d00a761d4e24 : SUCCESS in 1h 14m 12s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/dd665e93756a437e9c621bc093347bf2 : SUCCESS in 1h 26m 07s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/70aa7aa972d44f5381fb1fd6d38e6d00 : SUCCESS in 1h 15m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75383de04ac54bf9aaadea2a52cc7a8c : SUCCESS in 1h 36m 30s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/64d0552b1a974bd39daed574a4fe6c6c : SUCCESS in 1h 31m 51s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c2abb05df04e4bbea3d814784a2adc2f : SUCCESS in 1h 42m 58s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1bc0d52cf693461d8e82cc8bc2571113 : SUCCESS in 1h 34m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee2ceac47255479eaac4b1c5ef63d52e : SUCCESS in 48m 13s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/acdcf4bb2a1e47a7b637ebc57da5f739 : SUCCESS in 47m 41s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f9ff11fbf4c74f9b9ffb33cda09b817b : SUCCESS in 42m 47s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/446e67bc5bb64ecd98e5ee730e32ed21 : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b6629c5661e4f1b90330f39f623c68c : SUCCESS in 1h 37m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b70ff8a0f6704262ad0b7cd63b31e32d : FAILURE in 1h 00m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7414f4e14431422897fbf9fa6f7d7f35 : SUCCESS in 2h 12m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4063aa381c414ca3b8dd8a93ff91d158 : FAILURE in 2h 18m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/116054eb8eaf40ffa80b57156a4b6079 : SUCCESS in 47m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/d032446535e24bedb41dc2ab31b80bb9 : SUCCESS in 36m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6f423719853c46e480652580f18e6f7d : SUCCESS in 38m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d9536a3302074d77bfc061b7e33b3219 : SUCCESS in 54m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3afcdffe0c4943128b6f593432b5e406 : SUCCESS in 39m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/32f8b4c169ab4445b5df0e9cbc2b07a2 : SUCCESS in 48m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/7ccddb8c61014372a1f21abaef7944e3 : SUCCESS in 59m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/5edc9a1c60d245d3a15d47ec11998686 : SUCCESS in 38m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/647d8e32591f407680648c424caf51e6 : SUCCESS in 1h 36m 32s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/03b4cb3902c944f9956ac848f3a342e3 : FAILURE in 31m 48s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/9ed1bb19651e49468c896c63186cebce : FAILURE in 59m 02s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"0f5a17351589697bcd85f99ceb7bcdb4ee1d2d7c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-07 16:38:26.000000000","message":"Patch Set 30: Code-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"160431d0d98a4113f72c32f946466f3272a841fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-25 14:21:44.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"c3cd06a6cce0c5c727c91a64d1b11d53cea53a8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:13:29.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7979fe6dccd34c6481090dfad81db2c0 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1bebd7980194039802aaeb19e38a06b : SUCCESS in 4m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3df170defd7149fc97af98e4640f5623 : SUCCESS in 2h 05m 39s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/e96a9c4898e0413e942dfa39ddeb8d1c : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/251e8ac84a3e40dd879ea0c1ea6e330b : SUCCESS in 1h 25m 57s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/a520d3819a064630bdf660d1e3e6aa3f : SUCCESS in 1h 06m 20s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/3f8fa9d39bba4f85848e7450100701d1 : SUCCESS in 1h 03m 53s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/0d425ce999984405a7bcdeeb1e9e91c4 : TIMED_OUT in 2h 13m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/364c5d30a96c483a9871d15ea4c2d729 : SUCCESS in 1h 49m 09s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/888d262c271440e189f0d427311e7a18 : SUCCESS in 1h 32m 39s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/d0422d4606fe46978ad9d13c2021729f : SUCCESS in 1h 42m 34s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/47ab1ace13554cdca00ff73cdb3d764d : SUCCESS in 1h 29m 55s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6c93f81a792a451faed3edd96ff39189 : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b835c947ead6446eb65e94114975bd92 : SUCCESS in 38m 59s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/2eeac9a335d94dbfaf19193c767016c4 : SUCCESS in 35m 49s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/32ca5a31738446f6ae5806858030c992 : SUCCESS in 38m 20s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/c0adab475c6c4510b3de956f27eb0e05 : SUCCESS in 36m 46s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/0bc0ec8269864bcab76bdd7c21ef0a27 : SUCCESS in 37m 26s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/37d3585691724e2481b91d19e2adb3d2 : SUCCESS in 1h 27m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/10f3bfffcd6740febbfd8d1427e0e5c7 : FAILURE in 10m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/93c1b8a90aab4327b4388147831c154a : SUCCESS in 2h 05m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00b99b3c97d24fa5bc931bf3afa389f9 : SUCCESS in 2h 19m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04b7076f87f644c1a2c780f0a8b71811 : SUCCESS in 35m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/767f635f07c341b78c370aa320e26f0e : SUCCESS in 33m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/dd8eb4be72bf42bfaa5299d91a25cd63 : SUCCESS in 33m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/4e074a6370a24bbbbdf558baf0ea78b4 : SUCCESS in 36m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2e771bee0a424239a0cacc411d6ebf03 : SUCCESS in 33m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/796bcc1f4c4a467bbe5fc932c8607211 : FAILURE in 46m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/cd9ce85477914454b3bec7989a147cc3 : SUCCESS in 35m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/bd7d6b9ed7574ed284b3f19b2013fc1b : SUCCESS in 36m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/7b0a1ea9a4e44126806f73535391117c : SUCCESS in 34m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/13d707f2a6e543678fb3fd83052ca522 : SUCCESS in 35m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bf8191b0081741b18474ea548ff7560d : SUCCESS in 1h 31m 04s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e14f59be31174bb192de40d154e4e2fe : SUCCESS in 49m 42s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/73f17d564aa64649be78ce1158271a26 : SUCCESS in 50m 58s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"c7463a8d8e92b48fae9f8fea6b7da84c5a664e45","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-15 14:11:20.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"fa58dbe0c78177bb0b426080ab4086b4faa0ec78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 16:54:04.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3817c76daef94e77b91317c8b82fd214 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa2817539c704272beebbffb020a158c : SUCCESS in 3m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b72993733b55489fa5ae3e8dec06e944 : SUCCESS in 2h 35m 44s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/edd76f5f7f11401cab228a334f877e17 : SUCCESS in 1h 31m 11s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/1222ca3187934563abb5c65bc3c0756e : SUCCESS in 1h 28m 11s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/3c971150f5e449c2bfe58b33248feede : SUCCESS in 1h 03m 49s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/78f86ef346d04f0ba84bb6a24697aeb8 : SUCCESS in 1h 03m 22s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/efa6d26550d5452d8473bd5e906f5b4a : SUCCESS in 2h 17m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f63ba4cf48fc4a20baacd5a5cbc2429f : SUCCESS in 1h 17m 40s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/48f790bd0acc4cdda833a2089e549edb : SUCCESS in 1h 18m 34s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/a5e6ceebe7e44802a6cd0bf001157c76 : SUCCESS in 1h 28m 25s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/91f0ebefc85e42bd8da1ee75f9785113 : SUCCESS in 1h 34m 24s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b21fef0df84f4dbe8100f74d5efbf01a : SUCCESS in 1h 33m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab96922bc6434e9e8943807da848f0b1 : SUCCESS in 33m 58s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/6ced4069518e4d209d8b9db1c31aa08e : SUCCESS in 34m 47s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/11c152c5804d4afd9137a58c4f118b3a : SUCCESS in 51m 15s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/74a46f56bd5342d68c311799d0e2664d : SUCCESS in 34m 37s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f19f1c644d574b9ca4c234c4ea92e7e0 : SUCCESS in 36m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/003b163ec2dc4171a7408171c7534bc5 : SUCCESS in 1h 26m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/204c975a23cf4be98496b2b8240f8c58 : FAILURE in 8m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/988593ad272b4fcd8a5bb0478adfeadb : SUCCESS in 1h 37m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0a5f08e1241f495c938d40dd4032da79 : SUCCESS in 2h 03m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc2d9c9ac3d94ec79f8130147a5b40fe : SUCCESS in 34m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/e0563e4667654ecdbf65e31ce47444b5 : SUCCESS in 32m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/4e5e661c434e4129ab9f304796a2ff32 : SUCCESS in 30m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7c74ba443d7c4a0d92ac7529c2c409ba : SUCCESS in 47m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/430a8fae290f47d48df05ea8f4089b0c : SUCCESS in 32m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1d7c3504d648403c9bd470ffc148d4b4 : FAILURE in 1h 00m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/2fa7b1d1b6dd4b5ca652584014521ad9 : SUCCESS in 31m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/b9a24d0b63584ec0a0123036d6a08a58 : SUCCESS in 32m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/28258f9beb424ea492368dc3c85da778 : SUCCESS in 32m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/5f1ee18e0da44a26bed53d36dabdd9ae : SUCCESS in 31m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/da8d3de82f1d4863985eb8a4413fb87d : SUCCESS in 1h 25m 23s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6103e2826af143f8a5af1b7f6c2dff69 : SUCCESS in 43m 16s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/9568e3d7fcd64ca5bccb1e72d1680116 : FAILURE in 1h 04m 03s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"65ec30a4d10032cba1375d7ee9036f423598848c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-15 17:26:28.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"a87b171a1ffdc239097cf82e481a2b56dd360a36","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-25 11:09:35.000000000","message":"Patch Set 31: Code-Review+2","accounts_in_message":[],"_revision_number":31},{"id":"37d6e58966dd36f96f2e55fa91255af1c2ca2875","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-10-26 11:30:51.000000000","message":"Patch Set 31: Code-Review+1","accounts_in_message":[],"_revision_number":31},{"id":"b887520e358d913742497d5bd451758fc7fde335","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2022-10-27 00:59:09.000000000","message":"Patch Set 31: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"ae18badda6bb21e13685146c76be7326b1c05161","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-31 16:03:45.000000000","message":"Patch Set 31: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"520f6230ee06964dc72dfe8d25603b15e04ceb38","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2022-12-27 10:03:38.000000000","message":"Patch Set 31: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"e0b8fbe5cfb046df21211d6f187f3cd8c6bf7d17","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2023-02-24 14:27:17.000000000","message":"Patch Set 31: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"5ad0f522346273483f29b9218010c4c6f60549d7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-27 17:14:11.000000000","message":"Patch Set 31: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"6fe0376599f26628cdc1970a61b7ba5b23c765e9","revisions":{"aa44b2fe09aa01014da2aef3299f5c406e11017a":{"kind":"REWORK","_number":1,"created":"2019-11-21 04:53:41.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/1"}}},"commit":{"parents":[{"commit":"bcab0c17867689c720c34991e05d2c5d4aeaa0e7","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bcab0c17867689c720c34991e05d2c5d4aeaa0e7"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:53:12.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/aa44b2fe09aa01014da2aef3299f5c406e11017a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/aa44b2fe09aa01014da2aef3299f5c406e11017a"}]},"branch":"refs/heads/master"},"64a93440ac3cdaa21c0e8551050617d3f6fd8e41":{"kind":"REWORK","_number":2,"created":"2019-11-21 22:31:39.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/2"}}},"commit":{"parents":[{"commit":"46a4f10133e17570da583301da8c34d68eb51a89","subject":"[WIP] Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/46a4f10133e17570da583301da8c34d68eb51a89"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 22:31:10.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64a93440ac3cdaa21c0e8551050617d3f6fd8e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64a93440ac3cdaa21c0e8551050617d3f6fd8e41"}]},"branch":"refs/heads/master"},"9a00b85c04e914615185fc1d8a54b3d100770e2b":{"kind":"REWORK","_number":3,"created":"2019-11-26 04:46:51.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/3"}}},"commit":{"parents":[{"commit":"bf5161c63951639602578e56852d3171335f81df","subject":"[WIP] Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf5161c63951639602578e56852d3171335f81df"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-26 04:38:46.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9a00b85c04e914615185fc1d8a54b3d100770e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9a00b85c04e914615185fc1d8a54b3d100770e2b"}]},"branch":"refs/heads/master"},"ea7a4bd475dbb20f86d04fd606cf8747a90efb42":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-27 22:22:11.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/4"}}},"commit":{"parents":[{"commit":"b9cd994ddc52df1c30463faf503857b73740f369","subject":"[WIP] Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b9cd994ddc52df1c30463faf503857b73740f369"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-27 22:22:11.000000000","tz":0},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ea7a4bd475dbb20f86d04fd606cf8747a90efb42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ea7a4bd475dbb20f86d04fd606cf8747a90efb42"}]},"branch":"refs/heads/master"},"7a0d1204cae96475c94adb8f1adb1d9a7ecee001":{"kind":"REWORK","_number":5,"created":"2019-12-02 00:55:21.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/5"}}},"commit":{"parents":[{"commit":"f86df83a8bd04d3211d1d905bc5dcd244b1aa90d","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f86df83a8bd04d3211d1d905bc5dcd244b1aa90d"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-02 00:51:32.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepend-On: https://review.opendev.org/#/c/693762/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7a0d1204cae96475c94adb8f1adb1d9a7ecee001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7a0d1204cae96475c94adb8f1adb1d9a7ecee001"}]},"branch":"refs/heads/master"},"f8a9620a1f8997b17aaeb34649f4d1d6c3521e1d":{"kind":"REWORK","_number":6,"created":"2019-12-02 04:51:49.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/6"}}},"commit":{"parents":[{"commit":"f86df83a8bd04d3211d1d905bc5dcd244b1aa90d","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f86df83a8bd04d3211d1d905bc5dcd244b1aa90d"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-02 04:51:31.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepend-On: https://review.opendev.org/#/c/693762/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f8a9620a1f8997b17aaeb34649f4d1d6c3521e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f8a9620a1f8997b17aaeb34649f4d1d6c3521e1d"}]},"branch":"refs/heads/master"},"d549803d712306efb7b12f806248fb3ec0bfb5b7":{"kind":"REWORK","_number":7,"created":"2019-12-03 05:18:13.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/7"}}},"commit":{"parents":[{"commit":"f86df83a8bd04d3211d1d905bc5dcd244b1aa90d","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f86df83a8bd04d3211d1d905bc5dcd244b1aa90d"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-03 05:17:59.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepend-On: https://review.opendev.org/#/c/693762/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d549803d712306efb7b12f806248fb3ec0bfb5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d549803d712306efb7b12f806248fb3ec0bfb5b7"}]},"branch":"refs/heads/master"},"0298fe74059562ee41425c78f843122100eee90f":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-12-04 01:08:54.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/8"}}},"commit":{"parents":[{"commit":"f86df83a8bd04d3211d1d905bc5dcd244b1aa90d","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f86df83a8bd04d3211d1d905bc5dcd244b1aa90d"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-04 01:08:28.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepend-On: https://review.opendev.org/679779\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0298fe74059562ee41425c78f843122100eee90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0298fe74059562ee41425c78f843122100eee90f"}]},"branch":"refs/heads/master"},"0de8d42220ae25c1234366b9ed308ab73338aaea":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-12-04 01:10:24.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/9"}}},"commit":{"parents":[{"commit":"f86df83a8bd04d3211d1d905bc5dcd244b1aa90d","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f86df83a8bd04d3211d1d905bc5dcd244b1aa90d"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-04 01:09:57.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepend-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0de8d42220ae25c1234366b9ed308ab73338aaea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0de8d42220ae25c1234366b9ed308ab73338aaea"}]},"branch":"refs/heads/master"},"f5f485a3c6f873b0b669ff18fd8ae6fe8780cdd2":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-12-04 01:14:26.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/10"}}},"commit":{"parents":[{"commit":"f86df83a8bd04d3211d1d905bc5dcd244b1aa90d","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f86df83a8bd04d3211d1d905bc5dcd244b1aa90d"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-04 01:13:57.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f5f485a3c6f873b0b669ff18fd8ae6fe8780cdd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f5f485a3c6f873b0b669ff18fd8ae6fe8780cdd2"}]},"branch":"refs/heads/master"},"1a5c9bc1ae493d45aeb909eefd00b554662e018d":{"kind":"REWORK","_number":11,"created":"2019-12-09 00:44:38.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/11"}}},"commit":{"parents":[{"commit":"c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-09 00:44:20.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1a5c9bc1ae493d45aeb909eefd00b554662e018d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1a5c9bc1ae493d45aeb909eefd00b554662e018d"}]},"branch":"refs/heads/master"},"abaef6d12494315160f6b84d4634e9324d740864":{"kind":"REWORK","_number":12,"created":"2019-12-09 00:47:30.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/12"}}},"commit":{"parents":[{"commit":"c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-09 00:47:16.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/abaef6d12494315160f6b84d4634e9324d740864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/abaef6d12494315160f6b84d4634e9324d740864"}]},"branch":"refs/heads/master"},"bd3520d7fd824f10b665786fbfbebd1583fbdc4c":{"kind":"REWORK","_number":13,"created":"2019-12-17 20:28:10.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/49/695349/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/13"}}},"commit":{"parents":[{"commit":"c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-17 20:28:04.000000000","tz":0},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bd3520d7fd824f10b665786fbfbebd1583fbdc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bd3520d7fd824f10b665786fbfbebd1583fbdc4c"}]},"branch":"refs/heads/master"},"f7fefd6c76277157ea9a435283364eaecf972259":{"kind":"REWORK","_number":14,"created":"2019-12-17 20:50:20.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/49/695349/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/14"}}},"commit":{"parents":[{"commit":"c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328","subject":"Test the Availability Zone [Profile] admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c2aa20c24a9adf505d06f1fc4a4cb0e18d28c328"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-17 20:50:08.000000000","tz":0},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f7fefd6c76277157ea9a435283364eaecf972259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f7fefd6c76277157ea9a435283364eaecf972259"}]},"branch":"refs/heads/master"},"771b6170e2ade3fc5429812a498da3643497aa07":{"kind":"REWORK","_number":15,"created":"2019-12-17 23:46:08.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/15"}}},"commit":{"parents":[{"commit":"24eacd586d40680291be61b89127e3d32976120c","subject":"Merge \"Add octavia-v2-dsvm-scenario-centos-8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/24eacd586d40680291be61b89127e3d32976120c"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-17 23:45:53.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/771b6170e2ade3fc5429812a498da3643497aa07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/771b6170e2ade3fc5429812a498da3643497aa07"}]},"branch":"refs/heads/master"},"f51149450d1cd6983895f60210f682ad49991e99":{"kind":"REWORK","_number":16,"created":"2019-12-17 23:49:04.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/49/695349/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/16"}}},"commit":{"parents":[{"commit":"24eacd586d40680291be61b89127e3d32976120c","subject":"Merge \"Add octavia-v2-dsvm-scenario-centos-8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/24eacd586d40680291be61b89127e3d32976120c"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-17 23:48:13.000000000","tz":660},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f51149450d1cd6983895f60210f682ad49991e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f51149450d1cd6983895f60210f682ad49991e99"}]},"branch":"refs/heads/master"},"1508c8161bf944359cf5e0584a030452266e2bd1":{"kind":"REWORK","_number":17,"created":"2019-12-18 23:21:07.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/49/695349/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/17"}}},"commit":{"parents":[{"commit":"24eacd586d40680291be61b89127e3d32976120c","subject":"Merge \"Add octavia-v2-dsvm-scenario-centos-8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/24eacd586d40680291be61b89127e3d32976120c"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-18 23:20:44.000000000","tz":-480},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1508c8161bf944359cf5e0584a030452266e2bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1508c8161bf944359cf5e0584a030452266e2bd1"}]},"branch":"refs/heads/master"},"48dfcd47334e3208fb2e06b83e895f553ca8e9c3":{"kind":"REWORK","_number":18,"created":"2021-02-03 07:37:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/18"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 07:37:14.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/48dfcd47334e3208fb2e06b83e895f553ca8e9c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/48dfcd47334e3208fb2e06b83e895f553ca8e9c3"}]},"branch":"refs/heads/master"},"825d65b7da281b09a7c55fedc05875aadd982429":{"kind":"REWORK","_number":19,"created":"2021-02-03 08:52:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/19"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 08:52:01.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/825d65b7da281b09a7c55fedc05875aadd982429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/825d65b7da281b09a7c55fedc05875aadd982429"}]},"branch":"refs/heads/master"},"ea31ef3cbc8587869fd22500a808c5777b927d20":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2021-02-04 13:51:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/20"}}},"commit":{"parents":[{"commit":"e52f8ba64543c5febd837ac63cf164df61ed025b","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e52f8ba64543c5febd837ac63cf164df61ed025b"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-04 13:51:34.000000000","tz":0},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ea31ef3cbc8587869fd22500a808c5777b927d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ea31ef3cbc8587869fd22500a808c5777b927d20"}]},"branch":"refs/heads/master","description":"Rebase"},"546758078967853e2e5cf9be974b2bd9b100a7b0":{"kind":"REWORK","_number":21,"created":"2021-02-04 17:18:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/21"}}},"commit":{"parents":[{"commit":"e52f8ba64543c5febd837ac63cf164df61ed025b","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e52f8ba64543c5febd837ac63cf164df61ed025b"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-04 17:18:28.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/546758078967853e2e5cf9be974b2bd9b100a7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/546758078967853e2e5cf9be974b2bd9b100a7b0"}]},"branch":"refs/heads/master"},"578c466aa094fd310468b3528a0b7ed56b31d755":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2021-02-05 07:37:31.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/22"}}},"commit":{"parents":[{"commit":"4f899f87bae81d2d391117f58d2b3d31a22df7bf","subject":"Fix copy output in two-node jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4f899f87bae81d2d391117f58d2b3d31a22df7bf"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-05 07:37:01.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/578c466aa094fd310468b3528a0b7ed56b31d755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/578c466aa094fd310468b3528a0b7ed56b31d755"}]},"branch":"refs/heads/master"},"64af711263b7f15d6fadc9ec81f8c9cb005a30c0":{"kind":"REWORK","_number":23,"created":"2021-02-05 12:00:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/23"}}},"commit":{"parents":[{"commit":"4f899f87bae81d2d391117f58d2b3d31a22df7bf","subject":"Fix copy output in two-node jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4f899f87bae81d2d391117f58d2b3d31a22df7bf"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-05 11:59:48.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64af711263b7f15d6fadc9ec81f8c9cb005a30c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64af711263b7f15d6fadc9ec81f8c9cb005a30c0"}]},"branch":"refs/heads/master"},"0d159f0011d79caa918cd5d968225545bc2fa824":{"kind":"REWORK","_number":24,"created":"2021-02-08 10:34:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/24"}}},"commit":{"parents":[{"commit":"64e86b2f7d0ccbd8d128d3a7497d074b145c5e65","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64e86b2f7d0ccbd8d128d3a7497d074b145c5e65"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-08 10:33:49.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0d159f0011d79caa918cd5d968225545bc2fa824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0d159f0011d79caa918cd5d968225545bc2fa824"}]},"branch":"refs/heads/master"},"49a5e6e0b8e2a619a1fa1c4efbe6663337dd36b6":{"kind":"REWORK","_number":25,"created":"2021-02-08 14:56:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/25"}}},"commit":{"parents":[{"commit":"64e86b2f7d0ccbd8d128d3a7497d074b145c5e65","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64e86b2f7d0ccbd8d128d3a7497d074b145c5e65"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-08 14:55:59.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/49a5e6e0b8e2a619a1fa1c4efbe6663337dd36b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/49a5e6e0b8e2a619a1fa1c4efbe6663337dd36b6"}]},"branch":"refs/heads/master"},"a8501892b604689dc8495b63032499242db1dc68":{"kind":"REWORK","_number":26,"created":"2021-02-15 10:02:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/26"}}},"commit":{"parents":[{"commit":"64e86b2f7d0ccbd8d128d3a7497d074b145c5e65","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64e86b2f7d0ccbd8d128d3a7497d074b145c5e65"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-15 10:01:34.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a8501892b604689dc8495b63032499242db1dc68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a8501892b604689dc8495b63032499242db1dc68"}]},"branch":"refs/heads/master"},"565ec2dcded0067a52d95f50c25a6d7ea125dffe":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2021-02-19 15:35:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/27"}}},"commit":{"parents":[{"commit":"5f572c7c899261d5a4c15744cfb90c59581697ab","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f572c7c899261d5a4c15744cfb90c59581697ab"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-19 15:35:19.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/565ec2dcded0067a52d95f50c25a6d7ea125dffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/565ec2dcded0067a52d95f50c25a6d7ea125dffe"}]},"branch":"refs/heads/master"},"14488d162e2ac28b0de186c3c3d5da9c052e0fda":{"kind":"REWORK","_number":28,"created":"2021-03-19 08:17:16.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/28"}}},"commit":{"parents":[{"commit":"5f572c7c899261d5a4c15744cfb90c59581697ab","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f572c7c899261d5a4c15744cfb90c59581697ab"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-19 08:16:51.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/14488d162e2ac28b0de186c3c3d5da9c052e0fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/14488d162e2ac28b0de186c3c3d5da9c052e0fda"}]},"branch":"refs/heads/master"},"d5486e98ca0879232b163b575c3bf627b272044e":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2021-03-19 15:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/29"}}},"commit":{"parents":[{"commit":"c97b0dc51ece45387678aca4beee1f0d846e6f50","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c97b0dc51ece45387678aca4beee1f0d846e6f50"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-19 15:06:59.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nDepends-On: https://review.opendev.org/#/c/697211/\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d5486e98ca0879232b163b575c3bf627b272044e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d5486e98ca0879232b163b575c3bf627b272044e"}]},"branch":"refs/heads/master"},"1dca3e98e7a8d89126608036d8b5904ca7210c23":{"kind":"REWORK","_number":30,"created":"2021-03-24 09:58:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/30"}}},"commit":{"parents":[{"commit":"c97b0dc51ece45387678aca4beee1f0d846e6f50","subject":"Fix two-node job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c97b0dc51ece45387678aca4beee1f0d846e6f50"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-24 09:58:08.000000000","tz":60},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1dca3e98e7a8d89126608036d8b5904ca7210c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1dca3e98e7a8d89126608036d8b5904ca7210c23"}]},"branch":"refs/heads/master"},"6fe0376599f26628cdc1970a61b7ba5b23c765e9":{"kind":"REWORK","_number":31,"created":"2021-08-25 14:21:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/695349/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/695349/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/695349/31"}}},"commit":{"parents":[{"commit":"572ca619b9ed564c164900822a482ecb740f65ac","subject":"Merge \"Fix wrong protocol in UDP pool tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/572ca619b9ed564c164900822a482ecb740f65ac"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-21 04:51:24.000000000","tz":660},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-25 14:21:01.000000000","tz":120},"subject":"Add new scenario test to create LB in specific AZ","message":"Add new scenario test to create LB in specific AZ\n\nChange-Id: I03776d0b7f19066e85034020f374fd8102a0018e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6fe0376599f26628cdc1970a61b7ba5b23c765e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6fe0376599f26628cdc1970a61b7ba5b23c765e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
