)]}'
{"id":"openstack%2Foctavia-tempest-plugin~725758","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I9d5e9fce5d387c655c0bfc1bdcb699a1aa0f56a1","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I9d5e9fce5d387c655c0bfc1bdcb699a1aa0f56a1","subject":"Remove all usage of six library","status":"MERGED","created":"2020-05-06 05:21:35.000000000","updated":"2020-08-26 08:50:51.000000000","submitted":"2020-08-26 08:46:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725758-1598431565729-038e877c","meta_rev_id":"5954c5096a9442ad29be58ce46d6a354b52c3f33","_number":725758,"virtual_id_number":725758,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 08:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-07 06:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-08-16 13:08:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-08-25 21:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-08-25 21:32:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 21:44:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-07 06:55:34.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-16 13:08:39.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-08-25 21:32:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-26 08:46:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"127eb2fe317044dd86f419be297571981943a17f","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-06 05:21:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a569aeddc3e9d97ca86f1e7be3fa5e5f1a2481be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 07:59:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c7322fa41abe46299c9fedb5ccc92ef6 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35f7899a49943109201d95cbdd5f49d : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dafccfddc0b54146a02534743446796a : SUCCESS in 5m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5ba8d631a5d74ae0a3df632f91a2acfc : SUCCESS in 40m 50s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/9f003292bfaa44b483e69f7a426de90e : SUCCESS in 54m 17s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/8dbaeaf34b564d00bf49fbbfff6e0e61 : SUCCESS in 52m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ec4c2be529d41c99e6afdbe7e44b018 : FAILURE in 1h 50m 58s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8a9448e3f82d4fe4a80c52fd262a9a75 : FAILURE in 1h 55m 16s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/e9d4de22ad0540f3b5c9c63b7fb2bc68 : FAILURE in 1h 44m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b2abad901d2148429ba6853781c5f2c7 : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/51589292f37845eabc6879ab57ab0d99 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b3bcc9b32bfb4c55880fc1a1fb9da4e4 : FAILURE in 2h 06m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c612a6f6345f4000b4bc86c881fa837d : SUCCESS in 59m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/cfa58527330c4573b81a71fee35e7c38 : SUCCESS in 1h 05m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1442df6243324dbab12c8af178b53068 : SUCCESS in 59m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/432f732caefc4697b521bf6038fe98e3 : SUCCESS in 54m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/94cc111c46854a4bbb31a82166b49cfd : SUCCESS in 59m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/f8a59a484b5e4265b467ba8424abe142 : FAILURE in 59m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ddcbc64e0515443aa07fde7e389f0e72 : SUCCESS in 51m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/3357a0cc82f0491aa08ca2aad0fa94f0 : SUCCESS in 59m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/01d6641172f444528864f4097bc95801 : SUCCESS in 55m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c3b43de604204654aa2e53abf4a4ff62 : FAILURE in 1h 41m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b128dc573162755d6fbb4bb7b4256cf3ccf055f","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-06 15:26:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb1a14ca09d6a64a08c4e3ba1cd35e56f1cb9b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:05:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/17fb0ab2a4e4416fa1cf528be7560010 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe801c8b4e0d4fa7b662b290181c4b39 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6528660a733c4c0c922e87c5612a41ec : SUCCESS in 4m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0182c192c4484832860d2a3965a27bb3 : SUCCESS in 50m 19s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/56843cfc9ebc4bda81dd45daff68d191 : POST_FAILURE in 57m 53s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/cccc18c5a9fb441e9adcdf443532f2fe : SUCCESS in 43m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39b90b835e38489783dc49054410ac4b : FAILURE in 1h 18m 02s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8aaf75c259084f9e9e110a7544a15047 : FAILURE in 1h 35m 44s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b27aef322b34404080d7b3e106508ce7 : FAILURE in 1h 56m 05s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dff85bfc7ced42c98ed9a18e07281dbb : TIMED_OUT in 2h 35m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ebbfbb5442944c88b33aabb5fef32d11 : FAILURE in 1h 10m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b9974a36ce0f43be97b6bc95e0a8c4b0 : FAILURE in 2h 13m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fd2fac51422428c969510be672d32ba : SUCCESS in 1h 14m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/00e3712ce3d6477c83bdc5a2aeb6a8e5 : SUCCESS in 1h 16m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/dac7f9591e9e4acf9620a3774d0915ac : POST_FAILURE in 1h 08m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e1f5223781674418b5948c443b016c24 : SUCCESS in 1h 04m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/d4a8f381ee00453fa3e3eb49fe478d35 : SUCCESS in 48m 11s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/a02a22a7207d4d15a083b8d9ef054925 : FAILURE in 46m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/52c46bafce58452ab69d450cbbc53098 : SUCCESS in 1h 10m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b3db4ceecb3d4a5c8a2657ef79b5e6f5 : SUCCESS in 1h 02m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0d4ca559a912421a8a7a3c12c0eff863 : SUCCESS in 55m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1aa9ed70cbd84feb8fe816501f4ae7e5 : FAILURE in 2h 05m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec7e94026022ec6aa1bf44195b3a48ad0b163194","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-19 19:27:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"61156a6d2ea0655e968d728880bfa743d961b60d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 22:04:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/36026dbb9e0c4e6397cc23aeb2722fc1 : FAILURE in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5466592d33324676984f071d38e386b3 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2109873cc22c4e878c90f398cd27bd32 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18b25b81c40d43f387be7d2e6ec58cb8 : SUCCESS in 46m 46s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e4df59fe7f834064b943442677f143b8 : SUCCESS in 57m 45s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/c199860cc9764df3a670c8143e83cf47 : SUCCESS in 47m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79d06b5337a9423ab51cb03b949c7959 : FAILURE in 1h 51m 09s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c782f0e83483428f803bd06aa1aa95a0 : FAILURE in 1h 14m 14s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/45784c74c5354689868eb2065aae4e8d : FAILURE in 1h 07m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/24a85c69f30240958a1ce9de804ae5a0 : TIMED_OUT in 2h 35m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b04158c2790147c29e36cc6ff7d248e7 : FAILURE in 1h 45m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf198ec7626d412988bf1ef29819ca30 : FAILURE in 1h 21m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80046d982cf44bea93fba86a2984f05c : SUCCESS in 1h 01m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bed798ce686641babc862745e2bd7c92 : SUCCESS in 58m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3c91f7ef258b4557958d920f63a1d646 : FAILURE in 1h 08m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4ae940035c21427dbca9ad276758a378 : SUCCESS in 57m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/adb07c6d64394d5b837e32987810b876 : SUCCESS in 58m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/c4740fddb91e4825bfa2299d34351f80 : FAILURE in 1h 04m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c2bd5934f4844397af0ca8768e2870a1 : SUCCESS in 51m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/e70ac02506cf457999cce595ae475841 : SUCCESS in 55m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/4eeef02d7f2444bda25e7235f5b59f2d : SUCCESS in 1h 02m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/675daa46f27b4bc68d9b85702317da4f : FAILURE in 1h 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"062f2a4d95dedaebc5172b8a00c54cab407d6c60","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-20 14:58:23.000000000","message":"Patch Set 2:\n\nThis would need a rebase on-top of https://review.opendev.org/#/c/729377/","accounts_in_message":[],"_revision_number":2},{"id":"ec0d4e24cdf24d7052435557a9af29f6ed5f1b88","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-09 21:43:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"17924574385dd4eed3d2690eed0d20c5ffb19eb9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-09 21:44:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e2f0de4e40a8b61c689b255215d49bd288b300c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 23:48:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8663aedbef624ee2b82e67df2db0183f : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e85de07a3424806abb33ecf0d081f15 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31278b100414ac88a13be8982557155 : SUCCESS in 3m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a248651ef7347ab9e44256c1a922cd3 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/054d41783b184c0e81747561585e2154 : SUCCESS in 46m 25s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/ebf7fe3caca941718a3ca1b958d263a6 : SUCCESS in 44m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0def21021720404fa5dbd98824ebb63c : FAILURE in 1h 57m 30s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4c267b4a4b894b52a27f525118ad098a : FAILURE in 1h 40m 48s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/bb979602a7c647a68059d9c8f9fdb48c : FAILURE in 1h 49m 31s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1eeb4c7432744b3f960190cbc8f907cf : FAILURE in 1h 57m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/50e538d254f84fdba62761ea84670730 : FAILURE in 32m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/39786edd75544dff9497f3a61b3fcb60 : FAILURE in 1h 48m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1cb15424a234b94afa1fa53a2a342f5 : SUCCESS in 1h 02m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/da8061d173ab41b697940ffd35702382 : SUCCESS in 1h 04m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3494852db61e40ed8e13a0f799856f6e : SUCCESS in 1h 01m 16s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5bb6f92cdaba4007be6bcedc21a4bb77 : SUCCESS in 58m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/99685cb918154c5c9446fb894d70e54b : SUCCESS in 1h 01m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/d6c1372c19e14bee9fbe4bc7bd9d8615 : SUCCESS in 1h 05m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d589d6ddea9e4a048cdb0ef37668d3dd : SUCCESS in 57m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/56d3fa7dbb4c482ebd05c4b0552e4ec5 : SUCCESS in 56m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/dcebf078015d42b38dabf86aa83e7f26 : SUCCESS in 51m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/939a48571a78423ba96e92781955a002 : FAILURE in 1h 50m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b3c5a39d7f82a951b71768587a6e4597a475021","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-13 18:18:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7af49bab07487c5ad073733561e5a95d983bc69c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-13 18:18:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"34f705990b6effe96c913a343d36b21d5e8eeea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 20:57:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a8b236a670fc4db68a59f97ed2e5ae0f : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e385c1c70ca6458bb41a952cd04736f3 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd85ab3dcdd4f7da4fb51ddf983cede : SUCCESS in 3m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/193d8deef0e34b35ac2dc2d45bea6bd3 : SUCCESS in 46m 37s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c257d121e68349d0a09868fd15dc9b6d : SUCCESS in 52m 18s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/8fe415496489486a867ebd5f5c4b8da4 : SUCCESS in 47m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3394b73681d84b7d8c8a36a79fa9b39e : SUCCESS in 1h 58m 15s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2bae1f1eb428412daaef7d39b316fae5 : SUCCESS in 1h 55m 25s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/11407140793147b181c87fefe0051218 : SUCCESS in 2h 01m 59s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/70fd266237f4419b800070421321ef19 : SUCCESS in 1h 52m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b13fa00f1c774f2c99c6bd644e6a0cdc : TIMED_OUT in 2h 33m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c95245f5b86845eda52b3ca8fbedaf32 : SUCCESS in 2h 17m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62c1058bf01a4afbb11dde8d22d66d52 : SUCCESS in 1h 01m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0e21c74c9f574bd8b54a8ee9d345372d : SUCCESS in 1h 08m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/16715a1dc9fb46329404d0cc40d50045 : SUCCESS in 1h 05m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/42efd3d8320c47afaaef5838cfc06def : SUCCESS in 57m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/e1849ac81cca4f3fb808519f0a5424a2 : SUCCESS in 1h 01m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4ecef94f580e4f8bb62fab71dee9fdef : SUCCESS in 1h 01m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6f6a1f2da8f94ed0ae893d35d18788d8 : SUCCESS in 53m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/af2c02371b914a989d56c7b603bd4b8f : SUCCESS in 59m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/1275bb3b19e04140acc05213ba2878bb : SUCCESS in 58m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6d955b16b3734165b0149df677b1a50c : SUCCESS in 2h 08m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bb1ec51999d2df647fa56a9546345f776c7f1648","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-07 06:55:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8288d1625a068f5d5c2d94d573aa30b87069fea1","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-16 13:08:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"dd2325ccd00a3f0a3aefc61e846834ee95d60a47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 21:32:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"7f6c16391eaf70631acdbc5f93b80cce067631fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 07:23:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"662514ce37cba0de4de9dcdd28e1bd7e69ea4226","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:46:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e4cf7742daa84beeb3059a99c9bb4b59 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f080db426f9941fdba31c0c7af3f90a6 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a81c27451e343c09af9b2776c2aea9c : SUCCESS in 3m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86886f9a9f5a4ca1b1111617b1d9fd89 : SUCCESS in 52m 38s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/4a8c76354aec496294b9c351cc52fac6 : SUCCESS in 47m 15s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/0b4c68aaf7f541aebb96ef982176fa15 : SUCCESS in 51m 47s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/17ed06d6e12f476192da732480ab0fda : SUCCESS in 49m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a86f6783f98446bb852e2d8c82cc03a : SUCCESS in 45m 37s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0316f689e9db41bca101cc627beaa738 : SUCCESS in 46m 30s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c8037bbb3fa5465ea85ba6f0e5c3a9dd : SUCCESS in 1h 07m 43s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/2365e51bbf424788a1044d455c82a1be : SUCCESS in 57m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b1780d4de6d40a090315cc8212b7504 : SUCCESS in 41m 10s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/6b9f3eaec138437bbf079b70801f6dde : SUCCESS in 50m 49s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a058f97893374bcab78a0578a273482f : SUCCESS in 36m 05s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/732b4d9e115d44e79d0bf1800d3d19c1 : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":4},{"id":"dc1278242653601c983bbc2e9cf200a605a706bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:46:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5954c5096a9442ad29be58ce46d6a354b52c3f33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:50:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6122bbf4a31041ad87d04148fe50447c : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bd587263755716f3de4987ae177e97fc28e957c0","revisions":{"64ec4a837056d6ed12cd566e7955e6426ff54679":{"kind":"REWORK","_number":1,"created":"2020-05-06 05:21:35.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/58/725758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/58/725758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/1"}}},"commit":{"parents":[{"commit":"a97deefcb3c21194ec1e2b4f11cc0fbd3c16b720","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a97deefcb3c21194ec1e2b4f11cc0fbd3c16b720"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-06 05:21:06.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-06 05:21:06.000000000","tz":480},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nChange-Id: I9d5e9fce5d387c655c0bfc1bdcb699a1aa0f56a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64ec4a837056d6ed12cd566e7955e6426ff54679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64ec4a837056d6ed12cd566e7955e6426ff54679"}]},"branch":"refs/heads/master"},"eeea45b7e92ca5ac77891fed863733d817016bc4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-19 19:27:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/58/725758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/58/725758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/2"}}},"commit":{"parents":[{"commit":"0031cdfedf1d3114bfdbbac0cb1e226c0d2b6b44","subject":"Fix amphora list-show to run in a busy environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0031cdfedf1d3114bfdbbac0cb1e226c0d2b6b44"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-06 05:21:06.000000000","tz":480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-19 19:27:38.000000000","tz":0},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nChange-Id: I9d5e9fce5d387c655c0bfc1bdcb699a1aa0f56a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/eeea45b7e92ca5ac77891fed863733d817016bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/eeea45b7e92ca5ac77891fed863733d817016bc4"}]},"branch":"refs/heads/master"},"f6e3c549c371ea112e292a8873ebdf7b9fe3d654":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-09 21:43:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/58/725758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/58/725758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/3"}}},"commit":{"parents":[{"commit":"47e4589d453ad69d94b004a1e335032481300b85","subject":"Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/47e4589d453ad69d94b004a1e335032481300b85"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-06 05:21:06.000000000","tz":480},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-07-09 21:43:01.000000000","tz":0},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nChange-Id: I9d5e9fce5d387c655c0bfc1bdcb699a1aa0f56a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f6e3c549c371ea112e292a8873ebdf7b9fe3d654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f6e3c549c371ea112e292a8873ebdf7b9fe3d654"}]},"branch":"refs/heads/master"},"bd587263755716f3de4987ae177e97fc28e957c0":{"kind":"REWORK","_number":4,"created":"2020-07-13 18:18:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/58/725758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/58/725758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/58/725758/4"}}},"commit":{"parents":[{"commit":"47e4589d453ad69d94b004a1e335032481300b85","subject":"Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/47e4589d453ad69d94b004a1e335032481300b85"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-06 05:21:06.000000000","tz":480},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-13 18:17:57.000000000","tz":-240},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nChange-Id: I9d5e9fce5d387c655c0bfc1bdcb699a1aa0f56a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bd587263755716f3de4987ae177e97fc28e957c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bd587263755716f3de4987ae177e97fc28e957c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
