)]}'
{"id":"openstack%2Foctavia-tempest-plugin~733155","triplet_id":"openstack%2Foctavia-tempest-plugin~master~If7d4c0a0e8b0657c6edcb9a820844f3aaaf46cae","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"If7d4c0a0e8b0657c6edcb9a820844f3aaaf46cae","subject":"Turn up concurrency for two-node tests to 4","status":"NEW","created":"2020-06-03 14:08:18.000000000","updated":"2020-06-17 21:31:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"64a4e50ac2b95c202ce6fb8fb2ee8cc124401ff6","_number":733155,"virtual_id_number":733155,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-04 05:33:15.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"date":"2020-06-17 18:50:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-17 21:31:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-03 22:46:05.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 22:46:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-06-17 18:50:37.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-17 21:31:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4eb5c7da7888889f9ae6aa0071d4c21a94994a2f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-03 14:08:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6aa3906db9892d9775b00aa941f5acb22b17e6fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:48:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136d5365cc9d4ad0aef4356bac702f20 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e731dfbcf3d4dad9061b8fdb5dfe3fc : SUCCESS in 3m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ab8208705cac47a79c7f9796c70b5b74 : SUCCESS in 38m 56s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/6a9a2090c3504cc3ad42d8e69229c595 : SUCCESS in 50m 52s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/50f8010f26484215bbb7ac87e8f4e8f7 : SUCCESS in 54m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f3148c06f4c417285e4be4ab4743619 : FAILURE in 1h 16m 06s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c2de2f2af41f43d4a58918473777c57b : SUCCESS in 1h 50m 13s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8f6d93726b4749ec8864da5a969492eb : FAILURE in 33m 28s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c90e772467d845fdb0c834aa132375da : TIMED_OUT in 2h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2a0cc8ba205647d68add8bb4fc5ec47e : TIMED_OUT in 2h 33m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cb29dab06bf3471bbb52b8be83adcd04 : SUCCESS in 1h 20m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a3b92cda18a4423919e9f566b1bb94d : SUCCESS in 1h 23m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2832729a18da498dbb34f3710879dceb : FAILURE in 1h 18m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8fb4b07ec9064f3cbc18c802108729cc : FAILURE in 31m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db5feb05764d425aa2429cf7fb35d012 : SUCCESS in 1h 07m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/50ad1efec3864e74ae4b4efdcf051508 : SUCCESS in 1h 02m 26s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/3902d3dc46334ac9a87dd2804a6a1943 : FAILURE in 34m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e2658978fe7e43efa415555ffeac4446 : SUCCESS in 53m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/42f1640f08c641b292a702d4338f5633 : SUCCESS in 1h 02m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/ffc8a385a0b749a7b0077e24aa91ead0 : FAILURE in 31m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c41e77f083d143e695a9a86ddfc012ad : SUCCESS in 1h 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6376ce71b4e1e8245847636249a228de4160d374","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-03 17:30:39.000000000","message":"Patch Set 1:\n\nrecheck\n\nmore datapoints please","accounts_in_message":[],"_revision_number":1},{"id":"a6ac62164541b4c69f45c061fdc5db4e29e7b77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 20:06:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23d5d046c9942d483f0070c7e19660f : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03511a54797b4ceda6369466350e2b10 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44de259725e049fd839feecbe173f087 : SUCCESS in 40m 29s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c0d4f86d9c984a8b95f0b9df624b7719 : SUCCESS in 48m 15s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/9d2101be63d34528b42731a7d5649eff : SUCCESS in 41m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/163da1c82c1849cb957c96ba729843ae : SUCCESS in 1h 48m 26s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/23a052b447054099a18b1e64d8b2c1e3 : SUCCESS in 1h 49m 19s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/5d209404caed47de9b28bc49e6058388 : FAILURE in 27m 18s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/02733af54b9d4d028ccad701fa254b91 : TIMED_OUT in 2h 34m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/54a293bcb70b4207a5af192935bbe85c : FAILURE in 2h 16m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea565f6620314b6daf65d51ffabd3d88 : FAILURE in 2h 11m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f11862db6d904f7ca063b1218643c5cb : SUCCESS in 57m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9e2ee57ac9574f65b7d7796abd5b5e24 : SUCCESS in 59m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6e53b359701c448fb355422ad33098b8 : FAILURE in 23m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/683540d1bfda4848897ddf80d17f54de : SUCCESS in 53m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/70afb33e8d6941768c252f94155de8c4 : SUCCESS in 53m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/851fdf39b4e84ceca421b557e6e23a91 : FAILURE in 23m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9c8448b652fa4d93984c09f3c40e0b03 : SUCCESS in 51m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/c87a4315b68c4d86b668d648939a5c36 : SUCCESS in 51m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/bdf1409703aa494688d7ffcf063f539d : FAILURE in 26m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/805d5924ae284c1a9bb47fde82a3c832 : SUCCESS in 1h 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46ef4e017c9e5b0f82483e4580889a82011b2a5b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-03 22:46:05.000000000","message":"Patch Set 1:\n\nrecheck\nLooks like that run had a kvm issue","accounts_in_message":[],"_revision_number":1},{"id":"f7b4cc3548494c28928ba51cb479214c6ac6b6d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 00:55:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a777e56e36c4cb49cb0955912eb083f : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/789260629d4c45dea8f2639f22cafe9c : SUCCESS in 4m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c154905f65004edb8d7e5296bb8d4890 : SUCCESS in 44m 56s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/4d6cf5fe418f4da7b147509ff102a498 : SUCCESS in 1h 04m 27s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/440d93d567c2433d873db44ec7ad2fa4 : SUCCESS in 1h 02m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e2ba3ad62be4964b24dd8f28824f0af : SUCCESS in 2h 05m 55s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c5096f874f754205bfe189ed94009a44 : SUCCESS in 1h 18m 36s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/06ad80515eb54744947d40362fc5da8a : FAILURE in 46m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1033079e77524aa4be460eeb85f6ddf1 : FAILURE in 1h 32m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/50631dc4fee448db9257655346360c17 : FAILURE in 58m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e0edf8dd504e4b76b2fad0f8f31d31ef : SUCCESS in 1h 55m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fd6d9e560e545639d9f1609b8de279d : SUCCESS in 1h 07m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/dc4d33709ad849129f83aa0525d2248b : SUCCESS in 1h 13m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/06e6d726223c482c8cc74f66e5a77f6d : FAILURE in 27m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd3fe28ae8a34f2a9dfcc6d60ab9d41e : SUCCESS in 1h 05m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/e69fabcf35e7496c94217696a4ac024e : SUCCESS in 1h 07m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/08d90e5e6ee24e82b5b2a8a74af2112e : FAILURE in 33m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8882aa6394b1406eb4d6a78041bf0f9c : SUCCESS in 53m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/54e7324df0bc4c2493611b15a7d18383 : SUCCESS in 56m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/94034eaff5f0492ea8c248ec4d0c3699 : FAILURE in 28m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e7a94f2ad2444b0aa5e74be164048875 : SUCCESS in 1h 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f76c98ef4536c46d0b9678808e07c4b7c662bc7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-04 05:33:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a733bbeabcf9665c52143f3c16b38eae79d4bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 08:09:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75ab3acc780c4086be46f50271135366 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eed838eebf749a0a5194fadceb0794e : SUCCESS in 3m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3268241f247a4db9bfa3951634087ecb : SUCCESS in 42m 02s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/1eda5ec5f9cc4a76b3e73b7bc3783caa : SUCCESS in 48m 30s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/d2b5a314fac240bdb9eae028f7b89f93 : SUCCESS in 43m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01358f0de602418a8a47ebde332acb77 : SUCCESS in 58m 28s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/286f93050e16485a92125a7499843ef7 : SUCCESS in 1h 01m 32s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ab274142556a4368a7bb45fec8606a08 : FAILURE in 26m 36s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f41b577c2a5441d2abb6d24ab0055409 : TIMED_OUT in 2h 34m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2dc0e9f3e14e4aa9a67d01feb8f5d1cd : FAILURE in 1h 07m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86189a04c84249f89877a398b5ef8297 : FAILURE in 2h 09m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19a2e28148b14673aaefd47721b98750 : SUCCESS in 1h 06m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/02857dc50ebc4d52888dec6baead1e89 : SUCCESS in 44m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9b8e20c978b54965928bb439c4316f43 : FAILURE in 18m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/efc08eca64e949b093d4c59fa822ee8c : SUCCESS in 39m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/af7121c496164166a584b6ae46cf2a4a : SUCCESS in 1h 14m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/b11df7a1fb0a4977b1159a0d957c58e8 : FAILURE in 54m 47s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/91f3861e9d08484a98e33f7d38c24c84 : SUCCESS in 1h 10m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/16404879c8f445ec9d5c94d57e910b8d : SUCCESS in 55m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/e74f9d33e35a451f85e0c31e1df3e741 : FAILURE in 49m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9b0210fe7b0c476c91ff5c008bf0bc19 : SUCCESS in 1h 25m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2aa920801323c4b373ee54b7d5f2c75534e2635d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-17 18:50:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64a4e50ac2b95c202ce6fb8fb2ee8cc124401ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:31:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf67cefeed24ec58683a5eb10dd1596 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd7d218f280431b8e13f9ebc69f3423 : SUCCESS in 4m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f474e1a3003140208740fe5b4ba74bcc : SUCCESS in 48m 27s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/7892b748f9e54b3ba76ee071723e95e5 : SUCCESS in 58m 04s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/4aa40b469dec44d5a0754280694f3fe5 : SUCCESS in 55m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac8b0a5f70904b95b2c6dfdb2d2d02c9 : SUCCESS in 1h 48m 12s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/73d3b0573d314207a80243ad82aa5d4f : SUCCESS in 1h 53m 49s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ebc99b7c38a04c04bb033ac1d636640b : SUCCESS in 1h 10m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/656fb46a4c0842138b3f8d987b93ef12 : TIMED_OUT in 2h 34m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/67f94b11d016495eae2e31d72b463173 : TIMED_OUT in 2h 34m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/13e56ac09f4c423fb4dea585c9ea18fb : FAILURE in 1h 35m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fdc76e53a2743998de38c25f8cceed5 : SUCCESS in 33m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ec09d637a9b9463980aab55de212a23f : SUCCESS in 34m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/84e58a20d3994dfb9dc533eb079b489f : SUCCESS in 1h 16m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e739504c7cc540d893c1dbd89e63e945 : SUCCESS in 53m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/dac25e5a277c43d9b5088aeb71f4c63b : SUCCESS in 1h 02m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4cd2069d95214dd5b09c8632a7506260 : SUCCESS in 1h 09m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d5ff9759b6244a2ebb127f2ee1f0ea9b : SUCCESS in 56m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/c3e08bd304104252bc8027339346512c : SUCCESS in 1h 01m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/5e057a7d3d5644fb8a2aa0eaf75b9151 : SUCCESS in 48m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fdb7c4ebc7fc41d58a573ef1d9c371a8 : SUCCESS in 1h 51m 14s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"044f1620161d6c203e96cdd8c943281a495445af","revisions":{"044f1620161d6c203e96cdd8c943281a495445af":{"kind":"REWORK","_number":1,"created":"2020-06-03 14:08:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/55/733155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/55/733155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/55/733155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/55/733155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/55/733155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/55/733155/1"}}},"commit":{"parents":[{"commit":"d2c4881d07173a3a01fd2cae44a361ff8ac0281b","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d2c4881d07173a3a01fd2cae44a361ff8ac0281b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-06-03 14:07:59.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-06-03 14:07:59.000000000","tz":-420},"subject":"Turn up concurrency for two-node tests to 4","message":"Turn up concurrency for two-node tests to 4\n\nChange-Id: If7d4c0a0e8b0657c6edcb9a820844f3aaaf46cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/044f1620161d6c203e96cdd8c943281a495445af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/044f1620161d6c203e96cdd8c943281a495445af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
