)]}'
{"id":"openstack%2Foctavia-tempest-plugin~740242","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Iace86cd45bc8b4d7ff2dbe147f924c039278334c","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Iace86cd45bc8b4d7ff2dbe147f924c039278334c","subject":"Use memory_tracker instead of peakmem_tracker","status":"ABANDONED","created":"2020-07-09 10:58:33.000000000","updated":"2020-07-09 14:25:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c3530b40969676a590bad28cb5542d0c6bc5d2bd","_number":740242,"virtual_id_number":740242,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-09 13:15:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2020-07-09 13:53:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 13:15:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-09 13:53:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"ca742e88ff2aff7ad00c1a25ba9815ef0bb42767","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-07-09 10:58:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dcdd2d7ea41344480bc2c808f04431c3f75797d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 13:15:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47361ec726a54357a0ecb6c38dfe09b5 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5c58a36b2f4fcebc92df81c47fb79f : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/30db18f75fb64439b147854a9df172a3 : SUCCESS in 45m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/256a2486b4704b0b913e5b3d4bcef5fd : SUCCESS in 49m 14s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6f7db877652b4076a4d9066052032646 : SUCCESS in 53m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65cb2b077a8a478cac3f5213b37be303 : SUCCESS in 1h 48m 46s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1fbbf7e44d334002992ded5c0627ba50 : SUCCESS in 2h 04m 06s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/819c2d7a1b754189b2865ad5b52beed1 : SUCCESS in 1h 51m 14s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c83341a8ab204f16997b270681b78ba4 : SUCCESS in 1h 06m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d1b492cc1c01453bbb64571563aa3575 : FAILURE in 39m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/30c6e9d34e1e4c7c9397468fc6b620fe : SUCCESS in 2h 09m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da64689436fe4207bfb43c8576d718fe : SUCCESS in 1h 11m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b49105a7c86c41d8bacc928a44594354 : SUCCESS in 1h 07m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/21c67e208be7413c950031a1d25377e8 : SUCCESS in 34m 14s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82031e6f25c149d0998e013868cf2e5e : SUCCESS in 59m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/9258f76a4da8441e97df0f00fb45c1f4 : SUCCESS in 1h 03m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/d7a2ff60ba1743a8a01f5c927d0caba4 : SUCCESS in 1h 08m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a11c66231d394de1be0224c5f5b70003 : SUCCESS in 1h 01m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/47028b77e66040e8bf6b1b4bd0d65f18 : SUCCESS in 1h 02m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f88f9fb5185c44fa9e90b1ca2962bc69 : SUCCESS in 1h 01m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dcefc4ee840b4c169463ed907a46d7e2 : FAILURE in 1h 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"102d87441de657afaf815ac5332272a7a91a1bfc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-09 13:53:30.000000000","message":"Patch Set 1: Code-Review-1\n\nDuplicate of https://review.opendev.org/#/c/704202/","accounts_in_message":[],"_revision_number":1},{"id":"c3530b40969676a590bad28cb5542d0c6bc5d2bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-09 14:25:34.000000000","message":"Abandoned\n\nDuplicate to existing patch https://review.opendev.org/#/c/704202/\nAbandoning","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0edc7ce7116c92d15d31377b2343a304484f70c","revisions":{"f0edc7ce7116c92d15d31377b2343a304484f70c":{"kind":"REWORK","_number":1,"created":"2020-07-09 10:58:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/42/740242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/42/740242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/42/740242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/42/740242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/42/740242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/42/740242/1"}}},"commit":{"parents":[{"commit":"c4ace76900b150fa65d2f537329482f7fb58b9ee","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c4ace76900b150fa65d2f537329482f7fb58b9ee"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-07-09 10:56:42.000000000","tz":120},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-07-09 10:58:30.000000000","tz":120},"subject":"Use memory_tracker instead of peakmem_tracker","message":"Use memory_tracker instead of peakmem_tracker\n\nThe peakmem_tracker service has long been deprecated in devstack and will\nsoon be removed complete. Replace it by the memory_tracker service. [0]\n\n[0] https://review.opendev.org/638409\n\nChange-Id: Iace86cd45bc8b4d7ff2dbe147f924c039278334c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f0edc7ce7116c92d15d31377b2343a304484f70c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f0edc7ce7116c92d15d31377b2343a304484f70c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
