)]}'
{"id":"openstack%2Foctavia-tempest-plugin~740349","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ic6623994713abb2c8b7a79eac8e0b3d10b0f1571","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ic6623994713abb2c8b7a79eac8e0b3d10b0f1571","subject":"Add stable/ussuri jobs to check and gate lists","status":"MERGED","created":"2020-07-09 18:51:05.000000000","updated":"2020-07-17 20:26:42.000000000","submitted":"2020-07-17 20:17:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740349-1595017023633-ffcc4a9e","meta_rev_id":"fee3b7d6506ff460234c5d095e7a2c8ed1a821ed","_number":740349,"virtual_id_number":740349,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-17 20:17:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-07-17 16:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-07-09 21:46:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-07-17 16:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-07-17 16:50:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 21:46:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-07-17 16:46:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-17 16:50:58.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-07-17 20:17:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"481e3f5e56028f968e2d79a883bcb153f475f043","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-09 18:51:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79961b738d28d9e8f4b594ad69d3956871cdda03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 21:21:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d46f01084e4e47b64451466621ccbe : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9401a83619498fa7c7feabda8fa575 : SUCCESS in 3m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0667b92074d24c4fa25ab45583166bb0 : SUCCESS in 54m 34s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/676fa3ad4f274f6283dae7eebba61b16 : SUCCESS in 56m 49s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/f5488f7b318749f2a6771cda3fae209a : SUCCESS in 55m 08s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/00c8d9c0b28141a7bb9cd765961fec32 : SUCCESS in 41m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6800d14377c3429aae0cc7b225ebc97b : SUCCESS in 1h 22m 53s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6d2b96b35d5d4cf5bcea7d0688519cac : SUCCESS in 54m 28s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a34148719c33417eb226d9a6cb2c8200 : SUCCESS in 1h 22m 39s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8221f7763033459ca779f4e2214a9846 : SUCCESS in 1h 30m 18s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5ae9578cf78546bca1d8fea27b3ccc80 : SUCCESS in 47m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3149264c53ad48c6ba6c7261b90d3a39 : FAILURE in 24m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/491be577bde340019e1a9994fcb819e6 : SUCCESS in 1h 49m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44cf86a9d1eb458b92f75b8fa7035a4b : SUCCESS in 1h 00m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/dde69f43450c4d8aba8a3910fd062681 : SUCCESS in 1h 01m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5fcbbdb0ce0a491db322ce308d2e623f : SUCCESS in 1h 02m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4ce34562ffdf499ea24b6cc1ccfbfd2f : SUCCESS in 35m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d33b9a3b2e9c40cab998e8ec3a908e72 : SUCCESS in 33m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/6fc7cd522d4b45b7b5a776ca2e9e5f3c : SUCCESS in 1h 07m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/270f18b7a8614764afade1d552fae049 : SUCCESS in 34m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/1628ba53633548c7847528aff4b5be21 : SUCCESS in 1h 06m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a0497d9b72a14d9184b690bfb7f7570c : SUCCESS in 53m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/3c419e43ef0040efa648ef38428356ce : SUCCESS in 52m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/57fd8dd3bfbc44f78ca582e4a45b7411 : SUCCESS in 55m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/9c1f71e6007c4bc1812b55f1c691d55f : SUCCESS in 53m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e0ea6555dd443778d1a411b722195b6 : SUCCESS in 2h 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1118b4cbd9c5d762fb37374bdac7ab5ef29805b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-09 21:46:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"366bfc9617f253629901462d51ad96a693680201","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-17 16:46:25.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"dec2e50d91a99fa0f4c8912d5308de483a4118d6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-17 16:50:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5129a8fcea55857ef634b667ed8f588449bf1d9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:51:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e94b41b9a4e34c2f2c144b1ddc44f588b33bdc4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 20:17:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0c9de29d92dd87cbe43d18b5c19e4bf490c2ec81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 20:17:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815cbb28fda3430d81e276f1f54d5759 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/198a56ccafd54756b63e578a0a2a74ea : SUCCESS in 4m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1f3aa78a7bb941afae979e54ef82b19d : SUCCESS in 43m 23s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/67195e291cfe42eb94cec03a12a8fbda : SUCCESS in 44m 59s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/aaec1a285e4544b0918c49358432225c : SUCCESS in 45m 06s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/74494c3d385e4febae5df9f95d9c951e : SUCCESS in 52m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89f462b154d34eb8b73ad0ee6bdfbf74 : SUCCESS in 55m 42s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6d5cd9532ade47f785fbe0836d962673 : SUCCESS in 55m 27s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e7165bc5c293412ca5baeab706e6d4b9 : SUCCESS in 54m 08s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/519f261e97834b29ab3a1e2d3d42ada3 : SUCCESS in 49m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fee3b7d6506ff460234c5d095e7a2c8ed1a821ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 20:26:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303b38c943564a2d96b743fe16c2fb62 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2b5b447518c18e1705bcd4fbecd6405b8565c39","revisions":{"d2b5b447518c18e1705bcd4fbecd6405b8565c39":{"kind":"REWORK","_number":1,"created":"2020-07-09 18:51:05.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/49/740349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/49/740349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/740349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/740349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/740349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/49/740349/1"}}},"commit":{"parents":[{"commit":"c4ace76900b150fa65d2f537329482f7fb58b9ee","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c4ace76900b150fa65d2f537329482f7fb58b9ee"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-09 18:49:50.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-09 18:49:50.000000000","tz":0},"subject":"Add stable/ussuri jobs to check and gate lists","message":"Add stable/ussuri jobs to check and gate lists\n\nChange-Id: Ic6623994713abb2c8b7a79eac8e0b3d10b0f1571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d2b5b447518c18e1705bcd4fbecd6405b8565c39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d2b5b447518c18e1705bcd4fbecd6405b8565c39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
