)]}'
{"id":"openstack%2Foctavia-tempest-plugin~742195","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ie6443a694b3d6e835cc86cb1b8b85f8d6afc6fa1","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ie6443a694b3d6e835cc86cb1b8b85f8d6afc6fa1","subject":"Fix two-node active-standby job vars","status":"ABANDONED","created":"2020-07-21 13:50:47.000000000","updated":"2020-07-22 10:18:43.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b4997a7fa69acb0e2e356eef274033c54949480","_number":742195,"virtual_id_number":742195,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-21 15:34:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2020-07-21 16:15:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 15:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-21 16:15:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d787531f8208bbbb45baffbc58b53ac5edd80f57","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-21 13:50:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8333cbb9f13ea2829477183ca80a5f16f81fd8fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 15:34:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314d102b919d44fabe7d20eec5a1f86e : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/133c1128b0d8483ca60f0cc018e00aa5 : SUCCESS in 4m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d96bddf0d5994b73a82c83705371afd0 : SUCCESS in 49m 21s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/264b06f4477a484db7ed1265372c362b : SUCCESS in 51m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/27f19f92401f45d78b206a7952fd4499 : SUCCESS in 54m 33s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f599421f014646d6872f9e11d0f03f2c : SUCCESS in 54m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d893bd6cc29b4f00bc5c43d9ec07a32c : FAILURE in 1h 35m 57s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c201996c252a40f48dec59871557300e : SUCCESS in 59m 41s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/995916d7351748da82e653fee5d3abd7 : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8b2521ffaae84a168a582f21430b8893 : SUCCESS in 58m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/408773a984c747d7b8caf0993efee36d : SUCCESS in 43m 43s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/8b39ee22bc894c838ea3cc7b62283916 : SUCCESS in 40m 14s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/986d8eb59267422faeb39e4e2da2a236 : SUCCESS in 57m 41s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/90cbb6ce7fb34693a4966e6c48d2b1c5 : SUCCESS in 36m 02s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bfc9a4e456ff49ac81bba706992358b9 : SUCCESS in 1h 06m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e78e8789aa744d35852fad755d20d168 : SUCCESS in 1h 15m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3eea1c727e6141b880cd208d479aca13 : SUCCESS in 1h 12m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b60bfd4d92e24f738354ba99c4fffee2 : SUCCESS in 29m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/88ee6f05bd51414aaad6c01fad10f248 : SUCCESS in 34m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1648ee9f14d44abb8e00ba20bbafeb60 : SUCCESS in 42m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/da7121690cbc4b9a91f5b35b3c6940cd : SUCCESS in 42m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2608f737f41b40e7a7674a5210a8cf5f : SUCCESS in 29m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/8a2dc3e66bce4382930db6f22c38714b : SUCCESS in 51m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b8860b68987344a28788b93f8d6dc1fe : SUCCESS in 32m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/60e37e82575e406fb349af86136ee8f5 : SUCCESS in 43m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/29093fb731d74d57ab954e139d9a34e8 : SUCCESS in 53m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9fc64032e6a3e722ae0029d4bd3f1d5cb30d672","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-21 16:15:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b4997a7fa69acb0e2e356eef274033c54949480","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-22 10:18:43.000000000","message":"Abandoned\n\nErm, not sure where my mind was...","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a45d48b666f7df15ba2808182bd7b85c517728d1","revisions":{"a45d48b666f7df15ba2808182bd7b85c517728d1":{"kind":"REWORK","_number":1,"created":"2020-07-21 13:50:47.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/95/742195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/742195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/742195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/742195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/742195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/742195/1"}}},"commit":{"parents":[{"commit":"ae44e20514156e9365a28bc570d8dbfb93966124","subject":"Merge \"Gate on octavia-v2-dsvm-tls-barbican jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ae44e20514156e9365a28bc570d8dbfb93966124"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 13:48:33.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 13:48:33.000000000","tz":120},"subject":"Fix two-node active-standby job vars","message":"Fix two-node active-standby job vars\n\nOctavia is not running under the ACTIVE_STANDBY loadbalancer topology as\nit is expected to. This is because the devstack_local_conf section is\nnot under \"vars\".\n\nChange-Id: Ie6443a694b3d6e835cc86cb1b8b85f8d6afc6fa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a45d48b666f7df15ba2808182bd7b85c517728d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a45d48b666f7df15ba2808182bd7b85c517728d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
