)]}'
{"id":"openstack%2Foctavia-tempest-plugin~742199","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Iab24754bf6881671091a06047767f8b4dc8b5420","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"story/2007943","hashtags":[],"change_id":"Iab24754bf6881671091a06047767f8b4dc8b5420","subject":"Enable nova anti-affinity in two-node act-stdby job","status":"MERGED","created":"2020-07-21 14:02:12.000000000","updated":"2020-08-25 22:59:13.000000000","submitted":"2020-08-25 22:57:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742199-1598396232265-eee7de1a","meta_rev_id":"b24b73cce0dddacc15513e1b913f1ecc8069deef","_number":742199,"virtual_id_number":742199,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 22:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-08-07 06:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-08-25 21:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-08-25 21:34:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-07 06:52:22.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-25 21:34:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-25 22:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3782152607eb7152e324616c0858994ae163ac40","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-21 14:02:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8aaa8215ead4211ae3af06bee1d37ba00e9332b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 16:03:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4799fed798694281af977fa6c94f394f : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c7b21d9a7f74e088b988836d8ed1168 : SUCCESS in 3m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db117d90c229435f89909cabbd2759a9 : SUCCESS in 52m 24s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/68463865fc5240c8b5ee973c59f2e56a : SUCCESS in 42m 24s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/5160aed662be455db85481bc4db26889 : SUCCESS in 43m 30s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/1fb3e0b313db47b1872928ebe59e58a1 : SUCCESS in 48m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1f4893dee2249799944bed57a152bb8 : SUCCESS in 1h 12m 30s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cd83caef6cc94cb3a8282c467404671c : SUCCESS in 1h 11m 25s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2ab98f6240944619a014bf3e979f3d26 : SUCCESS in 1h 07m 37s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/98aa6142b754481bb66b10cfbf22ca83 : SUCCESS in 1h 08m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b74217576d54b50ac8bdfb1f7e195d6 : SUCCESS in 50m 09s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/2978ae3cce624cddb2b10c988a32b76c : SUCCESS in 33m 46s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/99d31ef3779f4363af5787e62e220639 : SUCCESS in 1h 00m 08s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/180c2f0c21b14e8d89480adf64a1f403 : SUCCESS in 51m 17s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d7b782fb037d44369b7a572f27cf667b : SUCCESS in 1h 29m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1ef0c465c7c144cb82e77f1e3a9c9c7b : FAILURE in 1h 43m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c30c66a674ca4cb0932e0e60394f661d : SUCCESS in 1h 42m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a6ddfc5110443dfb05d3c06b159feff : SUCCESS in 31m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/289eb33728de4364871f00b227b0f369 : SUCCESS in 29m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ce2f8ca5f7ab4caf870edd1d4fc60b55 : SUCCESS in 48m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/cace7cd35662459d9e34247038abafc3 : SUCCESS in 43m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a87945792f9845a0aa490b1926ae1ea0 : SUCCESS in 38m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/94bb0450b9f6451bbf78f1cebadc5743 : SUCCESS in 49m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/99a03c398f6b47d5ad64cbe8a3587d2b : SUCCESS in 50m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/24780cc4f72e4f2a8a57cbfb3460327e : SUCCESS in 43m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d786f322c5e48a69a8dffd361f462ef : SUCCESS in 1h 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92a1ac8bed39a1a89464849b169151c11d85226e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-22 10:17:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3a3017a8c421c4a1c1aa26d40bbf5514e3c67de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 12:01:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657cb3dcb87a46fcada1b6577814ba64 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe5cbc2763a48bbae316f20b2cee96a : SUCCESS in 3m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38a19e1a47e446078664254b1e7fd14a : SUCCESS in 1h 04m 21s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/933141f179a24f0db54627505150ac78 : SUCCESS in 1h 03m 25s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/2857db893a204433957c04128b3cb74b : SUCCESS in 1h 03m 48s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/842522887ef64b5c9f3c562ac93b8693 : SUCCESS in 1h 03m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d411a18fd374454f972aba6a94cadbee : SUCCESS in 1h 19m 43s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cf87a28222834288a13a72c8707db152 : SUCCESS in 53m 35s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/786e94cb6e3c44b7b397f00cff244c44 : SUCCESS in 44m 34s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/f2ccd3ce9baf4ff8b00b49040372434f : SUCCESS in 1h 07m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e368de7c50694f7db71275a67560af59 : SUCCESS in 49m 20s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/011793d0bbed4c8080e3540474ed5cb7 : SUCCESS in 42m 57s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/8a196bf99df04a4aab723c9f7141b0fc : SUCCESS in 42m 26s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/a8ef2aabdfbf4ef68b2f72deb12e4401 : SUCCESS in 38m 25s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9c64ba7f335f480b914172d83e15b2b9 : SUCCESS in 51m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c988881e48274d3c9ec22e839b758769 : FAILURE in 36m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0b04bce9aee44c1498cd25340fc4326e : SUCCESS in 1h 12m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4566baee4924058a01b12bc23928975 : SUCCESS in 49m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/399b3edc9f994371a0316aa2474047ca : SUCCESS in 53m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f8bef9f4feb34243b17e10e2965dcf23 : SUCCESS in 45m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7db62e39282048c498b86e9411041adb : SUCCESS in 43m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7a0709fc97a94bde85346cab1ea3735a : SUCCESS in 46m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/70c236670a31400f808323172bfb5c6d : SUCCESS in 51m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/e9128cfcd3e243398ac10a67970e0be1 : SUCCESS in 49m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/91b7a4a3faf549ffb306f44ea46a6e79 : SUCCESS in 40m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9ca99d2c3487414da2336ed998bd94a1 : SUCCESS in 53m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40aa4a137179b2aac874fd3e2fce2774d09e2606","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-07 06:52:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"89b2ad649b86820bd13aca2f0a36fb9ff8c3ab34","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 21:34:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"b8534a85c4f4bcfeb452fcec8814d31c70f64c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 21:34:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1c2370cd9b31cf1e44c7b4314d056647344ab15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:57:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73169b6dcb8c4f7ab600260c559ab34d : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb680cb5fd64455f8bdd6f364fcc966e : SUCCESS in 3m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a265d9067fb947e1a87e20c291230c92 : SUCCESS in 35m 00s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/41d4bdcea2fb4f078d6235a97ba3b41f : SUCCESS in 33m 34s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/56b4019a92f24f9b98ba124f720fbeeb : SUCCESS in 32m 04s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/a59d626338ad473fa482259cdd7e0aec : SUCCESS in 39m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc0c679a7ccd4afc9cde046d83fbfc8d : SUCCESS in 1h 11m 51s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/48cbfa9798c44e45b693083bea5a97e2 : SUCCESS in 55m 55s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/76f50fefa74d4e7f913bef96d3c06900 : SUCCESS in 1h 20m 35s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4c2c156df45c4721b8225453e9cfedb9 : SUCCESS in 53m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b19a3f22f0aa45a481f7165088406521 : SUCCESS in 1h 07m 04s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/743a34da2fd84f939e3c9be1a8819a70 : SUCCESS in 40m 35s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f9a566cc52784dde85d35552d6c1c4bc : SUCCESS in 1h 05m 18s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/d738617d9fe34f6aaaecd67e9176d99b : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e7a6ce4c8673b3f9535c037beecaabfd83cf4d9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:57:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d9c5749b15593ff658135d7f61573657280e0823","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:59:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7962d5f50f28408d8dc0a13a3d2739d3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5700f3cd93668bb6b7c9f9303782557d3d23bc3c","revisions":{"e0b37791d01681e646d12eab9608dd6caa4efb78":{"kind":"REWORK","_number":1,"created":"2020-07-21 14:02:12.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/99/742199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/742199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/742199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/742199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/742199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/742199/1"}}},"commit":{"parents":[{"commit":"a45d48b666f7df15ba2808182bd7b85c517728d1","subject":"Fix two-node active-standby job vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a45d48b666f7df15ba2808182bd7b85c517728d1"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 13:58:25.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 13:58:25.000000000","tz":120},"subject":"Enable nova anti-affinity in two-node act-stdby job","message":"Enable nova anti-affinity in two-node act-stdby job\n\nProduction deployments are recommended to run on active-standby\nloadbalancer topology and with nova anti-affinity enabled for higher\nlevel of availability.\n\nChange-Id: Iab24754bf6881671091a06047767f8b4dc8b5420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e0b37791d01681e646d12eab9608dd6caa4efb78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e0b37791d01681e646d12eab9608dd6caa4efb78"}]},"branch":"refs/heads/master"},"5700f3cd93668bb6b7c9f9303782557d3d23bc3c":{"kind":"REWORK","_number":2,"created":"2020-07-22 10:17:49.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/99/742199/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/742199/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/742199/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/742199/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/742199/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/742199/2"}}},"commit":{"parents":[{"commit":"ae44e20514156e9365a28bc570d8dbfb93966124","subject":"Merge \"Gate on octavia-v2-dsvm-tls-barbican jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ae44e20514156e9365a28bc570d8dbfb93966124"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-21 13:58:25.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-22 10:16:36.000000000","tz":120},"subject":"Enable nova anti-affinity in two-node act-stdby job","message":"Enable nova anti-affinity in two-node act-stdby job\n\nProduction deployments are recommended to run on active-standby\nloadbalancer topology and with nova anti-affinity enabled for higher\nlevel of availability. Job octavia-v2-act-stdby-dsvm-scenario-two-node\nwill test Octavia under such configuration.\n\nTask: 40413\nStory: 2007943\n\nChange-Id: Iab24754bf6881671091a06047767f8b4dc8b5420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5700f3cd93668bb6b7c9f9303782557d3d23bc3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5700f3cd93668bb6b7c9f9303782557d3d23bc3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
