)]}'
{"id":"openstack%2Foctavia-tempest-plugin~742716","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I85ebad6d5308f5f9ce4d8f0f160d954aeb5e7c94","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I85ebad6d5308f5f9ce4d8f0f160d954aeb5e7c94","subject":"Update statistics tests for response time metric","status":"NEW","created":"2020-07-23 17:42:29.000000000","updated":"2020-08-19 18:36:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"818684a43647a8c16e229d79e35ab82f7e03e8e3","_number":742716,"virtual_id_number":742716,"owner":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"date":"2020-08-19 17:14:01.000000000","_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-19 01:19:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"all":[{"value":-1,"date":"2020-08-18 22:45:51.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 01:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-19 17:14:01.000000000","updated_by":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"reviewer":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"state":"REVIEWER"}],"messages":[{"id":"b43a0775e909ed6ee578e6783f73e998a753b106","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-07-23 17:42:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"994b882a6de22feb622c40a645677cc88916d7e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 19:22:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718c244929564594a5775478c6f6bc61 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79ef0d3cbde34302abcb49eca891b081 : SUCCESS in 4m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e851ec09b6d496383d11dddccc0bbe8 : SUCCESS in 51m 36s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/e2f895a4f99a4f39bda83ef60a4b42d6 : SUCCESS in 44m 49s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/652e2fd0637d4d9f943d40221ae98ed5 : SUCCESS in 1h 02m 20s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/92e25295f1304cb78bae49697cf0b97d : SUCCESS in 50m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f9032963348468db94ec826a1ae3472 : SUCCESS in 57m 43s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a6077ab29dd54bafa5ac7fc57dd6eb17 : SUCCESS in 1h 23m 14s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/465a25cfa6d04ba4b9020c85ab35dd83 : SUCCESS in 41m 52s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d8df9198d81c4e2ca19de27193af83c9 : SUCCESS in 1h 30m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a061af8ad4ab417ba2bca74249743edf : SUCCESS in 41m 11s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/3055a24199924cc3a80b0fb853ba9101 : SUCCESS in 44m 39s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/16508c4c86074e5c8a093bfc4bdd947a : SUCCESS in 57m 54s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/45e55b296ea44ee68f4b978466aafda3 : SUCCESS in 57m 42s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e0f00187011346a198104bc153f32839 : SUCCESS in 51m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bf88fed44c2744af8402b9f154e47a8d : FAILURE in 1h 06m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/871e1514f3b243268e8c64ac636b11bd : SUCCESS in 50m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5c8ef546869466dacac571cb164a1e2 : SUCCESS in 53m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6759e62794e84441abbdbd8e3c014898 : SUCCESS in 49m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/27fe4e3707fb404093fe641ea7011046 : SUCCESS in 42m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a92ef92e22b641b49e5d80b07c605a77 : SUCCESS in 59m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/039f54f2ca9647418b3b4922637f12de : SUCCESS in 1h 01m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/186e390530da4a58a9bb07d9fa899649 : SUCCESS in 41m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/2bd3cc46217b47c8b77b17d5d80909ca : SUCCESS in 39m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/69d452e5f34d45ce8d3d0793c91c3690 : SUCCESS in 1h 16m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1681c52ccd5b4bfd82a6654f84bae4b4 : SUCCESS in 1h 20m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a64f4c210519b416074019160a75f7282b22e1dc","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-18 22:40:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbd3b71d0414885e475c629fe2f84e258c502181","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-18 22:45:51.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"8b09f5003e659d47feb6cadffd9adc6d25065771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 01:19:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7780357fc3294658a5a86268829e2241 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12c4461aefc34d70948151a4b492c2f1 : SUCCESS in 2m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad27a8c8a384440cad7812a662e30dae : SUCCESS in 42m 33s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/f07efb9a53d54902bf79aa45a35841e1 : SUCCESS in 2h 00m 38s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/46be9842fdaf401bb717f921865b620d : SUCCESS in 44m 28s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/e262232bcb55420c965296351be35c01 : SUCCESS in 45m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/662b68e027bd491bacbafab84039733f : SUCCESS in 2h 33m 23s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2d42d8f4f8de443599305834d16dff52 : FAILURE in 2h 07m 52s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9d4e3158ba114b75b8a0901f98e5dc30 : TIMED_OUT in 2h 33m 02s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3fb8f6bcc2084546af9d6b9a32a0e0a3 : TIMED_OUT in 2h 33m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/06715d4db95341f68d20f3f80c0358cf : SUCCESS in 2h 23m 25s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/6e301c4567ad4209bc76feddf77324a5 : FAILURE in 2h 09m 51s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a027214245094ff4932e869b02cd1bce : FAILURE in 1h 27m 38s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/d9e3d4f62d0d49dd96278959cbfb8e49 : SUCCESS in 1h 09m 01s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/28be00db0c9e401ab7fb2f3e6a11f64e : SUCCESS in 2h 34m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ccc23c5422fb4b37a73c309ccabd1576 : TIMED_OUT in 2h 34m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c645597654341ee8f37c763eb778fd7 : SUCCESS in 1h 03m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3113ced409144a5b9dec2c9505f4c0de : SUCCESS in 2h 19m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c3243a89c6b34407b8c0de2438d99f4f : FAILURE in 2h 03m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2dfe606d25644be5bd4c46bd0c3e924d : FAILURE in 2h 16m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/58cacd55194040c2aa65fee99f7e7922 : SUCCESS in 2h 24m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0f230834f2cc45a684783a71d36f03a3 : SUCCESS in 47m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/75322e1597924a0d80a62b510395fc3d : FAILURE in 49m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/1dafd4fa571242c481cdde8f9f7d42cc : SUCCESS in 2h 20m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/a34929a90f534a9d96a548e5db901e00 : SUCCESS in 1h 07m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2911241a2bdb4d4f8d22676a96585707 : FAILURE in 1h 33m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a71c85c161760d77bdceba3d662858848f0fca55","author":{"_account_id":32085,"name":"Stephanie Djajadi","email":"stephanie.djajadi@gmail.com","username":"sdjajadi"},"date":"2020-08-19 17:14:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"62b23414a958285af978e95bc4e40dd65592e6f5","author":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"date":"2020-08-19 18:16:50.000000000","message":"Patch Set 2:\n\nRecheck","accounts_in_message":[],"_revision_number":2},{"id":"818684a43647a8c16e229d79e35ab82f7e03e8e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 18:36:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950a9fb033f84f50a62a7a609642ff13 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fc9f0fc78ab4769a636f11548d9dc96 : SUCCESS in 4m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e063e6002d7743da97f89501a071074d : SUCCESS in 45m 03s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/bf7006a87b0342af87464c1aa4c30ab5 : SUCCESS in 49m 21s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/1ce2c6bf4cce44788d50422a05ad28eb : SUCCESS in 48m 19s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/b037b04df16d4f848aec3cbf0e0180f3 : SUCCESS in 40m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5834576805ca4001a9ab63d3eb410a62 : SUCCESS in 54m 09s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8f6e1d381f4849f99ecaf1782c450b19 : FAILURE in 48m 01s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/43a5562c5f0043eb83d15dddbe3bdee9 : SUCCESS in 42m 20s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4f3fe23be9f0491888b62eba50948502 : SUCCESS in 55m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b951c97b37f740f590fec8e896e56b7e : SUCCESS in 42m 58s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/8412a2b718f24fb1905eb981e28c7dc3 : FAILURE in 44m 26s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ee4bae48710c446aa1803f36095d2b4a : SUCCESS in 45m 07s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/01f05134e6834257ae2d766b5af4e684 : SUCCESS in 56m 15s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a1e92568beac4eee87ddaca607850438 : SUCCESS in 1h 16m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4d3eb15e9c7e44d6bdf2b52d3140a7f7 : FAILURE in 1h 03m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8b4cd5cb6440481bb953208a386a32f9 : SUCCESS in 1h 01m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78c6a41310d6475ea4588dc65e240906 : FAILURE in 9m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9c5819baf6d34276833dd33939b39957 : FAILURE in 27m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6b4355e814de4766a354141563b7421c : SUCCESS in 54m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d8565e37fccf46d6a6ab736e12477762 : SUCCESS in 58m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/210643a916934e6a986b70c8cd63d0ac : FAILURE in 14m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ed4601c8e5bd4a949ec693dbf1f11c83 : FAILURE in 51m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/c95596ae988646d78dc7e93c2e221336 : SUCCESS in 46m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f76cb7ab92f847fbad367dfd811af4f3 : SUCCESS in 41m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d8e1021a57a4eaba1fbf7f991ffee47 : SUCCESS in 1h 02m 26s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b019eaee25d79107b2114db8fb0cd930f673b076","revisions":{"3ed365d28f43e417abc6a7909f0542f6a00b0adb":{"kind":"REWORK","_number":1,"created":"2020-07-23 17:42:29.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/16/742716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/16/742716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/16/742716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/16/742716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/16/742716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/16/742716/1"}}},"commit":{"parents":[{"commit":"ae44e20514156e9365a28bc570d8dbfb93966124","subject":"Merge \"Gate on octavia-v2-dsvm-tls-barbican jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ae44e20514156e9365a28bc570d8dbfb93966124"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-23 17:42:08.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-23 17:42:08.000000000","tz":-420},"subject":"Update statistics tests for response time metric","message":"Update statistics tests for response time metric\n\nChange-Id: I85ebad6d5308f5f9ce4d8f0f160d954aeb5e7c94\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3ed365d28f43e417abc6a7909f0542f6a00b0adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3ed365d28f43e417abc6a7909f0542f6a00b0adb"}]},"branch":"refs/heads/master"},"b019eaee25d79107b2114db8fb0cd930f673b076":{"kind":"REWORK","_number":2,"created":"2020-08-18 22:40:59.000000000","uploader":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"},"ref":"refs/changes/16/742716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/16/742716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/16/742716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/16/742716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/16/742716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/16/742716/2"}}},"commit":{"parents":[{"commit":"7bd2f97221a369f379e15f33a3d3294f220c039e","subject":"Add an optional amphora admin log offloading test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7bd2f97221a369f379e15f33a3d3294f220c039e"}]}],"author":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-07-23 17:42:08.000000000","tz":-420},"committer":{"name":"asingh12","email":"anushka.singh.2511@gmail.com","date":"2020-08-18 22:39:56.000000000","tz":-420},"subject":"Update statistics tests for response time metric","message":"Update statistics tests for response time metric\n\nChange-Id: I85ebad6d5308f5f9ce4d8f0f160d954aeb5e7c94\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Stephanie Djajadi \u003cstephanie.djajadi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b019eaee25d79107b2114db8fb0cd930f673b076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b019eaee25d79107b2114db8fb0cd930f673b076"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32084,"name":"Anushka Singh","email":"anushka.singh.2511@gmail.com","username":"aannuusshhkkaa"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
