)]}'
{"id":"openstack%2Foctavia-tempest-plugin~746737","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I4ae7a03d2248c970d7bcd3fe8a43a43ca48d5084","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"http2","hashtags":[],"change_id":"I4ae7a03d2248c970d7bcd3fe8a43a43ca48d5084","subject":"Add ALPN protocol scenario tests","status":"MERGED","created":"2020-08-18 14:29:16.000000000","updated":"2020-09-22 02:19:15.000000000","submitted":"2020-09-22 02:18:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746737-1600741084001-e4987785","meta_rev_id":"e76577d77f9cc230b0c3da2a217812db2cbbe2dc","_number":746737,"virtual_id_number":746737,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-09-21 15:36:46.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 02:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-15 07:47:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-27 21:52:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-09-11 17:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-15 07:47:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 21:52:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-11 17:16:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-09-15 07:47:54.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-22 02:18:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe4aa893d412b442e1569dc9c5009f833fafbd13","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-18 14:29:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c96149a552771221f246f4268ae12e20cbffd56f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-18 15:52:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0fb38e7ceee74ea838365854c9ad995fb3e5a47a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 17:21:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5fb13e23244ea7a968841f3b519c1f : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df4c6cd01bba492298f064d88513858b : SUCCESS in 4m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b9a4f2d8bf845398f4d7de8ee059b35 : SUCCESS in 47m 08s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/6d1a7d06edbe4433a45297de195c9837 : SUCCESS in 47m 33s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/3f24b285988d43429cff5a824d893167 : SUCCESS in 45m 19s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/9dd92aa5dfd546a4a3c39bb60d408058 : SUCCESS in 44m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d65f2fb64dee422c8b6af23511f86237 : SUCCESS in 1h 02m 21s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/daa1dcbd51594aea981b5b7304d2d3dd : FAILURE in 47m 20s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/fdb0c9872b5140acaf251585dc19dfe2 : SUCCESS in 1h 06m 26s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4d95f76c7e60474dae58e2608e1ed90f : SUCCESS in 1h 25m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92daba48498541dda766f4a1021667aa : SUCCESS in 34m 23s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/d3ad7a82dc66421eb8edb93786cd67da : FAILURE in 27m 45s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/b60c11e7eed247feb1330cd851338f42 : SUCCESS in 32m 16s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/58462da083ef4e7ead59fea87a1b3428 : SUCCESS in 55m 54s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c03d0caff3344b3e99525314b2f1daf9 : SUCCESS in 1h 06m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a09d126ea56e413598adcc9773cc5288 : SUCCESS in 1h 15m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d759596ee53b4298b915ecd5b1ff9c3b : SUCCESS in 51m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8c564b9a48d4c06b9b7c1a7d658cd4d : SUCCESS in 56m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5dc1b64672144da7ab2830af1e280c94 : FAILURE in 46m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4d220f0088aa4029b2f5ca0d9f3523ba : SUCCESS in 47m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/53c55c0faf6941e28f15d0bee0ba334f : SUCCESS in 54m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3cde7d5bd28047faafd6257afa95b946 : SUCCESS in 58m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/abde7c9cffd048d9943c878720762825 : FAILURE in 41m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/52492b56f28a47f882901ada824ca1dd : SUCCESS in 40m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/21009ecaa3fb41fda1e2faa2c16bc974 : SUCCESS in 41m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/794eaca4775142dd91b65c82b215c3a5 : SUCCESS in 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6b73c4808f7aba9c0b51e03072334b8919c97a4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 13:53:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"384a1e8c01ac072003cf0edfe068792a050e9613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:30:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859ea287072a421dbff11c97a3f64f6f : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bde2069320d473990e09694e78d7424 : SUCCESS in 5m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16413cdc6d014463abb43094b7b07298 : SUCCESS in 50m 58s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/93896db7780048778c140bcaec9b3c0c : SUCCESS in 49m 57s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/81bbb6988b294192b88f03f23a2790b5 : SUCCESS in 44m 12s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/c96b90b1c0784789988ad8751f718731 : SUCCESS in 47m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f72e4179784477681ba81beedd50c23 : SUCCESS in 46m 07s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5629494c23c14165bbc507e33e896cfc : SUCCESS in 56m 36s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/133e6cbc519a4c3290b70b137d44222a : FAILURE in 57m 14s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/927e9302f23c4130a39094f7ed57728f : SUCCESS in 53m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe56f572cea74fd792b1ee29035b956e : SUCCESS in 35m 55s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/49d00a49101f4562a1eb56b7047aea84 : SUCCESS in 44m 42s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/5f360e3098dd4f3ba0f12a74b9f100b1 : SUCCESS in 35m 14s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/bbd68e125b744e1fb7e2a0aa543e6caa : SUCCESS in 44m 56s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae4503c6e9df45318eccb7de263545f4 : SUCCESS in 59m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7a064cb4244e45e8a02529361e8e2d1b : SUCCESS in 1h 09m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8863eb89f8e468b8e51a29575d6bdfb : SUCCESS in 1h 11m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c662ff33ca944205bd85867d8ab903a8 : SUCCESS in 55m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a1f9ef0960e94c3d96074d1108a3705c : SUCCESS in 58m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/455e43e940ea46f9bb89c33c82b3a845 : SUCCESS in 44m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/eecfdaffa469450faf90fc4fde3c4c6f : SUCCESS in 56m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/66df3689d21c41fbade8c0b23243f1a9 : SUCCESS in 55m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/1ac55c8382834bc19e077dd440a35ef5 : SUCCESS in 57m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/ead61737531d48e9a0fd2405ef618362 : SUCCESS in 45m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/926c1f84acdc4944ad88def28b4f0a48 : SUCCESS in 43m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d61dd553f5b24ded9f53083883a06d33 : SUCCESS in 59m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb160692c41c9396ebfe20deb9bfd649e5c851be","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 15:31:56.000000000","message":"Patch Set 3:\n\nrecheck\n\nknown udp health monitor issue","accounts_in_message":[],"_revision_number":3},{"id":"49109776df8b465cc0ff90b2d316b9c446f298f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 17:00:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b663049a037a4de3b1136b5cf47ece90 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e83e4f7073de436d925189529b60ffe0 : SUCCESS in 3m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc93786cd9864449a4feb307119cc690 : SUCCESS in 46m 10s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/65eddf830d784721b47304d4341f1fcf : SUCCESS in 49m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/791075c92b934f8881b68ed6da381ec6 : SUCCESS in 51m 29s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/04aa1eaa92134c5d8a60e9a5a0112ab8 : SUCCESS in 50m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9570d1ec65db474da0e404ec49694d0b : FAILURE in 44m 41s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c20977ddfdb14ac7bed53f411a95c940 : SUCCESS in 52m 53s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f883a498af4f4285bba9b1555a13f0cf : SUCCESS in 1h 18m 21s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/dead737a314a477493c1aab7221f8671 : SUCCESS in 42m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f34736befdb040f3b60bc4c85748fff8 : SUCCESS in 36m 09s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/35b91acb21e74059bf9ba0cf89677bf9 : SUCCESS in 33m 36s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/c03b43a9b12f42989dcf508948432197 : SUCCESS in 57m 23s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/88ddaecd1b8a472a809bb59648fc3394 : SUCCESS in 36m 18s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6c9704aaff0448f08a03179223576284 : SUCCESS in 46m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0b7009d2f8ca41338716ed7c6898bfcd : FAILURE in 37m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9949a6f848f040d1a25a187d9bbd5fc0 : SUCCESS in 54m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1322d2a3f3c4418a9554b413d03b5718 : SUCCESS in 47m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cba8ffdd0ebb462ca6dcfa250e58187d : SUCCESS in 56m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8bbd122b738941468c666624aad88199 : SUCCESS in 58m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/07589d98ed8a42189f5b45fde17e6305 : SUCCESS in 57m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b5ad7d1da12e489097dccd7a3cc56e6d : SUCCESS in 55m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/8e2cded0fd4b40de95340f86b2a318dc : SUCCESS in 40m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/82637c9c1d964d91bd6d6cf9347db3c8 : SUCCESS in 46m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/c2f3f977b21e4bb4bdca804f3a629aa1 : SUCCESS in 48m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/26a7f2c3d9a24da9a8294c3d9d2d0951 : SUCCESS in 58m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77fef89ca887eb66c1d46f873c1db2d3cde22a50","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 21:19:49.000000000","message":"Patch Set 3:\n\nrecheck\n\nudp again","accounts_in_message":[],"_revision_number":3},{"id":"491dd58582f87d3e7502a3df743c8f53176dd4a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:51:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a7ddb4d1457408583443e77e164fe3c : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b12aec71fb6475ab25892cd866f4105 : SUCCESS in 3m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c4da1c702e3469287b033456d5925e3 : SUCCESS in 42m 15s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/a6a7ce49470a4efb95ba8c2d402dc392 : SUCCESS in 48m 05s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/fc65da43d32b48f8b88c29cfa0f777b4 : SUCCESS in 47m 39s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f2b32e99bab24fe5affd10d7168f2051 : SUCCESS in 44m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1b6f9f4089e4aca990a02721b29d990 : SUCCESS in 1h 13m 42s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/bc9bed7eeca94b9da0fd2b7274636add : SUCCESS in 52m 34s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d676dcbbadfc44f2bd0979e4cdea3d2f : SUCCESS in 49m 58s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/508862781dc942c28adf898e390e1f07 : SUCCESS in 1h 13m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/efbb14d45809460180ea4ef90afc862a : SUCCESS in 41m 09s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/1703866e6cbb42d58e91fcd2fd150a19 : SUCCESS in 32m 07s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/467ba1a1a5bd457caaff0eca03e9b7a5 : SUCCESS in 32m 32s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/9e699688d257411bb3a6e7f79a41a19f : SUCCESS in 33m 30s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3dd32474e37f4f54a376eeaafe26aafb : FAILURE in 1h 14m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/77d10bae4bdd451fb8a8ba02b6d8a912 : FAILURE in 35m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e2895abd14e44ea2901cbd7c834ab3c4 : SUCCESS in 1h 06m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cf08c8aea8a4f09ab36a804d69196e9 : SUCCESS in 30m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1a5872e105d440edb5a2079aa4f05ba8 : SUCCESS in 29m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e1c358c35b2e4ad5a2615f9895417d45 : SUCCESS in 57m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/370ab84e0538458ab99d68c9968969da : SUCCESS in 31m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3b31b509f0894b8da89a34891da3ef20 : SUCCESS in 53m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2203e3341e754743bdda04a184c91247 : SUCCESS in 31m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/f230e64e4d8b42c18945b1ea6c017806 : SUCCESS in 38m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/2c8c36e33ad64351be8a5eaabffc1e5f : SUCCESS in 32m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d0ca52d2f8974cf586e0662d369d3e87 : SUCCESS in 1h 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"83c67c78cf480029339b7fdc5009a243c21fb891","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-27 21:52:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2783a793ae1ac6e791e24b5667477be03db16de8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-11 17:16:04.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"54b1996991db160f7eedd989443df5d479ffbc76","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-15 07:47:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f8e58cb682125811daee1b6abb06fe1f40c641e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 07:48:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5466e90c95150524bbf4c3aeae3a3a7e082975c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 08:20:01.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7471ec77a65a42ca9c2aefddf53f3fbd : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60aaa445089a4b41b989826c5bab2226 : SUCCESS in 4m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/35bd5c015bf94baeb6f941c74c26518d : CANCELED\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/39274f23c1d44829a79c11a6bfb2ba9a : CANCELED\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/bc083db2e27c4a1e92d4f8b2eea4c86f : CANCELED\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/7e2b527f85b64ff18c7c8b47526da62f : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef426b3203fa4c9aa0e1a37d35edf5d7 : CANCELED\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/de753b7c7ddd44e0951db2de04a192bf : CANCELED\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6c5df62a971049098ad329686e3e5692 : FAILURE in 16m 53s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c4bc32162a22400bbb0c6d1e9d4a0e46 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bce760b0d1c4379990e2c7df9a2a7e9 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/a2f4fa9828454b7f850cd8bb675925b6 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/142410a3b17d460faa64634db037ac81 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/3cc21079f2b64e66a67cdefbb0f5637d : CANCELED","accounts_in_message":[],"_revision_number":3},{"id":"851f207e534d62f167d30ca60faef78142a26642","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-15 13:05:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1d416564c6f1d6c364665a6f7996753556862d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 15:27:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101143a97ac14e49b7e963fb7de0795e : POST_FAILURE in 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49a5264cdc04a0dad7a3eac4981948c : SUCCESS in 3m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/511b94ce5ff84515bafcc8b241859083 : FAILURE in 15m 12s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/6c8af1daf184473291ecb10379e9a373 : SUCCESS in 47m 35s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e881b8990a8b4a93be974cf8664ca0e0 : SUCCESS in 46m 11s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/b71bd6897fa146b6815ba28e0f056dc2 : SUCCESS in 51m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9496ee9c98344a9a825724b3eec670a9 : FAILURE in 11m 31s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d8d8b01b10274a9b97be2a8a17b687a8 : FAILURE in 15m 02s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/913beba7958943a6adaf5bbf283160b8 : FAILURE in 15m 48s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3dae52d590a54924858a79acdb52639b : SUCCESS in 1h 35m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/582268eea68343318fc3d1a03ba203b4 : FAILURE in 11m 31s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/5e50578819aa40b2a0557bf723dc0480 : FAILURE in 14m 12s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/37246e950b364a4c93134be111d1df35 : FAILURE in 15m 52s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/af56caaacb1f49b1b78a8ea4abbaa421 : FAILURE in 15m 38s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d178df0b75d641599b64eac060275de6 : FAILURE in 11m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/47728ac836da4f1e99a6a6fb59af04c4 : POST_FAILURE in 1h 57m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e868474b3404c2989da6a675eb00743 : SUCCESS in 2h 17m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45f321d71505472cb0c0836b730db6d1 : POST_FAILURE in 11m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/adbaaf900f364058a8262f75ac17fabf : FAILURE in 14m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/40e946fc55f949d19d48827ee42759bd : FAILURE in 15m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b879c36e349d493eb1c4bc24043c5c46 : SUCCESS in 38m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1155ee685e1b40658bbcdf4462bbf6d4 : SUCCESS in 37m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/a246bd67e7a7444db24b2522487167b2 : FAILURE in 15m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/cf6899a002fd4a2bb08986c3c6624263 : POST_FAILURE in 16m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/92c6084e6e354f38a70f1e72093ba9ca : FAILURE in 15m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6cfc57ebfe6044ed968a8c78615e60d4 : SUCCESS in 2h 15m 09s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/23620b58411145abbf60decbfd7d55ad : FAILURE in 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"47fab40a8b156ce6a4b9fbc536ca88c3b6f8fd77","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-16 06:45:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2c5989d5591eb6b3a423e39c15ec9f31932d7a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 09:28:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f8bd0b2feb46829e2bf4937b0ce8ed : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b1d234df7842cda3f881716d35da83 : SUCCESS in 5m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c43168612aca43c9a4ac6bfd95e60533 : FAILURE in 45m 29s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/c2c4d9ee02b2402da48459b5638ac749 : SUCCESS in 39m 52s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/92dad70411df44ab97bf5e9a760b0662 : SUCCESS in 56m 28s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/2c332c28e56045488abf586e3d1de880 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7f500ea047f4ecd94bd1af8e1a40a26 : SUCCESS in 2h 24m 39s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5f49c6bfa5e440f38eea7fef2218fa4e : TIMED_OUT in 2h 35m 02s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/67fd4cf99b1e412ab1884ae4486dbfa0 : SUCCESS in 2h 28m 09s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ab20ff33f8ea4c778e02fe988c181941 : SUCCESS in 1h 25m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd882aad39e84bbebbe5356441470cda : SUCCESS in 43m 35s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/0ae7881e3ca948a89c2a5bd59aba1586 : SUCCESS in 42m 31s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/62edcc9231a64c8e9b095d4b06216128 : SUCCESS in 46m 22s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/2c130598e74b4b47b16c06a2a7da4108 : SUCCESS in 42m 34s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/921285bf19e84fde9ba54c1b128355ac : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3497509e9b1444cbab5aeb5d7ddebbb0 : FAILURE in 2h 24m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66d8afed0a8e43fa9bb0131912c226f1 : SUCCESS in 2h 01m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ad374f9fd8243fd9891be90080eed81 : SUCCESS in 41m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cda2ba95e9e24f1688a9b23b6b790ee1 : SUCCESS in 56m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3a218292063d4fc7b1745ad31b4e57d4 : SUCCESS in 42m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6592063fa50e4af3835ccef68388716b : SUCCESS in 42m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/14d5f69439d64211938457b6ee1e0730 : SUCCESS in 40m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/98fec020abb04b808906dd3b395f5d3f : SUCCESS in 40m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/523efcfada7f4015a0a78b09dd3d5b02 : SUCCESS in 43m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/c46a7c20402c47cda5593c88bfca8fd0 : SUCCESS in 56m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fb040ca392af40d18980fbdc247f2af7 : SUCCESS in 1h 56m 34s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c0e7a6effecf40228abdafdc80213097 : FAILURE in 43m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65fb22607b8c6b83f88e3b5002c56b3786cb01f7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-16 17:31:38.000000000","message":"Patch Set 3:\n\nrecheck\npypa fixed","accounts_in_message":[],"_revision_number":3},{"id":"1269ab4a2408c88aa13a98f04fe2c24a97433125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 20:02:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394c4defc21c48eea6bc3603fcc44cae : SUCCESS in 25m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18fddd559344884b7b3663bb0373ca1 : SUCCESS in 4m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/205d2d538409486d8b7609fb36fd7798 : SUCCESS in 48m 04s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/d35f85afc22549f6a02015e4e4761943 : SUCCESS in 53m 20s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/8486ce38777d4ed7af0fc162acec5b80 : SUCCESS in 55m 22s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/10dc497411b94693bc218b05286a7492 : SUCCESS in 49m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f09c674786b540b3ab805697ed10c8ac : SUCCESS in 1h 46m 13s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5e5c7cbcde0a4530891a1a34f2b3dd0f : SUCCESS in 1h 56m 46s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e787d90a531648b2afdf569e142eba55 : SUCCESS in 2h 12m 14s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/87fed9a910ff486dbe85dd3944243fd1 : SUCCESS in 1h 38m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7afff3e5bb74c6791fe59223106349e : SUCCESS in 1h 55m 33s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/449d1ac33dd54cad9246fe8de0389236 : SUCCESS in 1h 53m 49s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/806fcd8bca604d49bae8848393567b6d : SUCCESS in 43m 02s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/f56a2d9757384214a2670bb4850d9916 : SUCCESS in 59m 21s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ddf1a681a8ce4d92a40c27c07e975afd : SUCCESS in 2h 19m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4c87858a6f334fb498ffe6d16cd827cb : SUCCESS in 2h 23m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6524559170504b98a0752b2726873bd8 : SUCCESS in 2h 11m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a12be360b8c49648638e8dbb988d3a5 : SUCCESS in 1h 10m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/658c9c41de98492393b878cf122e1cf5 : SUCCESS in 56m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bc2eca1ea03944e1babac910bc8a1e39 : SUCCESS in 40m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3f8cc2fb82b8462bb887ea5171886186 : SUCCESS in 39m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5ebe426281d846f7b1d6caac32086974 : SUCCESS in 51m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/d62510ee88f54ffc9aa0b5df5ce0afca : SUCCESS in 42m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/23e9487a55cc471ba83ba72edd67d037 : SUCCESS in 39m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/7b8bca3049da47d99b95861a135c839b : SUCCESS in 39m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3be19179abd74d2f867825967a15738c : SUCCESS in 2h 00m 44s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/be974dbcf320469b92d178bb1c7ad1ef : SUCCESS in 38m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54675390166b04a4216e854dec99debc8e8019d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 20:02:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6087dceb13c7a9263b6e181dd549f68396b9e7cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:37:56.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311bf5e773ca4649b89e0fd450ce91a8 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5490400a1b24b4a8cf07d79b38addd8 : SUCCESS in 4m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e9a096b43c0644d8be41b92320cc6aeb : SUCCESS in 45m 56s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/72b072ae27b148a1a4b00c5c8b580cbf : SUCCESS in 50m 33s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/8b893879737a4584a0197a0b95790a43 : SUCCESS in 45m 18s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/29d6817bddee4d2eb2fe78859027b215 : SUCCESS in 44m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecc1e278e0224732a1fe48311ad2ad4c : SUCCESS in 2h 03m 24s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f71934849687448a843492898fa40888 : CANCELED\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ddd7ce14082e43e3a33e1d40bc2675e1 : TIMED_OUT in 2h 33m 13s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7d9130e32721421cb5d744b2b345582d : SUCCESS in 1h 51m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec4eac3a0ac944d7851d7d4173d76c41 : SUCCESS in 50m 00s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/365b858b78a74fd9a23d6dac3ccfd680 : SUCCESS in 43m 26s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/6896106e010243f392f5b96f2fac2417 : SUCCESS in 1h 03m 27s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/7d60b586da92405e95e2a59012051bdb : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":3},{"id":"b1429b14ae47aae13f022f34a05cb0a976517716","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-16 22:41:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5875defaaa6c711ed346d911da1dd316d468a9de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 01:12:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97e10295ec74ac2b65314cbd63bb749 : SUCCESS in 14m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af2fa30e6df40b19c6fe7a1121f71fd : SUCCESS in 4m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0ed3212c4de41ab93d11dc1c6058441 : SUCCESS in 37m 15s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/bae628ab87c34b9ea8e12b9b1e4e607b : SUCCESS in 39m 16s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c34a1a932d5b413c9ffb7f9f2a7a1455 : SUCCESS in 41m 34s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/24f58df15c7e444a90b3b0d028917732 : SUCCESS in 38m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56c7226294874a7e90613d5a2fa9678a : SUCCESS in 2h 11m 52s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ee61ded5db5046978da689772d74bf1c : SUCCESS in 1h 50m 33s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/872efba95e034e56836a37796ce840cb : SUCCESS in 1h 59m 59s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/eb89dbfbf2804f3783dbed40dfc8eb6d : SUCCESS in 1h 24m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f947347a50bc4043b8c377bf67cf869c : SUCCESS in 44m 23s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/10a212ef45e64511824c5fbdfe87532a : SUCCESS in 44m 38s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/151aafb4612348698008a1c4a1b0ee8a : SUCCESS in 43m 41s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/f830646dfd434ec9aad16d9ea7252060 : SUCCESS in 47m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/577a6be4068d418ea7eb77646e3903d4 : SUCCESS in 1h 41m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5a6934ce3604405cbd532ad78a012181 : SUCCESS in 2h 21m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac8f7484a2884310be38895322a7ab7f : SUCCESS in 2h 29m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/140cbe515d184e5ba1d7e8c6f19b2e04 : SUCCESS in 40m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8e347e1f9054442b8ac19a6429b343f1 : SUCCESS in 41m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0515b6a5357c45578c0e27e56497b9d0 : SUCCESS in 36m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/edbbf073452046d888539f36c6b6e8f5 : SUCCESS in 41m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/79bddc2a2b7e496dabb782b4c4033716 : SUCCESS in 35m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/9d6c409579794e78baa75cc79ee96448 : SUCCESS in 40m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/44024b0cb21743c0a1292c9a2ee594ae : SUCCESS in 41m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/40237faf2ede4746b897e314f4109519 : SUCCESS in 40m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0927c106aa7144f981d97e796986a069 : SUCCESS in 1h 41m 24s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f680a9ed4d7b4bd8be35c94269e7db23 : SUCCESS in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d0ed31a072053edb20d359bfce0ebc26f34bd437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 01:12:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"02452eaace809571fbae5923bae67d1295ec16d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 04:09:34.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc19ea7f5d4a4472854b887cd00bb55c : SUCCESS in 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34d9d6bea864b08b1f6f5352c41c756 : SUCCESS in 3m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0cb5a9c39051448eb55602344b241081 : SUCCESS in 42m 55s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/ab63fd1c3a38460ea5ed01e324b36535 : SUCCESS in 42m 14s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/12ba5bd8e4fe4d9aaadaaedc8f8d0b3b : SUCCESS in 42m 37s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/0a1bf2555d9c471087c5b287eb16038b : SUCCESS in 42m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0809fad6fe574615964a949e9b4de0fd : CANCELED\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e97593aa1ee14efe82b63824bc39fc72 : CANCELED\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/efa93f4f31b5494e854a74b0a3f4214b : CANCELED\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/63d4f9e76d8a4399a6f6676b97597250 : FAILURE in 1h 23m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3c172e344364c4a955eb7bfbfe305d2 : SUCCESS in 40m 53s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/873551eb9e5d4dd398e9c2eb7f61f3cc : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/61cc0aefb48840f283bd9a85e509630f : SUCCESS in 45m 30s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/47a59fa90f6b44649fca48178bbb1958 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":3},{"id":"d0ec20cfbf32194a00c991e5dc992ea5090200ee","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-17 11:26:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3bef183311cb5d934452d8e2aed0dbcc8625438c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:10:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f30d5055cc748aa925b094c77969e41 : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2635e6475d146a3a608d60d0b40127b : SUCCESS in 5m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be1e83a3d90547acb67986bcc31daf20 : SUCCESS in 1h 47m 20s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/00428aaf568746a9a39138c5f5f66558 : SUCCESS in 1h 30m 12s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/fef8e2e7b36f45a899acf643753cb35a : SUCCESS in 1h 45m 38s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f791320f38194a37a78375dce41aed20 : SUCCESS in 1h 17m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9265ce47e416436faf2baf36dd83465f : SUCCESS in 1h 48m 46s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/51db1b91748d4d3192bae3641c1bb40c : SUCCESS in 1h 49m 41s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6b978befacae42748459a7d16cbd5470 : SUCCESS in 1h 59m 10s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c18bf8ca21684cefaeb4b23c90d9c5ee : SUCCESS in 1h 48m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/632d21bb37624f25be7d14d8775ab36e : SUCCESS in 1h 06m 20s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/4b077ef8a881467a9762b76fdfa50567 : SUCCESS in 41m 29s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/18106e380e0b43ef8616bdad5203bef2 : SUCCESS in 1h 16m 08s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/c4bddf7fb77e4f00b75d007a20d09e36 : SUCCESS in 1h 18m 31s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a5379cf9fd1148a99a2ed49c47a33961 : TIMED_OUT in 2h 35m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/709bc8fd93a04bc1ae74717050899f0e : TIMED_OUT in 2h 35m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b2fba4a1070c463685e8e1279795f81a : TIMED_OUT in 2h 33m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/559cb55b63174b4d93e0cc0b05d57066 : SUCCESS in 39m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/659a3634df6b48329b3125e18a9a6c4e : SUCCESS in 39m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3218c013190f4c24b094cf688cd61dd4 : SUCCESS in 42m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b5bd5343cef1492a886129dfa3f22280 : SUCCESS in 42m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/030cba4fb86e43a1b1c4de80e46a9e4f : SUCCESS in 38m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/c73cea8454994a6cabc4b0a07fb48493 : SUCCESS in 38m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/1b8fc8dc01a64e0a9e53658d7a9d9f0c : SUCCESS in 49m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f002146c14e64650981c269cd2798f92 : SUCCESS in 41m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ffd137bbcb3b426d9923cdb2632f1d39 : SUCCESS in 2h 16m 00s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f83c243c051f428aaf403fdd1d2aa109 : SUCCESS in 45m 32s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/cdf8add377ef49adb77be638419fe21d : SUCCESS in 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"75938fdb90f79b2eb84d8f4f6fd366930d28f98e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:10:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b370f6512902e5a81dc5971443ca1245e9a06eae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 16:11:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c529e9b05e4af5836d99717f0eac6f : SUCCESS in 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e23cb0dda6d949f28c7a89ba90114c43 : SUCCESS in 3m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3aaf0ba938294491b035228a6ac5599e : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/ea9d0b915d374feaae46935a689e622b : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/582d883ac84544acaf0ce06d5e735194 : SUCCESS in 1h 40m 58s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6fdd8ec738204e8c9b8e9d7a4335e48a : SUCCESS in 1h 17m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c358ab15afe94531968fca74661a83b4 : FAILURE in 1h 55m 32s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2f9ff18951f742d2bb46b29e2c87c469 : CANCELED\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f2f5ec51f609462baed1fb4acdef6305 : CANCELED\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6134fa67073a43938a4cea1996a64f08 : SUCCESS in 1h 39m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e209512c5e0f42388db9c6acc70957fb : SUCCESS in 1h 10m 40s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/e58fbe79a39243ada9da68abd420e60b : SUCCESS in 50m 02s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/63526ca71d494139a0a1b3d421f5f8d0 : SUCCESS in 59m 31s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/a68ab33cd96d481f9e346459a9e8d01c : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":3},{"id":"c625d3b7b797fce718e553e0bd944bd7083e9152","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-18 07:33:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d3b01bb09a31d07f0efd38f29596f7e0b1e18c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 10:15:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc942ff85a15451e84c24fccb4b5455d : SUCCESS in 23m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf678638d0d438cb78b9eb5e03657eb : SUCCESS in 5m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7c8249a31fff4833bc775a34d86c8057 : SUCCESS in 1h 34m 40s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/a101ad0cf1ad4add9922ea20bb4cc82d : SUCCESS in 1h 53m 19s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/00f794a7b21f47ca9627f00ece7688c9 : SUCCESS in 1h 40m 29s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/2e4b15f58f3b4e90ae252551bad8108b : SUCCESS in 1h 22m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63cb14a1836b40949156b595b5a86e84 : SUCCESS in 1h 46m 04s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/86f95912d81d45bbb44c8937572bb05b : SUCCESS in 1h 50m 45s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/db95f91d7bb34d17b28757c6736e05f1 : FAILURE in 2h 01m 01s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6c70518fe82c42628028b5ca12a9ddc9 : FAILURE in 1h 27m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bb7e222fe91f4de9aa53e31f1f16d65f : SUCCESS in 38m 42s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/fc5cdd5541774e809ea2400b3d42cb7b : SUCCESS in 45m 52s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3227962ae5e54ed09dce8cfcc8ccabf5 : SUCCESS in 52m 22s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/3089e71d78384ac5bbb648d2d456ae12 : SUCCESS in 49m 41s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/864c3b870b704af48bae04891480b5f5 : SUCCESS in 2h 08m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ec7d25585ae14218bf5ab8d3be1f7341 : SUCCESS in 2h 13m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/abb870823e504c55971c665b7249e810 : FAILURE in 2h 33m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9f5dd002938485f9def56a4ba1dd908 : SUCCESS in 37m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/859676ac5e204190be2e516cdf0920db : SUCCESS in 1h 04m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4219832335e445779331ed40834ae28a : SUCCESS in 47m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/2eef45ad51e345f0a12647071cabbd31 : SUCCESS in 1h 06m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/316db0a5dfd04ddba1cad43f590c2ce2 : SUCCESS in 44m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/98b8ebbd40674dd8a1baf73ec7ecdc52 : SUCCESS in 58m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/0c8b8b0705df4718bf2e8fc97e1b4595 : SUCCESS in 45m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/29f5efdfc74c447d9dc9fd9912c7789e : SUCCESS in 1h 01m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/90f1f4c15f5c4904950352b8e134ef9a : SUCCESS in 1h 50m 59s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/7408a0b0bbc34a1094d61646cee03021 : SUCCESS in 46m 41s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7a3ec6c0543a41988d7ebc89c8b28447 : SUCCESS in 40m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd5aac6d8cd46a5f01eb40566971ef790a967b7f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-21 09:21:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0bcd8ca1877f8ea30ffd6a3b3f3894edc62c0ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 15:34:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81e70016816455f9e943ad6b10e7dd3 : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3036bc8d6e8b4db99c0cbb957d1320ef : SUCCESS in 3m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c13b4280dc2d48af9c241288f6aab4db : SUCCESS in 1h 59m 22s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/621f516d23df4a37810eae1ed2e5fe6e : SUCCESS in 1h 43m 51s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/1d7c6d72fcd64248aeb912047e959115 : SUCCESS in 1h 51m 26s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/70543864bc204dddaf712a340e0636d0 : SUCCESS in 1h 20m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7bc9a9329da04793afba4cd4e4d35f98 : SUCCESS in 1h 54m 54s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e4fedc07831148bd9ee54ee17cbfdba5 : SUCCESS in 2h 10m 44s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/82e57768f7d04540ac9d25eff6a006a6 : SUCCESS in 2h 13m 47s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/5dbf312005874a30a06f219c5a1815bb : SUCCESS in 1h 33m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c70b2a3653804a65a7e7bd989fd60cdc : SUCCESS in 1h 01m 14s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/3d6e2630e4924e9189cea018b5575581 : SUCCESS in 42m 11s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ca3d0879cd674ecabc7da71ca2c01082 : FAILURE in 48m 03s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/1ef89fba7bee4dd2a9bd53383747f9ed : SUCCESS in 45m 29s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a794bd2f48a442cabe1a7e7740df6060 : SUCCESS in 2h 14m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a5c78e31b2d4466eae8c3652f9c704b0 : SUCCESS in 2h 18m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad3ec776355a47778bd7b7abf3d206b4 : SUCCESS in 2h 25m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15008a60b1ee4cbfa1b09258af1facfa : SUCCESS in 37m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/81ed40208fbc4b109108756f52d980aa : SUCCESS in 37m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2d142ab4d8b84361952d64a1b023e006 : SUCCESS in 37m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3381f606a39c41849e5b6dd43b9573a8 : SUCCESS in 39m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/32705ff13d674f89a37d8a95afde576c : SUCCESS in 58m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/696ef4db04ca40dd90d5040e760c65af : SUCCESS in 37m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b66779f364074b3ca317ddda3e935f8b : SUCCESS in 42m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f1f39c8d1b8e4fe4bc41e155c1d94426 : SUCCESS in 35m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/46f325da50ee47ecb10adb96e1a53656 : SUCCESS in 1h 44m 57s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/77de84ca3c2d4cc093ba2f8c98c04a0d : SUCCESS in 1h 01m 47s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5848969d52ea4e5a9bff29995a135c32 : SUCCESS in 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"503b74357c6f0c72285ca6fc2c9f408956a91bd9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 15:36:46.000000000","message":"Patch Set 3:\n\nrecheck\n\nnetwork issues","accounts_in_message":[],"_revision_number":3},{"id":"d69ca22b38e5c13debe68d6195eb295caf63b7a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 21:42:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184ebc18754d496aa44f6ae7c7b3b0df : SUCCESS in 19m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abed2727b70446889c7b5e2141635c83 : SUCCESS in 3m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/caf347e24aed490eb378d114afa67ddb : SUCCESS in 2h 01m 34s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/6d11cb0c44814b3d9de4ec71dff87030 : SUCCESS in 1h 36m 52s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/61884e3adf404cd0a3d4dbd0eab2ba6f : SUCCESS in 1h 31m 22s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/7b9520bfbff54e60b1bd7d7888c1a5b1 : SUCCESS in 1h 33m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50ca0a7ca0744af281b99ffdca2b4155 : SUCCESS in 2h 13m 34s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/af7b04530b6c4ba2a7561fbba1c310ba : SUCCESS in 1h 54m 42s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/825c517541bf44b093bdc0942edfcd08 : SUCCESS in 2h 23m 47s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/49f75f556e2b4b9b9940324f8172174c : SUCCESS in 1h 17m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c1907b1da6f4604829a0bd1aaf61387 : SUCCESS in 40m 10s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/0c147176ff5740e2aef9ccce3ba9ec81 : SUCCESS in 41m 14s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f1231bf4527e42b6a7ae6f2f554d7a85 : SUCCESS in 40m 51s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/2d148262aade45f3a3a2f02e222fc7a6 : SUCCESS in 44m 53s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8614668d667d466f8759d93d2373131a : SUCCESS in 1h 52m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5151ec1e2d4241f29eee20f64bbe06af : SUCCESS in 1h 54m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d8cdb2ce7c144d7a908ce8737b17f301 : FAILURE in 2h 16m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b65d3e59b6c04cbda14cb5b8b7718d0f : SUCCESS in 37m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/47ed7e879906488c942bf3f1d87b028f : SUCCESS in 38m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5c22ffd540bb49c0924dde4cb7a6723f : SUCCESS in 38m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/959aa29cc3444fe0934ea6a13f410658 : SUCCESS in 40m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7711e0853beb415195b2d5a3191bfab7 : SUCCESS in 41m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/02d0f95719f24764a60c30b73ae29b45 : SUCCESS in 41m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/65818008a3884d73904d36dfab82060d : SUCCESS in 40m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/4ef6223ec3fc44d9936b5d75d524261f : SUCCESS in 36m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e7417d5e760444d80a8a2d13c538e25 : SUCCESS in 2h 16m 45s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b4de74c874a04dee8949e5b31fcc7c76 : SUCCESS in 39m 22s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/21f216b6ef304dfcba4006ce154dcb73 : SUCCESS in 39m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7b5f160aca29efe1a35b3ddb0677cab88e6f94b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 21:42:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e72f2ab4ff08560c3ec69c66b79a2ae7c47235dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 02:18:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b06277acc4f473aac3f30d3a758b034 : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/897b025d15074d0b8d57040f95bedc8a : SUCCESS in 3m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dedf35804595499bbb505894750db7c6 : SUCCESS in 1h 42m 32s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/9e5dcde69a7b420aba01de829a5be165 : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/9c2b0179b6fb46d592df76fe851661b6 : SUCCESS in 1h 32m 13s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6f859e7743284c879101541a622e915a : SUCCESS in 1h 32m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b91ffc20efe94573ae6ab00970e9aa28 : SUCCESS in 2h 05m 05s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ad910fa55c97447a8cdd9b5b6483f2b4 : SUCCESS in 2h 25m 30s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ef43a6fe527948b2ad488b58ed1acef6 : SUCCESS in 2h 21m 39s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/252069d6521e419984446b605da038ca : SUCCESS in 1h 16m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db264df17deb4443b5f5a3d985d629c2 : SUCCESS in 38m 41s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/51f6032f668e47f49623b00ed515366e : SUCCESS in 45m 07s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/4547147b39e9480ebaff144b90c185a2 : SUCCESS in 45m 48s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/7b9dc4c83a2b4d12a4426cae586d806c : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":3},{"id":"552555940e9eedadd3b048abc57ba2eb4ec1bb5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 02:18:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bdd589072060ed8e4f3a8caa15fd78683925d17a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 02:19:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c1f1bd89a34b8799fccd7d79a18d68 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f7718ef862c568f7c761c17ac9736e650ececb73","revisions":{"5dbf48a967084a65ef1b87e06d27296d379b2098":{"kind":"REWORK","_number":1,"created":"2020-08-18 14:29:16.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/37/746737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/37/746737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/1"}}},"commit":{"parents":[{"commit":"6fbfed03d31a3041a62ca621551f22829f42da2d","subject":"Add skip_if_not_implemented to the service client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6fbfed03d31a3041a62ca621551f22829f42da2d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-18 14:21:36.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-18 14:29:05.000000000","tz":0},"subject":"Add ALPN protocol scenario tests","message":"Add ALPN protocol scenario tests\n\nDepends-On: https://review.opendev.org/#/c/744520/\nChange-Id: I4ae7a03d2248c970d7bcd3fe8a43a43ca48d5084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5dbf48a967084a65ef1b87e06d27296d379b2098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5dbf48a967084a65ef1b87e06d27296d379b2098"}]},"branch":"refs/heads/master"},"8af18236c440d4213dd29070eb5f8e285f2d991f":{"kind":"REWORK","_number":2,"created":"2020-08-18 15:52:36.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/37/746737/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/37/746737/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/2"}}},"commit":{"parents":[{"commit":"6fbfed03d31a3041a62ca621551f22829f42da2d","subject":"Add skip_if_not_implemented to the service client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6fbfed03d31a3041a62ca621551f22829f42da2d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-18 14:21:36.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-18 15:51:37.000000000","tz":0},"subject":"Add ALPN protocol scenario tests","message":"Add ALPN protocol scenario tests\n\nDepends-On: https://review.opendev.org/#/c/744520/\nChange-Id: I4ae7a03d2248c970d7bcd3fe8a43a43ca48d5084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8af18236c440d4213dd29070eb5f8e285f2d991f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8af18236c440d4213dd29070eb5f8e285f2d991f"}]},"branch":"refs/heads/master"},"f7718ef862c568f7c761c17ac9736e650ececb73":{"kind":"REWORK","_number":3,"created":"2020-08-25 13:53:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/37/746737/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/37/746737/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/37/746737/3"}}},"commit":{"parents":[{"commit":"bb23855ce65500dda6a0e332eab7174ed493a0ac","subject":"Add tests for allowed CIDRs in listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb23855ce65500dda6a0e332eab7174ed493a0ac"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-18 14:21:36.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 13:53:30.000000000","tz":0},"subject":"Add ALPN protocol scenario tests","message":"Add ALPN protocol scenario tests\n\nDepends-On: https://review.opendev.org/#/c/744520/\nChange-Id: I4ae7a03d2248c970d7bcd3fe8a43a43ca48d5084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f7718ef862c568f7c761c17ac9736e650ececb73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f7718ef862c568f7c761c17ac9736e650ececb73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
