)]}'
{"id":"openstack%2Foctavia-tempest-plugin~747629","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I4dba477053cc71c54b9494c8f6dc6cd4b97687aa","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"octavia-arm64","hashtags":[],"change_id":"I4dba477053cc71c54b9494c8f6dc6cd4b97687aa","subject":"Add non-voting aarch64 scenario job","status":"NEW","created":"2020-08-24 07:20:15.000000000","updated":"2021-03-11 11:36:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"019aa08c298f3d8def96be406631b7c60b7d74ae","_number":747629,"virtual_id_number":747629,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-11 11:08:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-1,"date":"2021-03-11 09:07:38.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 17:05:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-22 15:38:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-28 15:40:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"677164357050c4ca2d7ad69d83ec4029379e1a0f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 07:20:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a3ad8bfc32cb025cd8a816412cc5a7a23e10b71","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-24 07:20:30.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5529780f2d48a19326ec6c0a450bd306f1aaf5a9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 19:20:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae7a94831cf8abe35732f04ceb66dd9e131a88a9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 19:20:41.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d5c7d77311659e0c7066fb64559b8384be36d761","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 19:42:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a1b4eb50c50d52970c8e9d6a4c121c0cb68268ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 20:57:45.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/cad75e561a6243cca2d82bddae6da464 : FAILURE in 1h 11m 10s","accounts_in_message":[],"_revision_number":3},{"id":"1c58136f7ecfa62568f730f8093b1d2f5935adcf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 21:04:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5b28b0a2a4b62346d22d5383893ef86ef657c50e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:32:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991ad050266b4b51aec0c27062445b7f : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b65ababc6e40d9b574f10f4f99a4f6 : SUCCESS in 3m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e04aa15b3cb4e40aa00c9fa47f16457 : SUCCESS in 30m 59s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/36f4f1abeafa4f268b94018650284a9f : SUCCESS in 56m 53s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/092d054fe27c4f67a6e0410b7b93657b : SUCCESS in 35m 51s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/2cfb30e2628240ed82a77dfec3d8c04f : SUCCESS in 57m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1b836efcc6c4767a7b44a03f86d4c8d : SUCCESS in 44m 35s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b88c33bf393c40749a5febbf99ccb5bf : SUCCESS in 43m 03s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/095d1d325d924ac0a6c601efcba38785 : SUCCESS in 50m 16s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c80c2781d07a4a40bee80578fe85b96b : FAILURE in 41m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bd518587934649b6bc7ef52309b5923f : SUCCESS in 35m 29s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/a7d23428ea7340ae9075440e7e6498a2 : SUCCESS in 30m 02s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/4d999fdeed094305a1de2b566e8a4534 : SUCCESS in 33m 15s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/809d7248ccb645a39d4cad205084a15b : SUCCESS in 33m 32s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a5f10cffdd184ef9a113fcb51f8fe26c : SUCCESS in 45m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9e83a4b338294eb08db650fd0ab40923 : FAILURE in 38m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f4467e2a93f54d17afa2a8c3ed8c6c1b : SUCCESS in 1h 25m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/478a1891dd694d9b8107623a3cb491f5 : SUCCESS in 37m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e2f3ef880fdf456e8be74565c250cbb7 : SUCCESS in 53m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f7615547ffc449d6b2d63e6185a7441e : SUCCESS in 54m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/62929211765c49e0ac9ec5a86413f149 : SUCCESS in 55m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5bdf82aad9ab42cb93e6de0dea0cdd1f : SUCCESS in 52m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2f2a9e80d16e4e50853dcc3287e340db : SUCCESS in 1h 00m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/8c4b43c1c20845c8a1a235c066cecc5b : SUCCESS in 55m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0239b1040f1d4976a1a9e22541b824e4 : SUCCESS in 42m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/546707ee05f14cc0bb08ab76ee10b815 : SUCCESS in 58m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"55da5ba9b646dc3113f69601d2ed0f20a8d7177f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:38:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/b7dd27a1ec38454dbd0277a1463f2a21 : SUCCESS in 1h 31m 08s","accounts_in_message":[],"_revision_number":4},{"id":"2a29107fe7d22c6f9a19e04f4de53ea25a0ae387","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 22:41:11.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f3f41254600f0bc32e47c0ac7db3b06be40c12e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 11:09:01.000000000","message":"Patch Set 4: -Workflow\n\nThe ARM64 Zuul pipeline is non-voting so no need to explicitly set this job non-voting.\nhttps://opendev.org/openstack/project-config/src/branch/master/zuul.d/pipelines.yaml#L410","accounts_in_message":[],"_revision_number":4},{"id":"17d9c3f31de69ffbc6f27c2ff74efa8caccf87f9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 18:59:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d27074157b60ba26b789f49c13937acbb0f7d603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:25:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04746c4dc59444c5af863bc97dc79865 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dea5a70e9344d1ba24b9f891b9dca40 : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a564fc95beab4a9e83907d08015e6aba : SUCCESS in 49m 59s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/92d5a51e18ba44f4b317b81b11b3197a : SUCCESS in 48m 02s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/593a30ea6c6e421193b0f31055130fdc : SUCCESS in 52m 30s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/14abbd816d8e4b6ea75f60ccfd52e5a6 : SUCCESS in 42m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3aa4d203367f4ecb999f1a114cc8860a : SUCCESS in 1h 09m 39s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6d3fc403769c497093078fcf6522bdcc : SUCCESS in 1h 20m 06s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a4cecc7436174718a2d41f5c7fa7d897 : SUCCESS in 59m 51s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/94a3ea27984f45608edcc063465c5bb7 : SUCCESS in 1h 01m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36225e242d69498181940ed1c24d5ebb : SUCCESS in 41m 25s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/bc0542d2720e44f5a46776d6e386146d : SUCCESS in 33m 18s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/7086acaf052a447ab0a1c1167ed8678c : SUCCESS in 40m 30s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/956153e33a4341bb99d285049695fe8a : SUCCESS in 51m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/550d4b99ac4c4d65b035b4335ed170c2 : SUCCESS in 1h 03m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fe551c3520d94bd2816f75dad4d29a1d : SUCCESS in 1h 07m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3e1773de48354b55b1a3132ce02e4379 : SUCCESS in 1h 02m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba010136e1604bbf8da917c2acdfbfb9 : SUCCESS in 38m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ddb141e87f5b48e1aac9332a19aa5522 : SUCCESS in 32m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0da8203e900e467cad0f246aecf33227 : SUCCESS in 34m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fd95ab6489b949078e88a4c5757d4659 : SUCCESS in 47m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8f03f940778a430c96530fb6f35c7adb : SUCCESS in 46m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/aa035ac8817a4db599ac1fba4690c194 : SUCCESS in 32m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/52d335a137ae402c9de6d543ec5cd83a : SUCCESS in 44m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/43cb494d99c24fb89ca132da02be36eb : SUCCESS in 51m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b37474b7ac124ad394750cce248f6483 : SUCCESS in 50m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6a7906877fc5143340fefd14ad34037fa453536f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:34:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/e192556c7fda4a08a5716d4a0dcc7d0f : SUCCESS in 1h 32m 19s","accounts_in_message":[],"_revision_number":4},{"id":"9bf89663440283dc9f6953e6050db6f94c280c33","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-04 17:05:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2b984b1185f9250b95d1f83c91a7d6420ca2ded8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-11 17:00:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fbc4672ee2fde4a270fbc9f5a3c348ec6b8385e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 17:04:36.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"816a07091dddfaeb7a610d9afd50926c53320ea4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-11 17:05:06.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c2ebdcc194a95d32dcc7c18d7e338c21e57890b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:17:03.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/3909dee7a1e447a8a950bead9a4f9bdb : FAILURE in 1h 09m 07s","accounts_in_message":[],"_revision_number":5},{"id":"553e25a7e510854f929be2951ded99230ff84679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 19:44:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492af8a94c8440eea12654b6d8c2a598 : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5630d717b7894257a90d8666f7f11b63 : SUCCESS in 4m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0ba1cb1b9b04a08a2c0661e833c0409 : SUCCESS in 49m 19s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8425ee0074a44f39bb9e65c3f18cd0b9 : SUCCESS in 44m 44s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/66abc6f52822443185f3653dcfa80155 : SUCCESS in 50m 43s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/93b70489340c4285973ba757c0b17577 : SUCCESS in 49m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/924312bec44044818beed91899ff3870 : SUCCESS in 1h 31m 51s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d92407c0124648fabf3fb886475bc6fd : SUCCESS in 1h 27m 41s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/fa63930eba7d4dd2ab004a7b7117a7ba : SUCCESS in 1h 48m 20s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/362d64b5cee84b368ff5323340961aa5 : SUCCESS in 1h 13m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/254451ae1e35412292c6aafb4943b0af : SUCCESS in 42m 45s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/1ea80f5391e24e04b7ceaaef077ce3a5 : SUCCESS in 43m 18s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/28e5cf5aeb214ff8938da229787dd527 : SUCCESS in 1h 19m 18s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/760eb5117284477098ca4b78a28945fe : SUCCESS in 1h 22m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/defd3139ec0148fba588a56ba11a435f : FAILURE in 2h 33m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a6803d18be094281bdec386e48a7b8ae : TIMED_OUT in 2h 35m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b07423f194454d858a0ae8306d67234e : SUCCESS in 2h 02m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fba88f5d49ee412c8bd879d11d39f420 : SUCCESS in 1h 09m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/24d0e5251ffb4b9ea2a4d13d70adc03d : SUCCESS in 1h 05m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/90b53757f63440b99abddcec4ac911eb : SUCCESS in 1h 05m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/64b859d189f1417989f145012fdff55e : SUCCESS in 41m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4e4db53a015740658dc6c97c0da170bc : SUCCESS in 34m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/cb16b192e04044ba80978304207ce9a6 : SUCCESS in 36m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/98d94e4f00424741a2fac654340775f2 : SUCCESS in 41m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0c9bd8b35bb148fea435a8a869a8742b : SUCCESS in 41m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5b3aa331cbe748efba3711f779965434 : SUCCESS in 1h 46m 49s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/3bdcd02b08f14c0c8c66cb6984e908e3 : FAILURE in 41m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a5ce02e9f40142922bb5b5f03ff8f2b76e5f5230","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-14 07:14:34.000000000","message":"Patch Set 5:\n\nrecheck\n\naarch64 job failed due to memory allocation issues","accounts_in_message":[],"_revision_number":5},{"id":"25339fd7402d4ff2c099175fc3390cad9f8ea3d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 07:56:23.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"962855dfa0d0fb9318d0a4e089777cf224bc8517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 10:05:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/279e3f5758534d0eb2d22ff786eb6160 : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3331d2372f1546329cc6935c363eca88 : SUCCESS in 5m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/37056497df6e40de9169bd962b24c3bd : SUCCESS in 48m 40s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/ba88cec6bd1e4505adfc785d116f8d9d : SUCCESS in 51m 12s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/84fcf900f89c4a9093dd04c5c05b0b98 : SUCCESS in 51m 56s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/1e9f0f19bce84a74ac1db5d133605d04 : SUCCESS in 51m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e265573394a648229d45fe36fd9806d2 : SUCCESS in 1h 54m 09s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2ec13f5379d041ceb41cc65338a2e2b6 : SUCCESS in 2h 08m 16s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d6cfe3e51e754928b19db2e8e88f1d61 : SUCCESS in 1h 59m 08s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/294b872094984194bd85c5ebc73cde59 : SUCCESS in 1h 21m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6740f5efa1c3408eaea5a69e9261dac8 : SUCCESS in 40m 39s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/eef8ce982a2c4373a395446f565d5b3e : SUCCESS in 40m 04s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/8006dcd7617c4209a54a34a83ec3d72c : SUCCESS in 44m 46s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/d343e6ebf9ff4d4081a0389096fb104e : SUCCESS in 46m 08s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9316df383ab040fea61f1ad761a0c8a6 : SUCCESS in 2h 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/73f1f6d708da4989a53600d72c2eaa7c : SUCCESS in 2h 04m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f97ef706b6c40fabc12a5d53ca1ea10 : TIMED_OUT in 2h 33m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/979118f73e804007a5d13282210d6069 : SUCCESS in 58m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/74be4681e12e4a7b9a03a4fdb5325e5d : SUCCESS in 33m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/43f7b3c7be2f496cb2ddb2f2a8a6c1d7 : SUCCESS in 40m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a94e4694baf6462c895f9d521409d235 : SUCCESS in 33m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d373287515964e5d8d2dcd649475cdc1 : SUCCESS in 39m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/bef71065242f431cac293494a1f87e73 : SUCCESS in 37m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/11417cfcab6342fd98d7e736c01cf681 : SUCCESS in 1h 26m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/d08337550a154dc4be5304dcb6ce6e83 : SUCCESS in 1h 30m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c5ef79cf438e4dee9a11abdbd67cc54d : SUCCESS in 2h 18m 48s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/90e44049acb54b7982697e055d14e45e : FAILURE in 41m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"451bf07e3047d627cf8ff26a2697b10b2e917514","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 09:08:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"059083290afab4e207d852bb258985b44598d031","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:44:54.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/de5b5c62de0d44a98cc558640e670201 : FAILURE in 28m 00s","accounts_in_message":[],"_revision_number":6},{"id":"ddebd28a7fa667311cb394b14c901037a4d49fdd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 12:24:49.000000000","message":"Patch Set 6:\n\ncheck arm64\n\nChange https://review.opendev.org/#/c/751829/ merged. We can now recheck ARM64 pipeline via \"check arm64\".","accounts_in_message":[],"_revision_number":6},{"id":"57fbfc557c848e4cc83cf5d43cd855361c41ba29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 14:31:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b3fcdff039414bb6dd87b544436724 : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6846fd62cc89490e933517717652c7c7 : SUCCESS in 5m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b9e4c1edbaa446b9fe0d8f171c4898f : SUCCESS in 1h 41m 09s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/d835238dfab248eea25874e9e45f5c1c : SUCCESS in 1h 38m 53s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/18bfee6895f742a59aaec06f32c904d3 : SUCCESS in 1h 36m 51s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/85b7afc74487481f84275ca1c195f5f2 : SUCCESS in 1h 18m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf818e215fd14b078eaf12ddbd86b403 : SUCCESS in 2h 03m 56s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f0592ca537534c2ca4c7e524d8ad5e09 : SUCCESS in 2h 13m 57s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/675c3d9c81734199844dd3447ebb3ba0 : SUCCESS in 2h 03m 51s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1682574ea5aa4ac78b30f5dd5e9f76c7 : SUCCESS in 1h 39m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/003752a4bc8f4b1b8760a13f8abe2727 : SUCCESS in 42m 04s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/57b8ca3a3e224fc6b45acaab0a47c8e5 : SUCCESS in 1h 00m 03s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/082f93efa2dc489cb070492e6a348c1a : SUCCESS in 43m 43s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/a111d81d2c9e46e88c2c25c5acc7b11d : SUCCESS in 41m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2b98e924407440f6b20d4f8c3db9fbea : SUCCESS in 1h 58m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ec64b935fe1d4f7a855b4f133d388e08 : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/52dc6ea868374174917d44ee947ad86b : SUCCESS in 2h 02m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74a8d10b893f47b2864bef4da5057443 : SUCCESS in 1h 00m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ed1dcc6d21144e15b0d7e48b68e3a9ca : SUCCESS in 36m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/33c15cdd2b0e4769bc9eec6500773571 : SUCCESS in 47m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/812aa06fbe9b42d8b49ca1121a8fa504 : SUCCESS in 36m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/68fca7dacc524991badf33988765ecc5 : SUCCESS in 33m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/4c57c91b150a481b854edbabe5ba71b0 : SUCCESS in 43m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/15912185f7a244dbba6bca82dd5b0fe3 : SUCCESS in 36m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/b4f8a0efa95c40c7a2dcc78b644d7672 : SUCCESS in 59m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/703d3e56dc6345e5a9b215394f86d775 : SUCCESS in 2h 32m 34s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/dfa7a9dca7e0428e940405faf97be196 : SUCCESS in 39m 15s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/1b1ca2f1df5d4c94b715ad23fd501db2 : SUCCESS in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bfcc946374d11fbc668e9664301cc24170d5e23f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 15:09:13.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/85c0e6be20654750907f20e03e51c3a7 : FAILURE in 1h 23m 19s","accounts_in_message":[],"_revision_number":6},{"id":"b0a6f0ab62383965d513153b528897bb4b60eacf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 15:32:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4c46fb991310920fad1d4811f0cd3c8a3654a193","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 15:41:20.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"2b70b676319497a4c6a3bfe29e66d340c1f032d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 21:44:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c700dc531a64b55b54fef0b20cc2c06 : TIMED_OUT in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e64e48d0adf4b4c8301a53e94cba9ee : SUCCESS in 5m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d15adcc20b248c0acdad6cdb8093673 : FAILURE in 1h 39m 06s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/29b5e04a599c4c1ea87113e7e66949e7 : SUCCESS in 1h 33m 13s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/82905c3d86b64aab9512930b5e22e3d9 : SUCCESS in 1h 44m 57s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/823aff59a9934a528d2cee4ce1d2438c : SUCCESS in 1h 29m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42c9b4d344e64a5f81fe7705b89f8db0 : SUCCESS in 1h 50m 33s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/633a32b595034c69b976264e01703822 : SUCCESS in 1h 50m 41s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ee798e83d8e3447ab0fb45ae6d830e1e : SUCCESS in 2h 17m 54s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/39dc98db90cf4d3aba8795cb6f64f6a7 : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d8de015edfcf46129eadc53f2a8155f8 : SUCCESS in 47m 12s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/a9e02954570740c38c908cc1289b43e2 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/528406cf1a7b4441919a8e05fe15cbb9 : SUCCESS in 39m 18s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4efea8d040e2481e8aeb635c77a3600d : SUCCESS in 41m 42s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/acd51ba4b93343adaa356493934ddf83 : SUCCESS in 1h 49m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c45f00dca1784890a88c4eed4bb734ee : SUCCESS in 2h 28m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a53b8f0bb9994ed2b7e73c5701f13c66 : SUCCESS in 2h 02m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa783023c55c414485d616bdb608075a : SUCCESS in 38m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3dddacaa64964775b5451527079f8af9 : FAILURE in 1h 00m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/7f5b21c0e4884a6a87b42392f41d437b : SUCCESS in 44m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/29efe3b6710641d6bff5c485a85376eb : FAILURE in 1h 03m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cb7a73a6a84f444db7814a9eeaaf26d2 : SUCCESS in 37m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/00dedcb4a8534d83b2a2cbb8df3ca350 : SUCCESS in 37m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/866d020e1255405a8924def4e7974acf : SUCCESS in 41m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/4a2ba0bc4e854660add506315e57a4e4 : SUCCESS in 39m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/edfd3d709cba4e1cbc1729403611b67a : SUCCESS in 2h 06m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d767e6cd7d2a42f69f6332a358984d3d : SUCCESS in 38m 32s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4253c89af68449198dc0705e2be20d5f : SUCCESS in 38m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0873bdd08e23ea118eded6451890e2549ba5f69d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 05:24:46.000000000","message":"Patch Set 7:\n\ncheck arm64\n\nChange https://review.opendev.org/753036 merged. We now have an aarch64 CI system with larger memory (8 GB -\u003e 16 GB).","accounts_in_message":[],"_revision_number":7},{"id":"e247685aabbc815549317660789937f7403157c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 05:38:26.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"6f52412a024534fe002b2a52a6527073164eb4a5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 13:49:15.000000000","message":"Patch Set 7:\n\ncheck arm64","accounts_in_message":[],"_revision_number":7},{"id":"41e52f13d39b51c105a65e861a7c51df1ee51c09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 14:22:02.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/c3b2618d610e4cb1b8773685f4a54c95 : FAILURE in 25m 07s","accounts_in_message":[],"_revision_number":7},{"id":"23db1b27e0eeb724a3ea65f77f5189c132a762b7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 15:38:22.000000000","message":"Patch Set 7:\n\nrecheck mirror failure","accounts_in_message":[],"_revision_number":7},{"id":"7d54a9588760b3311f41dfd9f281dc9c67505042","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 16:12:25.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/0a5235dbf4a54b189d3b08786b12e9e2 : FAILURE in 27m 59s","accounts_in_message":[],"_revision_number":7},{"id":"9341da7abfafbce68f261c47c1316f4ad788f024","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 17:14:38.000000000","message":"Patch Set 7:\n\ncheck arm64\n\nslow network","accounts_in_message":[],"_revision_number":7},{"id":"0c31a108fed286afc09b0daf3596a8a07533326e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 18:18:19.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/0b47cda88381452bad01dd226357530f : FAILURE in 57m 35s","accounts_in_message":[],"_revision_number":7},{"id":"f4c66ec2bf0771c59d1f27a7c79ce03fa0e25f28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 18:36:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa700adfff8b42988522aa2a67b8b98e : SUCCESS in 19m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de6f61ee56e4bbd9c0c94acc0f677a1 : SUCCESS in 4m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ecdb79b078a4b37ac21d99cbc312ff4 : SUCCESS in 1h 53m 48s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/f734677c5ee14ce7ba4cddf647e8f44e : SUCCESS in 2h 00m 19s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/a5249d910ab74147915b18b8af5b31a3 : SUCCESS in 1h 38m 26s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/5e0799e7a8594ca0a71d86375008ad3b : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b52c3a92c46d488a8b3a57b30cf3b885 : SUCCESS in 1h 59m 20s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/bbc3e621df424436947a8176d4b7a8d0 : SUCCESS in 1h 58m 22s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/70885a1b5cb0426d9ae5b4495a459b17 : SUCCESS in 2h 17m 22s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ae8e0707d5c843ba97e0852feaa0baf9 : SUCCESS in 1h 39m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d87c5e9a09c4463bb4b50672680bdc3c : SUCCESS in 45m 37s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/32a681b169844cf7add6c3d15c4e22be : SUCCESS in 40m 19s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/2fb77b26ccbf4529ad5c81d633bdf097 : SUCCESS in 41m 06s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/0eaaa613975740b98cb716c1a00a1c94 : SUCCESS in 41m 09s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a39a62068977421898faf4a1dbfc4cfd : SUCCESS in 2h 05m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1a00a748fd2b4937a73a02beb3a4ec02 : SUCCESS in 2h 07m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da7c87007b584e3e9c85812089bbcec5 : SUCCESS in 2h 00m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd64e2aae77045219535732a12b618e7 : SUCCESS in 40m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9496e197098d415a8ccd5abcf5cb3404 : SUCCESS in 40m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bebc3b01564f4d8ea69d9d89b9129427 : SUCCESS in 39m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b46eb36cb7bb4bd4bf09b216886cd68e : SUCCESS in 39m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d83b45d6503442099ff8a3940af94e22 : SUCCESS in 37m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2c559f2a685042db9cb78525ab309516 : SUCCESS in 39m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/02df84032f764dcca8efe2a15e4fe6d7 : SUCCESS in 37m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/71669350e0c1471fac1b555c797964f2 : SUCCESS in 41m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9fbf8f055b894116b25aeae8c18e69f4 : SUCCESS in 1h 54m 45s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b939c3e27658444bbb9e44209c9a3e81 : SUCCESS in 38m 29s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e5563296a93e486f8e3906bbd332caea : SUCCESS in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ae4c16a5a25069cf096f3a8006ed9c645c31e68b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 20:29:29.000000000","message":"Patch Set 7:\n\ncheck arm64\n\nReported issue to #opendev. Meanwhile, suggestion was to recheck.","accounts_in_message":[],"_revision_number":7},{"id":"c90e45ac30184e5c5ff65ec8af467788890a1e63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 22:18:43.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/d1e95d441f044211bcacb7568ba5da70 : FAILURE in 1h 44m 01s","accounts_in_message":[],"_revision_number":7},{"id":"2643708c60a2b9471dd76d296a937b265bf66a21","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-23 07:45:08.000000000","message":"Patch Set 7:\n\ncheck arm64\n\ngunicorn/ssl/socket errors in amphora","accounts_in_message":[],"_revision_number":7},{"id":"f435d9d72a842a3641b156ce6eae55cd844680ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 09:54:13.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/c24530e0dd5f4ea68365b447f71fe1bd : FAILURE in 1h 52m 16s","accounts_in_message":[],"_revision_number":7},{"id":"375918061fb8c80d6ab020a4dead76f038a6d418","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 08:04:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"36ab4b666533df92afc1978b803ef1ec0dd20e10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 09:27:17.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/d86fa3e452a949b8bf1f4dd112042856 : FAILURE in 56m 22s","accounts_in_message":[],"_revision_number":8},{"id":"2ed68aa80f251ccfdd658d100513bf981e9def46","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 10:27:35.000000000","message":"Patch Set 8:\n\ncheck arm64\n\nOperation too slow. Less than 1000 bytes/sec transferred the last 30 seconds","accounts_in_message":[],"_revision_number":8},{"id":"f3ea31f2844fe349e7f7451065ab835045a30b98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:31:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125fd46ac8e64f1184c0682c797643c1 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2614606ec394c9887b1e081bbc68bc9 : SUCCESS in 3m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53985b94ce644b09bb862bb337dc43bb : SUCCESS in 1h 37m 26s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/69843777e03c46769fb51188ae59179b : SUCCESS in 1h 40m 54s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/82f2fdc9457146949ad32f22eb281bd8 : SUCCESS in 1h 32m 42s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f8eb536613544382a5b981248ad04ad5 : SUCCESS in 1h 27m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3b5808e48bd4ed599064a4e95e07af2 : SUCCESS in 2h 00m 37s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ff5e90463ef049ed9eab61bbe75eb68a : SUCCESS in 2h 20m 04s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/adb60d12785f41f2a7dddf2d765b2e35 : FAILURE in 1h 48m 42s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7c9353e3fca148fa9a2abb6bfa894050 : SUCCESS in 1h 27m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c56ee6e39a84f97b880b7f1973d9955 : SUCCESS in 43m 43s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/89621ccc583b493dae96f8345a97068e : SUCCESS in 41m 53s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/1b80120b79cd48399dfedbd677cb97bb : SUCCESS in 46m 38s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/f260cb42c1994f009a5aefd1e930c3f1 : SUCCESS in 43m 31s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3eb423908e054f61ae1a7b15a1990226 : FAILURE in 2h 06m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5a327d2a6bce44dd84dac814cba55feb : SUCCESS in 2h 07m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/176d7798f22f4a90af7dff1f701603c0 : SUCCESS in 2h 21m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13f0f66d85284c3a9d882ba3a4e4e287 : SUCCESS in 41m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f991271d46de4101acad55c5b97988d9 : SUCCESS in 42m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c652fd96ac384aeab8cfd94b450a1062 : SUCCESS in 39m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ad0e0d77443941c28fdd3b9dc1340412 : SUCCESS in 43m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5e009cd7c4464b73ab3d8175c7bb6d65 : SUCCESS in 36m 47s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/a5c272679fcb4ed7b051e07b31e24ec6 : SUCCESS in 37m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/df6f7fbda4814a8ab03fb4f3695190d4 : SUCCESS in 37m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/2edfc50b06c746168d511eda20a92324 : SUCCESS in 43m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d52fc7f7e140411f8c58f06ed188f191 : SUCCESS in 1h 44m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2f9ec4c681ce4cb99cfd94a79455baa3 : SUCCESS in 41m 59s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/96f830e184614bc682ca1e843e10f0be : SUCCESS in 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8a9f0ead0078871f99ef8265171c86c46d59a194","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:23:16.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/4d7568cc79944fb7934e9b8ce3bbc3fe : FAILURE in 1h 49m 16s","accounts_in_message":[],"_revision_number":8},{"id":"6162125182634e60bf6f7c1118e24fa0cdc7ad8d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-08 09:11:01.000000000","message":"Patch Set 8:\n\ncheck arm64","accounts_in_message":[],"_revision_number":8},{"id":"ad94b0e5dd753bcff4b89a6cd1ff1e7a218a66fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 10:59:54.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/2842735c60e64a1ca02ae417d1784764 : FAILURE in 1h 46m 18s","accounts_in_message":[],"_revision_number":8},{"id":"c704e5e2c718770c88934f8031293df6cd71db57","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 10:06:05.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dd5b05b7ad92ac75900aadc325f3d2c6c16c0e42","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 11:12:33.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"3fef0a7ab7ad232d45192ff5fcf3a76887c1eed5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 12:28:58.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0f2d23079bc1ffe00b12d870502a729cb2ca6422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 14:17:05.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/cf2a4affa6d84bfa923429161fbb54f6 : FAILURE in 1h 39m 15s","accounts_in_message":[],"_revision_number":11},{"id":"7642dc97c2c95cfc0deb673fa5481710456da688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:09:30.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75624d4b3b404804b52f4d53f4f1dda1 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b09bab7a6be4a76b511494d882f26cd : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/51b13caf16ba40b6bfc98a35e2d95053 : SUCCESS in 1h 27m 22s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/b00f55d388c943c9872e5a01d3737c8f : SUCCESS in 1h 23m 59s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/ea57806ce5a649458acc7824db6854db : SUCCESS in 1h 37m 15s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/ba5425790dd14a88a070f3b8738c23f6 : FAILURE in 47m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e18ab7ec74e34687a2d55dcda59170f0 : SUCCESS in 1h 28m 02s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/fe27d65975ac4f8b822d36ce1dee8d7f : SUCCESS in 1h 57m 03s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/71455a4eb69147f19fe3703c297fc1f7 : TIMED_OUT in 2h 35m 22s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/27c7748267ec45f19f7cd54ae08678fb : FAILURE in 19m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ba2098f800554f62ba6efe916c4f918a : SUCCESS in 50m 15s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/fa6703b652b84c79a329ee83b1c7f87a : SUCCESS in 45m 48s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3bbb1c1352124c9985592c9a58e44b48 : SUCCESS in 44m 26s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/64d272e305bb4626b4ed30273f0f9193 : FAILURE in 18m 47s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a4b0587b5b424d80a50bc8ffd1b385e6 : SUCCESS in 2h 06m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/698a53e35eb043d09ca4b67dd125eb8c : SUCCESS in 1h 50m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/40df095d514846d780fd0b86e9282b53 : FAILURE in 1h 50m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53f991b11c5d41a19825f92858c0e405 : SUCCESS in 46m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/00c8a0425061445d8560488b65b14263 : SUCCESS in 45m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/77393f9e67e24e87a2cda86db429acfa : SUCCESS in 41m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/94c8a9a5866548ce98db64c886098e98 : FAILURE in 23m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/02a3d543c98f4d68bf59d01ed2130c2b : SUCCESS in 52m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/a2b563eb6eac40ae9fc87ec28a6d1f76 : SUCCESS in 1h 08m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/295250a6c2934aa485f61ff207e70ea3 : SUCCESS in 46m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/196ed0d9d42e41349e4ad035cbf87ba1 : FAILURE in 19m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8b636c5ba79d4dca8e5aad22a8548ab4 : SUCCESS in 1h 58m 09s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/65b26ca49a7b4b92a477763facd85bc9 : FAILURE in 1h 00m 35s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4ebd94bdf3914aa1a2cfa1f3efb86b0e : SUCCESS in 43m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b697fc934f55cdbd454ff453df186cce1d27905d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-21 10:36:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"4dce2a1a5b3821accda1794dcb9d0c331feabfc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 10:45:42.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"022729d8b3886d9e6d1dd127a9b2f1545c88a23a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-21 10:46:44.000000000","message":"Patch Set 12:\n\ncheck arm64","accounts_in_message":[],"_revision_number":12},{"id":"f17ff6d1c5907fe1905a527ff8231d000ac91596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 10:55:22.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"1a92e9a9828bce6b8bf91b62558b8ff419948173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:45:11.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dedfbb58d7042f583c1691b86cdf9f5 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8164e762687b47bd949b82c0364d1e5a : SUCCESS in 5m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f05ff6ff53f42cd9de69059012fb390 : SUCCESS in 1h 28m 25s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/34fecc33fd2d41479fbd94765ddc95f4 : SUCCESS in 1h 23m 59s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/69c8eda84f804a8f843831f85ed7721c : SUCCESS in 1h 21m 46s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/0aaca04a1b464bc78bccca178f0c301c : RETRY_LIMIT in 15m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/618d9e44ff524fdf88ea190d9dd0aab9 : SUCCESS in 1h 44m 16s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2a2bca50a7704c4abda2791fd7f52ded : SUCCESS in 1h 55m 28s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5e7c5715628a4a82b3ae751618fecf2a : SUCCESS in 1h 44m 53s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fb712921459644d3a289b267a7d6a513 : SUCCESS in 1h 17m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/77ec183c274d4f78bccfbe61b2f9648e : SUCCESS in 45m 39s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/4069af76e38948d4af6eeacd585a5b91 : SUCCESS in 44m 57s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/061517812fac4c519fc4541020987c9d : FAILURE in 1h 15m 46s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/8d0c57891df74126917641e3c8d60571 : SUCCESS in 42m 27s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9760c9a3b5654149b3f8d43bac37b0cc : SUCCESS in 1h 53m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a0615ffd083d4c00b28cd724a97dd644 : SUCCESS in 1h 42m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd85866855714e40bd0b68745071810a : FAILURE in 2h 21m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/377d0207f1624f138ddf66d1f7172a08 : SUCCESS in 45m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2118d0525d534ba1ad0111a46ba32a2c : SUCCESS in 47m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/58bebcba345a48338fcb4a636aab0bfe : SUCCESS in 43m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b18f5948f5e842e5be39c1d3e9bf6713 : SUCCESS in 41m 47s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/adb69440c4a04cbcb72f5da23c73d2b2 : SUCCESS in 41m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/86f6db4902fb4c5fadec3b55e583dadc : SUCCESS in 40m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/c64cfaddda624ff5be6f9c51e8cf42a5 : SUCCESS in 37m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/fee748f7702f475186c4372299e18275 : SUCCESS in 36m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cd5ec2406f124e35bc613ad9884a1d5f : SUCCESS in 1h 29m 01s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/700a7059e6b8449da5bf0cff59a9ba33 : RETRY_LIMIT in 19m 04s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/487b37d0b36e4c8886bf6f7e61244f99 : RETRY_LIMIT in 21m 20s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d4aea87d1590c7587db5941d1bddf85358461619","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-23 21:59:36.000000000","message":"Patch Set 12:\n\ncheck arm64","accounts_in_message":[],"_revision_number":12},{"id":"241248559030c52f0e51b6bd3ace5a8beebb4e5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 22:27:52.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/75f6df2540bd4aab877f5bc3577ae98b : FAILURE in 26m 44s","accounts_in_message":[],"_revision_number":12},{"id":"e0e6e4e4dfe411688988a51e01fde91d13755b48","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-24 09:54:10.000000000","message":"Patch Set 12:\n\ncheck arm64","accounts_in_message":[],"_revision_number":12},{"id":"1abc88be0c0a6b3f129bbdf19c63d74531db5f35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 10:49:29.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/0a6f0252286e4a6b926c7c9ac9994ab7 : FAILURE in 53m 46s","accounts_in_message":[],"_revision_number":12},{"id":"b7eb987b2dd9f6cc1bb30b5506b5808b652736ef","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-28 13:11:17.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"16bfd9d60b3216b054fcb97417cf3b7dd67ec833","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 14:13:19.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/6ccc2799613f40aa92e3b48a49b4886c : FAILURE in 59m 20s","accounts_in_message":[],"_revision_number":12},{"id":"dfdca556b9126d037c445f540196fbf15fb93361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 15:40:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c374b12e4a3f42d089a7685e2f529fba : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5da4777491140f6a58fbe3fdae2a1c5 : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48dcf8826db04fa1a63d0202275fb097 : SUCCESS in 1h 30m 49s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/d5ca15bc071341d5a99eee2baa352add : SUCCESS in 1h 12m 06s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d8658ccb02ec419899ea7ca58fa7d823 : SUCCESS in 1h 14m 41s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/8356523e45114307b518aee48c042014 : SUCCESS in 1h 01m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27e594c79b9b4240880061d2b782068c : SUCCESS in 2h 12m 14s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1926ed1456e94e8bb98b21349b584e9f : SUCCESS in 2h 03m 03s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/94b0278275384ce6858646c313378ee9 : SUCCESS in 1h 50m 14s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9d4f48a863bd4384ac7a4095adaddad2 : SUCCESS in 1h 25m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb644e062b144aab93ba1dcf00ff3ee3 : SUCCESS in 46m 34s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/e6efa3ceb1c24999b4c24d15b1429286 : SUCCESS in 38m 30s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/57501acecc644e6a8a8491155a3b8d02 : SUCCESS in 44m 02s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/7b711847b13a41f8b959adb66be7ce7d : SUCCESS in 45m 16s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/59cc9bc7b7ac45959c2e57f9747d1d9b : FAILURE in 1h 57m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/13c322c7cf2a44ceab03ccdfe9c0cc28 : FAILURE in 33m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c620a9c9db54e58b0804d9030079b0e : FAILURE in 2h 07m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e36e5772d8b34d8ba1f323c5f2d07bca : SUCCESS in 45m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b4e9d88b23794fcd8b5697f1ecad5a58 : SUCCESS in 38m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6a642b0376f84834bc17c1a5c1246197 : SUCCESS in 45m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/0df7826aa54446fab1ed0e5bd03a8720 : SUCCESS in 46m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b6ed2cfc5f14d4b9dff38dc7c44b399 : SUCCESS in 34m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/afc99b0d26da4e6da2a0d58f71942f0a : SUCCESS in 42m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/014fbcd388db4b1b9945712440e3eb0a : SUCCESS in 41m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/ebdc4bfe7ed546fea7fc991b2502660a : SUCCESS in 43m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/12440a1bbda64704b643ae577186baa3 : SUCCESS in 2h 06m 18s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/431bb0d82e9244dcb8d1f62e5b5123f4 : FAILURE in 1h 00m 47s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/117ef84d3d0d4f94bf50f6618cc29a90 : SUCCESS in 1h 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a4624fbe0ee239ac85ac05ea048415614dacc9bb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-02 16:36:03.000000000","message":"Patch Set 12:\n\ncheck arm64","accounts_in_message":[],"_revision_number":12},{"id":"8387395d3c3dc69b0b9e4c1f574e4884b885093b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 17:36:24.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/5265a5004f004e9fa16439f16ae1e164 : FAILURE in 58m 23s","accounts_in_message":[],"_revision_number":12},{"id":"c9875269b9494d39333c5424e9cb082cf305fced","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-04 09:42:33.000000000","message":"Patch Set 12:\n\ncheck arm64","accounts_in_message":[],"_revision_number":12},{"id":"5767f3cd78c9523d4db31f65b2b5c9a3f0ecb472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:41:32.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/6ed422fbedff4f14a1d80511422f1b57 : FAILURE in 1h 04m 26s","accounts_in_message":[],"_revision_number":12},{"id":"84e54ae25cb11060024933dd10ca1a2d2facc726","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 18:50:01.000000000","message":"Patch Set 12: Code-Review-1\n\nMarking -1 to ask a question.\nThis appears to be failing and not zuul -1.\nAre you proposing merging this as-is to work on this, or should we get this passing first?\nWhy is zuul +1 even though the job fails and it\u0027s not non-voting?","accounts_in_message":[],"_revision_number":12},{"id":"1990a1e47b9f6ea24abf61bc0e4ed78c94484c11","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-06 09:46:02.000000000","message":"Patch Set 12:\n\ncheck arm64\n\n\u003e Are you proposing merging this as-is to work on this, or should we get this passing first?\n\nI got the job passing CI twice long time ago. Local testing seemed to consistently pass deployment and smoke testing. The main issue I observed last time was ARM64 nodepool nodes were not stable and a regression in CentOS 8 (https://bugzilla.redhat.com/show_bug.cgi?id\u003d1862714). The later issue might have been fixed meanwhile.\n\nSo, I would like to get https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/747629 and this patch merged if possible. Other community members also seem to be interested in ARM64-based Octavia deployments -- see https://review.opendev.org/c/openstack/octavia/+/770937.\n\n\u003e Why is zuul +1 even though the job fails and it\u0027s not non-voting?\n\nARM64 CI jobs enter a dedicated \"check-arm64\" CI pipeline. All jobs in that pipeline are non-voting. See https://opendev.org/openstack/project-config/src/branch/master/zuul.d/pipelines.yaml#L365-L391","accounts_in_message":[],"_revision_number":12},{"id":"ee4a2a7605dd53ad15ae8f23a92543a5ce062a51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-06 09:53:18.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"581aef54c21ede6bdbbcf16bea4b7ce415483240","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 10:56:36.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/2885f14e438441d1ac4dea9ba30ef62d : FAILURE in 1h 01m 50s","accounts_in_message":[],"_revision_number":13},{"id":"b3a58581a109730dbfdf4e63f468c9d971fdfa04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 12:14:43.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c37326412b4337b747767e902f5c6d : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/075a616697234144a4600ebedf029660 : SUCCESS in 3m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e78dea3ab2b444c6b54bd0de57c1b0db : SUCCESS in 1h 13m 28s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/e1d4df8895874cc283a83be992863bde : SUCCESS in 56m 13s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e7bb46109d7a4d05a33300de277b04ed : SUCCESS in 1h 15m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/791acefee5394d79a069eedac981c25c : SUCCESS in 2h 17m 13s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/bf7fd31bcd214272aa4b377abaa466e0 : SUCCESS in 1h 32m 22s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8c1f0db5ee024dd298fe309b916d09d8 : SUCCESS in 1h 43m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/487c1a77ea2f4622a38d74c2683b70b4 : SUCCESS in 36m 55s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/33be172f35ea478d83904282b3ce8c20 : SUCCESS in 41m 22s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/76e6893a41e7467b91bce6177c5b040b : SUCCESS in 42m 40s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8fc1102831d04a1298c3734898ceb6d8 : SUCCESS in 1h 26m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/696a7711831e42649e8d8e3c5820e99d : FAILURE in 43m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/32b8e2e920414c56b88a47027f0661af : FAILURE in 1h 49m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/302caacf12294f5fa7af188a4b92befe : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8ec2ac57fa8d445baeeaef2f9aa11ad7 : SUCCESS in 43m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6722b892bc6742b38c3465d073c202e5 : SUCCESS in 37m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3fd5d7a24b4144b5b4718e977d247855 : SUCCESS in 48m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2f72bb36a6db413dba45ce0345f52485 : SUCCESS in 48m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/571cc8e946f9436d8d57360138926d97 : SUCCESS in 45m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6a43361191054a29be7dba844fc91a3e : SUCCESS in 2h 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0abb2756148e9117a91a72e3933c5068bbe70882","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-10 16:06:48.000000000","message":"Patch Set 13:\n\ncheck arm64","accounts_in_message":[],"_revision_number":13},{"id":"7ea12c765b705dc9cbb32e9481721e6039a918ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 17:04:30.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/ab5e7f37c9c64d5ea9504afb27ac0043 : FAILURE in 39m 35s","accounts_in_message":[],"_revision_number":13},{"id":"f4e8f14c786bb186a5381a62cda0393cff5ec887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-11 09:07:22.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"4f9f8abc1e0581294914213eb2b8b30fc7e67bbb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-11 09:07:38.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"6a3c6a763104ccffce2c1c99ed632ca93ff62fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:08:41.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2dec159ffa44edfb8e4ff0e1b55fbef : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b95c8bb746db455fbea4ed976474aca3 : SUCCESS in 5m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d825c2739a61453bae7d7dfb770dcbf1 : SUCCESS in 1h 46m 50s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/d85913cc7be14c2188bbcff6f15b3759 : SUCCESS in 1h 27m 02s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/2300b0c64e724871a25811a86d3ed568 : SUCCESS in 1h 14m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec8557fc4a774b408928c71ce716e68a : SUCCESS in 1h 52m 49s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/be06453dec3541fba8da7f92c3c2a523 : SUCCESS in 1h 24m 01s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2ddc0cc33cce4b229c53e76587fb63af : SUCCESS in 1h 29m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b6b357a54d534098894c18dde69606b2 : SUCCESS in 36m 50s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/79f54c65546149c7a5ce7e2fb9285e93 : SUCCESS in 44m 19s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a2c6e805643a44cb83487d42ba81a868 : SUCCESS in 43m 33s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/87acf8dae4dd41dcaa4d202a109b1e71 : SUCCESS in 1h 27m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/395ae5634c5b4f90acb3451cbe16a27b : FAILURE in 33m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/41ca26fdb8c9494db38ed48b6c13bbde : FAILURE in 43m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e22f5487edb44d6abbb79b9cdfdea2b : SUCCESS in 37m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/995d2482eed04a4eb7870d9bd8966401 : SUCCESS in 37m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3b9893ed7fae431f9a2f7aef9d111a01 : SUCCESS in 35m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/243e18b19f60493fa7f57019b3beb0e4 : SUCCESS in 36m 47s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/18ab88e555544c19ab22dc3268071982 : SUCCESS in 37m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/bcc4246796984b8b840605b565d23a69 : SUCCESS in 37m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2dc2cd2ad70644c1a54ea3c6bef684c2 : SUCCESS in 1h 55m 27s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"019aa08c298f3d8def96be406631b7c60b7d74ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:36:07.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\n\n- octavia-v2-dsvm-scenario-centos-8-aarch64 https://zuul.opendev.org/t/openstack/build/012e5ea33bb54ccd83bd1508f2069447 : FAILURE in 1h 06m 05s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"7ec4f20f44a9a9b31b1b52462217fa28cc39c93b","revisions":{"0d801f76958847ca1f6c0e4d2c738fdc04689931":{"kind":"REWORK","_number":1,"created":"2020-08-24 07:20:15.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/1"}}},"commit":{"parents":[{"commit":"7bd2f97221a369f379e15f33a3d3294f220c039e","subject":"Add an optional amphora admin log offloading test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7bd2f97221a369f379e15f33a3d3294f220c039e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-24 07:17:35.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-24 07:19:35.000000000","tz":120},"subject":"DNM: test_server.bin for ARM64","message":"DNM: test_server.bin for ARM64\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0d801f76958847ca1f6c0e4d2c738fdc04689931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0d801f76958847ca1f6c0e4d2c738fdc04689931"}]},"branch":"refs/heads/master"},"694ef41c34fd4baef473f60acc1077471842670b":{"kind":"REWORK","_number":2,"created":"2020-08-25 19:20:21.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/2"}}},"commit":{"parents":[{"commit":"bb23855ce65500dda6a0e332eab7174ed493a0ac","subject":"Add tests for allowed CIDRs in listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb23855ce65500dda6a0e332eab7174ed493a0ac"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 19:07:53.000000000","tz":120},"subject":"Add no-voting aarch64 scenario job","message":"Add no-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nThe test server ran by the backend members need to be compiled for\naarch64/amd64. Instead of distributing another test server binary in the\nGit repository, one is built via devstack plugin.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/694ef41c34fd4baef473f60acc1077471842670b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/694ef41c34fd4baef473f60acc1077471842670b"}]},"branch":"refs/heads/master"},"bb6b18dbaa7667e37efd1de42c751a24d92c7f38":{"kind":"REWORK","_number":3,"created":"2020-08-25 19:42:32.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/3"}}},"commit":{"parents":[{"commit":"bb23855ce65500dda6a0e332eab7174ed493a0ac","subject":"Add tests for allowed CIDRs in listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb23855ce65500dda6a0e332eab7174ed493a0ac"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 19:42:03.000000000","tz":120},"subject":"Add no-voting aarch64 scenario job","message":"Add no-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nThe test server ran by the backend members need to be compiled for\naarch64/amd64. Instead of distributing another test server binary in the\nGit repository, one is built via devstack plugin.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb6b18dbaa7667e37efd1de42c751a24d92c7f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb6b18dbaa7667e37efd1de42c751a24d92c7f38"}]},"branch":"refs/heads/master"},"a24a666c7a5efd9bbed84bddde8d88b0daf8830e":{"kind":"REWORK","_number":4,"created":"2020-08-25 21:04:20.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/4"}}},"commit":{"parents":[{"commit":"bb23855ce65500dda6a0e332eab7174ed493a0ac","subject":"Add tests for allowed CIDRs in listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb23855ce65500dda6a0e332eab7174ed493a0ac"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 21:03:59.000000000","tz":120},"subject":"Add no-voting aarch64 scenario job","message":"Add no-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nThe test server ran by the backend members need to be compiled for\naarch64/amd64. Instead of distributing another test server binary in the\nGit repository, one is built via devstack plugin.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a24a666c7a5efd9bbed84bddde8d88b0daf8830e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a24a666c7a5efd9bbed84bddde8d88b0daf8830e"}]},"branch":"refs/heads/master"},"4c60235c6ef5a2beb4c07f610c745f1032bd95d8":{"kind":"REWORK","_number":5,"created":"2020-09-11 17:00:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/29/747629/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/5"}}},"commit":{"parents":[{"commit":"89bdbcd1252dbce8645c24fa6db3e2d32fa779c3","subject":"Adjust scenario tests for NotImplemented skip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/89bdbcd1252dbce8645c24fa6db3e2d32fa779c3"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-11 17:00:10.000000000","tz":-420},"subject":"Add no-voting aarch64 scenario job","message":"Add no-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nThe test server ran by the backend members need to be compiled for\naarch64/amd64. Instead of distributing another test server binary in the\nGit repository, one is built via devstack plugin.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4c60235c6ef5a2beb4c07f610c745f1032bd95d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4c60235c6ef5a2beb4c07f610c745f1032bd95d8"}]},"branch":"refs/heads/master"},"24d680a0aad99e0c102b85b60a9ea02751288070":{"kind":"REWORK","_number":6,"created":"2020-09-21 09:08:11.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/6"}}},"commit":{"parents":[{"commit":"a80a185abef3234b035ec5d09e1748b743bc69e2","subject":"Merge \"Add duplicate uuid checking for tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a80a185abef3234b035ec5d09e1748b743bc69e2"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 09:07:58.000000000","tz":120},"subject":"Add not-voting aarch64 scenario job","message":"Add not-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nThe test server ran by the backend members need to be compiled for\naarch64/amd64. Instead of distributing another test server binary in the\nGit repository, one is built via devstack plugin.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/24d680a0aad99e0c102b85b60a9ea02751288070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/24d680a0aad99e0c102b85b60a9ea02751288070"}]},"branch":"refs/heads/master"},"56dea4234d6e49222a67be067a71f2243ef4b92a":{"kind":"REWORK","_number":7,"created":"2020-09-21 15:32:59.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/7"}}},"commit":{"parents":[{"commit":"a80a185abef3234b035ec5d09e1748b743bc69e2","subject":"Merge \"Add duplicate uuid checking for tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a80a185abef3234b035ec5d09e1748b743bc69e2"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 15:32:06.000000000","tz":120},"subject":"Add non-voting aarch64 scenario job","message":"Add non-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nThe test server ran by the backend members need to be compiled for\naarch64/amd64. Instead of distributing another test server binary in the\nGit repository, one is built via devstack plugin.\n\nDepends-On: https://review.opendev.org/#/c/738096\nDepends-On: https://review.opendev.org/#/c/753036\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/56dea4234d6e49222a67be067a71f2243ef4b92a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/56dea4234d6e49222a67be067a71f2243ef4b92a"}]},"branch":"refs/heads/master"},"052ad3f59bd116399426aa3bfab7a7f1f1dd8450":{"kind":"REWORK","_number":8,"created":"2020-09-24 08:04:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/8"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-24 08:03:38.000000000","tz":120},"subject":"Add non-voting aarch64 scenario job","message":"Add non-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nThe test server ran by the backend members need to be compiled for\naarch64/amd64. Instead of distributing another test server binary in the\nGit repository, one is built via devstack plugin.\n\nDepends-On: https://review.opendev.org/#/c/738096\nDepends-On: https://review.opendev.org/#/c/753036\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/052ad3f59bd116399426aa3bfab7a7f1f1dd8450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/052ad3f59bd116399426aa3bfab7a7f1f1dd8450"}]},"branch":"refs/heads/master"},"e2c60bf76182a8f2c89b4a2964108b4ae6523fc9":{"kind":"REWORK","_number":9,"created":"2020-10-19 10:06:05.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/9"}}},"commit":{"parents":[{"commit":"6128fc9fcfabdc6b72cf70111922d51a477832c6","subject":"Build the Go test server on devstack test-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6128fc9fcfabdc6b72cf70111922d51a477832c6"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-19 10:05:12.000000000","tz":120},"subject":"Add non-voting aarch64 scenario job","message":"Add non-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e2c60bf76182a8f2c89b4a2964108b4ae6523fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e2c60bf76182a8f2c89b4a2964108b4ae6523fc9"}]},"branch":"refs/heads/master"},"79206fd9078129f6489cd4b57eee2c076bb40887":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-10-19 11:12:33.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/10"}}},"commit":{"parents":[{"commit":"6426665b43f086f03e70ddc2ab68a11270db154d","subject":"Build the Go test server on devstack test-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6426665b43f086f03e70ddc2ab68a11270db154d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-19 11:11:51.000000000","tz":120},"subject":"Add non-voting aarch64 scenario job","message":"Add non-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/79206fd9078129f6489cd4b57eee2c076bb40887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/79206fd9078129f6489cd4b57eee2c076bb40887"}]},"branch":"refs/heads/master"},"35ff1efc84fa6e2fb515ab449c49b212f52c7f93":{"kind":"REWORK","_number":11,"created":"2020-10-19 12:28:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/11"}}},"commit":{"parents":[{"commit":"d840b6c2813ab148856f6cb3ee9dbab5df0ee520","subject":"Build the Go test server on devstack test-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d840b6c2813ab148856f6cb3ee9dbab5df0ee520"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-19 12:28:21.000000000","tz":120},"subject":"Add non-voting aarch64 scenario job","message":"Add non-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/35ff1efc84fa6e2fb515ab449c49b212f52c7f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/35ff1efc84fa6e2fb515ab449c49b212f52c7f93"}]},"branch":"refs/heads/master"},"0cb7ac41bd092ae8d58f138570a646e97242e7c4":{"kind":"REWORK","_number":12,"created":"2020-10-21 10:36:11.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/12"}}},"commit":{"parents":[{"commit":"f077c3a5d6f4d343dba8af7e0ba4c79c5500d2f6","subject":"Build the Go test server on devstack test-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f077c3a5d6f4d343dba8af7e0ba4c79c5500d2f6"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-21 10:35:49.000000000","tz":120},"subject":"Add non-voting aarch64 scenario job","message":"Add non-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0cb7ac41bd092ae8d58f138570a646e97242e7c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0cb7ac41bd092ae8d58f138570a646e97242e7c4"}]},"branch":"refs/heads/master"},"8f82d4a8fded851b7787686e8246b88c10ff2715":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-03-06 09:53:18.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/13"}}},"commit":{"parents":[{"commit":"3ff105bb5d84a05269f8e42918cd49ca4a80b23d","subject":"Merge \"Enable HTTPS on the test servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3ff105bb5d84a05269f8e42918cd49ca4a80b23d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Gonçalves","email":"cgoncalves@redhat.com","date":"2021-03-06 09:53:18.000000000","tz":0},"subject":"Add non-voting aarch64 scenario job","message":"Add non-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nDepends-On: https://review.opendev.org/#/c/738096\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8f82d4a8fded851b7787686e8246b88c10ff2715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8f82d4a8fded851b7787686e8246b88c10ff2715"}]},"branch":"refs/heads/master","description":"Rebase"},"7ec4f20f44a9a9b31b1b52462217fa28cc39c93b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":14,"created":"2021-03-11 09:07:22.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/29/747629/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/747629/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/747629/14"}}},"commit":{"parents":[{"commit":"59ec77d69d47411132569b2333726055351a750f","subject":"Merge \"Add neutron plugin to octavia-dsvm-base job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/59ec77d69d47411132569b2333726055351a750f"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 16:55:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-03-11 09:06:47.000000000","tz":60},"subject":"Add non-voting aarch64 scenario job","message":"Add non-voting aarch64 scenario job\n\nThe Linux kernel does not support nested virtualization for ARM yet, so\nlibvirt runs VMs on QEMU-TCG which make our tests run very slow.\n\nThis job runs a smoke scenario test on a CentOS 8 aarch64 devstack\nenvironment. Tests are limited to those marked as smoke because\notherwise the job would take a considerable amount of time to run.\n\nDepends-On: https://review.opendev.org/#/c/738096\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/779943\n\nChange-Id: I4dba477053cc71c54b9494c8f6dc6cd4b97687aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7ec4f20f44a9a9b31b1b52462217fa28cc39c93b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7ec4f20f44a9a9b31b1b52462217fa28cc39c93b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
