)]}'
{"id":"openstack%2Foctavia-tempest-plugin~748502","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I7615f03dc363e73a045b5a913020ad045eaa8b70","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"pass-subnetid","hashtags":[],"change_id":"I7615f03dc363e73a045b5a913020ad045eaa8b70","subject":"Add member API tests that do not pass subnet ID","status":"ABANDONED","created":"2020-08-27 18:15:46.000000000","updated":"2021-07-29 21:04:03.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7ea9ac1429722b5d235c59e5212cdb813a0a59ac","_number":748502,"virtual_id_number":748502,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-23 21:35:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 17:38:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-23 16:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69f64a7eac1d26e98f555867e7d80fc111a75fa3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-27 18:15:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3d5ac3d7cdaa95fe7f51f38e6c1695782fad06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 19:49:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f6da79c58648b8b373f6ac5caa9bd0 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2244a1992744afb385d5d7a4a0e96a : SUCCESS in 3m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d7cf9dd086c4471ca2dc56797337f99d : SUCCESS in 49m 00s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/bffe4c869ff647d99e414c5f29f4a9d1 : SUCCESS in 52m 01s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c2ee7ee661b043fe92d36adab507f6f5 : SUCCESS in 47m 45s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/9f2ebfd0a400439ea9a3f227d29ae171 : SUCCESS in 48m 41s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/640bf6f308e44c1baa1cb83a14709747 : SUCCESS in 1h 28m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/931f54672371480eb158b425e5ea4a4e : SUCCESS in 44m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1b437f1032f541f09e165b1b04f9bad9 : SUCCESS in 45m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bd05b45cf35a4d7488ad36554a1dea13 : SUCCESS in 43m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/86bdbd051f054c29b92d1799897088e3 : SUCCESS in 54m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40ef94f31c397b33e0efd891a50130dcb89e0779","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-27 19:51:33.000000000","message":"Patch Set 1:\n\nDoing this will allow the OVN provider driver to blacklist this test until it can be changed to try and retrieve the subnet ID via the DriverLibrary() class.","accounts_in_message":[],"_revision_number":1},{"id":"f749479f3a91593f08c13f6622216f4c426b55b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-03 17:38:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffe786bded1e740d83b2233b31a1c04cf58b6a83","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 18:04:03.000000000","message":"Patch Set 1:\n\nYes, there is still a bug in the OVN provider, I guess I think making all these tests run with and without a subnet ID should be done, I can re-arrange to not boot a second LB.","accounts_in_message":[],"_revision_number":1},{"id":"8e1f75fd4127fdd86e2b9c866093a59fbb61b171","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-04 18:45:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48d57cb3b90e20f0936df64c9286e4739209176b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-04 18:46:20.000000000","message":"Patch Set 2:\n\nChanged to not create a new test class.","accounts_in_message":[],"_revision_number":2},{"id":"8cf65dbb5cf6e7db82e8062ca56fb6b3ba09b2c4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-04 19:19:31.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2372cb684cd51bd26e847bd762d05739bbefb03d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 20:20:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf3757e3655a415399f6b086a9918d1a : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa8e600a8ea4449844bfdf3b6a4c966 : SUCCESS in 3m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0a47eff4927463497b5b1426fc7424b : FAILURE in 39m 35s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/e649736584e6493d94c75209439b8d6b : FAILURE in 51m 53s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/72d198ed6b71480880cf65804f5e4b84 : FAILURE in 41m 31s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/dbcd1518df3c468c89cd3083c86c96b0 : FAILURE in 41m 44s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/38d14be1ee5e499f948e0d434a6b0ea1 : SUCCESS in 58m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/375f77d72b6e493c83c4208589aa5276 : SUCCESS in 39m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/75e1f5b1c99541ef8a5879e393416dcc : SUCCESS in 55m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ec603d8aa3a54ee5b62744e86172bc62 : SUCCESS in 58m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/1ef99a9666da474dab68baaf1c3414bb : SUCCESS in 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"70946e17e07ee409f95873f9eb8214c7611bd73e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-06 16:21:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d8871fbe6e2f87ff4838a30c2a6fa4e6907e1b79","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-06 16:23:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"908eba140f914fc9725e96e48a29047bbbeecb30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 19:03:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e891435f574d87856eb456ac38b411 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6446f9e4d25441f2b3be5dda634ae2e4 : SUCCESS in 3m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b15732c08ae1484caa4e53f4f87f69d5 : SUCCESS in 1h 55m 06s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/78192c5ba10c405782188cbb569e13d8 : TIMED_OUT in 2h 13m 24s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/5e293f4b0d53450ab630ceea8a9a0eeb : SUCCESS in 1h 53m 32s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/954d1fb77cee479e852066251eff51ab : FAILURE in 25m 13s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7a537bf204184a108b809110d968d067 : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31c3396480394f8ca5af5ee2666965f5 : SUCCESS in 37m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9b12b7c59f0c4c2d90f84c5c3ee73d11 : SUCCESS in 38m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4a1ac6c340514c5f95282492b53369b2 : SUCCESS in 39m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/332d25593ebe42e898ba26634e13aef9 : FAILURE in 1h 23m 19s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/084057ed940f4474bffcf7857fda6402 : FAILURE in 33m 20s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b5aee9f4aea14f879f5ecb47735cf5fb : FAILURE in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e92f002dcf6881c430eca7f6c09817adf52a94f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 18:01:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"590b34f75f064bdbcc17210b7a2a8e5aeec8dcf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 21:07:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d71b3096d748b98647864fbe8bac58 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af66d329c73b46d285d34a1ec0943075 : SUCCESS in 3m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c46c1a3148343d3bca070c639733a5b : TIMED_OUT in 2h 13m 20s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8b96edfe0ae3482dac160adb80a71b83 : SUCCESS in 2h 01m 48s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/91c653c7c4ee449b84ff70edd588f78d : SUCCESS in 2h 02m 12s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/0b8698be8b7f4f0eb0bb47139a3d8708 : SUCCESS in 1h 27m 56s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4448222174aa42b2b55a4e3393a218e5 : SUCCESS in 2h 30m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ebec347abd084ca79f625a59b73b6c8c : SUCCESS in 47m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/41066f1e8db8450abaa8cc34cd5ed347 : SUCCESS in 45m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/091e00cf95a24bee8bc7592827efc3b9 : SUCCESS in 37m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/5648f000a6ff462caf6aafe41e7409e0 : SUCCESS in 42m 56s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/daf284909eab45f2a236882366ce4633 : TIMED_OUT in 1h 33m 30s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/db870bf0a6414e51aa93ae6ef8feca0b : SUCCESS in 43m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"37dd022cd272ebdbda76c25f4066cb55aec5e7d7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-23 14:25:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d7d3aee95722995fade9ab9ae1c8ee1387db7317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 16:36:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/369c2b3e80b2495ab59a4c4fe484a346 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90de807361be4e76a19133067ac0d4b4 : SUCCESS in 5m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed0a326c13c84429a9edc7d4f5705f15 : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/2eaca123a42340099841118d5e946b32 : SUCCESS in 1h 59m 14s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/8dcfeea4cec743718c2e11e7d71f663c : SUCCESS in 1h 49m 31s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/3080412ed15b46b3a24855bedacd2187 : SUCCESS in 1h 28m 51s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ce4351a19ba4b969d053a94f96249fd : FAILURE in 36m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a375cecb25e14c8e948d28784d6e83b9 : FAILURE in 36m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/41a4b1eb6cba4d60b28a8224cfc6512f : SUCCESS in 45m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/cbe8c35c092e41b29b49533660fe674d : SUCCESS in 38m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c8f9d45ba1b54363952ea45f744e0a87 : SUCCESS in 47m 02s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5a611f0c37a748c1afc7710ef676d7b1 : TIMED_OUT in 1h 35m 20s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b94eaa2d9d064eb2a7e05565c9ae01ae : TIMED_OUT in 1h 36m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8067b45c98579857c321b8e92fe3b0c8ba756e21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-23 19:19:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"e22780a415d21a71f6bd812690bc21c195defab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:35:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bd569c263234eb88b96f57cc899a5ed : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faccdfedf3c447cb8777c117ce27e2f7 : SUCCESS in 4m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74b1fda9fd774fd3946030548dd9ad4f : SUCCESS in 1h 55m 11s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/d4b8723fc08c4a4dbd35c38702c5d2d4 : SUCCESS in 1h 39m 22s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8b6e1a1e181d427f887bbf33c1b3113c : SUCCESS in 1h 37m 31s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/9f76ccbd6b814886b94841076c48b076 : SUCCESS in 1h 40m 27s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/75b1fbb4b773474aba42ca3ef280dad4 : TIMED_OUT in 2h 14m 32s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/415402de88b94124bcc5b145e82e4993 : FAILURE in 18m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f71bf66a0ff94938be5226c42c7bfaf5 : SUCCESS in 43m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/569912b9c98e41fa9e6787ec87005a8b : SUCCESS in 51m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b2ddf79d24ea40df8af7ad366534c72a : SUCCESS in 40m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bccbc0118fcb489899b9465721d0709e : SUCCESS in 38m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/8bf420a777f847d88215f418f8538646 : SUCCESS in 47m 22s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/8ffeb194ea7644b492f6b5bc4a4ce7df : FAILURE in 52m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"57996f6fa75b0d5b34b129a63d6726e1e13750ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-24 23:25:39.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d3b00b042b774be30bb346d6110d7f29729c0f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 01:46:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5434b6e5b38e4bfa908c9c6363984c2c : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea7c91adc4144a9984ef2fa3bf8dd5c5 : SUCCESS in 5m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0d30596b3b1461f91d7efb76c100b2a : SUCCESS in 2h 02m 41s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/87de41b2d3f947f88bdff16e9a3a4902 : SUCCESS in 1h 58m 24s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/e71c74ab98e34b1b8ab7544b10377486 : SUCCESS in 1h 49m 00s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/f68608e937d846379e28edb507397d4b : SUCCESS in 1h 17m 03s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/f4867e3b3b6e480f90f720c3ed6620f1 : TIMED_OUT in 2h 14m 58s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2d3e0cf7290437f9c5b9bbddfec1c8f : FAILURE in 18m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ae612d059b34d92b58399608e8cc5c6 : SUCCESS in 42m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/6c78fa3001504c73a943fa4b11b39c6a : SUCCESS in 42m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a92b00a55c914d91bf0d3e7fedd9bffb : SUCCESS in 43m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4f2b7b3f1bd74ac487333d7ebef5bd03 : SUCCESS in 37m 35s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/0800e5ffb58040548683281d6cde01a1 : SUCCESS in 56m 53s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/2a57c75f469a4f0ab6ce36bcee764d6d : SUCCESS in 58m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7ea9ac1429722b5d235c59e5212cdb813a0a59ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 21:04:03.000000000","message":"Abandoned\n\nNo time to work on this, reopen if needed.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"10883fca4c60bcf01d2d6464939204bc572a786d","revisions":{"e6d59770ab63126121d6f036f3f2e5c97a096493":{"kind":"REWORK","_number":1,"created":"2020-08-27 18:15:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/02/748502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/02/748502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/1"}}},"commit":{"parents":[{"commit":"e551af210421f46298268e648706dacbe362539f","subject":"Always pass subnet ID to member_create()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e551af210421f46298268e648706dacbe362539f"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-27 18:13:43.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-27 18:13:43.000000000","tz":-240},"subject":"Add member API tests that do not pass subnet ID","message":"Add member API tests that do not pass subnet ID\n\nSince the subnet ID is an optional argument for member\noperations, create a new sub-class MemberAPITestNoSubnetID\nthat omits it from all the member API calls.\n\nChange-Id: I7615f03dc363e73a045b5a913020ad045eaa8b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e6d59770ab63126121d6f036f3f2e5c97a096493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e6d59770ab63126121d6f036f3f2e5c97a096493"}]},"branch":"refs/heads/master"},"8af9c42edf0654fa8fe251e5731734285597de44":{"kind":"REWORK","_number":2,"created":"2020-09-04 18:45:49.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/02/748502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/02/748502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/2"}}},"commit":{"parents":[{"commit":"e551af210421f46298268e648706dacbe362539f","subject":"Always pass subnet ID to member_create()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e551af210421f46298268e648706dacbe362539f"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-27 18:13:43.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-04 18:45:26.000000000","tz":-240},"subject":"Add member API tests that do not pass subnet ID","message":"Add member API tests that do not pass subnet ID\n\nSince the subnet ID is an optional argument for member\noperations, create a new sub-class MemberAPITestNoSubnetID\nthat omits it from all the member API calls.\n\nChange-Id: I7615f03dc363e73a045b5a913020ad045eaa8b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8af9c42edf0654fa8fe251e5731734285597de44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8af9c42edf0654fa8fe251e5731734285597de44"}]},"branch":"refs/heads/master"},"acc3b96c60724d7cf9d4d08c2109a0bb2ce68117":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-04 19:19:31.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/02/748502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/02/748502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/3"}}},"commit":{"parents":[{"commit":"e551af210421f46298268e648706dacbe362539f","subject":"Always pass subnet ID to member_create()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e551af210421f46298268e648706dacbe362539f"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-27 18:13:43.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-04 19:19:02.000000000","tz":-240},"subject":"Add member API tests that do not pass subnet ID","message":"Add member API tests that do not pass subnet ID\n\nSince the subnet ID is an optional argument for member\noperations, create tests that omit it from all the member\nAPI calls.\n\nChange-Id: I7615f03dc363e73a045b5a913020ad045eaa8b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/acc3b96c60724d7cf9d4d08c2109a0bb2ce68117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/acc3b96c60724d7cf9d4d08c2109a0bb2ce68117"}]},"branch":"refs/heads/master"},"0bd481e6c0fed45dd6edf2a96bab7a89c4f4aa25":{"kind":"REWORK","_number":4,"created":"2020-10-06 16:21:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/02/748502/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/02/748502/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/4"}}},"commit":{"parents":[{"commit":"9d5a6eee1973f18a8e7c3b921ebb9a4e0e522fa6","subject":"Always pass subnet ID to member_create()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9d5a6eee1973f18a8e7c3b921ebb9a4e0e522fa6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-06 16:19:47.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-06 16:19:47.000000000","tz":-240},"subject":"Add member API tests that do not pass subnet ID","message":"Add member API tests that do not pass subnet ID\n\nSince the subnet ID is an optional argument for member\noperations, create a new sub-class MemberAPITestNoSubnetID\nthat omits it from all the member API calls.\n\nChange-Id: I7615f03dc363e73a045b5a913020ad045eaa8b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0bd481e6c0fed45dd6edf2a96bab7a89c4f4aa25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0bd481e6c0fed45dd6edf2a96bab7a89c4f4aa25"}]},"branch":"refs/heads/master"},"10883fca4c60bcf01d2d6464939204bc572a786d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-23 19:19:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/02/748502/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/02/748502/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/748502/5"}}},"commit":{"parents":[{"commit":"41ac74762cd09b89687c62666305fef898b803a4","subject":"Always pass subnet ID to member_create()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/41ac74762cd09b89687c62666305fef898b803a4"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-06 16:19:47.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-06-23 19:19:03.000000000","tz":0},"subject":"Add member API tests that do not pass subnet ID","message":"Add member API tests that do not pass subnet ID\n\nSince the subnet ID is an optional argument for member\noperations, create a new sub-class MemberAPITestNoSubnetID\nthat omits it from all the member API calls.\n\nChange-Id: I7615f03dc363e73a045b5a913020ad045eaa8b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/10883fca4c60bcf01d2d6464939204bc572a786d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/10883fca4c60bcf01d2d6464939204bc572a786d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
