)]}'
{"id":"openstack%2Foctavia-tempest-plugin~754386","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I2fd55d0328e567e03a43a1bc8581e47a76f4cd07","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I2fd55d0328e567e03a43a1bc8581e47a76f4cd07","subject":"Add traffic_requests configuration option","status":"NEW","created":"2020-09-25 13:15:45.000000000","updated":"2021-07-29 22:24:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"afab0a300eb35d4c07a8fce5871c691e0ec26115","_number":754386,"virtual_id_number":754386,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-25 15:54:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 15:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-28 18:18:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-29 22:24:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"52736563761f8125e5b07f19f89702830bc66add","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-25 13:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eea63de0c900f81573a66d5a130ca2f9b6770c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:54:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72bd60787826435d9081401434f1baad : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9aa01db32964e9fa1605d40b8ec1798 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/267b466db6ea410c9db168bb82dedb9a : SUCCESS in 1h 33m 50s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/fc15aa95602e40e9ad1839e59f497cb5 : SUCCESS in 1h 56m 02s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c0173c67ad9442ce81cf04c038b49f4a : SUCCESS in 1h 34m 07s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/a79300f33dc741f2b71435ba041e92a9 : SUCCESS in 1h 44m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12049de6b390453ab9e63f3a128b9a4b : SUCCESS in 2h 02m 40s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c5c41a77e22a470bae5de582a0804e89 : SUCCESS in 1h 51m 27s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/289f02e5071a46c19a08c8d472c25677 : SUCCESS in 2h 12m 24s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9b06939311fe4ef4b14a45d1028f2d4f : SUCCESS in 1h 33m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/083f0a1fd7da4aff8b31a693accea555 : SUCCESS in 38m 59s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/42fcef3151b34a96aa7b2eb399e1c213 : SUCCESS in 43m 20s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/240b14f2672e4faea6c97fb42a0bcdab : SUCCESS in 43m 41s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/993c8d7648394c1e880ac832ba011cad : SUCCESS in 44m 14s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f5900b5d2ae6495d960f3413b58685e0 : SUCCESS in 1h 41m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1b4d8cfb4c7b41dba23372a3d7c76fe3 : SUCCESS in 1h 58m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0b70aae407ca4ea480175dfcfa94e315 : SUCCESS in 2h 31m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/826f72b1270d46a6b1b9c4078d8daf93 : SUCCESS in 38m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/be5fa10ca01c4855a8f0caf7780997fe : FAILURE in 1h 32m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/35d9016efc6a41d0809ee2b66ca696ce : SUCCESS in 34m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/59a8bc312b6e4545a14df3042233fa35 : SUCCESS in 40m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c288ff2d771248eda8bd8483fcef59c5 : SUCCESS in 37m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/54980f33f6f442d6bda6c99cfb3f8114 : SUCCESS in 41m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/8ab3fb57ecf14764b6ea255bb8d2ac34 : SUCCESS in 40m 47s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/b2c32f0e88914bd189dbfed1fe895a10 : SUCCESS in 40m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71f6c594fc6f42afb90e230ccee592c8 : SUCCESS in 2h 15m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/784890de75d344289d64699edcd282ce : FAILURE in 1h 19m 14s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b4a1d71a8e8748e0977f6f925c0efa76 : RETRY_LIMIT in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afab0a300eb35d4c07a8fce5871c691e0ec26115","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:24:11.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9984e55a147a08ca5b796cd8d7885c9feb2fddf0","revisions":{"9984e55a147a08ca5b796cd8d7885c9feb2fddf0":{"kind":"REWORK","_number":1,"created":"2020-09-25 13:15:45.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/86/754386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/86/754386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/754386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/754386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/754386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/754386/1"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-25 13:04:06.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-25 13:13:56.000000000","tz":120},"subject":"Add traffic_requests configuration option","message":"Add traffic_requests configuration option\n\nThe test coverage matrix has recently exploded from a few dozens of\ntests to hundreds. Many of the tests send up to 20 traffic requests to\nmembers and load balancers to check consistency of responses. The number\nof requests is not tunable (harcoded), and now with hundreds of tests\ntakes CI jobs a very long time to run the full test suite.\n\nThis patch adds a new configuration option to allow tunning the number\nof traffic requests to members and load balancer VIPs. The default\nnumber is 4.\n\nChange-Id: I2fd55d0328e567e03a43a1bc8581e47a76f4cd07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9984e55a147a08ca5b796cd8d7885c9feb2fddf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9984e55a147a08ca5b796cd8d7885c9feb2fddf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
