)]}'
{"id":"openstack%2Foctavia-tempest-plugin~755777","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7","subject":"Fix scenario tests issue with immutable LBs","status":"MERGED","created":"2020-10-02 12:09:21.000000000","updated":"2020-10-30 04:18:16.000000000","submitted":"2020-10-30 04:12:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755777-1604031130194-692fdd93","meta_rev_id":"b5801ff2c59a543b3a1852222f7bd52592f4b34a","_number":755777,"virtual_id_number":755777,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-10-29 07:37:36.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 04:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},{"value":2,"date":"2020-10-21 12:23:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-10-19 19:33:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-10-27 20:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-10-27 20:47:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-10 01:55:42.000000000","updated_by":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"reviewer":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"state":"REVIEWER"},{"updated":"2020-10-19 19:33:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-21 12:23:09.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-27 20:47:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-29 07:37:36.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-10-30 04:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4cb47bc499ed026d21a4fe02709d84f2e9f4ec0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-02 12:09:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6212754e323d1afbc712e43d3ffc85255819618","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-02 12:17:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c584bcd9594ea48fb54a42eb8a8319debf806948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 15:16:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebd39434bdfe4b1d8bfae760b22df8ec : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7414d8e729949678ad2c061690f26d5 : SUCCESS in 3m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/430e3cb389344b69bc77478af8b7d82f : TIMED_OUT in 2h 35m 20s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/928523a97b3c4329a1bb13bad13f022b : SUCCESS in 1h 42m 20s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/cf1948c380a844d481217ed6d3513cb9 : FAILURE in 1h 58m 01s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c6d76c9f2b5542a69d4be5f211c448f5 : SUCCESS in 1h 34m 54s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d031c1f0c4e044a58db89cc4de73e378 : SUCCESS in 2h 02m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/01236baf4bd640a2848362068715a074 : SUCCESS in 2h 30m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb82711d8405475ebf42a1e78d1fe5c2 : SUCCESS in 2h 31m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79fa9f03c59d414d8bbe7d7ebf9eb3c6 : SUCCESS in 40m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1885686d37d542e5aead9dcdb541ce95 : POST_FAILURE in 1h 04m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/347f6be8b6fd4fe09e28febb1a93652a : SUCCESS in 36m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/dc40fefcbfd4490fa27635ae9d4fddde : SUCCESS in 43m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/87f92d3d1322429ca4b439a5ef4cb285 : SUCCESS in 35m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/d4f4605d2d3d45e5a8c32bbbc4b2e1df : SUCCESS in 41m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/fe638f89acdb41d694043d53854a543b : SUCCESS in 34m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/1c581754f1ca4af9b149ab884eb18aa9 : SUCCESS in 1h 15m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/db72824512c14418a89e16293d6b8110 : SUCCESS in 2h 20m 03s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/1c2aa5d0287c4c439815194d3720114f : FAILURE in 1h 01m 17s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f0751495b6c54f14b584a95d2562e93a : SUCCESS in 36m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9237b8e5071f7ef60d9aba673bde8ceae5e94af6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-02 15:41:29.000000000","message":"Patch Set 1:\n\nrecheck\n\nReadTimeoutError(\"HTTPConnectionPool(host\u003d\u0027213.32.77.242\u0027, port\u003d80): Read timed out. (read timeout\u003d60)\",) while deleting a pool","accounts_in_message":[],"_revision_number":1},{"id":"bb33462dcdc2e58dcb93d4b3c7b1f6b5ca951f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 18:29:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cfc2219223041b7a47b0e0cb8a075c2 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec6b8bdce9a4d00820b7a7c85542969 : SUCCESS in 3m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8077b1b460d64d9abe16560f74e3c9cb : FAILURE in 1h 45m 11s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b0441cf0b6bd4c1cac0beb1a9045784c : TIMED_OUT in 2h 34m 24s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a6ac19bd087645168a95abf301333a84 : FAILURE in 1h 55m 09s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/52178913b85d4174be0c43b24112609b : SUCCESS in 1h 28m 50s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e8f72ab932da467b8b43df5a9fa2f178 : SUCCESS in 2h 09m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4e280bb169f34e808936f187de5b8cb3 : SUCCESS in 2h 19m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7be8917e84ff4a64a070f28d56d4816d : TIMED_OUT in 2h 33m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05d965575c704dda8459c3282e5cf2cc : SUCCESS in 39m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/eba50a9f7c50438b87f242714b184b34 : SUCCESS in 42m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4f586747aa5d407eab4999a250d208e7 : SUCCESS in 47m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a097b83ddff34b7282fa391a6c694b47 : SUCCESS in 45m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bb1b0f34cf28403db8a834d57515d78d : SUCCESS in 44m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/9c1cbb4b3b3748b79dafaf601bf68d7c : SUCCESS in 47m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/6bcea0ac3a6942d88fb561306f4cbb71 : SUCCESS in 40m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f0b19eb648af438d87d7c6d54f08a56e : SUCCESS in 42m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5442016164354b09baa7ff5bee87261a : SUCCESS in 2h 00m 42s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/1204364f94d64192854e8fabb25f405b : FAILURE in 58m 22s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/54208fe6e4a946f6ab25631e1ece01e7 : SUCCESS in 40m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0c0b6bb679c423c3b0f6696514944e135969d89","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-05 07:47:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89c5e6f3f8cabca114212d8ebc4ebf7e07d2e097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 10:23:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4272f5241fe4802b2e503a36f700be7 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7297424815df461c8da35cc2b66b5068 : SUCCESS in 2m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66797861ffa447368d4d4b62a8ad364f : TIMED_OUT in 2h 33m 50s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b0b127475b904a91ad5e8ebec1876a42 : SUCCESS in 2h 23m 12s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/167628c33a574a94b4bd1773800e9f67 : SUCCESS in 2h 08m 12s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c53a689d0cb64fe9abc74997344c9eb0 : SUCCESS in 1h 34m 10s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9b930dcfb4f042d78d0011ac10efba17 : SUCCESS in 2h 04m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ada5b4c5d84343978ce0d5a3f784020e : SUCCESS in 2h 10m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad3fd76a15b548d59bf6bd9b42713025 : SUCCESS in 2h 18m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/780e8c39d82c495eb369e4ba85766e8f : SUCCESS in 42m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/aded6dc1b40d401c8ce68247d9072b47 : SUCCESS in 42m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/faadd3f0c5f7453ba6be1782b647c875 : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/008ce7fb3ee9476687fdd90c2d8becce : SUCCESS in 42m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/285f5f3103e448e2a9cecc9af7463297 : SUCCESS in 36m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/6e121e33f430494b915e9c9a8659ac7f : SUCCESS in 37m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/c56a874c96294dec9663203796ece3ef : SUCCESS in 39m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/149353f0a86848b6bb939615d8ddbcda : SUCCESS in 40m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1fcd0f3fba4a4ff986691117a507d664 : SUCCESS in 2h 12m 53s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/1eae283eb2354b0bb22ab6578059446c : FAILURE in 1h 02m 08s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/1bd09d20850848ed9146350141f5ce2c : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e16dfeb4256d2fdc6e1b4ea5f59b49a468f70887","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-05 23:30:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0dcbe5b806d6aca750f52190080780a299c53266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 02:15:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b3efe647f1344de8c7b79207a2c3d71 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa7ec8f2b994a9482738a2d736aaa4a : SUCCESS in 3m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/564906f3887c45088a3638dfbdc29f20 : SUCCESS in 1h 55m 10s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/c78036353fc34c65a62782e6b00ce352 : SUCCESS in 1h 38m 01s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/0de5a3f127a44800b7d8e2873f644b67 : SUCCESS in 1h 38m 49s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/bff0c007b5114a6a97ec775f9595d6c8 : SUCCESS in 1h 21m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2459a9ec8a194118bb50c65596cb1785 : SUCCESS in 2h 21m 23s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/805ce847f7994982a65ae15400313b64 : SUCCESS in 2h 14m 11s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4d0359f9a71c493f8e173e674f60fc91 : TIMED_OUT in 2h 34m 02s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9467289c0a1b4176aa0907f0e751e84b : SUCCESS in 1h 41m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f652bdb9b994701b063ecb8fac95545 : SUCCESS in 43m 42s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/1b6a2d20c5ca4418a37c88207d2aa3df : SUCCESS in 40m 56s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/94953417aedd4eb199dd68b932e4a245 : SUCCESS in 1h 09m 21s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/9bb2f420cc694cc99120c4765bd9ac90 : FAILURE in 1h 07m 36s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d35729ab06f4743b70ca7e81b5ccabb : SUCCESS in 2h 12m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/383930a4beb54c60ab3f4dd5fba778d1 : SUCCESS in 2h 01m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/459d6a89a1054d1f83fdc0d4c47c98fc : TIMED_OUT in 2h 31m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19110cad917c4b1dac7c5bc3d4a8cfa3 : SUCCESS in 39m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b0f19b424dd54cb193bf2cd4504f5e27 : SUCCESS in 39m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2d01c41153174ca3a8955112f8110615 : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/022c5ae1bef2496393e0694da1221341 : SUCCESS in 50m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5d2d2c81fdbb4048bc5c6d40c58d472b : SUCCESS in 41m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/1341404c9ead4909bad78f68725ae01e : SUCCESS in 40m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/8eec0568df19461595d34339815c171e : SUCCESS in 40m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/fc22cd73e5b346c884a6265238f34ea2 : SUCCESS in 43m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e5395d71cc047c286781c450278224f : SUCCESS in 2h 09m 13s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/17c654adc1a044d1bf4c26cc6e8529b9 : FAILURE in 26m 21s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/79d9a051fd934f7ca8e93bd128b47a56 : FAILURE in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7c9c52b3ea01332bcfc4ae7b8bbc3c339841bd42","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-06 06:43:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"acf39a1144d6325bc535a3541634889bf4e3d12f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 09:59:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e23c39370474aa596c91659a5b888eb : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c685c9acb00d453bbbbdf495c883047f : SUCCESS in 4m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/88068d77189d49e08a4d509fc7d1263a : SUCCESS in 1h 32m 17s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/a74abd33fa404992b0d8b6b9529cd388 : SUCCESS in 1h 40m 34s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d28e4d369062430eb80b1209cd83bcd8 : SUCCESS in 1h 33m 43s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/842671f0e6c748aca624fabe131e4b61 : FAILURE in 23m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95fb7a037de746c4b3794c981058cbc1 : SUCCESS in 2h 03m 04s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/fc167d66350243d0b252c06bba3024ac : TIMED_OUT in 2h 34m 49s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3cdaf2d6ac194d609b04535ccd3b6735 : TIMED_OUT in 2h 34m 28s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/f077c61b21e042ac873fd5189a41880e : SUCCESS in 1h 32m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3972e96d01354193b077d8ef7789c1be : SUCCESS in 41m 21s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/49fd2ab060d14a7d9477b18a886167e7 : SUCCESS in 43m 37s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/947183cba634483c8bf5fe21f372d152 : SUCCESS in 41m 26s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/f5db2bb37c6a477e863a9e870c072aa1 : SUCCESS in 42m 09s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dfa4887e43ca46a88236ffa1a55e3576 : SUCCESS in 2h 08m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4c94bf466a1a4faf995eda7e75ddf490 : FAILURE in 1h 58m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d92bc4f7c9741cebce00a7a347f5c12 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3205afc5fd73436aa4adbeac8ffbf842 : SUCCESS in 34m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cce4ff875e2640e18398349c7d022fc2 : SUCCESS in 41m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/547220ae6d9b47d38a3f78dbd53bcc31 : SUCCESS in 39m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a0d41c07bf6a4f2b9137f10e34dfb7eb : SUCCESS in 41m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f55031ef8a344626881cb3d6339ade8c : SUCCESS in 40m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/8d2f9de4abbb42e0bd168126fa964c0d : SUCCESS in 33m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/42a498353aac481b89e63f50298e086e : SUCCESS in 40m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/40c84ef7b9644854ad2c0c74f194ab5e : SUCCESS in 35m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e75cab2c79d4c7388ee918ce52d5ab0 : TIMED_OUT in 2h 34m 34s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/27bb7b008bb443a5a2393c394fd424da : FAILURE in 31m 33s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/627b5cbf63fb4c21a8e75102dd36ef90 : FAILURE in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e97e2e234cab18d79affbc6c8f9d7cc1616dec8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-06 19:56:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d3c5a5e6b09924dec48adb194488bdae5fe06c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 22:35:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1d765a5e094f0dae299435a9294a33 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592dbf74f84f4a979e37c2e5bde2fb68 : SUCCESS in 2m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8be79f0083da44ea9e24be80098c85e7 : SUCCESS in 1h 37m 22s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/cdac37cf54e14b93abbe47d1923a2e14 : SUCCESS in 1h 36m 30s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/f8614cd7c8044930b3274a6ed12c6296 : SUCCESS in 1h 33m 00s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/c67e23b52a9a428a8ceccc0b9195bdb9 : SUCCESS in 1h 23m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/badaf78394cf46d0a8caa19bed7ceefd : SUCCESS in 1h 59m 37s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d593de70397a40cab2426057fb564bfc : SUCCESS in 2h 15m 41s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0c78fdabb5b14257a98834bd2fbd3819 : FAILURE in 2h 24m 02s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/515ad4e9fe7b4f69beb3f26b1db26685 : SUCCESS in 1h 33m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c67077376d6a40d6934f5684102db4ac : SUCCESS in 45m 06s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/90411bbe33474976ada30da828f8816b : SUCCESS in 1h 09m 01s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/d69d8c7aa11f4ee89436a5f1a500f821 : SUCCESS in 35m 06s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/ab29a497aff64eb3a5585d75ef2bedc7 : SUCCESS in 43m 03s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6c1af08edba74e82b589870963863839 : SUCCESS in 1h 48m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/495e5e00fe694c4cbe934723e911e851 : SUCCESS in 2h 01m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1b51ab49ae94003a946358ac05d8604 : TIMED_OUT in 2h 32m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee06f667c0f64c2abd2478ea90aa8218 : SUCCESS in 42m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3af840ce81b44e40ad3e51ae7f12b483 : SUCCESS in 42m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d07098f41d19446ab6dde6138d8e1ad9 : SUCCESS in 42m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/83633748cfd24268a900d99511590645 : SUCCESS in 42m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/703a23b87e034fe1813a305e58e9b7ef : SUCCESS in 42m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/04219d63d9954d8882da98f3ace16d45 : SUCCESS in 41m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/806168d4e8324d57bad356ed94f25183 : SUCCESS in 43m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/128071432f9b4b6faadec4d010296066 : SUCCESS in 43m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1880bc00ea724fdf84a703fb9751ef87 : SUCCESS in 2h 22m 48s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/900894a3fab043eeb82bf3c7531c2082 : FAILURE in 33m 47s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/64f4d3fe21354c04b5f05370f91099a1 : FAILURE in 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7c7cb63d87a06f5c5a9444cff243c2795542213","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-06 23:44:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c351e9241787ead26d8d4eda54945ddfab1af155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 02:22:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f438a68746b044b08f396ba212d5cae3 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73e7ef103f8740e2b317bd655975fcac : SUCCESS in 3m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb526d2202c9407d9cd94cea4692e77c : SUCCESS in 1h 38m 09s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8d4782df1286466db62a53ff5de76a73 : SUCCESS in 1h 40m 21s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/478432d954354a029ef49d8296708059 : SUCCESS in 1h 38m 24s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/221160650a7b4c86969f4014786968a2 : SUCCESS in 1h 13m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c189939dbe14b179c60d408576937bb : SUCCESS in 2h 00m 49s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d7f0a7f41bb4476e881b3c27147f7809 : TIMED_OUT in 2h 33m 12s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d984bab2d7974b0e9a2056e0db61fc56 : SUCCESS in 1h 40m 45s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/0702e04a649040e7a8b289e5d0da9f04 : SUCCESS in 1h 15m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/059893bc1c9749c4b54592b695c59b28 : SUCCESS in 38m 33s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/289ce50d2e2d443286c1461829ab07a6 : SUCCESS in 44m 05s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a87b080b1cce43f2a7cfd461719e6fef : SUCCESS in 45m 24s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/356b82061222437cb4442e579296b037 : SUCCESS in 44m 34s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/56a1fe38f9db4a32bdfb16fa62c4de0e : SUCCESS in 2h 07m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6881aefa49a94275a27318abe1d21ecd : SUCCESS in 2h 08m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f7beec8d55ef460e817ee85b9b13110e : TIMED_OUT in 2h 32m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58ddf125242d4984ba3ea01b882c95fc : SUCCESS in 32m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/96bca3f154284d959163cd0f03e97d87 : SUCCESS in 55m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2f1c5f9af49a43f19fc32171e36702dd : SUCCESS in 43m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b7c6a8932e644ca6a49fd1eeca82bc85 : SUCCESS in 36m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8564e987283a4128878b2559ce36b714 : SUCCESS in 42m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/4b36e4fcdd804aa68b2770282dbc26b7 : SUCCESS in 41m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/bf57b19776e34315bd14afc02af1b0a3 : SUCCESS in 37m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/08ef6525b7a44d78aaea5b653c2e1131 : SUCCESS in 43m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7f3e849cf378425f92f688f0b9defa9a : SUCCESS in 2h 05m 52s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/cd1f0cb08aea4bf5be62088cc4fc3d85 : FAILURE in 34m 56s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/8207273926a1479ca8bb5be173315ea2 : FAILURE in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e76f6b79f578e87933a29778e1bc40bdbbe1415","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-07 06:43:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"71716a9abb5c9c1aec39476863da6297e6817f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 09:38:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d647e07b1604998972c9838dba12281 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce55d6b925e348f0bb2290ecc56e9b73 : SUCCESS in 3m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/608bbd3661d24f45b1fe7b41eb444b8b : SUCCESS in 1h 47m 41s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/c2ec40fed36c4d79a225cf04d9a9150a : SUCCESS in 1h 56m 50s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/28edbc0caae84447a68d3c69cf60fdb3 : SUCCESS in 1h 36m 45s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f91f87b99fae4d028bef9da9003b4f05 : SUCCESS in 1h 25m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2affe9d90ace4096a7e8a0d9609d7688 : FAILURE in 2h 03m 49s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/790e9949f1444a6d839254a4b13be195 : SUCCESS in 2h 06m 40s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6a12f892f27e41ce930d592e943eece2 : SUCCESS in 1h 56m 22s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8f575ebca2a949fe9bb699f2a5977dc1 : SUCCESS in 1h 25m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b39420ad89a451f836a9c178f28cac7 : SUCCESS in 39m 35s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/29546e1cbfe540e8a1ae521d285738cb : SUCCESS in 44m 57s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/74efeb0d0700443490ac59486e1a5221 : SUCCESS in 42m 30s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/96e1041042ba4d8282b5e4bd94f2c8dc : SUCCESS in 41m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/39afc83abd0d4c08b117c21802a352ef : SUCCESS in 2h 19m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8b7553ebde554900b8843d93225f6305 : TIMED_OUT in 2h 34m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/75597733aa5348cc98ff1e140d525526 : TIMED_OUT in 2h 32m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ba7d6810fd24cc598d2b4fcdb5d1ac7 : SUCCESS in 39m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/156348f13d2b4ac38fae8ddc9e52880d : SUCCESS in 39m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a6944cb0ecaf4267a566b86615e5b818 : FAILURE in 29m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4493b2a7c27445d1a325db48b439bee1 : SUCCESS in 48m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b048803f477c4b4cac444f5bad0294a5 : SUCCESS in 46m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/54818b8929aa4cb1b3ca7361aa6402f3 : SUCCESS in 41m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/19bc93d487f842f4bd1a41447ccbb929 : SUCCESS in 42m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/e4ba823c935746fc8cca0c918102e803 : SUCCESS in 40m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b85196e8b61e4cd198dca190caff1a1f : SUCCESS in 2h 23m 53s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/3c3b08a7e5bf48e380970d2e8ee52614 : FAILURE in 35m 06s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/940563d0b2c548ffaa29a4334953a8f2 : FAILURE in 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc703e50ba8c3bfe54fad99e93876a0b62db6646","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-07 09:45:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"10e591275c26acd375b9ce1f394fee23c32d5f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 12:32:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d703c483efe4bf6973a285dbd7809ee : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6123045c9474ab9b16fd6aab7f5e29b : SUCCESS in 2m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4944001ce21f44d997dc58ed986bee74 : SUCCESS in 1h 36m 47s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/66e9c46ee31949c4a6d12ec27bb7c710 : SUCCESS in 1h 44m 59s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/1fc0e25053f649a7a226cdf60157b1ea : SUCCESS in 1h 45m 35s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6dd081f4492c493a8e1a494c2099e617 : SUCCESS in 1h 16m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c506c573139f433bb1422710724819ab : SUCCESS in 1h 57m 28s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/4bc4bfd9c9074782a2cec2b7a638200d : SUCCESS in 2h 01m 25s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bec0fa81964a4bfd83ea48a0c7201e6e : SUCCESS in 2h 19m 57s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8fbad747de194bafb19fd0a1f91374db : TIMED_OUT in 2h 32m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ba8a447eec5452f83fd2fe1c3759dd3 : SUCCESS in 44m 28s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/3b1efa28d54b43f393877f8032aace1f : SUCCESS in 40m 29s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/66564b8680b04b3093b7b827d868f467 : SUCCESS in 43m 03s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/cf7213a4a7e74342b5fff75ffc136bc1 : SUCCESS in 45m 19s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/08fe2c35721c4964b990b9633c7f5924 : SUCCESS in 1h 53m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/932dee1cd30744bd867da0a51cc19090 : FAILURE in 2h 32m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/22bda218602043038eda4c9685009be8 : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bda7bfb2b7a04c5ca3dd08d8e667d6cf : SUCCESS in 43m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9f75cc9dea9a40cc939ae8b4819ba06a : SUCCESS in 1h 32m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e19eccc804634e6fb8b8a7d85407312a : SUCCESS in 42m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d30f4f765e844f6c976468ba62b526dd : SUCCESS in 45m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/77eec6ee3533413c8918a2415a60c6b2 : SUCCESS in 38m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/d6ed17795d2f41879a87b586a56f2bba : SUCCESS in 41m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/9358aeabbe5448e68a2f19ee1606f0da : SUCCESS in 45m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/2cfb2aff994c43ddbba0769bfa7cc03b : SUCCESS in 43m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/48e2592c407b479e943cfde4843a377e : SUCCESS in 2h 13m 07s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/46bc89a49b274552881eac1c347c33a9 : FAILURE in 34m 22s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/fa9756cf06674f46bea29355f660d795 : FAILURE in 30m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e089a64541b12434a79bbdc3b8cd07448f05d5c0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-07 15:18:14.000000000","message":"Patch Set 3:\n\nrecheck\n\nmember create timed out. no amphora logs to check (amphora log offloading added in train, failed job is stein).","accounts_in_message":[],"_revision_number":3},{"id":"b6f6d6a955f9b1a09ea1fb1e1b0f8d793c3e5f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:25:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e31290bcd2a4c9183cb202b6b5fc5e0 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29c54fc9644472b8d047e46c5c871b3 : SUCCESS in 4m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3679e105c117467f9071814f848b87d0 : FAILURE in 1h 35m 12s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/fa0ee36b361d40e79d721d5027ad7d6f : SUCCESS in 1h 26m 20s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/024eb038fcfe4c0c859d989437ca4470 : SUCCESS in 1h 33m 52s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/77a40dc6ca5a4681841db5dbda5aa625 : SUCCESS in 59m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3688b23459614f85a6f8b5d16455bf8a : SUCCESS in 2h 11m 33s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/4d7fbce4a71240b7b1c1a6252870cec2 : SUCCESS in 2h 15m 47s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/69eeecdc63a94e8c845b906cb47d5e09 : TIMED_OUT in 2h 34m 50s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7a23375328ea4eda90adca1c623f5135 : SUCCESS in 1h 25m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2d1935598bb4a74a2443072d78f58f9 : SUCCESS in 41m 58s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b92e00865c7545538b8ee5e3ebd5a481 : SUCCESS in 43m 36s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/9ae66ac9c7f64e32907ddac4783a2c28 : SUCCESS in 42m 24s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4b0bdd1b14224f60853bc549c0e2560c : SUCCESS in 49m 08s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f7c3a80dcb2c463ca5620ec8cfcf88cc : TIMED_OUT in 2h 34m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1e8be8696f2748e3b1b52e87ea6dfb0c : SUCCESS in 2h 15m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/98689d3595c0474d80bdcddad1ed2600 : SUCCESS in 2h 00m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b07fad70db7458da1abf81360865853 : SUCCESS in 41m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6f8b0f7b9e8d4b28aab628a8dfefeed0 : SUCCESS in 43m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8a12e879b2bd49308c83f07e4f4ac688 : SUCCESS in 40m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9c077f2f268b431bbf713f3947940bfe : SUCCESS in 49m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/19b6d2c4858d40f293dab6d3106adf34 : SUCCESS in 37m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2b8126e0c276466a9e3a571c2cf531f9 : SUCCESS in 38m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/529b825c9b2c4a6bb94bcca8dca27146 : SUCCESS in 43m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/9de16b62b6824d5aa062f8963367306e : SUCCESS in 43m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c0eeb1e9a27416bb16af5bd64753b67 : SUCCESS in 2h 05m 35s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a41fa5a5e8ee45c98dc7c82c4535b3aa : FAILURE in 34m 09s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5c3dc995381345a991fb550e558151d2 : FAILURE in 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2cbac74438624d67b563cd07e8220559cc5cdba2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-08 08:22:10.000000000","message":"Patch Set 3:\n\nrecheck\n\n2020-10-07 16:02:52.369 84269 WARNING urllib3.connectionpool [-] Retrying (Retry(total\u003d9, connect\u003dNone, read\u003dNone, redirect\u003d5, status\u003dNone)) after connection broken by \u0027ReadTimeoutError(\"HTTP\nConnectionPool(host\u003d\u0027158.69.65.104\u0027, port\u003d80): Read timed out. (read timeout\u003d60)\")\u0027: /load-balancer/v2.0/lbaas/pools/30ade6ab-f043-479e-9746-d2c795cc9e1c","accounts_in_message":[],"_revision_number":3},{"id":"e1c4b54bc85efbb1b305afa31d69517be9ed7fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 10:59:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454c47eebd554613b06a52fb096975bc : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ad7cc28985e444b86ff663a1a1aa201 : SUCCESS in 3m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6d07170cf06f4fde8ccbdda73a77761f : SUCCESS in 1h 12m 21s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/1f06b3dfec194a799e1c132fedf88c58 : SUCCESS in 1h 47m 51s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/3e8ff3077f3b4a7ba9d0bfff7cff1439 : SUCCESS in 1h 38m 18s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/9cef246f2c384729b85073aa4cc1bf03 : SUCCESS in 1h 22m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4cc649c403d4e70a7e7b7e5266aa205 : SUCCESS in 2h 05m 43s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ef2328ab3f1c4c0e85c573a9a283f5b6 : TIMED_OUT in 2h 33m 58s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/71bb61c2fb954c50bab5066d182b7df3 : FAILURE in 2h 06m 04s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/622ceb9cf5a341f182644612775dd5e1 : SUCCESS in 1h 39m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca146becbb49454f9eef22be3e545159 : SUCCESS in 44m 45s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/22ef70eae1d642aebc4c9946e5cca2cc : SUCCESS in 45m 25s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/8838552c44184d4e8864c9fcd7949a26 : SUCCESS in 44m 05s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/b5814c383b384d848a521b955f04e323 : SUCCESS in 46m 48s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/63b88bc76d314aacb1c1c50e792db4de : SUCCESS in 2h 10m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/29dbf04626f64ea39bbb6847f315b8dc : SUCCESS in 2h 10m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ec97adacb6e433b90bb341e7911ee25 : SUCCESS in 2h 24m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55a47c98099b4be4ab56aa8b7d2de31f : SUCCESS in 40m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3e7b9f1546634c58a925ade6bd41b4d3 : SUCCESS in 42m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9484eb3a544147d681662544f98e08b4 : SUCCESS in 1h 09m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/eb6cbb5f7ecf46c3be60837ac9ed25d9 : SUCCESS in 45m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f7b608d2607344c08dc696226d346d5d : SUCCESS in 1h 12m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/fc696b7aaa994934b9504ea4ed922027 : SUCCESS in 41m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/a7c8a0cae98540499c1e7cc786d5af26 : SUCCESS in 42m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/c1b3118b84d04038864989c06bb8dcc9 : SUCCESS in 1h 07m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/51dfa328dc604038bb77ab8d76bcac20 : FAILURE in 2h 28m 40s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/0cd97c0d8f3c4443b1800cf55f9b1bfb : FAILURE in 33m 37s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/3f2e4ee674244f579290c088f3bd8cb3 : FAILURE in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e333432f4f3227bb4b7f6ca5aa6509b1b8052d01","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-08 11:36:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"20a21fd0a9b7f1466797dbdeacc9a2df823e9b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 14:11:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c3f12e200049cbb4c47f03b4ff5e33 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/131fe33e5ee5418798d4d18216de83bb : SUCCESS in 3m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08cbe18ab5944ff88a81e618818dde68 : SUCCESS in 2h 09m 43s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8b247d4758ce43d49e8fcc854699aece : SUCCESS in 1h 45m 28s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/26dbf243dba34ae8ba1f508bbbe3bfe7 : SUCCESS in 1h 46m 29s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/d761d7fcae53461db9f5a449b2f6499d : SUCCESS in 1h 21m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7deb010181a4e7c8049f6bb2f333a0d : FAILURE in 2h 03m 35s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0de44d7937534e6da5659aca9e2361bf : SUCCESS in 2h 01m 07s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5f25438a5921436e81501c9d8a690487 : SUCCESS in 2h 14m 46s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/028df7b8882441feba30229f9e4da36d : SUCCESS in 1h 34m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e129930dabc04331a512bc790707c642 : SUCCESS in 45m 48s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b44cf597e6a2438cb6dbd6f8230994a1 : SUCCESS in 43m 16s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/1d67a947086b4424a0a1f4e97858eb6d : SUCCESS in 45m 58s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/220e811c297e432dbc973f1a40200247 : SUCCESS in 49m 24s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8306c3c141a64843821a6af594d4f7b9 : SUCCESS in 2h 04m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b2ab5cbed6f64583bb2398981683f306 : SUCCESS in 2h 10m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d8f90e56a67b435e86b969435332019b : SUCCESS in 2h 18m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf12ed1e45294677a1b2354458accbf2 : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b4a2c5e2a9d444d3975582079b33e957 : SUCCESS in 41m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2b5517cdb73140e59b63c52baeca78a1 : SUCCESS in 39m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3af6717b44694ad99451eab6121fa778 : SUCCESS in 40m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/050dc0b2bfc1455a9bf15e6699adf229 : SUCCESS in 52m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/9f8d0011451f4a7c8c36fb1a0822bb26 : SUCCESS in 50m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/e7001a64bb07431388356ee648802462 : SUCCESS in 45m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/61901e6204b94f4799986de3a2d1fbf0 : SUCCESS in 46m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dfbcf925f8f54cb49fe3e6b9c7638b01 : SUCCESS in 2h 05m 35s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/7985f6731b164383af60cce04c9960c3 : FAILURE in 32m 49s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/488d2ac1497a411abe01bbf17be61352 : FAILURE in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c491d82bfe1136e5d015f0714a0968968b37746a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-08 17:37:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d4148df608b510d68212a55b3c565fcd9a66bef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 22:29:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38e874f609f462da1c1dcee516cd553 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8486eed3e04d4993b59bf9fb79b06534 : SUCCESS in 3m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1bbf9485ca54b17b0d9b5d271befcf0 : SUCCESS in 1h 35m 23s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/ba987b720b9d458b92758563e986a8f9 : SUCCESS in 1h 39m 07s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/1a74a24970514976b480439f788cce8a : SUCCESS in 1h 44m 47s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/3c38fc1809494971ad094fe8b69580d9 : SUCCESS in 1h 14m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/582414b3b9ce48e894af88018d9078ce : SUCCESS in 2h 02m 35s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f78d8cfa16f848bcb4636b3e6ed2c944 : TIMED_OUT in 2h 33m 17s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0360630166704f6ab5a1a5af613dbf92 : SUCCESS in 1h 47m 05s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/f2bf4f7a4d434ba7a98c2ba61f42ee1e : SUCCESS in 1h 31m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8294ad5c3bb34e969ccfcdfed2a5366a : SUCCESS in 39m 35s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/685e11cf49a34df0a65da553780ad72e : SUCCESS in 46m 26s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/d5e14e0e25e84b368382be9f6b8f3026 : SUCCESS in 48m 36s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/49c5aa809f924f5aa45031472cf70c47 : SUCCESS in 47m 51s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/36e5f5495669459c8a6ea609c4183ee9 : SUCCESS in 2h 10m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d507009df8264c669f225dfbdda39692 : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/887482a863284875944ab7ad5dbdbdf2 : TIMED_OUT in 2h 32m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7a54ff8b4b043038b14963cc02a3258 : SUCCESS in 42m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8dd475efe7d446ddad9bc18f3a7f24a4 : SUCCESS in 39m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0804b0b788544de384cfe24b52a949c8 : SUCCESS in 1h 18m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/52ca9ffeb45646daba6aeb1e6f8903ff : SUCCESS in 45m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e034091fc7804cfe9b11e696f81c7e0a : SUCCESS in 32m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ced23f616331492193f396044b81b008 : SUCCESS in 1h 07m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b880f782868a48adbea4fd3463a3e47b : SUCCESS in 42m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/44a6c27092a54df6af9ab1b49160ad32 : SUCCESS in 41m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ed48d43e1d04dd6a578c6ab6e4abb04 : SUCCESS in 1h 57m 04s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/88b34d91d7c7496fbfd5dc3bea7f4580 : FAILURE in 32m 07s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b19e063e68974e2ba775794fcd57f96e : FAILURE in 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b517112c7876dc13bc46aa29bb15c779d13cf45","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-09 00:24:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"24f70e72858ac1cf9920260078bfc233b58d5af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:02:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0072a31381bb43bfaf2c9dc1f1f154ed : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40432c9b7647442da62d445e292ff459 : SUCCESS in 3m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ed73304097a4a78bae592cad69a2b9c : SUCCESS in 1h 45m 03s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/e8fc6177382c4432a7d4417e8de0bea5 : SUCCESS in 1h 44m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/1db6d27d58b744b0b7350baf4c23b74c : FAILURE in 1h 39m 12s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/027aa878412746c68414266cbdb6271a : SUCCESS in 1h 17m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cbab9a224544b17989653463a0757cd : TIMED_OUT in 2h 33m 15s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0e39f4bfedd64c729f87aac23fe57491 : TIMED_OUT in 2h 33m 45s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/428c750587fa451da2f3625020becead : TIMED_OUT in 2h 33m 03s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/2455cc0f2a7e4fbd90495d3434ce10e8 : TIMED_OUT in 2h 33m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/241e74e2230c4809a91fe5d0eb68c7a8 : SUCCESS in 39m 41s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/9da9cd476f5d4b02a9058c3781df9f31 : SUCCESS in 38m 24s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f6b8238908a744c18cf688c050b48210 : SUCCESS in 40m 43s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/68491df6356a457791c6bbb2d4df651b : SUCCESS in 45m 05s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/28c8a58a945d4a4d9dc973b3b6c41aa7 : SUCCESS in 2h 09m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d53a39749cab4715bca5f953fb3751cb : SUCCESS in 1h 43m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5624b56a367f44ee9cd030a998124182 : FAILURE in 1h 54m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/491f312c01b64d9bafbce2ee4e31e4cf : SUCCESS in 1h 03m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0247552df41a40e78579f3b7b4034fd5 : SUCCESS in 41m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/000e8a11edf24eecaa1308039022dd0d : FAILURE in 1h 09m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/83f8feac4bc44715ab7af64d39a4eb58 : SUCCESS in 1h 01m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/61623ad3dff24f17a1dda1842a5ae631 : SUCCESS in 37m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/24332ee5958b42369dd631b903961cbc : SUCCESS in 56m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/970011358c6f4e748481523dd4c52840 : SUCCESS in 43m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/38a428f87c9e4eeb8a322e329b2f9bc3 : SUCCESS in 1h 03m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b34cb2a59f91427b987837eb2388ef5b : SUCCESS in 1h 36m 34s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4400d1b8b00e4486a1cfcbce7492feed : FAILURE in 29m 13s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c98b08f9cb664ba89da12a349abf7424 : FAILURE in 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1462ab760715d65269fb5c258b37bb78946af904","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-09 09:19:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f6f67ee403340ab7b136080b146adb478d63b421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 12:36:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6cd59f41204396852bcd2487b12c8c : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e303caf9e17047aab9506c7801839b7b : SUCCESS in 2m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40f1bdd9d3644cc4993838256585b6aa : SUCCESS in 1h 45m 16s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/f78c459593714215b9fe7720fdbd3898 : SUCCESS in 2h 00m 50s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/0a3af3ff23e944ae9fb923d2ca33f144 : SUCCESS in 1h 42m 24s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f3a9bacd7a504244997bf605bc186773 : SUCCESS in 1h 33m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2868e6d7fd7d4f3cb7fee1d41c070777 : SUCCESS in 2h 19m 43s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/20d8f7cb8fce4c3f85f924a8a58c215b : SUCCESS in 2h 06m 39s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1bf92a9c9c84432bae993e01e556c752 : TIMED_OUT in 2h 34m 15s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fd897bdad0d044f0825fda6c05f358ab : SUCCESS in 1h 46m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2887ac911a254bd88b855ad8f5720b62 : SUCCESS in 44m 56s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/6fe512bd98894a2bacf7436330a7548f : SUCCESS in 36m 27s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/9317b6f6e0ce494881135fc6f6bcf16d : SUCCESS in 45m 34s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/ffab322eb9834beea3501ad191ce6e7f : SUCCESS in 42m 30s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e9bb95d9e0584b269d3274f00ea23c4f : SUCCESS in 2h 05m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1422f9079fff408ba68ad2b2337c7ca0 : FAILURE in 2h 32m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/45b25c0035c9429988fa965df5e87e88 : SUCCESS in 2h 26m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44685c77499645468df575f7690afb4e : SUCCESS in 43m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a6f04f8388234635921850e0afcc6eaf : SUCCESS in 41m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8fad841abe0c463f81e95ae3a72f198d : SUCCESS in 39m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ef9606fcb12c412c8fcb93e7ae2603ea : SUCCESS in 40m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8cb565b27fe84614b12799932fff5eca : SUCCESS in 38m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/be1566f64e2b405ba48c042ae8af6d7c : SUCCESS in 1h 04m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/9572ba0e2b4f4101b9e262ab69b9d782 : SUCCESS in 40m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/4330667a22064c739309ffdf69cedcc7 : SUCCESS in 40m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a041094e2f843708d79bc23d38062c3 : SUCCESS in 1h 50m 07s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5fb98f43ea4747f68d0ebe712da14fde : FAILURE in 31m 26s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5b75dab0f5994ba28ab37547b1d3b92e : FAILURE in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e26ac7c782abafa75cacda62cd987bd51c605b19","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-09 12:59:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fa80674754d7761e12d7ec4eaf788a3e05d29c09","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-09 13:59:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b9a331e7073834038528a765899f9c8c71645408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 18:36:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de386cfda0ca42b39d26c0919eab858e : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4545dc70036f4569ab378fa869975bf6 : SUCCESS in 3m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b9adbd735c034de98221696e233bf9df : FAILURE in 1h 29m 37s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/a67a728a3fdd4bbab1a062f3de679c5b : POST_FAILURE in 2h 05m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/ada06eb42e154e68b87ecf7c22b05620 : FAILURE in 1h 00m 02s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/26ed1d2da5fb470aab99897ac0be9c1d : SUCCESS in 1h 35m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e964837b5cc2488eaae3b36645dfe8c0 : FAILURE in 39m 57s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b42a8b32190949ab9dcaed86b4e6286e : FAILURE in 1h 14m 19s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a6ecc74110d34e4199cc0f8f14384a7d : TIMED_OUT in 2h 33m 34s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fe72bd4bf36946b5b2a826c3f723baa3 : FAILURE in 16m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f20f5e3635564dd49b8d1712632101c3 : FAILURE in 45m 58s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/04e91867f2e64b7a9969ae4eb88a7210 : FAILURE in 2h 34m 30s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ae9e5782b6154dda828170819d0196e7 : FAILURE in 2h 26m 16s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/c1ff7daa20924384b5eafc3dbbed743d : FAILURE in 2h 28m 30s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ef1d0867bf7c4cf3816d8c9187c94c76 : FAILURE in 14m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ca1b1c8ac4af432f9a24ea1a7d46a3e7 : FAILURE in 31m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d59b8df6d144dddb5d08f59e0a2183b : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa74734147fe4b7aabe6d233bff5b697 : FAILURE in 1h 56m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a8edc2901920490fb94849e59f58c19c : POST_FAILURE in 55m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a9d4b32ece994b1daa84b42c56086000 : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ada12cfa591a4044a2a7e4a991baf0b1 : FAILURE in 1h 01m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/65f2911c54bf40e5b14aac47372b0258 : FAILURE in 40m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/7f6400d95dee44f3935fa92bf02f7267 : POST_FAILURE in 1h 11m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/8aca990dd9374cea8f7e1819cce4a75f : FAILURE in 42m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/695859b5a47340ae938aae0a306de817 : SUCCESS in 33m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9cbbcbae634b4cbe8eecc96ca45493b0 : FAILURE in 14m 00s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/336c8d90bbb5427997a2a1adc301a6eb : FAILURE in 40m 50s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e49ac75a60ea462eac4d007e8a6a72f6 : FAILURE in 38m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1a7d80f3498ae4546803eb5e7276e8800c113bb2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-09 19:06:37.000000000","message":"Patch Set 8:\n\nrecheck\n\nCI issues. Affected region disabled in nodepool for now.","accounts_in_message":[],"_revision_number":8},{"id":"36200f5d198da1f5515e95e1430e3aee0bb166db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 23:24:54.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c973ea42bb094ee2acf020a909ac8dd3 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a79e65c11c4b4187660d9e22b3f107 : RETRY_LIMIT in 4m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0602143c2ed34b65aa8a07fd4bdb3aa0 : FAILURE in 8m 17s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/fa28ea273e1149459e76671118b067ee : SUCCESS in 1h 15m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/39213ef15616419e80908c5931b88a20 : FAILURE in 1h 17m 32s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/1203e4e2930d4f7bae66afc3d3173eab : RETRY_LIMIT in 4m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a70bd02241c4572a77d62b4e4065d05 : SUCCESS in 1h 58m 39s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f1a8659f7f8848c7bbb41759c3d76fc0 : TIMED_OUT in 2h 34m 45s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d2ac0f8b357d4225a732352e38a4535b : SUCCESS in 1h 32m 24s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/30d6449f12e3421db9fac81a2a6320b1 : SUCCESS in 1h 21m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dead75738a854b2ba4763807c816b40d : FAILURE in 1h 31m 17s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/4b6d20197c654e85aaa3222b9f681f74 : SUCCESS in 43m 33s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f616a99831da40e3a1b17be8dd7e7907 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/16ab68602e0d4221bf33e15f2330bad6 : SUCCESS in 46m 38s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e78e50f2692343edb0c9038552459f4d : FAILURE in 2h 13m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9852558fba44430f838f0cf129ea8a45 : SUCCESS in 2h 06m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f7619ca46248472da49c319dfa03d374 : FAILURE in 2h 21m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b49bf9eac614a4e870621e9addc48ff : SUCCESS in 42m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6a23e211ea8048cf8366f0c6719b8523 : SUCCESS in 44m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/44ee9bdbb3114f1da674f1a57788b464 : SUCCESS in 39m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/66be38603a5748ab8e511cbedc9a2186 : SUCCESS in 37m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/14238847955c45dc8e741da9745cbef5 : SUCCESS in 50m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/d3c9efb015a1466f8137203cef9b2905 : SUCCESS in 51m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/02f0cd3e987e42daac26170f2c6f9dc8 : SUCCESS in 1h 47m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/638fd8d8161548f4a6c7c8f056c50872 : FAILURE in 1h 51m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e4d0a858721c4bf6968b610e6ad2f2fd : SUCCESS in 1h 37m 00s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/7b4ecfecee5547249ae82cba4598bd90 : FAILURE in 8m 45s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7605bbae9c884f07b41419d234afb7ee : FAILURE in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9628387e520947b68179291bf84b60073db13dd7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-09 23:28:31.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f7ff7c008c73ab97caa58c8c7eff5bd5255477a8","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2020-10-10 01:55:42.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"713f1d60ac1bba2e19c8ad59cc2a29da113619ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 02:09:49.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ae87998af94218a88772305bef8da8 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e136008bcc6482f9edb372d5bbe3500 : SUCCESS in 3m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eae3ca841137464db9a09c55bcd26ba3 : FAILURE in 1h 19m 04s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/4569356cdcb748e4af941412529fb5af : FAILURE in 1h 21m 25s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d00cb9641d3a48419a7c8a07c385c518 : SUCCESS in 1h 11m 25s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/5114594e33394429a33f46a7225ab885 : SUCCESS in 1h 01m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d933094ae2947a8a417ea997367dbd0 : SUCCESS in 1h 54m 42s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/77d5d94d924647fa815de68aeba80274 : FAILURE in 2h 31m 11s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/edd16d699be34cd0b155b1b2a0460435 : SUCCESS in 1h 57m 01s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6ee433f9efc44256a022fbc618f18cd9 : SUCCESS in 1h 16m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8eaf166bc204c9b8fb62f22217bb8b9 : SUCCESS in 1h 26m 05s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/c212ab10958b4ecda0c6a9b6b8b7ac89 : SUCCESS in 44m 57s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f902b526ebf0486bb3c5c1c237412eb8 : SUCCESS in 47m 26s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/96fd4059448848bcadfa41d665991167 : SUCCESS in 51m 37s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6b25f8a523f647148f82975e3dc9b8af : FAILURE in 1h 49m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9fa88057200d46b083b2c65ae9c9f633 : FAILURE in 2h 19m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b80b9fba9e4646c7a1639395841db538 : FAILURE in 2h 16m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7716860c694d4f71af0cbae23000f61f : SUCCESS in 52m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1a09fdf958f54a1180a9df5fbf100020 : FAILURE in 59m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e1f7ea7394d74a8ea078ebc5f7589946 : SUCCESS in 36m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9f408a93f4da4a5993d138d55660567d : SUCCESS in 44m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9d86220b532c4014bce75b0e87eb3df7 : SUCCESS in 43m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/9f3c02c79c8e46cbb13ff5a6873c2760 : SUCCESS in 1h 11m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/edbd8be1b777400085c104a260864ba7 : SUCCESS in 34m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/44ac8c3ed86b4b8188f5509494d55976 : SUCCESS in 43m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/87da1e126ca84d2499952fd8d98f857f : SUCCESS in 2h 25m 29s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/cfabfc7d834341c8b93a8dd2125170d0 : FAILURE in 31m 08s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a1b3dc025d0144dda411c40a1a3d67d0 : FAILURE in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c7af6c8b25381490c6fd35a759ad0caaf2df3d74","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-12 15:41:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bc93b40b240bb5345b05a2701c8fea8c91eb1b98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 15:41:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":9},{"id":"63f3eb72cfcbb078f91db582c586370e38d9b5b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 15:41:54.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":9},{"id":"dab25ce3437357bc4e1a5e49b34dfc0cbbb82d13","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-12 15:47:04.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"06394b37ae015e01846f6da21907e4a5f2daded0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 21:37:20.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3bbfbce5e6c4d8aa2477b33e9f373a0 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb6b8fb4d534471833a1c43d871d5a5 : SUCCESS in 3m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ab6a58c866144cd9a3cbb1039410497 : FAILURE in 1h 31m 44s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/dcb81fd1ed804dd4950b62b3b16fc058 : SUCCESS in 1h 15m 34s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/262d3ac877f0419b81d82a68a80bdeaf : FAILURE in 1h 24m 04s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6432ca6271eb463c8d5f4df2053397eb : SUCCESS in 1h 08m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60709249399b45008cc8b0e20e725353 : SUCCESS in 1h 54m 11s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3b8719d80b454dc5bad18f07baa19f03 : SUCCESS in 2h 13m 36s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1fec9f52caf64f28ba58662c477b4715 : SUCCESS in 1h 55m 28s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/cb746aae17ce4ded84e8974ece96fde5 : SUCCESS in 1h 20m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fbc9b79dc452423dbaf46b51f4113a23 : SUCCESS in 1h 21m 59s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/2b25e1442d1d4ea6bdee0bc90db19562 : SUCCESS in 43m 39s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/b6e658119d294afdb6434163325f6436 : SUCCESS in 45m 48s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/b93dc985519e40f99a53d42f767f5e29 : SUCCESS in 37m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a5c2c4b9171842d097a95db20d036415 : FAILURE in 2h 01m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b21cf87473ad4521bcfef3483ab67697 : SUCCESS in 1h 50m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aac9c3bd17014fb0aa09a1c23be956b6 : SUCCESS in 2h 02m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d16c1d8132c481f88b958c9a0dec69d : SUCCESS in 41m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2598aa3f2c73400dbdf51a76b5e3dbf0 : FAILURE in 1h 13m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9b803f64e3f945f6b91c846a887c0146 : SUCCESS in 34m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/90e22650dec245f4b6322f9c50cb33e0 : SUCCESS in 45m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f00982c7737c40bb8b134c23c9de953d : SUCCESS in 31m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/8c31c172ad70462cbdbfa3d5c94bca43 : SUCCESS in 1h 17m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d3f8b208853443ab97563c74aee4c621 : SUCCESS in 1h 14m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/7755a7c697e2496ba35f1a993701290c : SUCCESS in 1h 22m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/69ff5962684c4b2e905638bbbadb8245 : SUCCESS in 1h 53m 46s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f2798d08752840a996152d558262ecc3 : FAILURE in 1h 10m 58s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f1be6414b536417db50d33b70f604a9c : SUCCESS in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1240c3a3c2823b71fc6cf6c21d84e065c2ad36a1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-13 06:12:20.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"4025cb9c355440081edf0b215e16ca731e2a8d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 09:28:50.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b4700eca724e91bd17eca678484891 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e3f81fdcd948df9d95fa58fe991774 : SUCCESS in 4m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/508eb714b8704ec9a96b9ce9d31a73ea : SUCCESS in 1h 20m 29s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/21ebdf68bddd434ebae6493b8db23212 : FAILURE in 1h 23m 36s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/39f401474e4c480b97dce1fac743af8d : FAILURE in 1h 22m 03s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/ffb4c49021494be1b16e4b8ee4ae5aed : SUCCESS in 1h 12m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4938dbbd0594555a4db021a9c5dd692 : FAILURE in 1h 58m 14s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d0acd498282a495c850b7282153060cb : SUCCESS in 1h 58m 30s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2be59cfd74d041c0b3eae539b71de077 : SUCCESS in 2h 20m 21s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/c9a0130069234e80a587fbb7436791f1 : SUCCESS in 1h 24m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ab4422cd24c40a8ba4cba05c153f6d0 : SUCCESS in 47m 32s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/48f700c5a53142898830c91ccdd16b36 : SUCCESS in 46m 17s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/e408b717b59c44819555f128323315f7 : FAILURE in 32m 53s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/1d7a6997449b4e4e9df8ad24a372ecc7 : FAILURE in 33m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c48a35f9187f4be792dd62a368d1aa28 : FAILURE in 27m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/af26208284c04c8194a45bd12edaf601 : FAILURE in 34m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8798c0e9346d4ff795be5f6ecdcf3d05 : SUCCESS in 1h 54m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83bfb7cf0fa44d57a9eb070ece12cf73 : FAILURE in 29m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2819e57750084901ae33bf36f88500a9 : SUCCESS in 43m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5f506d870a2e496d9383a1cc3652ddff : FAILURE in 31m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/14586f9ce3534a66aedec270199d83d6 : SUCCESS in 48m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b55cb515ef294be6b9bbd9738e2be4af : SUCCESS in 44m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/016f962378954e62a4c99fcf437d208f : SUCCESS in 58m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/3c88f9dddd6142f785162a45da25b18d : SUCCESS in 45m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/fc2745054b6c44f09a1972653adbb479 : FAILURE in 30m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f32ee3ab7cc64cb996ac8fbf79a081e3 : FAILURE in 30m 13s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/63a1b2762776417384133a0797c236ff : FAILURE in 1h 06m 58s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b31611299f0e4683b9593a2ed1f837d7 : SUCCESS in 43m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"49acd011cb099bd0e3af76dc75d838bb98945394","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-13 11:45:38.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"75cc8b3e334ad410c95bdd488ad70f4e32701ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:24:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228f20e0d7684187b2ed2cf723392efc : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee27ae1630f43779d1d7a0d3ab666bd : SUCCESS in 4m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ddeb83aec3634ed992c61ee0383044bc : SUCCESS in 1h 19m 18s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/bf268ee2e88f4b1b8683fe2b71b517bf : SUCCESS in 1h 19m 54s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c88d1c2356524713ad9bf753604cb36a : SUCCESS in 1h 21m 03s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/0c89ffcf69d8456eba4a7b145fb940e7 : SUCCESS in 1h 04m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de780eda3bfc461e9925288ec6ed38e9 : FAILURE in 2h 10m 29s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6d23a7ad2a774b0bbb0791477fd65000 : SUCCESS in 2h 06m 11s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0bfcfd761bee4e4b8bbac4bcfd9c8fa4 : SUCCESS in 1h 53m 29s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/88fd3f53ed114bdba58d28768e8bd1be : FAILURE in 1h 21m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9ada02f5865412fb003489edaf13084 : SUCCESS in 45m 58s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/11c96e674695449aa7d481d920eb7ac3 : SUCCESS in 1h 11m 29s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a959a211c25b4668b88f5b88301696c5 : SUCCESS in 45m 35s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/acfbda9efc5a4b03ab3ff604818f8183 : SUCCESS in 1h 12m 09s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98ecbfc2a480457b91d7e069f9f79f77 : ERROR Failed to update project None in 1m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4b3bd1176a534219baaa34e382a94161 : SUCCESS in 2h 02m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73f1a1322c594bd980cd9ad2f0b295e6 : FAILURE in 2h 12m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b1bbefa01844b1d8929fcf74ad13835 : ERROR Failed to update project None in 1m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/53fde0c7ca6e491fa534fb2a0f444a9a : SUCCESS in 54m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c92fb476f70f493c8433b22b522e23d8 : ERROR Failed to update project None in 1m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/db70322f130c4e509b80c64ddf53c7f4 : SUCCESS in 42m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/734e470d428646f6a46822404c5b6427 : SUCCESS in 38m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/6d9581da7bb34c4087079dffdc83d5f5 : SUCCESS in 40m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/73bfc8e33da245a8bc3b61e555cbb534 : SUCCESS in 42m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/48d503c809284a6b85f3fbae247712d8 : SUCCESS in 44m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/30d5f69159ec44f89d72847d09cedae8 : SUCCESS in 1h 56m 52s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/0c597889f70c456ebc81f2ef9cb8aeba : TIMED_OUT in 1h 32m 37s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c56127937c0542e5b7a455bedc3ea53a : FAILURE in 1h 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"44ece3e4e4fb0a3c58cfab079a714acba265ad2e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-13 18:34:52.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"baa9ae06a4b6d799bc6b131b4f8efd5ed9848884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 22:17:56.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fb96e679e24643aca0197b628c405a : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f7b41cd3e14448bedc213cc193e0df : SUCCESS in 5m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d9e43218ece64e6da7dd6100f072ff58 : SUCCESS in 2h 03m 28s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/c29cf3f06f474b68b4d7dcb5e6847e99 : SUCCESS in 1h 18m 15s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/06087e4173e140318e2778ab084b5c63 : SUCCESS in 1h 48m 13s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/0ceba2440e28442abcb0d42472d8d067 : SUCCESS in 1h 05m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec9ed48ebb44415ab87fb29bc0ff7069 : TIMED_OUT in 3h 37m 32s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/98b32edd63ad46ed9a66f02282cb4cad : FAILURE in 1h 40m 59s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/25f5d8c160684792961f74285cbc7006 : SUCCESS in 1h 58m 19s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fcf3d0c0db184700aa9548b64957bf4b : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6453f0e65774949a17af2e47c8575b5 : SUCCESS in 1h 28m 44s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b10c9723bfc54e83b640178365f9ae33 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ecff235f81304dc891eccc36bfee663d : SUCCESS in 44m 07s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/5203d362ae1748d79160d4cf19a8ead7 : SUCCESS in 47m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/324c3051e53b4daf9b894673d8b76d9a : FAILURE in 2h 15m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/081b2a2098484d2ba0bbbd84fb9a26c3 : FAILURE in 2h 01m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f99b2464e65e45e0969e65413fcfddc9 : SUCCESS in 1h 46m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b098e6788fd846a8becf30404a2b3315 : SUCCESS in 42m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/98221607c9234d2d802b6b150cdfa2d1 : SUCCESS in 44m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8839c1f180fb42ecb0a4c88957f8a62f : SUCCESS in 1h 14m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/3b9bce4888434d978807c578d3bd3b5c : SUCCESS in 35m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/57acd322c15c4b65a4fab2699084df60 : SUCCESS in 1h 21m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/9870170ce34f488bb6aa548073774e17 : SUCCESS in 42m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/a995c955103548c382aef4b7251d0598 : SUCCESS in 42m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/8e1ae6d543164b799e27b10da62a41d8 : SUCCESS in 41m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/03c386f1dc764ff381d6f3509dafd26a : SUCCESS in 1h 51m 29s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f381d83dab60499284e0f2d6bba14436 : FAILURE in 59m 43s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a74525eb6a9d498eaa39fbb7ed3e65fc : SUCCESS in 40m 26s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"13b0d4f86081a8b88a0a8412b8d384793a9777c7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-14 08:53:54.000000000","message":"Patch Set 12:\n\nrecheck\n\n- octavia-v2-dsvm-scenario: failed due to what appears to be Infra related issues (e.g. timed out collecting logs, etc).\n- octavia-v2-dsvm-scenario-stable-ussuri: gunicorn issues. root cause unknown at this time. may be that pending stable/ussuri patches help?","accounts_in_message":[],"_revision_number":12},{"id":"e70d6d86d2503d82ee8c6d0d7948dfce3b8f3640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:25:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c348a92e7346f0991bd0d0758f2568 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f358cc8bd3444dbb2a26ff3cdaadfd : SUCCESS in 3m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e58ac3cf014421e9ea78ed4e974b85a : SUCCESS in 1h 19m 57s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/a7854dd1674a416998122778138c3d60 : SUCCESS in 1h 20m 42s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/54d8e3c2858c4d359394f7ca40105b45 : SUCCESS in 1h 19m 44s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/61289265cc7c4bb291e9e6e5e25d1adf : SUCCESS in 1h 13m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1fc7853e3444f608f4880cd1d498d15 : SUCCESS in 1h 57m 40s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5253ccf9ce71404f8f015c970859a1b4 : SUCCESS in 1h 51m 21s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/504f675c6c774bdcaca965e087edf159 : SUCCESS in 1h 52m 57s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d74dff313afe41e3b4c31d15bdaac80b : SUCCESS in 1h 18m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2afab2e296d94e37ad6509f598dafc9f : SUCCESS in 38m 23s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/df6c9a23e7a240a88ee418428b4730e3 : SUCCESS in 47m 08s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/161c958aeea24e8bbec1854f6b96f6c9 : SUCCESS in 42m 01s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/b41e9798213449b799e397a85686b787 : SUCCESS in 42m 40s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/540299975bce463a9057c1ecfaae8d8a : SUCCESS in 1h 49m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0d7b45478e8b499788b3e8487ec803c7 : SUCCESS in 1h 59m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c87637255e04e02a9724cef4321836c : SUCCESS in 1h 49m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82c1c4800084464bbe21d706acb8cf86 : SUCCESS in 44m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f57f290c39b1487fb2cd11c458628619 : SUCCESS in 42m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3f285fd4a9b548548bebece6b1cabf34 : SUCCESS in 44m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8f8f69ceef2b4b03a1a3250ebc70050b : SUCCESS in 41m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/71cccb206f514f72acad9e49559eb6f1 : SUCCESS in 41m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/e27c2d61730c4f41acd1db9d3b20bf8b : SUCCESS in 43m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b0cefec65c7b44d9ade9ac61bf7d02cf : SUCCESS in 43m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/7271574a1df3437599b98d7af8e184cc : SUCCESS in 37m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a8d20338df02440a9ba5c79b39a1b8b9 : SUCCESS in 1h 55m 22s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/fd49313823f1418ba205ee0472748091 : FAILURE in 48m 46s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/ccb2c5d152494b2eb06b1a68a36a69e8 : SUCCESS in 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"37d4e2a046cefa0bb73c04db4475b5aedb5255df","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-14 14:10:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"40155a075fa49e2385c59c5590976c1e1cb186d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 17:06:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75780f27376547d2a069dfeaeb2b4315 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6701e0bb74514fa9b383ce9397575500 : SUCCESS in 3m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8f7054ad0276444f8ad0dc21d07d6782 : SUCCESS in 1h 19m 56s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/2e80b5aa97e742298d81249bd2119718 : SUCCESS in 1h 25m 50s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/3a2eab00730d4e7aacf5b432c2001909 : SUCCESS in 1h 25m 40s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/a4e543f7c80746c885cc775a5a3eee9d : SUCCESS in 1h 08m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97eb952e9657492693f049679610a2a1 : TIMED_OUT in 2h 33m 29s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9b65603f62924b97ac71ecdd29e477f8 : TIMED_OUT in 2h 33m 37s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b6cdcd3dfd444c3d80a1f09a9b05f5f7 : SUCCESS in 1h 52m 08s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8c105cfe30e140938cb88c2872df2600 : SUCCESS in 1h 25m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2a334e929bc47aa855e9443d6da2524 : SUCCESS in 43m 55s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/e6ac5b39e01c4df4938b65e9e4dc03e8 : SUCCESS in 44m 57s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/c34bfeee06ed46438003ef69943158b2 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/fc999f175c024000b80dff52bd814704 : SUCCESS in 44m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9427b52ba0624b82b2b149cab1779cc5 : SUCCESS in 1h 48m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/87019d1144d5417d818992b090a9f90c : FAILURE in 1h 56m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c8acde3ea39d498c9fbbdd731e048e2c : FAILURE in 2h 28m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3ae2783ae764d3d87bc34724e4c0165 : SUCCESS in 43m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0415049ed429489d9a6f444f87aa06b7 : SUCCESS in 44m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/89346bb1be784e64b6d3d5c064dd659f : SUCCESS in 47m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/bf864ce476494466993a1ce12e51ba31 : SUCCESS in 46m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1f3d7f158f1b4feaa0789c637217ee88 : SUCCESS in 41m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/1af193aa15ad480f82614c5fcb84b0b8 : SUCCESS in 1h 55m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/c2f8bc66dec54597ace2aeac7de1e1ee : SUCCESS in 43m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/19ed7d3230ae412ab3075ea0305e3d3a : SUCCESS in 43m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/56b9a69027a14e308dc1cd149a38cb9b : SUCCESS in 2h 10m 10s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/bd12d71d1d4e4d3b83788a69bea64b6c : FAILURE in 1h 00m 26s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/88d2a52774c146a89ad6590a8dc596e0 : SUCCESS in 49m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3528cb4608991aabd55ccadec724f1301759ca05","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-14 20:38:28.000000000","message":"Patch Set 13:\n\nrecheck\n\nslow nova in both failed jobs (\"Waiting for compute id XXXXXXXX to go active timeout.\")","accounts_in_message":[],"_revision_number":13},{"id":"01b5407ab9edc76e1cd48998ecea7779656d492c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 01:32:28.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718d144af3794acb89300139963a5cd4 : SUCCESS in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59effbd785704e5c819c6c056c2d4045 : SUCCESS in 3m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89422e148b05441ab998b8340a2755a8 : SUCCESS in 1h 46m 07s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/100e3dfe218e4bd190ff80e0115d38e1 : SUCCESS in 1h 14m 53s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e22c31bb45ef45278f692b35c9083043 : SUCCESS in 1h 22m 01s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/32c9f5508ea84e8880a606308aae527c : SUCCESS in 1h 05m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/703ce78f1b4043a689c264755ae3b05f : FAILURE in 1h 57m 52s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/203a0b25af38450590f988d3285377bb : FAILURE in 2h 12m 30s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/674f5876cdf84183b697095d1ef29425 : SUCCESS in 1h 42m 55s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a4e8e60fcaa14a31bfe69f3aa94aefaa : SUCCESS in 1h 52m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36a28eb535424cde8101ebfe6ca2d61c : SUCCESS in 52m 55s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/8273971cbdf842209d81ecd6fa71abd2 : SUCCESS in 47m 28s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/9af7dc9a9551445b8a9a179e3ef3d738 : SUCCESS in 48m 23s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/cbd935e6fd674401a200830949e3fc81 : SUCCESS in 47m 23s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/918fbdf9502b4dfeb57c6392ed713e3a : FAILURE in 2h 00m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dbe07f517ff34ac6afc34c5ad1c2a498 : SUCCESS in 2h 09m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8c64b52dd6a4c279da48ef0551ec56a : SUCCESS in 1h 49m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cccc3c26942046a99158cfa6d50595c9 : SUCCESS in 43m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c7be53e0e3124f469eb7af114058f05e : SUCCESS in 45m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/44983f63ae704ea6b9aa73ab9e0f2744 : SUCCESS in 43m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/440b52d600a44be3aa7d54f4c85cc055 : SUCCESS in 42m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bfd358a225954f9d894d5096d503d26e : SUCCESS in 40m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/799edf4d94254c43a345b71ab32f57b7 : SUCCESS in 42m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/306d652d42e841808e25561a3d036237 : SUCCESS in 1h 10m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/318611085ed040738d472efad8272c71 : SUCCESS in 44m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/18c178278ef1422480b5b9615b9c6b06 : SUCCESS in 2h 29m 57s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/dec3874db6a4473b804eb676a060f9fc : FAILURE in 1h 01m 24s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/713a730bda72452fb1058e04c3c5cee8 : SUCCESS in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0c78881cae53462ea1534050149d84d19d85ac18","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-15 08:49:48.000000000","message":"Patch Set 13:\n\nJobs failed on different tests but with the same cause: the gunicorn worker timed out after 3 minutes waiting for the reload API call to amphorae to complete. There is evidently an issue with the reload process within amphorae. I am thinking about increasing the timeout from 3 minutes to something ridiculously high just to let the reload complete and analyze the logs.","accounts_in_message":[],"_revision_number":13},{"id":"f282e503044565774f175f33ff4383ee2be529b3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-15 08:53:30.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"0ba371c103745a2264d2b3fc0b547e294c933517","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-15 08:53:48.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"262be603e7006ecf7bd8c46da83712fba246de89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 11:37:17.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613f75b82f6a4322bdd4eae81f074a66 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27930ccf1a2643e08ca96014bc8f6619 : SUCCESS in 4m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc2317052186409c9457216312d80217 : SUCCESS in 1h 22m 08s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/96aee93e1b9d4a5a913a4b9bc743cc3b : SUCCESS in 1h 17m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d4143a196411444ba1ab40be9c2b8af4 : SUCCESS in 1h 20m 46s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/c65ac307add842aabb94a19576297790 : SUCCESS in 1h 10m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/531a3cc74988433fbad15c64e181bbd1 : SUCCESS in 1h 59m 55s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/34157d5fd6e047a9b5f3be0c441599c2 : SUCCESS in 1h 55m 52s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0fab2b95c41a4a06a0029d549671b4af : FAILURE in 1h 47m 00s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/225988b5c00a497a88278c3f1d95ce13 : SUCCESS in 1h 22m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24ca0b4aaee64f9c9c3806a943ee3b42 : SUCCESS in 45m 08s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/0007f7061274476fab14045014ddc24e : SUCCESS in 42m 33s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/9f2d1c0f7e3d4c0786fbac80cab1b0d3 : SUCCESS in 45m 33s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/bd2372dce02441a49e183791843537ac : SUCCESS in 44m 53s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c11d2d932b7b4dc9a993ee146d5eb96c : FAILURE in 2h 01m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a0dfec6b329e44ebb14ebb8156a6c214 : TIMED_OUT in 2h 35m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cf7ec1d9b574a158d338fbb96aaed4a : FAILURE in 1h 51m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6160eb81d874031b280b6340873867d : SUCCESS in 39m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9c294c922be94efcaf279beae9bbd796 : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a01d55a6490143faaf5429a56b6f9fe1 : SUCCESS in 45m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8b3a1a56b69b4e50b96015626b37ee45 : SUCCESS in 42m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6d71790cb52d4ceb94a3f9fa14da91f1 : SUCCESS in 38m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/f5990dd02ac04b57b0cc344c4ac0d17b : SUCCESS in 37m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/47fd32f46cae409b9fb8f60c7860d6fb : SUCCESS in 45m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/5e29f202ad2b41d08502617ff65c14b5 : SUCCESS in 44m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d958e36178b24b5fa0d870265e5dd998 : SUCCESS in 1h 47m 49s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d07593de0a894da0b4f21e04efc905eb : FAILURE in 58m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/d4e9587ca2fd40bdb6dd721de958360b : SUCCESS in 45m 23s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b18a24e6aba64b2878941273d85f3aa1ad83037e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 05:53:45.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"1024c1a3c294c7a5577912b7d6de082301ffce35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 08:12:08.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3b66a3b37a54a3a87a9cd51ca10003e : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b6d5e0cc8e41f39908bcda59895d8f : SUCCESS in 4m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/348840e9ed71472c86170144f6ad41b4 : SUCCESS in 1h 21m 39s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/5150b7d6dd904682a88e0510f121de22 : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e66afc8c92254082a9f6baa86183aff5 : SUCCESS in 1h 22m 35s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/79ecd8d2ebb5447487d69e6d37dfdc0f : SUCCESS in 1h 16m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc92e21a544b4466a9d8838fa7398ed7 : SUCCESS in 1h 53m 22s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/496b4b7fe9dd4fafa4117ef9c7499e00 : SUCCESS in 1h 40m 21s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6c3229de2ad74a63a3d53a870375c519 : SUCCESS in 1h 47m 28s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/005d161356014e23ac57c6ffb8c6af5a : SUCCESS in 1h 23m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8e536728162c45d580eb9ed01c864f4b : SUCCESS in 41m 10s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/6791834f252045f6b3c0db9a2e13e010 : SUCCESS in 39m 10s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/cf58073bb92f4a44bd647201dfe6baff : SUCCESS in 40m 23s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/e5fe5e47fcbf4fd0ab4505fb2573ae7a : SUCCESS in 43m 05s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0a9014b014b24929964cc6e8ad728e89 : SUCCESS in 2h 13m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/439aff73883f43019fedc6dae24508de : SUCCESS in 1h 43m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa85f8d3fb66405f9b2464d79081573a : SUCCESS in 2h 12m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c0f5423e88642018c7029a34b15107d : SUCCESS in 40m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/49efb9518dab49a4ab34a1b3bbed9a47 : SUCCESS in 42m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c39c0a343cbe4e2f81f6522cd335cb7a : SUCCESS in 43m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/2432f3421e2b4e04a10705870632552d : SUCCESS in 40m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b1a98d242dcd4fcf9b5fb192537d0977 : SUCCESS in 39m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/61fabf384f76427cb383e274ebd2dd3e : SUCCESS in 38m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/94f19795ab2e4062bd83c94a3fdc1cc7 : SUCCESS in 50m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/d5007eec6f67404eb1aa87ca43917e79 : SUCCESS in 52m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eeca76f3b2a94a7baeff8bebcba2ac25 : SUCCESS in 1h 50m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/88e2590eb1114d148462a649fdac1bef : FAILURE in 56m 40s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/539dff6e7a74476dbe511ab162f0aa81 : RETRY_LIMIT in 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9ad61b077b5086744f49338f8e4258c88b08c771","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 08:30:10.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"f71a68f29278bf4ac55e3542aef978782566f04d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 08:33:01.000000000","message":"Patch Set 15:\n\nThere is still one remaining issue but is less sporadic. We can have a look in a follow-up patch. The patch should be good enough to provide much higher passing rates and unblock master and stable branch gates.","accounts_in_message":[],"_revision_number":15},{"id":"df7bea0bfbde83dfa3172dd911a783ea45e416a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 11:08:59.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88093971944e4689a144570f2adf8048 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3587b0f6d74b45f3ab50c3b4e410a9c5 : SUCCESS in 5m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/178375f8924849df85eb05317222e101 : SUCCESS in 1h 40m 54s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/25e597ab84e84ed08a3606f997a9eef8 : SUCCESS in 1h 19m 23s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/839bf98fcace43d882d3c45152916fd6 : SUCCESS in 1h 23m 32s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/40efade0e94e468990bf9eaf8dbf2615 : SUCCESS in 1h 24m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d9c96d0951a4f358d218e79df0095d5 : SUCCESS in 1h 47m 59s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6ce1cc2a35534cefa3351f2f73bf965d : FAILURE in 1h 34m 04s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a4deeef063ba4c57b61f44e7f96b293f : SUCCESS in 2h 16m 42s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/783a842f59e74684a08eda4f26766603 : SUCCESS in 1h 24m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5918611dd9f04ca39aead89132f8f233 : SUCCESS in 50m 31s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/ecd569c8325744fb82770001c137107a : SUCCESS in 44m 51s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/6473b72277f6405dad3e99160a52cc38 : SUCCESS in 52m 44s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/64f3bdd1aded4d618508c63f6700905a : SUCCESS in 48m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/14e27a25239648e6accaf0fb8621c152 : SUCCESS in 2h 00m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/196f3a81f06948a281ce1c6d3d539fec : TIMED_OUT in 2h 35m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9dd24b016284089b4fa1be51022bc27 : SUCCESS in 2h 12m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/924d762b12c34740a85226e85bea70de : SUCCESS in 44m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/27749938adf947e68fa7f33cd5b2afdd : SUCCESS in 43m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/82196586f97242149230137e6de625e1 : SUCCESS in 50m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/e23138a73bda4eb9b9faa03c77236672 : SUCCESS in 41m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/580c2741b58a4fb487790196d88fd253 : SUCCESS in 48m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/00e2e1d280b043bab1cb299f0743d52a : SUCCESS in 38m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/4e79b9738ffb4f88a6fa39e418df2b1c : SUCCESS in 48m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/e6029546834c4b3e9e836f8772bec2dd : SUCCESS in 40m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cc55f808850a4b55bdcfa72dc5777adb : SUCCESS in 1h 38m 26s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/ad07b419a15b4643a5227877bc8fecd2 : FAILURE in 1h 22m 11s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e08a4a49c48542afa32a5b76d0e67991 : SUCCESS in 47m 47s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"de0090818e93c6da773fb7b2b97e230d8a63f956","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 11:13:21.000000000","message":"Patch Set 15:\n\nreheck","accounts_in_message":[],"_revision_number":15},{"id":"07653fc8525e72b24733a97263627d998bd679e1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 11:32:13.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"abf9bee2713e595281e9322c6a84e22ff0d7e4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 14:09:18.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb3c1f8bc63447f95eacd2e4ec15331 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d5a6857a4547239b0af2d2c5a3daa3 : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d6ffd5c6f444f4899e47d0a4a7a00c5 : SUCCESS in 1h 21m 00s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/da033ebd0a334bbaa7e79137b54cf65d : SUCCESS in 1h 10m 44s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e296df52ff9e4f429d0ede67b568655d : SUCCESS in 1h 13m 03s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/7d0a95542bd64aa197c7f505cfc9ca38 : RETRY_LIMIT in 21m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09f2100888ce461ab6e437fa39a7bd19 : SUCCESS in 2h 16m 25s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6fd98407fd7f4d559d9fecea500d724d : SUCCESS in 1h 49m 22s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b8b60c5ea7df436c863fac4e582b4a7e : SUCCESS in 1h 54m 03s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/8de344a1f226479f9f7f59e6171137df : SUCCESS in 1h 10m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4bba2ca9f1a741b4bdd68ade62537cf8 : SUCCESS in 41m 38s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/53f8de53bc414fcfba0ce58d2bdc55b3 : SUCCESS in 44m 43s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/829d37b1bcf243039396063bd69660bb : SUCCESS in 43m 25s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/ce8b795fc19249dcbc05d43fef02856e : SUCCESS in 43m 18s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0f2858fe77f7409aaf4be608240fe47a : FAILURE in 2h 30m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ef40bddcf1ab45f0b983b7e4af28e7f1 : SUCCESS in 1h 48m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f2bff69364a499e924e9d18c6d5e80d : SUCCESS in 2h 02m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fe80a949b444b9db71d00a872c6042b : SUCCESS in 37m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6b5e3a41f3e348cda307b3dff379249b : SUCCESS in 48m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/565fc912ec7044b1955674928373e623 : SUCCESS in 45m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/dc4851e175ef4afbbce37c37744412c9 : SUCCESS in 44m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/77c19cbaa7ac49f6b00f6edbffa2ff4b : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/dc7d5417825d4d8e9fc9c17ef4e80449 : SUCCESS in 50m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/a73cbcecb861417d882b9dd40682e36b : SUCCESS in 41m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/1c7bbfc45ceb4232b3b3c416e4d7c7b6 : SUCCESS in 40m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/874b596210304afaaf36ffa007a72c8c : SUCCESS in 1h 47m 55s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/249df3dac86742f1b50bc60246d7eda4 : RETRY_LIMIT in 23m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f58086ed5f7f4bdcac73c14f4a1e0c1b : RETRY_LIMIT in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"670ba4a972a498b849ff0e76a9229c2afd6f4b99","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 14:14:55.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"306a6486909ee7869cd72c78042a8e9cce6f7969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:27:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec86ce0bb774fd6b32d8f7d3c049da8 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b898448a0ff14f43861e91337b7800ba : SUCCESS in 4m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64058eae07c24cc6aedf9abbdbb6943a : SUCCESS in 1h 20m 56s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/cf745c4d37bd440187881bc2c10968da : SUCCESS in 1h 22m 29s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/99b708616d3841dda875ded2ccdc7c33 : SUCCESS in 1h 22m 31s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/8bd74bc096ca4b419c7886c3460bfdad : SUCCESS in 1h 07m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b98fb30e618459b95f7c814c1e5daa1 : SUCCESS in 1h 54m 33s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6937178159364176890d09cdf5d24b7b : SUCCESS in 2h 13m 39s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/584a5cb9ed3e4feca37da22dd28740a0 : SUCCESS in 1h 50m 16s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a9072a72aa97407e8b73cce4ad35b0db : SUCCESS in 1h 24m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24241188f71b4a6eabf971a96b2cf03e : SUCCESS in 45m 42s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/1ffd8161ec634f0b942450223a527a93 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/2c4046b5f43944a889777cd340d63b42 : SUCCESS in 46m 14s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/790582d75906427394bc95c3cef13203 : SUCCESS in 1h 23m 47s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e8fa16ee747f488e8a87e3ebf5c70d03 : SUCCESS in 2h 00m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c60eb8ce59514788919581f956943d40 : SUCCESS in 1h 48m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/16331829c8284994b70cb94ecbcb8df9 : TIMED_OUT in 2h 34m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f522702da8924901b3e7a20ad037185d : SUCCESS in 41m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/41469149626d4631875587a321bbd542 : SUCCESS in 43m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/95504a792be2496a81e0189ec850af86 : SUCCESS in 46m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9607a9b02b8f4a1599146b429e5d9735 : SUCCESS in 47m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/25ec7f4dcd4e448f9bd90a48c1eada11 : SUCCESS in 42m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/54aa419715324e5ca33e53e0ff7f2a0e : SUCCESS in 41m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/2ed62660e19247a9b457edc8121e51a3 : SUCCESS in 42m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/6aa4a274158e421c824889caa13d478c : SUCCESS in 53m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1613eb39ab83486b914d15f3bf401f68 : SUCCESS in 1h 51m 12s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/c93c0dd31d4f498c9563a2306d4d8c3d : FAILURE in 1h 09m 16s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/801b201ca48d464eb5cee895764570ce : RETRY_LIMIT in 1m 20s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0f57228229195a394c0c7add9088be72c6f749c4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-19 19:33:50.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"81af248b8872a1d681cc2a9147a126368eea60c9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-21 12:23:09.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"882172e3c0db9e267911306f46c78249d9199a36","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-27 20:47:03.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":15},{"id":"903a321fa499e706e1a3fdd8e02a407bda90f565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 20:47:24.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"21cccfb82fb69fb1d356bf6bec646b9a5d4fec12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 23:12:47.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241271949d5142b3a75c4a32d4440819 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad162e480a814ffd9c52c4ba2bedcdcf : SUCCESS in 4m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8bea0e6f62564fa183b2dfdca3bc36f9 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/13b9b8dc900e48b99c9b78ed957ddda6 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/df9ef7f1e3a8478eb3b503e56ff6a834 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/410d313acc4248f6879fd264240327ad : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92d21c311a73427490fdfa97a1174d6f : FAILURE in 31m 19s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/860c9111467d44b39d3d62b64ec45188 : CANCELED\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5e7f91e918b942daa527889de9788bdf : CANCELED\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/36cce68e307646989f937ad36e7083bc : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4bdbd380b3434e928c460af7ce736e2c : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/6530ff5c7b414f98827c957c7bc4906e : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/e82ae69cf4fc4049bf41518386c8f0f5 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4e1afd9679a24fd4b9ae6dd0270114ee : CANCELED","accounts_in_message":[],"_revision_number":15},{"id":"6aab8512b2b1f9648a22b5124c42a1783a81645a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-28 13:24:52.000000000","message":"Patch Set 15:\n\nrecheck\nnova bug fixed","accounts_in_message":[],"_revision_number":15},{"id":"bfa80dbdf74148ca478b7950194b8ceb152f2faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 16:53:50.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b907537dd81a450d932df58a19b2c5ce : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a92ef4679e4096bff8ef27567c4104 : SUCCESS in 2m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/30378e551e1d463987bdbb6a860d07a7 : SUCCESS in 1h 23m 27s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/4b6e456f5a36467890e926893984c441 : SUCCESS in 1h 24m 25s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/ea4f4f6ae6b84549946a3c8a575a8977 : SUCCESS in 1h 26m 13s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/6902cd3eb67440ddbd77ed80322fcbd5 : SUCCESS in 1h 15m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2da7f64d48b94739a5017af67ec086d0 : SUCCESS in 2h 03m 12s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cb67b3f112e941da8ff9c41de6b9144e : SUCCESS in 1h 45m 41s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d78a3e1ff3d94fa293a71df78a07027f : FAILURE in 2h 25m 22s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fc9fe22778a84d2c887ed21bfea73f31 : SUCCESS in 1h 19m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d67ecb582564c228a39e7056ad52ee4 : SUCCESS in 41m 03s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/c8d96834f49a4f889844024db0467a18 : SUCCESS in 43m 18s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/71017ad431834fde9de5e2fc868fb679 : FAILURE in 34m 32s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/b2267f7113c14bf0a6f0ac88cb19982e : SUCCESS in 58m 49s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6c60ee3c8ca9413b85ccd2cfb12adce6 : FAILURE in 29m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/26dab0f6df5d4622b8f1b6815459337f : FAILURE in 29m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ca6e083ca5a4cbfaeb4b797cad41f92 : SUCCESS in 2h 08m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/267d66d9ae7f459680c2629c2d57cf05 : SUCCESS in 43m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e7ae6481dcef450f81980445b27bff48 : SUCCESS in 44m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9570e894e5024f5a9027b7b7498d6f67 : SUCCESS in 42m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a8493505082f4440bd38c411f32e0acb : SUCCESS in 47m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5fdab83f8c2a4a6f93df46ff1bfa4910 : SUCCESS in 42m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/a36e4355bb134fce9314a3241c2411cf : SUCCESS in 43m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/3c267e9af6e14d368874645ee4e5c55b : SUCCESS in 44m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/4c06c692b9af453e890cdc2c04e794fa : SUCCESS in 45m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5133998ed6314613afaa2d30afc8dfa8 : SUCCESS in 1h 35m 54s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/dace58549f154c009cc597c94e991275 : FAILURE in 1h 05m 56s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/bed628c18ce3404c8d9588005623befd : SUCCESS in 51m 16s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4ebf97722644c7785da2c46834d59068cebaddeb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-29 07:37:36.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"2f4b5a5c5934b700bdf8d7e9a9b7f4f117ee344e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 10:04:40.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10d4903f9c34ec5aa73c49276646cdc : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4722def11f144022989270e671b7128d : SUCCESS in 3m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5c7569e624d74d938997eef5eb5bdaee : SUCCESS in 1h 23m 44s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/f747a5ba572648518afc9eba68147b3e : SUCCESS in 1h 27m 31s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/83a06040b0084a0ba6fd023d406ad9ab : SUCCESS in 1h 27m 58s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/e72b0a97441541aea480a1c4ab5ed521 : SUCCESS in 1h 19m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27d432e46a524762884b813b60cea769 : FAILURE in 2h 21m 50s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9dd7daf132ae4c18acee143923af4e7f : FAILURE in 2h 01m 08s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/862b373d9ab74cb88430cb14ddc9235b : SUCCESS in 1h 56m 41s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/161f41baa2264447b01deadddaa968e5 : SUCCESS in 1h 16m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f401fd1e1a88426bb8bbed9c2f19e582 : SUCCESS in 46m 30s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/ea746bb81ba040ec830e923c0b3271f1 : SUCCESS in 41m 30s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/35bcf58424c340b0bea0729eabc87aca : SUCCESS in 44m 18s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/4e9966a69b9a4bdbb400473091f959b5 : SUCCESS in 41m 17s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7e941621c5e841c88c6fac7cc35f98a4 : SUCCESS in 1h 55m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9200fc0d39a4471a9cbfa5495d838946 : FAILURE in 36m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6352b7d8d952436797bd7d4490b9d3b5 : SUCCESS in 1h 38m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f26ebbdd2d34a6e8d1166e32930d3ec : FAILURE in 1h 15m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d1280fb88ebd4faa809fd8363a2e5775 : SUCCESS in 41m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ff1aa84b66314cb0a02503a8d7ba1ea2 : SUCCESS in 42m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6779a486e63d4d88b1ddca18f67a3a02 : SUCCESS in 44m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bc2d83e8f01e455aad4faa656543bc2b : SUCCESS in 38m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/fe8227d4afde498eaa9616127da361d0 : SUCCESS in 44m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/ad3dff9fee814f9a8fb266fef1affd98 : SUCCESS in 39m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/832a1a7a1ce84f68a0b36269c0b2adc4 : SUCCESS in 48m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ed48e57f0094043b645afe36ba49c81 : SUCCESS in 1h 46m 43s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e98623fbc17342f9a7282acef99af465 : FAILURE in 1h 09m 40s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/160075742447478182c7f37903e2cf07 : SUCCESS in 42m 07s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"28637f4eb98cad2345084acf6129587729c22605","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-29 14:03:58.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"ca79b1c9cab807eb000479c3aa7d869be8f1f84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 17:02:13.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4ff07949834679a5b3a59a4d5430a8 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a043099c0b3465594090efd4db85e9a : SUCCESS in 3m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/36711e86a6e3413589caf5bb51b5529c : SUCCESS in 1h 17m 34s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/fd31e53c64844f389e845464db3d1805 : SUCCESS in 1h 20m 23s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/bee8a0351fc84499a9cafa3f51c9fc96 : SUCCESS in 1h 21m 02s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/807d2e1d70844582b3f33e102155d06a : SUCCESS in 1h 12m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a6d0c76c2c54a9e8522aa1164e22aad : TIMED_OUT in 2h 33m 26s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/176dc328826443d28d2a0982ae0e212d : SUCCESS in 2h 07m 01s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/7fa2e4a9adf8489698d5fd7534769501 : TIMED_OUT in 2h 33m 46s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/622b3c9041ac4395b51be11902ab4bc0 : SUCCESS in 1h 23m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed165d86589643f5926c214fcec195b7 : SUCCESS in 54m 43s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/3c5606c0c3f64e21ba83fc22fe039405 : SUCCESS in 56m 12s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/284f781680554be481f9daeec4dea00a : SUCCESS in 1h 01m 44s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/ad22207217394409b7af0b841e4b1345 : SUCCESS in 55m 42s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/331e927f2b634d039dc9ce6d4db4a7a6 : SUCCESS in 1h 50m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9c7d04064bf544549de9ebe3d0f0ba59 : FAILURE in 32m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6379e7f064fd4fd6acf95c2a93dc388b : SUCCESS in 2h 14m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4f6bbe7ad924d2aa6ff7404da47b4d3 : SUCCESS in 45m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/88262a40ed774c0ea98ad28c9a62208f : SUCCESS in 46m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/201e1b53e4e64d4e8267c3d74619b548 : SUCCESS in 58m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/6e56fb771f114107bbda2c81b862c8c5 : SUCCESS in 35m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c4a8efa9c11d4550b7e88d85d29c3374 : SUCCESS in 1h 55m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ad4e10ed533a45d687f688afe0d7a35b : SUCCESS in 46m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/dee785c2f5f4486e86bb3acaa41e04d6 : SUCCESS in 36m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f21b876a80304a9ea0c4b91ff4f2cb06 : SUCCESS in 38m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/86ced17a6fe14b3c8a1ee93b6cd1ff69 : FAILURE in 1h 56m 19s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/af55eb20748043329c6d7cdb4d40308c : FAILURE in 1h 06m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c1ca9aa0563b4e5287c34f590fb95eaf : SUCCESS in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ea2eb159ed1acc4de4a8031c1ad2499c307edb60","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-29 21:31:54.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"7a101f4e9ad93b936612a79042a2937a2695b922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 23:51:28.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e013fbc155f43a5a7db51d27dbeb29d : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8b20e14e4a48dcad069f2cb6a56c49 : SUCCESS in 3m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/084db6fa212c4569999a2ff0b07b4fd4 : SUCCESS in 1h 19m 59s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/aeb2031cd2664b35b8844b80b39abbcf : SUCCESS in 1h 19m 13s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e8bce30e00604d72ab66ff925b1de27f : SUCCESS in 1h 20m 16s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/51d1b464154b4857880e0179ce4b2c73 : SUCCESS in 1h 01m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3760292bd7664c78a0f12ecc5df18e50 : SUCCESS in 1h 57m 29s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/bea88d8e964d466abd580e1677f3a20c : SUCCESS in 1h 56m 15s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c7800346eb2b487ab17d87bb363a8ebc : SUCCESS in 2h 15m 41s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/69fed08c05b64c7588ba5af11597e4a6 : SUCCESS in 1h 30m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71f06a1ce5c94791b618d0b07088d290 : SUCCESS in 40m 40s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/49b97917c5bc43aaa8427280f294dd60 : SUCCESS in 39m 19s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/282cd736097445f0a303a0e57b87ca6a : SUCCESS in 41m 03s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/7c334f69d19c45599d34256e72909d2e : SUCCESS in 42m 03s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/17678fad96a043a8b0aaf4c17c76d59e : SUCCESS in 1h 44m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2c2370347e6e498cb6e193f3bef99b7c : FAILURE in 30m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7b0b98a3858e4f799451f593d3652008 : FAILURE in 2h 05m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0ae0c1ca4814514b162716ce04d50c8 : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2a143eb15205457fbe87bbb589b92a56 : SUCCESS in 43m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/45742fec894f4a8794ab6429de729aac : SUCCESS in 41m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b420fb542761416db7ff98c17dd15269 : SUCCESS in 40m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bebbf0b1505a49d2a4e240b518193bad : SUCCESS in 36m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/c8fad4e36c66496eb1fa04239d997a5b : SUCCESS in 37m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/def3999881314a6b8081d1922523b9fe : SUCCESS in 44m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0de284b281a5439c92a5060dbf2a8eaa : SUCCESS in 38m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf34b16f3e114f30a4273db4d1491f67 : SUCCESS in 1h 57m 11s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/80c7b16a8fd64aaa9691925e2aefc2b3 : FAILURE in 49m 18s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e3b330cca46b48978169e8dc1af81a55 : SUCCESS in 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f6a6939e2b5eadf3a185be0c2f5a1bfdceb9a90f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 23:51:47.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"80fda431627cd6faa58e36d6e90a7f5d1b5850c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 04:12:09.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e962224941e1450688e2184771312df4 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e36dfd733a94f00845a385fc5e14c49 : SUCCESS in 3m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c8c972fcc384bd18861e128a6fb027b : SUCCESS in 1h 40m 08s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/807ef84f52304255b744170952405074 : SUCCESS in 1h 28m 12s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/8c5fb407db544bbc96407b18b46891df : SUCCESS in 1h 22m 53s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/86bec4b3b4b94b70bd7ed21ca92028ef : SUCCESS in 1h 21m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7bd777e965d4c4fa2ccec16f2c9d52e : SUCCESS in 1h 41m 20s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1274262f900543c3863bcc99a42fe3bb : SUCCESS in 1h 47m 26s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1ad653491c714348bb58682dbf36a55e : SUCCESS in 2h 04m 59s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/f4af1c89f81d41688586e533134640b8 : SUCCESS in 1h 18m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/394a6153c0274fd2955458d897a1c0fb : SUCCESS in 40m 14s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/222929e6f9a149b8b933e792e297aa06 : SUCCESS in 41m 16s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/2d5f0ab052424870a756257d3db80d2b : SUCCESS in 1h 05m 28s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/88e7ff184409413c858b5681f8c89922 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":15},{"id":"f5766d00d4bac16ce148851bf7340906ea72abe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 04:12:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"b5801ff2c59a543b3a1852222f7bd52592f4b34a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 04:18:16.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d857794fb543889a5e0b63ea39af84 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"3bc1f4decd285d455c928d19aa7cd4a81236576a","revisions":{"4e27bb42d76e7b14af8a383fc511f65806f7d6f7":{"kind":"REWORK","_number":1,"created":"2020-10-02 12:09:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/755777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/1"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 12:04:46.000000000","tz":120},"subject":"Fix healthmonitor scenario tests issue with immutable LBs","message":"Fix healthmonitor scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting the\nhealthmonitor, to ensure the next test uses a LB in a non-transitional\nstate.\n\nStory 2008219\nTask 41008\n\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4e27bb42d76e7b14af8a383fc511f65806f7d6f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4e27bb42d76e7b14af8a383fc511f65806f7d6f7"}]},"branch":"refs/heads/master"},"a61069e1bf4e8fad171a3e85588d698ce7c9ed71":{"kind":"REWORK","_number":2,"created":"2020-10-05 23:30:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/755777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/2"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-05 23:30:42.000000000","tz":-420},"subject":"Fix healthmonitor scenario tests issue with immutable LBs","message":"Fix healthmonitor scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting the\nhealthmonitor, to ensure the next test uses a LB in a non-transitional\nstate.\n\nStory 2008219\nTask 41008\n\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a61069e1bf4e8fad171a3e85588d698ce7c9ed71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a61069e1bf4e8fad171a3e85588d698ce7c9ed71"}]},"branch":"refs/heads/master"},"d0a99bc47eead68e9f28c8c471e744c06ee228cd":{"kind":"REWORK","_number":3,"created":"2020-10-06 23:44:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/755777/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/3"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-06 23:44:31.000000000","tz":-420},"subject":"Fix healthmonitor scenario tests issue with immutable LBs","message":"Fix healthmonitor scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting the\nhealthmonitor, to ensure the next test uses a LB in a non-transitional\nstate.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nStory 2008219\nTask 41008\n\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d0a99bc47eead68e9f28c8c471e744c06ee228cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d0a99bc47eead68e9f28c8c471e744c06ee228cd"}]},"branch":"refs/heads/master"},"05ab44d080576bd91f77bfd86330b5342956f011":{"kind":"REWORK","_number":4,"created":"2020-10-08 11:36:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/755777/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/4"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-08 11:36:09.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nStory 2008219\nTask 41008\n\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/05ab44d080576bd91f77bfd86330b5342956f011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/05ab44d080576bd91f77bfd86330b5342956f011"}]},"branch":"refs/heads/master"},"31be14a555908181a1d59ba8ec6fc47cb52d3ac4":{"kind":"REWORK","_number":5,"created":"2020-10-09 00:24:46.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/755777/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/5"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-09 00:24:45.000000000","tz":-420},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nStory 2008219\nTask 41008\n\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/31be14a555908181a1d59ba8ec6fc47cb52d3ac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/31be14a555908181a1d59ba8ec6fc47cb52d3ac4"}]},"branch":"refs/heads/master"},"17d27ee0f61a0b372145ee71327ca3f83662af3e":{"kind":"REWORK","_number":6,"created":"2020-10-09 09:19:31.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/755777/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/6"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-09 09:19:21.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nStory 2008219\nTask 41008\n\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/17d27ee0f61a0b372145ee71327ca3f83662af3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/17d27ee0f61a0b372145ee71327ca3f83662af3e"}]},"branch":"refs/heads/master"},"fd85abc398c9040d062336ae7deb781ef194b723":{"kind":"REWORK","_number":7,"created":"2020-10-09 12:59:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/7"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-09 12:56:28.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory 2008219\nTask 41008\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fd85abc398c9040d062336ae7deb781ef194b723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fd85abc398c9040d062336ae7deb781ef194b723"}]},"branch":"refs/heads/master"},"507c8665f00fdd9233fe73af8392a08e40e1661a":{"kind":"REWORK","_number":8,"created":"2020-10-09 13:59:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/8"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-09 13:59:30.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory 2008219\nTask 41008\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/507c8665f00fdd9233fe73af8392a08e40e1661a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/507c8665f00fdd9233fe73af8392a08e40e1661a"}]},"branch":"refs/heads/master"},"90196848dc196855d3608c69b244644ecc32a94d":{"kind":"REWORK","_number":9,"created":"2020-10-12 15:41:00.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/9"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-12 15:40:21.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory: 2008219\nTask: 41008\n\nDepends-On: https://review.opendev.org/#/c/755777/\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/90196848dc196855d3608c69b244644ecc32a94d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/90196848dc196855d3608c69b244644ecc32a94d"}]},"branch":"refs/heads/master"},"6741c495b422a9fdbaa552041ef21a304e7ad923":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-10-12 15:47:04.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/10"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-12 15:46:42.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory: 2008219\nTask: 41008\n\nDepends-On: https://review.opendev.org/#/c/757604/\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6741c495b422a9fdbaa552041ef21a304e7ad923"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6741c495b422a9fdbaa552041ef21a304e7ad923"}]},"branch":"refs/heads/master"},"f7dc82b7539d1e038cd94b4cd0f08b267317fbcc":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-10-13 11:45:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/11"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-13 11:44:35.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory: 2008219\nTask: 41008\n\nDepends-On: https://review.opendev.org/#/c/757604/\nDepends-On: https://review.opendev.org/#/c/757840/\nDepends-On: https://review.opendev.org/#/c/757841/\nDepends-On: https://review.opendev.org/#/c/757842/\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f7dc82b7539d1e038cd94b4cd0f08b267317fbcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f7dc82b7539d1e038cd94b4cd0f08b267317fbcc"}]},"branch":"refs/heads/master"},"a7df2036206176a5b1129e5091bc02b9b38acdb0":{"kind":"REWORK","_number":12,"created":"2020-10-13 18:34:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/12"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-13 13:48:36.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory: 2008219\nTask: 41008\n\nDepends-On: https://review.opendev.org/#/c/757604/\nDepends-On: https://review.opendev.org/#/c/757840/\nDepends-On: https://review.opendev.org/#/c/757841/\nDepends-On: https://review.opendev.org/#/c/757842/\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a7df2036206176a5b1129e5091bc02b9b38acdb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a7df2036206176a5b1129e5091bc02b9b38acdb0"}]},"branch":"refs/heads/master"},"db4a1723892e2dacc19d71955b7d48c72fbf5219":{"kind":"REWORK","_number":13,"created":"2020-10-14 14:10:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/13"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-14 14:09:26.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory: 2008219\nTask: 41008\n\nDepends-On: https://review.opendev.org/#/c/757604/\nDepends-On: https://review.opendev.org/#/c/757840/\nDepends-On: https://review.opendev.org/#/c/757841/\nDepends-On: https://review.opendev.org/#/c/757842/\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/db4a1723892e2dacc19d71955b7d48c72fbf5219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/db4a1723892e2dacc19d71955b7d48c72fbf5219"}]},"branch":"refs/heads/master"},"d9e081f667d03f5ffc35807c6954f56f6f0ecd1d":{"kind":"REWORK","_number":14,"created":"2020-10-15 08:53:30.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/14"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-15 08:52:41.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory: 2008219\nTask: 41008\n\nDepends-On: https://review.opendev.org/#/c/757604/\nDepends-On: https://review.opendev.org/#/c/757840/\nDepends-On: https://review.opendev.org/#/c/757841/\nDepends-On: https://review.opendev.org/#/c/757842/\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d9e081f667d03f5ffc35807c6954f56f6f0ecd1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d9e081f667d03f5ffc35807c6954f56f6f0ecd1d"}]},"branch":"refs/heads/master"},"3bc1f4decd285d455c928d19aa7cd4a81236576a":{"kind":"REWORK","_number":15,"created":"2020-10-19 08:30:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/77/755777/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/77/755777/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/77/755777/15"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-02 11:53:31.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-19 08:30:00.000000000","tz":120},"subject":"Fix scenario tests issue with immutable LBs","message":"Fix scenario tests issue with immutable LBs\n\nWait for the loadbalancer to become ACTIVE after deleting an\nhealthmonitor or a listener, to ensure the next test uses a LB in a\nnon-transitional state.\n\nThis also fixes the cidrs tests that were using the wrong waiter\ntimeouts.\n\nThis patch also moves individual deletion of children resources in class\ntear down to a more performant way by cascade deleting the load\nbalancer.\n\nStory: 2008219\nTask: 41008\n\nDepends-On: https://review.opendev.org/#/c/757604/\nDepends-On: https://review.opendev.org/#/c/757840/\nDepends-On: https://review.opendev.org/#/c/757841/\nDepends-On: https://review.opendev.org/#/c/757842/\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I6a4eed7269e4f502bd0fc8613cb4ec4da13890e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3bc1f4decd285d455c928d19aa7cd4a81236576a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3bc1f4decd285d455c928d19aa7cd4a81236576a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
