)]}'
{"id":"openstack%2Foctavia-tempest-plugin~759357","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Idb5d7fc23ccc72ba91548428e4befb726e46e05c","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Idb5d7fc23ccc72ba91548428e4befb726e46e05c","subject":"Fix request_interval uses","status":"NEW","created":"2020-10-23 07:18:59.000000000","updated":"2021-08-09 13:37:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a6725c25624c8239e78cfe2bd90742a7ddd5d171","_number":759357,"virtual_id_number":759357,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-09 13:37:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 12:36:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-23 15:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 22:24:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"51aa8e690bb6ed84908009b926a030c6c87223f7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-23 07:18:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d9a9adaa78b536e2d3a1be34e89f257798527f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 09:53:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/104012944b044f1cb06215b59d65c5b1 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4eef515d2d94147b55c3d6b92aae6e6 : SUCCESS in 5m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb8ec61846f445f9b6e519c637857335 : SUCCESS in 1h 37m 38s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/b30ef0d138454d02b17619ecb1d37187 : SUCCESS in 1h 43m 22s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e9bed9d164fa46f3b6926f97b9d69922 : SUCCESS in 1h 40m 04s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/2a94b9877ab74c2686948c12cffeff38 : SUCCESS in 1h 25m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b20843d856f4b77b64689701983fc43 : RETRY_LIMIT in 3m 34s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9fcf7ccdef2d4ee79fbf4cfa48696468 : SUCCESS in 2h 16m 58s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ebb05d0237c641478fdd3132abb9be01 : SUCCESS in 2h 18m 14s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d4c13fd3d3ae40f5b94eb4d608b8ee8a : SUCCESS in 1h 38m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29062ed4c7cb4835b552234b3434b43f : RETRY_LIMIT in 3m 45s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/d458d4a51dc04d06a37d0408b71ba785 : SUCCESS in 1h 08m 29s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/cff2517247c4471c8edd835f3d075c59 : SUCCESS in 38m 08s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/84580df7510247718abb2d94327b7be4 : SUCCESS in 42m 05s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b5850d748e5444c5ad55d137cd6a833c : FAILURE in 28m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/12131c79e8b245e69a5968528bb25ce6 : RETRY_LIMIT in 5m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d35c6e1f0c146c882e360b3851a9e9c : FAILURE in 42m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c93dc3c0c3c4150b7dd9e4acce2da89 : FAILURE in 35m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d972d494cb094f1bb13c0bda051110a8 : RETRY_LIMIT in 3m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/46650d9341904c55bb9cdbd33fa51976 : SUCCESS in 53m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/22554e0fd62e4d0d944fcdc5a375e155 : SUCCESS in 46m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/890107548263494f9d30346335b5b554 : FAILURE in 33m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/3c0896f213a742f7985ec96c3ef09458 : SUCCESS in 39m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/3d9bf275fad7445b9fb09fe6ce974cba : SUCCESS in 34m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/e2844ca4d08a47a7a7f39311ba5df121 : SUCCESS in 39m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6f22da00a49a4e318fd05a7b49d2c0a8 : FAILURE in 39m 10s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/9b3176dd9a6e41b790da425d3aecff5a : FAILURE in 1h 00m 07s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/47ae7802b9c746c7b51c42e12507ce7f : SUCCESS in 47m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8dc78371de0d5ab5fa7a34cb89b80da16d0c723","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-23 12:36:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a8ee778b932171cbffaa79f0e7675bfcb7b5a6f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-23 12:50:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cdb0553e56a086e1b9568bdf62c5ddf2562dbb7e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-23 12:51:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14ee17b7ef2629f7b60dc633a539ca21ae8ecd72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 15:12:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff1a53d98f94ccc844311cfb7fe0f58 : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b5dc973fe249508a1e25aead9322b6 : SUCCESS in 4m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcaf6a8b4ac7429cb66d85b3b6453723 : SUCCESS in 1h 31m 48s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/c527effd3e794e53aac7a532c9268cf3 : SUCCESS in 2h 12m 46s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/0b2aa3bfa1574991a0bf86f86636120b : SUCCESS in 1h 56m 21s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/9fa421d2adc54006a98e95307098ca62 : SUCCESS in 1h 40m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1946d3df7f7347b6a5e9dd293419d53b : FAILURE in 34m 46s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3c6fbb4f2a6743cead87ca70a2c13658 : SUCCESS in 1h 54m 34s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ca2d87210c964b0ea0c8545c3fc079a4 : SUCCESS in 1h 55m 08s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/a94854de7eaf45a695926affa56f48c3 : SUCCESS in 1h 36m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4ec5e93c4c9496795547012e3d6a9cf : FAILURE in 37m 29s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/405eb9e42d364d7995c49abe93430041 : SUCCESS in 44m 40s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3c99dfdf11e744a5b327c64c712cd13f : SUCCESS in 48m 22s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/45be0ee35cc94a6ea6df2fb0c8b25d3b : SUCCESS in 46m 46s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/60a5a89c80e24db2afbdc159348346d2 : FAILURE in 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1a2ef485a2b34017ab67ee08b4f5fadf : FAILURE in 34m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c916d44c6c94f5dbe3aa1c3b318ac3f : FAILURE in 37m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58d7fcd7ae5448139dd93f00c54eb34a : FAILURE in 37m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/269c3e3483f841c799591804773ebc2e : SUCCESS in 43m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5cf0ff61484644e3bc226fecfd29f330 : SUCCESS in 1h 12m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/46c192159c9c4d58ab48b6f98701db06 : SUCCESS in 36m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9a14011a96894a6aace2b5850af959bc : FAILURE in 29m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/84465f655ac14568b2347e5a71cc0e27 : SUCCESS in 33m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/7801e232be6a49529ec5a5ce59f31e3d : SUCCESS in 45m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/76821849cfa84d6faab9b9d84200ffae : SUCCESS in 44m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c5cc7b2382814272b988b7a750b9b69c : FAILURE in 34m 17s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b965c2f981054e969510d293802bceda : FAILURE in 1h 22m 05s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/8de679386ddc44dbb3e211fcf47c2efe : SUCCESS in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36f47a1b18b94e17f5b7e100e05a1ac8aa304197","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:24:38.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":2},{"id":"dc9d8c195df3618f51857afa017cc53dffc3dbe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-09 07:43:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8a318d68df9437161f20283e730128fd9e35b95b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 09:52:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad785c0f7ab94f2ebe7d4cee8c3ca367 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/322c3632467f4d6087992346bb444508 : SUCCESS in 3m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b13ae803c0b4882a1b4596975732eb3 : SUCCESS in 1h 46m 56s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/34fb428f5a1941129590cf441675d7ce : SUCCESS in 1h 28m 16s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/10a879de5cde4eb68c52b51ae5ad1540 : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/92d0cb9fa0df4ce9864a9daba7aa1f83 : SUCCESS in 1h 32m 01s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/4eccf2fb5ba848b0bab3b23a5270f07b : SUCCESS in 1h 01m 46s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/a8132673497a40389b6b458559e84cd9 : SUCCESS in 2h 01m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7af6b825fc3941e8b4fbcfc8504fec16 : FAILURE in 2h 01m 04s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/9b0b5342fc80420b828d7fbd800d8d0f : FAILURE in 1h 39m 28s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/b37a706a102d4b7a99c7946b1493aad2 : FAILURE in 1h 46m 49s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/52be82fda9494b83a05c1c2b87bcd7be : FAILURE in 1h 25m 00s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/af1574b911b74ccfb99fc202206217e2 : FAILURE in 2h 06m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e3f36b643dbc445ea691e16e6c5f11ef : SUCCESS in 42m 11s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/237a31e3d9e745d7a2088197cead8c5c : SUCCESS in 32m 30s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/9ecc69fa72c54f439a26e878e55f3040 : SUCCESS in 43m 47s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/64e0d701f5384287af0c4c2187e1a097 : SUCCESS in 35m 57s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/029df8e0b7684a45adcc4c468045ce6e : SUCCESS in 42m 47s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fc9152d44f384f26af569749becc0863 : FAILURE in 1h 54m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6ad3174af2cd414694c60f48a521b5f9 : FAILURE in 11m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/91d9a961b2714336ad514e1203cda0fb : FAILURE in 16m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b2f7c6f136c4b7eb648821fa8ebc28f : SUCCESS in 40m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/1f0c882eec384c91ab1d248ce64a82e7 : SUCCESS in 40m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/6518ed11e7cf48d6a6d41761e3a42440 : SUCCESS in 44m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2aab4dc35672487980c2e12703eb7196 : SUCCESS in 41m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5e6429f3047b46818e78367d3dd19f9c : SUCCESS in 41m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/596ee666b7554c11adc47f8c70c92488 : FAILURE in 49m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/e0566fb8ef27415d8c7e644aecb987cc : SUCCESS in 40m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/79c637ca662f4011a4eaf463df7373aa : SUCCESS in 35m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/dd25168359ff46ab8976f6a4e9a3ad34 : SUCCESS in 39m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/21a918de28024351a157c5c1266883b4 : SUCCESS in 35m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d1fc57c0a234ba19e5e1d9569c2eba6 : FAILURE in 2h 03m 09s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2ba38afb1a6448ddb693b2dbc023e7e7 : SUCCESS in 43m 04s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e7d19394f8c44bbf9d14ec0046ed846a : SUCCESS in 53m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2df37f1bcbd384a156de84b58911a876f141fdc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-09 11:36:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a6725c25624c8239e78cfe2bd90742a7ddd5d171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 13:37:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d95f49a2484a59afef0377a53b2f4c : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b69bec42874ade933f0661d90b92ab : SUCCESS in 3m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ffdcff7949d4cbea414e348243c3682 : SUCCESS in 1h 40m 55s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/bd0a2e08084845a993e4017bdef1cb57 : SUCCESS in 1h 25m 54s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/bdfea98d28b5493498d472c4e459415a : SUCCESS in 1h 36m 58s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/0996ade0881342d6a6bc6d405aab32fa : SUCCESS in 1h 33m 50s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/afc9dea52b9d4954bc5d074d5cd3fe33 : SUCCESS in 1h 01m 11s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/7dae6626efb34e258dcddb5d6662a079 : SUCCESS in 1h 56m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7051af93613d4277b1082a79d3ac7ba0 : SUCCESS in 1h 47m 06s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/b844161fa3f9472b9e0cfed37393628f : SUCCESS in 1h 43m 12s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/f00d9f1ecae24d50ac9e4aa1a183bf85 : SUCCESS in 1h 35m 09s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9a280b8f3c48468c88336a9046fee536 : SUCCESS in 1h 28m 56s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/152ee5a01cb642919a7885db8306b68e : SUCCESS in 1h 30m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/305fff393ad34bb08df441a63b6d1412 : SUCCESS in 36m 15s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/aa71447941e947c6b22f743e58e473c8 : SUCCESS in 33m 38s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/f5f370709d0b4767ab700abaae2fee62 : SUCCESS in 32m 48s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/2ce05dc06b23423dad38fe89209f0783 : SUCCESS in 35m 22s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/04329d5dd1c9486986ab3ff18676dc22 : SUCCESS in 32m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fad74e17c24b4e4e93f06f1d8441f35e : SUCCESS in 1h 34m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c43e7f33e0e84390994b426c18c4cb0d : FAILURE in 10m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/528bf542e7264fd187549a20f0b521bf : FAILURE in 13m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3c3e912a09049098124dc9c20367e19 : SUCCESS in 33m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/3e3e3312d4f64659b8c98feda905db48 : SUCCESS in 32m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5e71611adbbf41c797f14de3fd933183 : SUCCESS in 32m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9b2354d3173f454fa29a17056e2f468f : SUCCESS in 32m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/96e8d8dac2b34116985fd35f67f25278 : SUCCESS in 31m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ee373e3f2dda4b7faa176ce72fb339cb : FAILURE in 45m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/52aef22791fb4b94baa8ad4ceea80f54 : SUCCESS in 30m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/337fc2342ac3469e83923e20f5eb4fee : SUCCESS in 32m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/7b014b7e8c53446d88637e3846ec62a3 : SUCCESS in 30m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/549e4ca003874fb8969d61ebfbad1832 : SUCCESS in 30m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2f22aaec70844ed08753f3563b80352b : SUCCESS in 1h 19m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e8473b4f3a504b5d96733a1c9c6c3bbc : SUCCESS in 54m 27s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/d967deb04a234f7bbad41f813fb77ea2 : SUCCESS in 52m 13s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d61da47571e99198293088d0b34f80bc6cc8cf68","revisions":{"0e3aa53d4de6e1d76a2fac3489e93df99750d376":{"kind":"REWORK","_number":1,"created":"2020-10-23 07:18:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/57/759357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/57/759357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/1"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-20 13:55:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-23 07:18:26.000000000","tz":120},"subject":"Fix request_interval uses","message":"Fix request_interval uses\n\nThere\u0027s a mix between delay and request_interval parameters, sometimes\nboth are passed to a function, but only one is used, some other times,\nonly request_interval is passed but is also unused.\n\n- replace delay with request_interval\n- use request_interval parameter instead of hardcoded values\n- fix time.sleep placement, it should be called after a failing request\n- update some default parameters to sane values\n\nChange-Id: Idb5d7fc23ccc72ba91548428e4befb726e46e05c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0e3aa53d4de6e1d76a2fac3489e93df99750d376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0e3aa53d4de6e1d76a2fac3489e93df99750d376"}]},"branch":"refs/heads/master"},"eed5ffb1ad369f4cf0fe61ca909dea7f5a26007f":{"kind":"REWORK","_number":2,"created":"2020-10-23 12:50:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/57/759357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/57/759357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/2"}}},"commit":{"parents":[{"commit":"67249477d855d5be4470defcaa7a981fdb66ad49","subject":"Merge \"Add HTTP/2 tempest scenario tests for listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/67249477d855d5be4470defcaa7a981fdb66ad49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-20 13:55:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-23 12:50:26.000000000","tz":120},"subject":"Fix request_interval uses","message":"Fix request_interval uses\n\nThere\u0027s a mix between delay and request_interval parameters, sometimes\nboth are passed to a function, but only one is used, some other times,\nonly request_interval is passed but is also unused.\n\n- replace delay with request_interval\n- use request_interval parameter instead of hardcoded values\n- fix time.sleep placement, it should be called after a failing request\n- update some default parameters to sane values\n\nChange-Id: Idb5d7fc23ccc72ba91548428e4befb726e46e05c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/eed5ffb1ad369f4cf0fe61ca909dea7f5a26007f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/eed5ffb1ad369f4cf0fe61ca909dea7f5a26007f"}]},"branch":"refs/heads/master"},"fe2cd2b5d3adb6c4df59148f2471e110a0b14f55":{"kind":"REWORK","_number":3,"created":"2021-08-09 07:43:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/57/759357/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/57/759357/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/3"}}},"commit":{"parents":[{"commit":"058ceaf0e7df0a911a6b07eff74f068c2e54502f","subject":"Merge \"Fix expect_timeout_error in IPv6 scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/058ceaf0e7df0a911a6b07eff74f068c2e54502f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-20 13:55:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-09 07:28:22.000000000","tz":120},"subject":"Fix request_interval uses","message":"Fix request_interval uses\n\nThere\u0027s a mix between delay and request_interval parameters, sometimes\nboth are passed to a function, but only one is used, some other times,\nonly request_interval is passed but is also unused.\n\n- replace delay with request_interval\n- use request_interval parameter instead of hardcoded values\n- fix time.sleep placement, it should be called after a failing request\n- update some default parameters to sane values\n\nChange-Id: Idb5d7fc23ccc72ba91548428e4befb726e46e05c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fe2cd2b5d3adb6c4df59148f2471e110a0b14f55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fe2cd2b5d3adb6c4df59148f2471e110a0b14f55"}]},"branch":"refs/heads/master"},"d61da47571e99198293088d0b34f80bc6cc8cf68":{"kind":"REWORK","_number":4,"created":"2021-08-09 11:36:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/57/759357/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/57/759357/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/57/759357/4"}}},"commit":{"parents":[{"commit":"058ceaf0e7df0a911a6b07eff74f068c2e54502f","subject":"Merge \"Fix expect_timeout_error in IPv6 scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/058ceaf0e7df0a911a6b07eff74f068c2e54502f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-20 13:55:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-09 11:35:52.000000000","tz":120},"subject":"Fix request_interval uses","message":"Fix request_interval uses\n\nThere\u0027s a mix between delay and request_interval parameters, sometimes\nboth are passed to a function, but only one is used, some other times,\nonly request_interval is passed but is also unused.\n\n- replace delay with request_interval\n- use request_interval parameter instead of hardcoded values\n- fix time.sleep placement, it should be called after a failing request\n- update some default parameters to sane values\n\nChange-Id: Idb5d7fc23ccc72ba91548428e4befb726e46e05c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d61da47571e99198293088d0b34f80bc6cc8cf68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d61da47571e99198293088d0b34f80bc6cc8cf68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
