)]}'
{"id":"openstack%2Foctavia-tempest-plugin~771251","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I4f6c16a164e08996581ba60a48ba7c480b5c4902","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"change-771251-19","hashtags":[],"change_id":"I4f6c16a164e08996581ba60a48ba7c480b5c4902","subject":"Modified test_traffic_ops for failover testing","status":"NEW","created":"2021-01-18 16:40:01.000000000","updated":"2021-05-10 18:12:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"e3b7c85c9f1386a256979212fbe533a0fabacd37","_number":771251,"virtual_id_number":771251,"owner":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-10 18:12:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-05-10 15:55:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 20:57:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 21:12:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-04 15:29:51.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-06 14:41:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-05-09 12:04:18.000000000","updated_by":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"REVIEWER"}],"messages":[{"id":"49cc699ee4edfa572bba47fce4dd1cd88e5004bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-18 16:40:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eab0d370173aac17ba86b2061ff4d4ba0b544382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-18 16:45:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c943bf6c808c9ee932c348ed57e20f2b7c9e97f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 20:57:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95acc2302bcf446bb9e9daa23f2226e0 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5021629fb734a2cbdd888a7829f5185 : FAILURE in 5m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c7d6c484ca948aa9f7c450a960f1b25 : FAILURE in 2h 00m 49s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/634f07571e9249b583471ee09ab715aa : FAILURE in 2h 12m 27s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/66dad0c27a974b3182c7f74efa853c6d : FAILURE in 1h 35m 29s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/cf65d1f099624a9facf637d1522025ff : FAILURE in 1h 09m 03s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fb2c8b6bc60746eaa5c6e1582d30ddc8 : FAILURE in 1h 46m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b2dca6ca4bf341eeb22f5834ffabcff9 : FAILURE in 29m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ce71c116a1743678aea1791bb7f5fa8 : FAILURE in 2h 07m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e08a8b38192c4aadaece39fdb417f566 : SUCCESS in 44m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/85e1355e1ab448b087cb33302a17c5c0 : SUCCESS in 38m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c2d21457989744ee9c66e2dd95b8efa6 : SUCCESS in 42m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/40d8a43bfbad4900b095e866c7b52e43 : SUCCESS in 50m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/039779da7075410b91f608c6966387ca : SUCCESS in 55m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/db1e3eb7d6914f1dbd44368824974108 : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/93183525ca3c4dcb9817ba9478b40d77 : SUCCESS in 32m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/ac05373ab7bc4f818a1ba82ba89a03b2 : SUCCESS in 42m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f36e40f0b1544ae58dd62f59e00434f3 : FAILURE in 1h 57m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia_tempest_plugin/tests/scenario/v2/test_pool.py","accounts_in_message":[],"_revision_number":2},{"id":"f81e672f92e92c908b0719a1328ef11567bd68e5","tag":"autogenerated:gerrit:setPsDescription","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-19 08:12:43.000000000","message":"Description \"Edit commit message\" removed from patch set 2","accounts_in_message":[],"_revision_number":2},{"id":"76cc129111d5863e93b44467f162618e56fec715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-19 13:39:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b5d3562c05446f7af61b558fce717c595d6ab6ba","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-19 13:40:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"64a4af19555b76072af959e7bc9b6e8f7d2e1eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 16:15:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cdee7bac0734d619a0275766c6fd187 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f520f69a5e6a49b1a92f8e70345d3ea5 : SUCCESS in 3m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4523beaa72e843adab6e8dd9db82d692 : FAILURE in 2h 31m 16s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/dfa15b0d463f4327a0507c30b98c8faa : FAILURE in 2h 12m 41s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/14361b3e58bf43fcaf2ac84344cfac75 : FAILURE in 1h 52m 00s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/cbfe4da6253c4156937ec0186299b50f : FAILURE in 1h 36m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cb1fd09b94994c62b2d04d4990e8282b : FAILURE in 1h 53m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a7161595253348048263e7e719374a0a : FAILURE in 31m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c1bf9f93751419c9be4a1d6af3f3699 : FAILURE in 2h 00m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7364c075178443409ef8c1f5b9db4172 : SUCCESS in 41m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/513cfc4a8b1c4d169a4ebdd3c7d67dbf : SUCCESS in 43m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/950eeca689a449dfb10cee79cbe2f144 : SUCCESS in 42m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/06437de5f75f4a7bbeaaa321197bf4aa : SUCCESS in 43m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a3720388078c4655b44966a8c6659cce : SUCCESS in 42m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/e676800e9e2e4de4910e9f9b170a2d4b : SUCCESS in 38m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/461705ca17d648cea5752650087c8d51 : SUCCESS in 42m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/73f5bb3c179040b3855eb7f346606490 : SUCCESS in 39m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/244fb986a96a40d4ab638732798e88cf : FAILURE in 1h 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d9dd236e4532a43f0cf5a8d170a3c6179a42c29","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-20 21:12:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f063a3c0776effbaf9ca5bd8c27d4ff24115b1fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-24 10:19:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"97a7408341754bd6cf67b8054b892503cee5f98b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 12:17:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96366fb18e4444c9820e7dd92d166913 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b27e9178cb4366be7cf1b34dec454f : SUCCESS in 3m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d5773d8f78149bcb36a59055506343f : FAILURE in 1h 42m 31s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/de27cd90f4bb42959a2e34f43d26bc25 : FAILURE in 31m 34s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0effa476ae85429fbe4c19a8b07d2e35 : FAILURE in 27m 32s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d984f5efb1eb44eba266007758fdc698 : FAILURE in 28m 07s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4dfe9f3f59bf4536b2c339f2735174d7 : FAILURE in 1h 29m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/df0fbb9308d74e319cbf6c0be34bc9a1 : FAILURE in 33m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f632dac275d41b5bb11a728d7f257a3 : FAILURE in 1h 56m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6d4145e5100491998dc7688f9e50721 : SUCCESS in 48m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ff8a166309464efe80a0a8b46e45c439 : FAILURE in 28m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b69bf3150744444ab194f1a366f08af8 : FAILURE in 27m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/b2f7dee07a614a43a902649d63c79a75 : FAILURE in 27m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/23b5f03bc1554e6290d68e3534ba0d92 : SUCCESS in 37m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/69b6ea0442a748ac9c313e79fde6d022 : FAILURE in 44m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d0a40f0bf1c3468db64ab4271a748df0 : FAILURE in 41m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/60b1b8ac8d8d4915940c8f2a690d0e28 : FAILURE in 23m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0781abf7fa654a3483e541108fe669cc : FAILURE in 1h 56m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08fd83d729091b5039021d393e0aed8b6793e83b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-24 14:24:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4277bb40c4f74f6b8a9ee23c35ab07bafd19cc56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 16:27:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad84ead87af046c3bf9218f49e1decdb : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dfe0c3e8aa94924be54e4caf370ac0e : SUCCESS in 2m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b762355ecbe41cbb1e65e54afb31423 : FAILURE in 1h 44m 09s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a1236169b1e84ff288fe3b81911232a1 : FAILURE in 42m 52s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/dc24f2cd06d942b38e1255535a4301ba : FAILURE in 8m 49s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ef937ff105294fa1ad63b1f8a936b665 : FAILURE in 6m 14s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/16b17293b94541b8933f90ff4e79105f : FAILURE in 2h 00m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b8a89853137942b887ce331bd837e40b : FAILURE in 46m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e587456dcd0744b6b8dc1ab165255f5d : FAILURE in 1h 54m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d23b227e5e1e47bb9683c21f4a0eb354 : SUCCESS in 37m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f575ad150859467cbc89c54a4253a9a2 : FAILURE in 28m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ba781fb01a3346c280d0882abad7da60 : FAILURE in 7m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/e5a839ae56194cffbcf718c7961c0578 : FAILURE in 7m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/12d6708bd62b4f50aee1f040b4cbc32e : SUCCESS in 37m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/00fea7e99f2147e488b310020072ead2 : FAILURE in 45m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/95864474e64f4e10ac4948aa8fbd3299 : FAILURE in 7m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/6fb641a2340d4f6cae27a47e67269b06 : FAILURE in 6m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de83a00aa8394fd5a7e2fdd065ac4c66 : FAILURE in 1h 44m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bc2fe799101eb245a361c20672e0d9c64f91c640","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-25 07:47:28.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"239328c64385702e032bfc7662722a3d4ca1b731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-25 07:51:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e5736351e6ea768b4a05a7e766319222ecce431b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:04:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0625c16d97294e0ea102769734d59177 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db93f0f7ca244ae8af7c10bff3af02e : SUCCESS in 3m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b981b4ce50841a9a0e46aa6176f992e : FAILURE in 1h 51m 36s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/69b1cc5d2a18473a9c80ccf2125a4aca : FAILURE in 38m 45s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d4b697b6b616462e9092321ba7445b74 : FAILURE in 7m 52s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/465228f996f64ad7ae155bdf4ab94374 : FAILURE in 10m 45s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/01d528fb08ba44baa54f331a28a79d7d : FAILURE in 1h 50m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d282720dbb75433da0c72e1c6fe0b8d1 : FAILURE in 28m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f7106ca62fd480086431bbc939fbdf4 : FAILURE in 1h 45m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9114f611e43b407c9f7417aad497fa05 : SUCCESS in 43m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/399efa7b70ae4c868dbb6a7ca2592ca8 : FAILURE in 29m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6b351628c2c4452685755da9378b2ba5 : FAILURE in 10m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/11915a872c8b484a8e015b6e809aaa61 : FAILURE in 8m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e2c262cc96d0466c811637ba718d4c71 : SUCCESS in 37m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/30c3fe257ade4f7e830bbd92182fe474 : FAILURE in 50m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/e67f82db518f4a1b9bed2b992ad9b61b : FAILURE in 8m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/020afabf06984548b5b74ba518377c22 : FAILURE in 8m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43efea504b3f4956aaec7f0546a0c7be : FAILURE in 2h 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8ef9319adc7936ec03b3d56749279fca73969905","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-26 08:53:38.000000000","message":"Abandoned\n\nerror","accounts_in_message":[],"_revision_number":6},{"id":"3d2d1f9c28fd183845ed2d73960e05de28e2f44c","tag":"autogenerated:gerrit:restore","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-26 09:29:21.000000000","message":"Restored","accounts_in_message":[],"_revision_number":6},{"id":"e7999beb735cfe3120203ab16456555bf36f74ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-26 09:29:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1902e082faf3d076caf5c76e891a84c91fc8fff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 12:02:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782834921e934c2787ac14c9e884f6c2 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77d94e96f83490d9906ba018706e91d : SUCCESS in 3m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ce5c076131e4bab91f5e2886041e7ae : FAILURE in 2h 07m 04s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7009e36e200a46219a596549dc430a3c : FAILURE in 52m 21s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/09e03ac4c1184acc8c834cf3e833bd3d : FAILURE in 9m 08s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fe9eaedb7b22436f8fb6cbb36ded6b3b : FAILURE in 10m 05s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3bf02049433f421a9a42cb3f7253293e : FAILURE in 1h 53m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/52ca3d03819f441b8dcf484aa1d87311 : FAILURE in 38m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edcf6cf64fdc4455a313b0874ca32480 : FAILURE in 2h 25m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c70c26c68d884c8bbcf709ce8badc29e : SUCCESS in 38m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/353a1f95203342ba94770326ff110da8 : FAILURE in 38m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/dcf7598aea004afca34191bf7e9963ef : FAILURE in 7m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/0d5867fb1b8144e4995400ee0c5d3375 : FAILURE in 6m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1ecaa67b41de405b8eb9dfbb9ad7c6a3 : SUCCESS in 55m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/259840fb6db14e06b2bf7a412b74cfdd : FAILURE in 54m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/2fd583c530834b5882e0a290d0e41d9a : FAILURE in 11m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/0e1c6f602b394031a92467c570357556 : FAILURE in 7m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/404d63eb43fa40b0b99e1b66a757520c : FAILURE in 2h 03m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f86309e80c69fc91402da7378fa721aa9c6ff18e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-26 13:51:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"49836ebc5505d1de8c77c58bd18fef9f5a89bfc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:22:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c464f15e5cc4b9589d0ddd70f796e63 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263b856d612a45379eb2701110b0424e : SUCCESS in 4m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8444933bd7d04485956755ea2bfa2f78 : FAILURE in 2h 27m 00s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6ed7a2b1fdc24128b23485646543d812 : FAILURE in 36m 58s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b874447c2310425a85639176c38f8518 : FAILURE in 11m 04s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ef155ad0e71a4883be710cc080cef318 : FAILURE in 11m 03s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/683c1fb2506542e98770595d97dbef49 : FAILURE in 2h 17m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d60956ee65e6480da0cb8005ddebb4ca : FAILURE in 32m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f11cb640750478dadbb81b1e6603aa4 : FAILURE in 2h 01m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d2810256d394d8091a8d6b029d90d77 : SUCCESS in 52m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cd10541af54f4b50864b075b60687813 : FAILURE in 34m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c286227b784a4b299d4c280a81546329 : FAILURE in 7m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/611d747583574e8d8ae5a3b0b640a47f : FAILURE in 10m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/abb650bf58b246f6a187c94efdfae15b : SUCCESS in 58m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/f9f399727327409f9e87360aad88366b : FAILURE in 46m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/946143f5a39b47cbb7d1fea779efe5b3 : FAILURE in 11m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/9aa5d574d2d2454c8048df95c2bda834 : FAILURE in 8m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7f2bb55314f84cb3857f5b212723ec6b : FAILURE in 2h 04m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"609326882347541ed0fd4375956f626350acabbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-26 16:43:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"455507fdcd6cca18e596a82f6a544df729a0e564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:39:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/646b215ed165496a9d2d9c1ee841c371 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4854851f31f44e69ab7e23fa20242055 : SUCCESS in 3m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/517df421601a4f81930b84dc617f91a7 : FAILURE in 1h 50m 27s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1f8a99a0836d4f1c89902f6fdf515697 : FAILURE in 34m 35s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3297de76f288405f8835c554d5378e0e : FAILURE in 9m 32s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/eeae31b7947f4521a0cd4c0a16d8bf87 : FAILURE in 7m 20s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9dda627966ed434ab1afd94936a25b49 : FAILURE in 1h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/636b7a9eaa384d13aced41f16e084eef : FAILURE in 36m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d363b2ef1670425daccf87a0721ed5c9 : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e81d8a5c1e84c249d327f1c991e0ee5 : SUCCESS in 47m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2cbb2579f3494f9db347f96e08851808 : FAILURE in 40m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/7d06073ffe6147c0880c66940b83f5e9 : FAILURE in 8m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/314b8b7726ee4c67a27e2fbb01b419c0 : FAILURE in 14m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5ef7f3b490414fa988881d11b63d389b : SUCCESS in 49m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/bcea5fcc3f0a46b09bd5144c291b836e : FAILURE in 47m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d757575a0745432db9f1ab8aa18a7610 : FAILURE in 11m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/439eafb92ba7403097177f6bfe8cab66 : FAILURE in 10m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3940b598fa846079ba564b520272eca : FAILURE in 2h 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"07be066bf431cc7db0342257856d8c0cc0a7e2a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-01-27 14:16:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7a749ea05cc80e96391fbc0c57e0527624e97803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 16:49:18.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6eefa2a2db4b27aea37f851ce2dfd2 : SUCCESS in 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b38ec54389934ba997e1cd6c00df6168 : SUCCESS in 3m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08097bc622bc4df7b5ba40ff41825cf1 : FAILURE in 2h 04m 25s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/edbeee5a2e174ad18f90207b0329690e : FAILURE in 36m 24s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5d8dfa4f561a4581a2cf97b69995c05e : FAILURE in 41m 29s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/518ea67da09943b1a3ff5f1208afbb67 : FAILURE in 32m 01s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aefc9e706b774699b4ebc22c6a45ffd9 : FAILURE in 1h 54m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d10da9b890af449592a17b422341fb15 : FAILURE in 30m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b6240314acd74319978e8ad6118e5f10 : FAILURE in 2h 04m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94f6b7569db5456c98b2250294b833c6 : SUCCESS in 50m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d4ceb6a4bd3b40af94c78275dcd18031 : FAILURE in 35m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9814c16dad63443e823fbe61f358b36a : FAILURE in 27m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/bf25deb304184ed6a74b539b3fa04b70 : FAILURE in 29m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1f259903b69b48b9bc83fce239957320 : SUCCESS in 38m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2c5ebfae48924d1d8c87157146115e54 : FAILURE in 49m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/8c399d6440f14d66aef11b3910a32eb5 : FAILURE in 32m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/7a86e79cbc8147f493f04b5780b53ae1 : FAILURE in 25m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9f2919bf9fd2479f85a9f9b57b285a30 : FAILURE in 1h 55m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"10aac428fdd3da2bfc3a11ba1110dd2bed084920","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-04 15:29:51.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"32a142c455290dcd019636791210a8b7aa3c306d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-02-10 09:26:26.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"7d901e08134c353a58ec85880cf31abbfb043be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 12:16:19.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1946afb02b714b02b7a00d656baa2d17 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c82bc14f43487985ee5e5fb67e9bcc : SUCCESS in 2m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c302baf8a2048b0ab61d3dd6eddaea6 : FAILURE in 1h 44m 32s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/764c3526633b4e368a24c46db6995436 : FAILURE in 1h 59m 55s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/acfefb9aef7d4509bb1e098ce41421c0 : FAILURE in 31m 26s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/21d49d79348d4662bc9706237dfc1dbd : FAILURE in 24m 33s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eaac811da3ce47fb988e75c38f7d5f09 : FAILURE in 1h 45m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cd92846a458145f5b041f5dd6b1b47be : FAILURE in 28m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/55988ea9c4f14e95a26f734691c746e3 : FAILURE in 2h 12m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0352d0a3fd454fd8a65a161b4f744d8f : SUCCESS in 40m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/bcde6c80c68043f4aa80bd90ed274332 : SUCCESS in 40m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0c4dca1a0883412c90a4d4f3c124911a : FAILURE in 22m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9077ceb4af594346a225ec3f12a1935b : FAILURE in 31m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/17769c70554b4a9ba5c482cc9f10669f : SUCCESS in 41m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/673c69e45bc447fb94144a092ab0431d : SUCCESS in 47m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/596235e37301467e9198f34192e5c55f : FAILURE in 24m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/05f489a00d4c4892b009be20d7a56a6a : FAILURE in 25m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e80d9db77a9242b78552c92842158e62 : FAILURE in 1h 55m 21s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7d0af47005a0302e03689dae6f10b16ee2a124c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-02-10 14:25:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f68ff845c4007958bcae82d0745faf6bbb44c4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:44:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e69879ba96a40999b5ba6076a9989a7 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6460f26440492882f99a6a584aebe2 : SUCCESS in 3m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89bdd92030594a48856b41547807ff3f : FAILURE in 2h 15m 19s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ccd05d6da1c047819b553cdd75b862ae : FAILURE in 2h 09m 49s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f27f53a988b841b5885c344283076560 : FAILURE in 25m 18s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/9fd0a38b601c4b79a9bc90ad1af56f7c : FAILURE in 29m 23s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c9198614400f40fd8d42694723c5787e : FAILURE in 1h 48m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/00fb3b53e5164f24a580e7537715a277 : FAILURE in 33m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4715c3e7be0f41568cb4002148e5839a : FAILURE in 2h 10m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92301bf8f00d4e5ca05688f2f6bd1414 : SUCCESS in 43m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5da08de720a845bcb2ef3c80a3788c28 : SUCCESS in 45m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a1f11f87bd6c40cb8c2e72a0e153045b : FAILURE in 25m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/58274e394c424d009b16243767148b49 : FAILURE in 29m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/97c621e998924289872c3985459dd05b : SUCCESS in 43m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/d4ba0683709c4435a9d75c9a49559cb5 : SUCCESS in 38m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b229136fdff141549df5add24911bb78 : FAILURE in 28m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/eeefdc4185674f05b9ac125da78aa50a : FAILURE in 29m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dce27b24ead64bf8a7c72b4c18c79b2f : FAILURE in 1h 51m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d9be0f5be4b6bf6da96a4ac4809c2a38d4d8d473","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-03-10 09:49:54.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"02a586c3a9bac8c836baa8aa6b377970253a6223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:56:05.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/707430bb865b4e4c97503182437fb911 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7608c6e4fdc5479fa0fa142aeb09f54f : SUCCESS in 6m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e17bc8bfcfa4098b3976ada4a50f3c0 : FAILURE in 1h 19m 49s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2462fd1db3fc4742803c2c4202f16b64 : FAILURE in 1h 46m 50s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/51479487927f4b55a1e89e60ce0e4d03 : FAILURE in 2h 02m 15s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0df33aef4f8b43bb974f732d668bec36 : FAILURE in 1h 45m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b2ff154c61c646818a1df348738e15d7 : FAILURE in 37m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d581bc49c9494c32927c6d81594595c9 : FAILURE in 1h 59m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a86dc1c4bd14950a999a875db50cc1e : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d771bf5fd44c4d6da7c83598af1a443e : SUCCESS in 57m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2b1be61728714e66b38cd96233f7c43f : SUCCESS in 44m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/df485922a7ed43a0adc111f5f6a10439 : SUCCESS in 48m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ec1a0c053ebb4c039c6db855186731fb : SUCCESS in 39m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/f674154fded24c8a830c19c626bb6209 : SUCCESS in 38m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/92d37db8343e4d68a1d464351b81c719 : FAILURE in 1h 43m 48s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ddbaaab03f9b297d4a2dd6eab274de7526a57179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-04-07 09:07:58.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"db284afa57b4feebc2da996a3f465e8ab73b1a65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 11:15:40.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0f28d7f6cb4911aeea123454cc09a4 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74d8b468b080404d85e3228716018a97 : SUCCESS in 3m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f3743e530364d219a6876c4eb98cb6f : FAILURE in 1h 29m 22s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/ca1557d3915e4521b22915c24a1a596f : FAILURE in 1h 28m 52s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0ff84d8634e642e8bea576095ea671d6 : FAILURE in 1h 43m 36s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/594382097cfd4258b322e1bf2efc8445 : FAILURE in 1h 55m 46s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a1f81883233f45ee88dd8fe8b69c691d : FAILURE in 1h 58m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4a92b037a4be4f919a3305f63410c63e : FAILURE in 32m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/67c8ae7b70894dc7920496e00f0f869d : FAILURE in 57m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba9c09cd63c74a01ad8e7dac53d00857 : FAILURE in 22m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/9925419489894f51bb05d64d32b9fa7d : FAILURE in 32m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c62e4777b9ce4ad8bfa6df10e746ab8f : SUCCESS in 48m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9c53cf0ffffb4acca356e69cc854f5a6 : SUCCESS in 46m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2f16adee8e57460f803b73704a267088 : SUCCESS in 35m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/01de9797773a4a4bb2eac6451f12df8e : SUCCESS in 46m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/74ddb10063614d7c92406a11ff7daec1 : SUCCESS in 41m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/0774ce2fe1bb4edfaa18c1c5b90fda4d : SUCCESS in 38m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6f0bca4a47fc40e196cc8e3e5699cd9d : FAILURE in 2h 01m 26s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/150311085ac44b7c826c388dc78f86ca : SUCCESS in 46m 46s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4bc0052e75db4ad19ff1a125d45f561d : SUCCESS in 58m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1a4d7a6a73ec717fa08fbee7be7e6c59e61ebb8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-02 09:03:22.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"7a31d15a12869658499b38cf413ea31f9cd1f9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-02 10:54:08.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48592539bd946dabc6e297700874588 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c8f1cd8b93489384e0bb7c933de8e4 : SUCCESS in 3m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38157c42102d495c8198f7e6c6d7dc9c : FAILURE in 1h 40m 32s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/407455ff99714fa4af2eb7197dc6d98d : FAILURE in 1h 48m 44s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e79737482f83473b95593e1d215edfea : FAILURE in 1h 45m 27s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/305f8e6a79f54d3cbd5784bb5decd48b : FAILURE in 1h 24m 18s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/742d878f7e954b6d946e3b47e8714255 : FAILURE in 1h 36m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3c35eff096a2482887db1a5e7a0993de : FAILURE in 26m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/52c30901d2df410492a460a2aefdbb1a : FAILURE in 45m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85af25e08fa145feb927d272934c8eed : SUCCESS in 34m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/60821cb2896e4507a00d8494c1a6b052 : SUCCESS in 38m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/de498381d15e472f8cbf156cbc027e6f : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ba0321c038c24fb68f9f28dc5455054d : SUCCESS in 48m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6829ec044cf74f16a34c35f65f7a474d : SUCCESS in 33m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/7bba3a9e9b084589b1ab71e83515d132 : SUCCESS in 35m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/dbb6931faaaf4b3e9a32e2e4e3cca8f5 : SUCCESS in 38m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/c2eebfb9136c4ce8b0924ea32d96cbc6 : SUCCESS in 32m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d772c022ada14c6489e25dcf542beb1d : FAILURE in 1h 17m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/c310271ad53f497cbba1e2a638412e15 : SUCCESS in 35m 14s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/0674c1830b844abd9392c454819f1557 : FAILURE in 44m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"aac57e472fbe9daf5da51d09b2406982550e7df2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-06 08:15:11.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"064ba8df9ab43c227cefc5ad93e02756227ac249","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-06 08:16:42.000000000","message":"Patch Set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"bb86d8d6c9a9cee7414d5d829ac7de000397c7e4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-06 08:25:50.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"abc3716604eed629cadfa1ba84e7a278f006b633","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-06 08:28:45.000000000","message":"Patch Set 17:\n\nI think you better run the linters before submitting patches, you would be able to catch those errors earlier","accounts_in_message":[],"_revision_number":17},{"id":"b1f58c9b5061852830d3b06ceeffe34612ff28ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 09:45:14.000000000","message":"Patch Set 17: Verified-1\n\n(55 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb495fb411440f6ae9a147d68e24909 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16755067814a4db6b722e78091cf8e0b : FAILURE in 4m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f539ebf6cf284318bc60191c04b97dd0 : FAILURE in 1h 25m 51s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/0c0d3421290747b1a217e8ca073932ca : FAILURE in 1h 11m 19s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a6bcb06366d24cb6be63aba07b70b287 : FAILURE in 1h 08m 25s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f339f6768e084e76bd091f4d9978a88b : FAILURE in 1h 12m 20s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9d9e7a4cd0b147c3a28af4d05fc4484c : FAILURE in 1h 27m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/245be277fde24b7ab058b0d5fdda1e6b : FAILURE in 26m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd8ed196403741b097b5eeb99d8a9330 : FAILURE in 44m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61e6e3ed0e2b450a906013ba2d17f7c7 : SUCCESS in 38m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/3295561fc7764329b33c6cfc6b39ddbc : SUCCESS in 46m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cb07d4f333424a45a5c5f7910b0480ce : SUCCESS in 45m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bcb103d8ea8b4d5497d6f4415b8378b9 : SUCCESS in 56m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/96d158f44a634cda8d2be2e60bd273dd : FAILURE in 48m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/497389b3358644eca6a6cae44efcf759 : SUCCESS in 39m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/29612aadf8944923be1bcbd3f8e76bfb : SUCCESS in 39m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/16b94b4954ef48c4b59b099f8b0f46af : SUCCESS in 42m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1fa177941a524582811a6aa9e989fe4c : FAILURE in 1h 13m 38s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2be09d51803e4b4f97505c5c728784cd : SUCCESS in 35m 26s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/751b81c7ce524d1c85e7901fdd713052 : FAILURE in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e257a0e9486a8a64404af3a63444bb99a1c9f252","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-06 14:41:05.000000000","message":"Patch Set 17: Code-Review-1\n\nPlease test these patches downstream before pushing them to zuul with git review.\nThis is excessively using gate testing resources.","accounts_in_message":[],"_revision_number":17},{"id":"4e96d240dd3a1263213fd6c342814281a5cfe93d","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-09 12:04:18.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"9ce1c447f569bb5e08f496211c9937bb010abbc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-10 09:31:59.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"3315601fb4352a07231caaf91559a77510f6827e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-10 09:34:20.000000000","message":"Patch Set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"7a54ee0c99007f9b6126986401e882c6c9eda162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 11:50:04.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba39e3cb2f443f9bc43788dceb5ce86 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f50c441559424fe78e37eefc3391fe35 : SUCCESS in 3m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55a7836e0e2b448f9698284122e7dd5b : FAILURE in 1h 23m 09s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/42385313f5074ffc9354cdc71436f028 : FAILURE in 2h 12m 08s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/882cfc571a2842b39555c5766b5851c8 : FAILURE in 1h 40m 32s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/76e1d5fcbe6245c1867f2d1489e42812 : FAILURE in 1h 44m 24s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f92619fc010c4d24997e5904c74866b4 : FAILURE in 1h 27m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9bd28a4816764c6783905396dd409913 : FAILURE in 26m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d42f03dbf9ed4cb5a867ce4955d1cba6 : FAILURE in 38m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2539e630ed24d139652638fbba6bb80 : SUCCESS in 45m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/aea8f051c04d4275b248974695c10e0d : SUCCESS in 33m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a91b6e11c9084dd5b576e929e52dc75a : SUCCESS in 33m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a172d1d56de34fb7b16902993d4e22cd : SUCCESS in 52m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a455aab775db4464a7dd916c74ae98a1 : FAILURE in 59m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/aecffa7c891e4b2bbce1e65554e33548 : SUCCESS in 29m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/d1df5499873a4cf5a4ea731c56e9ac08 : SUCCESS in 46m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/9da2043cd5c049eb8096b9e51d33ccfd : SUCCESS in 34m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/12f02b4f58b34d0fbd81696e047d612c : FAILURE in 1h 16m 47s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5dce6c4bc4204cac8a412cb5723d42f5 : SUCCESS in 51m 27s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/006dacc5baa04473be900d8d1e678a8e : FAILURE in 1h 04m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"6155c5a940597cc260b9691f599a628371497447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-10 15:36:42.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"6b087d77117adc4cefbb5d98a0e3ecf54f01105a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-10 15:55:32.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"e3b7c85c9f1386a256979212fbe533a0fabacd37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 18:12:03.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90cf1ca7697646f29db4a4b60cb89de2 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/920cc4e667754286915a376ce5cb76de : SUCCESS in 3m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/813e4c29fdd044e08d52037b3a84f162 : FAILURE in 1h 44m 05s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/0805da0d70b34b788cb21ddd4ab9e60a : FAILURE in 1h 54m 54s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5e6bdfc196c742538dcd9043be2c5ae2 : TIMED_OUT in 2h 34m 14s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/dab767366e5945449d1e971b94cb1f19 : TIMED_OUT in 2h 34m 05s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/72a3086f473f449d84c5fa78a6b5e3d1 : FAILURE in 1h 37m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/558fb23f75204c1a904f811817f44225 : FAILURE in 29m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ee5049cbf014131b0da3aa1b772dfb0 : FAILURE in 48m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8097b5a5fd5144ddb36b169e9d08e14a : SUCCESS in 36m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/af0df3e7331e481e8aa0c31bb2dc446a : SUCCESS in 44m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0d43b6d65e974411bd9ede8848f2bd19 : SUCCESS in 1h 04m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/680f0e508423429ea5661cb13a3a6980 : SUCCESS in 44m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4b5bffe83617413eb8d99873b3d5e056 : FAILURE in 50m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/4e8491e771074de08449866586d2e69e : SUCCESS in 43m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/c15b4bb3231c41869f33041e5015e2f0 : SUCCESS in 39m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/4940b09a76b44e4da12d353b23873ff6 : SUCCESS in 39m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cbd15d98e93f435e933c45d868dc7e7c : FAILURE in 1h 56m 13s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d95cdff8822d40a0912c0b4e7dfb71d5 : SUCCESS in 44m 42s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/8be99f4ecfba4acab0fd19ce4f474d7a : SUCCESS in 51m 11s (non-voting)","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"1f708addb01b7f2a3ad9de95affe3f95a8c74ea9","revisions":{"460c0ea8ad0dc373e56e9a3baea788281b2e05ba":{"kind":"REWORK","_number":1,"created":"2021-01-18 16:40:01.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/1"}}},"commit":{"parents":[{"commit":"4c0f9b094c7a973dbad8651f3d208ed51390ac6d","subject":"Updating _test_pool_CRUD: Added a new parameter \"next_algorithm\" that defaults to None. If next_algorithm is not None then the update of the pool will also update the pool\u0027s loadbalancing algorithm. Previous tests without this parameter ARE NOT AFFECTED. New Test: test_UDP_SIP_to_LC_pool_with_listener_CRUD tests the result of updating the pool\u0027s loadbalancing algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4c0f9b094c7a973dbad8651f3d208ed51390ac6d"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:39:58.000000000","tz":120},"subject":"Modified test_traffic_ops for loadbalancer failover testing","message":"Modified test_traffic_ops for loadbalancer failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/460c0ea8ad0dc373e56e9a3baea788281b2e05ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/460c0ea8ad0dc373e56e9a3baea788281b2e05ba"}]},"branch":"refs/heads/master"},"16b43933bafef396f8966b4d003141df3d3afb83":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-18 16:45:48.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/2"}}},"commit":{"parents":[{"commit":"4c0f9b094c7a973dbad8651f3d208ed51390ac6d","subject":"Updating _test_pool_CRUD: Added a new parameter \"next_algorithm\" that defaults to None. If next_algorithm is not None then the update of the pool will also update the pool\u0027s loadbalancing algorithm. Previous tests without this parameter ARE NOT AFFECTED. New Test: test_UDP_SIP_to_LC_pool_with_listener_CRUD tests the result of updating the pool\u0027s loadbalancing algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4c0f9b094c7a973dbad8651f3d208ed51390ac6d"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:45:38.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/16b43933bafef396f8966b4d003141df3d3afb83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/16b43933bafef396f8966b4d003141df3d3afb83"}]},"branch":"refs/heads/master","description":""},"3a2923fb09b413fe6eaa754b3d2d8cf615a3b612":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-19 13:39:31.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/3"}}},"commit":{"parents":[{"commit":"efae4470eb91859b6957d9280f0c60de25da54e8","subject":"Merge \"Fix Go lint errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/efae4470eb91859b6957d9280f0c60de25da54e8"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-19 13:39:31.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3a2923fb09b413fe6eaa754b3d2d8cf615a3b612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3a2923fb09b413fe6eaa754b3d2d8cf615a3b612"}]},"branch":"refs/heads/master","description":"Rebase"},"1d64780bbc9b10acd6d3befa28b7c440da6f25a1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-24 10:19:02.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/4"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-24 10:19:02.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1d64780bbc9b10acd6d3befa28b7c440da6f25a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1d64780bbc9b10acd6d3befa28b7c440da6f25a1"}]},"branch":"refs/heads/master","description":"Rebase"},"2dc9df5ecb96bc6da9a0962c64d5a221d5628bf3":{"kind":"REWORK","_number":5,"created":"2021-01-24 14:24:54.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/5"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-24 14:23:01.000000000","tz":120},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\nMerged _test_basic_traffic and _test_disturbed_traffic into a single method\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2dc9df5ecb96bc6da9a0962c64d5a221d5628bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2dc9df5ecb96bc6da9a0962c64d5a221d5628bf3"}]},"branch":"refs/heads/master"},"da03009af3db3bbd3b97de6bf16d8e2000416754":{"kind":"REWORK","_number":6,"created":"2021-01-25 07:51:53.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/6"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-25 07:51:16.000000000","tz":120},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\nMerged _test_basic_traffic and _test_disturbed_traffic into a single method\nfixed typo on line 207\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/da03009af3db3bbd3b97de6bf16d8e2000416754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/da03009af3db3bbd3b97de6bf16d8e2000416754"}]},"branch":"refs/heads/master"},"16249c63d41da8e3a0186ba498694718c87a36c9":{"kind":"REWORK","_number":7,"created":"2021-01-26 09:29:40.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/7"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-26 09:26:11.000000000","tz":120},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\nMerged _test_basic_traffic and _test_disturbed_traffic into a single method\nfixed typo on line 207\nfixed udp port clash on line 262\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/16249c63d41da8e3a0186ba498694718c87a36c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/16249c63d41da8e3a0186ba498694718c87a36c9"}]},"branch":"refs/heads/master"},"47823f0728bb386cf23cf6f41d60bf85dfc44112":{"kind":"REWORK","_number":8,"created":"2021-01-26 13:51:44.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/8"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-26 13:50:28.000000000","tz":120},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n* Merged _test_basic_traffic and _test_disturbed_traffic into a single method\n* fixed typo on line 207\n* fixed udp port clash on line 262\n* fixed typo on line 263\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/47823f0728bb386cf23cf6f41d60bf85dfc44112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/47823f0728bb386cf23cf6f41d60bf85dfc44112"}]},"branch":"refs/heads/master"},"b6d34cb5f28ea8140aca9cc51ae7734450c93c9a":{"kind":"REWORK","_number":9,"created":"2021-01-26 16:43:16.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/9"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-26 16:42:16.000000000","tz":120},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n* Merged _test_basic_traffic and _test_disturbed_traffic into a single method\n* fixed typo on line 207\n* fixed udp port clash on line 262\n* fixed typo on line 263\n* corrected the class name on line 263\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b6d34cb5f28ea8140aca9cc51ae7734450c93c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b6d34cb5f28ea8140aca9cc51ae7734450c93c9a"}]},"branch":"refs/heads/master"},"11bf25563fae6ea544919afa51388e82145fae1e":{"kind":"REWORK","_number":10,"created":"2021-01-27 14:16:35.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/10"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-27 14:15:43.000000000","tz":120},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n* Merged _test_basic_traffic and _test_disturbed_traffic into a single method\n* fixed typo on line 207\n* fixed udp port clash on line 262\n* fixed typo on line 263\n* corrected the class name on line 263\n* corrected the invocation of the dsturb_func\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/11bf25563fae6ea544919afa51388e82145fae1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/11bf25563fae6ea544919afa51388e82145fae1e"}]},"branch":"refs/heads/master"},"ce889fca1f34456f1dcdecabd27e88e08603599a":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2021-02-10 09:26:26.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/11"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-02-10 09:26:26.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ce889fca1f34456f1dcdecabd27e88e08603599a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ce889fca1f34456f1dcdecabd27e88e08603599a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d5152bc50a255d1b500ab4c6c187e3bd1b8eb00d":{"kind":"REWORK","_number":12,"created":"2021-02-10 14:25:24.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/12"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-02-10 14:24:51.000000000","tz":120},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d5152bc50a255d1b500ab4c6c187e3bd1b8eb00d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d5152bc50a255d1b500ab4c6c187e3bd1b8eb00d"}]},"branch":"refs/heads/master"},"2267e7c3b89ce89ce5b2c92d18975699f847e6e3":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-03-10 09:49:54.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/13"}}},"commit":{"parents":[{"commit":"3ff105bb5d84a05269f8e42918cd49ca4a80b23d","subject":"Merge \"Enable HTTPS on the test servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3ff105bb5d84a05269f8e42918cd49ca4a80b23d"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-03-10 09:49:54.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2267e7c3b89ce89ce5b2c92d18975699f847e6e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2267e7c3b89ce89ce5b2c92d18975699f847e6e3"}]},"branch":"refs/heads/master","description":"Rebase"},"c614c22d985ff2823d5f17d6049a7bfce3a615d2":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-04-07 09:07:58.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/14"}}},"commit":{"parents":[{"commit":"434170a8e0740aae2d6c31a9063a5c8db7616809","subject":"Merge \"Fix testing IPv6 VIP network connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/434170a8e0740aae2d6c31a9063a5c8db7616809"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-04-07 09:07:58.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c614c22d985ff2823d5f17d6049a7bfce3a615d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c614c22d985ff2823d5f17d6049a7bfce3a615d2"}]},"branch":"refs/heads/master","description":"Rebase"},"560e9025e636fa1e0306c9f677f961571b84ab8d":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-05-02 09:03:22.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/15"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-05-02 09:03:22.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/560e9025e636fa1e0306c9f677f961571b84ab8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/560e9025e636fa1e0306c9f677f961571b84ab8d"}]},"branch":"refs/heads/master","description":"Rebase"},"959188d585e55b38243decaeb5c3526af5482f02":{"kind":"REWORK","_number":16,"created":"2021-05-06 08:15:11.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/16"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-05-06 08:14:01.000000000","tz":180},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\nswitched all self.lb_id to clc.lb_id\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/959188d585e55b38243decaeb5c3526af5482f02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/959188d585e55b38243decaeb5c3526af5482f02"}]},"branch":"refs/heads/master"},"750db84e2b210e64fbdc9c170af5df152478a099":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2021-05-06 08:16:42.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/17"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-05-06 08:16:42.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\nswitched all self.lb_id to cls.lb_id\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/750db84e2b210e64fbdc9c170af5df152478a099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/750db84e2b210e64fbdc9c170af5df152478a099"}]},"branch":"refs/heads/master","description":"Edit commit message"},"800daa3293d74e4082a1214f8f7e2bd39edeb03d":{"kind":"REWORK","_number":18,"created":"2021-05-10 09:31:59.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/18"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-05-10 09:31:08.000000000","tz":180},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\nreverted to version 15\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/800daa3293d74e4082a1214f8f7e2bd39edeb03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/800daa3293d74e4082a1214f8f7e2bd39edeb03d"}]},"branch":"refs/heads/master"},"5254431d3d12215d46e8256f2e880c7d0a16fb18":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2021-05-10 09:34:20.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/19"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-05-10 09:34:20.000000000","tz":0},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\nActually reverted to version 15 and built from there\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5254431d3d12215d46e8256f2e880c7d0a16fb18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5254431d3d12215d46e8256f2e880c7d0a16fb18"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1f708addb01b7f2a3ad9de95affe3f95a8c74ea9":{"kind":"REWORK","_number":20,"created":"2021-05-10 15:36:42.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/51/771251/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/771251/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/771251/20"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-01-18 16:20:16.000000000","tz":120},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-05-10 15:35:49.000000000","tz":180},"subject":"Modified test_traffic_ops for failover testing","message":"Modified test_traffic_ops for failover testing\n\nA new method \"_test_disturbed_traffic\" has been added\nthat sets up a load balancer pool and then executes\na fuction that has been passed as a parameter\nto cause a disturbance of the load blancer pool\nand then tests that the pool has recovered.\nNew Test:\ntest_failover_udp\nthat tests that a load balancer pool\ncan recover from a load balancer failover\nActually reverted to version 15 and rebuilt from there\n\nChange-Id: I4f6c16a164e08996581ba60a48ba7c480b5c4902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f708addb01b7f2a3ad9de95affe3f95a8c74ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f708addb01b7f2a3ad9de95affe3f95a8c74ea9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
