)]}'
{"id":"openstack%2Foctavia-tempest-plugin~773888","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I24976e93e9ea8f7f50c9da014c2627965512893c","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"two-node","hashtags":[],"change_id":"I24976e93e9ea8f7f50c9da014c2627965512893c","subject":"Fix two-node job configuration","status":"MERGED","created":"2021-02-03 13:34:37.000000000","updated":"2021-08-17 19:11:26.000000000","submitted":"2021-08-17 19:06:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773888-two-node","meta_rev_id":"3cea490bc094fdbc9ffecff3bd987dcaa914e2cf","_number":773888,"virtual_id_number":773888,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 19:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-08-17 16:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-07-21 19:21:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-21 16:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-08-17 16:56:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 16:11:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 07:48:38.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-02-08 14:36:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-17 16:43:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-08-17 16:56:43.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"680e7b427dd85c89fe90d7a028c30dad992ac5ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-03 13:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a32f1510babffd7ec717211e516259593202ee7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:11:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88dc64489a4945a691f4ab57ee6e622b : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeaf52ca62db4edbb55efc5607518e33 : SUCCESS in 3m 36s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e86c14c6959492e869016da774a281e : SUCCESS in 2h 23m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6191db1dd33fe174a30e463a963918d9d94e3deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-03 16:36:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"099fd3563101f1baf6bb492d812740d8d74750e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 19:02:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f251cad8404a88a752aef4dbcb082f : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6b2473a0574079be31ab4ef927770d : SUCCESS in 3m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54d4381322cc4c8a9a1f9ad8c81fe3e3 : SUCCESS in 1h 07m 08s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/ae86785f1e5141a09a2891e18830dc7d : SUCCESS in 1h 16m 40s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e361be6c5bc0418eb14d9c349ff394fe : SUCCESS in 1h 30m 17s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/1a8ea468a9dd4804aea8f5d900f4ab37 : SUCCESS in 1h 02m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf5e8ef653c74b69920d2d5d984f039e : SUCCESS in 1h 31m 29s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f1edef919ddb49f19a266b7df2edf6a0 : SUCCESS in 1h 45m 56s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ccf1648b305b4f27a09024ebf369f3d7 : FAILURE in 25m 27s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ed0719d4148c4ab4ae1805929141e186 : FAILURE in 25m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/941b14414529417780f1b22e396f2155 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/a8dfcbf1f6cb4d3a86ec1330ec4b66ab : SUCCESS in 40m 49s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/401c6cd866dc4aca9be5af9dae2f0b8a : FAILURE in 26m 57s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/666b84f866274100b57aad9687001c53 : FAILURE in 26m 32s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ca474c4c43d742eaa46ad3bb41fc3d52 : SUCCESS in 2h 01m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/69ce60841a1d44caaaa4f0c8f57341c1 : FAILURE in 30m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1fd39964d4b14b478650a2db73c3a6e5 : SUCCESS in 2h 21m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efd99a4eb6ae4266b3aefe8cfe031c8d : SUCCESS in 47m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8594501f86884ab9b6b3bfd8b438ba74 : SUCCESS in 44m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ea5634e7690940548abae35fe051744a : FAILURE in 23m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/50f4cc8fb44e45c0a2d7d241864986b1 : FAILURE in 22m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3f6ea6ab34114bcb8dc505f98dee7688 : SUCCESS in 51m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/7e8f2b3c2e2a48928d58a3d78ade22db : SUCCESS in 41m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/09623794a7bc44ef86db06450feb37ac : FAILURE in 38m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/8b31790b5c5445c593d8f82787dbe94e : FAILURE in 36m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a9b3b9b07c084509bc00040af40700ed : SUCCESS in 2h 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4aa6fecfb4180677b058194bc6475a3840b36f7b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-04 07:53:59.000000000","message":"Patch Set 2:\n\nPlease note that the two-node job is now passing:\nhttps://a96e38bf7825cd25c740-9f001e5f5ba768b81be8f04331a61031.ssl.cf2.rackcdn.com/773888/2/check/octavia-v2-act-stdby-dsvm-scenario-two-node/1fd3996/testr_results.html\n\nPrevious results: https://zuul.openstack.org/builds?job_name\u003doctavia-v2-act-stdby-dsvm-scenario-two-node","accounts_in_message":[],"_revision_number":2},{"id":"a64b17518fca1b1305c38bea7e33a55cb5a74980","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-08 07:48:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a02a357e78a8439940a9adfe234b65205f76712d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 07:52:59.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"e7a66398c5b277d5c860bb875d7201524f521b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 08:16:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb571f9a31c742f88b04159bcc68a7a62bb3bc2c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 09:01:33.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"54a28061735da0887a5f3869af246ca225d8d982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 10:17:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c7b44ce3a7462a886cb3d0f81450b6 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bf135ce4326451d8b4f1cbd20529d97 : SUCCESS in 2m 32s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c642124b1d840e5a4c3bde8e5ae19c5 : SUCCESS in 1h 58m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23dc81dfe2b119f381581bda6fe460ea93e10ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 10:32:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6e53988d8b8dfba48f0d706c07165e1096553853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:19:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a2146f81214c4cba351b544a9385eb : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719454c1b531418397e6724bf6ec2e5d : SUCCESS in 3m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea80c43167b34968a6a568cd57ae0bdf : SUCCESS in 1h 26m 40s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/fb4eb0e8b0194854aed5e4fb40305b56 : SUCCESS in 1h 24m 07s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/ba8f852ba7a9424cafdc9d94e771a6b6 : SUCCESS in 1h 16m 38s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/4bb24077984f4158b79af959f3ddc4ac : SUCCESS in 1h 08m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef4328efcdc9466eb3b258f64de13d15 : SUCCESS in 2h 14m 27s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ec5ca8a3d439459a8c598c09992873bd : SUCCESS in 1h 59m 02s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9cfe99cbd7ea4b0e88f7347b50d14069 : FAILURE in 25m 07s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/f15253ce0c99439b8aa5512fcb784813 : FAILURE in 25m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09c450226c6c4169a76c4c452715405a : SUCCESS in 42m 36s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/8b83011adbf44e91aa33c428c6d424d9 : SUCCESS in 40m 14s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ca551f28fd054360a8ad0fe648013988 : FAILURE in 33m 51s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/0ea3f65709a24b608d80792de8706cbc : FAILURE in 25m 11s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7830d2432f144529805cd9b07524d5d9 : SUCCESS in 1h 46m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c5e3cd78a96144fb9ad078848cb48ce8 : FAILURE in 30m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/51793ef4a3fe4558880731e68d363216 : SUCCESS in 2h 22m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/434128252d4041d39eae3a7e9eaf6668 : SUCCESS in 40m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ed58fe405521428bb3dbe4c5d5ff2275 : SUCCESS in 36m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5e2baa78ba6d4c57aa4977f5f973987c : FAILURE in 25m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/d295ba6176d64b64b894e83481c2981c : FAILURE in 42m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e8b51d602b29413ca1616cc645a1e307 : SUCCESS in 1h 05m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/b1f6e5803a83484a9211284cb9093cb1 : SUCCESS in 1h 10m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/00354c5c80424c90ac255bfb3bd86324 : FAILURE in 25m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/57a61591caf94c978a57d5dbd0484873 : FAILURE in 43m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b6e87bdcd4774ad08b28712648bde496 : TIMED_OUT in 2h 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d83186f787f8ef90344d8c1ac7c594ba27d57636","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 17:42:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ea8deb81c2c8c66a06e55155c70dbe7164f7142c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 17:42:43.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"2c1e0d14ab843fcd190da5da0e6f8ce4a5dcb4fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 20:20:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e83b47b381042d0b025745b4883fe7e : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a069bbbc43304efeb1a5aef455e8dcf5 : SUCCESS in 3m 19s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59d5ddff94aa4ffda6eb0337e48537c5 : TIMED_OUT in 2h 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ac95aa42d56f0e51094468706ea5bbb565364cc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 20:24:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"812d294ffe689ec401e3e895b0b81c26f6e59db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:00:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b95f65c6ecf6434c825a840a6b118d22 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62a48c105c24c8b985ab2c43584bc30 : SUCCESS in 3m 07s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da482cdbd2dd4daa97127f37628af61c : TIMED_OUT in 2h 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b8fc8b456a6595573e782a9fb61e78e32694b769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-09 06:24:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3e5c693b48dea4213d622fdfec23f3f151681d6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 07:17:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9dbd3d2bf24b19bb3d7450578fbbbb : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76b6a7ee42d6414a9989999c59b93da8 : SUCCESS in 4m 23s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/061802bc52334b2a9951169cd3e9c91b : SUCCESS in 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1281668ca3e05c69f7065360e04488df0417dfad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-09 07:48:27.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"a610b8b9e66f6bd10417445a3c9af846ff3bbb1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 08:37:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3be4c53bafe4a58820ae8efd1efb0f2 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8739adf5529945a6b79c880d0f68f10a : SUCCESS in 3m 21s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/abb9f038d9254a91a3904e1d1be30867 : SUCCESS in 47m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cd3581cc40103c4da032c13436ff90173c25295e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-09 08:41:39.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d19bb07311d998f165f1a077947adde3139f6633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 10:57:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97fe4dc7f49a48f6a21d66b4f6c93a01 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d8ca66191e4da0bf1478a78c24e23b : SUCCESS in 2m 42s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ede6df72a754e32827c2013b134ff08 : SUCCESS in 2h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c8088716b036b7d82f0470d6d3fe7659957d5c75","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-09 11:04:51.000000000","message":"Patch Set 9:\n\nhttps://zuul.opendev.org/t/openstack/build/5ede6df72a754e32827c2013b134ff08/logs\n-\u003e job passed\n-\u003e amphora logs are there\n-\u003e nova.enable_anti_affinity is True\n\ntempest logs and config are still missing","accounts_in_message":[],"_revision_number":9},{"id":"1fd6337f9d3a776a28141cbe8f1c3373330d4872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-09 12:25:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b314d69fbeca950847849144f086dbaa7daa5694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 15:05:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5f64b78bba49e2841fe9f822ee9bff : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/561f206d74fd4fe8a6a6f0516c9e20a4 : SUCCESS in 4m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3918233780c34afa83780dda70592197 : SUCCESS in 1h 11m 26s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/85ecf00addc14e6995b4fbe1b088149e : SUCCESS in 1h 12m 57s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/4669ea7c749c4101854f32c2e8fa13e1 : SUCCESS in 1h 11m 43s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/8ddf06bad8774a65af8358b409a95e69 : SUCCESS in 59m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6259addef12e4e3d8aa116e544ce11c0 : SUCCESS in 1h 51m 42s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3af3b86b8fd64c0aa82cea5e5fdec603 : SUCCESS in 1h 57m 03s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c416d7fe614149299c03c851f8c543de : FAILURE in 26m 16s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/7ef2136cfc614ef3a75b1f55c2696432 : FAILURE in 25m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5fc510a9ee046ee8961ee1b9af598b5 : SUCCESS in 48m 27s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/0938b1069e89491292f862ece2dcd795 : SUCCESS in 42m 41s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/e0a4f0ab4cbc4034aadc6c5d74dba108 : FAILURE in 52m 25s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/12a8e3fa732d4500a70beb4001ec0c90 : FAILURE in 33m 23s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/22a21074ef28409d8242bf0b1f15aee7 : SUCCESS in 1h 49m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/577abbfd4733455c9d6f9b3459b5bbe4 : FAILURE in 29m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5eaa56f00aba4b09b46a6eadf0c8b8b8 : SUCCESS in 2h 15m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18dd936c16704059902734afbc84741c : SUCCESS in 48m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ae39e0f4a3fb446e94aad87abd599ac8 : SUCCESS in 59m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6d8933719f5d4649bae3bdf949ee631d : FAILURE in 25m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/ea2054d69d27405b93e05f8ff585467c : FAILURE in 24m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5aa3433f6ede448fa0ed3992b17bc752 : SUCCESS in 52m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/4847a5d04b214028a4c0867261917fe2 : SUCCESS in 40m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/9d8e990640564634992b26b86acad12e : FAILURE in 24m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/271e1afea01044b78190221aa318fb6e : FAILURE in 33m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6803fd89d5d64d31b42cae7dd9ea60af : SUCCESS in 2h 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2e994d7676cc669fbf89f17182d573473498cfad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-17 16:43:42.000000000","message":"Patch Set 10: Code-Review-1\n\nThis is good, but now all of the other logs and configuration file collection is missing.\n\nSee:\nhttps://zuul.opendev.org/t/openstack/build/84bdf7d2cc6c4aecb8e1cbb5046b586a/logs\nvs.\nhttps://zuul.opendev.org/t/openstack/build/5eaa56f00aba4b09b46a6eadf0c8b8b8/logs","accounts_in_message":[],"_revision_number":10},{"id":"675b7d1d9c67650e7d5b94269394ead26a19eb30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-17 21:02:12.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"fe16c746e2939f992e36f8685ed8ac3585888a39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 21:51:17.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abf30ee91c24db88403ca43944a8a2e : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7399eb20b8fa466db2307547177734ac : SUCCESS in 3m 12s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/28e9c70d4a5b4d5b9a0f4c02b9c3fc63 : FAILURE in 46m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4a8572a1bf74621a5e858cf4c6c0395 : FAILURE in 42m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0299817f9abf461b62e90aa5cf9bcc6b7bc9f7ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-18 06:45:01.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2f1028bbda29b44a4d4603d43ced30244304705f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 07:53:01.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559cbbb51fca4c1cbfc1c80a9843bfcb : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094d67e4557e49f9b19ccf5763881ae2 : SUCCESS in 3m 38s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/07a080056a614a42943578b70bb9f566 : FAILURE in 1h 05m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4a7521305d04643988c23b1bbb93b8378774642f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-18 08:00:02.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d3059ccb19bdbfc0d215009e44a7094a0db13c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:04:29.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2bddffb6e72480b9e2c654c7c3c5341 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8496bf142840779a331d91ddaf4c6c : SUCCESS in 3m 28s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/022f6a73520644a9aa534e8f51ed4b8c : SUCCESS in 1h 47m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e6da6c7c5eb64e2ea7b0e7f961e92635 : SUCCESS in 2h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e49b20dfefa8e75511cf57d27d551345dd636ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-18 13:50:29.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"428a383c1b756abaf0f18ed9e97e02d3518f773d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 16:21:41.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e613fd1d014410c941f7a05317f46de : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7389cf1ce9084da68a4d88af283ce297 : SUCCESS in 3m 50s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea993550780945a78f0c255d5d4218d6 : SUCCESS in 2h 16m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ca5930d9a7d4d2fa36a8588d90f04c7 : SUCCESS in 2h 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"86a502b7e3d1fb5d1027c5208aa79692ee58843e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-19 15:33:52.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a2be814725c9be098f636f835cd3ba998ec6061d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 18:33:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8ed7e835dc4048a72d3d97e48aacd8 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b3b88cd75546c6b2395d7320afa0ee : SUCCESS in 3m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70cabfc204944ab3ac8a5ad7ae8f8234 : SUCCESS in 1h 10m 46s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/5b4a1180cc934cf484c9e4bea4ca0a1c : SUCCESS in 1h 00m 34s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/5cf37a1a061b4609a3618e7c4ffc56fc : SUCCESS in 1h 14m 24s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/f50d12510cb54f38ac5eedf1c6227f0d : SUCCESS in 1h 02m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4daaf7db901a4f52a8c66caf2b688c97 : SUCCESS in 1h 37m 14s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/fd8a07378b98456eb181e8c2538b2228 : SUCCESS in 1h 43m 02s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0e8a7713d90d4f1bb03886a357d323ad : FAILURE in 25m 22s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/4425434afc104c33b1f61862c70f48cf : FAILURE in 23m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eef599a72f2f4dbca2db4dc20bb71e11 : SUCCESS in 1h 07m 15s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/12e62bf16bc9406489e73a51a39d1094 : SUCCESS in 57m 18s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/616de67ba92041428f036b805051128c : FAILURE in 28m 51s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/db3ce15f175144d2899d3833f1255f6f : FAILURE in 27m 41s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/19114074dcac46e2b43265727c93a57e : SUCCESS in 1h 51m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/22b3843634c348e8baeb9ef242edf359 : FAILURE in 35m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d50fc026b34d438bb0e321afb34230a8 : TIMED_OUT in 2h 36m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71dd7fbbfd4e4e5f889c3f476f9d78a5 : SUCCESS in 40m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d52db998d0974792a40e03c050c938cf : SUCCESS in 42m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6fa7af02cdd641b1b61d887d1837c7a0 : FAILURE in 24m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/0baad37f09c14182b57fd4bd6b23bc0b : FAILURE in 24m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2d2765ffd5fe4eeaa2c0d849cd6e96b9 : SUCCESS in 38m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/66877ed7922d4bfdba7d9fd48100383e : SUCCESS in 37m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/dd3a11166d8b41b38e08dda48bf154b7 : FAILURE in 25m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/f61f989790d84a79b95022417027f089 : FAILURE in 30m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6501038975c54efbb38f3dff0d1d78e5 : SUCCESS in 1h 54m 13s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0a7fbfc36a9530192b5f3313a8b5c15568c5b375","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-09 07:37:19.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"02bc71667d19a654230499071bf4ad8aa9ad3a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:16:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b4a69a4b3c47a2b82dc43d826fc4eb : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f548249e14a4035a36d6f8fc349f63b : SUCCESS in 3m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd467b1521de4bc3921add0f47e94ee4 : SUCCESS in 1h 18m 23s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/9ce6f9eecbf048e9bebbd4180d38c8f4 : SUCCESS in 1h 23m 53s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/dd8f9ad6e8f24a1fa2230e53fd15746c : SUCCESS in 1h 16m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89caddce6c80498ba727765cc5f527b7 : SUCCESS in 1h 47m 50s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d1bff721dfa14fe2b6f90f52ee9365fb : SUCCESS in 1h 46m 18s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/73c050c763994187ae9b059ebfdbeedb : SUCCESS in 2h 11m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97096f7b86e04461a39e661107ef5121 : SUCCESS in 46m 14s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b0e196429dcb4d408aa0e4dcf93e35b0 : SUCCESS in 57m 32s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/74ee61f7157145b48b288c08e3f894cd : SUCCESS in 45m 41s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ab97cdea99b24c7087d55dddfa6beb73 : FAILURE in 2h 12m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c9b0c90d93d740ea944c7ec855373750 : FAILURE in 30m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/25a7b21e2b4f412abd0164f469354290 : TIMED_OUT in 2h 35m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/031cacd5f4504c1f8831f0adbafb2487 : SUCCESS in 46m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/62daef7333e84b959c7618bf956b5513 : SUCCESS in 44m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/197595651d0742a99bd1f9758b1b0f44 : SUCCESS in 47m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b5314742dec743399a9afd406c7137ec : SUCCESS in 39m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/15d0444f41a240b5a7cb3923d6964e1f : SUCCESS in 46m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/86327606be4d4b91935f79596c2b40a2 : FAILURE in 39m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/82082931d3644b85af6cce32f5d1cba5 : SUCCESS in 2h 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"070c4c2f680ee10fb743cce7fea930fd5910afcd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-10 02:58:39.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"0921e1c53cc6884f15933bae21e52e0d84186cb8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-17 07:35:59.000000000","message":"Patch Set 15:\n\nrecheck\n\ntimeout in two-node job","accounts_in_message":[],"_revision_number":15},{"id":"b43a90e21376cbc26cb7accbe50e96db3fed2cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 10:06:01.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bbe6e6e688494786b44a1902cdb128 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb16da914bd418e92f470fd5b053820 : SUCCESS in 3m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b858d2b0fed4c398a28db2d639bb196 : SUCCESS in 1h 17m 48s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/5da36ea8e919463da05767320097fac5 : SUCCESS in 1h 36m 36s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d8be2f85cede4b909a152848f34c3a12 : SUCCESS in 1h 15m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/032aae402b8f4d76b734b4696e182f66 : SUCCESS in 2h 03m 32s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cfaabab871e74f1ba24e2d83a86d4462 : FAILURE in 1h 47m 56s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/617629cc95aa4041b5f569fb9592dc57 : SUCCESS in 1h 57m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c8ffab6957784520a2098c2fe89c50f5 : SUCCESS in 44m 05s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/502b325ba64f4d81a0e45a89a7be284c : SUCCESS in 1h 05m 14s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/6b4a989603214ab2be7e481e9a81680d : SUCCESS in 40m 58s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e080cd1857664e3cbfdf6ca44bd454ca : FAILURE in 1h 36m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5bacc9ea04354218be0f5245c78832aa : FAILURE in 45m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3df8c94c89484a39b20bedc19b7c2d5c : FAILURE in 48m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/552255d16d254d5dada0ce50b526c7b7 : SUCCESS in 35m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c187bc1d7d6840cf9582c57c2ccd66db : SUCCESS in 41m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a208f4cc54184a54add2bc53769f3488 : SUCCESS in 47m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f08cc0b59b4b4dcf9d8b415813edc943 : SUCCESS in 47m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/6af70e5197c7480d97c3444e65cbfcc5 : SUCCESS in 42m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/43c36fc3f7c446b1b30eddd7e3fbd0d4 : SUCCESS in 40m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a573e7d29fda463a8c6245e0065b99fe : SUCCESS in 2h 27m 12s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/167aa8a880024b62a110177b8e744fe8 : SUCCESS in 47m 53s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f12020a3746b4cc3afc1d5b360c1f705 : SUCCESS in 57m 33s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"cf5579840eec205756f5d86498e6c63be9f3cbba","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-17 11:07:05.000000000","message":"Patch Set 15:\n\nrecheck\n\ntimeout","accounts_in_message":[],"_revision_number":15},{"id":"e30ad9fbd1fe915d6f1cfc950ea24497064cf6c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 13:08:11.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/703a154febe3423ca33b87c7488de776 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b538acabe3c348339830c9f15000a487 : SUCCESS in 3m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a87163d94bca4ede8a0cfc05ac96688c : SUCCESS in 1h 02m 13s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/57582b179dc44e87a419f084ad64d614 : SUCCESS in 1h 28m 03s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/4e1605402ecb4b798fcea1e3eab2cccc : SUCCESS in 1h 32m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a2a61f40d58451cacf7f91c3bdad6f3 : FAILURE in 1h 42m 45s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2319caecc76f4489a7a69fabbada87f9 : SUCCESS in 1h 46m 56s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/71f7dd03ca9640f081a01550c41d018b : SUCCESS in 1h 56m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d8a633c8108490ea7aefee3139ab746 : SUCCESS in 53m 24s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/197e1304cd514f5db76cefead40d5bb4 : SUCCESS in 53m 11s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/25e457b755de4d1ea5dae3ab97d087d6 : SUCCESS in 34m 14s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/014a1d637c4b4ed09633049a657dbc89 : SUCCESS in 1h 53m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/196bd603d8dc43f883727db92f1413d9 : FAILURE in 32m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aedccf91052a41b8af40acdc81506ee1 : FAILURE in 50m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7b97e3ea2d641c3999878db86da6da2 : SUCCESS in 35m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/edd51a1408ab4280a95b74c9fb1606f3 : SUCCESS in 40m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/84eb51e32f6141219aff59128bb16d54 : SUCCESS in 40m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/08ec3ae6649f4a8092808cf909796838 : SUCCESS in 1h 07m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/06aefd32da804ad9bd62d7665107d532 : SUCCESS in 41m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/4fcdef65d6d24a1fb4df51377f36ddd2 : SUCCESS in 40m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a626e83474a947f884504e7bef5708a0 : FAILURE in 1h 57m 07s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/454b6952481846119af5d39256b0a286 : SUCCESS in 50m 52s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/359017269cdd4c8aa2f6c617ac52fb76 : SUCCESS in 52m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ea08b4ea09cf14f67777d8d1acda9759de614f17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 10:55:52.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"0902fa73126a72ce91fa7fc867148946e8bbcd86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 13:32:47.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba66f7ab7d44eb0b97d07519e321a7e : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe12a231f3a4871bcdfaeb51fb5499b : SUCCESS in 3m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4bbe0b80d044fdd99a09523fde1caf3 : SUCCESS in 1h 16m 27s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/52a65ad1c1fe4eee909720959d2beff9 : SUCCESS in 1h 07m 43s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/7a952ec8d1a341dc91aa185e8b97243a : SUCCESS in 1h 07m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80fd20597e174265abdce7aaefcb0ad6 : SUCCESS in 1h 30m 25s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6f21ea907e994f138fefb0fd41503a43 : SUCCESS in 1h 47m 48s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6a6ef7a8dc3d466792e2b8c4bc9d85de : FAILURE in 1h 56m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39b6be04c6cf4744ad4a1134f6615e45 : SUCCESS in 48m 10s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/9c3eeac21d574914a5e398bfb82e0cd4 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/dfe56fb449aa47fcaf79e389c52fbcbb : SUCCESS in 40m 54s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/344922fc771d487a827fab33cb6147de : FAILURE in 1h 54m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6d1ae9465da34b76823c886769a6b09a : FAILURE in 28m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86b660b0cfb947a19c75df2f26fd4595 : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecfe6dab03ff426696176a398c04092d : SUCCESS in 43m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/dd5a86133bb946019b54e29f2e27ba4a : SUCCESS in 43m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a3d8d46e966b468ab364a8be0ca12b49 : SUCCESS in 43m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bd341899143c49b08cd00e5c3f14973f : SUCCESS in 43m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/81b6a643dd044542b77407e29b7aabb5 : SUCCESS in 42m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/40b04032e0a8459b9f5056863bc09049 : SUCCESS in 42m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/16b0a5e927024a2e89f5e10bbcc6cc24 : SUCCESS in 1h 48m 30s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/fdcc601a5ac14eb791bf2d497681117e : SUCCESS in 48m 29s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c0a45db40b7d43cc8cf66e07a54c25a7 : SUCCESS in 52m 53s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0c6670eceb5353458b60ff02081b48e8168ecc3a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 13:37:14.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"f98d3fdf5481e4454734f12afcbaee12f403d010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:16:27.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da09e97a86a450f99d799759010ca93 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c1bf869567403591f4f65303568641 : SUCCESS in 4m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/726b63a0a77b4a51bdce830fc2b01ced : SUCCESS in 1h 17m 35s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/87233b24d3214778ba7a1f28fc77656b : SUCCESS in 1h 17m 42s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/b00bc400af1846b4ba2871fed2bb7ea8 : SUCCESS in 1h 10m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e59de76b0f104847941f0334f1c45d9a : SUCCESS in 1h 59m 43s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a84d533adf614c79b8f9d90d07c0f58e : SUCCESS in 1h 50m 59s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/aa638e19a1774d06a764e1eeef758db0 : FAILURE in 1h 49m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/70133ef55fef40a3a01bd0b1d392a537 : SUCCESS in 44m 18s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/02528019166043249710c8907c547360 : SUCCESS in 46m 11s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3708cbe04b0f4b76aebeda2958a71f45 : SUCCESS in 44m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fa1b7562a444465e9685e7be95c561fa : SUCCESS in 1h 47m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/85864bfa499349a9964d5a866bf24fa8 : FAILURE in 29m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab2fecfc54394a1f94721367f77e4360 : TIMED_OUT in 2h 35m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/403cee312b1441b39b642daf3e84ac77 : SUCCESS in 42m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9e61cc3521504421928335c5692b75fb : SUCCESS in 44m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9276a4c4a9274cdc871fbfb040320f24 : SUCCESS in 41m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3844c481a85b4174b0dc5420297a408b : SUCCESS in 41m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/cab159e2ce9a4c8dae6f4dab463f6957 : SUCCESS in 42m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/368f857b79584985bc3d85053a80f967 : SUCCESS in 42m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6c6cd13720334ed890032e366256ebf2 : SUCCESS in 1h 58m 37s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/1ce5ec52290c4a69b0c76d25c915a5bf : SUCCESS in 52m 19s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c63ca3babd144cf094bb775b2a007980 : SUCCESS in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"8323d751999601a94823c82894b3b9bc1f2e2c27","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 09:54:17.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"5e97f49cf6bb823d47795b5f7560fc7eb6b39f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:32:31.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6fa33b4516a4544963dc731376aac52 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16d2fea62f46413f8628f5573b45b743 : SUCCESS in 5m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7542bb2d526a40d881710a0d373c1841 : SUCCESS in 1h 18m 10s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/ca400a85f7454837bedc1ae5a40c7202 : SUCCESS in 1h 21m 26s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/fb53704510f046bfbd155f532905a8c9 : SUCCESS in 1h 17m 47s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/646aa8aff9db4fd79a8e7b1da99016a3 : SUCCESS in 1h 18m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15a0e03ba25b4512ab5e7d2c7e0a7288 : SUCCESS in 1h 58m 55s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/57dfff267c5449a38802a667b1978b55 : SUCCESS in 2h 12m 27s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d2931e24517a462a9c087fd1a4c03d42 : SUCCESS in 1h 49m 25s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e309f46466fb41d4b6eed3151e89bfca : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b001c41293a9476d98a2431c70737af4 : SUCCESS in 46m 27s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/b22f2b32bd404489be8a9336e5a3a4ca : SUCCESS in 37m 02s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/053f5a207da043baa8cb433e99079c6a : SUCCESS in 45m 02s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/bf8aa96abb214332afe4cd172787d52f : SUCCESS in 46m 26s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cb8a014bbb1740cfa0b8ef01bc9cfd61 : SUCCESS in 1h 42m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f819def947d74563bd629c6856bd309b : FAILURE in 35m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fed16f250c5d4f6eb3ca9f5a5253b75b : SUCCESS in 2h 31m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afb10c3b98a0481788bf08da9d9b8aa1 : SUCCESS in 1h 01m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/1bc967bf3b3549cfaf5714edf757d3eb : SUCCESS in 34m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b60a75aa78a04a56bdf767c44b65cc07 : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a8358b3b798b451c8cb4cc79905bc6e2 : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9b1db59b36ba4f07814d15023dd4f689 : SUCCESS in 32m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/fd0f5c07d75441ec93c789d8b0832428 : SUCCESS in 32m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/f47e409eb4054fb393439f9143367fd7 : SUCCESS in 37m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/2f3c449335704ad48e795d61a85af8ac : SUCCESS in 38m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8cb8de1cef8745f795d056b5517f618e : SUCCESS in 1h 53m 46s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a2c90ffe2ccc42f5bd451d7ab09674cf : FAILURE in 25m 18s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/15b7810ecde14095a22f62b26d5701e1 : FAILURE in 24m 40s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6c67c6980d5b319bd19c90a2b37664901e3b6c18","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 12:38:35.000000000","message":"Patch Set 16:\n\noctavia-v2-act-stdby-dsvm-scenario-two-node passed the last run, but its duration seems really close to the max timeout","accounts_in_message":[],"_revision_number":16},{"id":"471e667a33ff75261c44e82603dad591c8bd4326","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-21 14:56:27.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"d35788baeece1a99ec64326bc555f98f351b548b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 17:21:12.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae2f9fa74bd4f718d4f710c06ed28c2 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f5820240b74c0daa5a63f8244da627 : SUCCESS in 5m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1839e6ab22c6440486064f8335879c62 : SUCCESS in 1h 28m 18s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/23266602eb964253b3b67c003a6ed3a0 : SUCCESS in 1h 27m 36s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/e0a2f1f1b0a449c684b69910b43d3695 : SUCCESS in 1h 28m 25s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/7a7eb9d7d05a4421af09049ee7ba0d1c : SUCCESS in 1h 56m 07s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/5651ee67877c4dd4b4ff678f4c92ead7 : SUCCESS in 1h 38m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57dd21c7fc2844949fd4b16debd0faf4 : SUCCESS in 1h 28m 04s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/c39f4d05e38c4280988082e48e950be5 : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/daaec98efbc64e79bd3af1f04f713770 : SUCCESS in 1h 46m 25s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c036567e4f054f838595de11381181ce : SUCCESS in 1h 49m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09cf6bd06e894b3ca83a77e8164a388b : SUCCESS in 37m 36s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/bfe2e93f2cfe493da3ce6c0118d6a179 : SUCCESS in 42m 51s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/43a436da9591471f84dd066a8d7e38aa : SUCCESS in 39m 47s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/6cb0172081ce4e6aabe28d4f03d64330 : SUCCESS in 43m 07s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/53f7f84451df48078cfca22041acd988 : SUCCESS in 1h 55m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/060374192ede4360a22abdcba53ef4aa : FAILURE in 31m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eac4d3a9a45b459985b2897cefe7177f : SUCCESS in 2h 11m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/acbc4e43545f4fc78db3fcb09af80f43 : SUCCESS in 35m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/00a8defacf4d4cfd8e0608e834b7fe58 : SUCCESS in 42m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e520938ba31a4e16948243f381349266 : SUCCESS in 36m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/62025bdfa254467aac70291a26f94468 : SUCCESS in 44m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c79b2d73de3e4e2ea9240ad6ae957f5e : SUCCESS in 48m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/8937b1acfbe34529b099bd2a09d0e44b : SUCCESS in 47m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/3852bc79b1834ca4a3f878ac8a4a6b60 : SUCCESS in 47m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/1dcb3a68185b499399b49241c51d8ba4 : SUCCESS in 42m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/92b3c5e2c5844abba9e378e8834f00eb : SUCCESS in 1h 49m 18s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5882780297074892bd7e23cbc1f61ad8 : SUCCESS in 49m 46s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/1a24eb4a072545419035153fa830f375 : FAILURE in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"2cfb572cbf77455ecb51c1f37092d2e8e5e44761","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-22 16:24:02.000000000","message":"Patch Set 16: Code-Review+1\n\n2h11m is better 😎","accounts_in_message":[],"_revision_number":16},{"id":"c3b7d311d2d6b08db25d200544d918247449041b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-26 19:41:32.000000000","message":"Patch Set 16: Code-Review+2\n\nLGTM\nStill seems odd that it is so slow. I know there is devstack overhead, but ...","accounts_in_message":[],"_revision_number":16},{"id":"9149ae1ce6daa6e56afd13e3b47b3cd1e71a0c2a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-16 09:52:39.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"4894afc53c23a44b3950269fca902348f26b1cf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 12:20:40.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05bb6d951e8457bb20641f2e9ae48d2 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a234ab8eca47ec9d8443e6954ee338 : SUCCESS in 3m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/968715225b1b476881ad4a531f81dfcc : SUCCESS in 1h 51m 36s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/0be8a9a0d83f4452806187240d96a7ef : SUCCESS in 1h 28m 35s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/3432dad6b812425bb508b405f5e164a4 : SUCCESS in 1h 12m 09s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/2c31ae43afc24f8899c39f4ffb9d8ccc : SUCCESS in 1h 10m 00s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/1c0154def3174ca1bfcd6231e4b3ccd9 : TIMED_OUT in 2h 15m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50274e385be44f89bc78aaef646c1f45 : SUCCESS in 1h 55m 46s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/6eaa4b35af7e464fa1e43577dc48309c : SUCCESS in 1h 22m 04s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d2046508cf7d48bab1dbfc09c0863df2 : SUCCESS in 1h 54m 51s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/881ceccf8c16453aa119ad9639f00368 : SUCCESS in 2h 14m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/56437369b6c94efd8307fbf874d58cfd : SUCCESS in 45m 11s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/a4154a913dd348b29c612224b3bc172b : SUCCESS in 38m 26s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/6f049621e07f4a59ae2aeee77214d2b7 : SUCCESS in 36m 52s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/186a8a209301488c88292a06c5f63558 : SUCCESS in 42m 26s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a6138816a06545ed8dc4354c39a04bd2 : SUCCESS in 1h 31m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/171c9a0d8f314b1b8f36908e912ea34f : FAILURE in 11m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b6db3ac9c71418dafce53c0dd934596 : FAILURE in 19m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/abbbb70c32c940b79f5f35708fd17e8c : FAILURE in 42m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/7568cfd987694812b13f803714be047e : SUCCESS in 36m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/68b5d839eaab4153bc6faed117305b64 : SUCCESS in 46m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5575a9b9515649568c812ba648fb9071 : SUCCESS in 38m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/58269a08128a47c5b55b2629344792aa : FAILURE in 52m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/95625b3ddd484ee282c930010a843fbd : SUCCESS in 45m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/fa5d58cbab97433a9d7730d70ec31224 : SUCCESS in 36m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/23cf41f6e6cf40d2af1d8848ccfe82aa : SUCCESS in 35m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8712f1681e4741b3ace290ff5115d9d4 : SUCCESS in 1h 42m 36s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/78d13021a00549e392d973fbda0fc64e : SUCCESS in 49m 25s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/2438cb33a99543c3b732a2924cf85909 : SUCCESS in 1h 06m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4298d372df11cfda755955a23ee4fc2ca1a3711f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-18 10:44:58.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"358ea6ebff0d6980a3a87b6d59eb9bdd4d1684b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 12:59:05.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ae45c996734b74804f257bd442e8b8 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818b6e8bcab8474fa743274440326dd1 : SUCCESS in 3m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db00995bae494f0ab53896b3c0165141 : SUCCESS in 1h 49m 40s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/3205f11d816e4379863d5bdf1816dd91 : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/02ca64b66e934ae59811171a8aea4425 : SUCCESS in 1h 06m 34s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/6151140aa41f48ba8d4e478810deced9 : SUCCESS in 1h 12m 44s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/8cd5bcdc9eca48b494b5315cd5e28b00 : SUCCESS in 2h 03m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5be501b8e12a47ffae1c46ff5947ee77 : FAILURE in 1h 50m 38s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/6afbec79a67f40f8bb274e96e6bd4d1d : SUCCESS in 1h 37m 21s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/25f3402ca12b4779a8e926b3c72311c6 : SUCCESS in 1h 50m 28s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/7b9632b1fa2e42b5b5c43dc9d0009953 : SUCCESS in 1h 27m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b74566cf8ef405cbe79b64eb4004e8e : SUCCESS in 44m 04s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/7c7d49c48daa41bd840dcc3d196a65bb : SUCCESS in 40m 13s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f47d22f85665425f9faddc25e1a78589 : SUCCESS in 39m 24s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ae87d112f99a4b54beffa90429a3c7bf : SUCCESS in 55m 59s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1a48ce0c44db41fda7007193a1c511e0 : SUCCESS in 1h 48m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/56f2225ebdf247da93af7da7025e293e : FAILURE in 11m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d850ea2eb814f8f9bed18070cbc763f : FAILURE in 19m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4bd961f0ab54987a0831cc31c13a88a : FAILURE in 38m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/08b84fa8957749a1984f3f8f8bd1c6a3 : SUCCESS in 50m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5aafb1f8b80a4023b652e49f6f2582ed : SUCCESS in 56m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/63c70e87751e4239a4cf4d29820a2fae : SUCCESS in 44m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ab48e3a75bb64227b7a41ddd3d8584a5 : FAILURE in 54m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/46ff286b06dc4b8bae73b104f5e7c2fd : SUCCESS in 38m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/3fdf6f8ea76641c08fbab55d28879378 : SUCCESS in 37m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/f65be858341a450db57273e6f454eb0b : SUCCESS in 32m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e140182516154dbdb6edde2ae383677e : SUCCESS in 1h 25m 09s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6e244f0a2cdd4224ae79dd6ecdcac50d : SUCCESS in 50m 31s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/98e40dfa3a76468590d77d80fe230eb9 : FAILURE in 1h 08m 03s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0d3b67680d773ab27a66e5a8dd6a4360a0e771b1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-18 19:15:54.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"0ccd81afaddd614f08f162e256d2001d8206997b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 21:31:00.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d4f1d9d9ab4dff98c01849bf58c107 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/999ffba29e1c46cc8970e70b87854045 : SUCCESS in 4m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03a3503853924371af2647626097fe05 : SUCCESS in 1h 37m 42s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/60514b9bab314af6860aada2dc40f463 : SUCCESS in 1h 24m 47s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/ae1546701fbc44bea57d6100dbd69b0b : SUCCESS in 1h 32m 00s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/6f2079731c9a4ced8a721b9b4d4fa2a2 : SUCCESS in 1h 34m 34s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/086f96b2ab704ab4b4b80b019eb80c4c : TIMED_OUT in 2h 13m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4285100afab4f31ae284673ef4d2558 : SUCCESS in 1h 59m 23s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/3f28ad1098f44c8cb15cb479d7c3afaf : SUCCESS in 2h 03m 57s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/4060b597e6e449ccbe1ad526d14aafcf : SUCCESS in 1h 43m 53s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/be664af2132f4dc9a857de09f5a54ace : FAILURE in 2h 10m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bb0b7a564dd9401cb1586d9ef41e5091 : SUCCESS in 44m 02s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/4a6d600243514c7eb5cd9d0e4fd54ac7 : SUCCESS in 40m 26s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f2ccda89b7044bffbba91b1cde4701a8 : SUCCESS in 41m 02s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/7b2769e680b14c20b4d26ec876701056 : SUCCESS in 42m 55s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d8a3939efd0d4b89a8f336f28996c828 : SUCCESS in 1h 53m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f056889f5f78432abfceb0337fd933bb : FAILURE in 9m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/572a431869594166b2dcfe230ffb4fbf : FAILURE in 1h 36m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf2415f67816436f98c514e5285f2e82 : FAILURE in 35m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/0165d8be2dc44197a054403fb4392eb1 : SUCCESS in 40m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8ffaee45448242f2a1c87b438a9b5c09 : SUCCESS in 42m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e86dc1ad11c5433cbd233f4344dd1336 : SUCCESS in 42m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aaef96f9979d4eb9b4d57f075172689f : FAILURE in 49m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/71b50f42197842028fe3833bd93fa7e1 : SUCCESS in 39m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/1f849c9b415b468b85e035b8d6be3394 : SUCCESS in 42m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/484d550f324d47948396cbde88c097d9 : SUCCESS in 39m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/55acd1f292ad49febc5291b0059ad461 : SUCCESS in 1h 31m 08s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/9c6af485153c4e758acbf80fb773134e : SUCCESS in 39m 27s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5873b5538c464811a406ccce86235383 : FAILURE in 46m 13s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"09d1f875961d9db3375b3db8cacb508bc20df70e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-19 09:25:50.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"a944db911d4118f96f7e1733ef87d134e8348ee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-19 11:14:38.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51f28f10c7c4b888042e5fececffbd1 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1af2f5324464bef87b503b2a7c4353f : SUCCESS in 3m 40s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f5070813cb74574bea81db59fcd0e5d : FAILURE in 1h 46m 46s","accounts_in_message":[],"_revision_number":18},{"id":"4f8c7079d089f057eca74ab69c6c83a5b710d719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-19 12:35:59.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"718211db48e4c650765cc720dc9691dfb02e30ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-19 13:59:40.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/601d646427994bb49762b9c204b5cf60 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc763a9172f04ce48daed6e6725f0e16 : SUCCESS in 3m 50s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea2e3e28bee9426a97819efbed51b6c5 : FAILURE in 1h 21m 51s","accounts_in_message":[],"_revision_number":19},{"id":"6b7392ea589c79b02e1fe80a49f742cfb5707bc5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-21 13:29:04.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"a5e9f9b9ce6ff42c033fc67511e288421c52a8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 15:17:31.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b563905e62714ad1a35d3a4dc02a782b : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740c96eaa2ae4e039bd36ccf8e168dd4 : SUCCESS in 4m 23s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73b3184323864e3eae05c0a07684853b : FAILURE in 1h 46m 06s","accounts_in_message":[],"_revision_number":19},{"id":"199a181f00aa6b59afd5d352c3a3588c1676c07f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-21 15:37:59.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"1f64eb7f0ce2ee847d25ce477fa724cc7a391e7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 17:00:30.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dba9d147b2a47c6a56be9b52c5d76bf : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3113bf2a7cfd42bd82890e31412039b3 : SUCCESS in 4m 37s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b23eecf27a84324ba161867a9699a15 : FAILURE in 1h 21m 08s","accounts_in_message":[],"_revision_number":20},{"id":"a0e4f7ea63b00768ed2a89452001d4244d454339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-23 06:37:43.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"2d5c2f4ffffc6c9a9b294abff9bc8e0db4b588bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 08:08:00.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7135f7cb86194558a897b825d0ea4af0 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a107a81392e4437988293f2707cf0246 : SUCCESS in 6m 08s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/03b514961be04d579608be08c17aac53 : FAILURE in 1h 28m 50s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a71f9452c19438194836a7df86d80ca : FAILURE in 1h 23m 35s","accounts_in_message":[],"_revision_number":21},{"id":"ed52cddd3704343f26d7b354b4b8c35e3f284159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-30 07:28:15.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"24d7beca027a15caa44e101b62f01b931564b4a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 08:31:01.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344d7d32d1aa4df8a402b122530c162c : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b63a711f1cc041d5b5cd388919c6516c : SUCCESS in 5m 47s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c6658b0099e4dfb96165154b342bb12 : FAILURE in 1h 01m 05s","accounts_in_message":[],"_revision_number":22},{"id":"52c8720d9a7aa00e3ba432c4f486699494d86203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-30 08:57:03.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"b57f60b6b5acb0fb263c612b18237cfdc7f84640","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-30 09:33:08.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"cfa87a29882439ddc416f7f1596dc1107c3c11ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 11:37:59.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664bdc0601d34ec5a5e25f6f84c1d239 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3b4a536f5e40edb66eae4d723ed9a2 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/756123836d6f4fd2956a153e32706b20 : SUCCESS in 2h 01m 02s","accounts_in_message":[],"_revision_number":24},{"id":"f3ff4966fe467bea84c56ba0fb4a51389a449bff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-30 11:55:26.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"368c01b31716bb1467a69dec462da279aff37734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:35:09.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07586a5be20d4fab80e28c9469a534aa : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4080c7cca6484a7b9df1814a9c18fa74 : SUCCESS in 5m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1df1524fd9604b22b43cc9ae5c42232b : SUCCESS in 1h 47m 35s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/b28f99606aab447aa3c04cf8639831ef : SUCCESS in 1h 32m 44s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/022d8a2d09c043ffa973e5790f6f283c : SUCCESS in 1h 44m 17s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/0feaa1f278fd4a26bc8a9a2c59221b81 : SUCCESS in 1h 06m 03s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/60bf5424c3cd44afab5dfa209b27f46b : SUCCESS in 1h 57m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21e29c9382cb4afc9499ef0179d214b6 : SUCCESS in 1h 54m 41s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/8e214af26d0147b3b9ed2cb84ce65de2 : SUCCESS in 2h 00m 45s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/185d0db6b21f44e79381568ac039a49f : SUCCESS in 2h 02m 40s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/59c2953d693a4db79193b4d35b362ca1 : SUCCESS in 2h 12m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9cb708793c3a409a9aa0f5ce94230e4d : SUCCESS in 44m 38s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/839d4432e7a84471900719c091831ddf : SUCCESS in 46m 28s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f50c89275f96401a9c570b859b22b783 : SUCCESS in 46m 48s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/c1ada54832bc4988a35b658b07c51d4b : SUCCESS in 45m 49s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1721951514fc412695232f60b60b9cf0 : SUCCESS in 1h 56m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8bdece892ab84ba48357afac4d2e6a1d : FAILURE in 13m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00246e25eeb54dbb91590f5b1308dd66 : TIMED_OUT in 2h 36m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/478a54937e394e02b0e88b1498c2d8cc : SUCCESS in 42m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/2d59a512c88043e5a2ad6be5d4820c05 : SUCCESS in 44m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8f8f83cb9a794fcbb8c3d913835957e2 : SUCCESS in 45m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d2bbfcfb6df24f0d97093bacfef33c55 : SUCCESS in 45m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/361359ef9a7f4aee8593ec1d01eaf886 : FAILURE in 52m 52s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/3a98638552074d17b409eefea5dbf8b3 : SUCCESS in 42m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/c055ebbe6b244deea0036e9736b99138 : SUCCESS in 43m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b088f43f1b9f4ea2bde75a051ed586b9 : SUCCESS in 44m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/60e37f4384774a068b09a4c155758810 : SUCCESS in 1h 53m 38s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/602642cb80de4ddeb93385070ac6fd74 : SUCCESS in 50m 14s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/d80c4b0484754874ae43b28d65d427f3 : FAILURE in 51m 55s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"38d94dcc35bcfc018cedf7b68efee16006b49e56","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-07-01 15:15:09.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"2b02a8fc01eecb73379bbd0e913f5effaf30e79f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 17:53:10.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367310dde5ed41a2a8f630b51079ba54 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a415841fa8164172a5aca71323a3e019 : SUCCESS in 3m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/664b3e3a3f554b21b242afe044d242e2 : SUCCESS in 1h 36m 22s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/4b0a689f350340cc8f1cd03a0c64b0c2 : SUCCESS in 1h 52m 47s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/d7ee3dc87d2a4b008517a7d2b0254c59 : SUCCESS in 1h 32m 38s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/a9220c0c5a0d4cddbac8a83dcf3b889d : SUCCESS in 1h 04m 20s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/2f59a22ac00d4fa6b40f5fd98c461cee : TIMED_OUT in 2h 14m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/410e867c790e4be7869f0a6f48bebd73 : SUCCESS in 1h 43m 39s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/ef7a06e2a04d437d816a5253fcf62807 : SUCCESS in 1h 40m 18s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/821292804f7a400d95782bfce8b48be8 : SUCCESS in 1h 47m 14s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1e146f6f0fd04ccdaac7d35b4eefa902 : SUCCESS in 2h 09m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2c879d46a013416f9d9f1450810d6788 : SUCCESS in 47m 35s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/97fe0a585d404023b6e9667533850187 : SUCCESS in 42m 24s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/db954508be684d82ac4c92917add7a34 : SUCCESS in 47m 32s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/99ab523c19e14f07bd8a972c78954839 : SUCCESS in 47m 16s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bc5f8f9380a84d04add50d407595b7f7 : SUCCESS in 2h 12m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/36c42dbe220648ee87091df9e678c662 : FAILURE in 11m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1688036739c4134b06d11403622c77f : TIMED_OUT in 2h 34m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfaac75277034c7cbaed2e315d53a873 : SUCCESS in 38m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/2027011049824c30a6f7171364597a45 : SUCCESS in 41m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d4e93bfc9a0a4a9294f9c9849e2b9124 : SUCCESS in 38m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ff954a39a829450b9c2ea2cc4f8ea398 : SUCCESS in 32m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d2f2a3fd84c740edbe8d59f7c5cfedef : FAILURE in 49m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/d275f29468bb46c1b1c42d5dea47d487 : SUCCESS in 33m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/7e91aacc5c1e49678676edbb47ec18fc : SUCCESS in 36m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d691877213d448849800bcf3f4d34909 : SUCCESS in 32m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3aba2e4acb884270bcc0bbcc693eb226 : SUCCESS in 1h 29m 43s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b64c9e15c40444ef97c57f799ced4a4e : SUCCESS in 50m 45s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/935d7e3522034379a7a698a869fe6161 : FAILURE in 54m 40s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"1542fb960c390f832ed189f55962adab446e51bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-07-05 10:29:24.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"be967a696bd9e0f653931302168bc4fee39a3dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 12:52:03.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d99d181ae194284a70d6371d77e0538 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee8d234bde9488da4dfd2243cb60214 : SUCCESS in 3m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49f4b914954b4176a447319d4fff9352 : TIMED_OUT in 2h 13m 14s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/cc4e88fbfa314ae5ac8ea2f52ab057bc : SUCCESS in 1h 44m 17s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/4db8068c16d248cb9f5022c79bc6ad72 : SUCCESS in 1h 02m 53s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/e4d5fe23407e40f38986207f51da78de : SUCCESS in 1h 05m 20s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/d5a8201a4c8b4dacb523db6052b31331 : SUCCESS in 2h 10m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/639421e8220e488f8ab2149289372c4a : SUCCESS in 1h 44m 52s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/0cfd76e4cdbb4a36b3252e6f6dcf1946 : SUCCESS in 1h 36m 57s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a0143795a64349c992ea39f44a07d045 : SUCCESS in 1h 32m 55s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ab86cdc0e5b642dc9962efb6e6ac2ea2 : SUCCESS in 1h 32m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7724f00b76aa4bbb8d8d691f489a0364 : SUCCESS in 41m 45s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/a8e2837903954dfdbe0449f6ceb0fc4f : SUCCESS in 44m 01s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/dac90cc858ca46e2b15db87165603b44 : SUCCESS in 37m 02s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/b45ec095e0824cb09d06ee87b5804d7a : SUCCESS in 41m 44s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6c5f7e44871d4b5cb387f233451c2ce5 : SUCCESS in 1h 53m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2a5636e6e09748efb0976e45c0c7b4b1 : FAILURE in 10m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74a4379c96824322ac4a01e9bbc5412c : SUCCESS in 2h 20m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/354e7b3a9bf44442b70f5ebe9177784b : SUCCESS in 1h 58m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a4e010580cb46bfa44234f59c347294 : SUCCESS in 37m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/867872339b9b45198c10dc3a4e48c12a : SUCCESS in 44m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b54b498918104d5b86ac2efc937eec4b : SUCCESS in 45m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8262bd40216b45139a3d3fd62489a626 : SUCCESS in 39m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f1d6f5eca5814a969c17786770a7b9fc : FAILURE in 46m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/d4639a06cbbf4e3dae5738c1f4475c99 : SUCCESS in 32m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ece47473f2ea41b7b12420e61f8ba8d7 : SUCCESS in 44m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/64106264e08445dd84aa1b346018cbc8 : SUCCESS in 43m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/afd68f9a47094692b31878eb891fb728 : SUCCESS in 1h 23m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/aaba744317d34882be272c7b805f7068 : SUCCESS in 50m 56s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/daaba7fe315b4705946b293783183834 : SUCCESS in 59m 45s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"9d35a6747727b85d38113ee91dcfe9b8c41f48fd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-07-19 20:32:36.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"8226c61a8fd5f2772cb34d23b4c24dbcc99b8c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 23:04:09.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe799ae9c88f4ff9b5c30673ed10b9ca : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ba334e09e74806ae71bb4533c73014 : SUCCESS in 3m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50e576e02f854c20985c1ab4033a3a1e : SUCCESS in 1h 44m 20s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/3ce40311b41e408391504c9990b438aa : SUCCESS in 1h 12m 02s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/00f97ffe2a06483c935eb27b9cd2653f : SUCCESS in 1h 45m 29s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/bf828c23ac474d4997a713c16f84fa8c : SUCCESS in 1h 00m 32s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/a4777faf127f4671b78c368eb5e78be3 : SUCCESS in 1h 56m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1b37b938cbe4ea1b64014d5959ddbe6 : SUCCESS in 1h 19m 29s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/439e95e5cf0447a28786445ba888392e : SUCCESS in 2h 08m 13s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f5c60d2720e7403ba82d8a5f3eb8a4e5 : SUCCESS in 1h 27m 33s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/15ca4c2a213b4dc78955d69f77f5b51e : SUCCESS in 1h 44m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c48fc788f34f47e58f5553f30b96d32c : SUCCESS in 34m 47s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/999d3c9e142648fdbc2dd4db898675db : SUCCESS in 55m 00s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/160aa5cfbc064262b27c9bd88eb82b71 : SUCCESS in 33m 48s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/73919e2b31e54555afc2894325664e8b : SUCCESS in 39m 29s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/781894a861cf424b9defd3e7bc8999eb : SUCCESS in 1h 33m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5f0236aa0335445d9d327d463d593939 : FAILURE in 15m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5c1998b5f51d4f74959a37defa347473 : SUCCESS in 1h 38m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bdac1e64e45b4997bb01e9c680db4e0b : SUCCESS in 2h 29m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f2b9411ed6c4b6991496a9545bed34e : SUCCESS in 33m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/ce921d4f7ab44e7095b35827baeb11c8 : SUCCESS in 38m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7d4e0ffe64594963b2f9faee2a7f9d5f : SUCCESS in 31m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/586e204f083d43898796ac94ab47a1ed : SUCCESS in 1h 00m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/54cedfa3e16547bfac76f5b554144068 : FAILURE in 40m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/41b31594cc594b74bfd4df3ca38b097b : SUCCESS in 37m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/fdd0697acfc94a4189e3c5e668086f62 : SUCCESS in 33m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/62d89a6cdb0e4ccfa2dc73b0b7ba8103 : SUCCESS in 44m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f99e8ac8b2e74adb8a348f302a96b75d : SUCCESS in 1h 23m 25s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f222c4943e354620add0e5ffdc47481a : SUCCESS in 46m 09s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a26f86fb32ec42e5b33954606bf8f1c8 : SUCCESS in 53m 01s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"dc59c3bb6b595550b6cf3abd00f444162a150a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-07-20 05:18:51.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"2fb8b01f8b3cd2f8d7870b8590c58aadfe11516f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 08:04:53.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9812f379fd444f4bb3d7f119f93b7f13 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71f6c2071ab48a285566722c7114a6e : SUCCESS in 3m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e45e934075b497db11767e438c85fc4 : SUCCESS in 2h 05m 57s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/b26af172066b465ea14009b72c6d77c7 : SUCCESS in 1h 32m 06s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8ea2fd5d68cf42ed97cbb72313b50a47 : SUCCESS in 1h 02m 29s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/95de1b8fe9714efb9d79af1886d0ce58 : SUCCESS in 1h 29m 06s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/dd22312650974e099206eb2ace47426e : SUCCESS in 1h 54m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f785ee1f29a54c978ab4a445d0b80d41 : SUCCESS in 2h 07m 14s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/68c5e8a8434044fd88803798e9999b40 : SUCCESS in 2h 24m 05s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d1fe6b8f336444de8f69166960b327a7 : SUCCESS in 2h 18m 49s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ddff5458db3945219c6f5792af9add03 : SUCCESS in 2h 22m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3889e94f181149c3bea6f5493183a5b6 : SUCCESS in 37m 35s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/58a12220c73449c187e1a236085158bd : SUCCESS in 1h 03m 31s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/2110a4b396bf4c4a99f8dc3413f0cc6a : SUCCESS in 39m 13s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/d33ab7261dd14a51ade6861a3c0cbc41 : SUCCESS in 59m 11s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/25b32c85753249448ac10b9a4adf4869 : SUCCESS in 2h 11m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9e2795d58e9d43e08fc48a342dd2bbe7 : FAILURE in 9m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c8a0243123e42ecb6b42dc88d5a2da7 : SUCCESS in 2h 40m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edd28342e4df4badbb3c0bd20f2bb4ed : SUCCESS in 57m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/9cdd0a3ff282440b860f63f1f0ae529c : SUCCESS in 53m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/15a2fd0113b0449e963fda3d73af1484 : SUCCESS in 32m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/eb1393e9aa9c4ebd84ada440f1e03e26 : SUCCESS in 38m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b744ec33e827401598a6ad3b215535fd : FAILURE in 46m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/509bcd825b164a918bd51f5ec5bf5a66 : SUCCESS in 54m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/44870254032b485b8299ae2eafad6baf : SUCCESS in 53m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d24781329ac945fdbb8aa14995173d2a : SUCCESS in 36m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/913725e08b784202927a92d4e3b86ed2 : SUCCESS in 1h 53m 24s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/869d5e0cba934db38be9dfa7c80ea3c9 : SUCCESS in 44m 54s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/19c9af79f8a4421b8868189beb60e48f : SUCCESS in 46m 16s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"26fd1e4a3f39564101564a3a98d3317ff4b35d53","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-21 16:53:54.000000000","message":"Patch Set 27: Code-Review+2\n\nAwesome, thanks Greg!","accounts_in_message":[],"_revision_number":27},{"id":"8260f7e4f83a3b09fde98311b883d97402c0c723","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-21 19:21:22.000000000","message":"Patch Set 27: Code-Review+1","accounts_in_message":[],"_revision_number":27},{"id":"217925e1d7f27d436ba641c3f234a015f163596a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-08-17 16:56:43.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":27},{"id":"5a6eac3332630a66743a011e4897a732bf1c552b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:57:35.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"17f777709351d63b8f2bb7fc5fa53d3eab52fe98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 19:06:56.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9338b40bffca4ce0aefe5893e5597935 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c6349182fd461fa242ea2cfe3d9c80 : SUCCESS in 3m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d85686b89274e618818b41f88f118fa : SUCCESS in 2h 05m 38s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/dbf46307fc6f4b67a06b451439cea362 : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/c798402aac7e45539ee3acf13183a798 : SUCCESS in 1h 26m 03s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/f1bf824c4017408cbb2689b380366e3a : SUCCESS in 1h 02m 37s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/91f322e9a4404765a10bdb7fbc673d49 : SUCCESS in 1h 30m 40s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/a849df9d33ba454db51138f1e1c3a3ad : SUCCESS in 1h 53m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21c5d8ba413f49f2900cc66db6bb0615 : SUCCESS in 1h 55m 52s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/9ee42f28e80241e6afe2920fa86b7a36 : SUCCESS in 1h 12m 16s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/9a6e6a744eef4ddab7fcaf6cf62a26c2 : SUCCESS in 1h 43m 37s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/dbccc98ab3a8443cb180501f4791f714 : SUCCESS in 2h 01m 16s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a22134c36b99462da5e21dbeaa5a6d5b : SUCCESS in 1h 31m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7204c000d5c14cf0a56b0bfd9678c41a : SUCCESS in 49m 18s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/7ea9418b6c384f31bfb80afac60c951a : SUCCESS in 41m 15s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/3861c74072374af1bbb802f16dcf63ca : SUCCESS in 39m 37s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/ecb9a740aeaf48c9b14b3fe3ae4f0c89 : SUCCESS in 33m 43s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/73675e43721349b886e661d95182c19a : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":27},{"id":"7567adcd4213e6a5e544a173859292097c040b07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 19:06:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":27},{"id":"3cea490bc094fdbc9ffecff3bd987dcaa914e2cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 19:11:26.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f070c7fb1541578e4694a96dc5144b : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"5cf1f61fbffeb901955f10469babce0b9905a093","revisions":{"20c02d230de70dbc9b44050edf3bb27b78671a1d":{"kind":"REWORK","_number":1,"created":"2021-02-03 13:34:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/1"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"subject":"DNM WIP Trying to fix two-node job","message":"DNM WIP Trying to fix two-node job\n\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/20c02d230de70dbc9b44050edf3bb27b78671a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/20c02d230de70dbc9b44050edf3bb27b78671a1d"}]},"branch":"refs/heads/master"},"e52f8ba64543c5febd837ac63cf164df61ed025b":{"kind":"REWORK","_number":2,"created":"2021-02-03 16:36:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/2"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 16:32:46.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e52f8ba64543c5febd837ac63cf164df61ed025b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e52f8ba64543c5febd837ac63cf164df61ed025b"}]},"branch":"refs/heads/master"},"1962fe269d8619655e3e9ce104491b62be99b167":{"kind":"REWORK","_number":3,"created":"2021-02-08 08:16:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/3"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-08 08:16:34.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\n\nThe \"zuul_copy_output\" option under host-vars overrides base\n\"zuul_copy_output\". This means only barely a handful of log files are\ncopied. Fix this by using \"host_copy_output\" option to combine both log\nfile lists.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1962fe269d8619655e3e9ce104491b62be99b167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1962fe269d8619655e3e9ce104491b62be99b167"}]},"branch":"refs/heads/master"},"64e86b2f7d0ccbd8d128d3a7497d074b145c5e65":{"kind":"REWORK","_number":4,"created":"2021-02-08 10:32:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/4"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-08 10:28:11.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\n\nThe \"zuul_copy_output\" option under host-vars overrides base\n\"zuul_copy_output\". This means only barely a handful of log files are\ncopied. Fix this by using \"host_copy_output\" option to combine both log\nfile lists.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64e86b2f7d0ccbd8d128d3a7497d074b145c5e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/64e86b2f7d0ccbd8d128d3a7497d074b145c5e65"}]},"branch":"refs/heads/master"},"2e7f93f4636a47045f94ec9f8dd36a4c411ceae7":{"kind":"REWORK","_number":5,"created":"2021-02-08 17:42:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/5"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-08 17:42:19.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\n\nThe \"zuul_copy_output\" option under host-vars overrides base\n\"zuul_copy_output\". This means only barely a handful of log files are\ncopied. Fix this by using \"host_copy_output\" option to combine both log\nfile lists.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2e7f93f4636a47045f94ec9f8dd36a4c411ceae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2e7f93f4636a47045f94ec9f8dd36a4c411ceae7"}]},"branch":"refs/heads/master"},"c258beec84dd23b2b8d0e4a99d8d3f1aa3d5e62b":{"kind":"REWORK","_number":6,"created":"2021-02-08 20:24:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/6"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-08 20:24:27.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\n\nThe \"zuul_copy_output\" option under host-vars overrides base\n\"zuul_copy_output\". This means only barely a handful of log files are\ncopied. Fix this by using \"host_copy_output\" option to combine both log\nfile lists.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c258beec84dd23b2b8d0e4a99d8d3f1aa3d5e62b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c258beec84dd23b2b8d0e4a99d8d3f1aa3d5e62b"}]},"branch":"refs/heads/master"},"357e9ce595f571d7e85284f054bec44b4781ccc9":{"kind":"REWORK","_number":7,"created":"2021-02-09 06:24:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/7"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 06:24:44.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\n\nThe \"zuul_copy_output\" option under host-vars overrides base\n\"zuul_copy_output\". This means only barely a handful of log files are\ncopied. Fix this by using \"host_copy_output\" option to combine both log\nfile lists.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/357e9ce595f571d7e85284f054bec44b4781ccc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/357e9ce595f571d7e85284f054bec44b4781ccc9"}]},"branch":"refs/heads/master"},"e4391f89e635ed93799b2583fa0f7b3e184fcb7d":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-02-09 07:48:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/8"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 07:47:54.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e4391f89e635ed93799b2583fa0f7b3e184fcb7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e4391f89e635ed93799b2583fa0f7b3e184fcb7d"}]},"branch":"refs/heads/master"},"7e72b4db3ec195c77bff8d0d740b6085cfc0b977":{"kind":"REWORK","_number":9,"created":"2021-02-09 08:41:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/9"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 08:41:20.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7e72b4db3ec195c77bff8d0d740b6085cfc0b977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7e72b4db3ec195c77bff8d0d740b6085cfc0b977"}]},"branch":"refs/heads/master"},"3356ebca5591c5ebe8d54d81cba065fac88fedf9":{"kind":"REWORK","_number":10,"created":"2021-02-09 12:25:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/10"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 12:24:34.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3356ebca5591c5ebe8d54d81cba065fac88fedf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3356ebca5591c5ebe8d54d81cba065fac88fedf9"}]},"branch":"refs/heads/master"},"21c581091ec997f516ca15f2e5b95ed4d8f82e5c":{"kind":"REWORK","_number":11,"created":"2021-02-17 21:02:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/11"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-17 21:01:49.000000000","tz":60},"subject":"WIP Fix two-node job configuration","message":"WIP Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/21c581091ec997f516ca15f2e5b95ed4d8f82e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/21c581091ec997f516ca15f2e5b95ed4d8f82e5c"}]},"branch":"refs/heads/master"},"cb71efdfec129260e3385feb641519360dcae121":{"kind":"REWORK","_number":12,"created":"2021-02-18 06:45:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/12"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-18 06:44:41.000000000","tz":60},"subject":"WIP Fix two-node job configuration","message":"WIP Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cb71efdfec129260e3385feb641519360dcae121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cb71efdfec129260e3385feb641519360dcae121"}]},"branch":"refs/heads/master"},"f5ccc17d5a306c3b607ea9b74184be9bbf610f3c":{"kind":"REWORK","_number":13,"created":"2021-02-18 08:00:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/13"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-18 07:59:53.000000000","tz":60},"subject":"WIP Fix two-node job configuration","message":"WIP Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f5ccc17d5a306c3b607ea9b74184be9bbf610f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f5ccc17d5a306c3b607ea9b74184be9bbf610f3c"}]},"branch":"refs/heads/master"},"e0017c8b182693d5b1f6f675e381b127c243e920":{"kind":"REWORK","_number":14,"created":"2021-02-18 13:50:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/14"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-18 13:50:05.000000000","tz":60},"subject":"WIP Fix two-node job configuration","message":"WIP Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e0017c8b182693d5b1f6f675e381b127c243e920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e0017c8b182693d5b1f6f675e381b127c243e920"}]},"branch":"refs/heads/master"},"5f572c7c899261d5a4c15744cfb90c59581697ab":{"kind":"REWORK","_number":15,"created":"2021-02-19 15:33:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/15"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-19 15:33:29.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f572c7c899261d5a4c15744cfb90c59581697ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f572c7c899261d5a4c15744cfb90c59581697ab"}]},"branch":"refs/heads/master"},"c97b0dc51ece45387678aca4beee1f0d846e6f50":{"kind":"REWORK","_number":16,"created":"2021-03-19 10:55:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/16"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-19 10:55:12.000000000","tz":60},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nDepends-On: https://review.opendev.org/774596\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c97b0dc51ece45387678aca4beee1f0d846e6f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c97b0dc51ece45387678aca4beee1f0d846e6f50"}]},"branch":"refs/heads/master"},"13e2d8f3c643d87e1822d3089bfa5f4d6d74335c":{"kind":"REWORK","_number":17,"created":"2021-06-18 10:44:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/17"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-18 10:42:12.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/13e2d8f3c643d87e1822d3089bfa5f4d6d74335c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/13e2d8f3c643d87e1822d3089bfa5f4d6d74335c"}]},"branch":"refs/heads/master"},"df0d49045001722de56de1bf39c57b048fd3c63b":{"kind":"REWORK","_number":18,"created":"2021-06-19 09:25:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/18"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-19 09:25:39.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/df0d49045001722de56de1bf39c57b048fd3c63b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/df0d49045001722de56de1bf39c57b048fd3c63b"}]},"branch":"refs/heads/master"},"a09a79ec3b6277fbf2d39d4d6e6786f0207bd863":{"kind":"REWORK","_number":19,"created":"2021-06-19 12:35:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/19"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-19 12:35:41.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a09a79ec3b6277fbf2d39d4d6e6786f0207bd863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a09a79ec3b6277fbf2d39d4d6e6786f0207bd863"}]},"branch":"refs/heads/master"},"89ea3bdc3090262a27b702629345bc1e1e89c835":{"kind":"REWORK","_number":20,"created":"2021-06-21 15:37:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/20"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-21 15:37:53.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/89ea3bdc3090262a27b702629345bc1e1e89c835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/89ea3bdc3090262a27b702629345bc1e1e89c835"}]},"branch":"refs/heads/master"},"814a048e629691416adf660e534b8cef4dd02653":{"kind":"REWORK","_number":21,"created":"2021-06-23 06:37:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/21"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-23 06:37:34.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/814a048e629691416adf660e534b8cef4dd02653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/814a048e629691416adf660e534b8cef4dd02653"}]},"branch":"refs/heads/master"},"047b763b3e365303158c1f152de6838920a50569":{"kind":"REWORK","_number":22,"created":"2021-06-30 07:28:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/22"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-30 07:28:05.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/047b763b3e365303158c1f152de6838920a50569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/047b763b3e365303158c1f152de6838920a50569"}]},"branch":"refs/heads/master"},"637beb67c266f78e0390c962d9b02da6c8b7b414":{"kind":"REWORK","_number":23,"created":"2021-06-30 08:57:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/23"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-30 08:56:44.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/637beb67c266f78e0390c962d9b02da6c8b7b414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/637beb67c266f78e0390c962d9b02da6c8b7b414"}]},"branch":"refs/heads/master"},"ad5e7b7e2660c8c018931e60b226441a5cf40af1":{"kind":"REWORK","_number":24,"created":"2021-06-30 09:33:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/24"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-30 09:33:01.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ad5e7b7e2660c8c018931e60b226441a5cf40af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ad5e7b7e2660c8c018931e60b226441a5cf40af1"}]},"branch":"refs/heads/master"},"c3d4a034d9e0529cd6ee8f23dea1fad02d94b74d":{"kind":"REWORK","_number":25,"created":"2021-06-30 11:55:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/25"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-30 11:54:50.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c3d4a034d9e0529cd6ee8f23dea1fad02d94b74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c3d4a034d9e0529cd6ee8f23dea1fad02d94b74d"}]},"branch":"refs/heads/master"},"99e41e95842b867a772302ee64d0c89b534fb7ed":{"kind":"REWORK","_number":26,"created":"2021-07-05 10:29:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/26"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-07-05 10:29:05.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/99e41e95842b867a772302ee64d0c89b534fb7ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/99e41e95842b867a772302ee64d0c89b534fb7ed"}]},"branch":"refs/heads/master"},"5cf1f61fbffeb901955f10469babce0b9905a093":{"kind":"REWORK","_number":27,"created":"2021-07-20 05:18:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/773888/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/88/773888/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/88/773888/27"}}},"commit":{"parents":[{"commit":"cac3eefc44afca458692e9dadb539c0cb17dc9e4","subject":"New test: test_tcp_and_udp_traffic_on_same_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cac3eefc44afca458692e9dadb539c0cb17dc9e4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-03 13:34:18.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-07-20 05:17:35.000000000","tz":120},"subject":"Fix two-node job configuration","message":"Fix two-node job configuration\n\nTwo-node job was failing because devstack tried to compile the\ntest_server.go file on controller2.\ncontroller2 doesn\u0027t have tempest installed, but octavia-tempest-plugin\nwas installed. The compilation step was failing and is not necessary\nthere.\n\nFix octavia-v2-act-stdby-dsvm-scenario-two-node definition, overrides\nwere not applied.\nFix the \"zuul_copy_output\" section.\n\nMove to ML2/OVN backend.\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: I24976e93e9ea8f7f50c9da014c2627965512893c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5cf1f61fbffeb901955f10469babce0b9905a093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5cf1f61fbffeb901955f10469babce0b9905a093"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
