)]}'
{"id":"openstack%2Foctavia-tempest-plugin~776817","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I6fad03f5a89c213562918ca258884aac34ba7ce7","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"octavia_new_rbac","hashtags":[],"change_id":"I6fad03f5a89c213562918ca258884aac34ba7ce7","subject":"Update tests for scoped tokens and default roles","status":"MERGED","created":"2021-02-21 01:58:21.000000000","updated":"2021-04-15 18:57:40.000000000","submitted":"2021-04-15 18:49:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"776817-octavia_new_rbac","meta_rev_id":"0d06edbba50239d8fa254c6f3bcfb834978b4e65","_number":776817,"virtual_id_number":776817,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 18:49:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-04-10 02:12:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-04-09 13:27:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-09 08:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-04-10 02:12:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-21 04:25:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 00:14:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-03 15:22:43.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-04-09 20:17:16.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2021-04-10 02:12:00.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"294eb84d6bd7976dbbaa98be9bb273337d79de89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-21 01:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fa78bc19612fea7f2c0174d7e7ba348be6f6b42","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-21 01:59:07.000000000","message":"Patch Set 1: Workflow-1\n\nVery WIP","accounts_in_message":[],"_revision_number":1},{"id":"5eeb996f17f649143aaf64e397a9a356422b036f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 04:25:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85c17148e68416da1e8c608246534b1 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a08e31987343b98a720fc9548dc785 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/279badf4c902428d9265270762646a77 : SUCCESS in 3m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/83b0067d85074eb8947935ba57b7a535 : SUCCESS in 1h 07m 44s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/3b180464787d4d578a67dc7d627d2293 : FAILURE in 1h 07m 03s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/cc514d9a1c204a7d9e1f05ce23977cd6 : SUCCESS in 1h 09m 00s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/3beeb30d1bc647d28f7f325bc07587de : SUCCESS in 1h 05m 40s\n- octavia-v2-dsvm-noop-api-stable-stein https://zuul.opendev.org/t/openstack/build/b99a91aae080446e8563b14e78a430aa : SUCCESS in 53m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d10412b23354024853ba482b2b2cb76 : SUCCESS in 1h 32m 52s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0a894e7ac0f34e97ad4c763811de2425 : SUCCESS in 1h 44m 13s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a2a3884afe9d4f93b7dc209e1a682a75 : SUCCESS in 1h 46m 40s\n- octavia-v2-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/0099029ecf9c4c8cad9734a737a660e9 : SUCCESS in 1h 16m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f67bed24a0ab4b7cb53bc0670293a111 : SUCCESS in 49m 52s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/7f27a12da97844a0880998def2f29a48 : SUCCESS in 42m 48s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a9209a5937334602bc57f31a3a9cb9d7 : SUCCESS in 41m 46s\n- octavia-v2-dsvm-tls-barbican-stable-stein https://zuul.opendev.org/t/openstack/build/2d51273de2c84c5393820305518ab83b : SUCCESS in 44m 16s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4300844cd16244839c8e32a02284b3dd : SUCCESS in 1h 38m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/32e4e22515254aecafe86717d401d575 : FAILURE in 40m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60f6898ec9444ba891bcc42032a2a37e : FAILURE in 2h 07m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b73b3c46fc0e437b843a1ddc9e47b95b : SUCCESS in 49m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ca49afd425b74cb9adfccbf0e656f950 : SUCCESS in 53m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2803869440ec45d284e597e8e30951f1 : SUCCESS in 51m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-stein https://zuul.opendev.org/t/openstack/build/fdcee99f3e7a4aa99cdb7796f846a22d : SUCCESS in 53m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9883c38054a84014aed83b83917c0570 : SUCCESS in 51m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/d6f631d7bad54f55ad661213477c79b7 : SUCCESS in 51m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/8f5ae910de1b4dc9a2fb15e2cd3d12ff : SUCCESS in 47m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-stein https://zuul.opendev.org/t/openstack/build/61876735443f485aa8c5d767a0c47c01 : SUCCESS in 42m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/844477ee468d4eca82e71943fae46f4d : SUCCESS in 2h 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7e33e4f63be7a21f83bfcadd113f0e399c72c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-28 22:39:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64ea539832429e3fdfac3d10180d8b2aad1f576d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 00:59:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1f92ab4d354c6ea922fd312aecedc9 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d70d46be14d4017b8a7f2ce3787f63c : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f045885ef2af416286ff62770a0378d2 : SUCCESS in 4m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a00b0ec7c487452db2578fb19392dfda : SUCCESS in 1h 44m 06s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/dbffad5cf0504b458227e35ce90672d1 : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/eba8a20821ae4386869d2090394d549d : SUCCESS in 1h 16m 59s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c1f5c4d627ff4e90aabb00f9e4b0f140 : SUCCESS in 1h 29m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db180906066b4595a469fcfd1fc76302 : SUCCESS in 2h 17m 22s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/32bc475d52cc4f49b5998239dcd3da08 : SUCCESS in 1h 32m 53s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8be2b60ae96d4b54b707238addaeba17 : SUCCESS in 1h 54m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/04d9dd8341814a4bbe43606db481987f : SUCCESS in 50m 03s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/2373d000fe464c97aed1298001525c7b : SUCCESS in 38m 41s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3091137a7b514e998fe85c672c73f4d9 : SUCCESS in 38m 40s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/92300b072c63494480fd42db88c8b113 : FAILURE in 1h 38m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/37a63ef3e5c94997929be05a4044fd3c : FAILURE in 27m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d4964ac63a7946378ba2ae4b68f8853c : FAILURE in 2h 15m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20ef00e1fc694c998307453aefa04e5f : SUCCESS in 38m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/84a04dab5df74a6cbea30d2dce6b6680 : SUCCESS in 59m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5bc7dd0af7a04781ae9d2b173e3c0c35 : SUCCESS in 49m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dae70b418d9c4cdc8624491a1fd09332 : SUCCESS in 37m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/a59dfc4f50e847b28a4ae57131eade8d : SUCCESS in 57m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/f2d3c45e4b8c4dc193a9e7cbcef0198a : SUCCESS in 48m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4fc96cfbabde4f45a0b877b97d4b5ac2 : SUCCESS in 1h 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ddfd583d5fd29ea3598ac84de4ce5a97b3af2d15","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-08 21:49:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"740dafed534c25b0e8b46b9ba1383a6d4259f288","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-10 00:28:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8aa36bd45e732e7074a1b94a66346367bd1de21c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-10 02:01:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e193991a988bc7911098a6ca9abf2a2806c5fa3b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-17 16:32:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43c2c3e610b91d8f879db37ad7a146889fe8bd0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-17 16:39:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f5111ab0581d9ef4876eb13c94dd2c74819696f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-17 16:46:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"73ba5e1a48153744f2b627a449ba510d98b68fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 18:43:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ff3629f000494da46b3799cb57561a : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/333a2d67f47d4727bf21d8447c9890a6 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc2ebfad6e89455f8dd50ad25d892d00 : SUCCESS in 3m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c227a1b24b54707a7d9870caaae5c52 : SUCCESS in 1h 32m 52s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/a086f9dd042b4a3a9f8f22d0af2fb2b7 : SUCCESS in 1h 27m 53s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/c995be01ce414cda92b8e5904e21c317 : SUCCESS in 1h 13m 08s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d64c9ece024f4e509cd0c353bfb6806e : SUCCESS in 1h 18m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0573c68da50749a5b4b173758cb11225 : SUCCESS in 1h 52m 19s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2748ef79391c41a993d6add08b03a7db : SUCCESS in 1h 28m 32s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/10a1029b935f4396b7e4eb9c19708ced : SUCCESS in 1h 59m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee38eb400f5c4d35a8c2080b3a4bf7dd : SUCCESS in 35m 08s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/32575f639c2b462f842105cec8387b47 : SUCCESS in 40m 26s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/b98ddccb8b8d4d809318152daf0739eb : SUCCESS in 33m 26s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/60f27892a87c4fcc96fed76ab3b4e57f : SUCCESS in 1h 37m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a9617c7cda614a9b880f754c86f24423 : FAILURE in 29m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/98493942602d4fd0a8ae8c0139596eaa : FAILURE in 54m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15e592d4c2b742b2b47692b1a62f9cbc : SUCCESS in 34m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c032fd878d34473fa1f8449c787eaffc : SUCCESS in 33m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/638ba405334f4e80bad26eab3af23e13 : SUCCESS in 51m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d93c964776404c009b2f903d74edad30 : SUCCESS in 36m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/fee38e6afd8842e3aa627e505fe4e927 : SUCCESS in 41m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d6e3342902264df59bf3473b93b6afc5 : SUCCESS in 51m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3b54ce3f90344798a04cfff696bb0c51 : SUCCESS in 1h 26m 43s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/3f71c7d3966b44e393657e913dd8d67f : SUCCESS in 44m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7ae216a00243411cbbb63d17474bdfa3 : SUCCESS in 52m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d808ad9ecf304350e01bec83d751baed7d7f803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-25 17:05:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"307fc383c74950a18a16932560b6488b7e975d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 19:19:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1feb04c91134d759128629f51bbe84a : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3b9733c07a486e994a01019d842019 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25f0b39ff7e34ecebdfd2041ab437ecc : SUCCESS in 3m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29291b6104f8498483915358c35b6251 : SUCCESS in 1h 39m 46s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/89af4bbbd04d430c8ac79ec6b31c76ae : SUCCESS in 1h 28m 26s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/1fb88c2564ca4528b7ed91dd91ae5459 : SUCCESS in 1h 41m 00s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/6378bbc1caad44ffb177df33206477c4 : SUCCESS in 1h 18m 05s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/b51c74b62129449e984089ec3a24d885 : SUCCESS in 1h 32m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e371bc81d93e42b2a18efbd75ffc969e : FAILURE in 1h 55m 40s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/25f285853ead4116bbabfda0ea19045a : SUCCESS in 1h 45m 29s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/90de6d715dac4eb3b50431f93d5bdc9d : SUCCESS in 2h 09m 47s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/11e9fe8a8a774c08acb434fc34957b2c : SUCCESS in 1h 44m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86866a7e55bb4ecaae7a8d056710ce32 : SUCCESS in 42m 29s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/27cdb5f116a84f5b96a69ad4bf823863 : SUCCESS in 1h 05m 18s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/c2f9fc77fc7548f2ad43137ebcfe3883 : SUCCESS in 37m 04s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/304ee294e43748c285bf88891dc386da : SUCCESS in 40m 51s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f5edfa66c9c412898bd746d6d3a29fc : SUCCESS in 1h 44m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/46a8bf9498a84e2e832165b2e66973ec : FAILURE in 35m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1e2aed536a84104918e197a8a7132e1 : FAILURE in 1h 02m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a0b5043ca4b49cab28a4836805073bd : SUCCESS in 42m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/df73120a3b5d4a4a86d9fe88a3922694 : SUCCESS in 42m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/867e405af49840b4a7f201561ea34754 : SUCCESS in 38m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5c80a352cec746df94da0b309f44a49f : SUCCESS in 46m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dc2958ac5e4d4d1fb116343e1733efa6 : SUCCESS in 39m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/5708518e3ec0425191a2cc58da0d82e7 : SUCCESS in 41m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/19ace09a7fd04bedae402ea02e4e3e4f : SUCCESS in 58m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/5c3fea0a920f420d810ee7ddde1f3e0f : SUCCESS in 33m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5bbef3db4f89410e89ab829e30f541de : SUCCESS in 1h 49m 30s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/55ec93f7ccae48e2a128bd0338c99c7e : SUCCESS in 46m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/36537faf3e614815afb7ceae60215097 : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2dcadd3bf782f299b1fbd96d16af2b232874e473","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-26 20:35:37.000000000","message":"Patch Set 4:\n\nrecheck\nunrelated LC test issue","accounts_in_message":[],"_revision_number":4},{"id":"d376474b4a4c445335f55dd2b6fc33e2d372124f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 22:56:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec186a03b8f04afeaa0568c9fd81cae7 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41fef469596747db8bd4cc5c0c79dad7 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/319a018c7e4d404c977e373006485aa6 : SUCCESS in 3m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ad8eb6eca0f4759afc2b76cced5d13a : SUCCESS in 1h 42m 06s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/08717df7630745849bf9ef80502b23a4 : SUCCESS in 1h 12m 26s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/b5dd1d5a041d413db082c835cdaaba1d : SUCCESS in 1h 30m 55s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/4cb8be3c5e704730beee21108065d088 : SUCCESS in 1h 10m 52s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/b962f3ee77334edbbdd1dbcaa30ca27e : SUCCESS in 1h 21m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0a1c56a1d58470e95d11d44e6600ad3 : SUCCESS in 1h 31m 50s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/557ef9cdee564e798c207e7f7ba0e639 : SUCCESS in 1h 42m 28s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/16c627d6c7df477f93bd35e3757e34c2 : SUCCESS in 1h 34m 51s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3c66f8bb1be747e6a8f614fc6df5d8b2 : SUCCESS in 2h 17m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02b26e0d862a4d4fb8a4f7cce47fcf9d : SUCCESS in 41m 47s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/26af627d16454f2ba743c91d716117b4 : SUCCESS in 1h 17m 58s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/7736895564a1466a8f4d9c22297a73ee : SUCCESS in 39m 29s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/19c49b83ffbc4d8887e9c31a2a67407d : SUCCESS in 36m 24s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5973e6822f6f43e89918478e27a259b6 : SUCCESS in 1h 38m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5b7f5af17d4f403ca0e446b77129c5e4 : FAILURE in 25m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0dd7be5f18124ce686e54bf91232454e : FAILURE in 1h 17m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed9ad5f5480d43ac99d607b9c3fecf0e : SUCCESS in 34m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/a4a2310651844294a31b3228cc43b7f0 : SUCCESS in 1h 13m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ab9a138383894694aaef620e27783859 : SUCCESS in 55m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/fe9f209c0dfc44ecbb2d196d3072d9d0 : SUCCESS in 40m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/54c256a6bd0b4b4193409c3ccf052bb4 : SUCCESS in 53m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/a408e7fcd2d947a58ff67ae878102ced : SUCCESS in 48m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/0d99ae79f8a04e279f26dc46ce488f41 : SUCCESS in 39m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b09a7ae29ef741bebba248beefe2975a : SUCCESS in 45m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f04cdf47558845a28b7c6180deb08741 : SUCCESS in 1h 50m 29s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e83e75250b384b0ab1d28ccec7318493 : SUCCESS in 55m 35s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a4ba195e933b4753b7c6b6cd7a0e2dd6 : SUCCESS in 43m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7eaf0696e7d31ff3c977f710ff79821ab1bb9fb1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-27 02:09:46.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nThanks for making the change.","accounts_in_message":[],"_revision_number":4},{"id":"5eaf4d5f519dd4a3218c916010f423ae33953681","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-29 13:40:39.000000000","message":"Topic octavia_new_rbac removed","accounts_in_message":[],"_revision_number":4},{"id":"f545774b3fd5b477b29e191f2a1c57ba804b0d33","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-29 13:40:45.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":4},{"id":"329f2583b061205c7a9cfe18aaaa25d11fc79f1b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-08 07:43:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nIf I\u0027m not mistaken, I thing we can remove os_roles_lb_member2 from the expected_allowed list in most of the \"create\" tests.","accounts_in_message":[],"_revision_number":4},{"id":"3539fed040e6706d5e4c70425a2d275e75811a8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-08 22:07:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"66cab4f99246d3eb20161edd12a2f89a0ac7f26d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-08 22:08:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1042d531f846040912b6fb5147fdfd72a7e3e14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 23:55:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b58f9d002fa40d9958defc6c74a131e : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfeae55403d94a49a3d3969a972f2a5f : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b82282bcd4e4c5db54eaa1353473e0d : SUCCESS in 3m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1a3d188b98443cab461a49305c5c3dd : SUCCESS in 1h 08m 46s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/a8d14aa7271b477d833ef36666f1e136 : SUCCESS in 1h 26m 15s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/26b19790281240399fe419090b72514f : SUCCESS in 1h 06m 25s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/605355a2e7914bee8727e1936682cf1f : SUCCESS in 1h 10m 59s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/675b4d9e12bf46e08a462209b8b1c6a6 : SUCCESS in 1h 20m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62da5243479d4fa8bfee27544503f2eb : FAILURE in 24m 18s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/7f709a03032e43879bb144258bb51e33 : SUCCESS in 1h 34m 09s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cd74d6a621d54633a3f0f5c2ade334df : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d2ff89659f5b487b927adf4d7b48deb3 : FAILURE in 1h 44m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c543e95bee7d4a7aaf8b5857b74ff132 : SUCCESS in 35m 18s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/06a1252cbcc044ea8fd677dd6d43f1f0 : SUCCESS in 37m 26s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/66dddd17b8b54ee6b7102848f62d744e : SUCCESS in 37m 16s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/c61a051234f946b9b201609b7b11b9fd : SUCCESS in 36m 54s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82aa9065366b41f5a3574a459e494de6 : FAILURE in 29m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b57cf820f5f1409ca6fce3211055947c : FAILURE in 27m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1ea66ae14bd486eb0aa06fea8c97575 : FAILURE in 39m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/782de15a79414b678c8670e1bfb59849 : SUCCESS in 33m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5180782070924bbe92e9f182f2b9171d : SUCCESS in 34m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a516b1dbae08442682dd17fe9e0ac217 : SUCCESS in 35m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/cb020ea23d334d98b601805d779797f4 : SUCCESS in 35m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/196ad7b5a7844aa3aa4fd81f2cb0f4ac : FAILURE in 24m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/ff1ceb8d1cfc438d88fdcc5a34f59730 : SUCCESS in 35m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2f55b453232a461faf700be49e01c213 : SUCCESS in 36m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/66a2ad3790194dbebe767ff1ab125fd6 : SUCCESS in 33m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f8279811439f4a41afd43560738d8bd6 : FAILURE in 24m 39s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/23d2f38361c545ddb6b2a2ee4934cbd3 : SUCCESS in 34m 28s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/9cabe4bc8b7b41b89c6fc249f4969b2c : FAILURE in 47m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d8cb291fef526450816b92455383336e0e64cc85","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-09 02:18:26.000000000","message":"Patch Set 5:\n\nrecheck\ntempest failed to install","accounts_in_message":[],"_revision_number":5},{"id":"466f86a46a77ead4817856160b8a01e781d9169e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 04:35:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e537c732544cec96108e557abfec97 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1253aacb0ed84cc0b91658e3fc301c0f : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18b9f72f68b04b3999af2470bc653e94 : SUCCESS in 3m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e1e094345a84fc591eb6641168a0b12 : SUCCESS in 1h 34m 18s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/3a431c0efc844c8abb6f78e80e8fe015 : SUCCESS in 1h 48m 10s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/ea95e3184f8045ada1528a2bab7ddaf0 : SUCCESS in 1h 35m 11s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/5085e7c7479848e8bacf7f0f4605f359 : SUCCESS in 1h 38m 26s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/7b6b7b4af70248d086b6d0e8897212c4 : SUCCESS in 1h 49m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ba0c56268ab40ef8bd63552bf92b0f8 : SUCCESS in 1h 33m 19s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/ac498dfcf43542d19c50fe8506b19a70 : SUCCESS in 1h 36m 13s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/18849c3a83964cd7bc7c6efa04976902 : SUCCESS in 2h 06m 46s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e218d6a8921f456d8ee8b0da3d4f210a : SUCCESS in 1h 34m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51e5c20338474893b1ea87e1c695bc8a : SUCCESS in 45m 11s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/bd0b9e885d1f44b9bef39b6af9694f0a : SUCCESS in 48m 02s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/46642ef2f1de4e599cd1ec6322141637 : SUCCESS in 38m 32s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/4f393027d7f0416f87844082571d7221 : SUCCESS in 37m 17s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eca180f538bf4067ab798f6b16a836a5 : FAILURE in 27m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d6b5c95a7173427b9f2a0beac0a56277 : FAILURE in 26m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f23178c9dd9343c88d7394671af9038b : FAILURE in 44m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/742492861ea240758aab8c228c9d7bb5 : FAILURE in 50m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/1365dd99eb5646b29cca52ef5775c1f9 : SUCCESS in 52m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6e6cc3d721654756b58b7ab6e243e90d : SUCCESS in 1h 02m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/936a20b0a0124701abee4d66d228c63d : SUCCESS in 53m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/40276366dfea445594bf6332c6b2dd2b : SUCCESS in 35m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/2fd38c75f63b4eabab2f4453ee3a826a : SUCCESS in 54m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/661c94cf9cdc4835bd677ed1b116b0c7 : SUCCESS in 42m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/35681b54cf0a4796abc82b25935e2630 : SUCCESS in 43m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e58b687b68c942f7a280018a0d9c506e : SUCCESS in 1h 42m 00s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/98b46982af4645c3aa00cda397409076 : SUCCESS in 43m 39s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/470051609ca746c4ba8b39f3f101e535 : SUCCESS in 53m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3102195b43225b00f9b0fc4fcccbce4ea9765ff6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-09 08:57:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a219d959b83b606ad31dcb9a868ed6e9d3cf47cf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-09 13:27:59.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78c80c9a5c41c3db08d5435de8be15ce9d9292e1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-09 20:17:16.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nThanks, Michael.\n\nI have a couple of questions inline.","accounts_in_message":[],"_revision_number":5},{"id":"e83d3470c4cce2b7fd2ff4f8806ae5c150c85b0a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-09 23:03:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a35dd3a783a690697eec3317a3f1fa5fbe1e6f97","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-04-10 02:12:00.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9c353079ef76b61809e093c1fc22544ac1df8222","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 02:12:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b46da6229fadc6111789dc6a277cc005d9e57487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 02:43:58.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/992dbb6ee7a74eeb95ca8696ba12680b : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76260100d584e2aa672ee6f90bbcbf8 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2596e59d5d3642a5ab9b295b69432ce8 : SUCCESS in 3m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/605b882c456042978004601b1dae54b9 : CANCELED in 30m 19s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/3cc6c4344d8f43c78f814ad7a9ef5b63 : CANCELED in 27m 30s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/363e7225e78a425fa04096c3a5e71148 : CANCELED in 29m 08s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/cd9ba98887564cad9461eb8aed8a4fb1 : CANCELED in 30m 27s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/e836c4934af24a648529a235d5af500e : CANCELED in 26m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e72cb08d23f34aef929f3abca0f75708 : CANCELED in 25m 26s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/b931cd02a09143258d1e2c453e276945 : CANCELED in 30m 11s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/63f09b4a66cd4e61b8e90db2bc684881 : CANCELED in 28m 13s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4e87e957d2f64bd4858d51cb2ef05875 : CANCELED in 27m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d979b8b6dae74c7db7a7ebe9974d77a4 : CANCELED in 30m 28s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/fc26f6ade0a74172a090df2ccc002c3c : CANCELED in 28m 02s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/1232b9aa04e84a6aa45b9656fdb66f38 : CANCELED in 26m 49s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3ce89d1e54a1416a99e6444eb412b914 : FAILURE in 30m 28s","accounts_in_message":[],"_revision_number":5},{"id":"1400b6a58cb666ed872123685a0ad1bd59f00cf4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-12 05:45:22.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"cffa94c555bb03b0733e1dcdbd36f0fb94c76d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 07:49:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9651b21236b47de941c2b7f767cc477 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64e7dc97dfc84b7687f76192aae2e4e4 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0225be8360794a718ca84efbb1bc9da1 : SUCCESS in 3m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6626aa6c0d67416aa63ca5533e1ad421 : SUCCESS in 1h 31m 44s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/bcc4c2ca3c0748a388e984d309d05b17 : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/3544d5549ec943019c1cc0589534e163 : SUCCESS in 1h 33m 38s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/baeb5179c87344778fb6108a201418f9 : SUCCESS in 1h 54m 18s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/8448b54a47fc41eba9b80340e492ccdf : SUCCESS in 1h 40m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c056b1e65e140f7b42cf88898d46346 : FAILURE in 30m 27s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/77274c75bba94c9d87ba83a4e36b7d38 : SUCCESS in 1h 58m 40s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b1c1b457014a44679a7f1bc7913f8c10 : SUCCESS in 2h 01m 31s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/44150ba132c94de48332ea2fc0822348 : SUCCESS in 1h 47m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/277166cfa643434cac5b218cb45bb623 : FAILURE in 32m 11s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/04041d49270940b4a312b9d510c0d974 : SUCCESS in 46m 57s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/5e3fd9e27a524c8c944a64a893f38d88 : SUCCESS in 43m 07s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/8f60368833894225afba34a3695ed8c5 : SUCCESS in 44m 56s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4e7ea6f6bc4740ad8c8155e5e3ab1aba : FAILURE in 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7d63f2ec24914f52a55572941465b50c : FAILURE in 29m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/090ebe685a6147b9b017c36888763bbb : FAILURE in 46m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a99f120fb6ab4d7f8147eff103c47872 : SUCCESS in 50m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/e1ae06cc90a144228465bde1f3e56dbe : SUCCESS in 45m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/793c0c9e5d364e6988757d1871b355d9 : SUCCESS in 57m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/77e75ae72fb94be0a51f88054c264f63 : SUCCESS in 58m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/95f9bf7ea4394f0886441590bc74fb94 : SUCCESS in 39m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/1d7485b8e31b4d8ab471d8f370c1f51c : SUCCESS in 46m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ac4947e5876848439eb53f1851f3ced1 : SUCCESS in 55m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/6e2bac23a2444870b543e2a69eb71e5f : SUCCESS in 50m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf5c596899264d4c8b7512a27dced0e6 : SUCCESS in 1h 28m 25s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b9c216564f124fa180d5dc3ceead3e3a : SUCCESS in 56m 26s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7c5f04482d064053b30b5b5334db4e7f : FAILURE in 50m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dd730815b2b4fda77a572aaf687d06a097b49ea5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-12 14:06:51.000000000","message":"Patch Set 5:\n\nrecheck\ninstall tempest failed, tests didn\u0027t get to start","accounts_in_message":[],"_revision_number":5},{"id":"fa761d87a12922342ed407e768b3bbc824ef2be7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 16:07:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f17452121555474382a0d6a9c8e3b61f : SUCCESS in 14m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2fd650483eb4ffba50a22b80475ac92 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4215b1d220240c49e2d76207a67725e : SUCCESS in 4m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb92b306d2d1479e85872bb1da04b095 : SUCCESS in 1h 21m 11s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/af6fd15e56804abeb3c8baf7abe2cd74 : SUCCESS in 1h 21m 17s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/b2480bc7f9f54ae3af9a9889cb0b0f48 : SUCCESS in 1h 33m 33s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/6c7b70bde6014484b452c8611dcdf0d9 : SUCCESS in 1h 32m 31s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/09955d68e5b549508f4918add860438e : SUCCESS in 1h 25m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8ad0d48504343478e224e9bdf6757a7 : FAILURE in 32m 51s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5c2c5cd6c9f44da681c1c9389da1d2a0 : SUCCESS in 1h 52m 47s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7e3751cd3a824411a08277bcb15912e8 : SUCCESS in 1h 54m 37s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/469b533daa3b4ce4b8d058b433cb8e82 : FAILURE in 1h 57m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c26f361041f74f9895c864e5ba1f033d : FAILURE in 33m 12s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/423a3a50ac474bd6a8faa850473b319b : SUCCESS in 40m 04s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/39ee14cb047c455eaa7b994650d0cac1 : SUCCESS in 37m 51s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/55fd2f474c3a423d8669410f91de82bc : SUCCESS in 37m 58s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7e80a56936c141e7ab797a9b0ce07703 : FAILURE in 24m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bfe385556704486bab7cad9a26a2b4be : FAILURE in 27m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/152770c6ae544e54acfa2b2cf4a30bf0 : FAILURE in 41m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8bb56afad1c471e87dd9d04aeae5387 : FAILURE in 25m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/c0ab92436a98481ba195ca0cb31b70d4 : SUCCESS in 36m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/fa57acdf8353450a901504cbf761552f : SUCCESS in 36m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/20a3feb50d4d4deba85c56874bcf80db : SUCCESS in 36m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9aa5c7a782d544018885649187ef88c5 : SUCCESS in 32m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/bfa539721234468bae427a1800f70273 : SUCCESS in 36m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/f945a15483c14d9a9f497535488bc87a : SUCCESS in 44m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/861ad26c102e43dc82961ad3f43b862a : SUCCESS in 44m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/853710e67ee441638dbf9c591b5d513d : SUCCESS in 1h 38m 59s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4595207d0b134e059e3f81009ae21851 : SUCCESS in 33m 32s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4146139cae504f838c1b5dd9102747b3 : FAILURE in 1h 00m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a59766740d765c5f747ab0af01f88bb211385244","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-12 18:14:37.000000000","message":"Patch Set 5:\n\nrecheck - tempest failed to install","accounts_in_message":[],"_revision_number":5},{"id":"5f53df9f012e275e9948d7cb8eb42c59c1c45fe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 20:29:42.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e157cea2dfe74a5ab208bc9ccec77900 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624e4c3905df4869a57ce1fb777c132a : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6112572309e4454a527a5d6739b33c6 : SUCCESS in 4m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/17d0d76d46c044d0b37a851d6480d4ee : SUCCESS in 1h 17m 10s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/ed2835f0d71047af8987e41b678a625e : TIMED_OUT in 2h 12m 59s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/2fe8b2ecf4474604bf0864585a1ee4e2 : SUCCESS in 1h 39m 13s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d808475e84764ba1992ff968ecb31174 : SUCCESS in 1h 34m 56s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/00d3823798eb4d1a9cc906ec9d07543c : TIMED_OUT in 2h 13m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf475c490912443e99cb4b6bd9931ee3 : SUCCESS in 1h 54m 38s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/b40c0d1326fb44cb8b303a9c80c880f5 : SUCCESS in 1h 38m 00s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a3ce8327a7484a72b5b1524f802d1097 : SUCCESS in 1h 24m 48s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6ea2c7d9d95b449aa8c4b55109fb0a49 : SUCCESS in 2h 01m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f8194e753f04ca3ad31cee1f8027c39 : FAILURE in 35m 08s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/7b041053531746258c31be31215a0a59 : SUCCESS in 44m 52s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/19488e3e5b1141e2b4fe647f32fddbee : SUCCESS in 1h 00m 44s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3b4701b4249248068453600a17c5813f : SUCCESS in 36m 59s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c7fb85c6dae442658a0554c94691250b : FAILURE in 25m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fbabb7ec94664d889b6c6ed06f8e995d : FAILURE in 25m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0799b891be8f4ae492792aec832ec9b6 : FAILURE in 27m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c36e02251bf48af926d9d48c7a523e9 : SUCCESS in 31m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/7b9538c96a2845b7b166c928a0e90c0b : SUCCESS in 35m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1e66eb40a6aa452f85b9d0b45ed38bbb : SUCCESS in 38m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/32011eef5b1b43b4abe3d35f9e6edd8f : SUCCESS in 34m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5d3e800a437d45fc9c028051b5b5a392 : SUCCESS in 43m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/2117144d19d04b09b8bcb7ea67e6c957 : SUCCESS in 50m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/51b8c0f582264cd7b86593e32dfb9d6a : SUCCESS in 34m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/96f4b55f68fe43ba844147664ef46322 : SUCCESS in 33m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/663c25c1657a4cba97ac8ff5d7453268 : SUCCESS in 1h 42m 16s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6a3668224e324f21b58e94d73d9169fb : SUCCESS in 51m 45s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e44de350224345e3b834dbc5ba7b0a14 : FAILURE in 53m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5a3bd36d1620e7a391e139f8c8a35255853b8728","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-12 21:04:04.000000000","message":"Patch Set 5:\n\nrecheck - timeouts","accounts_in_message":[],"_revision_number":5},{"id":"f45e517799a4d70949062f0ba3c6e14fcae5df64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 23:04:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f684f746cb0c4ceca917d4fed9a42d59 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a2bd2d5a384eeb85a7200bb8b936be : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/777d9659cedd42f58f19909927f1ebbf : SUCCESS in 3m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/729133baa9184a4f8a4e9e70f03bcfca : SUCCESS in 1h 30m 36s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/2f05400eefe94e5587f8867dca070d92 : SUCCESS in 1h 42m 11s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/07b3e88250de43b693ee33a71c8a9dc7 : SUCCESS in 1h 36m 47s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/682fce5e301e42a6a48603713746e7e9 : SUCCESS in 1h 39m 29s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/6491de37f1cf4b73884f6df889bbfd12 : SUCCESS in 1h 55m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61f680e7bc6d446696780a818200790a : FAILURE in 55m 56s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/a508c8af5c724982bd185d82077c57b1 : SUCCESS in 1h 48m 16s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/30bdb9d360974b2ba642d5c0e9b98582 : SUCCESS in 1h 17m 42s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4288e26fb9be404f8bda2e82349509e3 : FAILURE in 1h 45m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7befc13c5244dae8e7bd8d63df085d8 : SUCCESS in 39m 51s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/21fecfe14f724b1e90339146cf89e6c8 : SUCCESS in 30m 49s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/fa4ab079d92947d2b39744667121f770 : SUCCESS in 43m 26s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/452e1b029da34ea7b8d19636598f4e4e : SUCCESS in 46m 18s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ed7d13c3926843a48bd370f93868f7d1 : FAILURE in 30m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1ad28f4a6e0c49e9b3d8aab564af0e18 : FAILURE in 20m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a1808d5ef184c098b8e0017f9ac6a5d : FAILURE in 40m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d498560f760447ebf1c72950141c9f3 : SUCCESS in 1h 03m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5b1127eeca064c5c8da20e2ab3fca1af : SUCCESS in 44m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ab116f21761042c4956009babacf107e : SUCCESS in 38m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8e4060ce284644cbad6beec3ef78ab2d : SUCCESS in 43m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/436700a121d14d1082806b1d95d0b7d1 : FAILURE in 55m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/7e77439dfbf64668ba25c6a2276e5b65 : SUCCESS in 42m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/5af4ddd1f9ea466594f3f7a4dcdddb1f : SUCCESS in 38m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/591768502d624e02a44caace8ad0feb5 : SUCCESS in 40m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1fc1ab7558c34bee9521fd80fd4cdfc9 : SUCCESS in 1h 21m 39s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f584f9c70d74466fa3719a1397a0965a : SUCCESS in 49m 31s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c9fab105faaf4631b03691f2c068d595 : SUCCESS in 50m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"643dfa60b9d3b834887ed130df9ddeb0db3e83d1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-13 05:42:58.000000000","message":"Patch Set 5:\n\nrecheck\ntempest install issue","accounts_in_message":[],"_revision_number":5},{"id":"c692e962ff8542ca309c4fa4768ce0e097fbba59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 08:16:15.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7a2b94ffcc4c00822aeafc78a8e4de : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da26392e4b1a45ddbd1f09a913779c1f : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a8eede55364433dbafb580697931079 : SUCCESS in 3m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7341729d3d284f279df3d1fdac6eb1dd : SUCCESS in 1h 36m 59s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/cb922c6f0a064e42b97147d8ddcf3783 : SUCCESS in 1h 49m 41s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/dbfc6700480840a0b4fa8528a5860530 : SUCCESS in 1h 40m 50s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/746cc9cbea7443f0a345ecba18c482df : SUCCESS in 1h 37m 41s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/8c54fea19c4941db9ae01ec7c9967778 : SUCCESS in 1h 46m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be713b875d6d4b108f9fe9db8f3eb90d : SUCCESS in 1h 55m 36s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/19f6de8880a54e068a8434c04085f6f1 : FAILURE in 2h 28m 51s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/012b2865509d4461818ac137941a7257 : SUCCESS in 2h 29m 31s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9f16555d7d35400a83ef642d5fbe0950 : SUCCESS in 1h 43m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4e9ffd5a114e41f983450116fd256b88 : SUCCESS in 52m 54s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/ee107f989e41499aa20565d75705b8e8 : SUCCESS in 45m 29s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/8d8c7e1425984267bec844403e74467b : SUCCESS in 39m 28s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/6c85a9d718c64625acaa63b52b23fdd1 : SUCCESS in 39m 07s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/17a5d4351e7248629ea9ea97c6790e81 : SUCCESS in 2h 31m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fdcfb51467084fd89ee6aff0ba0cea93 : FAILURE in 28m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/21bc93f1b0b04371bf0952706a63fc49 : FAILURE in 44m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59e43e6cfded48f4af3d8a847252329e : SUCCESS in 35m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/a95140d0554f43939f5ef33647ff9e0c : SUCCESS in 40m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ec7c78c320c2491c93ac6edda1cc9e8b : SUCCESS in 55m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5866bef2307e42e7a5a0fa2b2d5e2c43 : SUCCESS in 53m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/15f4ffe298c149878830c3fc4913f21f : FAILURE in 26m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/e9253bd0b96147d2883815664eefed75 : SUCCESS in 51m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/e1fcde2e613245708fd4ff9136aeef6c : SUCCESS in 54m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/350e3950f70048d593b3fac9595a1a91 : SUCCESS in 38m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/efe0b0185f0e439dba9765be7b2a5676 : SUCCESS in 2h 27m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/c37c4e8a2b4249eeb35142922d1975df : SUCCESS in 53m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/8350150d5f934c5c883fa7fa97bd5435 : FAILURE in 1h 06m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7f9b76b24fe91796acd12f81a68fa3eef81cec2b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-13 09:36:06.000000000","message":"Patch Set 5:\n\nrecheck\n\ntest_ipv6_tcp_LC_listener_with_allowed_cidrs failed","accounts_in_message":[],"_revision_number":5},{"id":"e1dfe96e4146bcd0a48240b7d1e1ecf95ebb6d57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 11:44:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4c15ec4bf240adaa5c3df978d1f2ed : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56e41ddc33e4882852a6a9a3a269758 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95d90bc0bc6340bf9f9a3c7c1e5a03f4 : SUCCESS in 4m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4ec3288a80a44758fdc8a93600c05d7 : SUCCESS in 1h 39m 14s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/984784ddf9fe4743958f11e1793120a1 : SUCCESS in 1h 40m 31s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/fabe67283204461c83a7ac2e6ec8218e : SUCCESS in 2h 04m 57s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/3dc561928b0948f4b32593f487e976de : SUCCESS in 1h 42m 49s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/ba61c978fa6d461a9000bca0053ce3f2 : SUCCESS in 1h 53m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/243d90bd2f644a7baa0ead312cc2f3d1 : SUCCESS in 2h 00m 10s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/473e48a09bb146d3b9c8e365d075bb99 : SUCCESS in 2h 00m 36s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/eae021f2dd944c6cbee95b7e29a8346a : SUCCESS in 2h 06m 45s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/199948d4cb274f69b5ab3278303b1874 : SUCCESS in 1h 25m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9f74395fce94072823cb729b4cae43a : SUCCESS in 39m 03s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/f04c06416bc641f7b1f066a52d2bdf6e : SUCCESS in 40m 07s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b73565bb4a72443282cd2ace79e51e52 : SUCCESS in 1h 08m 32s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/bead417f745f44ccb6ca738a40232283 : SUCCESS in 45m 08s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d51452a507b464f9bab8b62181b2bb6 : FAILURE in 35m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/94967e6079a641c6bd156b60ac2bf98c : FAILURE in 36m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edbc506f18084acb8ca56112102d2822 : FAILURE in 54m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2193c34e4874ae7846212e5119ab1ed : FAILURE in 25m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/9db9589e4083427b9f865a7229cc42ce : SUCCESS in 46m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/aa43ba6240b041cfbeb7591d363a2135 : SUCCESS in 46m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/fe5bc89dbf02431fb78a7793ecf56842 : SUCCESS in 46m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dc058bb7e76c48a09159cffc56839a36 : SUCCESS in 55m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/370aa05d8e8f41e69fafaef7e690a70d : SUCCESS in 44m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/21181ad3ffd64b5981a0242bddc6be17 : SUCCESS in 47m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/df817c9a1e10439bac63c80ef3a31022 : SUCCESS in 44m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d2cac79a4a1547c0b1fc2401c00c54d2 : SUCCESS in 2h 04m 09s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a9a1659b855645f69d785a9d0a676fd8 : SUCCESS in 56m 27s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e59f46bfa5bc4f31a95599aabe4dfaf6 : FAILURE in 59m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"634cb03a395f4f88faff71e952540c6a85eefdb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 11:45:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"258a53ddc3412471f3a06eb89c6fe316e36e1094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 12:12:37.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50356e2d45e647f8acefb6b5fb2b51bc : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9078c62e63194d71bd1ecd03666aff0e : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e146999219d46938e71a86289637b18 : SUCCESS in 3m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f09091d4e5749aaaa8842e389263bd1 : CANCELED in 22m 12s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/0c0d1a58b76e459fb4bc1bc14551ec0c : CANCELED in 25m 18s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8b0610b1e32246c0930f79436bb55905 : CANCELED in 26m 22s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d1467e7c15934ac28a2909e5e4e32554 : CANCELED in 26m 30s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/4be2a387459f4c3391ac418acf02c2df : CANCELED in 22m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d054351fd79846b59cbb1dc6ab20d35d : FAILURE in 25m 39s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/87a884041eac4dbca4a7be8b26553b2d : CANCELED in 25m 45s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/fda81d8888044514b3b1ade2d824e69b : CANCELED in 25m 29s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/065a75d19fb248cdb8d46f3b398caca6 : CANCELED in 25m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5077efa3c4a94419a92d24557c04709c : CANCELED in 25m 08s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/aa845192c2bc436ead44ced91014c8fc : CANCELED in 24m 33s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/af600155ba81476abb3721960d20b5f6 : CANCELED in 25m 17s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f810bc75a18a4d93b12aa886c1fa44fd : CANCELED in 25m 09s","accounts_in_message":[],"_revision_number":5},{"id":"c09663e6149feb3f610acd06742ef81095a9e21c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-13 15:07:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"832caff4e1bb3c1cc3c033ba31a42e5d42ffa7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 17:31:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36c2cab61414443b244121f76a53323 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc0b864a83624ee39d2afb0f3ce8bcb3 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4684fa2777b744bcba3c138b270efbf5 : SUCCESS in 3m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98d935ad60194115a0c67babb50ea54c : SUCCESS in 1h 28m 55s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/187ae75a97e4403f9f22c8607e746004 : SUCCESS in 1h 25m 57s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/46dfe858d06f43e3815d6daa722fe12e : SUCCESS in 1h 29m 59s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d24eb1b2964f40fb8f15073ae4f3dbc3 : SUCCESS in 1h 39m 58s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/9b9d00658d7a47afa973a1aafd5c070e : SUCCESS in 1h 49m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60b19468cd414b4b83d55fb91b693a81 : FAILURE in 27m 31s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/73a6ec8f2d0b4017b4a5646da42a5858 : FAILURE in 2h 21m 17s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e49dde4699114c949efb36dcb6df8a8d : SUCCESS in 2h 05m 11s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1849d4bcc6a44be2af0b96fb325e8451 : SUCCESS in 1h 56m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/156ace71e60947a8b320891406c20250 : SUCCESS in 56m 43s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/1bae1f7c3ff44d7cbb01f379d88377eb : SUCCESS in 1h 03m 02s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/aed62e3b79cc43308cd05807450b214a : SUCCESS in 46m 36s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a25332f21b534dde93887b85d8695ebf : SUCCESS in 45m 32s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/75de8174df4d432c909b91b2eea20e1d : SUCCESS in 2h 00m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/18334d042a464976b4aa3504524ba04f : FAILURE in 42m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b4ec4b9cc6d1474c948e546c116703a8 : FAILURE in 31m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b2b64453ece4bb3a70c3690618a2562 : FAILURE in 28m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/1fd48e5b006a46bc939e8b71350b108b : SUCCESS in 35m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6857d3b38e7f4c3da9dd0dfcda3798f6 : SUCCESS in 57m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/554e064cfed34f6ab88b724872f20551 : SUCCESS in 34m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bbed586218174eb3b18388497d9d98a6 : SUCCESS in 40m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/d6e05a04bbe14d14ab6b96bc85f7a5ac : SUCCESS in 40m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/e53a4ca7cccc487cb7e3b989817edba9 : SUCCESS in 57m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/c4b68b117fd64d02b27045dd7575545a : SUCCESS in 50m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8581e00383294518ae49298b8434b560 : SUCCESS in 1h 52m 47s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/1424ce346c5a431d9c236978de23d3ca : SUCCESS in 43m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c13b661c51fa4cdab69300259e8c00c5 : SUCCESS in 50m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"89354699702cbd45b3feef4f6d0fb45377f3f389","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-14 13:04:37.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"470b009e07ea83d9c1b45ef78bb7612009cafb71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 15:13:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c2ad19e4b5497b879fa44f56e307ec : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90f3cb6f93dd4a0b92aafce01d241925 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/215973edf5844d3d838bf6cc062df73f : SUCCESS in 3m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6543d8c34a034c61a4eda3b18db1ffb9 : SUCCESS in 1h 27m 37s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/82865e4ba92948ed994daa0fa7819041 : SUCCESS in 1h 33m 08s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/c43512b899134e97882d5f89fa7fe11a : SUCCESS in 1h 31m 33s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/7b0c80b1163f41a48a7a73d2330ff016 : SUCCESS in 1h 31m 10s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/83a872fb371141f18312171ad570feda : SUCCESS in 1h 43m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8c67c82a4ae49d286306a2135afedb1 : FAILURE in 27m 38s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/74d4df132e9a48b9bdb2099573121355 : SUCCESS in 1h 52m 22s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8c66afdfaeda46eb87e4cedb3ed9df6d : SUCCESS in 1h 46m 35s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/31b72539bde34ffd9e53e88f5a429fa7 : SUCCESS in 1h 50m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b7b5f0aa9d534e218474cbc61a39cc8a : SUCCESS in 39m 54s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/349450f6243344e6ad3dafd1b2452e60 : SUCCESS in 40m 13s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/0e9dde0686da4b36b69259fad296ac37 : SUCCESS in 40m 27s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/cf4aa3d09ccd4e74a2c0a0a2ff7fdc99 : SUCCESS in 43m 53s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/826b7eb5ec364639a34e9f9ed8d305ea : SUCCESS in 2h 05m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/201386e7f36a4a9db05c0a13f2723961 : FAILURE in 30m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4cc6102b70cc47c490c312e2fce62909 : FAILURE in 30m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be2aee988953450fa208838c7ad00c1c : SUCCESS in 41m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/01b9567bbf004e1493f6dd34d716f919 : SUCCESS in 42m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/debb63e5bddf40c0a1ba830af33509b8 : SUCCESS in 42m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8bd2b569c91648189e0c122aaf56920a : SUCCESS in 38m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/df00b1b868f648ceb41a8821622fea13 : FAILURE in 27m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/12c3e8b8b4ba4410bca613c2f5aab368 : SUCCESS in 43m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/106dc47d93a842b6af600686cdd7eb92 : SUCCESS in 59m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/7cfe24845faf4129929106dc349f2516 : SUCCESS in 36m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/85db3994b1f64812ad5b3c1ae592182f : SUCCESS in 2h 02m 49s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/533ab4f49e9f43c28add9488820f9b42 : SUCCESS in 46m 32s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b2235043029b4bc1a6deef7c5a9c7117 : FAILURE in 1h 08m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4f45f1933c56719b77c9056cf685d11bdb7de6d9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-14 15:19:08.000000000","message":"Patch Set 5:\n\nstill hitting the install_tempest issue in devstack","accounts_in_message":[],"_revision_number":5},{"id":"d8a20727e7148676658af15063a83ea22ad11087","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-15 14:33:27.000000000","message":"Patch Set 5:\n\nrecheck\nDevstack fix merged","accounts_in_message":[],"_revision_number":5},{"id":"e552db89864293a8855860a4e46f0acdd47b6a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:37:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a804203fc774c9aad207074ba79381b : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f66af811e87e4f58b5933a3772919073 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c28c8495f8a94fbab5a90b730c71aea4 : SUCCESS in 4m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38fed3f1846d4d16b05bad9c504c71ed : SUCCESS in 1h 30m 21s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/ebddc45a30ed430ea1c2785c1b40a2a2 : SUCCESS in 1h 29m 32s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/9991c8b9fe32488d9696107379c11e3c : SUCCESS in 1h 32m 36s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/ddfdfa89bcb941ceade67f1519151b3b : SUCCESS in 1h 18m 58s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/e3f0032f42814a73b89de94e9d590e00 : SUCCESS in 1h 35m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/048c8d961c9b479fba2ce47e3924397a : SUCCESS in 1h 33m 35s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/8aa62d644392406781f5bf56a794e8b3 : SUCCESS in 1h 59m 42s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/29f360288d3441d7a2b1fcae48c8ff9a : SUCCESS in 1h 59m 16s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/67b2fca8f9594785b72a856acbfea9dc : SUCCESS in 1h 59m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0ebadf137c54c4fb72830232786dfbb : SUCCESS in 39m 12s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/8ed421d3cc754fa4b603691ecda19b62 : SUCCESS in 45m 14s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/c7129947ddaf4c629c669bcffe0a0466 : SUCCESS in 46m 39s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/05f46db3e2b44549baefaf6858babf99 : SUCCESS in 45m 50s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1a281ed4322c4e2e989cc3d965c85baf : SUCCESS in 1h 36m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/76e979a767e04ef28e09ee6ce96f3cb7 : FAILURE in 30m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af66a4fef024413cb7d9cd1aea3372f1 : FAILURE in 49m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06536f1ab2ae494e85ad388208a4d9e9 : SUCCESS in 47m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/6bfa0a5acf824fd3a01d235039d0ad5e : SUCCESS in 39m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f132f37efeae4b5cae89dfdd6b816cf7 : SUCCESS in 45m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b8afcc234f5c4a2dbcaa340898980ea5 : SUCCESS in 46m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4d5c68181a434a3b9ea4c28e35f5a582 : SUCCESS in 37m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/989bdcb86e0941e0b7c2a899716855c0 : SUCCESS in 33m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/f572a12138484dc282b338ea0131884d : SUCCESS in 34m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/5cf0db84bf2a494082b0b31ba483e723 : SUCCESS in 39m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d8824aa2c2514839a3cf85279aaa05e0 : SUCCESS in 1h 34m 55s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/24befbeeed37423aa043863aab581f2b : FAILURE in 46m 41s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/66009758026d4a3b99adc07575e174ee : FAILURE in 1h 01m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"521d41f062a6c69fb2a7ec6603aace3ab9984bef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:43:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"64c0f21b1dbbb2e44f0fca412574c701cdee58c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 18:49:20.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ada7e021c07487ab3348fd0767c0e9f : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/271b411315e544b2933caf595837d537 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8aa28a1aaa2468f8f2117d90498f8b6 : SUCCESS in 3m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3df7efa6de24180be834c3b69f63171 : SUCCESS in 1h 36m 41s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/6e5db921ef4246dba264702a9270c2a8 : SUCCESS in 1h 24m 07s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/26051fd6723d4853bf2273de0841232f : SUCCESS in 1h 20m 55s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/7f3cb5bf42bb44579ae336b2ca38cb0b : SUCCESS in 1h 20m 56s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/138156ad11a2427f83ecd746e8a02435 : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9e8a700ed804bb2a8e1263cb1e38ca0 : SUCCESS in 1h 36m 47s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/913b0bdab31140f0b389bb01da58f0cb : SUCCESS in 1h 49m 15s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f2c8892a59ca440294f08a9a78c69735 : SUCCESS in 2h 00m 09s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d9ff441642244d74bf0ec15c7c18d07f : SUCCESS in 1h 50m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80b620a00ed94080ba8685e9dda2d8b8 : SUCCESS in 48m 08s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/c3a4c6ff928b4c2f8198e2af8d0f2473 : SUCCESS in 46m 21s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/c39aa74e73b1449d9d4cc188772c354a : SUCCESS in 43m 32s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/2555523250bd44ce9ab230e474c689e7 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":5},{"id":"9bf32e2c5317681627f9df7defb9e4580a05d566","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 18:49:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0d06edbba50239d8fa254c6f3bcfb834978b4e65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 18:57:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c483c3f0a4ff4b96836f2e28b78904da : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4d0b6cd0d714712a632c7729d3138da : SUCCESS in 52s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6006de75a749556a071d23e12ca3a1e7e0ec58c0","revisions":{"8501a690c9405af8ad80f9fa60344e194e300bb6":{"kind":"REWORK","_number":1,"created":"2021-02-21 01:58:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/776817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/17/776817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/1"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-21 01:42:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-21 01:58:20.000000000","tz":0},"subject":"Update tests for scoped tokens and default roles","message":"Update tests for scoped tokens and default roles\n\nThis patch refactors the RBAC enforcement checks in the API tests.\nIt also updates those test for keystone scoped tokens and default roles.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/774524\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/775957\nChange-Id: I6fad03f5a89c213562918ca258884aac34ba7ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8501a690c9405af8ad80f9fa60344e194e300bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8501a690c9405af8ad80f9fa60344e194e300bb6"}]},"branch":"refs/heads/master"},"d7a1eb12419c6fddfd341718ea4a20beb1e3163d":{"kind":"REWORK","_number":2,"created":"2021-02-28 22:39:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/776817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/17/776817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/2"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-21 01:42:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-28 22:38:50.000000000","tz":0},"subject":"Update tests for scoped tokens and default roles","message":"Update tests for scoped tokens and default roles\n\nThis patch refactors the RBAC enforcement checks in the API tests.\nIt also updates those test for keystone scoped tokens and default roles.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/774524\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/775957\nChange-Id: I6fad03f5a89c213562918ca258884aac34ba7ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d7a1eb12419c6fddfd341718ea4a20beb1e3163d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d7a1eb12419c6fddfd341718ea4a20beb1e3163d"}]},"branch":"refs/heads/master"},"fab6e442f50a5cec15a79c42c855baf05898bb31":{"kind":"REWORK","_number":3,"created":"2021-03-17 16:39:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/776817/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/17/776817/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/3"}}},"commit":{"parents":[{"commit":"2d19795b84de70c1c64f7027caddf6dc19ad1678","subject":"Merge \"Change pool create scenario test to wait for operating status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2d19795b84de70c1c64f7027caddf6dc19ad1678"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-21 01:42:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-17 16:39:40.000000000","tz":0},"subject":"Update tests for scoped tokens and default roles","message":"Update tests for scoped tokens and default roles\n\nThis patch refactors the RBAC enforcement checks in the API tests.\nIt also updates those test for keystone scoped tokens and default roles.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/774524\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/775957\nChange-Id: I6fad03f5a89c213562918ca258884aac34ba7ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fab6e442f50a5cec15a79c42c855baf05898bb31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fab6e442f50a5cec15a79c42c855baf05898bb31"}]},"branch":"refs/heads/master"},"f4b5769f06cf492f96762b785dda0da0210d628c":{"kind":"REWORK","_number":4,"created":"2021-03-25 17:05:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/776817/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/17/776817/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/4"}}},"commit":{"parents":[{"commit":"41b3aac6b0386dae202abe8cc03455ea710e468e","subject":"Add stable/victoria jobs to check and gate queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/41b3aac6b0386dae202abe8cc03455ea710e468e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-21 01:42:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-25 17:05:55.000000000","tz":0},"subject":"Update tests for scoped tokens and default roles","message":"Update tests for scoped tokens and default roles\n\nThis patch refactors the RBAC enforcement checks in the API tests.\nIt also updates those test for keystone scoped tokens and default roles.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/774524\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/775957\nChange-Id: I6fad03f5a89c213562918ca258884aac34ba7ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f4b5769f06cf492f96762b785dda0da0210d628c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f4b5769f06cf492f96762b785dda0da0210d628c"}]},"branch":"refs/heads/master"},"6006de75a749556a071d23e12ca3a1e7e0ec58c0":{"kind":"REWORK","_number":5,"created":"2021-04-08 22:07:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/776817/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/17/776817/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/17/776817/5"}}},"commit":{"parents":[{"commit":"41b3aac6b0386dae202abe8cc03455ea710e468e","subject":"Add stable/victoria jobs to check and gate queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/41b3aac6b0386dae202abe8cc03455ea710e468e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-21 01:42:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-08 22:07:40.000000000","tz":0},"subject":"Update tests for scoped tokens and default roles","message":"Update tests for scoped tokens and default roles\n\nThis patch refactors the RBAC enforcement checks in the API tests.\nIt also updates those test for keystone scoped tokens and default roles.\n\nChange-Id: I6fad03f5a89c213562918ca258884aac34ba7ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6006de75a749556a071d23e12ca3a1e7e0ec58c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6006de75a749556a071d23e12ca3a1e7e0ec58c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
