)]}'
{"id":"openstack%2Foctavia-tempest-plugin~781461","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I68baab227f844a057f2abf77b295c3af287a333e","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I68baab227f844a057f2abf77b295c3af287a333e","subject":"Fix dispatching issues with least connection ipv6 test","status":"MERGED","created":"2021-03-18 13:07:04.000000000","updated":"2021-04-14 12:56:04.000000000","submitted":"2021-04-14 12:54:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781461","meta_rev_id":"f8baaf2eadaa87c313b2a9c540c9ae248437bd5a","_number":781461,"virtual_id_number":781461,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 12:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-04-14 00:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-04-13 19:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-04-14 00:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 15:30:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-13 19:15:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-04-14 00:29:47.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"6c0d16b9c265845d34b95b0c10d16822aa07a5fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-18 13:07:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f966a021fcfaf424f83f17d0518444b0087a0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 15:30:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cda3a1172b74eee83f33d9354f28c62 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a3483048a9e418491da672315519953 : SUCCESS in 3m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12aa8a1897f145c0b32e069af88253ef : SUCCESS in 1h 41m 08s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/307051dbae2d49dd9908b210895cbe2e : SUCCESS in 2h 12m 40s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/619e81ebe57e486ab93b2554a3c1f308 : SUCCESS in 1h 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a8cd82905ba03449e9444e5f4ad6e682d29bb903","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-18 15:37:56.000000000","message":"Patch Set 1:\n\nrecheck\n\nverifying that the test is fixed","accounts_in_message":[],"_revision_number":1},{"id":"ce9f1134a78b8a5b99a6d390673c2256aae0209b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:38:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1e3f8ffc364a06ba77c8d78e436f34 : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e1ac74d64c498496a2daf0d9fe2ddc : SUCCESS in 5m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64253a01a540444385acc747bae9ab11 : SUCCESS in 1h 25m 47s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/18ec9922cca84b718bd8e71d8707d77b : SUCCESS in 1h 31m 15s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/47156670337d4d4e9eeaad3192c19e05 : SUCCESS in 1h 39m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f95108290e5b4365952b5f3e1ee343da8eff15e7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-18 20:21:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"37cad0302b0b571f9600c4e0d7dc3f08ed42c59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:21:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/875045452999477ca965a6de522b20a7 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9d6e46324224431bbcfbc8b9a419c68 : SUCCESS in 5m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ac96bda0aed4840bab4a8e92277e7c7 : SUCCESS in 1h 33m 31s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3f43464dd2e047348becb7cf162e09ec : SUCCESS in 1h 55m 24s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/da648b37d4914b9aa8dd560dc3d86d9b : FAILURE in 1h 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9e86c97f705f79c1afb32e3d9577f6e4cee171bb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 06:29:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9ec019cbe912783809831ef0fda78051d4e11d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 08:52:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6f6f6bbdb74a309f99062901824d09 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b96e8eb24d4b9fa4298ce8a732cf6d : SUCCESS in 6m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d51884bd72444cb0b63700db47ef0414 : FAILURE in 1h 46m 20s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/8874fa69cc9e4f2599d6929dd487c4ad : SUCCESS in 2h 03m 10s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/14edd23891734d4089b645affe4c56fb : FAILURE in 2h 19m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c615d0eb3192dd80369676c0355c85fac840996d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 08:53:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbcacc1b141c1beeae8621058e96fba6349a96c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 10:52:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39c993293a2346e5b1703458c297cb6a : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518d0d01519849859832c2e6cb558b21 : SUCCESS in 3m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d6c208452a847ed809f593c6eff5e73 : SUCCESS in 1h 35m 20s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6ecc9e83b08b41fe9445bcbb701a257f : FAILURE in 1h 57m 19s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2a2cb5fffa5d40cfa2e325ac7f570e6f : SUCCESS in 1h 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6056fda06fdf45cb9440a53d4851b946432605cd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 10:58:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79074d429c5f6882025ab1e51ff52a96afac9f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 13:07:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091af0f4008147df99578932e98a2612 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5216ff90a7b341ebbd6be7cd67926896 : SUCCESS in 3m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cd1053752b84967816d007dca6b2aac : SUCCESS in 1h 24m 20s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6f5f889ac7eb42a5a1517239fa6711d8 : SUCCESS in 1h 51m 48s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ee04358fa1e74885a8f66eb9e0f6cc71 : SUCCESS in 2h 06m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0b754804d535bd2d0d44f290f5f6d8a1569f7e3d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 15:09:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4ee5b4c2d1ef522cf7b6b16f9732a6198c44f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 17:51:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8e1abdfe1a4e42a44d5e05db24c8c8 : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0faa233e9f1d438396f60a19bdecbf06 : SUCCESS in 3m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/480a073f9aaa4bd091dacda7f95a8eb9 : SUCCESS in 1h 28m 06s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cbad35f62ec7420a9e46212daf038a6c : SUCCESS in 2h 23m 24s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/12836dada8f64b77837f74902d05f159 : SUCCESS in 2h 20m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6bdec7f81f16b5f7ecc342d0e2cc6e9b13fe1fd7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 09:13:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"418cea58f7bd42d6c3168348ac5a2dd5f157a2c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 09:19:06.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"1c2741b16145d04650bee25ce253b87731c959b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 09:25:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"503881771e3d26c5bf50fb746d7b5b0b79c6f787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 11:49:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb42d7e3c6bf4949afe127808917db6b : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b7b50c85fb452499a3ac9d8815aaca : SUCCESS in 4m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bfe28af0cae148a6a630b2e87f3ccdd0 : SUCCESS in 1h 50m 59s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/334437cdbebd42458d464cea809c6b64 : SUCCESS in 1h 48m 34s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7d6c86332f9948f2a8b55c719fd4636d : SUCCESS in 2h 10m 10s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2a77194aa0dc4b048b253e9c103107f5 : SUCCESS in 1h 31m 03s","accounts_in_message":[],"_revision_number":2},{"id":"262fa02a6e525ff6dbb26fa79aa75b7f017e5305","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 11:56:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a8bcbf8c03cefc289a898182ba26a391dfdb5dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:18:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ca778998a24e64864a063a5d7ba99e : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a777ca301e4214952685166f3615e7 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1a3aa401fdd4b8e9fcb4f4fe7592ae5 : SUCCESS in 2h 04m 31s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/cf9b3d7ba426484397a15b40ba2bfcab : SUCCESS in 2h 12m 07s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5a00d9739a2545858c32a8d8fa284e41 : SUCCESS in 1h 46m 31s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/914e711520d64e64b0529a22081fc006 : SUCCESS in 1h 49m 45s","accounts_in_message":[],"_revision_number":2},{"id":"2b146df74e337e200705eb7bfa04195b5b25d53a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 14:27:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e9b16e1b53275764cd45a03ed535ee46efaca5dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:35:55.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876e023cf4a045279d56d2150581585b : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acac2a96aa17402f831a1dbce86cbd81 : FAILURE in 3m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/592ce1c9e66f483d9f543aa48d1e3124 : SUCCESS in 2h 05m 44s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/1301c66818d0435c98ca48044fa08406 : SUCCESS in 2h 09m 24s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a937fe98643342b8bffb951d72b3537d : SUCCESS in 1h 57m 23s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c08be8597dc946608a129d9eb93b6178 : SUCCESS in 2h 07m 39s","accounts_in_message":[],"_revision_number":3},{"id":"538f9873d61ef7fbb5ee6887295c589ff59ca5b5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 17:43:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d76874123f00d12999a3fe981b4ee87b70e296bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 20:04:34.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37c7d375c6b493ca26ccdf8a0e26d9b : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0abd5bee6bf4a0ea40c1114963fa70c : FAILURE in 4m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bebdaa862c804274819e15f23b533828 : SUCCESS in 1h 58m 10s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5f8dffce2ff944ada7b82ce660a7a366 : FAILURE in 38m 49s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9f88a58b257c4338b7bd9a1df8c3fd14 : SUCCESS in 2h 05m 28s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/41db3fc03a6546ec870607175f1ed992 : SUCCESS in 2h 06m 22s","accounts_in_message":[],"_revision_number":3},{"id":"e2b6dd9e37e665c4a07e7b15cfc4b892687bf99f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 20:16:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8705a9fb851375fa822517eeec83c227c38b60eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 23:20:23.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86be60850f954f6ab58ee16feba5fa58 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6a1ca2779243d8ab8f48e43586b9ce : FAILURE in 5m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3d91c030e96426e97428862a342036a : SUCCESS in 1h 46m 12s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/28f7db97b18544d8bdffdf2255166eba : SUCCESS in 1h 49m 36s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7552226b92a647c7b5e27bcd5f6c2fbd : SUCCESS in 2h 00m 14s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1ca0527c45c84cb981f65a18496afe26 : SUCCESS in 2h 01m 30s","accounts_in_message":[],"_revision_number":3},{"id":"c17151d579642316dbf0f1929c5691d824506d33","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-23 09:56:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"513e0ed32db1e7170e927f4d760f966f70e4345f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:56:44.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae785c56aff54efeb48db49b4fa0770e : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a1124e1bb241999ef2499212403df3 : FAILURE in 3m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbbdc25ad8ee488ab41efc087188c91e : SUCCESS in 1h 56m 25s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/453d5a76cab34549a03f938fbceb6c67 : SUCCESS in 1h 58m 14s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/840dbea75ec246b99fb25563857ba30c : SUCCESS in 1h 51m 31s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4477aaabc873493c85f7476bf14c178e : SUCCESS in 1h 51m 51s","accounts_in_message":[],"_revision_number":3},{"id":"b1bfb7c4335fd6361e0456b4914f07f52d1db88c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-30 09:38:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"145da99e469dc26f424478f9da3fda2f0045c646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 12:01:31.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880abcc94be24e5780df985913a56164 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e5d2d699374524adc167a968b2efc9 : FAILURE in 3m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/108ebd7c301b42aea9a4a0d5f3e54b10 : FAILURE in 2h 19m 57s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/79eb4081f5ff4dea91f6f364013e8c6a : SUCCESS in 1h 40m 56s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2c0e2f673ee14956a73f207e72216cf3 : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8f11fca869d140fdae4f1614eb5855e8 : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":3},{"id":"784feebeeb0ff05511149eb3044d89ba71327181","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-30 12:11:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5cd4514bf1cfc5688a4039014f902b94ad624142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-30 14:06:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"71cdc8c2839dd4819ba9e204751d8dcd77f89f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 18:03:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e14f5e5fa98414c84d504d2ab2ee3e3 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e0d658f519479187b839105943f2af : SUCCESS in 3m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/423581be95b04064aa52f4e010097923 : RETRY_LIMIT in 14m 21s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/4752e4a6818e4c9382ec3594d56e0692 : RETRY_LIMIT in 14m 42s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0633d2e5554b47369268c9c0f7af0a1f : RETRY_LIMIT in 14m 58s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3425ef4a51e04fe3b0090fdf0479f5d1 : RETRY_LIMIT in 15m 17s","accounts_in_message":[],"_revision_number":4},{"id":"e3f9b645c6d3ac44f51a0a1497334c8caf19b9d9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-30 18:07:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"197f464bdca608f04fcc6206666ca7fef6a173fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:03:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c16adb8e0e24517b6583d075b418234 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee7ddb6e9c04a20a80e2d4fab870576 : SUCCESS in 4m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edeb30f18c764877a0474fe85070632e : SUCCESS in 1h 52m 41s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/6b14d0d869b34480a3991debff79d00e : SUCCESS in 2h 32m 42s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/493cf072f0aa4953a02f4113f48b0f51 : SUCCESS in 2h 06m 00s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ef1597087f80498d8e064d860581dc76 : SUCCESS in 2h 03m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5648aa12de519499df405bc37acebf1aaab98f52","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 05:29:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e7e437fafda087c3651e565e53c932d3cd280474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:17:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de12b172321421eaca00fe05dce34a5 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1c1abceb844eeb84db31b425d4e376 : SUCCESS in 3m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f32615c1867e4c8798efe9b4be10e9b6 : SUCCESS in 1h 26m 11s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/407af881f6a142d6be533795778d153d : SUCCESS in 1h 21m 17s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/77e46a8edd774583bbd6ada0fb98b001 : SUCCESS in 1h 26m 08s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/32e8a1addb3542dc8eb107555216e8c8 : SUCCESS in 1h 42m 49s","accounts_in_message":[],"_revision_number":4},{"id":"ec74081b230108fa7961cd2a839e0e8e09f34a59","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 07:25:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"55aa1a17807b59a5815eac0e3a1c5fac6e21620d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:48:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6991184f51ce4ebe8a3d9f18ddb58332 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508c8ff243c64bf1a8366c49941c4d39 : SUCCESS in 4m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da9d45266e744e3cb62422a561062c25 : SUCCESS in 1h 52m 34s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/ac3fa035bed8473da11c85a13298b6cd : SUCCESS in 2h 01m 11s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1c6147c4b0f44dc19da767d712c4d202 : SUCCESS in 1h 37m 28s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9b77551418ee4b828dc7740f4de04f73 : SUCCESS in 1h 54m 08s","accounts_in_message":[],"_revision_number":4},{"id":"389877de2668aa6ba861f682da6da604e986b03d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 09:59:21.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"cab5777fb97086a0403f482dabcbff094a971c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 12:25:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0508fed7f95a4a7b9fada657ef17d4e6 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e2faee828b41beaae499d1583a9cd3 : SUCCESS in 3m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86145445ac3e4bc3b67f9d09d79454cc : SUCCESS in 1h 25m 48s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/19aeae6bd27a437ba24a48d0e3b8c62a : SUCCESS in 1h 33m 00s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3f880613ce834adbbf3f71512838cd24 : SUCCESS in 2h 09m 01s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a62b9ca5d25c4d778ee72a6e999d4956 : FAILURE in 2h 02m 49s","accounts_in_message":[],"_revision_number":4},{"id":"5a6b3c1277d88d52fc054371850511ac5bbf1cfb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 12:53:53.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"58c3027c8cfe5b63c809492049a3bf7ee8b08bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 15:09:51.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d96e492a664754bc6db358fb6905ea : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e25760cc04c44299f33554c0777e242 : SUCCESS in 3m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/054c17bac97b47e88dfd205b07f2c9b3 : SUCCESS in 2h 12m 58s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/72e4c55370284e89af4a10588646e059 : SUCCESS in 1h 52m 01s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/20ee5909e5524430981d0638ee3a669a : SUCCESS in 1h 54m 15s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b5606d1a8eb44c8bb6e07a006adda39f : FAILURE in 1h 48m 00s","accounts_in_message":[],"_revision_number":4},{"id":"0257e4b1e048c7f874a3b78ce519007b6628a50f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 20:41:03.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"80a0fd6035a332c02eb384a090733c1f87edba56","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-01 05:13:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"67cc391d80af9042d9149401873b8c3df6f4eed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 07:21:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bab749f710c4b62bc9139cb721317a0 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b7ac37312a44b8b60b6dfeedef0951 : SUCCESS in 3m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84b62895142d4c09960cc967b5d2af52 : SUCCESS in 1h 33m 22s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/dd3e637208b541c28fbba34057de2c9c : SUCCESS in 2h 05m 34s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/08d4d936ee9549b5a960f2a17c0d6404 : SUCCESS in 1h 41m 39s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6e75cc2693b7490d84aceca03ebe72c5 : FAILURE in 1h 43m 29s","accounts_in_message":[],"_revision_number":4},{"id":"99016ef489784d3d25b36ac3a41d9810e89f97ed","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-01 11:01:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4589e78537a279a63a1d3eb137ab5c5472732434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:27:48.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da57c79f1b3848fabdb8c2e36de627c6 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc06c0e05b8f4559a7618054300bfeff : SUCCESS in 3m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc8512cc43534721817ab59ea2112eb4 : SUCCESS in 1h 48m 07s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/082077a525ed4d989890dd8ad0d4c6d9 : SUCCESS in 2h 17m 45s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ee879079e3c744abac582946d1ef7e04 : FAILURE in 2h 01m 11s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/49f253977c6f4a20b6004763ac4aaee3 : SUCCESS in 2h 12m 54s","accounts_in_message":[],"_revision_number":4},{"id":"32143c7f60c0da64daf240d14bf2e7decd4164d7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-01 13:58:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d2e1f9b0864b6079aa02863fc960f2f8265afa1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:27:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c5fc86d77da4505bf6a0c9e366f3b3a : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8e46593520449c86d0615593296c9e : SUCCESS in 4m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20363be3e3e44b5aa027503da9b79aba : SUCCESS in 2h 07m 45s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/118083ce3e344a918f9ba4cbb75ce15b : FAILURE in 2h 11m 08s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/aaf0a1d3f07b41528f8c4dd6f03468d3 : SUCCESS in 2h 01m 19s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e84cc3ba18ad4d60adfe79429fbba4be : FAILURE in 1h 41m 24s","accounts_in_message":[],"_revision_number":4},{"id":"4eebe4a261b435bf2a17a97b510925fd99378d87","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-02 05:52:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bd8b333cf348ceaecdf0b0740228400b4f983847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 08:02:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521ac0bafda74fb9ad5b6aa252a5f14c : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b2d2a42e5a46289a6144ec62718ae4 : SUCCESS in 3m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac2e22a9c1bd44dd9dcfa4e6d051736e : SUCCESS in 1h 26m 32s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/d947347895b14d37bbbba3f3869bf0c8 : SUCCESS in 1h 45m 58s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/afcf85c2220b433a811facd89911098c : FAILURE in 2h 07m 45s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0aca1c100ed04a87890b15a2391390be : SUCCESS in 1h 45m 41s","accounts_in_message":[],"_revision_number":4},{"id":"d65cd914ed3892da52853e4551d8bac7cb59aefa","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-02 09:46:29.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bfb0e482b5990c222866dc54bd5e0bbf351d9993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 11:37:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b016a21f4a9d401aa3bf8626a72fa42b : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/916f434f0fde450d8005ab6a1073ae4d : SUCCESS in 3m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f809bdaaf9ac4c98b1934e803ff7a5a7 : SUCCESS in 1h 31m 27s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/14de9973c8604e1d969fca6b8cc1171c : SUCCESS in 1h 47m 33s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/945011cb5ec74d6695d9c5d73383b68c : SUCCESS in 1h 22m 36s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9ef3910d06804b62a85997cad53d4234 : SUCCESS in 1h 45m 08s","accounts_in_message":[],"_revision_number":4},{"id":"c8e160c8e002938ac8a95577d46a7008c4f981d3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-02 13:39:25.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"887fa358af058ce432256687e039149d498abc99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 15:50:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba234589db2448d6a98af9d91c3c7e29 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/127604a146f74a6da4d136709268d2bf : SUCCESS in 3m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d16807108e394766b66d10907619d7d2 : SUCCESS in 1h 39m 30s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/103396de444c4366a1dadaec9620c5dc : SUCCESS in 1h 33m 53s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c7b25b6e25714443b3e36ab3180729da : SUCCESS in 2h 08m 05s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b6dea30d42e04b2989b016f4596aaa20 : SUCCESS in 1h 43m 25s","accounts_in_message":[],"_revision_number":4},{"id":"48c4e58b9a178cae6bf6ba286cfe4e7955432761","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-07 08:25:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"50531d45eced4681b92a84dcee75f5e642f6ca94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 10:37:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1007db00271a4d7cad5829cb16c73d48 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151b12d9b9bf4825b4c17af3d4811bf0 : SUCCESS in 4m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4af9b0668561416c8208fe2446c5e2e2 : SUCCESS in 1h 46m 41s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/b9c966512d944850b8069771cb9b9769 : SUCCESS in 1h 46m 24s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/36db880a388b4feebbb33a990902a7a9 : SUCCESS in 2h 09m 41s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/59e98f8c8d374eb4bea8f55ae69581ab : SUCCESS in 1h 50m 23s","accounts_in_message":[],"_revision_number":4},{"id":"aea31bbdec9f96914101e0927018a187ddc53e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-08 08:14:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"77f65a185493fb5b5f50eeeb5cf39bee4da28373","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-08 08:17:49.000000000","message":"Patch Set 5:\n\nRemoved the DNM/WIP flags. I was never able to reproduce the issue after adding the delay.","accounts_in_message":[],"_revision_number":5},{"id":"c9c37b01bb0fb55369991355f37e1f76def08579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:36:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d320327030e04c679d438ed5aacce388 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3d899989444a55bdbd15a68ef5e659 : SUCCESS in 6m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c78c1677fbe431c8684183fdfbd87e2 : SUCCESS in 1h 55m 46s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/a2a4d5daae1949ea9fc150fc9168e240 : SUCCESS in 1h 58m 00s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d88b44b8ada64441a1aa55ee061b9495 : SUCCESS in 2h 15m 09s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/c641dc653496423eb3e1ca438945034c : SUCCESS in 1h 48m 27s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c3a37d27721a41f29d0d9b90cb03fdb2 : SUCCESS in 1h 39m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3c53760e33af4830a63f8af3834d3fff : FAILURE in 30m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/26f0c5c7c4714381a130e0ec684817ab : FAILURE in 52m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9864199229948d882ccde1ac0083936 : SUCCESS in 45m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/a29a11370e654d6784ba424b11886962 : SUCCESS in 50m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1ddf6aafa9c049fb8865639694d122b6 : SUCCESS in 49m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/638df92125d74e1096c7219bbe6dee88 : SUCCESS in 41m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0cf30f17e42543cb86154df495e1a34a : SUCCESS in 42m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/ae2390bb5a50459f8027bf23c7032d43 : SUCCESS in 38m 41s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/5f805d5f406446aca8c762044d180115 : SUCCESS in 33m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/df80cb37e907454aa39c2ce9f7573635 : SUCCESS in 36m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d61488c3a9ad41b1aa5faca347e14dd4 : SUCCESS in 1h 59m 21s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/742b29d41aee4d0189d4e9c93d5e26e6 : SUCCESS in 1h 01m 14s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/653fee34a0724f4ba2e0e4001f72d18c : FAILURE in 1h 03m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2b254cb75dacdbec90fc3cfc56fe023355940510","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-13 19:15:01.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM, Thanks!","accounts_in_message":[],"_revision_number":5},{"id":"b5679efd7f4a80cafedc1c450a97227840ba5f7a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-04-14 00:29:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nI hate fixing things with delay/sleep, but fixing the tempest-side issue here seems out of scope. 😞","accounts_in_message":[],"_revision_number":5},{"id":"14cc57d915bd09b2e243c103f2f8653c08938b74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 00:30:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c5b2a4b7df396cd28d4346c29ac45cc55d7580e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 01:01:18.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325aa62c8a0240f9948af0f655e68898 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1196cf02fdae41d8a1398bd46b66da32 : SUCCESS in 3m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2537ed31b314349a156fc411648b5d1 : FAILURE in 30m 04s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/b0502ac84a4e41338cb964b3948ad09a : CANCELED in 29m 58s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6d310f78a1194d6a831acde8c58d3804 : CANCELED in 30m 14s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/fc7307528994460e8a5920b9294d8257 : CANCELED in 30m 01s","accounts_in_message":[],"_revision_number":5},{"id":"00596bd49671b5c50b913a8fdf70f5b77159619c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-04-14 08:23:49.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3b094e54683b48eaa96fe69feed582a76a348d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 10:53:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36628ea334664496a4fcd856013e7855 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf907ae665f4550a8b8d558de702a6b : SUCCESS in 4m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d411e88a2394062994dbf3660172919 : SUCCESS in 1h 40m 20s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/358d3bfd441046a8b9fa9c2f3707fd2a : SUCCESS in 2h 00m 45s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a7d245d1d9f9499b88189e7f2739695a : SUCCESS in 1h 48m 56s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d443b7d6d9824bd3aa1a5ef6e3c55a47 : SUCCESS in 2h 11m 40s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/71ec7133bf754b1b8e3b24a79e072b58 : SUCCESS in 1h 59m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1332aa80b6834ad3a4152ee210d79b9d : FAILURE in 26m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1359106d023c4c8c8f55f04601caf009 : FAILURE in 31m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe7125a5c173404b929df96412041747 : SUCCESS in 37m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/788d18b863004a17b213c08c7accb01d : SUCCESS in 40m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/36805ed9720442118571283aac9ed820 : SUCCESS in 41m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/708c7ba1ddec48789800834cd2005d09 : SUCCESS in 48m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e59f2b33e94e49098c8cf5dfa507b7b2 : FAILURE in 42m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/9004011cb9ff406b8cff34174aa22110 : SUCCESS in 41m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/63b5ed1745614f028dafa8248f7a9c93 : SUCCESS in 58m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/8a4cf8d9fc2a4c3599f6960775c79a29 : SUCCESS in 40m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d805178ffd894eac9e4a64c35e95ee26 : SUCCESS in 1h 41m 45s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/679ba3dc6e00404b9f89075313823d23 : SUCCESS in 46m 07s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5e63216f45cb46e69ca115a76ef0c076 : FAILURE in 51m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bfbddf1a0289ecf65c2c73bb63e79fab78aadd2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 10:54:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"35b2d42fa34d209e9bbd7e62530432ae2c5224f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 12:54:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21fb1ec054b840d486ae3277d6455578 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc45a04115f84bf0988134224de86e17 : SUCCESS in 3m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce3b0270b229464d926c83cc5c838666 : SUCCESS in 1h 23m 24s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/bd2119ecd7604f78b6176997b5713658 : SUCCESS in 1h 28m 14s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/53575eb56da04a8f819cfcf836dcaa4f : SUCCESS in 1h 49m 58s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8c6d57dfe4774d968933d9880df9956f : SUCCESS in 1h 56m 43s","accounts_in_message":[],"_revision_number":5},{"id":"aac412a0fc150e62ba7360582c6f33a189b4ef41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 12:54:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"f8baaf2eadaa87c313b2a9c540c9ae248437bd5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 12:56:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13fb87a6671439ca9fbca4ee885fcf3 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b0b6767f1c9e8a4c68b569badd0425db5a2736a7","revisions":{"917984cca1916f823784cd0c95e801c7f0e52696":{"kind":"REWORK","_number":1,"created":"2021-03-18 13:07:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/781461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/781461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/1"}}},"commit":{"parents":[{"commit":"38c2a65c2720fd5786479a1b281b87561f90eef4","subject":"Merge \"Fix OVN provider jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/38c2a65c2720fd5786479a1b281b87561f90eef4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 13:03:05.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 13:05:24.000000000","tz":60},"subject":"DNM test adding delays between least_connection requests","message":"DNM test adding delays between least_connection requests\n\nChange-Id: I68baab227f844a057f2abf77b295c3af287a333e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/917984cca1916f823784cd0c95e801c7f0e52696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/917984cca1916f823784cd0c95e801c7f0e52696"}]},"branch":"refs/heads/master"},"02c3c42a8853931f6c6764892a6f14f69e7de6c0":{"kind":"REWORK","_number":2,"created":"2021-03-22 09:25:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/781461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/781461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/2"}}},"commit":{"parents":[{"commit":"41b3aac6b0386dae202abe8cc03455ea710e468e","subject":"Add stable/victoria jobs to check and gate queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/41b3aac6b0386dae202abe8cc03455ea710e468e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 13:03:05.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-22 09:25:02.000000000","tz":60},"subject":"DNM test adding delays between least_connection requests","message":"DNM test adding delays between least_connection requests\n\nChange-Id: I68baab227f844a057f2abf77b295c3af287a333e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/02c3c42a8853931f6c6764892a6f14f69e7de6c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/02c3c42a8853931f6c6764892a6f14f69e7de6c0"}]},"branch":"refs/heads/master"},"ca919473df706ed3d818ae3ab688c0f91c116579":{"kind":"REWORK","_number":3,"created":"2021-03-22 14:27:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/781461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/781461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/3"}}},"commit":{"parents":[{"commit":"41b3aac6b0386dae202abe8cc03455ea710e468e","subject":"Add stable/victoria jobs to check and gate queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/41b3aac6b0386dae202abe8cc03455ea710e468e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 13:03:05.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-22 14:26:44.000000000","tz":60},"subject":"DNM test adding delays between least_connection requests","message":"DNM test adding delays between least_connection requests\n\nChange-Id: I68baab227f844a057f2abf77b295c3af287a333e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ca919473df706ed3d818ae3ab688c0f91c116579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ca919473df706ed3d818ae3ab688c0f91c116579"}]},"branch":"refs/heads/master"},"5f99682f0b854848c38e3dfd579881d190c42e28":{"kind":"REWORK","_number":4,"created":"2021-03-30 14:06:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/781461/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/781461/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/4"}}},"commit":{"parents":[{"commit":"41b3aac6b0386dae202abe8cc03455ea710e468e","subject":"Add stable/victoria jobs to check and gate queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/41b3aac6b0386dae202abe8cc03455ea710e468e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 13:03:05.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-30 14:06:18.000000000","tz":120},"subject":"DNM test adding delays between least_connection requests","message":"DNM test adding delays between least_connection requests\n\nChange-Id: I68baab227f844a057f2abf77b295c3af287a333e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f99682f0b854848c38e3dfd579881d190c42e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f99682f0b854848c38e3dfd579881d190c42e28"}]},"branch":"refs/heads/master"},"b0b6767f1c9e8a4c68b569badd0425db5a2736a7":{"kind":"REWORK","_number":5,"created":"2021-04-08 08:14:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/781461/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/781461/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/781461/5"}}},"commit":{"parents":[{"commit":"434170a8e0740aae2d6c31a9063a5c8db7616809","subject":"Merge \"Fix testing IPv6 VIP network connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/434170a8e0740aae2d6c31a9063a5c8db7616809"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 13:03:05.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-08 08:14:03.000000000","tz":120},"subject":"Fix dispatching issues with least connection ipv6 test","message":"Fix dispatching issues with least connection ipv6 test\n\nThe IPv6+TCP+LC scenario test have randomly failed with an incorrect\ndispatch. Adding a delay between requests seems to fix this behavior\n(shamelessly inspired by the IPv4 LC scenario tests).\n\nChange-Id: I68baab227f844a057f2abf77b295c3af287a333e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b0b6767f1c9e8a4c68b569badd0425db5a2736a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b0b6767f1c9e8a4c68b569badd0425db5a2736a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
