)]}'
{"id":"openstack%2Foctavia-tempest-plugin~786275","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I6eb81eafce5c3bdd95b23c41497a8622d8432735","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I6eb81eafce5c3bdd95b23c41497a8622d8432735","subject":"DNM - Testing devstack debug","status":"ABANDONED","created":"2021-04-14 16:01:10.000000000","updated":"2021-08-24 22:22:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9554c4f0852a099f13b067308aa67e5d38f257e9","_number":786275,"virtual_id_number":786275,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-14 21:38:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 17:24:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 18:38:50.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"}],"messages":[{"id":"081fe69666c9bf53a5aa3319f6a1e6aa8e24f832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-14 16:01:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5545aee9fad346473adb5c1f35965d26ce5bf7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:24:09.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"a68b008b9fda34bd36defd007df274243ad14c63","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-14 17:24:29.000000000","message":"Patch Set 1:\n\nrecheck\nNew patch in devstack","accounts_in_message":[],"_revision_number":1},{"id":"18fc81a51522e0ffd648421535577ea5a32370b9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-14 18:38:50.000000000","message":"Patch Set 1:\n\nrecheck devstack changed","accounts_in_message":[],"_revision_number":1},{"id":"256ae6f6ca6c4aab003ab15b386cec9eb26a2fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 18:38:54.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f72eff6d222a66909c2b8044fa90552236d0170e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 19:38:05.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b5a5ebe17724351841ab1c63de1426244c0840a6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-14 19:38:07.000000000","message":"Patch Set 1:\n\nrecheck devstack changed","accounts_in_message":[],"_revision_number":1},{"id":"20e270f5ccb899e1353bce46258db8ae74623deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-14 20:46:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c83049dc3ce0ad8467aa618e161de7c689488369","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-14 20:47:15.000000000","message":"Patch Set 2:\n\nPrevious patch ran through devstack successfully, so making a change to start it again for another sample without waiting for completion.","accounts_in_message":[],"_revision_number":2},{"id":"36554ed4b9672f90717a1a0ce8ccd34fd2c8154e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-14 21:36:28.000000000","message":"Patch Set 2:\n\nsecond run in a row succeeded","accounts_in_message":[],"_revision_number":2},{"id":"c7a5f5a3f2df0cff41b6bc5eccde2c8821987350","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-14 21:36:35.000000000","message":"Patch Set 2:\n\nrecheck devstack changed","accounts_in_message":[],"_revision_number":2},{"id":"2709e63c6141f8522ece443597366b799b5dfd0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:38:18.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"7726a767816f0f00efa44ea6bc37c78303af5c8e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-14 22:14:47.000000000","message":"Patch Set 2:\n\nrecheck went too quick","accounts_in_message":[],"_revision_number":2},{"id":"9b849f7470598f820127465b77a3b96e74ec1705","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-14 22:14:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c210353aa025f40ebcd9d8db2d68cd489395c00c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 00:10:58.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98e68769b6b6428cb70071eac3d2b9e9 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f39fafd35eee461d9843981b5eca9496 : FAILURE in 4m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e5f2b2b37da4662b382b1d4e2d474f3 : SUCCESS in 1h 25m 16s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/c8ea7aa1852a498d9d3d55051453db53 : SUCCESS in 1h 19m 33s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/f1b9fb900b2d490989be9f808005c90b : SUCCESS in 1h 26m 39s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c3cd2c89f2b94029a94c817dea7c97ca : SUCCESS in 1h 15m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8130645e158d49919137e8ae105bdf6b : SUCCESS in 1h 52m 03s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/80e4c5cc324d405da85d45b31771a20b : SUCCESS in 1h 38m 23s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/da1c9c3bcb224465a5e3140e8a4daa44 : SUCCESS in 1h 26m 01s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d92f92389842499894781b47477e769f : SUCCESS in 1h 46m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ad0682167634466d99726d413ca55430 : SUCCESS in 37m 03s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/759e2936436740fe93f456b858743084 : SUCCESS in 41m 29s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/ec5d43e68ad545fea2abac1dd4df100d : SUCCESS in 37m 50s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ef87ca24b3ce42b89ffbc3a0e8370412 : SUCCESS in 41m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0defbf04aaff43c195623e6432b676b6 : SUCCESS in 1h 32m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9aec3402ecea41baa78ccfe3393672f9 : FAILURE in 22m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f8f0eb395e04d958c07f76cf25bb1d5 : FAILURE in 42m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/951afb2bc383412c861afe2bfa510f13 : SUCCESS in 36m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/66f4f31e9cfb458c87142f473dbf6f9e : SUCCESS in 36m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/457e89e63420445c9e26a33503854f27 : SUCCESS in 41m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/989f8a1bf6e94c7eb2c23796b592e562 : SUCCESS in 33m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/964fb8b55885465ea5f297e342926999 : SUCCESS in 34m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/61af1f65b2994c77862a27f4c68e6bf7 : SUCCESS in 39m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/dc0fe3e8029a4462a935ab3bd544a048 : SUCCESS in 40m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/28221d7ac277442fa4f320500a671107 : SUCCESS in 30m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/58e2246d352a42f6b0b62ae38eb567f0 : SUCCESS in 1h 41m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/cef6ba93d2eb4b9a902efa6f8b95d678 : SUCCESS in 39m 14s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b14b8ea7ee0d46609859d4ea50b75c6a : FAILURE in 52m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9554c4f0852a099f13b067308aa67e5d38f257e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-24 22:22:57.000000000","message":"Abandoned\n\nDevstack async is now merged, testing complete","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4c296863ab35fb4b5fb2f021cd6ec27df6026da4","revisions":{"6a6b4167d5ea03a0d0abe10f3e123e6ad6d220d2":{"kind":"REWORK","_number":1,"created":"2021-04-14 16:01:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/75/786275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/75/786275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/75/786275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/75/786275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/75/786275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/75/786275/1"}}},"commit":{"parents":[{"commit":"b0b6767f1c9e8a4c68b569badd0425db5a2736a7","subject":"Fix dispatching issues with least connection ipv6 test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b0b6767f1c9e8a4c68b569badd0425db5a2736a7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-14 16:00:28.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-14 16:01:07.000000000","tz":0},"subject":"DNM - Testing devstack debug","message":"DNM - Testing devstack debug\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/786250\nChange-Id: I6eb81eafce5c3bdd95b23c41497a8622d8432735\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6a6b4167d5ea03a0d0abe10f3e123e6ad6d220d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6a6b4167d5ea03a0d0abe10f3e123e6ad6d220d2"}]},"branch":"refs/heads/master"},"4c296863ab35fb4b5fb2f021cd6ec27df6026da4":{"kind":"REWORK","_number":2,"created":"2021-04-14 20:46:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/75/786275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/75/786275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/75/786275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/75/786275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/75/786275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/75/786275/2"}}},"commit":{"parents":[{"commit":"b0b6767f1c9e8a4c68b569badd0425db5a2736a7","subject":"Fix dispatching issues with least connection ipv6 test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b0b6767f1c9e8a4c68b569badd0425db5a2736a7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-14 16:00:28.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dms@danplanet.com","date":"2021-04-14 20:46:29.000000000","tz":0},"subject":"DNM - Testing devstack debug","message":"DNM - Testing devstack debug\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/786250\nChange-Id: I6eb81eafce5c3bdd95b23c41497a8622d8432735\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4c296863ab35fb4b5fb2f021cd6ec27df6026da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4c296863ab35fb4b5fb2f021cd6ec27df6026da4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
