)]}'
{"id":"openstack%2Foctavia-tempest-plugin~788954","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I4181a989514389ba18ff7874f8f4332f392d1189","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I4181a989514389ba18ff7874f8f4332f392d1189","subject":"Move ubuntu jobs to focal","status":"MERGED","created":"2021-04-30 13:50:42.000000000","updated":"2021-05-20 21:01:54.000000000","submitted":"2021-05-20 20:58:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"788954","meta_rev_id":"e717e746dc0c8a3b44507b558d78093fba540436","_number":788954,"virtual_id_number":788954,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-20 20:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-05-19 07:46:38.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-05-10 16:03:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-19 16:04:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-19 16:04:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 15:58:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 19:50:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-04 22:15:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2021-05-19 07:46:38.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-05-19 16:04:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-05-20 12:58:34.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"}],"messages":[{"id":"2bd24b85ff5534cbea463b2fd2a9027dabe68a66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-30 13:50:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a5ec1ef7537b98dee3a8459afd9905dedba106f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 15:58:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5482df687e1e41058bab39037e95527c : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73be4d8466d44f3197366e2b2ec51cec : SUCCESS in 3m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac89894efa804552946700d245222440 : SUCCESS in 1h 40m 17s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/73fa569d06da47b48d3fe6d88495ea7e : POST_FAILURE in 1h 39m 23s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/88662b1e962249818b747d9258959629 : SUCCESS in 1h 50m 25s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/49237f729e084a3b9734fd035dee1214 : SUCCESS in 1h 27m 41s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/3d12a060d0e4417495ed651df4c2604b : SUCCESS in 1h 42m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a31d26969084e74988d8b76dbf8ed59 : SUCCESS in 1h 28m 07s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/103f4f5a2333480c8acc107726e361f7 : SUCCESS in 2h 04m 18s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d61383d02b544286acfb77f751395ad3 : FAILURE in 5m 56s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ba99d516ef984405b223ecabfaebbb52 : FAILURE in 5m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4b7c85f92bf46e2b02dfcf0b0f91df3 : SUCCESS in 46m 41s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/9cb4e8f356ea43638af01624b41c7f0a : SUCCESS in 31m 39s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/58d03bc68bc9408785681c261c922ee3 : SUCCESS in 34m 18s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f1ad3a8e39024e5ca84d917d9619b877 : SUCCESS in 38m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/92b5116922c0408297ea719bbd584ac6 : SUCCESS in 1h 51m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b0fe176b041649f58d2baf0f94dec439 : FAILURE in 28m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2596d7f44414dbbbe3a555619cf54b2 : FAILURE in 47m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f07ae8142b3642c6bb0d56a7d4440a48 : SUCCESS in 45m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/306dbd7157d44d988ee90a65a5cae14f : SUCCESS in 47m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e1f4b7e1fbfc4f8ea3dc6ede1a3e1ff3 : SUCCESS in 49m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a90ae569817d4adb98beb1d1b955c952 : SUCCESS in 42m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d9fba68228a34da684cb330598ca2635 : SUCCESS in 38m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/943f6070f9d946c2a078246dd3b93929 : SUCCESS in 42m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ec3c93891d584f798d124f4496f47ad1 : SUCCESS in 50m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/19e906722d794227b262ae8124e355eb : SUCCESS in 40m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/75e0ca7bec344a96afb918c5c910380a : SUCCESS in 1h 43m 13s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/729942daa93a4216bd38d0c74252b285 : SUCCESS in 49m 27s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/9203ddc662ff47628d8402e179b8b3dd : SUCCESS in 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0186214d0afaad80d90893d4ab87403577815db6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-30 16:38:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1792b1f1e77a973d5429bc7a85ca2f10964339f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 19:13:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959730822f2f4b5ca4db2f631f1bb81e : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa195b858c8041049b69e2bb5da91d79 : SUCCESS in 4m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae768d6c3d6b437bb813a6c01f0497e0 : SUCCESS in 1h 07m 11s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/38928fc7f85a4a858489c0ecc94bcfa8 : SUCCESS in 1h 34m 14s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/bacd5d0661b943e09c79810d6f917fda : SUCCESS in 1h 27m 54s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/923ac440277e4fe9bd1f256560a05468 : SUCCESS in 1h 27m 29s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/c4d47c54c3ed4a958720cbde23627391 : SUCCESS in 1h 45m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fae5f4228004c8ba578672997028da7 : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/69ed5b84c0fc4dc69242e87d519d1d6b : SUCCESS in 1h 39m 19s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7a2f460378e44b8fae2f8f747a94fd33 : FAILURE in 1h 37m 38s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/bd38e45285b64e0f85e8c41426e4240c : SUCCESS in 2h 26m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/28f3a166f59e4313ab1794f5215772de : SUCCESS in 43m 55s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/f5fd9b1eb30349feb8b7bca5a2847edb : SUCCESS in 47m 57s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/e784b397614e4add9fa6feac663b6459 : SUCCESS in 43m 19s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/d502616aa60b4470b5b1afc539472881 : SUCCESS in 40m 23s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ffd429c805c7407a93e632cc1a692ce9 : SUCCESS in 1h 46m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d0a03cce869b4b8da911f5e6b8519e33 : FAILURE in 27m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1aea04cd1a6e4797b1cf5e233cd2deb2 : FAILURE in 1h 02m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f29269da916e4f0bb8d434172a9eaddf : SUCCESS in 59m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/a1f56916bb7341388c383daf36f528d7 : SUCCESS in 43m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e6015aea60e846a585fff117e842d72a : SUCCESS in 45m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8d978223148f4fb78a91d7ebb39133d9 : SUCCESS in 36m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9cea902a7a1143c98941c3565432334a : SUCCESS in 42m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/9e7c56150c2a4604a450215974ff83fe : SUCCESS in 45m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/63a1911a48464690b3456f2652d13962 : SUCCESS in 37m 24s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b6478ba919bd4c0c966faedd7a6a8d3c : SUCCESS in 33m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/470c134148324c3d9caf9a3988566e68 : SUCCESS in 1h 42m 39s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2eab6f4c385240a1a2bfcd91f9274a27 : SUCCESS in 47m 48s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4495968dd71342b8b005cf3b8c5d56a7 : FAILURE in 1h 08m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"766867c9caf2feb2a7e4bb3035d5ae5e4c5d6e79","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-02 11:40:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"58a61d4201fad74f39034b1f8bb009f2df915898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-02 13:49:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f44607546c047fcaa46aa1c5dab24c9 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6944d1b5a340410c9e78f1dba588fada : SUCCESS in 4m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2037fac08ac743eeb26f5a0225eb1bed : SUCCESS in 1h 23m 38s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/441a2c021fb8464da84523149292dfc5 : SUCCESS in 1h 30m 13s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/3c0ab42e91e44ea58dcf4985ab626b18 : SUCCESS in 1h 30m 23s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/bafd35c2e12b451b8a8b0dcebc80de17 : SUCCESS in 1h 11m 47s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/a28942f3163c4a698c20a4e6b3ea6551 : SUCCESS in 1h 30m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bea77b71709a45439a88bb4b4096f24f : SUCCESS in 1h 44m 21s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5dd2733b5c8b441fb531df44fa1b164c : SUCCESS in 1h 28m 43s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/384ab20c6ade43aa8c721f401aa420f9 : SUCCESS in 2h 00m 57s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0de6276ac8984b93a0bed7b187ea9b93 : SUCCESS in 2h 07m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/525d680a84ef4ee98716f61492066c14 : SUCCESS in 52m 23s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/e8e949042c35419dbc11c8a57ee8195d : SUCCESS in 47m 56s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/68b6e58512c640d38ef681defbe2a8aa : SUCCESS in 54m 01s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/6adf9f326e9c49d5a680e36c22a2c1a0 : SUCCESS in 38m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2c1bbb2760a84af5b8fd929b8fab191f : SUCCESS in 2h 04m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5fdc102761bd44cf921b20c1ff6a3d6c : FAILURE in 27m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bac4ddef65a748ea91084d58e30fe9eb : FAILURE in 50m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3073e0f24a84454b6f3efcd883017f6 : SUCCESS in 33m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/df56813d20e943ef9dbb9505efe393b3 : SUCCESS in 45m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/21bcad5d2bb343aaa8d4bcaf93f6f393 : SUCCESS in 38m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/18029301a9934bb1896234fb1e9563da : SUCCESS in 36m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ca0f3669c0a14982ba375c44959d226a : SUCCESS in 45m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/685afd7ee3d148dcb604160f8d67912b : SUCCESS in 51m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/2c165efdd7d849c39793e1b549790616 : SUCCESS in 50m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b93350ffdac040528b587b7abbb8058e : SUCCESS in 52m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/50df6aef47724808bbcad8ec38639182 : SUCCESS in 1h 39m 05s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f7377f577be4476fada5b2414e694623 : SUCCESS in 50m 18s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/03ca34eb2d7a4310b4b0863a27739382 : FAILURE in 54m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"adbc789e74d48798e80026597deed05ca718b67f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-04 22:15:47.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nIs this no longer WIP? (The title still says WIP)","accounts_in_message":[],"_revision_number":2},{"id":"492a88ebfd4fb7a0984a29c253f9c3dee6f79799","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-05 06:54:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4b77558bdee5b4b0e4b28a5dfe2af9852ef90beb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-05 06:55:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e1025ddebbb5ea93462ed2ec6ee50465613286c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 09:04:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4cb960bf6a41c2aa45eb5836e67d67 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78be5a4dc0d494fa4da33e98949ef1b : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10a317bc87ee431e81134f4feb6a49a0 : SUCCESS in 1h 52m 59s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/9a3776a0269740848626bfe6b0502495 : SUCCESS in 1h 36m 54s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/2ff2c05be24f4ce59c6bff06bb6689aa : SUCCESS in 1h 19m 17s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/250f55722d5b44bf9a17b1b31719ccec : SUCCESS in 1h 26m 49s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/bfe9241896c64aeb8856bc84fdc19fd4 : SUCCESS in 1h 51m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62f9607aceb44e24a414c8b6dd223c7c : SUCCESS in 1h 26m 31s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/35233341bb3d483bb5581e15a9aa9a66 : SUCCESS in 1h 35m 41s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7adf72e1301145ff86be17afd014bbfb : SUCCESS in 1h 52m 30s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/9165068610e74ffcb251af3fe40b19d6 : FAILURE in 1h 22m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82a2278cbc5e4d24b3ce870f5b5a8d81 : SUCCESS in 36m 27s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/d3db5fcec1924f14a71dcc5c18060b91 : SUCCESS in 39m 38s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/3b180109156f479490958aaba9de667d : SUCCESS in 45m 57s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/eca53d27b09741ebb4b0e38a20ffe349 : SUCCESS in 39m 27s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a024e52da0845fe982b527a6cc92dde : SUCCESS in 2h 01m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/42d04a14801346debd583e7e8fdbdb53 : FAILURE in 31m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac059dc8e9924b31851b3bd8eb085ef0 : FAILURE in 39m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/666c50da7a9844a3adc06cd529775e97 : SUCCESS in 1h 05m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/01866c63b5404c06abd08841324066d6 : SUCCESS in 45m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a1c3a29625da4e8f926abe814a18c6bd : SUCCESS in 48m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ea35de0423a6450ba7e446922194e0b0 : SUCCESS in 57m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9a9b79762a2c43f7860bcca6330bf5b7 : SUCCESS in 38m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/6ea8b28ea6104ebd9d0a39026982fceb : SUCCESS in 40m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/39f69e644eb840b38a10b0ce740bb420 : SUCCESS in 1h 05m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/e9c23b19793f40e09586b0e97b4c36bf : SUCCESS in 51m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f76f77122a8d439ba3bb804754d57062 : SUCCESS in 2h 08m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/8fc3dabef23a4a0e80c17634e99233df : SUCCESS in 40m 13s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/50f42b3e352148c8969893a4c6c10080 : FAILURE in 53m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb883a72f563b48472d8e3ef1fa4cd7cfd58d5c5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-10 12:46:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"761d76a7d24c490296790e07a5976e2f279e8fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 14:49:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46fdd0ab2aa44ec82999b56e70db25a : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7273af661304b51b82d0c9318392d2a : SUCCESS in 4m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/415f48916f324d2482ff96c721c13ef1 : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/356001ce133d4f39aa4633eca0b3128f : SUCCESS in 1h 31m 02s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/4a286374d61340b19393a99ebefae339 : SUCCESS in 1h 11m 03s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/25064a1b5db34a83ba06210aea155dee : SUCCESS in 1h 04m 09s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/dc0cb9099e014d86a1df03e3cb897d1f : SUCCESS in 1h 19m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a316118f3d164774adb5f5dd230efaa4 : SUCCESS in 1h 15m 18s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/f2fc54badbd846c5bc938731ec3d478e : SUCCESS in 1h 34m 35s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/4d95ba71bdba4f4097a5cf0af6db7419 : SUCCESS in 2h 01m 54s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/387e77e6aff94b24bafb3c0860d89a6d : SUCCESS in 1h 27m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c6e42f204f34455b85ec94d27ae6c7ec : SUCCESS in 35m 05s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/9ee6074d9bf244d2a14aed9c093b55eb : SUCCESS in 58m 10s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/668ad78022c24e3ea147a6b192157354 : SUCCESS in 34m 35s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/5caeb4757e7146a0a2bfefaae295922c : SUCCESS in 37m 09s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0cddfd0d5ddd46ccb022868a5fccc795 : SUCCESS in 1h 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/867dd4472b9642048c3b927857312b49 : FAILURE in 23m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61b5149434b04ad4843d19567e5c802d : FAILURE in 42m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87ca2235d80b48f7ad61a72b92a85112 : SUCCESS in 30m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/0950ddc2c2d84e8cae06356fb346616b : SUCCESS in 39m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5ec9553d0ffc4247908156d08c052268 : SUCCESS in 40m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/81f568fa673a45cc8ab0472de6475cc4 : SUCCESS in 37m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/14a4a5f672954ce8a1ab1486a6de417d : FAILURE in 41m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/9ffda8b51aa9499f9debcb3eeeacfb49 : SUCCESS in 38m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/c26fddca86e54383834de2120f7993dc : SUCCESS in 33m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/2b82e1484f194c8dabc315b1a1f68b6e : SUCCESS in 35m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7f4854d52df64456bc6eb26bbea84ae5 : SUCCESS in 1h 28m 02s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/70907023d37f49c090e8ee9692b0d05c : SUCCESS in 39m 22s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/10ebdf3742954c8a8e248c4e669a6ea2 : FAILURE in 58m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4f28a12def53c4ee2eb73a1bff2aaf47ab8d74a0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-10 16:03:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fdd15408b638dc2ae035c33d1092f49a66c04ac7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-19 07:46:38.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"141ac53a80f346d469982459fe73e10db854f2e2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-19 16:04:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"dafa025153801df16d8099abfe65ed298c3d3cf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:04:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da99f9e711d63ac0af360e2c1b770225a1aed3c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:22:17.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2bc4ae5cb5f404e976b8075088ee8bb : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f17686f291d14753894e702268dfdf00 : SUCCESS in 4m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95a8379c98684c49951c4f6c73659be6 : SUCCESS in 1h 47m 57s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/5d3c6030aca34777afe81666d26ba0b6 : SUCCESS in 1h 48m 28s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/7fd8a803301348ab99a123486925ab9a : SUCCESS in 1h 27m 32s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/ba5134f45b8e4b3bb6d0e247cd572682 : SUCCESS in 1h 34m 37s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/171c39f7ec7c415bb2b934fc2d7a6704 : TIMED_OUT in 2h 13m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e46987cdf2349db967146290279d1b0 : SUCCESS in 1h 32m 19s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/40c29f4e82484d88a78a5b02465c56e4 : CANCELED in 2h 16m 56s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5f3994cd1b644f4b9f571dc90216af67 : SUCCESS in 1h 42m 51s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1783e2ae5820403f9146098083aba6ea : SUCCESS in 1h 51m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3053660a3de4985a8768e804df9b2cf : SUCCESS in 44m 37s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/d72335979c2c49e49a76dc0652997998 : SUCCESS in 37m 00s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/149f581ebbf8499686d1051a2b5b0830 : SUCCESS in 43m 20s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/a9eb7934db044a5a9708704016876450 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":3},{"id":"076a6b6bc2cbe27ff0c84992464d2540e1117b20","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-19 18:23:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1833576c55f1198205125c3ddcbf752240fe7875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:25:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f48a2d9cfea43eaa0519b34477ffaac : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6237c8508e79464ea8c19506f700cdaa : SUCCESS in 4m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a37f6ffffcf1468c9ae991e6c9cdafb4 : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/cbfc9faf8f5f4d798f6ad94a8ecc2395 : SUCCESS in 1h 31m 01s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/57783436b0284142a30919e01216fc7f : SUCCESS in 1h 12m 40s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/2552eb2570414b97adce7b0eb05fbd31 : SUCCESS in 2h 01m 30s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/6e1bc76f4c35420680e216b2ec3825f8 : SUCCESS in 1h 53m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ae75d9d3dc04fa7825b501edbeb1b00 : SUCCESS in 1h 26m 47s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/9498c816cdb54504acfa5f7f93be4b1a : FAILURE in 1h 32m 31s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a17f4a5c5d7c4005bd6c7dd98fd0c118 : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4fd5684993224542831b0b281a4737be : SUCCESS in 1h 56m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/50ff8d877e444fecac640921d9d162c2 : SUCCESS in 34m 56s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/d88d6dfb07e44ce8aa3c22ddd0c59acb : SUCCESS in 40m 03s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/c43d0e9de102445e927c0a03bc64bfb6 : SUCCESS in 37m 52s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ddbddf479aec433da27e9fab17019b95 : SUCCESS in 36m 24s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/879dc9d004294215988004c07a01ea5c : SUCCESS in 1h 25m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3bfd18778b1346eca69290ec8967d2d4 : FAILURE in 9m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62ad05a8b7d04474b0bbc5b15a7d17d0 : FAILURE in 44m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c4239cda36448c4b77b320c580d90c5 : SUCCESS in 36m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/95008aebe0e84c05a2c188e36b6f370f : SUCCESS in 35m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/de2c55de1175491fb27ea43212b9c142 : SUCCESS in 39m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d259b23d00da4bcab93c6ce07f0195f2 : SUCCESS in 34m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b55f92b996194c5092317eb9412c17f7 : FAILURE in 43m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/0db4836587094a458efddd828e4acc00 : SUCCESS in 34m 02s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/701262926ed54a8c98370f4ea31fdf38 : SUCCESS in 37m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/cb03105931dd407f95f261aba5c4e05c : SUCCESS in 36m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0969de9171e149ddad16fee5cfa30774 : SUCCESS in 1h 36m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/459f0d5a03cf4170b9b407be62f7cb35 : SUCCESS in 40m 04s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b3e1bccee26d4e858faaafc9d79ce42b : FAILURE in 59m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a9713c20ecb2d769a29e97a6b9b722a0c887cb12","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-20 10:45:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d84f14c7765b8a8d06fe6a1d5f87f0b4b23d39e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 13:29:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee662ec8227c458aa991452b4a670a25 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9284c46ec874f4f8fe7a06fe04cecf5 : SUCCESS in 3m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/348473c114a4432fa9c5c4b1e574ff3f : SUCCESS in 1h 24m 55s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/3719028f86c84b57a74d506cd00b1224 : SUCCESS in 1h 26m 11s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/d9f076793b5d4fbdaf1e372eb0ec2fb3 : SUCCESS in 1h 06m 02s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/f31a90b1aeb742c3931d298223cb72a1 : TIMED_OUT in 2h 16m 13s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/53e6774a06b04603a6bf3d8fd4205a29 : SUCCESS in 1h 34m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb8a0f4108fc4c3aa4a54b2eae6b30a2 : TIMED_OUT in 2h 41m 13s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/43c792377d0343b985f476376edcaebe : SUCCESS in 1h 52m 02s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d1fbe6367c64451a85ba395a4c805b2b : SUCCESS in 1h 52m 24s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/ef78832ef8a246bb8c2b6a32b327491d : TIMED_OUT in 2h 35m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a35de1547b2844dfb1285ac016de68c2 : SUCCESS in 43m 56s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/dcf4e497c0ac4a398d28f346b50dd182 : SUCCESS in 38m 39s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/a12672af86cb4db3a40b1c761a21a61e : SUCCESS in 45m 56s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/e4da0996be1548ac9c9c5d88a7b9aa13 : FAILURE in 2h 04m 41s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d8b037ff1aa94487a5204a9be176f08e : FAILURE in 2h 21m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3d6a63434b9b425288d85918a359cd28 : FAILURE in 9m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da25f935f47440ee9a822b2e34b9e6d8 : FAILURE in 1h 05m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b501f5f3fe4432fbd7522d4dccc4281 : FAILURE in 1h 04m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/6d9592d9317f446b886f572bc5a36237 : FAILURE in 2h 18m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7e727ad9c331435dbc0b60f4f4ea39b3 : SUCCESS in 46m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/e9d92fa5ca1c4eeaa146c3798ec66bc3 : SUCCESS in 45m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3e002387f72a402a90219c7e53587a89 : FAILURE in 47m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/b27b5bb171df40deb515e5cb31f5d23f : SUCCESS in 36m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/8fedb3e8627341ec807c3cc058e80945 : TIMED_OUT in 2h 42m 21s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/0940752f3a4343deb5d976d60122e635 : SUCCESS in 38m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ac0734d153914d8db8ebd73d62216441 : SUCCESS in 1h 55m 31s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/0489ac9b18d4441ca383d9d094def0f1 : SUCCESS in 42m 50s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5f53edb22b3d4ce8a028221a3ea31816 : FAILURE in 1h 02m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49cfcdbe364701b0b7f6896656f61c99e4d85a23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-20 14:42:05.000000000","message":"Patch Set 3:\n\nrecheck\nSlow node","accounts_in_message":[],"_revision_number":3},{"id":"3d63533a27005326da55caaea93afc12b1613ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 16:43:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddcbe27f16b3474eadecfaf0383bacd5 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03b9f86b1bb545d49551fcf69169f5e3 : SUCCESS in 4m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fee8e877fbb0468ea7e764aa9b104c96 : SUCCESS in 1h 28m 17s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/a74f6c1b890f42efb5391bb4a94cd60c : SUCCESS in 1h 32m 15s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/b04ea08e16194b3c9cd342acad585878 : SUCCESS in 1h 17m 49s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d151bb3fcda04ad38acfc2afbd662cc2 : SUCCESS in 1h 15m 22s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/dff0a5203c374d80b3c6429823a36a92 : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ebabd97cbfc4f5091af61ae6d1d5a28 : FAILURE in 2h 00m 59s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/8a57d96f416c445d9283b568d55d921b : SUCCESS in 1h 43m 46s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7b5ce3b93bb44a2fa4f7c677af519017 : SUCCESS in 1h 42m 08s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1ea46ff34770430abe91c835f25234fd : SUCCESS in 1h 35m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4ac40d1b50774adfbaeaa6a9d8847ee7 : SUCCESS in 50m 18s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/2308236e21d94d0a94493044b6273c44 : SUCCESS in 36m 19s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/670e265723db49f685bd654ec355e4ba : SUCCESS in 35m 46s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/8c7d053013e74e34992ccf0dd6774a57 : SUCCESS in 32m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c6c7974135f4c0d8cb5cdada13f5f83 : SUCCESS in 1h 55m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9e2c625f20174932a07adc759ef641e5 : FAILURE in 9m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5bb7c9d857449388b7614c2870d8fb5 : FAILURE in 58m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad9c3a9e38fe4d179f58f348c87ddf6a : FAILURE in 55m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/c30b636f2d3846f7a2226ca4b0a7f828 : SUCCESS in 37m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/847e03d207e0492fad08fc016cf8fe3f : SUCCESS in 36m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/286436adff95465dac6290ede911c733 : SUCCESS in 35m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6489159fea7d400aa7816f2d68713462 : FAILURE in 39m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/7c3eb902f2fb4fca86b5a96922200102 : SUCCESS in 33m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/09d92585a8a848d0b6b3e65d8970cf1f : SUCCESS in 33m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d045c16218654756984d4acb8bac06c5 : SUCCESS in 35m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4082cd87ea9c464481c7a62398f56024 : SUCCESS in 1h 25m 41s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a719dd4948d44e88b8c949986d0e2d18 : FAILURE in 51m 57s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5a38ecbb3f3b4a81bafec67b95df9df9 : SUCCESS in 55m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"903a1ff11d98a6cb582eaf2d7415ee3c77b5745d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-20 16:44:50.000000000","message":"Patch Set 3:\n\nrecheck\nneutron issue","accounts_in_message":[],"_revision_number":3},{"id":"b807a380d19b61ca29b95bd9869785f893ad973c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:46:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b2e7aabf7a44d19d44de39518debe5 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/209df469bc7a4ddc84d16e0d6fb89dff : SUCCESS in 4m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f020154aa0be4af5aa6b2551f49da23e : SUCCESS in 1h 19m 12s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/08dfface2102490ba53ca016fba4339d : SUCCESS in 1h 23m 39s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/2c7bcf83f83f4aa296a2a71aead243fb : SUCCESS in 1h 30m 48s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/a3d64fa5096349059e2f57eb620d18ce : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/799f46442e6e4304bcca8888d1815ec8 : SUCCESS in 1h 33m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89e45a41a8434b48868453d25a0b6ed8 : SUCCESS in 1h 32m 11s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/db82ad12beab4f6c986097c9661e0e67 : SUCCESS in 1h 59m 08s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/f6c389c293894ab1ab8e2c6933e377e5 : SUCCESS in 1h 45m 18s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/babda5c161ae4f9792f3b7dc6388ea29 : SUCCESS in 1h 22m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d050b8b45064d0e89a76011b8198955 : SUCCESS in 47m 09s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/97b312157b2b4cbab76411f3baf0cdf1 : SUCCESS in 47m 10s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/201b3e0a9bf049189d826c12bfc92a5b : SUCCESS in 37m 56s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/0d188943ee9c434e9ee604f3c1eb53d6 : SUCCESS in 42m 30s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0e20979e2caa4e4f81abba7a6dd302a4 : SUCCESS in 1h 42m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d2f7552fc92f4c928d1e593d027ea421 : FAILURE in 7m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/54574b8bba3a4718837b51fa197c8511 : FAILURE in 44m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71c7eb2c3b7a4223a4abb23c905832e0 : SUCCESS in 32m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5330df94d79d4c8ea64c35560c4c1190 : SUCCESS in 32m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/866b012f7b2947c0993fdc535fc5c885 : SUCCESS in 58m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/93e5ed5b28d94b9f8b50d29c6ef18027 : SUCCESS in 40m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3eb2a177a56446a1bad75e3d036829dd : FAILURE in 1h 08m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/9958a2e3dbb94f72990d7935b44b219a : SUCCESS in 39m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/e16d1aa118eb4f61a22b227de748bfe8 : SUCCESS in 36m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/78a2622f85df4065a889733194b5ac52 : SUCCESS in 39m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/751fc1d0d405417a8294e3ef822c25d7 : SUCCESS in 1h 47m 09s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/ae5d1f68adc347abadfd25401114b371 : SUCCESS in 43m 05s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/fa3c5e838f054d94a51270dbd5fd7135 : SUCCESS in 48m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5566ce5951276480663dfd7add7d5583a50d644b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:47:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"60e1ae39ebea1af20d9b801827b23e7801b6011c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 20:58:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f69d9dcdc94f40a39fdb20e022937c : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8556aed69e14e3dab5c8b70a53c9d0b : SUCCESS in 3m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1cc7d2c2f33c4d858a08c7fe8044524c : SUCCESS in 1h 21m 37s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/9c519802f3d64afe8d66ad3d06348c0b : SUCCESS in 1h 33m 21s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/9cdfe851ac87486688b822f9fb2d66c3 : SUCCESS in 1h 23m 38s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/8e17e3004c534b3cbbe2900a9d028db7 : SUCCESS in 1h 17m 26s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/5f1fc49fa8ce48af94a9662f4cae58c1 : SUCCESS in 1h 21m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f229ffadba3471aa00b8e2c1487aac0 : SUCCESS in 2h 09m 23s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/cd7a479e708a4e948b61e79a5c200681 : SUCCESS in 1h 38m 27s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/191d17bf980d4e81bb9a56e4dd7dd29e : SUCCESS in 1h 47m 44s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/7dd9fcc818364a42b90a35d055b72115 : SUCCESS in 1h 42m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf596584d9d24936a10dbf6be5a4dad1 : SUCCESS in 29m 47s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/af048d639a5045a98c4f3c1301455df6 : SUCCESS in 34m 30s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/e033974541a3498090f51885777c981b : SUCCESS in 41m 59s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/3188291bfba4438d80bb45130d520625 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":3},{"id":"66e6faf477a871fbf324436245ee39f34e677136","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 20:58:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e717e746dc0c8a3b44507b558d78093fba540436","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 21:01:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb39f52a8134dc7a1d0b648cd493a9c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3079d00ee6d5f3b5eed61317e1802aeb97c04d45","revisions":{"b78f7e3a4a540271170bbb0a05f41ea34b2250cd":{"kind":"REWORK","_number":1,"created":"2021-04-30 13:50:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/54/788954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/54/788954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/1"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 13:38:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 13:42:21.000000000","tz":120},"subject":"WIP Move ubuntu jobs to focal","message":"WIP Move ubuntu jobs to focal\n\nMove master and victoria jobs to ubuntu focal.\nussuri and train jobs still use bionic.\n\nChange-Id: I4181a989514389ba18ff7874f8f4332f392d1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b78f7e3a4a540271170bbb0a05f41ea34b2250cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b78f7e3a4a540271170bbb0a05f41ea34b2250cd"}]},"branch":"refs/heads/master"},"6fa05310f55bb6fe751499e0f1c83cc8f2e07b33":{"kind":"REWORK","_number":2,"created":"2021-04-30 16:38:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/54/788954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/54/788954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/2"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 13:38:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 16:37:41.000000000","tz":120},"subject":"WIP Move ubuntu jobs to focal","message":"WIP Move ubuntu jobs to focal\n\nMove master and victoria jobs to ubuntu focal.\nussuri and train jobs still use bionic.\n\nChange-Id: I4181a989514389ba18ff7874f8f4332f392d1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6fa05310f55bb6fe751499e0f1c83cc8f2e07b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6fa05310f55bb6fe751499e0f1c83cc8f2e07b33"}]},"branch":"refs/heads/master"},"3079d00ee6d5f3b5eed61317e1802aeb97c04d45":{"kind":"REWORK","_number":3,"created":"2021-05-05 06:54:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/54/788954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/54/788954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/54/788954/3"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 13:38:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-05 06:51:53.000000000","tz":120},"subject":"Move ubuntu jobs to focal","message":"Move ubuntu jobs to focal\n\nMove master and victoria jobs to ubuntu focal.\nussuri and train jobs still use bionic.\n\nChange-Id: I4181a989514389ba18ff7874f8f4332f392d1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3079d00ee6d5f3b5eed61317e1802aeb97c04d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3079d00ee6d5f3b5eed61317e1802aeb97c04d45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
