)]}'
{"id":"openstack%2Foctavia-tempest-plugin~789829","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I0ac83c5fc9446db700f122b8ee7adaa79c67c95b","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"change-770163-17","hashtags":[],"change_id":"I0ac83c5fc9446db700f122b8ee7adaa79c67c95b","subject":"Updating _test_pool_CRUD to enable testing of updates to the load balancer algorithm:","status":"ABANDONED","created":"2021-05-05 13:12:01.000000000","updated":"2021-05-05 16:03:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c0dc0fcf9708d4c58cf92692de18b4abbef080f","_number":789829,"virtual_id_number":789829,"owner":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-05 15:24:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 15:24:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"324c978b4f7766448d5c00634c8fd00cb2976d5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-05 13:12:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afd42586417acdcd354315d035b21068abd906cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:24:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/666b114d5f56471192cb86bf07b21836 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e97e4919fc4e09ac65f705819b7f11 : SUCCESS in 3m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe3257d9ba5b4b2ead5cfe8ba5016b0a : SUCCESS in 1h 25m 36s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/68ae8a4301a84eef999a3781cf7fb500 : SUCCESS in 2h 10m 10s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2aa8051157e9440f9c20f0e1d7a5542d : SUCCESS in 1h 42m 34s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/7b7bb9622a2d49dcb28b71d6187a0398 : SUCCESS in 2h 02m 36s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/03ccc9c30ef345d398a2d235f75b5eb5 : SUCCESS in 1h 44m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a7b04329524742d5a66806ccd0a5ce6e : FAILURE in 31m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff68f9714298404aac4012f833c7df77 : FAILURE in 54m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e3fb718c395475c8e05ccf526efb094 : SUCCESS in 43m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/2d78aadabbc340cb9b9d33a15f5ad54e : SUCCESS in 52m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/15cfeeb76ade410f9ecce2f763dca58e : SUCCESS in 44m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f943019cf6db4b9caee1f6cd20d82c13 : SUCCESS in 48m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1f61bb42ca08420ab6354449c62b95da : FAILURE in 50m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/e0156c1d8aea4f818f8262b104c3490f : SUCCESS in 41m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/4620886fa18940e5ab1af58029b50e67 : SUCCESS in 44m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/f0fd4d6de65a48579aab5deeb1c00d8e : SUCCESS in 40m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9a6027b3cdeb4a2293419e6915b91889 : SUCCESS in 1h 14m 46s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a2b21ea73d384c7b8567457547622a3b : SUCCESS in 50m 17s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/66ceaf27604a44df94a9f508a81d5f28 : FAILURE in 53m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c0dc0fcf9708d4c58cf92692de18b4abbef080f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"date":"2021-05-05 16:03:51.000000000","message":"Abandoned\n\nduplicate","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5766b521b6eecbb95a5121c95727304aa48b89c1","revisions":{"5766b521b6eecbb95a5121c95727304aa48b89c1":{"kind":"REWORK","_number":1,"created":"2021-05-05 13:12:01.000000000","uploader":{"_account_id":32137,"name":"Arieh Maron","email":"amaron@redhat.com","username":"amaron"},"ref":"refs/changes/29/789829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/29/789829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/789829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/789829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/789829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/29/789829/1"}}},"commit":{"parents":[{"commit":"3ceae99ba91cdf15335f3b7d53e6cc3263939702","subject":"Updating _test_pool_CRUD to enable testing of updates to the load balancer algorithm:","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3ceae99ba91cdf15335f3b7d53e6cc3263939702"}]}],"author":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-05-05 13:10:04.000000000","tz":180},"committer":{"name":"Arieh Maron","email":"amaron@redhat.com","date":"2021-05-05 13:10:04.000000000","tz":180},"subject":"Updating _test_pool_CRUD to enable testing of updates to the load balancer algorithm:","message":"Updating _test_pool_CRUD to enable testing of updates to the\nload balancer algorithm:\n\nChange-Id: I0ac83c5fc9446db700f122b8ee7adaa79c67c95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5766b521b6eecbb95a5121c95727304aa48b89c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5766b521b6eecbb95a5121c95727304aa48b89c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
