)]}'
{"id":"openstack%2Foctavia-tempest-plugin~792081","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I89fa52503738bddb50d2bf31394131f07d2ac435","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"I89fa52503738bddb50d2bf31394131f07d2ac435","subject":"DNM Debugging webserver issue in CI","status":"ABANDONED","created":"2021-05-19 08:05:55.000000000","updated":"2021-06-18 12:04:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c5b1c42d405587fe25a28714cfb60e4ba4639709","_number":792081,"virtual_id_number":792081,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-01 09:54:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 08:06:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-25 14:14:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8ba68ef785e3ff66d73903061a74ebbfb620c675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-19 08:05:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8037cd4bde2255f5d5668277e95d6915f65ef391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 08:06:13.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error\nwas:\n\n  expected a list for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - publish-openstack-docs-pti\n        - tempest-plugin-jobs\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          #- octavia-v2-dsvm-noop-api\n          #- octavia-v2-dsvm-noop-api-stable-victoria\n          #- octavia-v2-dsvm-noop-api-stable-ussuri\n          #- octavia-v2-dsvm-noop-api-stable-train\n          #- octavia-v2-dsvm-noop-api-scoped-tokens\n          #- octavia-v2-dsvm-scenario\n          #- octavia-v2-dsvm-scenario-stable-victoria\n          #- octavia-v2-dsvm-scenario-stable-ussuri\n          #- octavia-v2-dsvm-scenario-stable-train\n          #- octavia-v2-dsvm-tls-barbican\n          #- octavia-v2-dsvm-tls-barbican-stable-victoria\n          #- octavia-v2-dsvm-tls-barbican-stable-ussuri\n          #- octavia-v2-dsvm-tls-barbican-stable-train\n          #- octavia-v2-dsvm-scenario-ipv6-only:\n          #    voting: false\n          #- octavia-v2-dsvm-scenario-centos-8:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario-two-node:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario-stable-victoria:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario-stable-train:\n          #    voting: false\n          #- octavia-v2-dsvm-spare-pool:\n          #    voting: false\n          #- octavia-v2-dsvm-spare-pool-stable-victoria:\n          #    voting: false\n          #- octavia-v2-dsvm-spare-pool-stable-ussuri:\n          #    voting: false\n          #- octavia-v2-dsvm-spare-pool-stable-train:\n          #    voting: false\n          #- octavia-v2-dsvm-cinder-amphora:\n          #    voting: false\n          ## Third party provider jobs\n          #- neutron-ovn-provider-v2-api:\n          #    voting: false\n          #- neutron-ovn-provider-v2-scenario:\n          #    voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-api-stable-victoria\n          - octavia-v2-dsvm-noop-api-stable-ussuri\n          - octavia-v2-dsvm-noop-api-stable-train\n          - octavia-v2-dsvm-noop-api-scoped-tokens\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-scenario-stable-victoria\n          - octavia-v2-dsvm-scenario-stable-ussuri\n          - octavia-v2-dsvm-scenario-stable-train\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-v2-dsvm-tls-barbican-stable-victoria\n          - octavia-v2-dsvm-tls-barbican-stable-ussuri\n          - octavia-v2-dsvm-tls-barbican-stable-train\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":1},{"id":"273cf2624e0ca9fa7f86671166592226980f4140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-19 12:41:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a823640bf0e10a8e7f95051adede226f1320c2dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 12:42:23.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error\nwas:\n\n  expected a list for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - publish-openstack-docs-pti\n        - tempest-plugin-jobs\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          #- octavia-v2-dsvm-noop-api\n          #- octavia-v2-dsvm-noop-api-stable-victoria\n          #- octavia-v2-dsvm-noop-api-stable-ussuri\n          #- octavia-v2-dsvm-noop-api-stable-train\n          #- octavia-v2-dsvm-noop-api-scoped-tokens\n          #- octavia-v2-dsvm-scenario\n          #- octavia-v2-dsvm-scenario-stable-victoria\n          #- octavia-v2-dsvm-scenario-stable-ussuri\n          #- octavia-v2-dsvm-scenario-stable-train\n          #- octavia-v2-dsvm-tls-barbican\n          #- octavia-v2-dsvm-tls-barbican-stable-victoria\n          #- octavia-v2-dsvm-tls-barbican-stable-ussuri\n          #- octavia-v2-dsvm-tls-barbican-stable-train\n          #- octavia-v2-dsvm-scenario-ipv6-only:\n          #    voting: false\n          #- octavia-v2-dsvm-scenario-centos-8:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario-two-node:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario-stable-victoria:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri:\n          #    voting: false\n          #- octavia-v2-act-stdby-dsvm-scenario-stable-train:\n          #    voting: false\n          #- octavia-v2-dsvm-spare-pool:\n          #    voting: false\n          #- octavia-v2-dsvm-spare-pool-stable-victoria:\n          #    voting: false\n          #- octavia-v2-dsvm-spare-pool-stable-ussuri:\n          #    voting: false\n          #- octavia-v2-dsvm-spare-pool-stable-train:\n          #    voting: false\n          #- octavia-v2-dsvm-cinder-amphora:\n          #    voting: false\n          ## Third party provider jobs\n          #- neutron-ovn-provider-v2-api:\n          #    voting: false\n          #- neutron-ovn-provider-v2-scenario:\n          #    voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-api-stable-victoria\n          - octavia-v2-dsvm-noop-api-stable-ussuri\n          - octavia-v2-dsvm-noop-api-stable-train\n          - octavia-v2-dsvm-noop-api-scoped-tokens\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-scenario-stable-victoria\n          - octavia-v2-dsvm-scenario-stable-ussuri\n          - octavia-v2-dsvm-scenario-stable-train\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-v2-dsvm-tls-barbican-stable-victoria\n          - octavia-v2-dsvm-tls-barbican-stable-ussuri\n          - octavia-v2-dsvm-tls-barbican-stable-train\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":2},{"id":"62aeead719601445cc4ae962b71d8fbc5acc6a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-25 14:14:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"10622e546dfb77baffafd79572b9a429b46aad33","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-25 14:15:49.000000000","message":"Patch Set 3:\n\nChanged zuul jobs definition to have a single jobs, which I think was the intention?","accounts_in_message":[],"_revision_number":3},{"id":"9c0ed996ba77d9bb73eee0da0f872350fb2d90a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 15:43:52.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/616a6b7a242848619c8f699643696018 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904b0445629d42f2a478f46418e3c231 : FAILURE in 3m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7481f582a49745caab2d9855fd61014e : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":3},{"id":"edd49f54ed1954cbeee1ed907a57d717af996d99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-25 16:18:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba7053e958891df59dc219383ae1efcee958c91f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 18:52:48.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74a6995500846f284b5e86d62e9d2f2 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f719a622323c4e2ea8d3711fb0bdf7d0 : FAILURE in 3m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c5557693212e4b3abb52fcf8a3c03949 : SUCCESS in 1h 31m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4177c10ea2344b7a81552e6e3fe1d44c : TIMED_OUT in 2h 32m 57s","accounts_in_message":[],"_revision_number":4},{"id":"fe657e4e7f8f460f4612dc2cc4f1fe8aa4e79355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-26 19:16:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"64e0ecf715acf0487fdf6b95c8d2d99ea08691b1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-26 19:18:08.000000000","message":"Patch Set 5:\n\nAttempted to add a server start where everything is logged to a file we can examine.","accounts_in_message":[],"_revision_number":5},{"id":"551cef049e4b7933d856822720067ee6f23541c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 21:51:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411c72d26feb43f2abd8f0bc8a41b6a2 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da66958e9ed74bfc997029658347580d : SUCCESS in 3m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b0f67b0cb7440b98a5d5cab1cb9a5fa : SUCCESS in 1h 33m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ae2a16e333941c78a2f4e982b4a47c9 : TIMED_OUT in 2h 32m 53s","accounts_in_message":[],"_revision_number":5},{"id":"c029fd1738824a62b32f45c3a51e9d972600ddee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-01 08:12:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6d0653df97b951f13f734cba4771ea701af8989f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 09:54:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79c1abbcc544247b3b5adbdf0b86211 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce2d8fd2cc04136a1d965e280028f86 : SUCCESS in 4m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a63a6159af3d457c8573abc07272df12 : SUCCESS in 1h 33m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93d7af901bdd4ec58475a1a4f1115e3d : FAILURE in 1h 31m 22s","accounts_in_message":[],"_revision_number":6},{"id":"c5b1c42d405587fe25a28714cfb60e4ba4639709","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-18 12:04:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d84c38c46d408b46d789d4bfbe80f47a79f3edfb","revisions":{"4214087643b1e6d46d369242d4d52255faf7104b":{"kind":"REWORK","_number":1,"created":"2021-05-19 08:05:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/792081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/81/792081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/1"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 08:03:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 08:05:26.000000000","tz":120},"subject":"DNM Debugging webserver issue in CI","message":"DNM Debugging webserver issue in CI\n\nChange-Id: I89fa52503738bddb50d2bf31394131f07d2ac435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4214087643b1e6d46d369242d4d52255faf7104b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4214087643b1e6d46d369242d4d52255faf7104b"}]},"branch":"refs/heads/master"},"c185f0b9812d4c90117f712defa08b2443738035":{"kind":"REWORK","_number":2,"created":"2021-05-19 12:41:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/792081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/81/792081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/2"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 08:03:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 12:41:51.000000000","tz":120},"subject":"DNM Debugging webserver issue in CI","message":"DNM Debugging webserver issue in CI\n\nChange-Id: I89fa52503738bddb50d2bf31394131f07d2ac435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c185f0b9812d4c90117f712defa08b2443738035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c185f0b9812d4c90117f712defa08b2443738035"}]},"branch":"refs/heads/master"},"7a02d5e752b5ae5290e95d6e6ac12672a907bd44":{"kind":"REWORK","_number":3,"created":"2021-05-25 14:14:45.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/792081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/81/792081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/3"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 08:03:59.000000000","tz":120},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-05-25 14:14:36.000000000","tz":-240},"subject":"DNM Debugging webserver issue in CI","message":"DNM Debugging webserver issue in CI\n\nChange-Id: I89fa52503738bddb50d2bf31394131f07d2ac435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7a02d5e752b5ae5290e95d6e6ac12672a907bd44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7a02d5e752b5ae5290e95d6e6ac12672a907bd44"}]},"branch":"refs/heads/master"},"5f29661dd83bc91a18e0bed6ceb8e75abff6ba05":{"kind":"REWORK","_number":4,"created":"2021-05-25 16:18:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/792081/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/81/792081/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/4"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 08:03:59.000000000","tz":120},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-05-25 16:18:34.000000000","tz":-240},"subject":"DNM Debugging webserver issue in CI","message":"DNM Debugging webserver issue in CI\n\nChange-Id: I89fa52503738bddb50d2bf31394131f07d2ac435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f29661dd83bc91a18e0bed6ceb8e75abff6ba05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5f29661dd83bc91a18e0bed6ceb8e75abff6ba05"}]},"branch":"refs/heads/master"},"e720fc31a24c4dc0f91ac59e94a17dfd2a2e5ed5":{"kind":"REWORK","_number":5,"created":"2021-05-26 19:16:54.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/792081/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/81/792081/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/5"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 08:03:59.000000000","tz":120},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-05-26 19:16:25.000000000","tz":-240},"subject":"DNM Debugging webserver issue in CI","message":"DNM Debugging webserver issue in CI\n\nChange-Id: I89fa52503738bddb50d2bf31394131f07d2ac435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e720fc31a24c4dc0f91ac59e94a17dfd2a2e5ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e720fc31a24c4dc0f91ac59e94a17dfd2a2e5ed5"}]},"branch":"refs/heads/master"},"d84c38c46d408b46d789d4bfbe80f47a79f3edfb":{"kind":"REWORK","_number":6,"created":"2021-06-01 08:12:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/792081/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/81/792081/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/792081/6"}}},"commit":{"parents":[{"commit":"3497f6cb7067eab4e37aaeb376e56ef3fa7cf395","subject":"Fix owner_or_admin API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3497f6cb7067eab4e37aaeb376e56ef3fa7cf395"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 08:03:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-01 08:12:31.000000000","tz":120},"subject":"DNM Debugging webserver issue in CI","message":"DNM Debugging webserver issue in CI\n\nChange-Id: I89fa52503738bddb50d2bf31394131f07d2ac435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d84c38c46d408b46d789d4bfbe80f47a79f3edfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d84c38c46d408b46d789d4bfbe80f47a79f3edfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
