)]}'
{"id":"openstack%2Foctavia-tempest-plugin~807127","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ie2fb0b5b2e793ec84a68c81d1573294409c4ea59","project":"openstack/octavia-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ie2fb0b5b2e793ec84a68c81d1573294409c4ea59","subject":"WIP DNM testing act-stdby jobs","status":"ABANDONED","created":"2021-09-02 12:26:24.000000000","updated":"2022-04-15 08:01:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"833bf28c832adc231f30bb0877f0169aaecac496","_number":807127,"virtual_id_number":807127,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-02 13:06:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 13:06:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9208b891eada83b9e87d5a8dc3050832a32aab73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-02 12:26:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03cd39553996f62eadeb4df8270cb51a3a92cae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 13:06:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f393bbfa23fe418e9267e88edfa40525 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5fcdabafcee43efa18175c52f5827c2 : SUCCESS in 4m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13e86888020e44e3bfb573920964a2ae : SUCCESS in 36m 21s\n- octavia-v2-act-stdby-dsvm-scenario2 https://zuul.opendev.org/t/openstack/build/c676bdccbc9f49f7875ed33fc4b4ff35 : SUCCESS in 35m 32s\n- octavia-v2-act-stdby-dsvm-scenario3 https://zuul.opendev.org/t/openstack/build/8422b254baed44e0b3b4a20028cf9846 : FAILURE in 30m 59s\n- octavia-v2-act-stdby-dsvm-scenario4 https://zuul.opendev.org/t/openstack/build/c9374ce731074f688160af34b43214d4 : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":1},{"id":"833bf28c832adc231f30bb0877f0169aaecac496","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-15 08:01:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae5798de670795d9f99bc7a2003781c98e5125f4","revisions":{"ae5798de670795d9f99bc7a2003781c98e5125f4":{"kind":"REWORK","_number":1,"created":"2021-09-02 12:26:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/27/807127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/27/807127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/27/807127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/27/807127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/27/807127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/27/807127/1"}}},"commit":{"parents":[{"commit":"605899c1a92756b568ed14478158ddafbcff3d18","subject":"Increase job timeout for noop-api tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/605899c1a92756b568ed14478158ddafbcff3d18"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 10:25:49.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 12:26:03.000000000","tz":120},"subject":"WIP DNM testing act-stdby jobs","message":"WIP DNM testing act-stdby jobs\n\nChange-Id: Ie2fb0b5b2e793ec84a68c81d1573294409c4ea59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ae5798de670795d9f99bc7a2003781c98e5125f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ae5798de670795d9f99bc7a2003781c98e5125f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
