)]}'
{"id":"openstack%2Foctavia-tempest-plugin~824999","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I43bff02cf318a30f20c74fbabe09abc005f36661","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"tw-experiments","attention_set":{"28609":{"account":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"last_update":"2022-11-29 13:48:48.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-12 12:30:22.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-02-11 13:52:58.000000000","reason":"Tom Weininger replied on the change"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-01-29 17:29:24.000000000","reason":"A robot voted negatively on a label"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-02-11 13:52:58.000000000","reason":"Tom Weininger replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I43bff02cf318a30f20c74fbabe09abc005f36661","subject":"Add tests for creating a fully populated LB","status":"NEW","created":"2022-01-17 17:28:35.000000000","updated":"2024-01-29 17:29:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":76,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"cc92a16e3ae0f0b2a6af2c380ebfb34b4a5a84f3","_number":824999,"virtual_id_number":824999,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-20 16:44:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 20:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-20 15:02:06.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-01-20 15:02:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-01-24 09:43:00.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-01-26 10:32:00.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-11-29 13:21:49.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"CC"},{"updated":"2022-12-02 16:46:57.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"708cd8a157756abffc1da0785c4520ca55b53a67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-17 17:28:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f82d2ff71cb74b342ed1d87c83cce14c1ef1c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 20:06:14.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d866d9941ab34b0db20e59f2a6b9f291 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f34665b55324c108df76bb4a54a8b7c : FAILURE in 3m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a14804748614e76ae374ef0685558db : FAILURE in 2h 04m 35s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/4de007d471ea4a9e8fb5545b6dcd3a15 : FAILURE in 1h 43m 49s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/a70e5e905ee54f3aad7214c84befff7e : FAILURE in 1h 35m 23s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/9278d0c4f6c34df1a87e345ec078d425 : FAILURE in 1h 36m 46s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/7cd1b30ecb74432fb7f63325914ba19a : FAILURE in 1h 38m 25s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/bd3b1023cffc44ad8dfb53a73185b3c8 : FAILURE in 2h 30m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec7d95d8956f45d680b2a387c44a27a0 : SUCCESS in 1h 54m 32s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/ec0cc9bfd7984ba882a9a7c10d353156 : SUCCESS in 1h 14m 03s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/30431fa726df40b3a9b5c8325a707330 : SUCCESS in 1h 37m 09s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/52c06ccffac1454f8252ef83a9ca0878 : SUCCESS in 1h 52m 10s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/f2297c594f154566b28efe1db518e146 : SUCCESS in 1h 31m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b7385a075dba42589af9dcb5bcc335ba : SUCCESS in 42m 16s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/892aa84d538c4e839669fda42906e287 : SUCCESS in 34m 06s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/2794ff31326a4e4ea1e2ec82745563e0 : SUCCESS in 45m 03s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/cc97b0fc70084f5f80f03d688752a012 : SUCCESS in 47m 04s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/acc2092566894103bf28dc8152070b08 : SUCCESS in 52m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/06809c3eda6d4b7199c91f1dbd56c22c : SUCCESS in 1h 50m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/d83add2587464aff98e273f52a5be024 : SUCCESS in 1h 56m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a75b36e6a2fe4d7e9af0d670df2808ee : FAILURE in 2h 25m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ccebacb5e22d4ccab1d315353f8315a0 : SUCCESS in 42m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/5834046e322d4b1ba06f4fbd7da1e9e1 : SUCCESS in 31m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/2220864d9b15463b979c50f61c4d950c : SUCCESS in 38m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a18c7a942b67465c9ad588d78ef66e7f : SUCCESS in 33m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2c76c18b3361486382c17f7062c99c9e : SUCCESS in 44m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3434622ab6484137a13efbd1e352c104 : FAILURE in 45m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/818be91c1bf045e1951e15969dc814ed : SUCCESS in 34m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/c789db02f068496386e217432466ae1a : SUCCESS in 31m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/31a07d0a7fa24f68b75163dcca723d7c : SUCCESS in 35m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/2dcfee6339c449348dd53a3b7378888c : SUCCESS in 36m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61425ad5b20140bda274037ce10e7e1d : SUCCESS in 1h 54m 51s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/fa766246d9634ba98e08b5b676651239 : FAILURE in 47m 34s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/0a689a4f39c14df58754a687ca8e51e9 : FAILURE in 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d98dbca5781acad77531e0fe862250333fba2beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-18 15:06:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16f161a762639cafceabe99db9df5a30aa6eb4d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-18 15:08:15.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3fbe6bec060d07d4097e6e9badbacfe5522f9f01","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-18 15:15:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"ea8faa185aada973c8bcfaa85cb187735a46236b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-18 15:48:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b29b6f6c4ce9b621107593a52fc717a54bdb1e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 18:19:00.000000000","message":"Patch Set 4: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc4449602674fe7ac406f63f014cf4a : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2de28ed6e5e477893235adb3c0ca913 : FAILURE in 4m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08c4d91ac8e745568536c427c30f97cf : SUCCESS in 2h 06m 53s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/cd127a3b78c24884800d5238444ef303 : SUCCESS in 1h 31m 27s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/5916eb12faa14a48b8be9cea9f2439a6 : SUCCESS in 1h 28m 46s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/17dfe87b079f48d4a8444cc58c9159d0 : SUCCESS in 1h 13m 05s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d0c3f7ddaaf04664ab2abde284508fbb : SUCCESS in 1h 20m 16s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/668406a5123248e6a24176ece0ae9325 : SUCCESS in 2h 25m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/271009e29b9d4d7daec5c4aa74805068 : FAILURE in 1h 50m 54s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/cd5e00394bf64b719e327a0c28d3a561 : SUCCESS in 1h 36m 36s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/58594b3340914201812da2b52a7821a6 : SUCCESS in 1h 26m 19s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0183b210b4ac4dd089351ed04adabbb2 : SUCCESS in 1h 32m 42s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/a73c89933418490f89762aeb0b2ac6ac : SUCCESS in 1h 38m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/07f61bf74fc7465aa18d3cfd755a8717 : SUCCESS in 39m 49s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/7475fd76aed84127bbc8a2837cb679d5 : SUCCESS in 37m 31s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/86c88ea4bf064f1c8d6d52dde98b3a7a : SUCCESS in 34m 13s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/751d37abf3da482391f77ccb41d66b7d : SUCCESS in 36m 18s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/0acab1ccd7244f4ca8a2b85accaf85e2 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c054d4028bf84441b3e8a5769b35d6e1 : SUCCESS in 1h 33m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c750979a16e4021a86c9e6e751ee23e : SUCCESS in 2h 05m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c8f0e50abb5d457cb503a36b336beacb : SUCCESS in 2h 04m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b29e7b3cf1ea41dd9d78ac04af4783e1 : SUCCESS in 36m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/53f0236705734489b945f53ffcb34f12 : SUCCESS in 33m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/4967318970b94d9e9fdb6399f71d4bfe : SUCCESS in 34m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3892d21c89ee4645b8faea73691981e0 : SUCCESS in 33m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/595a0aab05024a01b0bf133deeb71224 : SUCCESS in 46m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c8a83a63bf344929b940b86477cd5f90 : FAILURE in 49m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/c13f54aa064d4a4cbe4694ddb5037b27 : SUCCESS in 36m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/a973de4adeb84cd898eb7f8712134a8f : SUCCESS in 47m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/b8f5da2dc4834e849f3c1d1668212284 : SUCCESS in 37m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/d2262c2705cc4da9adcbd584bd3307e2 : SUCCESS in 46m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b9bdd68f16024f7980c3f922ecaf0a9c : SUCCESS in 1h 39m 25s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/524b5c1ee2e642a6a24200ff068e2740 : SUCCESS in 47m 13s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/129005794d444a4ea15156acb725d2a8 : SUCCESS in 1h 07m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eb645dd3252ebe19b53fd40b22970b8fcf689c60","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-19 09:20:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ca5b6fc70ad04a7e92c774ef9bfa676e0aafe949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 11:41:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897c99bbeccb4c718be82243ec66f5f9 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b14e48dbfc483da47d3fb8ffe7c5a7 : SUCCESS in 3m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7157873b5494c1380f9fddc180bb9a7 : SUCCESS in 2h 17m 42s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/4a35dd34df5248a889ce5a7021d7f674 : SUCCESS in 1h 34m 34s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/6f3173d46be244aa9fb5adf68beb9292 : SUCCESS in 1h 40m 02s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8c4dd837b97b43a9a01c617c1f375642 : SUCCESS in 1h 38m 15s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/4b501cb4bef64391bed2a70b73097ec9 : SUCCESS in 1h 09m 17s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/8fed0ddebb1048d88748bf57e1f2d463 : SUCCESS in 2h 16m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/111e7dade0864a0fb3395f2bed4422aa : SUCCESS in 1h 26m 46s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/c540f9c0b4e94a0489f00e519fc4fc77 : SUCCESS in 1h 25m 49s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5f3524fa492d43aaa6c2ff394c2c8d3a : SUCCESS in 1h 26m 20s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3603b157cdf94a6794689fdff0d3a666 : SUCCESS in 1h 31m 22s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b0621c6f0d2c44b38171da20cadc7fd7 : SUCCESS in 1h 46m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f923ddf3a72c4e2aabd90ec8c94d08eb : SUCCESS in 34m 06s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/9dd8d39bd86f4ee5abfb580c56d5647b : SUCCESS in 35m 47s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/811c7d4e729647f285c90dda67a37afb : SUCCESS in 38m 51s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/d90235ebd12045b6915b5b32ac89146b : SUCCESS in 39m 32s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/40822b86da944ac4bb107b6182907735 : SUCCESS in 47m 09s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/507d8630e9964e40a8f63aa084919197 : SUCCESS in 1h 55m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/050e4ce958b14ef9b873f15d53ebc2d0 : SUCCESS in 2h 12m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab54ea11c724460596a3f753737495b7 : SUCCESS in 2h 02m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f4e73b97f3a4e46b0397cf47093f9c5 : SUCCESS in 34m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/9410bf2233f14c4f9ae651c7c2a3a8cc : SUCCESS in 33m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/6fe78b4391d34da2aaa64be1d94ec9be : SUCCESS in 34m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5a91e00bf92342da9d8a9c27f96b9666 : SUCCESS in 34m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5ddc51d764b648bdb3edc5d15e0aad46 : SUCCESS in 52m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/953dab3600214396a2404db5df9570dc : FAILURE in 46m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/57007df2d0874e69bd58fa8a39b9b35e : SUCCESS in 45m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/c5b18429be7d49deb67a29cba9897173 : SUCCESS in 31m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/32d6cf74b51e4bf59e01fe9f7b31b6d2 : SUCCESS in 51m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/3b71e341b6734a768c5c6687fb129f26 : SUCCESS in 35m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/20aacf53470e40fc8578dbc73dc9df28 : FAILURE in 2h 05m 39s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/086e9794d8204d738bc3521b7134b6b8 : SUCCESS in 55m 09s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/886f27de022446dc9c7e095085b96157 : SUCCESS in 1h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd6306d37ff48e92ac53abb8c95756fbcb12f92f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-19 13:52:41.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"0e6acc1ed50f26688022652c0d68dc30e6ffcde2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-19 13:56:34.000000000","message":"Topic tw-experiments removed","accounts_in_message":[],"_revision_number":5},{"id":"689a423deafe55eeb3ea7667a3b5dfdc555a744a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-19 14:05:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"4f464b8e1dd4757879ba9beb4c3dcb013a2a48f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-20 10:05:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2f15b4aa0788a1aa175ba503db144b655e4871ae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-20 10:55:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8641b3431a5368830973dc641a472417982a51d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-20 11:12:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2c233eb3b7c9bc1c47b9c5832e55820f10633754","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-20 11:39:05.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bc00813e5597347d503b531c0befb7e5d2ae598b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-20 14:03:55.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"f3f13ee713f6f0da9bc1a36658d784d3bec88fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 14:07:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816229570f094341ad2772a74998bebc : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc7e228fb71943b58586a8b8a8d4523a : SUCCESS in 3m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1baa591ca4a0458bb294f6cd2abe4ea8 : SUCCESS in 1h 46m 32s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/b288ed6e160d42e2a0c5204560e449da : SUCCESS in 1h 32m 18s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/d8a4dd5d7dbd47138c94d20a11ed4259 : SUCCESS in 1h 26m 17s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/55ab09ff67cd46f1b39e5242e523e94a : SUCCESS in 1h 13m 49s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/c67d84bb9ea147459b71dc4d725c8979 : SUCCESS in 1h 46m 12s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/47fe324066984d3982f519d9cd408c90 : SUCCESS in 2h 03m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc3f42785a834f0c91b55baef4d41deb : SUCCESS in 1h 46m 46s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/a9ddce13293846d8bc6185fa9c2c9198 : SUCCESS in 2h 01m 50s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/7ecc117fc9e447c7965284f94b4f4879 : SUCCESS in 1h 50m 00s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b9d3a2f24d6d43ed9551ba8c5a9e7763 : SUCCESS in 2h 13m 20s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/7ebef00ff5cd440e896ccd8f4db30774 : SUCCESS in 1h 49m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68a4ddbcc4424983abdace2b755c3e54 : SUCCESS in 36m 14s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/0da2517ac6ed40df91e7ea336b3f314a : FAILURE in 28m 29s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/6ada400c178a4a498ac5e835c79043bd : SUCCESS in 51m 06s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/dc61c8551fdc402ba120e85bae8faecf : SUCCESS in 53m 35s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/51b82d59d9ad4a07a2ec1f298abfc2c0 : SUCCESS in 53m 40s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d1f6fe3c942a4199b27932c18c41d064 : FAILURE in 17m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1696dfb92b745f59e132200a94c2f87 : RETRY_LIMIT in 5m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b2ee844baa1e424b88a6773b499ce615 : SUCCESS in 2h 24m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ea7391e57aa49e990f76e8eeef10515 : SUCCESS in 46m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/4b5b8b6de0d043c584c365873e740286 : SUCCESS in 44m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/e7d8609cdee043c180e9abf0c3802f5e : SUCCESS in 45m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ebdaf922c69a43e6a22728e038c797b6 : SUCCESS in 49m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8a971fdb17364d9d97bbbc0965a11a3a : SUCCESS in 50m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d5f103bf602e401791717f4fd1b9bf9c : FAILURE in 44m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/8a73abd0882a4f44898f0ed63f990673 : SUCCESS in 34m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/1182a8d6135f4335bf2c783fa80e2885 : SUCCESS in 31m 47s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/5eed294e48fe42c998c915a32a78fbed : SUCCESS in 34m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/e2de886c287b4509a14cfc04b3a02489 : SUCCESS in 52m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7a98ac85de9f40b5ab9af8c384faf26a : SUCCESS in 1h 45m 43s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d828cbb81cda454381815f937bf269fa : SUCCESS in 41m 55s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/fc82f5e31a264de8aae23483c12a87ce : SUCCESS in 50m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"120ac261cbd102216b437d74e9b26514e7f669e4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-20 15:02:06.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4383f541133637f84fbb36f2dc9c1fbdb324ad15","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-20 15:02:10.000000000","message":"Patch Set 9: Code-Review-1","accounts_in_message":[],"_revision_number":9},{"id":"fca503107f1554fc2f5d9033c50cf63ea1f0d680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-21 15:23:17.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a6031cb18af05063471a5128a93d85dd84be8b8f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-21 15:26:55.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7f6caf734ea425235ae25dfd37ee5e7e6dc2edcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 18:34:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e228761b562477eb91af2d308159615 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479f266261364392a9d01bd20d2e4d97 : SUCCESS in 3m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b98d965cf02f4f7f92fb26fc27ce1145 : TIMED_OUT in 3h 04m 25s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/c2b0b19a09984350b7a03841cc05b8db : SUCCESS in 1h 31m 58s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/13679f0f810c4594b3142ed36f83fc25 : SUCCESS in 1h 40m 55s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/0d0fade036824888a9bf689c2e129ca3 : SUCCESS in 1h 18m 26s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/5e8c4b392c1d4ee4a3eadde8cbf7126f : SUCCESS in 1h 42m 47s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/9a70d9906b884298a04d8922807f9b3b : SUCCESS in 2h 58m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd652c8858ed49fcb047477954a69ee8 : SUCCESS in 1h 13m 14s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/45db9a9abb4940778a98015760b31be7 : SUCCESS in 1h 39m 07s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/3f543132bc4f441788842a7e29956cf1 : SUCCESS in 2h 28m 08s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/e4804251fffa4be2903b142ed3ee3a1b : SUCCESS in 1h 42m 22s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/d5667005b3e842c687ac56f42574794e : SUCCESS in 2h 00m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/faff164eea864f9694d95e7bc8ef3548 : SUCCESS in 36m 30s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/9cc015a8b7cf496d87836bcf0b0532da : SUCCESS in 52m 06s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/ea20f691c5e346308ee5b296ffbb3a28 : SUCCESS in 32m 12s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/8d666f433f5b45ad9a458d5a1754387b : SUCCESS in 39m 04s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/91272d935e5a4eaba683573be4f7f1f6 : SUCCESS in 45m 01s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5b482282f2ff44dea6335459be60f00d : SUCCESS in 2h 12m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a0a3b560979470a81736d4b25d9429d : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05ffd9b6561b49778e946d1f68fcaaa1 : SUCCESS in 1h 54m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25f0a077fe3243feb19b013f861f2bf8 : SUCCESS in 32m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/73c76e6a2d9545a9bda74776a1287f8a : SUCCESS in 33m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/a692a800847d40f0a67a6c8a1d0311a7 : SUCCESS in 33m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/63b0b1e9e7a34fadacbbc4b10d4f8278 : SUCCESS in 33m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/4139f4ff916b4186ab6181548bdcf0ce : SUCCESS in 41m 42s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9a196cd7690043f0a24ac3251ad5e4de : FAILURE in 42m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/5195ffa2a3f54018866ac09fd2456d7f : SUCCESS in 35m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/8287c995dce64e4ead497a982683b6c8 : SUCCESS in 31m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/daabbbd491ca496cb6fafeeefafe80f7 : SUCCESS in 35m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/2aa0722b4b4b4329be2ce79b5d572004 : SUCCESS in 38m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5160ff26db4c407a90b4109825d16e62 : SUCCESS in 1h 18m 40s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/35156a5e45e54f16a22a175aabd3b08f : FAILURE in 47m 14s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/9a3aa25700024c9195cf716733022c9a : POST_FAILURE in 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"29fcbfed7c2c706ff22575a25c8ac537884c60c3","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-01-24 09:43:00.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"08e198c7026646d0d3b84e0399155086242197e1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-24 09:46:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"95a788abde3e0648671a0958b611386fa98f022a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-24 09:54:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f188dde877545952d87b622e212abefcf0141b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 12:28:04.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9622b7a974a549e5bd5b8d3cc67b534d : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/483f098e30ba424ba4c82d654bc0e26a : SUCCESS in 3m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb0edb977bea4b6997a403bd11a979d0 : SUCCESS in 2h 07m 01s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/5781a40a839d47debd646eafb987f8e7 : SUCCESS in 1h 31m 04s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/92187f035b2544068b4336939a31a6d6 : SUCCESS in 1h 30m 11s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/81a74d0cfde143bca1a615b843c0cdde : SUCCESS in 1h 49m 04s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/1d94dd56d41d4921a07cc87b9266c846 : SUCCESS in 1h 45m 20s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/198d63939a7a4fdda974ad1a1f7ebe82 : SUCCESS in 2h 17m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12f30e578285425e83bd369dc63288e4 : SUCCESS in 2h 04m 18s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/1d035c92181d4e99a45518184788b665 : SUCCESS in 1h 44m 50s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/2648513ffe9142168da1ecf3485500a1 : SUCCESS in 1h 27m 39s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c40743bb364541fb953a1ad72e7781d7 : SUCCESS in 1h 41m 50s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/2a30f85db9484e6d8fb7153d11bb5337 : SUCCESS in 2h 04m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d31aa1babfe348b7881ed75e0efa8129 : SUCCESS in 33m 53s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/7406788e790d443fae67f1b25fb7106f : SUCCESS in 33m 30s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/635914b30ead49ef96c5e4e77aa3a90a : SUCCESS in 32m 44s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/9cede2bfae9e4172b79928784f36e411 : SUCCESS in 34m 32s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/b87136771adc414f94b5c105440615dd : SUCCESS in 37m 47s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d4d71f9d5fd470da9bce7bd5617e1de : FAILURE in 1h 50m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec06825bd125428dbaf0d06e069dc15f : SUCCESS in 2h 08m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce82155c062848f89b2c446f1f38b019 : SUCCESS in 2h 07m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3716913cbd7d42b9ab6634b0dbe82f26 : SUCCESS in 37m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/2e75650ce6074849be0afaef2f6b4ae0 : SUCCESS in 32m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/4227509c31ab4bc7aa6ba29501e7f61e : SUCCESS in 37m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2698266e9d0c4aa2bc7d026e4ebbd569 : SUCCESS in 32m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/8e1b056cd1e24dda9282b39f00ed3d98 : SUCCESS in 53m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f8c95454d9764735ba351a9bb027c36d : FAILURE in 42m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/41c5d9a040bf4fe1bf0b1db2f8d03dd5 : SUCCESS in 31m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/1ba85c7d565541eabaa023c35976f1d5 : SUCCESS in 29m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/532f6d3df567441a92c1c0554e8ee69b : SUCCESS in 31m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/19a0fd486b7a47d188b1c51d8ff7b822 : SUCCESS in 35m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/888c884017be43f2b56f90827c12d52c : SUCCESS in 1h 20m 26s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/65105e8ae6dc4b25ac5beef7d712eaf3 : FAILURE in 48m 54s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/2f67a94eacbd4a2c8f0fa1154ab89031 : FAILURE in 48m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"07907f76ac13aaa028b6ef5ce6e8f6d7cbf6b610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-24 15:12:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6f3ffdfc338bfae8e8f7189b2a3fc2730401c331","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-24 15:18:07.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"19d38a1ef28b89b7d3b404e2e8103c90336b0190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 17:47:03.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77caf64082164c8a917c33cef93d42c3 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09034c4d040a4476969bd032f4d4ad26 : SUCCESS in 3m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/907975cdc7f040f1be8e68e9cc2c0d2e : SUCCESS in 2h 00m 26s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/49755c08c6e741f3acb12b3833843ba5 : SUCCESS in 1h 19m 17s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/1bc9c72f4d8243a7a0061e1d4f63af29 : SUCCESS in 1h 31m 26s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/35495fd3777f4b41b3a43bb17182d178 : SUCCESS in 1h 05m 19s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/57a89fb3026e4adf8d67008c204efa96 : SUCCESS in 1h 06m 54s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/98de38e8753c4e5099fa860843866ade : FAILURE in 2h 09m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6eecf2b615ae4c05bd70f550be713fcc : SUCCESS in 1h 51m 23s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/34f88b64d3754d9cab74ad82315d393f : SUCCESS in 1h 42m 03s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/946afca593b844658f0d2c986f918f51 : SUCCESS in 1h 35m 35s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2e1649c733b149d5835b5dfd612f9212 : SUCCESS in 1h 29m 10s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1ff8bcb9df434695995566ed8d939cc3 : SUCCESS in 1h 52m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85448048efaf44bbbe012bf79d1e5760 : SUCCESS in 36m 01s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/252b1b5bbd4f4863bc9605d50e4e3c97 : SUCCESS in 38m 00s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/83f556b09db44417a29297a77df5223c : SUCCESS in 33m 24s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/fa68d527b0af4edc97d0e9c0642c38fe : SUCCESS in 41m 44s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/f8d6f2ac56bc48908945b2f6b2b08ff2 : SUCCESS in 39m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d06bc790448543379dc830636544301d : SUCCESS in 1h 42m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c30c01e560d46939ccd0c1f9805781b : SUCCESS in 2h 13m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/661661de30c5455e80c4d46e61edd305 : SUCCESS in 2h 29m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59dbceb5b18b48dc9ff94dd4c00948f5 : SUCCESS in 34m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/35fbc760bb6d4b56b2aecc4cd2bd2004 : SUCCESS in 31m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/b6608b022f0d4e42b1a0ebb8b7adfc4c : SUCCESS in 35m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/bff0efe969004a9bbc3da9d0e2f0b55f : SUCCESS in 35m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/b4122a68ab594ba288416543678a1d60 : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f59e879fe7764e009d5c56056f46b885 : FAILURE in 51m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/dc84efba20d84c3a9ba8748a94bf68d7 : SUCCESS in 31m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/9ddfc52a7b7447f7877c66dc4f117d87 : SUCCESS in 33m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/808523dfce8f487f9efebfd0a37f58fe : SUCCESS in 31m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b3f2211bea5d463492331a95521036cf : SUCCESS in 49m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/60b28f0d681148128ae126a4f4739694 : SUCCESS in 2h 00m 16s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/34080d5db50545709ea7520bb5128452 : FAILURE in 56m 33s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/9d304f204b5948c88ee8efa30287b13a : FAILURE in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"24e30412c56e547cf25cef31b0736f77b62ac404","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-25 07:43:06.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"409d75630216f8cae81d41ca23f7c616a78d0fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:22:59.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f160035e334cf2ba4a52652f4654c0 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0a7ac874da74e42afc80e0c163d9109 : SUCCESS in 3m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4c5f48781224a609513ae9e5db79aa7 : SUCCESS in 2h 13m 55s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/6f1bcbb582f942b2b1895b70adb0ad8b : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/5c8beb1f52d24731ac764d444a7d66dd : SUCCESS in 1h 31m 58s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/f26ee5210f544eeabb77403e1801a338 : SUCCESS in 1h 38m 03s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/78bcaad1e05c49908e01882e78db1b71 : SUCCESS in 1h 05m 06s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/322a2bdff0ef47778292127a328ae768 : SUCCESS in 2h 27m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2aa9cdd51d3b42e8a43d6776609dd56e : SUCCESS in 2h 03m 42s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/35e1dd8d5df949399e52102b5c9ceaca : SUCCESS in 1h 19m 27s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/d4ef7559b1a140ca8e68e32c0681053a : SUCCESS in 1h 28m 33s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c050d3e9025c461da1f84832bea1a6fa : SUCCESS in 1h 59m 18s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/5824e89ce4964aba81f393a87e4cf6f0 : SUCCESS in 1h 29m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0849d201c7b41cbbfcb5c476d9115c1 : SUCCESS in 44m 09s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/252b8fa7a64541539c037e9edc8989aa : SUCCESS in 34m 19s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/8864e3da04b1462fbc4ef5a6f103f27b : SUCCESS in 37m 19s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/020d025875004c6ab7f223df0011b26d : SUCCESS in 50m 05s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/ce27e604c9eb4a23b790de9f71d70a35 : SUCCESS in 55m 31s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/145eb3a92790429692e87593ea2ed8d3 : SUCCESS in 2h 07m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d57680ae36245f78495562f6624fa5a : SUCCESS in 2h 04m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b528f10464a942efb8cb8d5e2b5ac715 : SUCCESS in 2h 07m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c2897e429424092bbbe94db43fb950d : SUCCESS in 42m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/dbfa97fcb8f4468fa36f8935cba4c662 : SUCCESS in 36m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/d7f8f884fca84df3a5823c6e6de41215 : SUCCESS in 31m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/355387d6c6e945cd9e86726bf5b7a3cb : SUCCESS in 48m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/cc535325c4824274b0b8ceb453015077 : SUCCESS in 37m 00s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eb4f06e3c50c427f88468989f246b497 : FAILURE in 53m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/0b9c2b5809bc436a9523c0e7f9dfe08d : SUCCESS in 35m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/24f761c8f7554a568d438274d650d297 : SUCCESS in 42m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/b140e1819b824684bf6f80d8633e6e02 : SUCCESS in 31m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/765b83280553480cb1f73c7ef54e5eab : SUCCESS in 48m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/85bd7f945f76459c9e8bec0829e99b0f : SUCCESS in 1h 48m 43s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/8397a93dde3747d5b53291b6d6b40883 : FAILURE in 54m 26s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b9db22f0749341ba84c542f8ec758059 : FAILURE in 57m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"30e67ad0f4c096c525e57fc2ba7cbf1bc3a31af9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-28 23:55:47.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7615e3977e18a6d0410dff29a6dbdb5247c25bac","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-29 00:52:21.000000000","message":"Patch Set 12: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6e220bae69f3c9834b8ffa667d20068ffa487003","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-31 10:09:42.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"80b6b3d5a06b8e125afe17cdcecbf83ee0207ed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-31 10:55:08.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7bc6adfc3338942f51d9d6c7be783fa4201352ed","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-31 11:00:25.000000000","message":"Patch Set 13:\n\n(10 comments)","accounts_in_message":[],"_revision_number":13},{"id":"100344cffec7416fbf8b2e7b074f3d841a0c068a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 14:05:33.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681d71615d11430b980f42f1e5a9263d : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496b55337fa7498187e71d91e4ef8e3f : SUCCESS in 3m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6131815ff77b40a686828f19f8af4385 : TIMED_OUT in 3h 04m 22s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/f59227fffcde4b3cb303db59b7aae4c4 : SUCCESS in 1h 43m 21s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/cb23a738c23e4e72954930513221fb55 : SUCCESS in 1h 31m 12s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/0320cf0a21ae417592b28f0e38847b24 : FAILURE in 12m 15s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/d163a257f2fd4e41bba218593a2ae43a : FAILURE in 10m 46s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/1104bfdd525247248d0284f20b556860 : SUCCESS in 2h 47m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/252307fad70b40e3928df3335888ab2d : FAILURE in 1h 40m 58s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/be2eac14141845548a1c1890cda82e57 : SUCCESS in 1h 47m 52s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/df92161c7d9c45fabefea60beb3d46f4 : SUCCESS in 1h 55m 53s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/31ab180befc14ff7ab9cd3bf3d02f33f : FAILURE in 11m 56s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/0ef885115cde4991b2c8e6a09478ca45 : FAILURE in 11m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cdf1a2fad5174904b07783e9aefc10da : SUCCESS in 41m 13s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/2d745cd6b6574759a942ecc6cec06e7f : SUCCESS in 38m 42s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/c76471d527d44de28614df9e40db1a9e : SUCCESS in 42m 56s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/455018bc579e4555855729505570ad83 : FAILURE in 9m 32s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/445f0f9e34374c3cba227430d03f2b88 : FAILURE in 10m 30s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/64a0a44b300242e4b2791dee06a2f3f7 : SUCCESS in 1h 35m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0080cb0c59844e9b3f153dc0f86eb52 : FAILURE in 11m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/01bcc317297a4e8193549ff16949da16 : SUCCESS in 2h 11m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/317b15a248dc4ccfbca9e48fdd233410 : SUCCESS in 41m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/75bb7e877cdc4bccb7ee1c01f371d555 : SUCCESS in 33m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/60ad424580bb4a7d9201c75ddb183c53 : SUCCESS in 31m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/037243f1d5ad4c009de701356238c443 : FAILURE in 9m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3ad2ddf24aac4faebab8d3746c7d89a6 : FAILURE in 9m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/73c77dca34fb42fc83db00088dfbe61f : FAILURE in 45m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/4cd702a10de74eb5af2de26864720f8e : SUCCESS in 31m 47s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/c75659851c6041829604deafa4369743 : SUCCESS in 31m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/c5fd1fe7fa784f19a12248d00b9b162f : FAILURE in 11m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/3bc19384353c458ebdc8bd00a27b02ed : FAILURE in 9m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/418ab559006f48959a7439174a24fcee : SUCCESS in 1h 39m 25s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6f398833d06a48c5b0a0ed73254084d7 : FAILURE in 55m 51s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5364701ded0042388c1e325aff58f44a : FAILURE in 1h 07m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cc9725a590a797cfdc20054b83d653e9dfba560a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-02 15:40:29.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d59ef2c43a89169b8e1fec5b10f64e053caa7be7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-03 15:01:20.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"e3017060d78ec3e7e9024f1908eec882d0eec377","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-03 15:01:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":14},{"id":"8fdbef82f74a62ca5156590d29c31d1c0d1669bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:07:19.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0d4b4deda544b17931d7a06b9ea6b4a : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba868c0799f4694a13bef02cfc7c422 : SUCCESS in 4m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a96a01d906da4c0ca91c53bc19020bc5 : SUCCESS in 2h 11m 24s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/919f201911ff4ff1998911062efe5606 : SUCCESS in 2h 27m 58s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/1b25ef558d15443c9e8be5b576afb974 : SUCCESS in 1h 38m 04s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/e3c59e278ae342ed89fd1289eb524ae1 : SUCCESS in 1h 05m 57s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/b2580b45ea9446aa8942cf8c6265e96d : SUCCESS in 1h 03m 39s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/b8717aaa80bc404cacdd6bbe2d6cefd6 : SUCCESS in 2h 35m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbd1271b88294c468655c3a0f5de4f75 : SUCCESS in 1h 20m 27s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/017caf654a9c4a8eaeb0c7e73823d73e : SUCCESS in 1h 11m 39s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/22a6004dd2c94b88adcc28a6b008d0d3 : SUCCESS in 1h 20m 09s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/7eb4884c03674dcbb1e79b8c015f2ce5 : SUCCESS in 1h 26m 36s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/6340f3171b3e4aefb770b4be143ea9b9 : FAILURE in 53m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6d16da6cc71a44bebd325c5eb45219d6 : SUCCESS in 33m 48s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/23d34fafb0864d42af468c86285fc724 : SUCCESS in 40m 55s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/8bf9f8e4ae294c4693013e302f0e2eeb : SUCCESS in 35m 44s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/0360abcccb9243d99683023083d77df9 : SUCCESS in 52m 51s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/1b361e06c3594ce9848f6cf6407929af : FAILURE in 30m 58s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0fa99f754393431095463f805b3ffbbf : FAILURE in 2h 18m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb0388fa87e04e37980b8a9af8ccf764 : FAILURE in 29m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f8d6a27acab4a998983f0d48060f10f : FAILURE in 2h 58m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2e02f589d644e26bed2a3cc49812a76 : SUCCESS in 32m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/bd435aa879f64355a336120d6f0ce55f : SUCCESS in 40m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/3a48b7eb1d0a45aa8bde84cf418bd65d : SUCCESS in 34m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/94848abfa9fd4637bdc04a7dc777e42f : SUCCESS in 42m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/59f0d0be12164560bea197a00ad946e4 : FAILURE in 30m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e6f29e626ce244c7bed416582ecba8e2 : FAILURE in 49m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/7f4220ed16c84c91ab8e27310cf2eac4 : SUCCESS in 32m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/d7c6c3ea089647e0868e445b89a03cdb : SUCCESS in 38m 49s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/fe16ceffb9ce4ad697cf54275865faad : SUCCESS in 33m 10s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/323dfaf68cc44f049ad8683bbdc786d1 : FAILURE in 36m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/db105a8e1cb541e3b463c9ed0934d565 : FAILURE in 1h 28m 07s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5b26dde3699447bebce3898dd8f5d681 : FAILURE in 53m 16s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/ada5a1e35efb4395bab636eaa585c014 : FAILURE in 57m 05s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"fd700f512e9866d3b372b56db875a9481d42b710","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-04 09:34:28.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"decf2c73cf5185b26f6bf74a462c103f907f88a0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-04 09:35:27.000000000","message":"Patch Set 15:\n\n(3 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":15},{"id":"ca66c7e09b5307c9be839d28a859c4edf36606c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:00:14.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdde98a858a645e8b712425a8464f3d5 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb9d6d305614992947b15150fea7a2a : SUCCESS in 4m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f2cd22bd5ec4841b7b6739786756e33 : SUCCESS in 2h 07m 27s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/32069ed926d14f4289ae143f2a1e0dc9 : SUCCESS in 1h 33m 28s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/fbfa6c39cf704b4d80be5b351cbc20f2 : SUCCESS in 1h 35m 00s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/84995e9aec7d4b8f8c620e5dcd6a2cef : SUCCESS in 1h 53m 48s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/42040b981e234f08bd42098b6c9e0e89 : SUCCESS in 1h 42m 47s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/03aa00a1f89f4914b12f1125f9c3ac65 : SUCCESS in 2h 17m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51655403f9ca41e3824bf8958bc5e3bf : SUCCESS in 1h 25m 11s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/5b58c1ea649f4aa8bcfbb59466c2aee8 : SUCCESS in 2h 01m 37s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/3e9930607498493eab771ebcb25d1bcf : SUCCESS in 1h 29m 14s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/64f8f43044db4befa25f525295c365bf : SUCCESS in 1h 29m 37s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1f17b732e6894140b0aa7b0979d5badc : FAILURE in 1h 02m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0a58fe87539045e8bc504ebf2a9fc125 : SUCCESS in 36m 03s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/6a111fc03975466d9b4c542687ae1042 : SUCCESS in 35m 17s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/ccb0db43ddc14383a179548196c47289 : SUCCESS in 35m 34s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/904dc4b9ef0f47358cdd97ac1c60bd8a : SUCCESS in 36m 06s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/6840189501304b5780f485bb431e038b : FAILURE in 31m 24s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0d04c9edf4844f81977f9673420af725 : SUCCESS in 1h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9beacc9b7be346d28261ab4a4daa39f9 : FAILURE in 36m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d8ee3525e0e648bb9208ec06fae5811e : SUCCESS in 2h 00m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc48e6249bfb471494ea1929442e878b : SUCCESS in 36m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/94a94d7af8be46ef9065544ebc40fa6e : SUCCESS in 35m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/af0a9cf856ac4f49abe763164727be1f : SUCCESS in 33m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/113470030769422a9e2bacf8aca3f541 : SUCCESS in 34m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/dca9662f922a41d19ed02948ad3e164f : FAILURE in 31m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b39ca77b4cde4fb8bf486fc78e507af4 : FAILURE in 45m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/1d28eac1be2b4144bb669e8c95ab9e9c : SUCCESS in 35m 37s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/d37f490dab9b46758bed1c02754d2c14 : SUCCESS in 33m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/63531d0919b345808a05d94b53267cb4 : SUCCESS in 32m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/f26e35ea34b24ff4a4a91c38af259db6 : FAILURE in 30m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/253f29ea6e124a43aef1f2f32d8bffcb : SUCCESS in 1h 21m 28s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/609db0f785a3499db8ad08c3df56611c : FAILURE in 44m 46s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/ba2198c934384aa99845d3337a972999 : FAILURE in 54m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e7a90f5b840ce761f34039d6de8ff9e979b23dec","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-07 08:34:12.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"83703212d3ce38df2252539cbe9ac782fc28a33a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 11:05:40.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/158ba9cef5fc4e93b807c359e8b249f9 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b313f52b9c8a4cfe92d084f0a623a259 : SUCCESS in 2m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c97a31d5668f447e9de3707ceda12682 : SUCCESS in 2h 06m 18s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/195de61e2d06469c975b3dcbda967a21 : SUCCESS in 1h 29m 57s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/c6a200b692c84a509dbe39701c0cd293 : SUCCESS in 1h 31m 26s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/ef651127c3884757a31bbff6b6ccf53c : FAILURE in 1h 31m 40s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/0678f57291d540beabf0e94ad94198dd : SUCCESS in 1h 14m 40s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/26d8b2b4212a415d84b1c40dab38330b : SUCCESS in 2h 14m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61d8ada3fc7c487981b6ab44900db9cc : FAILURE in 1h 41m 51s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/4ef7cc624f444792b511c89a05f6f79b : SUCCESS in 1h 14m 11s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/1cda98bb2d244ebe85942263ea667c1e : SUCCESS in 1h 42m 10s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/68aed9528bec4d75815b4a0aa2f9dcb2 : SUCCESS in 1h 44m 46s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/3bb0fa4a0cca4ae08850a831cc8734c5 : FAILURE in 1h 05m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb70baa87cc2445590316596ab5f94da : SUCCESS in 36m 18s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/66cd982dd230409bb15c89a7f284cc0c : SUCCESS in 43m 31s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/fcea8791fadf44ebb135125632c7a90d : SUCCESS in 35m 24s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/87196c4531c842ce85c71fdfa60689c9 : SUCCESS in 47m 25s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/da297bb0e795497592191eb724c8fb57 : FAILURE in 28m 20s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/811601360c014431bcccd0906c2d0fd3 : SUCCESS in 1h 25m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9a115c5043041909ce6bc4ee7580b71 : FAILURE in 30m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5effda32c7b4424db96e49e09a29bcc1 : SUCCESS in 2h 16m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b349474642543d6917e293ea5ea58df : SUCCESS in 29m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/b290637bb371411f9add0649b49fa043 : SUCCESS in 31m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/54c4858cfc554566bc3540c1cae5f596 : SUCCESS in 30m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9a3f19ea19ba4e2b809f3cf2d12df489 : SUCCESS in 44m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/eb0599ff1d5046d79da42e33fbd6ec9a : FAILURE in 31m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bfc85ec4a5054b43858bcb4d162bb2f8 : FAILURE in 42m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/97a85083f52141a79462a47b5901dc82 : SUCCESS in 31m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/a414eebd2f5d4295a022a35c424c107a : SUCCESS in 41m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/52711af7cc7c49ecb2d387a799af6fde : SUCCESS in 36m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/b337b5ce662849eaacb5cafbc4b2b9d4 : FAILURE in 27m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7692ef4b74814eb594c53dfc711ce18d : SUCCESS in 2h 02m 05s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/9df6da96bee44043b27bf27d0618a83c : FAILURE in 51m 16s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a073e80b4e124554beeefb363f8f26db : FAILURE in 55m 18s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"532a2e97e0829cfbf18cfc0eda8397dfd1780b6d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-09 15:31:17.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"670c9a75109a88c4e36e19ef0307d74970e2020d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:45:10.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52d6d76af4547478dc108e2bb552bf9 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3493d69eb34147abb40d48af52f6429a : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/833f4ddb2f604d99b1c45b05d41e30db : SUCCESS in 2h 05m 26s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/6e89832c80284884bb3e2467933a2b55 : SUCCESS in 2h 52m 09s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/0d74b005d4a04a49a6c206189bdac5f7 : SUCCESS in 1h 41m 44s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/0b2a399c95384df2ba6238184d9b5321 : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/a97acf4545fa4dcda45818394ebc0c1d : SUCCESS in 1h 15m 36s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/5faaae1528b94186beb1b40536d549b4 : SUCCESS in 2h 27m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02666ee3e2854f61b6423789a64f8943 : SUCCESS in 1h 19m 45s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/272d2fe813c54f4ebfa7616334338588 : FAILURE in 1h 52m 18s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/4afc20e76b91492b9aacf09041678f92 : SUCCESS in 1h 38m 42s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/178e10aa416846a48ba0f7209d28f427 : SUCCESS in 1h 27m 22s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/5cce79404a1645779ac68cf32a2010e1 : SUCCESS in 1h 34m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/312e45ce629c487bb4c088801c40ac88 : SUCCESS in 36m 53s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/299c496e60a9456a93882acc72e17111 : SUCCESS in 34m 15s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/b3e6b8b39dac4f3185881de3acb450cf : SUCCESS in 35m 21s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/1aa0f6aef6f646bc87e15381e11984e8 : SUCCESS in 36m 03s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/a2e01fe275084847becf11ea155b36bd : SUCCESS in 37m 21s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ee07277960dc4d7fafde0142a3eb578c : SUCCESS in 1h 32m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd6892d9cc9a4cc983d75a6124e7be05 : SUCCESS in 1h 55m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bc944ca15fe475eabe6f18246a095b8 : SUCCESS in 2h 01m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ab9025faeae4e698dba116f2bd17a34 : SUCCESS in 1h 58m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf4d14d8396a47138d0b611cb7e1e6b0 : SUCCESS in 37m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/f7440c843d52421799bd7c14c22aac1c : SUCCESS in 37m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/dedb591fa99e406fbc0e421bd1a7a60c : SUCCESS in 36m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/cd8106124b2e407eb5a870b05b74934f : SUCCESS in 35m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/22f250e3417641febe4b03a46de58ffc : SUCCESS in 36m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/24ccbee9ac97483cabafd5155fef7c5d : FAILURE in 47m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/39d940f43a8a4aefbaa008e75fa83cc9 : FAILURE in 47m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/88313c07180646bc84f64a4803a06a06 : SUCCESS in 32m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/ec44e3308bd944669fcaf7ed7ae863d2 : SUCCESS in 33m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/28c352786a1142a39268869d8b8d8399 : SUCCESS in 33m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cabe462e27c74a9c9546b8bedd288c14 : SUCCESS in 1h 32m 38s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/77b0476d9be84dc08604c6063f37a80a : FAILURE in 42m 33s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b9a72f55d61d428cb70da5ac6249e89f : FAILURE in 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"fdf4a2f4a076ee590c8a6ce5f8ed390627889cdf","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-10 08:00:51.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3fab6b1dfc06044781de4e0021a3c437bd707cff","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-10 08:07:25.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a5a79db9450b483cd7677a9015aa2253abf07c6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:52:34.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d48508f2dce46588d761299ffb3f00a : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa1ce1a7c12467e809cb1ce1bf003cc : SUCCESS in 5m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c2c46b24fec34f3d88bd50b8430e2236 : SUCCESS in 2h 19m 34s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/0212865552f8449783e0f6ebdf38adf1 : SUCCESS in 2h 16m 59s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/71bd5575bade478b91d7e4e3d417334e : SUCCESS in 1h 39m 16s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/454348de58b24c838e50c4537bfd9d90 : SUCCESS in 1h 31m 14s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/04ba266b4e944b979b93100d67283722 : SUCCESS in 1h 38m 22s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/8b0efd139ce44b0ab79a1a3c5f701c3c : SUCCESS in 2h 45m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7e06b114ac542189c32baaf5565bbd9 : SUCCESS in 2h 09m 38s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/2c7a8d21d9eb40fe9881e2bb0b188c99 : FAILURE in 1h 58m 24s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/91a7c93a8e1d4536b6e70f39470947ba : SUCCESS in 1h 56m 08s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/76b4a9eff78e443fa71311f365b65e0b : SUCCESS in 1h 54m 43s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/d7ccf316321b4ed3969d09500c9bc779 : SUCCESS in 2h 20m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fbc554f854cb440dba0e5690825991f2 : SUCCESS in 42m 50s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/708c8fa1607e405d9d75bbdee3719736 : SUCCESS in 42m 36s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/c4d6015b0c434fd3b8b5316d3c62f4e5 : SUCCESS in 50m 27s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/6cc63dbf745547778223e50f95ac2faf : SUCCESS in 33m 55s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/304520a6360045b58ae9310cef344587 : SUCCESS in 48m 11s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fefdc80aa757492a903343186544ec6b : SUCCESS in 1h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/70c086b4636e4740968b5ff8cc3e4899 : SUCCESS in 2h 15m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a767e160478447bb64fd10e1d41b005 : SUCCESS in 2h 15m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4bed0d3bd8454441bf296e9f7bcd0fd1 : SUCCESS in 2h 46m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ebe4fea9c1a748c9b194ab6bad08552a : SUCCESS in 33m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/389e00e2b4034f6ebdce364b9b9d1e94 : SUCCESS in 43m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/1aaf65a480d643ef8276985dab7cc9d2 : SUCCESS in 43m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/063cec7e74b24cada73aa4a34d8b067f : SUCCESS in 44m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/3f8d21f5767247a5bdda1d1462240638 : SUCCESS in 44m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a5a46485109244b5a9b60896ef16ed76 : FAILURE in 49m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/58fb2bdaebd641f590f5e9a5b77c4a6e : FAILURE in 50m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/c2dcecc76c0146cfbb7db7d5171d5cb0 : SUCCESS in 30m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/01d9b645611e402889e68d4c962bd870 : FAILURE in 23m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/27e0330067a6428b9653dc4ed14f08e0 : SUCCESS in 31m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07bd4dab85cc4fb5952b5d8c36b65b19 : SUCCESS in 1h 24m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/7224ae1c92524c6b988de1fa7766c07b : FAILURE in 45m 17s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5ab857a10a364e5a970e52037722a025 : FAILURE in 1h 08m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1fac5ccdf804c3b5e0905f73f6f88903fef32a74","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-10 11:16:26.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f4faf28f4b4bf6370bae630ee12b9e9efe0092d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 14:29:31.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e399141b910b4b999e0cc5bbb59bea50 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31cdf951b6af4de7a57f954c3fc8fecb : SUCCESS in 3m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2903730249f641b9b3a5979b0a5ec6c9 : SUCCESS in 2h 07m 39s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/620c1a52f8c44a44b9433f3187535028 : SUCCESS in 2h 05m 07s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/547e1f4b2420460292309f13ba6b7452 : SUCCESS in 1h 29m 42s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/f3b05d1459b543f396ecce11423d1a9c : SUCCESS in 1h 30m 01s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/cfad36154e5b41dfa3700033bec04f31 : SUCCESS in 1h 04m 52s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/d553d6a58d78492895e38340be586fc0 : TIMED_OUT in 3h 04m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d913febdaacc494cb4862506a1df75bc : SUCCESS in 1h 25m 49s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/732843fb5fb14f95922958d569a22b08 : FAILURE in 1h 11m 20s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/d92f40fde98046c693ac63d29e94ea0a : SUCCESS in 1h 26m 21s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/668d41ce612142508c5500619aa43866 : SUCCESS in 1h 31m 18s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/4d2a85363b1a4b8caa9a01f7fef36c40 : SUCCESS in 2h 00m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b6e68fc0ec894099876e3d984d91a460 : SUCCESS in 35m 18s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/4e9ae9ed6b6f42d5badde168971fd1d2 : SUCCESS in 37m 13s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/7aa8b578063b4387a3c2878da3067cc5 : SUCCESS in 39m 36s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/df04f9fbc7434814a7ecf4da3ee6727d : SUCCESS in 38m 57s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b6001288206d43dfa89ab9f8aede1120 : SUCCESS in 34m 07s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6aa868ac0ba40b087a8930ff14adb4d : SUCCESS in 2h 08m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/c42336d45ab742f6b6c7e0b2d473bcf9 : SUCCESS in 1h 41m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/36dd1e5be7504a968623531cc53ecea6 : SUCCESS in 1h 51m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/710d1d172a6a4917b6b9260c5260d526 : SUCCESS in 2h 13m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f113a620df449b5a66883d4cb8a0a45 : SUCCESS in 31m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/f92941f170dc4e9c8534913cf7854dc8 : SUCCESS in 34m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/62b55818c9ae481d8a575d0c331eda59 : SUCCESS in 32m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/882477d0a17748f29fec76e3b0b4836f : SUCCESS in 31m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/de1a2ea82e854025a479ae9f08bc6941 : SUCCESS in 32m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/afbb15d9cd564266a5fa4696e842365b : FAILURE in 42m 19s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/d6f895bfb65d41a38077174c87742d59 : FAILURE in 46m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/5308808bbec0439298c8a7662913d0f2 : SUCCESS in 37m 30s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/429ff034b1044876a102148854ca918f : SUCCESS in 31m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/1c14d91e43404a30a32bf52d250a23c0 : SUCCESS in 39m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/12cbadcfb85f478b949eff8e33e8d5b7 : SUCCESS in 1h 52m 21s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/3e51b2287080425daef0591294c54bbf : FAILURE in 1h 00m 20s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/17c5ae776a4d4b4b991e6f2884d71101 : FAILURE in 1h 00m 34s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c2137cadbd2d763f867d23c5d7d043ec70cdbef9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-11 10:18:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":15},{"id":"6cb9c1df0c53742d1d15a768188a9c0969cc3abc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-11 10:22:33.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"16c483e18063e0864411f7d01bfaf8d95c5613f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 13:27:08.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0502885cd12f433383d31a21c010504f : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7abc281321f14d72a0c547a083500d61 : SUCCESS in 3m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf8f59d9e68f40f2aedfdebd142bb7ca : SUCCESS in 2h 44m 31s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/9994301a356f4c4384fd468eb2b2c33d : SUCCESS in 2h 56m 52s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/0d47c4c609fc4c8cb466d9304ec1e508 : SUCCESS in 1h 26m 52s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/f03e95a07b454a2eb07c990d27754ca1 : SUCCESS in 1h 36m 40s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/1c1bcb0bbbd641ab8d8a5a0804ed4da1 : SUCCESS in 1h 02m 04s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/baf9f5fd079543e89d5d505cac085de2 : SUCCESS in 2h 37m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18bf7dea38fb4107951b001c77027978 : SUCCESS in 1h 50m 29s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/3f9edfe37a8c4551a3b0a781e61081b4 : FAILURE in 2h 02m 30s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/9a493cd7a13f4025be42d6e567d858e6 : SUCCESS in 1h 53m 02s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/faf105b8a0f841cc817c66e37adba093 : FAILURE in 1h 45m 39s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/08ba0b9a71694e0db63844be719b6edc : FAILURE in 2h 27m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48dd15c750544399a4d61a624f12f488 : SUCCESS in 48m 25s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/c1efb5fe0f5742e396a0a0ffaae9c84d : SUCCESS in 45m 21s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/05f974879eb64cd68d3ebdec05b5e83e : SUCCESS in 52m 08s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/eada7b2a85e1443aa3c63f2eb431ec15 : SUCCESS in 32m 27s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/3329240723c2439997957c0580807569 : SUCCESS in 33m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d8eb3cc071f340d8acb4749979995559 : FAILURE in 1h 32m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc14227303b34e09859ae5077aaeb53a : SUCCESS in 1h 54m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a059ae7bd8c413ba9bd2d02dd233c55 : SUCCESS in 1h 58m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1972d959ba954c05939fa47cd84180ec : SUCCESS in 2h 12m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a895bcac11144d3bca934ad2abe7778 : SUCCESS in 33m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/f7bf7de0a0534389addf3dbf113f5ae4 : SUCCESS in 34m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/e1ce36ca5f5b488c9e2a1ca4c567176e : SUCCESS in 32m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/ca7cae5ec38c479ebc8c58eb0c687fef : SUCCESS in 31m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/29ad2838313d4b9381ae5e8d0f17785d : SUCCESS in 31m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5856e57bc9044fc081018bec7ba655ee : FAILURE in 42m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/824b1765d20346f688930827330e4857 : FAILURE in 43m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/3818f4b205554ce0835d555cee67f3a5 : SUCCESS in 31m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/762a27a93b32487cb929da547370a993 : SUCCESS in 30m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/489d9a67c80e4a078eb01fd935326723 : SUCCESS in 31m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4b500f12ee53442fb950d6790177e79f : SUCCESS in 1h 36m 07s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2d76415d9cdb4397a9fbe6c00a5bae04 : FAILURE in 46m 13s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/943fd8e0eb4242adb20b004a896d5b96 : FAILURE in 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"81a72019c9bcdc7b4d52cc28b9971920ef305754","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-11 13:52:58.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4dc0bb6a435c9e1da06ad5c88bfb9d3966fab1fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-11 16:09:47.000000000","message":"Uploaded patch set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"ecf700561734df58fb347b81e2fb130e267b7302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 17:20:18.000000000","message":"Patch Set 17: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":17},{"id":"a6e0113549ee6bf14009e8016073d0ce7fd56fee","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-11 17:22:17.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a8bbac6d74661c3c0925da78c770b10a2257a526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 20:33:39.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e63c3095f524321a82a87405eeeeae9 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa47318586ee4955a78d29005af4afc7 : SUCCESS in 4m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7c0fcf68332041eab92cffcf4fed6271 : SUCCESS in 2h 35m 31s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/0542826f251d4511a80acc9c7b525c1b : SUCCESS in 2h 50m 35s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/5ea5f691fd044f1583dbfb187d44bd1a : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/787ef9bf120f46e99756a37cf0bb13f3 : SUCCESS in 1h 35m 35s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/47864023036d427abb051f23ffd02f73 : SUCCESS in 1h 18m 36s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/290e3d93429646c6ae238daa4415a195 : SUCCESS in 2h 37m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a2c0b608aa143219b72b865ec317907 : SUCCESS in 1h 19m 06s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/c43302db59784c0894920cadff2c740f : SUCCESS in 1h 37m 05s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/788e4ba040b043b8b9e391a35698f474 : SUCCESS in 1h 43m 53s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/c362a7a043c14b0e9d673866abeee3ff : SUCCESS in 1h 53m 30s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/afc7834935df408fa332fc300d33afe2 : SUCCESS in 2h 02m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a280ff8a0fdb4bcc873702214387c985 : SUCCESS in 36m 10s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/ef0fd4db0784474eb7f0c9bdf8494e22 : SUCCESS in 38m 42s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/ddfe7ccdde0d4a3cabffebacfc2a70fc : SUCCESS in 39m 41s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/4bb188d9de68448d9660c13ddeb8a5f2 : SUCCESS in 34m 02s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f3cba28cee334cc992630458125e0149 : SUCCESS in 43m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3db900d80a474348a1d1237eb96e1d1f : SUCCESS in 1h 33m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/129777201cf44d159bef7e1bb20aadde : SUCCESS in 1h 43m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4dc0d2da02e4744a1d5afc48f951eeb : SUCCESS in 1h 38m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cafbd47e7a2b46bfad853b3f4ff46441 : SUCCESS in 3h 00m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e273f1d21e2648438456468bf811204e : SUCCESS in 32m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/7f9f4d12236b4fa9ba1b7d02bac6bd52 : SUCCESS in 37m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/8fab0f0f0d454b7a928385c49c1c1e98 : SUCCESS in 39m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/cbad0c7e16b642bbae951560aeb75c42 : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/b902582eeb0c45b8b092670152dea617 : SUCCESS in 35m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4eb08da04d724079b38e93a43476eb2e : FAILURE in 43m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/a19c45cb1b8c427c8aac25a7aaf7937e : FAILURE in 44m 16s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/1b2f9695ba2d45a68211fb8abdefa42f : SUCCESS in 34m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/c6bd44b9c2674c07ac5f9ee2e589df7b : SUCCESS in 34m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/6fff5bb21da845cfb28c157d68e161a4 : SUCCESS in 38m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/95fb3f7c292f40f0a82b9e77ff2bf39e : SUCCESS in 1h 38m 00s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/fbdfbb4b6953463db7fbfb60f7ea05fb : FAILURE in 58m 41s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/aa475e2830614d6ba1998a22c14d0ffa : FAILURE in 52m 26s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"03fbcfb1e74ddb3e0da21d4019cb5184381a0475","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-14 07:28:06.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"41ff20b3701e4b880c72737d585eaf37d7e0e15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 10:04:38.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3924ffb0e0145d1a7d222f5abf062cd : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f56e1c70ce424ab8b7e9ac66572a29 : SUCCESS in 3m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f29b95ef8f944d028f8a8371bd6247da : SUCCESS in 2h 31m 06s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/1140b8abb8ed4ea3be97c2727d8922b3 : SUCCESS in 2h 07m 57s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/1b6a6a4e3efe4cd898cec611f4b3a73a : SUCCESS in 1h 25m 29s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/f239bdb3fc4046788741fe66810e163d : SUCCESS in 1h 24m 57s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/61a4e9e671834cf8b2e2dfc4ce3dc3b2 : SUCCESS in 1h 04m 41s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/498cc4db9cf44f7d8744cfe3b0546a9a : SUCCESS in 2h 28m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df21617d5f884ad3990fdaaa682ddf56 : SUCCESS in 1h 30m 01s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/b0a7eee70bef42e9b49c3375d3f24c3f : FAILURE in 1h 39m 01s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/23683161cb7741f0a83449028232b918 : SUCCESS in 1h 13m 33s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/9665356a933e4ff7be727acd9f869464 : SUCCESS in 1h 16m 26s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/57966dd3dbcc40f1951079d0aae1d7d7 : SUCCESS in 1h 22m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0cb908266659409eb89405c6c0e5b0f7 : SUCCESS in 34m 21s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/5ca241a46a8446d1bf08c03d398a5bd0 : SUCCESS in 36m 20s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/e94ad930052049028a0351efd4c2c856 : SUCCESS in 33m 47s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/6997354261d44e7aa42cb0d017cd902e : SUCCESS in 33m 50s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/52868bc7772a4102a4f8c36a4a4b9b7e : SUCCESS in 34m 07s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ec54120c21374af484b5b91fa9dcd9f0 : SUCCESS in 1h 32m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a30d1ef1099d4cb7aefc124c709bc953 : SUCCESS in 1h 43m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc07a2593982450bbc3a69121ce3245b : SUCCESS in 2h 20m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/79ae8236cd824767a4c477df3ad0f91e : SUCCESS in 1h 55m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5cf6c8518b642b3bf5b554127bc85ab : SUCCESS in 30m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/1133cd63aeea4368b0a548037b353b00 : SUCCESS in 32m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/388e628b1a31411c9b02626719fae91f : SUCCESS in 43m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/055ffc24b2b4429a86f56ec0ae927345 : SUCCESS in 29m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/9a69b4e569dc46c5b916ec582f51d2e9 : SUCCESS in 34m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/837cdd2e8a1840bca5f3ee0808e2ce23 : FAILURE in 40m 35s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/b012d72eef47438bafd0d21256b65804 : FAILURE in 45m 29s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/30168eec3c6d4fdfa3a4d332e2784a0e : SUCCESS in 43m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/2db6a93f77434d2597bb352fc2e6f09b : SUCCESS in 36m 31s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/ff34e2f3fb164ac2a7ec16f6f2052e43 : SUCCESS in 31m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6b1b842f16eb45089ec0ff4396caedf7 : SUCCESS in 2h 24m 50s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5b6afb4a8da64081893bfa543fe158eb : FAILURE in 40m 27s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/080bdbe7c9e5472386c3163b68274b71 : FAILURE in 50m 46s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"db09f2bd8f6b257a2005c6ac96b555c7e4201a6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-21 16:05:43.000000000","message":"Uploaded patch set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"59f5d9b10c8b6c65ed7499817bde396cc539289f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:48:00.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34be1149895a462a9aa26431fdf9b766 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fce12fae15d474db074d5c8725e7877 : SUCCESS in 4m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1eb1ab42ab6945b4ae21cddf81fc5eea : SUCCESS in 2h 25m 49s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/21c2d0a974884ed3a0c2bde63bb1d622 : SUCCESS in 2h 26m 17s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/e2a6647e3c154d77afa6f77d9c20389e : SUCCESS in 1h 36m 22s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/575adcf21d654574a6d73901217775e9 : SUCCESS in 1h 53m 31s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/0457b4f4647346e2b5ffe1e3959be7ff : SUCCESS in 1h 39m 14s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/565ae445fcd044e2a63ce0ebae90d528 : SUCCESS in 2h 35m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f748226b0f40441da1b6c43f109f5b01 : SUCCESS in 1h 26m 50s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/ca62e56bfc0e48798ba923a5360f239b : SUCCESS in 1h 59m 27s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/0304ba4370ee46cf9ba7079c6843000e : SUCCESS in 1h 47m 02s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/e470c87799464e55b31760c7ef0d26b9 : SUCCESS in 1h 41m 45s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ae24d272d5cf454eaa053754007eedd7 : FAILURE in 1h 29m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9525495c89345868cd95ef3583eafd7 : SUCCESS in 37m 07s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/9029b0bdf9314e90a72f7163230cc522 : SUCCESS in 35m 57s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/6196901c776248669e5561513a645524 : SUCCESS in 37m 05s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/3098bf40ca1448cfbfa3f0a28a413808 : SUCCESS in 36m 51s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/13aac281fbcd42eab6689f0f5566aa2d : SUCCESS in 41m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/53f3ff400e0c4a71905c461c25858483 : SUCCESS in 2h 00m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/49e613c7ab1e4599a8a084642a54f48f : SUCCESS in 2h 00m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/be52924cce2e4691b768396fcb9b7d6c : SUCCESS in 1h 56m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c9de8afbc134a82b0950e79ade67be6 : FAILURE in 2h 23m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b21d328192f64d35805a0f74651da147 : SUCCESS in 34m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/0f0987a729bf41b986cd1791ef19527f : SUCCESS in 34m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/dc178293488747fb8505b159be3fec73 : SUCCESS in 34m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/3951f0635af448f0a65fa025f7e74488 : SUCCESS in 33m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ccc49cf5ec0a4af48e37cab34bb8613f : SUCCESS in 35m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/71338dd8623c49088ca6e0354dec33de : FAILURE in 44m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/442b80215f684453b7a2e51c62d4ae77 : FAILURE in 42m 53s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/77a7d910639b4a4482237933445e037a : SUCCESS in 31m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/ec3168b348fb4417b0d6b06ab62f7491 : SUCCESS in 32m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/1e8ee35d3eeb49368344a83e8002bed0 : SUCCESS in 33m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba8e32c9ddf0456aaa91c38fcd8fb6fe : SUCCESS in 1h 30m 30s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/58421723e973469a9a1a30f5bfa52172 : FAILURE in 50m 32s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a19b3f58d95e461dab629ae4b3eb1b7b : FAILURE in 1h 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"c936cbbba5c5af1d103f79e64f4da4e499ac461c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-14 17:15:31.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"a82504a8ac6e1799040faa041654d91b7e07fa0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 20:15:26.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/059050d9175b442b8bf2eb0560a508d4 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02bde1f1c267467eb8e17e824750b300 : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9457f2b719064eeb85293733aa82e14c : SUCCESS in 2h 31m 57s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/fe6e7854735d40b288d4bfd4bc0b3ca7 : SUCCESS in 2h 33m 21s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/0b10228fefca468bb8fa3204cc038c45 : SUCCESS in 1h 33m 24s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/9eb1f5570a1c45e69faec8d5357190e7 : SUCCESS in 1h 32m 17s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/adc92e25af3546e3b9ce56a13eba50f4 : SUCCESS in 1h 11m 24s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/d7ccdc567b634549b5f559e39adef1c6 : SUCCESS in 2h 18m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ade974f30af47c19aacfa0cd3f80f70 : SUCCESS in 1h 29m 21s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/e5c3fb11648a453e98af24fca93af719 : SUCCESS in 1h 32m 56s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/59d218a7323544eea2719ee59d68c0df : SUCCESS in 1h 22m 24s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/803fec1873bb4bd1961ba781f97438c9 : SUCCESS in 1h 43m 11s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c59b6b024aba45cd8334e8aa8de4ca6e : SUCCESS in 1h 30m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef1aef08f2aa4ee485ce79076acb8ee2 : SUCCESS in 38m 02s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/4aa695ee6ac84afdba5e2b72c10d7bcc : SUCCESS in 40m 26s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/3a0340bf3e66491fbed402bcf37f79ef : SUCCESS in 38m 39s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/e3db55301a6a402698a38f779dd1cda3 : SUCCESS in 38m 42s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/7b38cfbcf1bf41769d051f5a2f81ab31 : SUCCESS in 41m 42s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4ce2995031f5423fb3d4bc86972d5d70 : SUCCESS in 1h 46m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a348763d6bc45fda137ef7657367cdf : SUCCESS in 1h 46m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b53b1e5cf48a42b68cb7952983dbe817 : FAILURE in 20m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/07bb299f73924a2081ac8c2c18e79ed8 : SUCCESS in 2h 56m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8eaf0e3e1c1f4cc98e2955bde5baa080 : SUCCESS in 34m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/0fc46d4eaacd48aeb5b00aa45b1753b8 : SUCCESS in 37m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/0fb9103cf92d4716bb8cc991dd2d4e99 : SUCCESS in 41m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/4834b0839a1942a4bcacdeca1ebf2133 : SUCCESS in 36m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2c60e9a40e54404d89f48dbb00650b70 : SUCCESS in 1h 04m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1ba9ff287b4d4da2a5f9b036037d343a : FAILURE in 45m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/54f7b890e1b24ae38ff369fdf161a234 : FAILURE in 44m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/3e96e9cb9a0d440b8361861e32e1273f : SUCCESS in 37m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/44bdf7ba8ec6427484c8c1efcce05eee : SUCCESS in 39m 17s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/563c8d7fc282464b9284a3af81e4814a : SUCCESS in 37m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3ac4f1c529f4a3fa97ca33a4cdb7a64 : FAILURE in 2h 06m 28s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/3d9e6148cb51493eb0f7f71b00cf0764 : FAILURE in 58m 25s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/37b873f3605a4396be28bd4e24037c8f : FAILURE in 1h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"0c69537f858041ccc5c0c90d0269fb91107c3a9d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-15 08:41:32.000000000","message":"Patch Set 19:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":19},{"id":"9935bf6ab2d353763ab343d61ff4c38027ac5161","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 12:51:59.000000000","message":"Patch Set 19: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":19},{"id":"905998444807f741bafee9b58ed1c00e9979d10e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-06 14:15:29.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"a623f0b173617e4d8908624d0bea073e8ad0e228","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-06 14:16:19.000000000","message":"Patch Set 20:\n\n(6 comments)","accounts_in_message":[],"_revision_number":20},{"id":"e17a059680a0399328cd528135ac8e15752bf006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 16:57:53.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2b89fd3d7d4e1d9c78afc78532dbf9 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d33ed5c7e4d4ae890790f44a682979f : SUCCESS in 3m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4290e01b5418401f85d2432b99947013 : SUCCESS in 2h 40m 26s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/17a53f9db91a48f4a6e362b9096edf2d : SUCCESS in 2h 09m 01s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/9f3fee19ffda4d9384725be9c5ebb7dd : SUCCESS in 2h 19m 16s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/60435435b33245e482a49f13d57ff033 : SUCCESS in 1h 52m 21s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/fb856acca241452f921bee7d6b2fc44d : SUCCESS in 1h 36m 46s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/6ffa90cf14454a2188c0ab9c34243317 : SUCCESS in 2h 23m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df8de2889d504b73bebfa3297c8648bd : SUCCESS in 1h 22m 29s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/06727afe7a884396a88bd7b98bc0de64 : SUCCESS in 1h 23m 06s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/b4703bd5a3564828b911971e8208e20c : SUCCESS in 1h 32m 30s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/c9d125db2aff40df947c04062474b29f : SUCCESS in 1h 22m 18s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/53f5a8bb4e9c4cbda077ae522f899ca2 : SUCCESS in 1h 28m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/332351955c94444d9f26dc88695164cd : SUCCESS in 46m 54s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/9610bd07cccb4073b5f4c90bacf98451 : SUCCESS in 35m 40s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/a0fff0bd1d8142a29f2882b94fa1e495 : SUCCESS in 39m 10s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/b613e882f6444b0d99ae6db8a08727cb : SUCCESS in 35m 20s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/e301a2687d8b44e1959ec5fe13dc8652 : SUCCESS in 34m 54s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/67fbe509839f4a00aeb61d83e2a60b8e : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/74344a358ecd4878973d571e18c236aa : FAILURE in 17m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/40c1633fcf5e4bae908119bb8f0534c2 : FAILURE in 1h 28m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1f9b8b5ce9746dfb34a9133cda7fd70 : SUCCESS in 2h 00m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f463d00e73a245a99eca666bb3694240 : SUCCESS in 51m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/12c74cf4d14b409ab35ad2ba554c704d : SUCCESS in 32m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/f09c6540dec345bf8fa52534f3dae0ba : SUCCESS in 36m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/70cd7c6e2cae4523ac46335f008e5b06 : SUCCESS in 37m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/0a32ded6304f495991ea231872ef56c1 : SUCCESS in 35m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fce8aeb4fe784a2d8724e489af7e7da4 : FAILURE in 44m 47s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-yoga https://zuul.opendev.org/t/openstack/build/60679446179f496b9c1468fcb2674546 : FAILURE in 44m 48s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/672ed9239e2a4a568ffbb603cf45bae9 : FAILURE in 44m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/4d08213569f3491788acae03bf720f53 : SUCCESS in 35m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/be1714e8b2b4467492dbde34a1939748 : SUCCESS in 35m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ec06c0406ac479b839fca9ecbe7d287 : SUCCESS in 1h 31m 56s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/fec0366210034b118afde668c49da555 : FAILURE in 56m 02s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/ae98e61d3d1c4b249cb59c407ecbc5bd : FAILURE in 1h 00m 42s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"a97219e29c0916d570c75cc0536ca04596537a39","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-07 13:23:42.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"1ea0b445b1100cc76df4d72c0bf5f6c65cddfb2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:09:41.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ed2ec556a064eaf839874171e09a130\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44cba0d875a341b4bc782c2e606287fd : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e7065a777d480c8fd0a87dadb171ab : SUCCESS in 2m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/faa4e4ae67464c87b54b875479fd0628 : SUCCESS in 2h 10m 18s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/3f496d1b50924568af85128e9588931d : SUCCESS in 2h 13m 43s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/bec5c25645a8465ba594b1bd2f889882 : SUCCESS in 2h 13m 43s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/1f821af99cb64f41a4b9b409a0b6ee88 : SUCCESS in 1h 29m 38s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/27e0c7c41f824b22b6f4f4601ea2c7de : SUCCESS in 1h 29m 58s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/4dd0c8ebc59f498d9837dd2e7bee267d : SUCCESS in 2h 31m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00905d42a2b84d1fbe7fee1015eb88e0 : SUCCESS in 1h 39m 58s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/5509215888814d78bd16d8a749c2dcda : SUCCESS in 1h 32m 04s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/3c84d186d2b14d27bb9517f574432ad4 : SUCCESS in 1h 38m 12s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/03d890edbba844b1bcb4b56cf59b0813 : SUCCESS in 1h 46m 04s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/46813fa1509d41f68c43a3da897023bb : SUCCESS in 2h 14m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0f39a7e87b549cdac8f5d4ac17e967a : SUCCESS in 36m 46s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/6364211a3eca45b9b2a3bd047ed83225 : SUCCESS in 42m 02s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/b62ff79ebd13436a814b80092098711a : SUCCESS in 35m 57s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/9e8832f4096b43deb4d6e0c826bfcfa9 : SUCCESS in 43m 26s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/4d120f082fb34bbf95a62b995d948c08 : SUCCESS in 43m 26s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/26073f1e9ec94661b4d77df67b5a6dfc : SUCCESS in 2h 07m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/80f25c62172b436388d544816a1532aa : FAILURE in 10m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b949e1f63c324d369365b24934a82ede : SUCCESS in 1h 52m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c211803764940928bb7b2ab6a9eab2c : SUCCESS in 2h 38m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d47ffffa48274b04a696f2a3dd263be5 : SUCCESS in 41m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/c66c44b8160946549056ec61bf203c62 : SUCCESS in 40m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/cbf1ff680b59432ab6be22dc9f0812ab : SUCCESS in 34m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/45daa96c97084021be9e3a805549f06c : SUCCESS in 44m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/0e8f27f2bc964c47a5f8d35cc01bc706 : SUCCESS in 31m 59s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3edd2e8f24a74f98b8861d965dc6aa4d : FAILURE in 48m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-yoga https://zuul.opendev.org/t/openstack/build/70f3a8bfb85242a8b8116e2d632f61c3 : FAILURE in 46m 22s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/32d0af57cfaa475585d19d66457a8662 : FAILURE in 49m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/001129f66eb749aca86759b18ef6b7e4 : SUCCESS in 35m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/b9bda8cbc142431bb6ff9d4ec5e1a243 : SUCCESS in 39m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d5abaf55c6924c63b71c306cd2698f0c : SUCCESS in 1h 40m 06s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b065995f85904fcfbf896e5105af1d09 : FAILURE in 48m 55s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b02bc01f9de14613b3a1f57957aaa4e2 : FAILURE in 49m 56s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"efe6ba0f85c0576f7dab38d0a289b85ffa03f68f","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-11-29 13:21:49.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"6b3a081b87b26e7f99374449e511c51302952d27","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-29 13:48:48.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"66dad901a5e89210f165c2207e63904d0241bf11","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-12-02 16:46:57.000000000","message":"Patch Set 20: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"2fb04585c009644d19c5029693d9ae11465ddd7c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-12-02 16:47:39.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"10fff720551ebd9c0dad39d4744b0d9db49dc583","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-22 14:57:13.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"c288897d96a667d06d97e79257b92f09a92c31bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 17:39:54.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e25bba3c4e3d41c086392c9fc4b88af5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef185b501cfa4d048e2ab2f24c918afe : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db642b6f0cd4da5a4d815a5d9d05312 : SUCCESS in 2m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f7a1d99d2304e609023dc4b353b55d6 : FAILURE in 1h 41m 01s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/599852c774c74260b1f92db17ce9ceb1 : SUCCESS in 1h 52m 50s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/c0c3b4a91c50459fac07cbb62e7baebb : SUCCESS in 1h 13m 04s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/2a32eb523d5e44ba84e186af4eb6197f : FAILURE in 1h 31m 42s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/60c139bdfb234c049a4cd9d31b58e4c1 : FAILURE in 2h 23m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03737ff6574c41b2a0f56af2f8af972f : SUCCESS in 1h 57m 37s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/115a44fa5fbd405ba59744a87a2a3a4e : SUCCESS in 2h 33m 22s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/6cc30b530758447589d5306784992db0 : SUCCESS in 1h 41m 22s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/1ad367a6161e457087f498b14c62af23 : SUCCESS in 2h 02m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/34b31d5182c047d1a9d7fe4978c20294 : SUCCESS in 33m 40s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/462a856c76494cacaefc89ff20a960cf : SUCCESS in 42m 07s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/0a55436309b54a1ab47455d71d4ea3de : SUCCESS in 33m 50s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/1594a7ff918546a484afa7457b3e0fd8 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c34d2f8ea02546b481df772907ed9a75 : SUCCESS in 1h 45m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfe3426e8faf4913b88343d81663b84b : FAILURE in 12m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a42076cad124a42ac5dfb0b658aabcb : TIMED_OUT in 2h 34m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c8c5caa0840446b498b05c6dce49d8e4 : SUCCESS in 2h 20m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49070348ea334e5fbaeba7e5d4f9b872 : SUCCESS in 31m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/c53912d9516540ce9582a8da882ec1e6 : SUCCESS in 33m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/2ff2c570db924e1a9532c696dd38850c : SUCCESS in 39m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/e6b7ca53fca7485d9f05ca67a678fb3e : SUCCESS in 40m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/0ef636f2f2e64371b35e03f988a32c35 : SUCCESS in 36m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9a294f03a2014ae9b4b29b3e038e3f61 : SUCCESS in 1h 45m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/398628b5a30946d4933704414c231d13 : FAILURE in 44m 34s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/336446af39644453a64695384715e161 : FAILURE in 48m 23s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"21df6e0a5da846a133d1bea84cb8057a0d25dc3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-22 17:46:24.000000000","message":"Patch Set 21: Commit message was updated.","accounts_in_message":[],"_revision_number":21},{"id":"d51aeee4701c3315a592f24d04521d96089c058e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 20:24:24.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d3e91037c734fa28f955ba61b411209\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac349b3b2cb47eda500dafffb1e7cb4 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39264c86a1941afb6f0fb7a1c62ff8e : SUCCESS in 2m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a232f0b7faf4709aea84788f544b78b : SUCCESS in 1h 22m 28s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/cca80f55a9814d0ca356ecc52baffc7b : SUCCESS in 1h 11m 28s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/80803af4a4c948618286c24b68a6c392 : SUCCESS in 1h 50m 02s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/8eb3b532b1a64c9187eeef08a914a3fe : FAILURE in 1h 14m 51s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/bee966b590ef4c549f90704df4c0899d : SUCCESS in 1h 22m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d0749259f5d4bf487153ab95877eca3 : SUCCESS in 1h 59m 23s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/0f52a92b868242008b563e1b66103351 : SUCCESS in 1h 50m 35s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/d26aa2ab184a41628c43efd35dbfcf10 : SUCCESS in 1h 52m 36s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/4553a1df7be24eab86d451480b91d1cc : SUCCESS in 1h 29m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5bc76996479487c81ee97ad47c46695 : SUCCESS in 37m 00s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/488a833a2beb487189147bb9875509ba : SUCCESS in 36m 23s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/8313ca2486644ee481d1637eb65810ab : SUCCESS in 34m 53s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/511c4b86659e49e4a5dd3c2dd00a2416 : SUCCESS in 33m 20s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e931e94605c1473999d0cebb609e1fce : SUCCESS in 1h 58m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/6dd89f2a36a44a3090cb058b9b8f5cb0 : FAILURE in 11m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ab0f72b88aa44f3bff3c58f271bc73f : SUCCESS in 2h 27m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/993434fa2ff247dbbe8a1c4358d38737 : SUCCESS in 2h 30m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fc86c0abcfb499ca089cd3fd80cfbab : SUCCESS in 37m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/5b9c373600dc4dd3aee36ecb4650b122 : SUCCESS in 36m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/4a03830e5ad64a8c94f4923ef5b5c73b : SUCCESS in 39m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/9d172632ed4b44be9019fe8e6faf3f37 : SUCCESS in 37m 33s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/4b92a308a65a4881b6e307a7d85afb98 : SUCCESS in 37m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6aeb6f5ef896449aa705e275bb72e3f0 : SUCCESS in 1h 59m 10s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/c5e0469ee768411f9da339860f5f9345 : FAILURE in 48m 54s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c7b376c62f684797a2542c8512a4a574 : FAILURE in 53m 03s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"c58f984d4ebffbc0635f9096f801fc95a302caf3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-23 08:09:04.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"096ff0588a10936576cbc3f658a31ce6d978986e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 11:31:58.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/148f19ce9b52420b80a2bff2cbc65399\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526f7a08657e442d8eb6c70aa539f5a1 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/747b58f40f4f4cfa9e59347524e59ec3 : FAILURE in 2m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0187e89bfcc44eeeaf5570440ca07651 : SUCCESS in 1h 33m 42s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/9bd07b62791d484fbf92d4abdf466403 : SUCCESS in 1h 11m 40s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/0b8f1d3bfda94735949af151cc7e68b5 : SUCCESS in 1h 26m 38s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/8eb2def0bcce4843bf49aae523450634 : FAILURE in 57m 34s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/aa188d580824492387f3844b10d324a9 : SUCCESS in 2h 20m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b91417a350048338375bc5b86817f52 : SUCCESS in 1h 53m 53s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/2bd2a494241b4eeaaf250e4bf2373e20 : SUCCESS in 2h 07m 20s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/2904e533437c40d0b1c8037741720899 : SUCCESS in 2h 05m 58s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/020fab53495b4e5d9542559affbe2acb : SUCCESS in 1h 44m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b5dc343535e4c2ba0c14fb41992cf07 : SUCCESS in 41m 43s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/de87b63747f8422482ca3573721c23fb : SUCCESS in 37m 07s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/cb8657f75f4644668989aff72deb4130 : SUCCESS in 42m 34s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/d736e714ef714db08419163e11b20f6c : SUCCESS in 39m 50s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5233a83c88ad42b384cdd11c7eecfcda : SUCCESS in 1h 50m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/54a53613e7ec4f01bcfdcd57e2579d47 : FAILURE in 8m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/75b0d40f8f1b478aabdc51cb7c8680f8 : SUCCESS in 2h 26m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c16b511d56a9484cb31dc0d74b7ad2cc : TIMED_OUT in 3h 04m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f093a999b8e549458c5ca6a5da7e080b : SUCCESS in 37m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/c61c3d1d512e4a64b32c17206847e2e3 : SUCCESS in 37m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/3897cad7316c472498970f9781aaff77 : SUCCESS in 33m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/42425f4890e5401eae88e8f100a3182c : SUCCESS in 37m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/813dee600d5647b7819609f9fcd72b9e : SUCCESS in 37m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/446f27831f76471a9fc42cb9aec0fdc7 : SUCCESS in 2h 19m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/9dc165abe8e44fa89d5cc8a6469f4042 : FAILURE in 49m 57s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5fa13e4c06c94ba993f2ae8bbe1488f5 : FAILURE in 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"4e34cfb80123af61b9bc280a3abc157121e0d6dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-29 13:44:36.000000000","message":"Patch Set 22: Commit message was updated.","accounts_in_message":[],"_revision_number":22},{"id":"b3e30bb8108f09486d296230c81e65c31f8ea807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-29 15:39:33.000000000","message":"Patch Set 23: Commit message was updated.","accounts_in_message":[],"_revision_number":23},{"id":"fc0639e9168a80737a4f0d30febb345817783555","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 18:14:23.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44aaf527d17944c5910480b4fc877d43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824e262e1783481ca866cbfa9149afe5 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a1830216fc45b48b3ce343afbe0f7d : SUCCESS in 3m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d2cfc65a6aa488ba5bd1c0836ed8c75 : SUCCESS in 1h 11m 03s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/0cbd1106f30c4750a0eebd30bf0a8627 : SUCCESS in 1h 12m 14s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/63ac7a43e93f4c96bc3cf8682ae8221b : SUCCESS in 1h 03m 31s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/34114cf547ad40af828d785202cf3684 : FAILURE in 1h 22m 44s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/c0b66fc8ca314ecdbd15ce2fa0516f8f : SUCCESS in 1h 19m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2131992a05224b45a92b4e0b5248199a : SUCCESS in 1h 49m 52s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/cc5c57a0687d428db8cd57358ead5100 : SUCCESS in 1h 43m 17s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/8fef50aed86742ea89b4cb3f74e49b5b : SUCCESS in 1h 33m 54s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/49023d9940f94c6f8469b0e8ae62538c : SUCCESS in 1h 33m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2014b1ba20b541f3921a43d328ecf4a2 : SUCCESS in 37m 48s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/d1aba5b2959e46b9b52a4af2a4ef8d6f : SUCCESS in 36m 00s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/c21da7b7b35d48b7b17e7679f4bdd278 : SUCCESS in 34m 33s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/cb6bda8a64604a21bab607236c0ac5ca : SUCCESS in 33m 33s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0d5ba86f33e6441bade41d4612851c5c : SUCCESS in 2h 22m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b552da4d00f446c801da0d1ceae6458 : FAILURE in 9m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc93fb2d96c741d29a104e722cab063f : SUCCESS in 2h 28m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6904fa105eb14528a9f51a5fa54b2f05 : SUCCESS in 2h 31m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39d54ad8c3aa4b5399905650843ee3d1 : SUCCESS in 30m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/fd23b892ed1943b5978d212d3a581c93 : SUCCESS in 31m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/6c3bc48b4bcb414d82574cb1acf90a19 : SUCCESS in 29m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/5f37d580351c4e6b9087c1c6fe19fa76 : SUCCESS in 39m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/d42e8fa7ff27444fbd1903a61e01b241 : SUCCESS in 30m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/31bccd16d2a846fea1bd1913d6811fbd : TIMED_OUT in 2h 33m 50s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/55e166ecfaac486abf2f9c3addfb2e0b : FAILURE in 37m 38s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/39dd8fe633194eed81160e8c0dda443a : FAILURE in 43m 00s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"014e7ecc18a081198b8c2844fb2ac4acc98024e5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 10:27:13.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"c686b6fd97e79c2dc1b3dc2f165faf45d9a8bdf6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 11:02:38.000000000","message":"Patch Set 23: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"2095f44069f26c426305e83845ce22df7ab339c1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-12 12:30:22.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"e6a7b243877db9820e776242f296c79c6e78dab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:34:03.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca0a257ae40442ea8cc0e6cbcd9b332a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d110882c9e34fa1be25a5229a14dd5c : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e041c7fcfa38453fad31a84b2efcd7f5 : SUCCESS in 3m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/362e3f3d9ed7467fbbc7a38907dd4af5 : SUCCESS in 1h 41m 54s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/e4e9e91294c6496bb181b2f17ab8568e : SUCCESS in 1h 57m 50s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/76731bcb0e644eb2a973dea932ae1102 : SUCCESS in 1h 32m 32s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/c70347ab34a84bcebf33d41ca9278cda : SUCCESS in 1h 11m 37s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/b05cdc2534f4498bb4ee7b972f0e42b6 : SUCCESS in 2h 19m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04abecbc820547a99ff3773580166104 : SUCCESS in 2h 07m 17s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/086c78700bee42208101fe292fc3dea9 : SUCCESS in 1h 47m 49s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/8fb7b04d21224f27a6408951a84eff9c : SUCCESS in 1h 41m 10s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/ea0df293bae14caca1acee7252308682 : SUCCESS in 1h 32m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1abce4650d044287aceb03ba76a3c289 : SUCCESS in 34m 35s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/31117a0e5b184b56a259d19fafd40f40 : SUCCESS in 37m 02s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/c4c0a5e0e375408491a6f29f3c2111f9 : SUCCESS in 35m 35s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/7e7150041c2d49eab6adcceae9ad8c36 : SUCCESS in 38m 36s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b8761291279548239343257ae2e02fc7 : FAILURE in 16m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/44072d859db64af098ecd00b97c0d6fd : FAILURE in 10m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbdf0998c6994328a8fd9abacd141337 : SUCCESS in 2h 14m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d5f7e1c5ddf42349d9326f278e2f772 : TIMED_OUT in 3h 03m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/911f3fd5a76f410aa409a276aa4e3eab : SUCCESS in 35m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/90d77901625444e88b0ea58b59a2eb43 : SUCCESS in 38m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/176d00a4955b4a758a703930f05504d8 : SUCCESS in 37m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/5655174e43cd477dad14ef1e4956a53f : SUCCESS in 36m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/984e218ae506438c8a3dfeb64e0cfd89 : SUCCESS in 34m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/359eb8f7825e423c86b4209d8fb6677e : SUCCESS in 2h 01m 19s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6f5beeac25ee4d4a83f4c52366be62a4 : FAILURE in 52m 41s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/cb936dbaa6094690a7d143e309f808b8 : FAILURE in 1h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"3354d74c7b7877e0ba87b8dee60da3d63205a494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-20 13:38:58.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"75dc28c2cdb03a3f7bfcfc630fb41530a6475d78","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-20 15:43:53.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"1e53707ddb5c836fb3eff2158fe1aa60c2e202cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 16:44:59.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89f60d94ae22485db6f5b898cc713dd0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6455e2712c3245dd8fc627af3a165dd5 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4b9518553546448ef4b963a9ff1e7f : SUCCESS in 2m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20bdb04cad4d4e99ad7de1f2539a105e : FAILURE in 1h 24m 08s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/42f79ab619734cfca0bb8aa214343000 : SUCCESS in 55m 21s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/ec97c03f6fe148ad8e451da0cb72ad8d : SUCCESS in 1h 10m 28s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/aceee0892c37477fac0817f15a63f68e : SUCCESS in 1h 09m 09s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/42bc0054a8fd4cf3a3585d6caa8fd0b3 : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/db53713d89e74f668bd66b861dae097e : SUCCESS in 1h 37m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a5fcf64c235e455f962175ea1c0746ef : SUCCESS in 1h 01m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/84a176b2c58e421da7030491883b85ff : SUCCESS in 1h 16m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/608ca2a82fc4439181a09249ca166bec : SUCCESS in 1h 07m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/ae21fb03124e44928a56abf80e6190e6 : SUCCESS in 1h 16m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/30a9d1e2e28b4d109f2dd6b14b79743d : SUCCESS in 1h 05m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/b869b41077c9413b8c2063dbb85c41a1 : SUCCESS in 1h 18m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/7e861741c97b4a3ba90b6616bcfbd1cf : SUCCESS in 58m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af8c81c40a2f4d3ab3dbe4d116b3aa75 : SUCCESS in 37m 34s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cb80934664514f84ab033cc79e630f48 : SUCCESS in 36m 43s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/eb2493bda7f244958ab1f95409d302ec : SUCCESS in 49m 03s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/fc53340c503543e6b9a33af393aedca8 : SUCCESS in 47m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/026028a506a541f28e2848eb2810cc6d : SUCCESS in 2h 29m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b646d62db8de414ea89146baa3dc7ae7 : SUCCESS in 2h 04m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/62fd78b38a004e419d964e6939523eb9 : SUCCESS in 1h 32m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df74e60d315545b0844188b2fa710587 : TIMED_OUT in 3h 03m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d3ef5a5b4a740a78dfebee491921ba9 : SUCCESS in 43m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/15423f849de7403186182b1274268136 : SUCCESS in 43m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/932ce3ec3eb840eda4b7b8f4a6673997 : SUCCESS in 43m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/af20ccf985b54955a14ec783f58ba164 : SUCCESS in 41m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/06852b1328f7428689dcade0b891fb8a : TIMED_OUT in 2h 33m 50s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/82b118e95bf7471799d9a46e28255319 : FAILURE in 48m 28s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/21343ff7c93a4defbcbb60bb2299ae14 : FAILURE in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"24ef3c6221caf0424dbd89eac4834cc39a3dc42c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-29 15:04:31.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"cc92a16e3ae0f0b2a6af2c380ebfb34b4a5a84f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 17:29:24.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a867685ea634f2b9a1958ea167e7b7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4d9943d85a445d8b1c7395b8cc9c13 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2362fdb5d25d4e45a082c489ddeeef2c : SUCCESS in 3m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/14285a76d8bc465a97fd496ebaf316ad : SUCCESS in 1h 33m 46s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/85d2a85b2e694b1cbe3ab4307076e278 : SUCCESS in 1h 26m 42s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/c530cffee2a44a4597b9faf6d6afaae6 : SUCCESS in 1h 22m 45s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/80bb74196719461c896a723a60668637 : SUCCESS in 1h 36m 13s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a7abe10d4d2a403395a9013ddc7e06d4 : SUCCESS in 1h 21m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a98334c06e7f4456b52e67719fc922db : SUCCESS in 2h 19m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/93d7bd0aac1042a3adf659f84d1263f3 : FAILURE in 1h 25m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/447dc8fbfd004ca5b6db40c518c8d968 : SUCCESS in 1h 38m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bdd816b2833e46b3a1103e1dc669de82 : SUCCESS in 57m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/a2029a00179d44ac92320ca080421054 : SUCCESS in 1h 15m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/ba45258b3ff84342a80af54b6e6b3a28 : SUCCESS in 56m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/53e6e41cdc394ac099a2108dfc9f6201 : SUCCESS in 1h 42m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/ab6b6007549941c798d7ffd88eb96e20 : SUCCESS in 1h 29m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/63128bb8ac2244108ec5747f93a4d96b : SUCCESS in 56m 25s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/384a534e6e78432ca73bdd0c2e9c7bd9 : SUCCESS in 35m 04s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/e9e0d3f1a8ae45568329cab4ffa0879c : SUCCESS in 34m 32s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/0f07c25039a9402ab040c1ee8564a3d9 : SUCCESS in 53m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c8bdfb8858724e25987070d29c739382 : FAILURE in 22m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/a7f458095e51432790b83c8a3a19a900 : SUCCESS in 2h 16m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4c05fc4271014b81a38627b8e77addc4 : SUCCESS in 1h 34m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba6013d1cee04f0a82920016e05c8fee : FAILURE in 1h 51m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d898f9862574ec98cec515d6e0c607a : SUCCESS in 33m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/5e214af4c23b43afb7d02eb74cc9fbea : SUCCESS in 49m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/96d10a9cca134f4caf94b3dce15b872a : SUCCESS in 31m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/b3272f5afad9487f9e55c4b2a706dcc2 : SUCCESS in 32m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/da8243386606449ea2e15a73fe32906c : SUCCESS in 2h 00m 30s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/add259e5321c464fa21d63e54d54c12f : FAILURE in 51m 11s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/86a4cb8707734426b90b15f91a17ff02 : FAILURE in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"852b58210742e67970fe9db71e47cc327b05a130","revisions":{"ddc9cfb07c0b2668726d447084c91236bc62314f":{"kind":"REWORK","_number":1,"created":"2022-01-17 17:28:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/1"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 17:28:23.000000000","tz":60},"subject":"WIP","message":"WIP\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ddc9cfb07c0b2668726d447084c91236bc62314f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ddc9cfb07c0b2668726d447084c91236bc62314f"}]},"branch":"refs/heads/master"},"396f860bc4b5f653f640c82f9b4c4b586440574a":{"kind":"REWORK","_number":2,"created":"2022-01-18 15:06:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/2"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-18 15:06:16.000000000","tz":60},"subject":"WIP","message":"WIP\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/396f860bc4b5f653f640c82f9b4c4b586440574a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/396f860bc4b5f653f640c82f9b4c4b586440574a"}]},"branch":"refs/heads/master"},"3a79d41127dc0e044fbe0502c8dbd2778453cff3":{"kind":"NO_CHANGE","_number":3,"created":"2022-01-18 15:08:15.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/3"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-18 15:08:06.000000000","tz":60},"subject":"WIP","message":"WIP\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3a79d41127dc0e044fbe0502c8dbd2778453cff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3a79d41127dc0e044fbe0502c8dbd2778453cff3"}]},"branch":"refs/heads/master"},"1287dcb3c1e396c7224fb2ae8b66605bcc765588":{"kind":"REWORK","_number":4,"created":"2022-01-18 15:48:04.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/4"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-18 15:16:17.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1287dcb3c1e396c7224fb2ae8b66605bcc765588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1287dcb3c1e396c7224fb2ae8b66605bcc765588"}]},"branch":"refs/heads/master"},"40662a0da4358d44a7f96f9ca6e9a9e2043d917b":{"kind":"REWORK","_number":5,"created":"2022-01-19 09:20:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/5"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-19 09:19:16.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/40662a0da4358d44a7f96f9ca6e9a9e2043d917b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/40662a0da4358d44a7f96f9ca6e9a9e2043d917b"}]},"branch":"refs/heads/master"},"e6f55b10f9b47730ee46ac3cea53706ceb3697b5":{"kind":"REWORK","_number":6,"created":"2022-01-20 10:05:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/6"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-20 10:05:39.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e6f55b10f9b47730ee46ac3cea53706ceb3697b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e6f55b10f9b47730ee46ac3cea53706ceb3697b5"}]},"branch":"refs/heads/master"},"8d69b380e62cb97202e5e3ed67fc3aee2f7f2c05":{"kind":"REWORK","_number":7,"created":"2022-01-20 10:55:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/7"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-20 10:54:54.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8d69b380e62cb97202e5e3ed67fc3aee2f7f2c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8d69b380e62cb97202e5e3ed67fc3aee2f7f2c05"}]},"branch":"refs/heads/master"},"5a919cecfc248e3c52f7ddc45b6752f59e5e898f":{"kind":"REWORK","_number":8,"created":"2022-01-20 11:12:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/8"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-20 11:11:54.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5a919cecfc248e3c52f7ddc45b6752f59e5e898f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5a919cecfc248e3c52f7ddc45b6752f59e5e898f"}]},"branch":"refs/heads/master"},"ed2771e598788587f36e521e423eec98881d2160":{"kind":"REWORK","_number":9,"created":"2022-01-20 11:39:05.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/9"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-20 11:38:56.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ed2771e598788587f36e521e423eec98881d2160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ed2771e598788587f36e521e423eec98881d2160"}]},"branch":"refs/heads/master"},"634d09cf77461e803bb3d86a4d4ba87ab5bb541a":{"kind":"REWORK","_number":10,"created":"2022-01-21 15:23:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/10"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-21 15:23:09.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/634d09cf77461e803bb3d86a4d4ba87ab5bb541a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/634d09cf77461e803bb3d86a4d4ba87ab5bb541a"}]},"branch":"refs/heads/master"},"3d906445be56c8dadc5e527dfa2b2022706405e8":{"kind":"REWORK","_number":11,"created":"2022-01-24 09:54:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/11"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-24 09:54:40.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3d906445be56c8dadc5e527dfa2b2022706405e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3d906445be56c8dadc5e527dfa2b2022706405e8"}]},"branch":"refs/heads/master"},"a2ad1e3ab457ed6a0dca7b8a4d23026a150f7ae3":{"kind":"REWORK","_number":12,"created":"2022-01-24 15:12:37.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/12"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-24 15:11:41.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a2ad1e3ab457ed6a0dca7b8a4d23026a150f7ae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a2ad1e3ab457ed6a0dca7b8a4d23026a150f7ae3"}]},"branch":"refs/heads/master"},"68eb64e83d92f4c722920603637c6871ac210862":{"kind":"REWORK","_number":13,"created":"2022-01-31 10:55:08.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/13"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-31 10:54:55.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/68eb64e83d92f4c722920603637c6871ac210862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/68eb64e83d92f4c722920603637c6871ac210862"}]},"branch":"refs/heads/master"},"586f979ee10d2d51abcee7fee08974f7b116025e":{"kind":"REWORK","_number":14,"created":"2022-02-03 15:01:20.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/14"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-03 14:38:54.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/586f979ee10d2d51abcee7fee08974f7b116025e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/586f979ee10d2d51abcee7fee08974f7b116025e"}]},"branch":"refs/heads/master"},"7d590ecff023f392bc62f4b672d1cce949422e16":{"kind":"REWORK","_number":15,"created":"2022-02-04 09:34:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/15"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-04 09:32:40.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7d590ecff023f392bc62f4b672d1cce949422e16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7d590ecff023f392bc62f4b672d1cce949422e16"}]},"branch":"refs/heads/master"},"9fd9ddaf8ebac1f577f53fbbab9bb958f217db55":{"kind":"REWORK","_number":16,"created":"2022-02-11 10:22:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/16"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-11 10:21:16.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9fd9ddaf8ebac1f577f53fbbab9bb958f217db55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9fd9ddaf8ebac1f577f53fbbab9bb958f217db55"}]},"branch":"refs/heads/master"},"ebe073d9b1600ecc77ae91d640c42fdd7a12bbe1":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2022-02-11 16:09:47.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/17"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-11 16:09:30.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/828606\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebe073d9b1600ecc77ae91d640c42fdd7a12bbe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ebe073d9b1600ecc77ae91d640c42fdd7a12bbe1"}]},"branch":"refs/heads/master"},"1422bef2f3f87e63904f32c5f5594328d929c16f":{"kind":"REWORK","_number":18,"created":"2022-02-14 07:28:06.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/18"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-14 07:27:57.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/828606\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1422bef2f3f87e63904f32c5f5594328d929c16f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1422bef2f3f87e63904f32c5f5594328d929c16f"}]},"branch":"refs/heads/master"},"e5caa69e71fb66993414ada2994bb6e947a42f0b":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2022-02-21 16:05:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/19"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-21 16:05:06.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/829805\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/828606\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e5caa69e71fb66993414ada2994bb6e947a42f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e5caa69e71fb66993414ada2994bb6e947a42f0b"}]},"branch":"refs/heads/master"},"f5704f47e435ed12d80b7472774351239f51f774":{"kind":"REWORK","_number":20,"created":"2022-05-06 14:15:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/20"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-06 14:13:35.000000000","tz":120},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nThe new test cases cover the load balancer API that allows to\ncreate a fully populated loadbalancer in one single request.\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f5704f47e435ed12d80b7472774351239f51f774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f5704f47e435ed12d80b7472774351239f51f774"}]},"branch":"refs/heads/master"},"ccaf24c1e23ba3f507e8f04db6711fa6f8be6672":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2022-12-22 17:46:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/21"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-22 17:46:24.000000000","tz":0},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nThe new test cases cover the load balancer API that allows to\ncreate a fully populated loadbalancer in one single request.\n\nDepends-On: I98cce51a2536bcf204e5753ce4984af57e7e69e3\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ccaf24c1e23ba3f507e8f04db6711fa6f8be6672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ccaf24c1e23ba3f507e8f04db6711fa6f8be6672"}]},"branch":"refs/heads/master","description":"Edit commit message"},"15c8d5e583b3770ff4af280e941a3a308ff9a85d":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2022-12-29 13:44:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/22"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-29 13:44:36.000000000","tz":0},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nThe new test cases cover the load balancer API that allows to\ncreate a fully populated loadbalancer in one single request.\n\nDepends-On: I98cce51a2536bcf204e5753ce4984af57e7e69e3\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/868527\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/15c8d5e583b3770ff4af280e941a3a308ff9a85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/15c8d5e583b3770ff4af280e941a3a308ff9a85d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"00d0bf91c67400a2b47c1017daa9fc8563d5a8b9":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2022-12-29 15:39:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/23"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-29 15:39:33.000000000","tz":0},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nThe new test cases cover the load balancer API that allows to\ncreate a fully populated loadbalancer in one single request.\n\nDepends-On: I98cce51a2536bcf204e5753ce4984af57e7e69e3\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/868870/\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/00d0bf91c67400a2b47c1017daa9fc8563d5a8b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/00d0bf91c67400a2b47c1017daa9fc8563d5a8b9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"852b58210742e67970fe9db71e47cc327b05a130":{"kind":"REWORK","_number":24,"created":"2023-11-20 13:38:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/99/824999/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/99/824999/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/99/824999/24"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-17 16:44:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-11-20 13:38:30.000000000","tz":60},"subject":"Add tests for creating a fully populated LB","message":"Add tests for creating a fully populated LB\n\nThe new test cases cover the load balancer API that allows to\ncreate a fully populated loadbalancer in one single request.\n\nChange-Id: I43bff02cf318a30f20c74fbabe09abc005f36661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/852b58210742e67970fe9db71e47cc327b05a130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/852b58210742e67970fe9db71e47cc327b05a130"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
