)]}'
{"id":"openstack%2Foctavia-tempest-plugin~825208","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ia1ecf95d269af2f6edbd562f31fb69c2cc884cf0","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"xena-stable-jobs","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-02-07 10:58:45.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-01-27 20:14:49.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"Ia1ecf95d269af2f6edbd562f31fb69c2cc884cf0","subject":"Add stable/xena jobs on master gate","status":"MERGED","created":"2022-01-19 01:43:55.000000000","updated":"2022-02-07 11:04:17.000000000","submitted":"2022-02-07 10:58:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825208-xena-stable-jobs","meta_rev_id":"ab6fbdc3ac88d2b65340ff4d4f48b1958989b5e8","_number":825208,"virtual_id_number":825208,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-07 10:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2022-02-03 12:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-27 20:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2022-02-03 12:44:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 01:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-26 07:05:51.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-01-27 20:14:49.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-02-03 12:44:42.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"9aef998c4b9b2964704f828a11c0767449385919","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-19 01:43:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd9ad83244920356bae44561ba248b0605df9473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 01:44:54.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/barbican on branch stable/train.  The error was:\n\n  Job octavia-v2-dsvm-tls-barbican-stable-train not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - barbican-simple-crypto-devstack-tempest\n          - barbican-simple-crypto-devstack-functional\n          - barbican-simple-crypto-devstack-functional-postgres\n          - barbican-dogtag-devstack-functional-fedora-latest\n          - barbican-vault-devstack-functional\n          # TripleO jobs that deploy Barbican.\n          # Note we don\u0027t use a project-template here, so it\u0027s easier\n          # to disable voting on one specific job if things go wrong.\n          # tripleo-ci-centos-7-scenario002-standalone will\n          # run in Pike and beyond.\n          # If you need any support to debug these jobs in case of\n          # failures, please reach us on #tripleo IRC channel.\n          - tripleo-ci-centos-7-scenario002-standalone:\n              voting: false\n          - octavia-v2-dsvm-tls-barbican-stable-train:\n              voting: false\n      gate:\n        queue: barbican\n        jobs:\n          - barbican-simple-crypto-devstack-tempest\n          - barbican-simple-crypto-devstack-functional\n          # TODO(redrobot): add this back once job is voting again\n          #- barbican-dogtag-devstack-functional-fedora-latest\n          - barbican-vault-devstack-functional\n      experimental:\n        jobs:\n          - barbican-kmip-devstack-functional\n          - barbican-grenade-devstack\n\n  in \"openstack/barbican/.zuul.yaml@stable/train\", line 148, column 3","accounts_in_message":[],"_revision_number":1},{"id":"8391ca8d49aad950bfb2b945fef35206d18baad3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-26 07:05:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32b2c351def0bdef631985836c07f8ca8b8fec78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 07:06:19.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/barbican on branch stable/train.  The error was:\n\n  Job octavia-v2-dsvm-tls-barbican-stable-train not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - barbican-simple-crypto-devstack-tempest\n          - barbican-simple-crypto-devstack-functional\n          - barbican-simple-crypto-devstack-functional-postgres\n          - barbican-dogtag-devstack-functional-fedora-latest\n          - barbican-vault-devstack-functional\n          # TripleO jobs that deploy Barbican.\n          # Note we don\u0027t use a project-template here, so it\u0027s easier\n          # to disable voting on one specific job if things go wrong.\n          # tripleo-ci-centos-7-scenario002-standalone will\n          # run in Pike and beyond.\n          # If you need any support to debug these jobs in case of\n          # failures, please reach us on #tripleo IRC channel.\n          - tripleo-ci-centos-7-scenario002-standalone:\n              voting: false\n          - octavia-v2-dsvm-tls-barbican-stable-train:\n              voting: false\n      gate:\n        queue: barbican\n        jobs:\n          - barbican-simple-crypto-devstack-tempest\n          - barbican-simple-crypto-devstack-functional\n          # TODO(redrobot): add this back once job is voting again\n          #- barbican-dogtag-devstack-functional-fedora-latest\n          - barbican-vault-devstack-functional\n      experimental:\n        jobs:\n          - barbican-kmip-devstack-functional\n          - barbican-grenade-devstack\n\n  in \"openstack/barbican/.zuul.yaml@stable/train\", line 148, column 3","accounts_in_message":[],"_revision_number":1},{"id":"4e02f3dd14ee0d2b8fdff383e4ff92501fa8bb45","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-26 07:08:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d23049dea8a40647e26de75bdd57fac388516dc9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-27 17:36:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffd606f695cf218215a618cc23a2d83c071ad5fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-27 17:37:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"83650a19b39f084f1e28eb2b04b19a409afc72e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 20:00:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e14ffa287e704b95ae456aa05bd24e90 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5082a3ab6cf4d91b221213028474546 : SUCCESS in 4m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f0242347af44808a92aad0ebb365c15 : SUCCESS in 2h 11m 06s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/0021c68c856648f0a9b819690cf2f18b : SUCCESS in 1h 52m 23s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/0ded15cb176b4e81abc3dfc1573d65e9 : SUCCESS in 1h 23m 20s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/9461e88191d246a9ba4967e122c808ab : SUCCESS in 1h 40m 16s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/23d571f52f644d07b26565a86472d73b : SUCCESS in 1h 05m 33s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/539379e6a9d24fb09f651abe0b4273e1 : SUCCESS in 2h 03m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e7c309829b745e0a140548d78d6f8fa : FAILURE in 1h 13m 53s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/296bcdba2b9847abb9e3528918578238 : SUCCESS in 1h 23m 51s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/69a5ef7db69e4b08ad6d4edec0d8137c : SUCCESS in 1h 20m 18s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/9584cc7522e9497a98bf3643b3273c97 : SUCCESS in 1h 27m 52s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ec46bdbb258f4bff9dd19376644debf8 : SUCCESS in 1h 26m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4449a7fc9db447509966df144d83a4dc : SUCCESS in 38m 16s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/5d18d88d71d7425d9de2d1a63a15e946 : SUCCESS in 37m 41s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/8ebbf97eab1b4755b7f8014d2b2471ba : SUCCESS in 36m 26s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/ece1400eb08a4722a764f99ac15533c4 : SUCCESS in 35m 30s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b5eeb4d3be454ce3a91bfc28dd73e359 : SUCCESS in 36m 51s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b562284feb4349a1befebde9edcdcaaf : SUCCESS in 1h 28m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/cdba2330545c45bb83f5ee7ed6589db4 : SUCCESS in 1h 45m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/196c00876d764efdb33ef371b93e1189 : SUCCESS in 2h 10m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c43569b6d76546608929049389377b52 : SUCCESS in 36m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/b8cc51fd1625462c9f50f275d7a9f269 : SUCCESS in 36m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/3f9a212ec5b944338d673cb2a4d0d66a : SUCCESS in 35m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/72685b59ed854e5ca6330d6cfe0d1939 : SUCCESS in 33m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/77b697e5655347269c4c44db34d4e160 : SUCCESS in 34m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a3bca4704dc8473db7de9522ed38c399 : FAILURE in 45m 01s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/2731e4afdf1d4c5a8b084216a9071edc : FAILURE in 47m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/2d6ec6bb38434f0bac1e47faf8c7c6c3 : SUCCESS in 32m 07s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/3570b90960964c629647fbc9c13ae228 : SUCCESS in 33m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/f5456b5c3a4d47288fae8e5a164dfd76 : SUCCESS in 34m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/08238ae4ae154154b7693b90bc174369 : SUCCESS in 1h 23m 49s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/69699ec1575a4dfd80558e7587c5d26f : SUCCESS in 41m 29s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/89fcf63d3f2f4af68141af32168e16b2 : SUCCESS in 50m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5ae7fb7a2e5eca1766981126f8f6525c090459e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-27 20:14:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"206eaee7f5b469961cd1d5c53e9eacae0a72f822","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-28 06:38:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b13a5b7eb3cc09caf07bc79e56a94d5eb755336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 09:32:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47cd04351a754f45b4d3ac140f58b809 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3474bfb63624052b1122ab6f69bf62e : SUCCESS in 5m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/538e994a7a144e71baaf1669c1f8ca5d : SUCCESS in 2h 09m 46s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/fc22d63ca3e24035a145dc280b076a31 : SUCCESS in 2h 18m 17s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/552a0d0e46bf4e8b8edd4cffacd45d88 : SUCCESS in 1h 11m 46s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/9704d9e6630344a89b5dbd3a69fd1a48 : SUCCESS in 1h 40m 31s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/db512033e624401dba9645f56af3fedd : SUCCESS in 1h 08m 01s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/c8a8dfe4338c42e382bea87218ae2fca : SUCCESS in 2h 43m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8567f291887b43a99023bc4a07fdb17d : SUCCESS in 1h 29m 17s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/1c93ccf7780e4a269f0ed490613273ac : SUCCESS in 1h 30m 47s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/af9a66e8f9b44d65ba548f5046452f22 : SUCCESS in 1h 24m 26s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/99699c49f8744f9db07f2794113a2cd7 : SUCCESS in 1h 39m 35s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a0412eaed2fe4d84880904928c12f5a2 : SUCCESS in 1h 33m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b8cfdd19acc4161948f3aeff7b60788 : SUCCESS in 35m 36s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/1058c86495ab49588c59fac0f3644e4d : SUCCESS in 45m 12s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/68d5e163e73c41fcbdfc7d54ea2a0c68 : SUCCESS in 37m 38s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/0e9e926537a34725b2053aaf09e58535 : SUCCESS in 47m 53s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/2c599ea5bab547a6a361d03f1d53ba17 : SUCCESS in 36m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/577587b792514fc89c33599b260eef92 : SUCCESS in 1h 51m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e56c43d287e64597b2ddcefd89d94f3c : SUCCESS in 1h 45m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/539173fd8c2f4f90bc84e648eef8bfe5 : SUCCESS in 2h 49m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7ac0a1556804dd89bfc6e0320060dea : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/b91f650a404341a496804ffb33e9441d : SUCCESS in 36m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/b512b18dbdbf445f9503bc5a706b6b68 : SUCCESS in 36m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/c04f62472ae9432abc5da7b1220f3e22 : SUCCESS in 30m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/6cb6f950311e41f59fc4500625be9ea8 : SUCCESS in 32m 57s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/31a5e30ec98940df8a8129585bec6093 : FAILURE in 47m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/c4bdbb358d11437b96e5bb7e9d187ae0 : FAILURE in 41m 38s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/0fe5cf3f80374113848255d55350c250 : SUCCESS in 43m 27s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/669fa409861b444293574eea2c0ad0a3 : SUCCESS in 35m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/3a71e991de754796912ce50a7a2c762b : SUCCESS in 31m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80e803e1f0d545828aba05344eb0068a : SUCCESS in 1h 27m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5cdd2e4b62894d248cf041ed12de8003 : SUCCESS in 42m 26s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/1c34853dfe704bb598b693439b9bef6f : SUCCESS in 40m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1115e9bf4c67f3fedd6c50b07e3c3facf610ab58","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-03 12:44:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"17bb3a8df42bbe32ba2ac759e92e1d20d1860efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 12:45:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"566528426a70c650e93f4ab72bdfb8ac72b96b36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:17:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752c6eaeb2fa42cb941e92cea40bafa4 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19a7132fab84bd190316fb1e95fcf30 : SUCCESS in 3m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99ad87664cdf4261b01203f668606b69 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/55875454986240459b8b5be3d063839e : CANCELED\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/12ebb3d3eda64da495f70d4f3f50ed3c : CANCELED\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/57218a6666db49e68ea5d4e3fd6b1160 : SUCCESS in 1h 28m 35s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/7c7788d38964455d9a43f9a91ee9461c : SUCCESS in 1h 27m 14s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/7ddd350d43904a04b2d87608d3de3374 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46a10b232495447197ebc55386f7ecb1 : FAILURE in 1h 46m 21s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/57d5be708b3e43ceb657498765df3a02 : SUCCESS in 1h 16m 08s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/790e3003405d4f08b26047e6768708bb : SUCCESS in 1h 14m 39s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/48333c9a4f184e9f909cc535910aab70 : SUCCESS in 1h 23m 00s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/73b87c4ecab443769fe3536171cf3fc9 : SUCCESS in 1h 40m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e535c0eb699041eb92dd5d594c26c0a3 : SUCCESS in 37m 20s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/e8cc82a2e9814eaf801c3d94c7afd19b : SUCCESS in 35m 58s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/5f9a5c9717d8430a809fe249cd5885b6 : SUCCESS in 50m 40s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/f6dfb2685a3a492f9d4991932a0d005d : SUCCESS in 41m 39s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/a9973989f0c84ff390ddd0c20928cc79 : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1e565d8df7106b045eb224ccb830dff6b4dd3ffd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-03 15:22:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8669e8635a5a03693e16e1e7631dbd58e7bfa37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5c2a63afcd4e34918f7a7ea5c30851 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/831c82763ccf484291deb3c2020624bb : SUCCESS in 5m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/adad516f08d74475ac05eb61495b72ee : SUCCESS in 2h 29m 04s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/bd89ecacaa7040c9af1d4f7c48eb8753 : SUCCESS in 2h 17m 24s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/e30ac209dff34e4e926c857d65bec7b5 : SUCCESS in 1h 31m 32s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/f83bb85774b04a2a8ce7e450924c7dd4 : SUCCESS in 1h 34m 28s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/fe935ae6648346a39de9f1f347cb3e87 : SUCCESS in 1h 04m 59s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/7fb30c36d31f48999bc0ae93de585986 : SUCCESS in 2h 35m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aef0cab249e94a258025dcb21fa8ef0a : SUCCESS in 1h 36m 50s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/14a1dccc10324dfc8aaeddaedb32fe2b : SUCCESS in 1h 42m 39s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/2c45618036e0466591fe774dab311117 : SUCCESS in 1h 12m 45s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/01e140f0393c44149b555b174a30ba51 : SUCCESS in 1h 27m 59s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/34e19007e73140c29ebe34f89014f791 : SUCCESS in 1h 41m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9958c5e656504f9788620c5c519587fc : SUCCESS in 35m 32s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/94504aeb774b4f1b9beca424f0f08774 : SUCCESS in 38m 16s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/e039b4d4bc39429ebc86808d0a419a84 : SUCCESS in 37m 11s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/aac600fd6a9e4eaab3d74969470505aa : SUCCESS in 35m 09s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/dfa1184dcfa14d22a5f7ff317f64435c : SUCCESS in 36m 58s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/94106f0a8ce54c159ab4f30c3dc4bf33 : SUCCESS in 1h 26m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2273e6f11a544d29ab249bce4fb9865a : FAILURE in 37m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00d0f749ef36456c96345792ab50ebdb : SUCCESS in 2h 07m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74f88b21db804ba0b5cb6e4c280dc547 : SUCCESS in 34m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/71e4f7233bfc48718754619ec737f4ac : SUCCESS in 33m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/1dfb3aac1b31495898210de188e846ce : SUCCESS in 33m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/810b9136b7ff4202b959f460717185b6 : SUCCESS in 33m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/18bdbd74a3864d428b8810827ecfac52 : SUCCESS in 34m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/57532c9ad45849078f5d4700a4d702d1 : FAILURE in 44m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/d207b856063a439595361df850329d5f : FAILURE in 47m 45s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/bab1d244db2f4615b8f98fc1ec32d023 : SUCCESS in 34m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/bbaa5e8db51744d29c83afc5a77d5005 : SUCCESS in 33m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/e595809614ae4704927aa10facd758d5 : SUCCESS in 34m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/161d4573eda04b76bdd8b7e228491237 : SUCCESS in 2h 06m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f2b3d13ec17145e4b8ea2ceeb34b46e7 : SUCCESS in 46m 25s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/ba9b3a6c7a2e43bf81dcb6f041d220aa : SUCCESS in 52m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"690e53420c862bf97b96aa6c1849016af98f97ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:32:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e44c76488de71d18d4065d76c7c252fc09d93567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:06:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb2e282e0ef4c9eb9bf88a6aa19695e : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74ce07b71b444d2a9377f2546decc760 : SUCCESS in 4m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/abaaaabfb087493594092188b6d8c5ca : CANCELED\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/2dc5376b9aee467db2d78d0c58320115 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/5b6d5f414ac34c37848ea4510f26d9f0 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/eafc1fddbf664ff0924d9e8527670e41 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/4648ab1c3df94f3ea44d00a8af88f3e3 : CANCELED\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/293113f316fb4529a47f3e30be5f2274 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6bf9a0b8768940be8c11a88882f8dfb4 : CANCELED\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/1fe7a0d11ad54acb9e549157d220673a : CANCELED\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/037754e9c599454490762d4afb96b797 : FAILURE in 1h 30m 31s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/98deed80e7724a6c975ce8cc4a22f8ea : CANCELED\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/335069491f0f424f832d90f27f8aa533 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3f7c84547d84b24a403c5c6368aa2c5 : SUCCESS in 47m 55s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/4846851690a24911a3cb52e89ed0d2f2 : SUCCESS in 52m 22s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/ccad4ef9e9d8441fba7341881ef7712d : SUCCESS in 45m 49s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/f9e08750fbd34568b08de4e3a68c4126 : SUCCESS in 44m 19s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/870bb954880b4e04ad4e2b4faed45d3f : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":2},{"id":"0f3fdcfbad583f586a97893a1c0d09a3d75daf6c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 06:07:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2715c0c739863bd05febcb65de7d4dbc79cf213a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:47:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b45717729864393b0eb142c0ed852cb : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35fcfe40044f44729cdeaa965d812184 : SUCCESS in 3m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c91e4f701cae4e18b9ad3708cb889d5e : SUCCESS in 2h 01m 49s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/acafc1ce20e148a1b79d5631f92dbd55 : SUCCESS in 2h 20m 34s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/4df73b1818ba43b79b4765d80712e005 : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/01e18ea28a824a03b6319cdaed988dd9 : SUCCESS in 1h 20m 40s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/abe8259b601b419dbad4fa15f340355e : SUCCESS in 1h 37m 28s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/6656b7bd5ed8460588ca4ccb5e56681d : SUCCESS in 1h 59m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a5defcb944b458badccc8ae775cc4da : SUCCESS in 2h 02m 29s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/d19a9b84eed241c18a6e3bf218bcdcf9 : SUCCESS in 1h 41m 17s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/3e6373e166554406ac58d25f0daf0b0f : SUCCESS in 2h 11m 07s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/e486379bea194984b7effa655cbb4596 : SUCCESS in 1h 27m 53s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/ff21026a0ac345e5b1eaee82a226e16e : FAILURE in 2h 11m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67d6c41d5bc1484a8feca6ffee39c448 : SUCCESS in 33m 17s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/2ae649a10e5c40efb4450d56005208e5 : SUCCESS in 44m 26s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/288aeccf7b334865a71dd05047a75a28 : SUCCESS in 33m 51s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/cf3615476ad5406d8b4961026724ef34 : SUCCESS in 39m 16s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/0218a9c8d5004afa929320c70ddbf957 : SUCCESS in 34m 17s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b5949c61b0a04e93b9b782ff5a9b3edb : SUCCESS in 2h 00m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d9a79e9a9cb40bc802d97759e919b9e : FAILURE in 27m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/acf5b8de8b6c48c5906f1dc065f6242e : SUCCESS in 2h 02m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43f8cca72abc4b3e893462883e7898be : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/fa1cf0e287df45f6823620099b6fa493 : SUCCESS in 46m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/91970925bd0f473480057f160458a72e : SUCCESS in 43m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/2bd21340cd864fe1b016e8abc08c0c0e : SUCCESS in 38m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/bd8fb1eecef1478bb8b3b3470b9966a8 : SUCCESS in 33m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/04568c8917774579a92de35649f3c513 : FAILURE in 54m 14s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/04f30dd34dba4fdfbe22eb3fb264f40a : FAILURE in 52m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/25cda94141754e6f9e9ec9de94ab0d5c : SUCCESS in 31m 08s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/d421c32a56f74ee1a4ed03d0e5c5307c : SUCCESS in 37m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/a6c9d9f5de6b4c9fbecf102051aaa80a : SUCCESS in 44m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ca5a7f0c0a534de797c5b0c6c882d86e : SUCCESS in 2h 00m 06s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d4d27b1e35d643c68cf870ddde89e2ea : SUCCESS in 34m 18s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/1bc4fe91263d4f528cbfdbf461be4c1e : SUCCESS in 1h 03m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4d2ead48991b3b3cbfb9b4ec08dc00ee138880ae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 08:53:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c6e28afe651a7eb7c56b23ef128140576326766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:35:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3377d0b2a7fa46fea97409ce97b6634d : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5f63cff30a64d82959bb256416451a4 : SUCCESS in 4m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6f70ea80d444845956b191e5b2a8e48 : SUCCESS in 2h 06m 39s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/1e789878b25f493eb4d10a0a0fcd678c : SUCCESS in 2h 37m 47s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/3c75c8031e8a477c8a549e762cc6cacb : SUCCESS in 1h 26m 17s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/8b00f9b415ec44ca99850d16c12072be : SUCCESS in 1h 41m 47s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/26cc173399c54090a0dcced35a5fa8dc : SUCCESS in 1h 51m 36s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/7f80c603b8e24b1d93206ee1b47abd04 : SUCCESS in 2h 18m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ab763896c194cbfa6abd0849720e0fd : SUCCESS in 2h 25m 27s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/d62f32e5909f4567be5e590221c80801 : SUCCESS in 1h 51m 49s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/c8b6ec8a1dc84631b737c7cb77ea86b4 : SUCCESS in 1h 27m 45s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/5da42fe6914d44198dc3eafa2feac84a : SUCCESS in 2h 05m 46s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a91b3d09fcf1410683b74c9933eb2bb1 : SUCCESS in 2h 01m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a478b07504b4e7c8210145377c2dd92 : SUCCESS in 49m 19s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/abb385ea4ad345db9b2d94d212e85401 : SUCCESS in 51m 26s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/546567944a4645ddade9213ee3939605 : SUCCESS in 52m 27s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/6ecfee51b6c7439f9b215365dc795c9b : SUCCESS in 37m 42s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/b24b18ed3f404f0ab5c955ddeedf8ebe : SUCCESS in 34m 50s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/af4066bb3ec24108ac5231d38233a806 : SUCCESS in 2h 02m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1257c4f6f77486b84f7b0e003fd8445 : FAILURE in 27m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d74cb36c761428c9ed2dd6093932def : SUCCESS in 1h 58m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93066249fe2544eabc9dd339003b8085 : SUCCESS in 31m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/cf637e0892f54424bcdf1c95bcd7eea7 : SUCCESS in 34m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/5bf63fe093db437099b2c63c3b8de214 : SUCCESS in 31m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/b029f14fcfde405f91935b13153fbb4a : SUCCESS in 33m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/1db517de727847648455c3ed6165ce16 : SUCCESS in 35m 15s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/48842116d93c44d7a8b3942ee0e251ec : FAILURE in 41m 40s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/b62139a4a1b7414e8f9eec25d15d15f2 : FAILURE in 45m 44s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/e25479b0380a4fe88ac776d94a61901d : SUCCESS in 30m 46s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/bcc7173305814e029b8bfc0e5979d1a7 : SUCCESS in 30m 50s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/16296f0f039b4370a87b30efee3c41de : SUCCESS in 30m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c242867fc0e74b429e322ccea4d7dbf8 : SUCCESS in 1h 31m 22s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/bab3bddadfea4cef9b48e7272e622289 : SUCCESS in 47m 21s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/20f0e02b13c5444ab4a3a0cbe13c6154 : SUCCESS in 53m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"008749335f48dfd7b2642dc037fb1bdbe310299c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:46:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76109293f130eb8093195c019772a228d09aa07d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:05:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a220212576ed4207a5e246ddf9040973 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/992e5bb62e434e42a003077a0ffacf9e : SUCCESS in 4m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b82e3f6992a94e8ab45144b73f4d75b1 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/9ceaed557e754498931a412e7069b674 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/327ecf1050154ac3a4967163094d8312 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/6cf6db5c4458403e89dac219bb0997c1 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8a3e087c921f4a539c8561ff301920a5 : SUCCESS in 1h 12m 54s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/28735177aae34b72b5242df92e9598bc : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba1be45237054d3d8d08f2e1af81b99f : CANCELED\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/bc19c91085ef42c9b660dd4795c86fcc : CANCELED\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/94b253ee0bb942c589e78bdf9b2682b8 : FAILURE in 1h 18m 46s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/67597142babf4846bcb171adba1dede7 : CANCELED\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/a82cc1641eee4339aebe2ef8705f50dd : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/921f5348fe3b4afca6379b66ad6c92ed : SUCCESS in 35m 53s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/6a2f585999da415d97129c97ddff4f7e : SUCCESS in 38m 43s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/21201ca6243c4de79790995174f2a415 : SUCCESS in 36m 29s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/267e6c81b63a401ca54c12d8f4e2f960 : SUCCESS in 37m 18s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/776a478c86574e2a917969c43a89d275 : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":2},{"id":"83aa93e5418fc1cb92df1d0a58043474b53f4a48","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 18:11:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"239bf00d175e03abb61cb2cf959ca0b94a6a5ef2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 20:59:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9112040eac2e4552b14d5fee9cfc14f5 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8529b61afbda418ca5dddd94a0d10ed2 : SUCCESS in 5m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c089fe6b2fa94819b981389c1d22f1fe : SUCCESS in 2h 14m 34s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/25f1aef32ba74e889d5e789fe57c760e : SUCCESS in 2h 42m 12s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/367c6e4240714f269b2a9c1710183a61 : SUCCESS in 1h 44m 16s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/2fa4f7cd9185405bac1e6378faf53e02 : SUCCESS in 1h 23m 26s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/5ddb4dbc193e4f77bc3675b9a6229d24 : SUCCESS in 1h 04m 59s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/4ba6d572d3634a3eaefe72ae7354517a : SUCCESS in 2h 37m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e4acea355c848dda30b6c7c238957fd : SUCCESS in 1h 45m 32s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/a8a87fd3e1c942c784a77ec3d7805af5 : FAILURE in 2h 08m 09s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/50bf3fb6af4844f080f044b90b3a751d : SUCCESS in 1h 15m 03s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/d7977264ac0c41d6b7f3f7ce80c74b70 : SUCCESS in 1h 42m 59s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/65355e800a53468b8132698f902770a7 : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3f24089c125b4102a44f948cd1b5f13d : SUCCESS in 46m 29s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/5deeba92459e4bee8d56c984c2f0fce3 : SUCCESS in 38m 09s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/aecc17b4713e4517876ce439b19d9e64 : SUCCESS in 49m 35s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/b2974634f56748ba9f599c8d59311a36 : SUCCESS in 41m 16s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/3d2f6eba16d448d69d7c0541656236d9 : SUCCESS in 51m 08s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a2d3629f5734e9daacd74ea6c9bc4a7 : SUCCESS in 2h 06m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed53efe565b945d3a58cba20f08e6c76 : FAILURE in 35m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6efa20c8216e4aeb9d9eecf0ec070132 : SUCCESS in 2h 32m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4dd68f2b0e7b42d4b9c50e51debbd27c : SUCCESS in 45m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/a53d5482af5f489b97de9ecae7098d13 : SUCCESS in 42m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/637f45e8d7f64119b706551d93934ebd : SUCCESS in 35m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/9b2cb10d33ff42f6aff10c33bb1362b4 : SUCCESS in 38m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/c9153b2c85184712b87793ff58974c3f : SUCCESS in 47m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/abdcd6246afe4d78bbc5ce379b1ac1dc : FAILURE in 55m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/ab455d631e4d4abba65cd50dbdeaaaa6 : FAILURE in 44m 12s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/0f88ce3ef3434d2ebdb2ed8e119041f7 : SUCCESS in 31m 28s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/52e318a71c584ff698af26898edb6097 : SUCCESS in 45m 23s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/f6de95a4f6504aafa3355ea0c8ef8c43 : SUCCESS in 40m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0574a9bbdbd84075bc9ace8b2c43e463 : SUCCESS in 2h 11m 48s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/72626e3e586d4e32bb33991f2c4fc8d7 : SUCCESS in 41m 45s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/d007e8b397c04e918e1d8e7f9867b0d5 : SUCCESS in 48m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"677ead4f9c7a7175192c0afbc3479bf8d5379fe6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-07 06:07:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b90512e11f4f4d4406ee0c3b8cf3356fb0855dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 08:30:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afeca1a2863e43d69615646af5dde899 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348a5f853237463b95c7d5d0d8496fca : SUCCESS in 2m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f536ca07fd4341a9a74da14e52a351e9 : SUCCESS in 2h 16m 47s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/566db0b21c994de58e87e5b085501215 : SUCCESS in 1h 56m 26s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/05ab57441e144ab98ed2dead1ca5648e : SUCCESS in 1h 40m 12s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/65e5d9334b4b44249d65e5a1d1609645 : SUCCESS in 1h 31m 47s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/8d8299a3bd424e54ae229d8e33c69808 : SUCCESS in 1h 05m 15s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/3faaa7f6db8d45d09660ae2705cd10b7 : SUCCESS in 2h 15m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89eb48262467403e943fcf7b8591317f : SUCCESS in 1h 16m 44s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/ffffd315e7924601908995d07ad38ae1 : SUCCESS in 1h 35m 33s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/daed2748d2194fe1babbe4406080e710 : SUCCESS in 1h 44m 30s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/7460cd5cd8c345fcbd11a53f2bd013a7 : SUCCESS in 1h 30m 41s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/cd1416d7c27b4fb8a8bfee50b7a4c43c : SUCCESS in 1h 16m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6701c45b19d343b796bc819a8cb156b4 : SUCCESS in 33m 29s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/f722add5b7d1409a81818ec3c6b8db36 : SUCCESS in 32m 52s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/51c7a8b0dd4745f7af668b98b97ef008 : SUCCESS in 33m 26s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/99f0377be9b04452a72336626022241a : SUCCESS in 33m 08s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/35746164955a4ad9bbc1008cfcc55c2d : SUCCESS in 39m 36s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/22078e95f413411bb51c6397c990dd57 : SUCCESS in 1h 25m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/550dc9f0b43f43628c15a8a4c7224a96 : FAILURE in 26m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/209bf3f15dfe4566831cfe3f69528bc2 : SUCCESS in 2h 03m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f969e8c406f41d7a549a442939f753d : SUCCESS in 31m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/cf93f8498abb49f4b2de713d04520ef4 : SUCCESS in 32m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/d05a536d792e42e1a10c04855b376e6a : SUCCESS in 30m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/963e7fab8d854a4d86bb2c6fef853326 : SUCCESS in 32m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/39591e5e81be4071950f976feb3afd0d : SUCCESS in 31m 09s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cdca58ff7ca34c3d99ccd7f8a041b05c : FAILURE in 43m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-xena https://zuul.opendev.org/t/openstack/build/9c6de390ef26411eb65c9e1b6e107775 : FAILURE in 40m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/30f632afd68048a2b612034d45179152 : SUCCESS in 30m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/00de54f462964fb78687c5dee8a24488 : FAILURE in 23m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/fdba2623aa3948978e78feaa314fc228 : SUCCESS in 31m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/57c9f145224f4753950a2e60ff583c12 : FAILURE in 1h 27m 21s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/1cc7eed0b3bc4076b871181abbce68bc : SUCCESS in 48m 55s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/14537111b10f4481a9306a702739b92c : SUCCESS in 52m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a01703ef6ab13f6664244ddc3a174c6ebec1e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 08:31:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebca7809dd1bc0c660a62c86566f6b4b9fa9b07e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 10:58:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d132d81eebf4b82943dfc6170e99d8e : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca6dee3b6c1481aa4a3cd03ae9b285a : SUCCESS in 2m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff12ba9000e745ab8189f3edc5b95f6b : SUCCESS in 1h 59m 26s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/86fb3a77d8f24e35b0b23460c27366ad : SUCCESS in 1h 42m 19s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/f2c522cd7f0545c88ef82a6a48acf293 : SUCCESS in 2h 00m 09s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/8d5faa5842934bdca318f2e3d6e080e9 : SUCCESS in 1h 19m 12s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/f4b905514bfd42f0ae000a5d0059f27c : SUCCESS in 1h 31m 25s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/f3b074fc0d864331af4818e9d1bb88b9 : SUCCESS in 2h 11m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d0e4ffb2af54bfaaa6cbfbd54ece65c : SUCCESS in 1h 56m 24s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/c8e9550241764995b43496e0b81d4b70 : SUCCESS in 2h 25m 46s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/1076c2377ac64c96b05a59024dd5f7cd : SUCCESS in 1h 58m 12s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/c9ad636bb4a74153baeb0fd330d7bce0 : SUCCESS in 2h 24m 44s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/2baec5a939fe4c14af49ec97502ee99c : SUCCESS in 2h 03m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/27b41ee6d80c4693880a7b9a4a7f3f2e : SUCCESS in 44m 41s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/2ec1cfec621c481384e36a866951db8f : SUCCESS in 44m 33s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/1acbddf3628441009fe713dce3e82440 : SUCCESS in 47m 08s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/77ab1c2a94764254a18cb977e6318dea : SUCCESS in 50m 59s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/bd595aa49931480293b7558b78f3883b : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":2},{"id":"15818a4ec8eff14ef072893eb5ea80f3887a474a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 10:58:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ab6fbdc3ac88d2b65340ff4d4f48b1958989b5e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 11:04:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5596d205e1b460cb9e062a621b83a68 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b04038940d5b39539f55555c8bfe7f9d1646ccb","revisions":{"4b7852e63ce1bcc0f64ec222b337cddfd8cf2492":{"kind":"REWORK","_number":1,"created":"2022-01-19 01:43:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/825208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/08/825208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/08/825208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/08/825208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/08/825208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/08/825208/1"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-19 01:37:30.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-19 01:37:30.000000000","tz":-360},"subject":"Add stable/xena jobs on master gate","message":"Add stable/xena jobs on master gate\n\nWe have stable/xena released so we should add\ntheir job on master gate to keep branchless tempest\nplugins compatible to stable branch.\n\nThis also removes the stable/train job as that is in EM\nstate now.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: Ia1ecf95d269af2f6edbd562f31fb69c2cc884cf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4b7852e63ce1bcc0f64ec222b337cddfd8cf2492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4b7852e63ce1bcc0f64ec222b337cddfd8cf2492"}]},"branch":"refs/heads/master"},"6b04038940d5b39539f55555c8bfe7f9d1646ccb":{"kind":"REWORK","_number":2,"created":"2022-01-27 17:37:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/825208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/08/825208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/08/825208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/08/825208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/08/825208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/08/825208/2"}}},"commit":{"parents":[{"commit":"7de0490e79a450a3fabe3188bf7a9af16c554c2d","subject":"Merge \"Switch centos-8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7de0490e79a450a3fabe3188bf7a9af16c554c2d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-19 01:37:30.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-01-27 17:37:57.000000000","tz":0},"subject":"Add stable/xena jobs on master gate","message":"Add stable/xena jobs on master gate\n\nWe have stable/xena released so we should add\ntheir job on master gate to keep branchless tempest\nplugins compatible to stable branch.\n\nThis also removes the stable/train job as that is in EM\nstate now.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: Ia1ecf95d269af2f6edbd562f31fb69c2cc884cf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6b04038940d5b39539f55555c8bfe7f9d1646ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6b04038940d5b39539f55555c8bfe7f9d1646ccb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
