)]}'
{"id":"openstack%2Foctavia-tempest-plugin~827811","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I5c2db87975803b30ea230c3bbf5dab4b96da4614","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"28609":{"account":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"last_update":"2022-02-09 15:30:24.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-09 15:30:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c2db87975803b30ea230c3bbf5dab4b96da4614","subject":"Fix octavia-tempest-plugin on old tempest releases","status":"MERGED","created":"2022-02-04 07:59:10.000000000","updated":"2022-02-09 15:42:53.000000000","submitted":"2022-02-09 15:30:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"827811","meta_rev_id":"7be996c358aacef0c14e34c26c5386458fb06926","_number":827811,"virtual_id_number":827811,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 15:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"all":[{"value":1,"date":"2022-02-04 13:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2022-02-04 12:13:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":1,"date":"2022-02-09 10:08:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"value":2,"date":"2022-02-09 12:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2022-02-04 19:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-02-04 14:52:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"value":1,"date":"2022-02-09 12:40:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 12:13:19.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"REVIEWER"},{"updated":"2022-02-04 12:50:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 13:58:38.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-02-04 14:52:00.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-02-04 19:36:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-09 10:08:13.000000000","updated_by":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"reviewer":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"state":"REVIEWER"},{"updated":"2022-02-09 12:40:14.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"1c4a6224f18d3df70054c5a955e659fe408c236b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 07:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f92b43626d3aa2712fe0d50004c7816cbe2f71dc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 09:48:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"6cf059c9cc5575e40a90d0204252f18e69c20d64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 09:58:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b401d716b05e0a4a4f3d062414fcfa536e397915","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2022-02-04 12:13:19.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f781bbd0689aeb6260f759b1a0a6a642a9d8b2bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:50:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8af63e9c5341ce96e14332ee7193e3 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27466ed2e35042c0bc78818761928bba : SUCCESS in 4m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f7a09754b39b43ae8d0e29f46ed54e68 : SUCCESS in 2h 36m 35s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/44248795a18a4777936db0a65fbcea92 : SUCCESS in 1h 43m 06s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/a7b40b1185514780bac1da01c9fe005a : SUCCESS in 1h 42m 11s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/186580a9091c49af935c31cc23b3d1e2 : SUCCESS in 1h 04m 08s\n- octavia-v2-dsvm-noop-api-stable-train https://zuul.opendev.org/t/openstack/build/2e1a506e105343018a08d10b6338a186 : SUCCESS in 1h 03m 00s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/98bebe595cc84ac9bdcfd2d618142cce : SUCCESS in 2h 42m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa7a368f59d94b52a2af3919625ad7a4 : SUCCESS in 1h 53m 08s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/26bb0047b87340569954dd6905dfe347 : SUCCESS in 1h 24m 52s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/653cd607e3ae49fa9e789fdf60425e51 : SUCCESS in 2h 02m 14s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/25edbbfb7e8348c8a7181fdaa9718206 : SUCCESS in 1h 46m 15s\n- octavia-v2-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/1913deecd6a049b094ab39d110ceabba : SUCCESS in 1h 42m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/716f0d83b14742cc874af1042e105b41 : SUCCESS in 42m 08s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/7bdc6aa860804edcbb60c2b24924eafd : SUCCESS in 57m 13s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/2f51b1adef1241bda97ec6b709190b54 : SUCCESS in 41m 33s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/f4d085a4634e47b28d47957b64429594 : SUCCESS in 43m 53s\n- octavia-v2-dsvm-tls-barbican-stable-train https://zuul.opendev.org/t/openstack/build/56fefb6740e5419fa6dfa7c26659afaf : SUCCESS in 44m 25s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c2b2de56dc4844fe8f16df83f78c16a4 : SUCCESS in 2h 08m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/1184b73f0367447993cd2e993887d55a : FAILURE in 30m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47b3a981616746adbd38f02a845ca5da : FAILURE in 2h 33m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf7e42140e784ad29b312a81a7c471d6 : SUCCESS in 49m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/73119b56a6c74271bbdd27638baa96c9 : SUCCESS in 38m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/7223a05de43a448fb45fcea58249efd4 : SUCCESS in 46m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/0e37bb94bbf2466daff60c184b78ba60 : SUCCESS in 39m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-train https://zuul.opendev.org/t/openstack/build/cbe48508b15e47b6a5721b9be7f9568a : SUCCESS in 32m 43s (non-voting)\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a818062f8588474a8d9ec2cd1dc558e9 : FAILURE in 46m 56s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/3c13cff42b2d453681638fe0728b2a86 : SUCCESS in 49m 18s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-victoria https://zuul.opendev.org/t/openstack/build/34aada829ea24e868cdd96c4f82bff58 : SUCCESS in 35m 13s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-ussuri https://zuul.opendev.org/t/openstack/build/7c06f96b04714a35abd5e2ba5d9cff88 : SUCCESS in 54m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-train https://zuul.opendev.org/t/openstack/build/8f06cea5641f485ab589e3c2d9030029 : FAILURE in 34m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ed6ff4b07ab84ef19e623f6743f852d0 : SUCCESS in 1h 25m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f01b919d53b944a18fe5a7a0d985e622 : SUCCESS in 50m 59s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7d3f7c9636534e76bde3841802f5fd71 : SUCCESS in 51m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"559c85908177412079f252c3bbe08d655694fb25","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-02-04 13:58:38.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cadaaa0927a68d72b5dae6de5b2c70989555fa43","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-04 14:52:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b614c19b4fa6c0fb9ffe78098f4356aaea62c5ac","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 16:31:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7bbd087c21494a416ade1824147d9b5ab0b88df9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-04 19:36:58.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"95a14a721931bdcee345533c698b9e4a59da09ba","author":{"_account_id":29681,"name":"Bruna Bonguardo","display_name":"Bruna Bonguardo","email":"bbonguar@redhat.com","username":"bonguardo"},"date":"2022-02-09 10:08:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f3c75576e9a2cd9bf902732d32ccd55b2788f581","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-09 12:40:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d494b5cbe2b2bf91f906c668c5ac12ae078aef27","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-09 12:40:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7df2551a10aa13b813ac19dd3c169a2e89bc1647","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 12:41:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b481596a9cfb94e898638a82e49d739c804e4e06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 15:30:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f1a93295614f2d9a44ae339e1efa8c : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5c30ca528d4ad585202989a0cdf855 : SUCCESS in 3m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46b96e88135946a8b35608c2adcda946 : SUCCESS in 2h 28m 14s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/3977c27c9f02411d936efd4a16c5c9c7 : SUCCESS in 2h 20m 22s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/78852886f46b485aa4b104ade7c6097d : SUCCESS in 1h 33m 36s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/f6de95170e5e4bae91577aab7227460b : SUCCESS in 1h 38m 20s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/dc81522f845244d5a11d867fd7ded2e8 : SUCCESS in 1h 19m 03s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/9abf83932da04ef9b3e16faea0861789 : SUCCESS in 2h 42m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffbedbd4da974c2898a70d06d7b79d04 : SUCCESS in 1h 40m 59s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/f77aa278b91f427387563f14b7bc27f6 : SUCCESS in 2h 00m 14s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/d0662f8cb4aa4fcc90d044b90e6dee9b : SUCCESS in 1h 23m 17s\n- octavia-v2-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/1ce11d69f92a4c328807aa4608238541 : SUCCESS in 2h 01m 33s\n- octavia-v2-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/925041abe8d54d7fb6816a4f13dc1b7d : SUCCESS in 1h 30m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29a534f282c74cfabce45f704d5a8fe0 : SUCCESS in 36m 28s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/a5f1450f490148deac6ef1c24e0e4caa : SUCCESS in 35m 48s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/97d9c074c3cf4acfa9a8d93e9bf12fc5 : SUCCESS in 33m 26s\n- octavia-v2-dsvm-tls-barbican-stable-victoria https://zuul.opendev.org/t/openstack/build/f4f161c0a22d4f2fb0f98cc003828c40 : SUCCESS in 41m 31s\n- octavia-v2-dsvm-tls-barbican-stable-ussuri https://zuul.opendev.org/t/openstack/build/9782152f161a4aadb7adfbb02c0a6f9b : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a390870a8ecc0becdb7d21426b2c3010a84dd720","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 15:30:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7be996c358aacef0c14e34c26c5386458fb06926","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 15:42:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cba1b3af65347469eb634a84335e763 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b0da4f3ce5142a41c1b20651f5d457fc9c100054","revisions":{"625ba80b32e096c9e9e93040bf68a73e9438edc1":{"kind":"REWORK","_number":1,"created":"2022-02-04 07:59:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/827811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/11/827811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/11/827811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/11/827811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/11/827811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/11/827811/1"}}},"commit":{"parents":[{"commit":"d0ea406ee45399bf329d82e403889a4ceeaa0a39","subject":"Add type to allow ECDSA keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d0ea406ee45399bf329d82e403889a4ceeaa0a39"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-04 07:58:06.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-04 07:58:06.000000000","tz":60},"subject":"Fix octavia-tempest-plugin on old tempest releases","message":"Fix octavia-tempest-plugin on old tempest releases\n\nThis is a follow-up commit for \u0027Add type to allow ECDSA keys\u0027\n(I34ac429ab5442cef056ee8b63fcb2ba41e8b9b27), this commit allows\nusing octavia-tempest-plugin with older tempest releases\n(ex: in our jobs on stable/train).\nIt catches the exception if [validation].ssh_key_type is not\nsupported and returns its default value \u0027rsa\u0027.\n\nChange-Id: I5c2db87975803b30ea230c3bbf5dab4b96da4614\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/625ba80b32e096c9e9e93040bf68a73e9438edc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/625ba80b32e096c9e9e93040bf68a73e9438edc1"}]},"branch":"refs/heads/master"},"b0da4f3ce5142a41c1b20651f5d457fc9c100054":{"kind":"REWORK","_number":2,"created":"2022-02-04 09:58:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/827811/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/11/827811/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/11/827811/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/11/827811/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/11/827811/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/11/827811/2"}}},"commit":{"parents":[{"commit":"d0ea406ee45399bf329d82e403889a4ceeaa0a39","subject":"Add type to allow ECDSA keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d0ea406ee45399bf329d82e403889a4ceeaa0a39"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-04 07:58:06.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-04 09:58:09.000000000","tz":60},"subject":"Fix octavia-tempest-plugin on old tempest releases","message":"Fix octavia-tempest-plugin on old tempest releases\n\nThis is a follow-up commit for \u0027Add type to allow ECDSA keys\u0027\n(I34ac429ab5442cef056ee8b63fcb2ba41e8b9b27), this commit allows\nusing octavia-tempest-plugin with older tempest releases\n(ex: in our jobs on stable/train).\nIt catches the exception if [validation].ssh_key_type is not\nsupported and doesn\u0027t pass the ssh_key_type argument to tempest\nfunctions.\n\nChange-Id: I5c2db87975803b30ea230c3bbf5dab4b96da4614\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b0da4f3ce5142a41c1b20651f5d457fc9c100054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b0da4f3ce5142a41c1b20651f5d457fc9c100054"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
