)]}'
{"id":"openstack%2Foctavia-tempest-plugin~828963","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I2d2cf910bd5801bfeb92c1ede51bd120ced3b4f9","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-12 14:14:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d2cf910bd5801bfeb92c1ede51bd120ced3b4f9","subject":"Spliting member API tests in 2 classes","status":"MERGED","created":"2022-02-12 18:02:10.000000000","updated":"2022-09-12 14:16:08.000000000","submitted":"2022-09-12 14:14:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828963","meta_rev_id":"ff2d459e9de1a6fa3ea23c505cad524a291fd944","_number":828963,"virtual_id_number":828963,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 14:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-24 19:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-09-12 11:49:08.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-09-12 11:49:08.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-12 19:32:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 16:57:55.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-02-24 19:55:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4ed303ea692828ccedca748048f3b084917ac4a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-12 18:02:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a1d41bcdace429b3187677fd162a6c3e5ab478f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 19:32:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5cb7252aeb34d0e8b988f73400903c1 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6358a693bb44e018bc119bdc898e4df : SUCCESS in 3m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f4b4dca8e8644beb5793cc0f0bc08fc : SUCCESS in 1h 17m 51s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/9ac5f5ff3ee5481792f2d5d19ffee7ba : SUCCESS in 1h 15m 26s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/072af4a29136439db92b2734b212267a : SUCCESS in 1h 05m 50s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/37cf0f1ad2d7452280796d724fbb1a6b : SUCCESS in 1h 19m 29s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/a36ee4ccc7004b27954ed6ab6d175052 : SUCCESS in 56m 02s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/0613e5af00b549d99bf1ec3a4c968e0c : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4dbb1fba4472470012798d326cd3ba3f7132d765","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-12 19:37:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9585407403db70e3b5da3a2dd7bd8f65cddcf97b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 21:07:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80b42c99ccf444fbdb16769ecedabe7 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/546073ee42004868b3e502b2a9958ed2 : SUCCESS in 4m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9be318daff0f4665a7c28787786c21f7 : SUCCESS in 1h 06m 41s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/f0190eb9b5d94815b5348b65c374e19a : SUCCESS in 1h 29m 00s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/42aabd2cca214765850de3b7bdae2c32 : SUCCESS in 47m 08s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/d697994a4ef6467385465981335383f1 : SUCCESS in 46m 09s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/56a86039ce7e414bad956a1766117542 : SUCCESS in 40m 46s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/338ed30ee73f46c2871623aa2ad68383 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f5bb140dd1f52ea49fdbc317a35bc51674d0f3bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-14 07:09:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd33fed312c271180622690671b47ab201fa67bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 09:49:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59464dd8af8465c8bc5b348577f09ca : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b649ca0da65f496ab038f7348c29a76c : SUCCESS in 3m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/976429c2c2ad4f1eb79d21c646530580 : SUCCESS in 1h 10m 55s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/d101a2447d4f4b569fd85f186fff0544 : SUCCESS in 1h 24m 21s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/c21ac17693cc40fa893ba7ff19f01f23 : SUCCESS in 1h 01m 21s\n- octavia-v2-dsvm-noop-api-stable-victoria https://zuul.opendev.org/t/openstack/build/a5ee590680ea437db48e528ebc81b006 : SUCCESS in 1h 11m 51s\n- octavia-v2-dsvm-noop-api-stable-ussuri https://zuul.opendev.org/t/openstack/build/050dadfd05ab42aa9e6af4169b1f3075 : SUCCESS in 1h 06m 43s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/3e4f001c6bb9472db85002653cd05e80 : SUCCESS in 1h 35m 29s\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e7c22c139932411793d6b978d0d5a583 : SUCCESS in 2h 35m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b3522b24a7547b39826726d9deb48d2 : SUCCESS in 40m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/e28fd600107d4907928d0bf765e76e93 : SUCCESS in 39m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/41d7e249f7c140afa5524014cf7eee18 : SUCCESS in 42m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-victoria https://zuul.opendev.org/t/openstack/build/56b0a936e0bd4040967348c30765c9bc : SUCCESS in 42m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-ussuri https://zuul.opendev.org/t/openstack/build/516b990c261b47ce8ea3025373172b84 : SUCCESS in 43m 39s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4b04efc056414efcb9aba829d3fd75a9 : SUCCESS in 37m 50s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5418d65ace0e478e8782ed2800dc527a : SUCCESS in 47m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f820188106412576d5f3e188f3b631da11b1f415","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-16 16:57:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"48f3e7e9632ccf18c067dc92018894ccb6fb3d69","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-24 19:55:01.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f95a6a6f9ad226610ecaac8583603d09955ce2a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 11:49:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"206cf5b6f51260f9976b07a09ffe132ba37896a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 11:49:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a97032b317bba327455c037929f1fce8bb56718e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:14:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b25e4acc859744d98f00e34e1a0aa0a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2760d8e0b1d34de09beb6c3f14cf4286 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d244215bf1414ca439c895396855b5 : SUCCESS in 3m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63324d5e1c5143bf8b46aeff5da63acd : SUCCESS in 1h 42m 45s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/39ba4e422c5242e880b11cfe2db60a6c : SUCCESS in 1h 27m 00s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/890c7bdb2347444a8ecbc1f1d6617362 : SUCCESS in 1h 17m 13s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/0d4f78fba79b4e5fb6eed5e099d68fb8 : SUCCESS in 1h 05m 42s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/2b711075511d4a67a6149ba6330a839c : SUCCESS in 1h 43m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c93155680dffd6957f1bbe2f56cbf454779eb0f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:14:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ff2d459e9de1a6fa3ea23c505cad524a291fd944","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:16:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af19c2eac70f429eb6ac75a727618918\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5313792a6448469e1a9eff0df95efd : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16b6906d5341cdc0c6d340a2ce6640ffd6d24f2f","revisions":{"2bb36c756b9e6e23310b3e0d54e542b8fa5c38a7":{"kind":"REWORK","_number":1,"created":"2022-02-12 18:02:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/63/828963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/63/828963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/63/828963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/63/828963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/63/828963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/63/828963/1"}}},"commit":{"parents":[{"commit":"a3a95b132eac0b37fe85835d89b195ac08436501","subject":"Merge \"Fix octavia-tempest-plugin on old tempest releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a3a95b132eac0b37fe85835d89b195ac08436501"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-12 18:01:51.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-12 18:01:51.000000000","tz":60},"subject":"WIP Spliting member noop-api tests in 2 classes","message":"WIP Spliting member noop-api tests in 2 classes\n\nChange-Id: I2d2cf910bd5801bfeb92c1ede51bd120ced3b4f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2bb36c756b9e6e23310b3e0d54e542b8fa5c38a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2bb36c756b9e6e23310b3e0d54e542b8fa5c38a7"}]},"branch":"refs/heads/master"},"16b6906d5341cdc0c6d340a2ce6640ffd6d24f2f":{"kind":"REWORK","_number":2,"created":"2022-02-14 07:09:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/63/828963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/63/828963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/63/828963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/63/828963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/63/828963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/63/828963/2"}}},"commit":{"parents":[{"commit":"a3a95b132eac0b37fe85835d89b195ac08436501","subject":"Merge \"Fix octavia-tempest-plugin on old tempest releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a3a95b132eac0b37fe85835d89b195ac08436501"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-12 18:01:51.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-14 07:09:16.000000000","tz":60},"subject":"Spliting member API tests in 2 classes","message":"Spliting member API tests in 2 classes\n\nSplit MemberAPITest class into MemberAPITest1 and MemberAPITest2\nclasses. MemberAPITest contained 240 test funtions and it took 1h45 in\nthe octavia-v2-dsvm-noop-api job to run those functions sequentially.\nBy spliting this class, the load now spread on 2 workers and it reduces\nthe duration of the noop-api jobs, avoiding frequent timeout issues.\n\nChange-Id: I2d2cf910bd5801bfeb92c1ede51bd120ced3b4f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/16b6906d5341cdc0c6d340a2ce6640ffd6d24f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/16b6906d5341cdc0c6d340a2ce6640ffd6d24f2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[]}
