)]}'
{"id":"openstack%2Foctavia-tempest-plugin~838168","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I8201da19d09e8fdabe50bff1e3bb5cc2d2e41b0e","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"multi-vip","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-08-26 07:19:59.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I8201da19d09e8fdabe50bff1e3bb5cc2d2e41b0e","subject":"WIP Add a job that uses octavia master with stable amphora","status":"ABANDONED","created":"2022-04-15 11:31:44.000000000","updated":"2024-10-24 07:16:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2257080a5448fa96cdac3dcc712e722329c5cb8d","_number":838168,"virtual_id_number":838168,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-17 15:05:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-15 11:33:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e8fb83ab4e6dd0799a36ae69d9ae8a05ed8e7ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-15 11:31:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e414df3f020abaab17b230e6009282d4eb430c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:33:07.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error\nwas:\n\n  Job octavia-v2-svm-scenario not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: octavia-v2-dsvm-scenario-stable-amphora-image\n      parent: octavia-v2-svm-scenario\n      vars:\n        devstack_localrc:\n          OCTAVIA_AMP_IMAGE_BRANCH: stable/yoga\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 571, column 3","accounts_in_message":[],"_revision_number":1},{"id":"5721405f7d29436892492e06f29195834f96dbb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-15 12:20:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4694e8b0770e1098f3ae2c3360c515a11c1b8523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 14:27:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-v2-dsvm-scenario-stable-amphora-image https://zuul.opendev.org/t/openstack/build/5360ebc169534f75acd3d21528407372 : FAILURE in 2h 03m 36s","accounts_in_message":[],"_revision_number":2},{"id":"bc16ef47ffa603b0a6e461b81cc40e7e536c729e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 12:57:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"df7f03e5f5f1b3a0ddf9ea7cc8da2b9b402af5ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:06:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-scenario-stable-amphora-image https://zuul.opendev.org/t/openstack/build/03ce417496a442caadb538cbd08f2ea3 : SUCCESS in 2h 02m 54s","accounts_in_message":[],"_revision_number":3},{"id":"a2c83e98e3b448d65730311562d82ba34112017e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-17 12:54:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"818ecaf34f4a55f744fac142324a37a95cf9d9c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-17 12:56:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9ccaf14ea31e620a1ca511f8f05b7877a2f59311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 15:05:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e4802ef77e048cb93fe6ceef2926286\n\n- octavia-v2-dsvm-scenario-stable-amphora-image https://zuul.opendev.org/t/openstack/build/5ec7afcd9910425bafa64e9c26e27339 : SUCCESS in 1h 55m 47s\n- octavia-v2-multivip-dsvm-scenario-stable-amphora-image https://zuul.opendev.org/t/openstack/build/ec020f46e4b74b03924d9b6aae1801ca : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":5},{"id":"c9986a801dc8409035102e9968819ba6dbdef2a6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-26 07:19:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a36971b36e8e167371d717188969d795c3139650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 09:05:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27686163097f4baaa85c8abb03830d34\n\n- octavia-v2-dsvm-scenario-stable-amphora-image https://zuul.opendev.org/t/openstack/build/c84b7f109ea449c097eb707486823c82 : SUCCESS in 1h 43m 41s\n- octavia-v2-multivip-dsvm-scenario-stable-amphora-image https://zuul.opendev.org/t/openstack/build/f72817898ad04a3e93db67590119e2be : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":5},{"id":"2257080a5448fa96cdac3dcc712e722329c5cb8d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-24 07:16:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"48fa53e2adc348e78b6f1011961fda7f3edb3153","revisions":{"bea5dac0c24f29e125e65ad410ff95204b2f7e97":{"kind":"REWORK","_number":1,"created":"2022-04-15 11:31:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/838168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/838168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/1"}}},"commit":{"parents":[{"commit":"4ef077b0594265d4c35902deab3512f76aef0e7a","subject":"Add jobs for stable/yoga, remove old jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4ef077b0594265d4c35902deab3512f76aef0e7a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 11:31:16.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 11:31:21.000000000","tz":120},"subject":"WIP Add a job that uses octavia master with stable amphora","message":"WIP Add a job that uses octavia master with stable amphora\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/838167\nChange-Id: I8201da19d09e8fdabe50bff1e3bb5cc2d2e41b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bea5dac0c24f29e125e65ad410ff95204b2f7e97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bea5dac0c24f29e125e65ad410ff95204b2f7e97"}]},"branch":"refs/heads/master"},"b33a42082ba921c5a395c9b2bf274f043e5f7e5c":{"kind":"REWORK","_number":2,"created":"2022-04-15 12:20:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/838168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/838168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/2"}}},"commit":{"parents":[{"commit":"4ef077b0594265d4c35902deab3512f76aef0e7a","subject":"Add jobs for stable/yoga, remove old jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4ef077b0594265d4c35902deab3512f76aef0e7a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 11:31:16.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 12:20:10.000000000","tz":120},"subject":"WIP Add a job that uses octavia master with stable amphora","message":"WIP Add a job that uses octavia master with stable amphora\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/838167\nChange-Id: I8201da19d09e8fdabe50bff1e3bb5cc2d2e41b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b33a42082ba921c5a395c9b2bf274f043e5f7e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b33a42082ba921c5a395c9b2bf274f043e5f7e5c"}]},"branch":"refs/heads/master"},"1f416f86504ab5d4a025ca1654115161ef8949c7":{"kind":"NO_CHANGE","_number":3,"created":"2022-04-27 12:57:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/838168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/838168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/3"}}},"commit":{"parents":[{"commit":"de762fa29e90d28ac05eca0e5db8a6223f7f4195","subject":"Remove old unmaintained jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/de762fa29e90d28ac05eca0e5db8a6223f7f4195"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 11:31:16.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-27 12:57:50.000000000","tz":0},"subject":"WIP Add a job that uses octavia master with stable amphora","message":"WIP Add a job that uses octavia master with stable amphora\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/838167\nChange-Id: I8201da19d09e8fdabe50bff1e3bb5cc2d2e41b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f416f86504ab5d4a025ca1654115161ef8949c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f416f86504ab5d4a025ca1654115161ef8949c7"}]},"branch":"refs/heads/master","description":"Rebase"},"82f6722330b01ef3695e34e57100cf5796ef568f":{"kind":"REWORK","_number":4,"created":"2022-08-17 12:54:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/838168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/838168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/4"}}},"commit":{"parents":[{"commit":"8e4e7593a2637fc1482dc2d92f5219e43e5352c6","subject":"WIP: Test Additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8e4e7593a2637fc1482dc2d92f5219e43e5352c6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 11:31:16.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-17 12:54:13.000000000","tz":120},"subject":"WIP Add a job that uses octavia master with stable amphora","message":"WIP Add a job that uses octavia master with stable amphora\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/838167\nChange-Id: I8201da19d09e8fdabe50bff1e3bb5cc2d2e41b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/82f6722330b01ef3695e34e57100cf5796ef568f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/82f6722330b01ef3695e34e57100cf5796ef568f"}]},"branch":"refs/heads/master"},"48fa53e2adc348e78b6f1011961fda7f3edb3153":{"kind":"REWORK","_number":5,"created":"2022-08-17 12:56:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/838168/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/68/838168/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/68/838168/5"}}},"commit":{"parents":[{"commit":"8e4e7593a2637fc1482dc2d92f5219e43e5352c6","subject":"WIP: Test Additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8e4e7593a2637fc1482dc2d92f5219e43e5352c6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 11:31:16.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-17 12:56:30.000000000","tz":120},"subject":"WIP Add a job that uses octavia master with stable amphora","message":"WIP Add a job that uses octavia master with stable amphora\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/838167\nChange-Id: I8201da19d09e8fdabe50bff1e3bb5cc2d2e41b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/48fa53e2adc348e78b6f1011961fda7f3edb3153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/48fa53e2adc348e78b6f1011961fda7f3edb3153"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
